)]}'
{"id":"openstack%2Ftrove~970825","triplet_id":"openstack%2Ftrove~master~Iec12b7f6c51fa8812d61787547803c4ed01dec66","project":"openstack/trove","branch":"master","topic":"bug/postgresql-fix-user-password-change-and-access","attention_set":{},"removed_from_attention_set":{"36080":{"account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"last_update":"2026-02-12 15:43:39.000000000","reason":"Change was submitted"},"28691":{"account":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"last_update":"2026-02-12 15:43:39.000000000","reason":"Change was submitted"},"31737":{"account":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"last_update":"2026-02-12 15:43:39.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-02-12 15:43:39.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2026-02-12 13:59:47.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-02-12 15:43:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iec12b7f6c51fa8812d61787547803c4ed01dec66","subject":"Postgres: Fix user password change and access","status":"MERGED","created":"2025-12-12 11:45:01.000000000","updated":"2026-02-12 15:44:50.000000000","submitted":"2026-02-12 15:43:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"970825-bug/postgresql-fix-user-password-change-and-access","meta_rev_id":"a4a77eab814a77fd0ba0617b50fa879617ad1378","_number":970825,"virtual_id_number":970825,"owner":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-12 15:43:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-02-12 13:59:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-02-12 13:59:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-12 12:00:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-12-12 12:03:08.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2025-12-12 12:03:08.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"state":"REVIEWER"},{"updated":"2025-12-12 12:03:08.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2025-12-12 12:03:08.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2025-12-12 13:28:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-06 13:57:13.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"REVIEWER"},{"updated":"2026-01-08 13:24:27.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"8ca1372e4a3e1e3e5339e6ee2385a7b24b7ef05c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2025-12-12 11:45:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"06668e64a7ca6ce31139192c787ed3c561080126","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 12:00:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5beab101b88d41d8b9f877ab411924e7\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c55ced8202df44dca0c044ada41f9903 : SUCCESS in 7m 21s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bdef219071564c86a3d988fc4332e895 : SUCCESS in 9m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ead7a65e6fcf4da893327b00023a4280 : SUCCESS in 12m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c3b9b7d37bb00ed2bd43948ddf0a333272be0063","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 13:28:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c2cf086f5a9744b1a29246795e950e16\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8ca2fde3a784dbe969959dcb05fc3a3 : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ea35440a7604938a010342cb1498f1e : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/46a0deaf7d664efaba2bd4f5b3cbe887 : SUCCESS in 5m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3534ef90933244b8b4c61976033bc396 : SUCCESS in 6m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77094e15214046c08e6b29dfa1b6eb26 : SUCCESS in 6m 14s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/81847e0cb4454232976c1bb42d62dec1 : SUCCESS in 1h 42m 53s (non-voting)\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/dbafac0d52ec47a5989ee1c2f6408120 : FAILURE in 58m 00s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/400716e074dc4c3782dd61beaa6e06fe : FAILURE in 43m 15s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/99c50e1c77cb4764aa7223cb58aa7b21 : SUCCESS in 1h 29m 23s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/3c0d14449ede4540b1248cce14bd038e : SUCCESS in 1h 16m 21s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/24f95ef99af94eceb83f7bc0eeb06033 : SUCCESS in 1h 15m 29s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/95ad33c9e9cb4bcaafc5ead4fea90211 : SUCCESS in 6m 12s","accounts_in_message":[],"_revision_number":1},{"id":"6c7031aa169279f440ad2f38881dafb1711d0bd1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2025-12-15 12:51:02.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"77702178cd50447c9b918adcb1c2a6fbdb5536f1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 13:02:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/373740711474473b98ddedb4323f8dac\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b8deda12ecb24b8b90ec784c5865c181 : SUCCESS in 9m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7e37f6b2730c4c94a812f57c1fd46aa6 : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/85235de833f94b76bf448676370a1d01 : SUCCESS in 8m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"703ed56ea88de7fe03d8035690b6c1f7ab904373","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 14:31:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9193b1b90f7439992b59af955190b99\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38c766dd35b749ebb30f752249cc21e6 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03ac6f38fb3148adb9ea04f2d61dd924 : SUCCESS in 5m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/13f38f6f47f1489da20325ad590936dd : SUCCESS in 4m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/13b062dfe10e4865bcabae88fac3c527 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef0a98d70e5945b89cd09192fb14e8b3 : SUCCESS in 8m 56s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/06e335a444a6442c8c11459bbc2fe940 : SUCCESS in 1h 39m 12s (non-voting)\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/75cdecf87f4349b7b836dbda001f5940 : SUCCESS in 1h 34m 48s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/919bdf1f82aa4090afa84127363bc432 : SUCCESS in 47m 38s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/7f7f06e24af74e42b3ff9b2b9ee96b7d : SUCCESS in 1h 28m 18s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/e0fe1086b7574ccd9e92eeea1dbb59f6 : SUCCESS in 1h 13m 12s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/4e3ef7138b8f4309b196d453ca288cb3 : SUCCESS in 1h 19m 19s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/37e17ca5057a49ff9f8bc3c3b50752e2 : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":2},{"id":"c62c33c89cdf576ce59d55dc748d692eca05ebf3","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-08 13:24:27.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"814ca6562f3e59dabbd33ec17d1f8d1f8b61de46","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-01-12 06:17:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3871b1efd39a470d19c4b028a8fabd7e7d611b68","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-12 06:21:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6caed224264c5152e5d0891268fa88932b66912d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-02-10 09:05:50.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"548f508dc93393c4d567df47d0ba052a889788c7","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-02-10 09:12:08.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f433ed34fb89f4ad3a63ac871e4aabf8533dda3e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 09:25:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f92865bd8054a799c10fe321f2cc109\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e5d1051783de4915afe9b7f902c5b2c7 : SUCCESS in 8m 09s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/82cf7e0b72a448d2a9b984fd31402961 : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/336fb46e56e8495b8b054a8d490a7773 : RETRY_LIMIT in 3m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ee2ae5e62cfac6b72f4af29ad82702f8e5b00fc3","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-02-10 09:37:23.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3a461ba83ad361d2f445a8d67c3027d66e7a7fa1","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-02-10 09:37:45.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"09a87585232e58088a50ff3bf888e7ab9d58412d","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-02-10 09:43:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"69e9683bed5906e32f7dd78cda61031978555304","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 11:08:56.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/861a6c8df2b14835a9e959998e4a13b9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe08a94676154c24a0690b19283626f8 : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fb12bf6e76b4cc3a96ce2d587f90732 : SUCCESS in 5m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/df0275c368da477598f96a248c8e3ac6 : SUCCESS in 5m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c6ac36ad3c6e4db2aec0e8caa2b18630 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6bb33a8f4284eb7958ae86623893417 : SUCCESS in 5m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c9401bb5f7b54e2d9b62f387c0380ac3 : FAILURE in 5m 55s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/31ac8b02bec24717983c376a625694f5 : SUCCESS in 1h 46m 54s (non-voting)\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/f9a43aff9df04cc4a7aba5a99344034f : SUCCESS in 2h 02m 07s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/49692c701089446eba88ac5b2c7bc5e3 : SUCCESS in 46m 42s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/38507f1e604c4b2aa673fe7e0dc95e0e : SUCCESS in 1h 51m 54s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/6316db9cedcf48aca289c02bad711e42 : SUCCESS in 1h 12m 21s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/0bb1ed02225c40a18aca10d7cf599c3a : SUCCESS in 1h 21m 46s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a10275e2229b4e64b4a8d0f6d147fb80 : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":3},{"id":"8317d7e5ff6d459abbc3d4f0fd3fba73427ddeca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-02-11 12:37:27.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"4b79b7be0ac8ec935133c9b8b0c88654e8276a9c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 13:21:33.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e4ca0cdccbc44cbb7f4871d21949ccd\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a52a9ef0708c4a87bb7cc515176fd20c : SUCCESS in 10m 51s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0d97e0107278460288cf3fbbc528edb0 : SUCCESS in 11m 34s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5109c560acc248d7bc8cf0a507c40496 : RETRY_LIMIT in 3m 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8ce0aadcca9fd17ecd7e7b589d2d93aca7366ce9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 14:20:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef22deb1ec9e4b40a9ca111a06f2607a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/489558bcda4d4f2f9f11250e39337d27 : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/758e90c76eb44a52b3971982d942a696 : SUCCESS in 4m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/693f148a439e433b9c77ed33357bab93 : SUCCESS in 5m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/28c3607b27ca4641b2ab67c26a22ca73 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab6750f34e434d2685d11ac3d5cfd874 : SUCCESS in 4m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f31cb7e656bc4ffdbf00939ec7245ba3 : SUCCESS in 5m 34s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/eeb1263d0a5b40b380fa217cc70fb431 : SUCCESS in 1h 42m 21s (non-voting)\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/06afb31afe0a44e8bb1d163c3ff8c3eb : SUCCESS in 1h 27m 48s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/ce153365f98041e5bc5a5675e9493d09 : SUCCESS in 1h 03m 42s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/067a277b40e948a59f04cffaae4da171 : SUCCESS in 1h 29m 01s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/b10b1867877845bf9c0298761e71e3da : SUCCESS in 1h 13m 27s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/646a642cbc194483a19cdc15e32cde6c : SUCCESS in 1h 20m 13s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0c08171624db47de8e0747bef07305d6 : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":4},{"id":"2d5003b0cd3cc521f5b72af0e8920e3af0a8b362","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-02-12 13:59:47.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"c721a05f12cb8655e062fda850f3ca631ecb2c0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 14:00:09.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7a729151f9905059fc984f9f8ffd4302a71b4b1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 15:43:39.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1a9834b4f3440a2b289f0f6419fc5e1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/418b76e931d74dc2bb6f7e847d5043f0 : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/077d6a2e53eb4db4a8df45c9918b1535 : SUCCESS in 4m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/02bd5f50a09b45f5a8c9ff897a293fcc : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba91fbbf30be4c1ea59640ffec84d298 : SUCCESS in 6m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee7c176584d440079256b6e0c1fcebfb : SUCCESS in 3m 51s\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/47231ec107e9420d9f177d7e4f8c36b6 : SUCCESS in 52m 23s","accounts_in_message":[],"_revision_number":4},{"id":"73130a539ba71a94e1138c4dbd3da0b3baa25254","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 15:43:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"a4a77eab814a77fd0ba0617b50fa879617ad1378","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 15:44:50.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a85a01de358c46fcb7c6c0ea9773c857\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e90cb087e9b4e25bc80578fbdcb92bb : SUCCESS in 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab1d85a4de5e4e479c1a5a188984d29d : SUCCESS in 56s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"665b55309a6dad392283e8f83d9389ae8fadc3ba","revisions":{"f84ca189c81ded26243af1aa730cadbe7fff8334":{"kind":"REWORK","_number":1,"created":"2025-12-12 11:45:01.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/25/970825/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/25/970825/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/970825/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/970825/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/970825/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/25/970825/1"}}},"commit":{"parents":[{"commit":"3ec0c5f8dd634525e2606ec1945a070cb69690fd","subject":"Merge \"Return X-OpenStack-Request-Id header in response\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ec0c5f8dd634525e2606ec1945a070cb69690fd"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2025-12-12 11:43:55.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2025-12-12 11:43:55.000000000","tz":300},"subject":"Postgres: Fix user password change and access","message":"Postgres: Fix user password change and access\n\nAddress multiple issues in PostgreSQL guest agent.\n1. Fixes an \u0027AttributeError: \u0027PostgreSQLUser\u0027 object has no\n   attribute \u0027_name\u0027\u0027 when changing a user\u0027s password due to\n   improper deserialization in `PgSqlAdmin.change_passwords`.\n\n2. Updates the `grant_access` and `revoke_access` methods to\n   handle PostgreSQL changes in versions 14+ and 16+:\n   - Explicitly grants/revokes schema (public) access (PostgreSQL 14+).\n   - Revokes CONNECT privilege from PUBLIC to ensure proper\n     revocation in `revoke_access`.\n   - Ensures connection is made to the specific database for\n     grant/revoke operations (PostgreSQL 16+).\n\nCode in the patchset passes functional test scenario:\n- create user\n- check user can connect to his own db\n- change user password\n- check user can connect to his own db with new password\n- create database2\n- grant access to the database2 for user\n- check that user can connect to the database2\n- revoke access to the database2 from user\n- check that user cannot connect to the database2\n- grant access to the database2 for user\n- check that user can connect to the database2\nagainst postgresql of versions: 12.22, 14.19, 16.10, 17.7 and 18.1\n\nChange-Id: Iec12b7f6c51fa8812d61787547803c4ed01dec66\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f84ca189c81ded26243af1aa730cadbe7fff8334"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f84ca189c81ded26243af1aa730cadbe7fff8334"}]},"branch":"refs/heads/master"},"0192be0da8a8b2a85d66979dfc15bf4298cd18fc":{"kind":"REWORK","_number":2,"created":"2025-12-15 12:51:02.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/25/970825/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/25/970825/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/970825/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/970825/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/970825/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/25/970825/2"}}},"commit":{"parents":[{"commit":"3ec0c5f8dd634525e2606ec1945a070cb69690fd","subject":"Merge \"Return X-OpenStack-Request-Id header in response\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ec0c5f8dd634525e2606ec1945a070cb69690fd"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2025-12-12 11:43:55.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2025-12-15 12:28:27.000000000","tz":300},"subject":"Postgres: Fix user password change and access","message":"Postgres: Fix user password change and access\n\nAddress multiple issues in PostgreSQL guest agent.\n1. Fixes an \u0027AttributeError: \u0027PostgreSQLUser\u0027 object has no\n   attribute \u0027_name\u0027\u0027 when changing a user\u0027s password due to\n   improper deserialization in `PgSqlAdmin.change_passwords`.\n\n2. Updates the `grant_access` and `revoke_access` methods to\n   handle PostgreSQL changes in versions 14+ and 16+:\n   - Explicitly grants/revokes schema (public) access (PostgreSQL 14+).\n   - Revokes CONNECT privilege from PUBLIC to ensure proper\n     revocation in `revoke_access`.\n   - Ensures connection is made to the specific database for\n     grant/revoke operations (PostgreSQL 16+).\n\nCode in the patchset passes functional test scenario:\n- create user\n- check user can connect to his own db\n- change user password\n- check user can connect to his own db with new password\n- create database2\n- grant access to the database2 for user\n- check that user can connect to the database2\n- revoke access to the database2 from user\n- check that user cannot connect to the database2\n- grant access to the database2 for user\n- check that user can connect to the database2\n- remove database2\n- check that database2 was removed\nagainst postgresql of versions: 12.22, 14.19, 16.10, 17.7 and 18.1\n\nChange-Id: Iec12b7f6c51fa8812d61787547803c4ed01dec66\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0192be0da8a8b2a85d66979dfc15bf4298cd18fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0192be0da8a8b2a85d66979dfc15bf4298cd18fc"}]},"branch":"refs/heads/master"},"a900939aa902f1f21dc2280a422a9ef0b683d2cf":{"kind":"REWORK","_number":3,"created":"2026-02-10 09:05:50.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/25/970825/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/25/970825/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/970825/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/970825/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/970825/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/25/970825/3"}}},"commit":{"parents":[{"commit":"2979b80978684154c28cf74d69c4419131d0e960","subject":"Merge \"Fix PostgreSQL cinder backup failure due to lost backend connection\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2979b80978684154c28cf74d69c4419131d0e960"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2025-12-12 11:43:55.000000000","tz":300},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2026-02-10 09:05:11.000000000","tz":480},"subject":"Postgres: Fix user password change and access","message":"Postgres: Fix user password change and access\n\nAddress multiple issues in PostgreSQL guest agent.\n1. Fixes an \u0027AttributeError: \u0027PostgreSQLUser\u0027 object has no\n   attribute \u0027_name\u0027\u0027 when changing a user\u0027s password due to\n   improper deserialization in `PgSqlAdmin.change_passwords`.\n\n2. Updates the `grant_access` and `revoke_access` methods to\n   handle PostgreSQL changes in versions 14+ and 16+:\n   - Explicitly grants/revokes schema (public) access (PostgreSQL 14+).\n   - Revokes CONNECT privilege from PUBLIC to ensure proper\n     revocation in `revoke_access`.\n   - Ensures connection is made to the specific database for\n     grant/revoke operations (PostgreSQL 16+).\n\nCode in the patchset passes functional test scenario:\n- create user\n- check user can connect to his own db\n- change user password\n- check user can connect to his own db with new password\n- create database2\n- grant access to the database2 for user\n- check that user can connect to the database2\n- revoke access to the database2 from user\n- check that user cannot connect to the database2\n- grant access to the database2 for user\n- check that user can connect to the database2\n- remove database2\n- check that database2 was removed\nagainst postgresql of versions: 12.22, 14.19, 16.10, 17.7 and 18.1\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/976212\n\nChange-Id: Iec12b7f6c51fa8812d61787547803c4ed01dec66\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a900939aa902f1f21dc2280a422a9ef0b683d2cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a900939aa902f1f21dc2280a422a9ef0b683d2cf"}]},"branch":"refs/heads/master"},"665b55309a6dad392283e8f83d9389ae8fadc3ba":{"kind":"REWORK","_number":4,"created":"2026-02-11 12:37:27.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/25/970825/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/25/970825/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/970825/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/970825/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/970825/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/25/970825/4"}}},"commit":{"parents":[{"commit":"2979b80978684154c28cf74d69c4419131d0e960","subject":"Merge \"Fix PostgreSQL cinder backup failure due to lost backend connection\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2979b80978684154c28cf74d69c4419131d0e960"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2025-12-12 11:43:55.000000000","tz":300},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2026-02-11 12:37:21.000000000","tz":480},"subject":"Postgres: Fix user password change and access","message":"Postgres: Fix user password change and access\n\nAddress multiple issues in PostgreSQL guest agent.\n1. Fixes an \u0027AttributeError: \u0027PostgreSQLUser\u0027 object has no\n   attribute \u0027_name\u0027\u0027 when changing a user\u0027s password due to\n   improper deserialization in `PgSqlAdmin.change_passwords`.\n\n2. Updates the `grant_access` and `revoke_access` methods to\n   handle PostgreSQL changes in versions 14+ and 16+:\n   - Explicitly grants/revokes schema (public) access (PostgreSQL 14+).\n   - Revokes CONNECT privilege from PUBLIC to ensure proper\n     revocation in `revoke_access`.\n   - Ensures connection is made to the specific database for\n     grant/revoke operations (PostgreSQL 16+).\n\nCode in the patchset passes functional test scenario:\n- create user\n- check user can connect to his own db\n- change user password\n- check user can connect to his own db with new password\n- create database2\n- grant access to the database2 for user\n- check that user can connect to the database2\n- revoke access to the database2 from user\n- check that user cannot connect to the database2\n- grant access to the database2 for user\n- check that user can connect to the database2\n- remove database2\n- check that database2 was removed\nagainst postgresql of versions: 12.22, 14.19, 16.10, 17.7 and 18.1\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/976212\n\nChange-Id: Iec12b7f6c51fa8812d61787547803c4ed01dec66\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/665b55309a6dad392283e8f83d9389ae8fadc3ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/665b55309a6dad392283e8f83d9389ae8fadc3ba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
