)]}'
{"id":"openstack%2Ftrove~970826","triplet_id":"openstack%2Ftrove~master~I4abee725caa762d031172d484ea93fb5fd34d6c4","project":"openstack/trove","branch":"master","topic":"bug/configuration-parameter-float-value-error","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-01-11 08:45:08.000000000","reason":"Change was submitted"},"36080":{"account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"last_update":"2026-01-11 08:45:08.000000000","reason":"Change was submitted"},"31737":{"account":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"last_update":"2026-01-11 08:45:08.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2026-01-11 07:32:07.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-01-09 16:14:10.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I4abee725caa762d031172d484ea93fb5fd34d6c4","subject":"Fix: Allow float values in KeyValueCodec","status":"MERGED","created":"2025-12-12 12:01:28.000000000","updated":"2026-01-11 08:45:58.000000000","submitted":"2026-01-11 08:45:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"970826-bug/configuration-parameter-float-value-error","meta_rev_id":"089ed5eab70b70841273f86348a2a89e00b1b286","_number":970826,"virtual_id_number":970826,"owner":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-11 08:45:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-01-11 07:32:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":1,"date":"2026-01-09 16:14:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-01-11 07:32:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-12 12:04:10.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-12-12 12:04:10.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"state":"REVIEWER"},{"updated":"2025-12-12 12:04:10.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2025-12-12 12:15:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-12-12 13:55:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ed2772fef9ba5992177a4176c4fd417ff2d14648","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2025-12-12 12:01:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b60ec753c8e2d02e427373ca5a5110fbe55155b1","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-12 12:04:10.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a0425d80ce2259b725321e48c3fa3898c0f23299","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 12:15:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0cc303e08b84222abc5a70010a39af8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/526a2101e5af44d6a5733ee0fe2de454 : SUCCESS in 7m 43s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/747b6ed4d74942039aed753f90907da7 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2a3d6859b9cf4228be392deb8e7b484b : SUCCESS in 12m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2daad32ad11fdb6e5220a4a3151a9922f156fd30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 13:55:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25b6dfa73c734ad7b8ea2d2dafc9a27a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3756e5a616d44786a11d68e66921264f : SUCCESS in 7m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8826188ba300414e97b082238269cb14 : SUCCESS in 5m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9634497c460d47e3957a0b8f0bf74cfa : SUCCESS in 6m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/24b6b20fdc48405e84388faa3aca0721 : SUCCESS in 8m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63eeb2d5ee184854af36f88774ab8dbd : SUCCESS in 5m 27s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/ca09e3ed5d964e1e93c3f70d933f8907 : SUCCESS in 50m 29s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/e4b282865b774589a19e1554aaf2c2fa : SUCCESS in 1h 40m 17s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/3e7b891ce4294a369bc0604ece9b2d0a : SUCCESS in 1h 28m 01s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/32aeda4720ae419aba573d1bb46c8dd1 : SUCCESS in 1h 52m 09s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/a319d09e9b964c1cb37b5ef42eff55df : SUCCESS in 46m 36s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/c3ae9053dc5d4aa8a66d8d96fcc0e62a : SUCCESS in 1h 39m 10s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/e29abed7f61345ed9fa121674045e875 : SUCCESS in 1h 21m 13s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/46080caca40647b08bb23b1610414368 : SUCCESS in 1h 31m 57s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/b8c09cca058b4b7ca8bfb043718f1edb : SUCCESS in 46m 00s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/fd0b2756982b4ae184a3afdb12421f6d : SUCCESS in 1h 22m 59s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/55f387dd70dd49e985dc6633903e7abe : SUCCESS in 1h 18m 30s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/d3b4094ce2f94afe8b683de15d38ee7e : SUCCESS in 1h 18m 55s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fedc561b90e145928ed364b36be2b28b : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":1},{"id":"19b74f41a578537f838b7a0c45df058322efd9a7","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2025-12-14 23:13:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a83c5b1dc88a8456bcfb5e587823a58ae708d7a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2025-12-15 04:34:33.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"30ea8c03a644a08c97419c4a12e9617d9fbf5ec6","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2025-12-15 04:35:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e9eeb86d46ce4b039f5cdc93fa4280de581fb754","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 04:49:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95a93a6174d64437b222a0f4dce1416e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/97e8f4cdd0de48339e59fc2a01293efa : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7b81acaf01e44eb5af9f1f8858d28925 : SUCCESS in 9m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7d6f62d061064a4c8e9ea7a6dc76574f : SUCCESS in 12m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b4b60c2d9f62c7a6b1c352b1f613ea51bf9aa7a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 06:24:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23291326b8cf4c429fedf394828e85cb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f42c0eb8cb364fc5bb0e200498242ba3 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dc439e5374c4d988cf9eb5ac0a9a5e7 : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e8f3feb88ac642c297f494c6a3d67635 : SUCCESS in 3m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7d14b4699d9d40788e0dc8f86b62f2b8 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a448cfa2f864bc8b8604a419fdfcc25 : SUCCESS in 3m 25s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/e699c7121cdf4a49b3753faf15b28b69 : SUCCESS in 1h 02m 22s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/c418573f3f26466797c825a34216242f : SUCCESS in 1h 48m 52s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/f8d11638292f4abdb68cca485cc68029 : SUCCESS in 1h 26m 41s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/8c21d42df16a42739fbeb67884d49852 : SUCCESS in 1h 46m 22s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/d514d1a17d9849ec995cf2401ed30a7b : SUCCESS in 43m 26s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/d093669bb36b4f1f941711676c30f254 : SUCCESS in 1h 27m 36s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/ab4b938f01984ea5b7205d369bd8c656 : SUCCESS in 1h 20m 11s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/555444e1ebd34ef2af90a1eeb64afaea : SUCCESS in 1h 29m 06s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/7bd4b398f5c64a6eaae06e0773b56655 : SUCCESS in 48m 19s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/729769e1882444bab74ce04b82ed4fd7 : SUCCESS in 1h 31m 58s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/df95b6cc9a3f4afa9c5278bb2add3539 : SUCCESS in 1h 13m 50s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/bdd96555ac344b5bb9ebe387baefe60e : SUCCESS in 1h 16m 52s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4657f01eb4dd49f596230a3f20922756 : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":2},{"id":"e08246fc148026e020b762cdda31dd3fbb65f345","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2025-12-15 09:25:59.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ef4d1eebd48abd8e0b8bf6ffa222a19e8a907ed6","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-06 14:09:25.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9cd6a7f3380d6f0b2b2b932627f0ba23d38ad420","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-01-07 13:39:51.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"f649c741a3b0a1a7310dfc21398e66b737972c13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-08 13:11:40.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"6defe58951ebac80a584eff21490886d09e24506","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-08 13:11:51.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"eefb874d0f322a69bc1a46e0e24c4eaf56e5fc68","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-08 13:11:54.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"38a4806161198f7ed487f9953ffe99bc8a0f49e6","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-01-08 13:19:29.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2bfe0842b9e50af5685d7b7bab7688e0f1e519bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 13:29:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f202669e05d44f61bdc17651099f5ebb\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/43d336e917ca46f8924eaeaba07e0f71 : SUCCESS in 10m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b8287466e47f4d4fb91a336a144c60fb : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/901b5ce1f295402bbda2886838515331 : SUCCESS in 14m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"33dfde4a3c6bfa0c654555ba597ed6695369e723","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 15:09:33.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c6a0477fb2df4cc08c094bf4f90e34ea\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7cdae08fcdd04817a14a6e27deacd223 : SUCCESS in 7m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ec9dde7e2e6406894481ae826962a37 : FAILURE in 4m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7f12b4bb507e44ae8395ab47f20b6cb2 : SUCCESS in 7m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b9af86599db441d8a1d9f7c536bb301b : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7b15b60f63c401690e3e02fc861ab42 : SUCCESS in 6m 14s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/469b57ccaa874da0b3613a952d78731b : SUCCESS in 53m 04s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/5f6b2292eba74628b4b2fe816950876c : SUCCESS in 1h 49m 31s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/52db2b3639a04514b353b3ac16604c60 : SUCCESS in 1h 30m 11s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/0a708ff8be5b4cdea92f70f8e043ecd7 : SUCCESS in 1h 52m 14s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/b9c71275e4c54e52869c87cf51d0a318 : SUCCESS in 47m 22s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/becf7fcd89fb4b9d8266f69cd7604437 : SUCCESS in 1h 46m 54s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/5cb999f7e6704cc88ab0d515ae8f7f39 : FAILURE in 20m 45s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/aa398e9860054c14a29d5b756f8ea280 : SUCCESS in 1h 30m 21s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/0ff74d7fae5747d68c51689fc88fef05 : SUCCESS in 52m 34s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/763d6b1ab0864851926c97417eedf710 : FAILURE in 21m 19s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/d882ae684eb64703be4323ff60c4cdac : FAILURE in 25m 49s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/23e29facabbc47bcbd6c8895350f2a6a : SUCCESS in 1h 29m 42s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5c696ed2683347b78670b1d7335198e3 : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":3},{"id":"a142f4056214dea84ec9dc27b67ad0fb1e39f3d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-08 21:59:31.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":4},{"id":"fd97d0e2d9b690f66b1f43891001727135e3956f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 22:43:56.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/857346c7eb6e46c18c9ccb3bd158b5ce\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b4ef3de2f34e4c299539fdb116585fb8 : SUCCESS in 7m 59s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4d6353f3fb79492ba61a52d4d4076a0c : SUCCESS in 9m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c91348d170794d4ea0b5df60ff8216e0 : SUCCESS in 8m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2fc738a371461c1fdea4462d2ff1ad2cc1adbe10","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-01-09 08:02:26.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"273855b76bc9f114387e93d5968d3b82172ec946","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 08:19:32.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68fb6ce01e4c462a9d8ea0f4867922ec\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1d8af007c0ed41edb9dd9aaa7ffc12c8 : SUCCESS in 7m 18s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5a8847565db84b72b1f58290572ec10e : SUCCESS in 10m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6dff546f5ab5426f9f9816c223c13ad2 : SUCCESS in 11m 00s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"736948bb7f53d8e2342da96e23903638de7236d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 08:26:40.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a5154c14c20c4dbfb0215eb2e25ec22c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/00fd0683dead413082169ba928c13c19 : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/870a000645ed47b5b152a7d5cec280fb : FAILURE in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ffa3c1d820f746e384e321760a884ac5 : SUCCESS in 5m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ce7a5c4e7d2443588df9f2723f51f586 : SUCCESS in 10m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e27c5d1e0707426db95c64ba1c6fc679 : SUCCESS in 5m 06s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/894f4e2210c947ee8fb322059a0cca81 : RETRY_LIMIT in 34s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/e5844f1887ef426dbb9ac68ca8dc0a4f : RETRY_LIMIT in 57s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/92e02c1d112f4bd9976948713c64bf7c : RETRY_LIMIT in 59s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/5b842f4f488b48ddb0227796dbe37a76 : RETRY_LIMIT in 54s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/36649e147d0142ca846286434c293162 : RETRY_LIMIT in 56s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/5d6c246db03d467a87f6c9094009c139 : RETRY_LIMIT in 1m 01s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/b60a4351614e472e95d2b68317cd6c22 : RETRY_LIMIT in 1m 01s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/6833e1c2e35b488bab28ff517a515771 : RETRY_LIMIT in 55s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/34316b87c368423c84fd08074936f489 : RETRY_LIMIT in 31s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/f6c0b518aa44414eb6cba131fcc3b9ed : RETRY_LIMIT in 39s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/229e544811874da6b6ccf6d7980b4966 : RETRY_LIMIT in 35s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/021b2a19b21c4e94af5a2e0ceb9df9cb : RETRY_LIMIT in 49s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/612f66ee5caa4c9a8201929a501754e8 : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":4},{"id":"7044172d4f22fbfb285823122e3768c7a17f94cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-09 11:16:29.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"d6eb4c3f8b0305432f289d176166b5c2243b5a52","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-09 11:16:42.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0680e3a954bae9859776191bcba22dca32d17b0e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 12:29:11.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4060b8a79a4847089c6af668ea3a0c0c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dfb7ac2757404754acb1eca61e74e76e : SUCCESS in 11m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8ca5beb24483432c87fc07ef1d8b9d4b : SUCCESS in 11m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/358b3a99ab2246ae952bcc06818b929b : SUCCESS in 14m 21s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"90ab7b8c5854a36dcf4e3ebe1bceb9ea17d9aab4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 13:25:55.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/10c7f9a4f0994ed589dcc21ca9463a53\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e62b67450bfb485b8f3690402825f086 : SUCCESS in 6m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41459fd57d9242079e444ad7486dbcaf : FAILURE in 4m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e42f3ca7293c4423980bde528fd776e6 : SUCCESS in 5m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fc3dec28adcd4ec7834b0bc7669fa1a6 : SUCCESS in 8m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a70a0e0a990647cd91cdc6212950cd7a : SUCCESS in 5m 37s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/61c1d91c1e37423ab526eebb466e98ee : SUCCESS in 48m 12s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/3b6ea87b3b9a4145a0a0f28756ba97d6 : SUCCESS in 2h 07m 19s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/f9f1d8912f964a6b9435219c16724dff : SUCCESS in 1h 31m 05s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/a52d8f6eb7de400485104ade09f0b1ed : SUCCESS in 1h 42m 46s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/266a7a6eadb34979ac0dfa9e98e2cfc6 : SUCCESS in 42m 53s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/f2f591b45a2a4e9284e1367e5a69e278 : SUCCESS in 1h 53m 32s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/a19711216e1749c697cd7b891efda673 : SUCCESS in 1h 26m 19s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/73e08553f521427683f401dde92f6a4a : SUCCESS in 1h 29m 44s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/3832dc0a04214eb3a81862c133d5f16c : SUCCESS in 52m 28s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/84685a3ac3604859a5135f596a828d76 : SUCCESS in 1h 34m 58s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/7e5eee3db45c498cb6b703b80575c235 : SUCCESS in 1h 11m 43s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/d4d5b0c8c0264949a153e829bc6bf046 : SUCCESS in 1h 24m 30s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/12b94c0382b748e2b440fa018d628279 : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":5},{"id":"e9f3c9faa339988f0cfe290c599c940307d4fb75","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-01-09 13:38:03.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a623bdd0b97ba6b337adf00c139c2f6ae11e71fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-09 13:58:27.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"212909421d0d028db8718a5f50b052669d0fb625","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 14:36:27.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c66dcf4f27484522b63b8064c0148195\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/88ebc47bd92e4aa09462e2f3a228ce0b : SUCCESS in 10m 59s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3e22d0a7968e43a980740490fb40c8a3 : SUCCESS in 11m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2354115a435b48a1931d92efffa86e0e : SUCCESS in 11m 13s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"6b6b1248cc3fe8a66091ad2a65dd0c2248b1759b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 16:12:43.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00393ee5054f47be8816cd8da7e46344\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b718eeed051846d18dab1f996d887dfc : SUCCESS in 7m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34804a9daafb4b5187b9ddf30f9ca648 : SUCCESS in 3m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dd57c2ac250d41699aa60b1a7f06aba6 : SUCCESS in 5m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f4772ee6e1c846e7b310381aeafd6b03 : SUCCESS in 8m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f4d1c6b0cd44d498300c068d2559eec : SUCCESS in 5m 54s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/62993b7317a442648f02d0172549be0a : SUCCESS in 51m 41s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/cfebae19635a4c8c9ce8311a19733057 : SUCCESS in 1h 59m 54s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/dd901185c8f24ce789dfed5c601343da : SUCCESS in 1h 28m 13s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/90876b6be03e477faf758202c28180e2 : SUCCESS in 1h 53m 21s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/64ca70267de04413be754547cf259029 : SUCCESS in 48m 14s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/95a44f0b3acf4812a895d1fefa15f137 : SUCCESS in 2h 04m 30s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/1e4ef4c7b49f460ab2d66ae59fa54b06 : SUCCESS in 1h 43m 28s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/39373367b64b4116b91df6f0455776e6 : SUCCESS in 1h 34m 57s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/40ad162651a64082b0c3b7b0469874d7 : SUCCESS in 56m 53s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/e5e50067bf39429482bc93149dd8ea9f : FAILURE in 1h 22m 23s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/9da734c38fd0476b8d20b9d17d120d42 : SUCCESS in 1h 16m 46s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/086c0b2292014e18b2e32404bb25c73d : SUCCESS in 1h 27m 32s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a31f4ef22d40441ebb39213b788f51ec : SUCCESS in 6m 08s","accounts_in_message":[],"_revision_number":6},{"id":"ce9871f1998d42c44e91b694f9b38631d39d68a7","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-09 16:14:10.000000000","message":"Patch Set 6: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"ff18656817c06f871b8b2d3b7d64ac7b6abd3672","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-01-11 07:32:07.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"c2e34b9dd5b5dd6c83a3b1b51e5c85eead774244","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-11 07:32:30.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"23d6fee907bd559f161aeffbed54ee08ed97f763","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-11 08:45:08.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/977b2178ff434710972abbe52b755898\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a34a7f4f60ae43399c0ea7bd14b00e73 : SUCCESS in 5m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0604ab3b8e1f4786b3317647d0ecf408 : SUCCESS in 9m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/348d5ebaf3a14509b9e9071d5af78a1c : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01860066b6774e3f8063454cc5f6f8af : SUCCESS in 2m 53s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/5d7d75f22de5432abc1069afe9d5b3f6 : SUCCESS in 46m 24s\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/23fff3c240384c9bad9e1662ff06bedc : SUCCESS in 50m 23s\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/dab7cd6031d64708a498a055dabce56f : SUCCESS in 1h 06m 35s","accounts_in_message":[],"_revision_number":6},{"id":"5a806e0f00cad3fffc9e2e7414f80dc85b7d0651","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-11 08:45:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"089ed5eab70b70841273f86348a2a89e00b1b286","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-11 08:45:58.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69c0547c2f3049389f305ed4dee2c438\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebc7396b368d4e2b887df7cdb3255e8e : SUCCESS in 38s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"ece29cea9dbae80f2b809663a6cd379ec2383344","revisions":{"849fad64115b5b8da5b5bc1e5beb6c24d0fcdfb2":{"kind":"REWORK","_number":1,"created":"2025-12-12 12:01:28.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/26/970826/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/26/970826/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/970826/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/970826/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/970826/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/26/970826/1"}}},"commit":{"parents":[{"commit":"3ec0c5f8dd634525e2606ec1945a070cb69690fd","subject":"Merge \"Return X-OpenStack-Request-Id header in response\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ec0c5f8dd634525e2606ec1945a070cb69690fd"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2025-12-12 12:00:08.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2025-12-12 12:00:08.000000000","tz":300},"subject":"Fix: Allow float values in KeyValueCodec","message":"Fix: Allow float values in KeyValueCodec\n\nFix \"AttributeError: \u0027float\u0027 object has no attribute\n\u0027lower\u0027\" in KeyValueCodec when attempting to set a float\nvalue for a configuration parameter with \"float\" type.\n\nThe `is_valid_value` method was incorrectly attempting to\ncall `.lower()` on a float, causing a crash. This patch\nadds an explicit check for floats to prevent this error.\n\nChange-Id: I4abee725caa762d031172d484ea93fb5fd34d6c4\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/849fad64115b5b8da5b5bc1e5beb6c24d0fcdfb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/849fad64115b5b8da5b5bc1e5beb6c24d0fcdfb2"}]},"branch":"refs/heads/master"},"77c372f6d87461d57f6768157a61a6d5a2248520":{"kind":"REWORK","_number":2,"created":"2025-12-15 04:34:33.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/26/970826/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/26/970826/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/970826/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/970826/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/970826/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/26/970826/2"}}},"commit":{"parents":[{"commit":"3ec0c5f8dd634525e2606ec1945a070cb69690fd","subject":"Merge \"Return X-OpenStack-Request-Id header in response\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ec0c5f8dd634525e2606ec1945a070cb69690fd"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2025-12-12 12:00:08.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2025-12-15 04:18:36.000000000","tz":300},"subject":"Fix: Allow float values in KeyValueCodec","message":"Fix: Allow float values in KeyValueCodec\n\nFix \"AttributeError: \u0027float\u0027 object has no attribute\n\u0027lower\u0027\" in KeyValueCodec when attempting to set a float\nvalue for a configuration parameter with \"float\" type.\n\nThe `is_valid_value` method was incorrectly attempting to\ncall `.lower()` on a float, causing a crash. This patch\nadds an explicit check for floats to prevent this error.\n\nChange-Id: I4abee725caa762d031172d484ea93fb5fd34d6c4\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/77c372f6d87461d57f6768157a61a6d5a2248520"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/77c372f6d87461d57f6768157a61a6d5a2248520"}]},"branch":"refs/heads/master"},"b166dc52f8fbc406bad93665eac649f67cb1ce72":{"kind":"REWORK","_number":3,"created":"2026-01-08 13:11:40.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/26/970826/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/26/970826/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/970826/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/970826/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/970826/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/26/970826/3"}}},"commit":{"parents":[{"commit":"3ec0c5f8dd634525e2606ec1945a070cb69690fd","subject":"Merge \"Return X-OpenStack-Request-Id header in response\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ec0c5f8dd634525e2606ec1945a070cb69690fd"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2025-12-12 12:00:08.000000000","tz":300},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-08 13:11:40.000000000","tz":0},"subject":"Fix: Allow float values in KeyValueCodec","message":"Fix: Allow float values in KeyValueCodec\n\nFix \"AttributeError: \u0027float\u0027 object has no attribute\n\u0027lower\u0027\" in KeyValueCodec when attempting to set a float\nvalue for a configuration parameter with \"float\" type.\n\nThe `is_valid_value` method was incorrectly attempting to\ncall `.lower()` on a float, causing a crash. This patch\nadds an explicit check for floats to prevent this error.\n\nChange-Id: I4abee725caa762d031172d484ea93fb5fd34d6c4\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b166dc52f8fbc406bad93665eac649f67cb1ce72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b166dc52f8fbc406bad93665eac649f67cb1ce72"}]},"branch":"refs/heads/master"},"5a44f6390967cd8f60ae2ba96ee4ca7e5fc1b282":{"kind":"REWORK","_number":4,"created":"2026-01-08 21:59:31.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/26/970826/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/26/970826/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/970826/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/970826/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/970826/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/26/970826/4"}}},"commit":{"parents":[{"commit":"3ec0c5f8dd634525e2606ec1945a070cb69690fd","subject":"Merge \"Return X-OpenStack-Request-Id header in response\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ec0c5f8dd634525e2606ec1945a070cb69690fd"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2025-12-12 12:00:08.000000000","tz":300},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-08 21:59:31.000000000","tz":0},"subject":"Fix: Allow float values in KeyValueCodec","message":"Fix: Allow float values in KeyValueCodec\n\nFix \"AttributeError: \u0027float\u0027 object has no attribute\n\u0027lower\u0027\" in KeyValueCodec when attempting to set a float\nvalue for a configuration parameter with \"float\" type.\n\nThe `is_valid_value` method was incorrectly attempting to\ncall `.lower()` on a float, causing a crash. This patch\nadds an explicit check for floats to prevent this error.\n\nChange-Id: I4abee725caa762d031172d484ea93fb5fd34d6c4\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5a44f6390967cd8f60ae2ba96ee4ca7e5fc1b282"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5a44f6390967cd8f60ae2ba96ee4ca7e5fc1b282"}]},"branch":"refs/heads/master"},"cfdee480c5750681c8e69db6de5024835f97334a":{"kind":"REWORK","_number":5,"created":"2026-01-09 11:16:29.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/26/970826/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/26/970826/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/970826/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/970826/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/970826/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/26/970826/5"}}},"commit":{"parents":[{"commit":"3ec0c5f8dd634525e2606ec1945a070cb69690fd","subject":"Merge \"Return X-OpenStack-Request-Id header in response\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ec0c5f8dd634525e2606ec1945a070cb69690fd"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2025-12-12 12:00:08.000000000","tz":300},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-09 11:16:29.000000000","tz":0},"subject":"Fix: Allow float values in KeyValueCodec","message":"Fix: Allow float values in KeyValueCodec\n\nFix \"AttributeError: \u0027float\u0027 object has no attribute\n\u0027lower\u0027\" in KeyValueCodec when attempting to set a float\nvalue for a configuration parameter with \"float\" type.\n\nThe `is_valid_value` method was incorrectly attempting to\ncall `.lower()` on a float, causing a crash. This patch\nadds an explicit check for floats to prevent this error.\n\nChange-Id: I4abee725caa762d031172d484ea93fb5fd34d6c4\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cfdee480c5750681c8e69db6de5024835f97334a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cfdee480c5750681c8e69db6de5024835f97334a"}]},"branch":"refs/heads/master"},"ece29cea9dbae80f2b809663a6cd379ec2383344":{"kind":"REWORK","_number":6,"created":"2026-01-09 13:58:27.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/26/970826/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/26/970826/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/970826/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/970826/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/970826/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/26/970826/6"}}},"commit":{"parents":[{"commit":"3ec0c5f8dd634525e2606ec1945a070cb69690fd","subject":"Merge \"Return X-OpenStack-Request-Id header in response\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ec0c5f8dd634525e2606ec1945a070cb69690fd"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2025-12-12 12:00:08.000000000","tz":300},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-09 13:58:09.000000000","tz":0},"subject":"Fix: Allow float values in KeyValueCodec","message":"Fix: Allow float values in KeyValueCodec\n\nFix \"AttributeError: \u0027float\u0027 object has no attribute\n\u0027lower\u0027\" in KeyValueCodec when attempting to set a float\nvalue for a configuration parameter with \"float\" type.\n\nThe `is_valid_value` method was incorrectly attempting to\ncall `.lower()` on a float, causing a crash. This patch\nadds an explicit check for floats to prevent this error.\n\nChange-Id: I4abee725caa762d031172d484ea93fb5fd34d6c4\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ece29cea9dbae80f2b809663a6cd379ec2383344"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ece29cea9dbae80f2b809663a6cd379ec2383344"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
