)]}'
{"id":"openstack%2Ftrove~972157","triplet_id":"openstack%2Ftrove~stable%2F2025.2~I2a7c69776e2a41f9ac5fd9da93f5772746d87a2f","project":"openstack/trove","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2026-01-07 12:31:21.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-01-07 13:20:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2a7c69776e2a41f9ac5fd9da93f5772746d87a2f","subject":"Set ipc_mode to host to prevent potential shared memory exhaustion","status":"MERGED","created":"2026-01-05 14:42:18.000000000","updated":"2026-01-07 13:21:14.000000000","submitted":"2026-01-07 13:20:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"972157","cherry_pick_of_change":971705,"cherry_pick_of_patch_set":2,"meta_rev_id":"573c4aaa0188007685c872788cb1d1ee621c1e5a","_number":972157,"virtual_id_number":972157,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-07 13:20:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-01-07 12:31:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-01-07 12:31:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-05 14:42:18.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"CC"},{"updated":"2026-01-05 14:43:12.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2026-01-05 16:43:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d067dc2fa84ca054aff63d1f1b787ad197137c02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-05 14:42:18.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a00cf35928b627e26d80ecb7b1e956a48284a2e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 16:43:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c11944a3b614bac9913375c9e9ca92b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c3e4ff0ada85471cabf532696643e11a : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27127e76d24644ee965d8d48b744b9ac : SUCCESS in 5m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bd72203e7feb415aa29a27d22c66efb8 : SUCCESS in 6m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5763406c22ac4c03a26a72481cec1a9d : SUCCESS in 5m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0c7642c2137441cb973d70ecaabac51a : SUCCESS in 9m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/020ff8fb771f4491a6a989e5391acf96 : SUCCESS in 5m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b353ef357fb4d72bd9177db7da1e7f6 : SUCCESS in 6m 15s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/289b0804f4184328bd4858e4db52ba4f : SUCCESS in 47m 37s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/c9ead560a3e84dd5835c753857cfb13e : SUCCESS in 1h 47m 18s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/cfb173cd425740b19884f7a1b7dc94f6 : SUCCESS in 1h 21m 12s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/bdb14cf7f8b7484c8098b16c642eaffb : SUCCESS in 1h 39m 53s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/7794bb9d54834dbf81e886b7925cb80f : SUCCESS in 42m 02s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/cc32b23288b5434e86f5c1564c88b44e : SUCCESS in 1h 33m 01s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/fdf23318741b4a9eb9532ff624721c2e : SUCCESS in 1h 17m 58s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/a7c86d0bd91a4508a84d2817c16d299e : SUCCESS in 1h 19m 47s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/59963503ee6d4b95b59ec882dd63d311 : SUCCESS in 44m 47s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/fcb027fdd02e42e0a6e614a92254ca95 : SUCCESS in 1h 31m 00s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/abe8c7f9a6ac4ff1869e6e83afe9e6c8 : SUCCESS in 1h 14m 39s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/49c90cb39de4471baa8509918a1c7965 : SUCCESS in 1h 19m 09s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3edc8c71c378474689007c9c88200110 : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":1},{"id":"fbdc4dbdd5b2773aae5bcbef53960146fbc3fd00","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-01-07 12:31:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"91e78d96ae887a9bd89ae557e9911955f365915d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 12:31:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f47061c3b430ffa623ad532462fa9fe149892ad1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 13:20:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3366a8e8f4c430ea429e9906274a66f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/345282af091345ac94fce806e9f38a5a : SUCCESS in 5m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6968890e7a8f4e1e9b1c6a5abaead01e : SUCCESS in 7m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7c5b809271fe4e3d94096d7e95016ed2 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60460b71c8494bfa8b3eede5ecf2dc54 : SUCCESS in 5m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/757d1306c0184f36aeafb32f75b180a8 : SUCCESS in 6m 23s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/5b524d18112b4cd78c5651dbbab17456 : SUCCESS in 43m 43s\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/97df3a4b10d84767913de75b3e83c826 : SUCCESS in 43m 25s\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/ecb3a384f1de4ad290078eee2f2a88a4 : SUCCESS in 44m 26s","accounts_in_message":[],"_revision_number":1},{"id":"e1f201c405ed0374e0ed68dbffe4a21006d319b8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 13:20:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"573c4aaa0188007685c872788cb1d1ee621c1e5a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 13:21:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f423ed82603345b4b05c2f3726898b05\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddaa657273c64f93a756accc649aab46 : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/098c70e2425f4bb3b0b08136a439d783 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"015a8a7cb49df65640f5b382660c9af4c094b173","revisions":{"015a8a7cb49df65640f5b382660c9af4c094b173":{"kind":"REWORK","_number":1,"created":"2026-01-05 14:42:18.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/57/972157/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/57/972157/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/972157/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/972157/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/972157/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/57/972157/1"}}},"commit":{"parents":[{"commit":"8a864a0360af3ecf52edf82b48fcfb1dfd8aed9c","subject":"Change settings for stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8a864a0360af3ecf52edf82b48fcfb1dfd8aed9c"}]}],"author":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2025-12-23 12:18:16.000000000","tz":480},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-05 14:42:18.000000000","tz":0},"subject":"Set ipc_mode to host to prevent potential shared memory exhaustion","message":"Set ipc_mode to host to prevent potential shared memory exhaustion\n\nTrove is launching Docker containers without specifying shm_size, which\nleft the default 64MB limit for /dev/shm. PostgreSQL uses shared_buffers\nof 128MB by default, so any database running with default settings could\nexceed the default /dev/shm size. In this situation PostgreSQL may fail\nduring shared memory operations with errors such as: \"could not resize\nshared memory segment: No space left on device\".\n\nThis patch uses ipc_mode\u003dhost to pass the shared memory from host to container.\n\nCo-Authored-By: Eric \u003cmangust404@gmail.com\u003e\n\nChange-Id: I2a7c69776e2a41f9ac5fd9da93f5772746d87a2f\nSigned-off-by: wu.chunyang \u003cwchy1001@gmail.com\u003e\n(cherry picked from commit 124d89a896d80e6ae9d6cc9b3792de9cc8f730ee)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/015a8a7cb49df65640f5b382660c9af4c094b173"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/015a8a7cb49df65640f5b382660c9af4c094b173"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
