)]}'
{"id":"openstack%2Ftrove~972460","triplet_id":"openstack%2Ftrove~master~I6e331f4f578ec0f6882ea90dc2c0f06ab54c09f3","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"31737":{"account":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"last_update":"2026-01-12 07:01:20.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2026-01-12 06:16:38.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-01-12 04:54:29.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I6e331f4f578ec0f6882ea90dc2c0f06ab54c09f3","subject":"[CI] Disable swift installation for cinder storage jobs","status":"MERGED","created":"2026-01-07 12:04:14.000000000","updated":"2026-01-12 07:03:07.000000000","submitted":"2026-01-12 07:01:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"972460","meta_rev_id":"465da352fb899c082806fb1ad61591d907f1878e","_number":972460,"virtual_id_number":972460,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-12 07:01:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-01-12 06:16:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":1,"date":"2026-01-12 04:54:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-01-12 06:16:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-07 12:39:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-01-07 13:41:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-07 17:09:44.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-01-07 17:09:44.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"state":"REVIEWER"}],"messages":[{"id":"866f4082ef5ce7c51b03d001f013e99ff4dffcde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-01-07 12:04:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"85ea17aa3cfe6a2f8ad3fdfb623ac9aec09dd4d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 12:39:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53562f197b7e4fdf8471862f95672e08\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/892b25d0cedf441ea311b943bb47731b : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2a595537ffb44d8a81fc9612f937bce2 : SUCCESS in 9m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7074088f170c4b1bb2d049a6aab1e2ee : SUCCESS in 11m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f0cf0a7398e9e3f5345cac1c6483ffa1116da6f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 13:41:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b9f323a55c64522a1cec18d6ff0e03d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60bee70154934f108b0ba5f69cb88cd7 : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38f223cf3a85438f9e1e7bd96f22b4df : SUCCESS in 5m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/48af3ce5a47e44cd90a3810212bd6f45 : SUCCESS in 5m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/644fd91c02da4041b8c1bace16f899bb : SUCCESS in 8m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c52bc22e8a24e19a6a9c4274c941cbd : SUCCESS in 6m 44s\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/c1d5a8d044044f8e8f9cbdf80ecbcf2c : FAILURE in 1h 30m 49s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/664ef30ead974451853d5763df82fd21 : FAILURE in 1h 11m 49s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/93a15701a5994498995c9b6f31170544 : FAILURE in 1h 21m 22s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a2e5b679e9b54144980419bdc92ceb18 : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":1},{"id":"db3ebe1d333a40914f1aa86674ce61c632285934","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-01-07 14:34:49.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"9f58207e9ce9cbbc6784af9971604c979b4dab9f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 14:47:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49f7913bf7f947f3a9db885ac1d69dc0\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a9f7a25bf38641799cb168db5c1f3045 : SUCCESS in 7m 04s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/66ad152a1a824d57b95488024e3d8b0c : SUCCESS in 7m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d495388259d243ab9db55154826b43fc : SUCCESS in 9m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9c97873886917a5dd935809694e4f008f3ab71c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 16:00:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55351b6725a8440a8ea5ebafd266f257\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a1f411e196dc40dd8ab0032b89c7a4ca : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc6adade61da402da0eea44824690257 : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/00f49ecd9a6d4cc981d9eeb5b488006c : SUCCESS in 3m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/acebaf3796ce48218ef1d7a1029881b2 : SUCCESS in 7m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80b3210388d44ee6a1df2fb0f313f8ce : SUCCESS in 3m 45s\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/4baed9c07991416a8633fd7b47461973 : FAILURE in 1h 24m 02s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/da4bc59e07be487f9e54dd93cd34da22 : FAILURE in 1h 06m 38s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/a7d8c1aa3dd5472a8e698ba2e3aba2ba : FAILURE in 1h 07m 44s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/31098afa2a4e4fbb90663932ad6d171c : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":2},{"id":"064fe8a23aec4cdce3e12d25d69235391ea0f321","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-07 17:09:44.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"96bd464ab40116f6c1d0d68792b87bc870c76307","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-01-11 07:34:25.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0859af7471f93df79ec98bf739f9cf6e2b5c678b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-11 08:45:24.000000000","message":"Patch Set 2: Verified-2\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/trove on branch master.  The\nproblem was:\n\n  Decryption failed\n\nThe problem appears in the the \"publish-trove-images-quay\" job stanza:\n\n  job:\n      name: publish-trove-images-quay\n      description: |\n        Build and sync trove images to quay.io/openstack.trove.\n      pre-run: playbooks/image-build/pre.yaml\n      run: playbooks/image-build/docker-registry.yaml\n      nodeset: trove-ubuntu-noble-single\n      required-projects:\n        - openstack/trove\n  ...\n\n  in \"openstack/trove/zuul.d/jobs.yaml@master\", line 281","accounts_in_message":[],"_revision_number":2},{"id":"ada4d992da3e44861faff803015be5e7f25e4351","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-11 10:30:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a63043a74cdc9fd08f4d18093325383f6bc31fdf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-11 10:31:18.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/trove on branch master.  The\nproblem was:\n\n  Decryption failed\n\nThe problem appears in the the \"publish-trove-images-quay\" job stanza:\n\n  job:\n      name: publish-trove-images-quay\n      description: |\n        Build and sync trove images to quay.io/openstack.trove.\n      pre-run: playbooks/image-build/pre.yaml\n      run: playbooks/image-build/docker-registry.yaml\n      nodeset: trove-ubuntu-noble-single\n      required-projects:\n        - openstack/trove\n  ...\n\n  in \"openstack/trove/zuul.d/jobs.yaml@master\", line 281","accounts_in_message":[],"_revision_number":2},{"id":"4a30ab88bc8cfc44d119c8221b76679ba61319d2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-11 10:31:31.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/trove on branch master.  The\nproblem was:\n\n  Decryption failed\n\nThe problem appears in the the \"publish-trove-images-quay\" job stanza:\n\n  job:\n      name: publish-trove-images-quay\n      description: |\n        Build and sync trove images to quay.io/openstack.trove.\n      pre-run: playbooks/image-build/pre.yaml\n      run: playbooks/image-build/docker-registry.yaml\n      nodeset: trove-ubuntu-noble-single\n      required-projects:\n        - openstack/trove\n  ...\n\n  in \"openstack/trove/zuul.d/jobs.yaml@master\", line 281","accounts_in_message":[],"_revision_number":2},{"id":"e3372c80b012da4a46aba390ef0ac236d17424ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-01-12 02:33:16.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"19c6a108aa8e49f626e4d80c227374598ee53d4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 02:33:36.000000000","message":"Patch Set 3: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/trove on branch stable/2025.1.  The\nproblem was:\n\n  Secret trove_quay_io_creds does not match existing definition in branch master\n\nThe problem appears in the the \"trove_quay_io_creds\" secret stanza:\n\n  secret:\n      name: trove_quay_io_creds\n      data:\n        username: !encrypted/pkcs1-oaep\n          - RikBORnOJ0l12Tb4KmXm7HM1P12ig8WsNy8ANfpMBY4AnrsETdw/RfQ35pyTimisu6dkl\n            L25+6049BOQYT3qHYCtf5nMj7AT9/KtNWTT93fFnj2YvbH8wNJAsCYqCodxa2YKg+rAN3\n            xF00+7NDKU3wdl7eXC946ypwBmVZTC8U+chEINNlV0iyHQK1Jv9peifl+HfPbMZtB9Xrb\n            /3tLcAqSZkWkjx/u1tW0FYCUUyTHpPeTkoSV6Xjd/iMp2IINkp8aDszTAl6nFnre+Cqxi\n            vWjOb9Phfb8j3U6L84Z83BgTAgZX3L+J/lgFLPCVMc2R3WMfHuYs7pYNTxFRCTlPpjQpu\n  ...\n\n  in \"openstack/trove/zuul.d/nodesets.yaml@stable/2025.1\", line 27","accounts_in_message":[],"_revision_number":3},{"id":"4f8d715d0113999f70029404bd67d2406aa4c2e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 02:34:10.000000000","message":"Patch Set 3:\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/trove on branch stable/2025.1.  The\nproblem was:\n\n  Secret trove_quay_io_creds does not match existing definition in branch master\n\nThe problem appears in the the \"trove_quay_io_creds\" secret stanza:\n\n  secret:\n      name: trove_quay_io_creds\n      data:\n        username: !encrypted/pkcs1-oaep\n          - RikBORnOJ0l12Tb4KmXm7HM1P12ig8WsNy8ANfpMBY4AnrsETdw/RfQ35pyTimisu6dkl\n            L25+6049BOQYT3qHYCtf5nMj7AT9/KtNWTT93fFnj2YvbH8wNJAsCYqCodxa2YKg+rAN3\n            xF00+7NDKU3wdl7eXC946ypwBmVZTC8U+chEINNlV0iyHQK1Jv9peifl+HfPbMZtB9Xrb\n            /3tLcAqSZkWkjx/u1tW0FYCUUyTHpPeTkoSV6Xjd/iMp2IINkp8aDszTAl6nFnre+Cqxi\n            vWjOb9Phfb8j3U6L84Z83BgTAgZX3L+J/lgFLPCVMc2R3WMfHuYs7pYNTxFRCTlPpjQpu\n  ...\n\n  in \"openstack/trove/zuul.d/nodesets.yaml@stable/2025.1\", line 27","accounts_in_message":[],"_revision_number":3},{"id":"8d287770dcc29da940ec75f45125db0758b786c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-01-12 03:02:11.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"b5b5d8bd989b533f7cc578c7ff86668430ac3acd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 03:16:59.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a5dbbe7cb4e4f60b8a1a58563d86ae4\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f7f6112c9ad6433db5000fb9bf437c3c : SUCCESS in 10m 34s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0e56e06df7814e7e994dbbeaea5fed35 : SUCCESS in 9m 00s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/04fb8ff594464e76813c8d0a2aaff5e8 : SUCCESS in 9m 35s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"71f1006ec6745c52f797a061846e2c9f2cbac357","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 04:50:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/510506f7894b498d821fe426c885dff6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9fe82fc25d154099830eda8541361903 : SUCCESS in 7m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72211db04bb74a0da5496a7d702453be : SUCCESS in 6m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/62e5c98d9d27441e8757802df8d02f50 : SUCCESS in 7m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4fcc5d8907414b03bc75223d0aece700 : SUCCESS in 8m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ab6b4ebc599411cbc8144c74f49561f : SUCCESS in 5m 30s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/838cb4aa6d3c42dab5ef87806100bee9 : SUCCESS in 44m 31s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/15cdaaa374ae4bc080c24b83194a16d1 : SUCCESS in 1h 47m 46s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/0e10ab056f5e4fd1ae88037cf6a914f4 : SUCCESS in 1h 21m 11s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/1d1d47f0c1b942f7a6edc05531563670 : SUCCESS in 1h 40m 06s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/a02f578b5ad6495b9b940ad66290c7f1 : SUCCESS in 40m 10s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/62d59acfb1d64a3283160b3a7efe065e : SUCCESS in 1h 38m 06s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/e075fe792edb43478fa15b78dbbacc15 : SUCCESS in 1h 11m 06s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/cd9234d95af14487acd44c8451251208 : SUCCESS in 1h 23m 53s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/c11566c44a664963bfd9b55e7eab5e50 : SUCCESS in 48m 07s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/461e331cb9c94a259c125d2d77321450 : SUCCESS in 1h 30m 54s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/7f9d2993642940fa95c27ef6672f3a09 : SUCCESS in 1h 11m 05s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/cfe982f371d64c6d9380d81fb71b53c4 : SUCCESS in 1h 21m 24s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3d9dd336496e46979882d16bcd651f0e : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":4},{"id":"e20e2c5bff44506f3993c5351a41df04fc4e0116","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-12 04:54:29.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"a23dcd63a0b74dfd12c87b89b60ed889f0e810ef","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-01-12 06:16:38.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"d9b5d8329dfc501350799f2a8328a1c48ee846ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 06:17:13.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c9fe1266a2e8abdbb0584383bb9b3b581e23ab45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 07:01:20.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b96587bb55c4464bcbb2149887eb881\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0852a1e992b4ac3a8648034e4b7fbfd : SUCCESS in 5m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b1a9b89d53b143a5ac57208234d7a7f1 : SUCCESS in 3m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/da7777583d5c4696a0e7f454df4c34c7 : SUCCESS in 9m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5e2f7fac60c47e6ad1c1fef532b195d : SUCCESS in 5m 17s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/073b7c8b137f4838b8ab57984dfd91de : SUCCESS in 42m 20s\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/49ec4ff94b7847eaa88fcadc71873442 : SUCCESS in 39m 08s\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/ce092eadd93a4ce6aca161a89d2d5117 : SUCCESS in 43m 43s","accounts_in_message":[],"_revision_number":4},{"id":"932cfb00a7ded066452eaee3503b4a7fbc026a6a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 07:01:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"465da352fb899c082806fb1ad61591d907f1878e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 07:03:07.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef1b70dffa7f479e80d02e9050f2ecd1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dde9e77b57c4117af44dc1e7411e374 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c5db82ed3397e19283c5c2b7230bdfe0d1330f11","revisions":{"87c2ab0ae52408694823e406cdf53ba186d9b779":{"kind":"REWORK","_number":1,"created":"2026-01-07 12:04:14.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/60/972460/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/60/972460/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/60/972460/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/60/972460/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/60/972460/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/60/972460/1"}}},"commit":{"parents":[{"commit":"0243ca14069331630d94a5dfd8a42f6eb64ea05e","subject":"Merge \"pg_hba.conf not re-creating after rebuild in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0243ca14069331630d94a5dfd8a42f6eb64ea05e"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2026-01-07 12:01:22.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2026-01-07 12:03:53.000000000","tz":480},"subject":"Disable swift installation for cinder storage jobs","message":"Disable swift installation for cinder storage jobs\n\nThis commit also removes the deprecated jobs.\n\nChange-Id: I6e331f4f578ec0f6882ea90dc2c0f06ab54c09f3\nSigned-off-by: wu.chunyang \u003cwuchunyang@yovole.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/87c2ab0ae52408694823e406cdf53ba186d9b779"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/87c2ab0ae52408694823e406cdf53ba186d9b779"}]},"branch":"refs/heads/master"},"373698100e5875807f1ecdc1dc6fc0a68af858d7":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-01-07 14:34:49.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/60/972460/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/60/972460/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/60/972460/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/60/972460/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/60/972460/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/60/972460/2"}}},"commit":{"parents":[{"commit":"0243ca14069331630d94a5dfd8a42f6eb64ea05e","subject":"Merge \"pg_hba.conf not re-creating after rebuild in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0243ca14069331630d94a5dfd8a42f6eb64ea05e"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2026-01-07 12:01:22.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2026-01-07 14:33:40.000000000","tz":480},"subject":"Disable swift installation for cinder storage jobs","message":"Disable swift installation for cinder storage jobs\n\nThis commit also removes the deprecated jobs.\n\nDepends-On: https://review.opendev.org/c/openstack/trove/+/970640\n\nChange-Id: I6e331f4f578ec0f6882ea90dc2c0f06ab54c09f3\nSigned-off-by: wu.chunyang \u003cwuchunyang@yovole.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/373698100e5875807f1ecdc1dc6fc0a68af858d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/373698100e5875807f1ecdc1dc6fc0a68af858d7"}]},"branch":"refs/heads/master"},"214a0d94d0b830735ba4e13790fec5f7bdbf0512":{"kind":"REWORK","_number":3,"created":"2026-01-12 02:33:16.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/60/972460/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/60/972460/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/60/972460/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/60/972460/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/60/972460/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/60/972460/3"}}},"commit":{"parents":[{"commit":"35b336b43a8bf082288150e9d23a8a2be2a4a83d","subject":"Merge \"Fix: Allow float values in KeyValueCodec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/35b336b43a8bf082288150e9d23a8a2be2a4a83d"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2026-01-07 12:01:22.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2026-01-12 02:33:00.000000000","tz":480},"subject":"[CI] Disable swift installation for cinder storage jobs","message":"[CI] Disable swift installation for cinder storage jobs\n\nThis commit also removes the deprecated jobs.\n\nDepends-On: https://review.opendev.org/c/openstack/trove/+/970640\n\nChange-Id: I6e331f4f578ec0f6882ea90dc2c0f06ab54c09f3\nSigned-off-by: wu.chunyang \u003cwuchunyang@yovole.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/214a0d94d0b830735ba4e13790fec5f7bdbf0512"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/214a0d94d0b830735ba4e13790fec5f7bdbf0512"}]},"branch":"refs/heads/master"},"c5db82ed3397e19283c5c2b7230bdfe0d1330f11":{"kind":"REWORK","_number":4,"created":"2026-01-12 03:02:11.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/60/972460/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/60/972460/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/60/972460/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/60/972460/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/60/972460/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/60/972460/4"}}},"commit":{"parents":[{"commit":"35b336b43a8bf082288150e9d23a8a2be2a4a83d","subject":"Merge \"Fix: Allow float values in KeyValueCodec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/35b336b43a8bf082288150e9d23a8a2be2a4a83d"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2026-01-07 12:01:22.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2026-01-12 03:02:03.000000000","tz":480},"subject":"[CI] Disable swift installation for cinder storage jobs","message":"[CI] Disable swift installation for cinder storage jobs\n\nThis commit also removes the deprecated jobs.\n\nDepends-On: https://review.opendev.org/c/openstack/trove/+/970640\n\nChange-Id: I6e331f4f578ec0f6882ea90dc2c0f06ab54c09f3\nSigned-off-by: wu.chunyang \u003cwuchunyang@yovole.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c5db82ed3397e19283c5c2b7230bdfe0d1330f11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c5db82ed3397e19283c5c2b7230bdfe0d1330f11"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
