)]}'
{"id":"openstack%2Ftrove~972699","triplet_id":"openstack%2Ftrove~stable%2F2024.2~Ia29363ba9f7e543702fa29a497098e39d9ccedaa","project":"openstack/trove","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2026-01-28 01:16:01.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"Ia29363ba9f7e543702fa29a497098e39d9ccedaa","subject":"Fix CI jobs for stable 2024.2","status":"MERGED","created":"2026-01-08 13:58:42.000000000","updated":"2026-01-28 02:38:11.000000000","submitted":"2026-01-28 02:35:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"972699","meta_rev_id":"a85ce8ce3ab46fa37eabc398c772f055ffc82094","_number":972699,"virtual_id_number":972699,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-28 02:35:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-01-28 01:16:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-01-28 01:16:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-08 15:10:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"683670b860b26f3b422c01fcd929bf5bb9957e51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-01-08 13:58:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2542881fb2c42a5c5264cf4b913e0a8e4b7ff416","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 15:10:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2463eb5486cb44d7b527ba5e01975af3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4fe693044b2c423a8ad896684bb6702d : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4126694f8f334011ae558c1d67d4c591 : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b6a18900400e4e9e80fe656bf464c935 : SUCCESS in 6m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0acf30c8630a458288b4763abf16f41b : SUCCESS in 4m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/41c79cc3e55c4e4992c710b9ffc3b0ca : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/591d07fe158b4a57a33bbd423670e944 : SUCCESS in 4m 48s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/e82d2c97976241f394881ccb50bda23e : SUCCESS in 40m 26s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/a5ba386fb2f14204b467ff13b0d833c5 : FAILURE in 44m 22s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/e5bb4f4540954de6a68ca3313a3b40ff : FAILURE in 48m 21s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/bb2ef8cfba0149598bfa645fcdff1f1f : SUCCESS in 40m 33s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/8469c01231e04eed858c00fbf1a28167 : FAILURE in 41m 45s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/d1c981ea18bd4405963ef1c2b68eb93f : FAILURE in 49m 42s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/b9502b3814234750af51dbd7f5bf757e : FAILURE in 43m 44s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/bea09e1510004928a8aa6171eb9d3a6f : FAILURE in 44m 12s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/65657caf594d4ea59651f7348e99675b : FAILURE in 43m 08s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/34e644b20f37442ba55c43ffaf4a919b : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":1},{"id":"b8ff8aef7e6428b642761e4895236147d54d4c6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-01-26 13:48:50.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"9fffb9200fea56d15bfdc446800ce31680c72a7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-01-26 13:51:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"72b82780190a8e3e28deeff6209613390118b9df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 15:07:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b256ede2d904247bfcd27eddfbc2fd7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/583e329fad944913ae78224e4ea5ab02 : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/869a1b545c2641b09900a545508d70f3 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/31494f2137844a2a945dcf481003302e : SUCCESS in 6m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/855ac913a8084f3982462a81cc7a605d : SUCCESS in 4m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bbed21e5b635491d9613e59c19067f4c : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01d08a24849f4e87aadd409867e25fca : SUCCESS in 5m 19s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/c774cb8ff71b4887b5f24d4753187ee6 : SUCCESS in 56m 02s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/e61d2d615f794296a80941d199a71c39 : FAILURE in 1h 01m 44s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/135dec4cb04d4e35a7d6e286ebf5b8a7 : FAILURE in 1h 06m 11s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/471ddf33cb3e4fd4b5f00040ac37eb50 : SUCCESS in 41m 38s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/dadce1011c534b0ead41a162bd84c5cd : FAILURE in 53m 24s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/3b762fa1d55b4d9cafadfc6d1fe1e19f : FAILURE in 52m 08s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/a5a038f17d0449c7b7e4090779f494a6 : SUCCESS in 51m 20s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/2028f154bbf34ea7ae6f10eb5036d8e8 : FAILURE in 59m 16s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/b7e400222317485bad32cf8812445a95 : FAILURE in 53m 38s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/542b572a809747698fc171bc05031d18 : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":3},{"id":"ced8a75af67898a31c33e07d0f510af68145e140","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-01-27 02:51:57.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"d6c1e79ec8ba04c68c913c2afcac56b004dcc0ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 03:43:12.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/708add40f8284d4e88f3b1b70e13373d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c706d3ea7bc4d439af57f10f5bfe9e8 : SUCCESS in 8m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdb692945aae4f079201ae5cd95af2fb : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37ce5fedee024c1fa8205d3db4c047eb : SUCCESS in 9m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e96cc3f947144644b1c6a362f740e088 : SUCCESS in 5m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e72d5f12c353465e9f285f42c6f67cbb : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8517a50e4b24a4a944ff568452e45e1 : SUCCESS in 5m 15s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/8aba7b88ba324274ac84ba777162c98a : SUCCESS in 45m 04s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/e93adc0e243943da931dc57bf332ebdf : FAILURE in 24m 38s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/422c461031114331844b9d3f9a2253b7 : FAILURE in 40m 29s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/5255056b331a4707b397fcc013523713 : SUCCESS in 39m 26s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/9c12734f17314c23991a6c0c9a9a963a : FAILURE in 24m 07s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/b915bdc4b7214bcb9d5370fce78e3ed7 : FAILURE in 25m 43s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/2ceb74cfe49f47fa9a1acae6da764faa : SUCCESS in 44m 19s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/23fdd6207a2b47d0b3aed3342a60de87 : FAILURE in 23m 42s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/1acb19dcc68840fcb94dc5651b6db98e : FAILURE in 24m 53s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0b8469331f2141aeb29b49642e7204b2 : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":4},{"id":"bfe2358dda70f30583b5b6e06a6de737407975b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-01-27 06:32:40.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"a8a1804de52d630f77aa870e47e05804edef9404","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 08:51:00.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b45babc1cea44276a8b2ab155e98e2dc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f4d495f308d644b1845dfb43cce43f0e : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d506b3603cec44438aa05f8afd5a9693 : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/029fc10bdddb4ccf87559aacc0dc6e6a : SUCCESS in 5m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4e1620dd83924e7fb2f8c647e2a4cb43 : SUCCESS in 5m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/61d763b55d974479a25fe027f2c277ea : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b822049c2b894cd78adaa0ea0382e840 : SUCCESS in 3m 32s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/76fc6010fa0c41f68b2768771e327cc6 : SUCCESS in 54m 48s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/eee1740efd90438a905eb6534190c67b : SUCCESS in 2h 15m 35s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/90cd6a8aaf5e453eae8d96f518c11b3f : SUCCESS in 1h 47m 35s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/49bad742e833432690d4f5da153a1952 : SUCCESS in 57m 37s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/26cd9e9a6cab42fabd7c57f62adf596a : FAILURE in 1h 34m 56s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/c505aa5b796d4a82883c9f712ed0995f : FAILURE in 1h 13m 30s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/2747b0f53d9d47cdad80527b7581cddf : SUCCESS in 52m 33s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/cdcc8b4abde14a63a66adf8cd07f8917 : FAILURE in 1h 11m 30s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/3cbb9316ba23431c80ee0548250132aa : SUCCESS in 1h 36m 02s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0ecadb23ab4b428abdbc7b22ead2cc3d : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":5},{"id":"8f673d40cff1d983eba5e4f62fc7e014cd2e786f","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-01-28 01:16:01.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"19e6573b7a6d71bc25a35b00fd37a55506d7b873","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 01:16:32.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"5a6e868c28c7dbde84bb262eb4e4dd443abc8312","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 02:35:43.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16e7ce6e61514c29b3b84119c8ca4b97\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2939528f94b24cedbebacf435d394a89 : SUCCESS in 2m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd315d65f53d47488ba77ba713341006 : SUCCESS in 3m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a7c656ae6ac0456885223a7be0038429 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c39abc2259694cef9bfec0e41b374f43 : SUCCESS in 4m 44s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/29dfe01223744498a0da44aa63cbec91 : SUCCESS in 44m 11s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/920fc66cc8114529a09a7adae3abcef6 : SUCCESS in 1h 18m 32s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/37354ec34c1a4138afa95b539329d871 : SUCCESS in 48m 50s","accounts_in_message":[],"_revision_number":5},{"id":"25f355d3c654c131b6e38b1d815b6fe759131cc9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 02:35:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"a85ce8ce3ab46fa37eabc398c772f055ffc82094","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 02:38:11.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82df9f793488483dbb2579684580b059\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/869e06d02a1e413c880c20027a8bfc66 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"52b1feac8da430082ed6b739c88b4d082ea0f3be","revisions":{"9ade889d145dbd3dbf5dcf339a4d7975a2340b89":{"kind":"REWORK","_number":1,"created":"2026-01-08 13:58:42.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/99/972699/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/99/972699/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/99/972699/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/99/972699/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/99/972699/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/99/972699/1"}}},"commit":{"parents":[{"commit":"93efb5f57f7bc6f683c5a5829ba9d7c955b30219","subject":"[CI]: run perodical jobs on master branch only","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/93efb5f57f7bc6f683c5a5829ba9d7c955b30219"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2026-01-08 13:56:12.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2026-01-08 13:58:21.000000000","tz":480},"subject":"Don\u0027t build backup images in CI","message":"Don\u0027t build backup images in CI\n\nAs Ubuntu 20.04 is no longer supported, we use prebuilt\nimages for CI jobs.\n\nChange-Id: Ia29363ba9f7e543702fa29a497098e39d9ccedaa\nSigned-off-by: wu.chunyang \u003cwuchunyang@yovole.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9ade889d145dbd3dbf5dcf339a4d7975a2340b89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9ade889d145dbd3dbf5dcf339a4d7975a2340b89"}]},"branch":"refs/heads/stable/2024.2"},"214ef34f44eb1da58f281f4937e47107308a1fac":{"kind":"REWORK","_number":2,"created":"2026-01-26 13:48:50.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/99/972699/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/99/972699/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/99/972699/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/99/972699/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/99/972699/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/99/972699/2"}}},"commit":{"parents":[{"commit":"93efb5f57f7bc6f683c5a5829ba9d7c955b30219","subject":"[CI]: run perodical jobs on master branch only","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/93efb5f57f7bc6f683c5a5829ba9d7c955b30219"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2026-01-08 13:56:12.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2026-01-26 13:48:17.000000000","tz":480},"subject":"Don\u0027t build backup images in CI","message":"Don\u0027t build backup images in CI\n\nAs Ubuntu 20.04 is no longer supported, we use prebuilt\nimages for CI jobs.\n\nChange-Id: Ia29363ba9f7e543702fa29a497098e39d9ccedaa\nSigned-off-by: wu.chunyang \u003cwuchunyang@yovole.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/214ef34f44eb1da58f281f4937e47107308a1fac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/214ef34f44eb1da58f281f4937e47107308a1fac"}]},"branch":"refs/heads/stable/2024.2"},"00c3cb65904c93144463e1149e599bd5720e73c1":{"kind":"REWORK","_number":3,"created":"2026-01-26 13:51:01.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/99/972699/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/99/972699/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/99/972699/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/99/972699/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/99/972699/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/99/972699/3"}}},"commit":{"parents":[{"commit":"93efb5f57f7bc6f683c5a5829ba9d7c955b30219","subject":"[CI]: run perodical jobs on master branch only","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/93efb5f57f7bc6f683c5a5829ba9d7c955b30219"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2026-01-08 13:56:12.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2026-01-26 13:50:47.000000000","tz":480},"subject":"Don\u0027t build backup images in CI","message":"Don\u0027t build backup images in CI\n\nAs Ubuntu 20.04 is no longer supported, we use prebuilt\nimages for CI jobs.\n\nChange-Id: Ia29363ba9f7e543702fa29a497098e39d9ccedaa\nSigned-off-by: wu.chunyang \u003cwuchunyang@yovole.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/00c3cb65904c93144463e1149e599bd5720e73c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/00c3cb65904c93144463e1149e599bd5720e73c1"}]},"branch":"refs/heads/stable/2024.2"},"20edad7d6909d21b2392f8c040f0141ff85a5ed9":{"kind":"REWORK","_number":4,"created":"2026-01-27 02:51:57.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/99/972699/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/99/972699/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/99/972699/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/99/972699/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/99/972699/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/99/972699/4"}}},"commit":{"parents":[{"commit":"93efb5f57f7bc6f683c5a5829ba9d7c955b30219","subject":"[CI]: run perodical jobs on master branch only","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/93efb5f57f7bc6f683c5a5829ba9d7c955b30219"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2026-01-08 13:56:12.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2026-01-27 02:50:54.000000000","tz":480},"subject":"Fix CI jobs for stable 2024.2","message":"Fix CI jobs for stable 2024.2\n\nAs Ubuntu 20.04 is no longer supported, this patchs uses\nUbuntu 22.04 to build the images.\n\nChange-Id: Ia29363ba9f7e543702fa29a497098e39d9ccedaa\nSigned-off-by: wu.chunyang \u003cwuchunyang@yovole.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/20edad7d6909d21b2392f8c040f0141ff85a5ed9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/20edad7d6909d21b2392f8c040f0141ff85a5ed9"}]},"branch":"refs/heads/stable/2024.2"},"52b1feac8da430082ed6b739c88b4d082ea0f3be":{"kind":"REWORK","_number":5,"created":"2026-01-27 06:32:40.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/99/972699/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/99/972699/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/99/972699/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/99/972699/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/99/972699/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/99/972699/5"}}},"commit":{"parents":[{"commit":"93efb5f57f7bc6f683c5a5829ba9d7c955b30219","subject":"[CI]: run perodical jobs on master branch only","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/93efb5f57f7bc6f683c5a5829ba9d7c955b30219"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2026-01-08 13:56:12.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2026-01-27 06:32:24.000000000","tz":480},"subject":"Fix CI jobs for stable 2024.2","message":"Fix CI jobs for stable 2024.2\n\nAs Ubuntu 20.04 is no longer supported, this patchs uses\nUbuntu 22.04 to build the images.\n\nChange-Id: Ia29363ba9f7e543702fa29a497098e39d9ccedaa\nSigned-off-by: wu.chunyang \u003cwuchunyang@yovole.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/52b1feac8da430082ed6b739c88b4d082ea0f3be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/52b1feac8da430082ed6b739c88b4d082ea0f3be"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
