)]}'
{"id":"openstack%2Ftrove~973012","triplet_id":"openstack%2Ftrove~stable%2F2025.2~I4abee725caa762d031172d484ea93fb5fd34d6c4","project":"openstack/trove","branch":"stable/2025.2","topic":"bug/configuration-parameter-float-value-error-stable/2025.2","attention_set":{},"removed_from_attention_set":{"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2026-01-12 01:52:22.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-01-12 03:33:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4abee725caa762d031172d484ea93fb5fd34d6c4","subject":"Fix: Allow float values in KeyValueCodec","status":"MERGED","created":"2026-01-11 10:28:57.000000000","updated":"2026-01-12 03:34:46.000000000","submitted":"2026-01-12 03:33:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973012-bug/configuration-parameter-float-value-error-stable/2025.2","cherry_pick_of_change":970826,"cherry_pick_of_patch_set":6,"meta_rev_id":"16125b086f3ec4b8ce7b0ca9211a249a1bded175","_number":973012,"virtual_id_number":973012,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-12 03:33:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-01-12 01:52:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-01-12 01:52:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"CC":[{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-11 10:28:57.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"state":"CC"},{"updated":"2026-01-11 10:29:43.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2026-01-11 12:21:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8c83a52a16df3afe649284b38b9e7745ea5e685b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-11 10:28:57.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ec6b9d2f3ec573b2224c6176aefc4bb0496e1d1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-11 12:21:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2340e416ca2a4ffc88745f0feaa48a29\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/178b1cfe31d0407481004c7d8bb586c7 : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba7f2388d9844480b1806412722c28ee : SUCCESS in 7m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ff027f47ccef449595ce56cee47f23b3 : SUCCESS in 3m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e0715beea0074593bba068233fb86122 : SUCCESS in 3m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bdec93804834494393e9e7076fac33a1 : SUCCESS in 8m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27fa6dd53e5b479cb8d913c7531bf354 : SUCCESS in 3m 39s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/aeffc23e04ac4da2854dfbe3444758a7 : SUCCESS in 52m 08s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/8b01174b03284c9291307fb3dd77adb6 : SUCCESS in 1h 51m 19s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/b94a3007c69148f58cf21b2e99c4c9f5 : SUCCESS in 1h 31m 09s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/b7ce3751769b49e893e267efbd978b14 : SUCCESS in 1h 45m 13s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/09d2ffd6184f460eaccf1188116ec8a0 : SUCCESS in 48m 08s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/bfdb85f563894601abfc5aa4be82cb33 : SUCCESS in 1h 32m 35s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/2ba9b2feff604c9dba078a6d62adb736 : SUCCESS in 1h 14m 32s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/474008ebb40e4fbf82994356b7e71a4f : SUCCESS in 1h 34m 23s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/8f2d25caad7e4fe38e783ac51ca1f82b : SUCCESS in 42m 58s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/43e5c24340da4ab39f02b231181828f5 : SUCCESS in 1h 29m 05s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/2a6f0770dda64b2b8425d15e42189cb3 : SUCCESS in 1h 13m 29s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/cebe59b683954839b65b251906dc85d8 : SUCCESS in 1h 17m 08s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/683afe3c83534041a9176e3cf4901d5e : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":1},{"id":"646ba062fc1c1e19146e8727f45b2198b69e80cb","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-01-12 01:52:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"20f56f55874a409043ffac0ffbb52b2862c8d827","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 01:52:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"384b3fd5f9261b25b452de1f26aa2862b15d27a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 03:33:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/311e23d8a0bb4770ad2ac7227b29c487\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8480cc4384b4ce9bb77c072ab6d781d : SUCCESS in 7m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7450cbc0c3a24448852d9de1ba545e7d : SUCCESS in 6m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0155f8da7246487595d5e63b536d5d39 : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a81007bb56dd437a99e76ce284966034 : SUCCESS in 5m 30s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/df53e03ff94c47e1ac379540fdb1d503 : SUCCESS in 46m 19s\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/63c5384f2e29437fbb334db982222b76 : SUCCESS in 41m 59s\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/5685871e01e24fe08bae78839b743d1b : SUCCESS in 44m 32s","accounts_in_message":[],"_revision_number":1},{"id":"5a04cb212c5ba238a84db847f308ae73352900c4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 03:33:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"16125b086f3ec4b8ce7b0ca9211a249a1bded175","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 03:34:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c07fb119b9b4566833ede4e7fbd152e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fc994e1500940839fbdebd9ff9eda65 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a7fe2edf5a8bfa235111edf0839cf64ca49ac520","revisions":{"a7fe2edf5a8bfa235111edf0839cf64ca49ac520":{"kind":"REWORK","_number":1,"created":"2026-01-11 10:28:57.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/12/973012/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/12/973012/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/12/973012/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/12/973012/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/12/973012/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/12/973012/1"}}},"commit":{"parents":[{"commit":"3b6026a4eaaed02492c0ba65e85c358e61c7c497","subject":"Merge \"Fix periodic image upload jobs\" into stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3b6026a4eaaed02492c0ba65e85c358e61c7c497"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2025-12-12 12:00:08.000000000","tz":300},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-11 10:28:57.000000000","tz":0},"subject":"Fix: Allow float values in KeyValueCodec","message":"Fix: Allow float values in KeyValueCodec\n\nFix \"AttributeError: \u0027float\u0027 object has no attribute\n\u0027lower\u0027\" in KeyValueCodec when attempting to set a float\nvalue for a configuration parameter with \"float\" type.\n\nThe `is_valid_value` method was incorrectly attempting to\ncall `.lower()` on a float, causing a crash. This patch\nadds an explicit check for floats to prevent this error.\n\nChange-Id: I4abee725caa762d031172d484ea93fb5fd34d6c4\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n(cherry picked from commit ece29cea9dbae80f2b809663a6cd379ec2383344)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a7fe2edf5a8bfa235111edf0839cf64ca49ac520"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a7fe2edf5a8bfa235111edf0839cf64ca49ac520"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
