)]}'
{"id":"openstack%2Ftrove~973105","triplet_id":"openstack%2Ftrove~master~I5d52370c38022a95dc2cc58c55a221104adb0a03","project":"openstack/trove","branch":"master","topic":"fix-misleading-volume-usage-instance-show","attention_set":{},"removed_from_attention_set":{"36080":{"account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"last_update":"2026-06-23 06:59:03.000000000","reason":"\u003cGERRIT_ACCOUNT_36080\u003e replied on the change","reason_account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},"31737":{"account":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"last_update":"2026-06-20 00:17:30.000000000","reason":"\u003cGERRIT_ACCOUNT_31737\u003e replied on the change","reason_account":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2026-06-23 07:45:14.000000000","reason":"Change was submitted"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-06-23 07:45:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5d52370c38022a95dc2cc58c55a221104adb0a03","subject":"Improve data volume response in PostgreSQL","status":"MERGED","created":"2026-01-12 14:52:45.000000000","updated":"2026-06-23 07:46:22.000000000","submitted":"2026-06-23 07:45:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973105-fix-misleading-volume-usage-instance-show","meta_rev_id":"33371cffed75a2a2d5b2b26295472eb46e5b68e3","_number":973105,"virtual_id_number":973105,"owner":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-23 07:45:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":2,"date":"2026-06-23 06:59:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":1,"date":"2026-06-20 00:17:30.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":1,"date":"2026-06-23 06:59:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":-1,"date":"2026-06-23 06:59:03.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"CC":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-12 15:05:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-01-12 16:38:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-12 17:27:14.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-01-12 17:27:14.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"CC"},{"updated":"2026-01-12 17:27:14.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"state":"REVIEWER"},{"updated":"2026-01-12 17:27:14.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2026-04-16 08:48:50.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"0f05aea4181706683356411abf99099b827b8e20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-01-12 14:52:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"73f63f63bf4bbb0d09017a3d251e6c1e00d4034d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 15:05:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8757fb3e61384ffea9fec031c922ab2c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/01cbdef336504a56bd41cd4cf211b694 : SUCCESS in 7m 34s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fe67477e2fe14af2bb9ad52c73422282 : SUCCESS in 8m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/40a77739e6c2487ebcb1a3ecc34de57c : SUCCESS in 10m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4a3f880e74906ae870247610b8a8b56ade78a027","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 16:38:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3e6828b8149439481eae8d1741bde20\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/67c0af4a511f49dca04e80bdeaa512b4 : SUCCESS in 3m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4263542b12004c71b5143d9a1af9dc1b : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d43a64d41653488ca31c615be7866368 : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0d3c5e1ffb274014905ad78636fb296d : SUCCESS in 6m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/945db4f9d9b945cd9c8b74a939fe42c6 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07eda4b15e964f9090e21cb522fe7979 : SUCCESS in 3m 22s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/9a39683a4f2d48bd90eb2f4b0a1b2103 : SUCCESS in 47m 25s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/b825a395ba1940b8a1dfec7d04af4712 : SUCCESS in 1h 45m 02s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/c69b4dd7cac24a33b9caeca80b2ffd4f : SUCCESS in 1h 25m 56s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/a17270550a00430e9ae2eb53029166cf : SUCCESS in 1h 44m 42s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/9929c1b9b7294264bf05f9fb368da1bf : SUCCESS in 43m 33s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/c4315d05ddd447a383db0473404237f5 : SUCCESS in 1h 35m 33s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/c5b0db8f34c74b82924a29a4ff923d2b : SUCCESS in 1h 34m 12s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/1d69f3afa6a248e88a96c35ece025f58 : SUCCESS in 1h 25m 04s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/270e5dd7670842b6ad4ec7a7b1b01d48 : SUCCESS in 1h 01m 16s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/8ba31a1d29af4a168f90c1a98a95da7c : SUCCESS in 1h 30m 30s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/5a6dae219d7544e1816c24b8e648194f : SUCCESS in 1h 11m 13s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/e0f157c6afe347a4bd631789baf80a06 : SUCCESS in 1h 23m 21s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0adb9b65dfdd44669b4ecb0d1970565e : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":1},{"id":"defe6aec9dcf5a54aeef45cd28914c288bfed32c","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2026-03-17 19:58:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eb9e269017b4a575f24d84c8bb9eff1e5688d9b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-03-18 14:15:18.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"70ca0e372c3c9764dfbc2307199b28e39e48c290","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 14:58:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39df143495ff4681b3484718d604daf5\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/649c92097f40480b97185ef83b83cf34 : SUCCESS in 8m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2a5191fcf35f4c4198bc969ed47091fd : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/710feec665f14096bd349db93485c514 : SUCCESS in 15m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c6fb9219f0572c01ab9c54a90de78f8222d8ac0e","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-03-18 15:45:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b7fe266b764d1c7c965323edffb3a6ff7fec9b97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 17:32:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6ed1b5a9a829470e94862ccdd04762ca\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/cf233e0761bd4f558f8b848deac6a838 : SUCCESS in 5m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6884bd885c9346d9972699432f53369d : SUCCESS in 8m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40fdb0b55d7a4ff6b8024d2bc4529def : SUCCESS in 6m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c0bb2320175147c782af670e7d9f98be : SUCCESS in 5m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9c380ce5e7ff47c9bfacf497a92bc277 : SUCCESS in 5m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/944a9867de404da7afd207e16c6f77e8 : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/007f8d638c2c481b919b02c0e75232cc : SUCCESS in 5m 27s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/805ff760b7bf49e0b41e34fe8e9ccb8c : FAILURE in 2h 26m 36s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/86d6b8b05d45418283adbba8633c83f0 : FAILURE in 1h 17m 01s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/118acac767e349eab4772cbc8286dc85 : TIMED_OUT in 3h 05m 59s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/52943f996b834cd2a9616f9396ec6f08 : TIMED_OUT in 3h 05m 19s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/922d3c19efa940c89fb7ecfe9c64a1e7 : SUCCESS in 28m 42s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/b24395cf3a9f4632b2531291788f8078 : TIMED_OUT in 3h 03m 21s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/827a08a63a9c43669684b480a4ab9935 : FAILURE in 1h 28m 01s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/f5aaf008ee4a4c91aedf6e6f94da6356 : FAILURE in 1h 39m 58s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/b1be6646efc046d0850e46d587f9f73d : FAILURE in 2h 45m 14s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/54457113a3884e4cb112c5047ef92c05 : TIMED_OUT in 3h 04m 53s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/b672007a7f93441f8bd6a39a0fe53d51 : SUCCESS in 58m 18s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/8a3d9cdc69cc4cf0bdc70b57f5a6e93a : FAILURE in 54m 52s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ff4fc474c4224f65a1e156a5a05e3309 : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":2},{"id":"ee8510148d203f87efd79abe32960628873793ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"real_author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-15 20:02:13.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_36080\u003e\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}],"_revision_number":3},{"id":"b427b5474dc8f726925dc6f4d6bcf7992b9b0c27","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 20:38:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c18714832d704b78aff7b9b016a304cd\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4900bfcfe8cb441185f2b05a16ffbfc6 : SUCCESS in 9m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cb1106bf019b4c62b53c072bef5eee4f : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b37d3555d9104edcb343064b1071bc67 : SUCCESS in 8m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"77ece1ab55c603e957617a4b6f40ad773d28814d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 21:54:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf21570eab5a4d19965dc112117a6306\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9efdbe3846b3403997eae702b656ba7e : SUCCESS in 6m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d2ef454d9c04f5bb8eae1fdefb9dab3 : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/758da0642cfb4469803adaffcef52cbf : SUCCESS in 6m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d2959fc701e745e58ae64ff3de191be1 : SUCCESS in 4m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8e88a97a989c4c76abc74cecd924c9e7 : SUCCESS in 6m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d1a30fbd55024c34abc198ebff90a0c2 : SUCCESS in 9m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d3bb7830b3c4db09b5e757cd6a1f5dd : SUCCESS in 5m 28s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/22742495662c43be8d4ecca801be4205 : SUCCESS in 50m 07s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/50e0100546164f2888fc7f3cf6e9c5b6 : SUCCESS in 1h 51m 26s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/83f43a2fae0243e39555c9d54c23c137 : SUCCESS in 1h 31m 58s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/e1d23f43b2af4f7980656ec99ca741dc : SUCCESS in 1h 42m 34s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/bcbf22350056456f812b3515f99f2d77 : SUCCESS in 40m 22s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/68ec3de84d7a43c591b33003bf364ddd : SUCCESS in 1h 29m 03s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/71ce6f1438304a4898b978a940905265 : SUCCESS in 1h 13m 55s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/92f6ab015a984c309dd6649b813d522a : SUCCESS in 1h 23m 10s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/36539355cd014bb4910fc658e3931f49 : SUCCESS in 48m 31s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/927c4508ec564f3d8ff970a605034fb7 : SUCCESS in 1h 32m 45s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/3394b13c72ea4bd386603f9913ec0241 : SUCCESS in 1h 11m 55s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/089b07aba5294d5e971df095efd21af7 : SUCCESS in 1h 22m 08s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/03738784a38241f2b3f29786ea3f4040 : SUCCESS in 6m 48s","accounts_in_message":[],"_revision_number":3},{"id":"ea11955fcc07b9d17599536a44016e9efef9ebe2","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-16 08:48:50.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"f354957a551077026cde6bf04c7ca16e1b1052b0","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2026-06-16 12:39:51.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cd1cb6425f3aaf806374be6c4d53518f98665207","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-16 12:43:48.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"064b1fe640023b0c8ff5f83331e94f7abac39338","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 12:43:55.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a04e1d69bb5340ffb9bb271c70544550\n\n\nWarning:\n  Dependency cycle detected and project openstack/python-troveclient doesn\u0027t allow circular dependencies","accounts_in_message":[],"_revision_number":4},{"id":"45f4b20fc04ebb958b3ed26212246fefffdf7e88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 12:43:56.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/82747f3ca7964fa694ad4f38100b72a9\n\n\nWarning:\n  Dependency cycle detected and project openstack/python-troveclient doesn\u0027t allow circular dependencies","accounts_in_message":[],"_revision_number":4},{"id":"3e5970962027c5fa97e2b2ab9f8f16a45cb0449e","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-16 12:45:44.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9fceef2d43b5f85d230776775d7cc519d12bf1af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-16 12:46:16.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"a32085bef2edac0e4103f615d3940da6148295c7","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-16 12:47:03.000000000","message":"Patch Set 5: Backport-Candidate-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"fcdfcc92d41bf12de3cb16b905a5d3cee3e31625","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2026-06-16 12:51:53.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"78e5e369539b6826a7ca17a4f7fb242993ebb23a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-16 12:54:01.000000000","message":"Uploaded patch set 6.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR **is:MIN**\")\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":6},{"id":"bde5ff99819ff6a05e6268efda775715c2080b7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-16 12:57:27.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR is:MAX OR **is:MIN**\")\n","accounts_in_message":[],"_revision_number":7},{"id":"16e7eb7f1b27700f72ac3331e42bff5eaf30d644","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-16 12:57:38.000000000","message":"Patch Set 6: Backport-Candidate-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b1c3c0e33a8028fed73f6d6c705e47327c6db6ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 15:59:45.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3afd981243f846e38428daac40ca4fb9\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1357e2ecfd5740b6996d1fef0943e490 : SUCCESS in 7m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ae227415355a4bbabbd35f932a34dff8 : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/93ac55ec0d164d4db3ca254aeada2f6e : SUCCESS in 12m 38s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"7014a84185f83ae03e4d72ccbd2534bb80730add","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 16:13:42.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59e92344d4aa406ebeccc1ea7a77997d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/15740457f706487f9af0ec92485f8f1e : SUCCESS in 5m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/009d16fc30654fce9c7007437ace7cff : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/368a197728eb48359084bdfc87256d25 : SUCCESS in 6m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/792e92743cac41639003b8c60741ea20 : SUCCESS in 5m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2c270f7c5d444d0ea0685667b1782981 : SUCCESS in 6m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a9e0ffc5cd6b4a0db2f79e4ef542e72c : SUCCESS in 7m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43a6bb2f99b84da98805c039ef3af4db : SUCCESS in 11m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/887be8f754a44e49b4559b7ec349b7a0 : SUCCESS in 4m 01s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/8a19ac4007f64c2fa80de244e2c648c3 : SUCCESS in 1h 05m 57s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/d273b0c098a5460fb794a05a119f5766 : FAILURE in 57m 46s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/6aabafb1277e423bbb77d9d3f776f9c7 : FAILURE in 57m 34s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/f8e3345aa21d4f66b2fa65b69357577d : FAILURE in 1h 04m 07s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/6effa3dddaa2494088c89472ca5004d3 : SUCCESS in 42m 07s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/3a18989c66584a7fb35b2d419c94e416 : FAILURE in 1h 01m 40s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/dd24897180bb44f9850179c3a066899b : FAILURE in 49m 52s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/9eec20d6b96c409ca37b8f418d68ed16 : SUCCESS in 1h 27m 50s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/b3d6d297b7c648a88409d6c1c9b321b0 : SUCCESS in 45m 41s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/a26b8589d208432a920ac569490e4e2b : FAILURE in 59m 36s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/aecb8255e21447c68dac94b0583a3bfe : FAILURE in 55m 40s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/aa9969749f1e4bd8aedac6614281827d : FAILURE in 1h 46m 57s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/486e778dda80498e8f49bdba9b4f54af : SUCCESS in 7m 12s","accounts_in_message":[],"_revision_number":7},{"id":"72f571dbd09a6dd462e9893cb0700ccc6249fb07","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2026-06-20 00:17:30.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"5792aebb1419e7fc1b34859c077146e9fe9b7dee","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-23 06:59:03.000000000","message":"Patch Set 7: Backport-Candidate-1 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"047ae3a3ee6f2748bbd002b57a790fff20b7c378","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 06:59:27.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"0fe3abd3ef6d92b632a049c3149b0231a035b641","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 07:45:14.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2bfcc34defb4243aa0bc9c984615d8b\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9bb4cadbf8714fc98b0f86c2882e93e8 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/195837c952df428e8827bf30a50107cc : SUCCESS in 5m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/549e0f65ee3745bea4990528631b2951 : SUCCESS in 3m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/46fe5459961248ea9730caf9628a3bb5 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/130ecbb4d31c4151bcf649fa3de7fdc8 : SUCCESS in 6m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9374e1748923408d921f63790477e04b : SUCCESS in 7m 16s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/d82f3b66e16641beacdf3545ecab8385 : SUCCESS in 45m 24s\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/613ea5ab88fa4d9bb0a8bdc42060f237 : SUCCESS in 40m 44s\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/fe021c9548d94349a8411a1c8030afa3 : SUCCESS in 45m 08s","accounts_in_message":[],"_revision_number":7},{"id":"51c2a159820b1cfa7c2832d3feb875481e729fbf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 07:45:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"33371cffed75a2a2d5b2b26295472eb46e5b68e3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 07:46:22.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f30350f1b17c40aa96a142c97db9d63b\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/50e801e2046a4b7b900bc52b1c65dc30 : SUCCESS in 54s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/603eb060f1ad49e4a8806f0b36295b21 : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/402ba2751a5047dfae7ffc2f7d45f7bd : SUCCESS in 50s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"e66c2f417e50b32519aa9b9212c898f9a63bd7e8","revisions":{"35dceef2b6b3d1f3edea6131cdc254b5518b9c0c":{"kind":"REWORK","_number":1,"created":"2026-01-12 14:52:45.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/05/973105/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/05/973105/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/973105/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/973105/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/973105/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/05/973105/1"}}},"commit":{"parents":[{"commit":"35b336b43a8bf082288150e9d23a8a2be2a4a83d","subject":"Merge \"Fix: Allow float values in KeyValueCodec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/35b336b43a8bf082288150e9d23a8a2be2a4a83d"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-01-12 14:30:03.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-01-12 14:50:28.000000000","tz":300},"subject":"Fix misleading volume usage reporting and expose real disk exhaustion","message":"Fix misleading volume usage reporting and expose real disk exhaustion\n\nPreviously the API reported only total/used volume values calculated\nfrom f_bfree, ignoring blocks unavailable to unprivileged users.\nAdditional free space was effectively lost due to GB -\u003e GiB conversion.\nThis often created an illusion of remaining space while the real\nfilesystem was already at 100% usage and databases were close to\nfailure or read-only mode.\n\nThis change adds a real used percentage based on f_bavail and exposes\nextra PostgreSQL volume details (WAL and WAL archive sizes).\n(Changes in python-troveclient will follow)\n\nStory: 2011633\n\nChange-Id: I5d52370c38022a95dc2cc58c55a221104adb0a03\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/35dceef2b6b3d1f3edea6131cdc254b5518b9c0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/35dceef2b6b3d1f3edea6131cdc254b5518b9c0c"}]},"branch":"refs/heads/master"},"69444dd22016f834f3b8068a8806a4eb29bcb4e6":{"kind":"REWORK","_number":2,"created":"2026-03-18 14:15:18.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/05/973105/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/05/973105/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/973105/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/973105/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/973105/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/05/973105/2"}}},"commit":{"parents":[{"commit":"d8bd151836576035e80170ffac18305d1635a896","subject":"Merge \"Allow to have less specific versioned backup images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d8bd151836576035e80170ffac18305d1635a896"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-01-12 14:30:03.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-03-18 14:15:02.000000000","tz":300},"subject":"Fix misleading volume usage reporting and expose real disk exhaustion","message":"Fix misleading volume usage reporting and expose real disk exhaustion\n\nPreviously the API reported only total/used volume values calculated\nfrom f_bfree, ignoring blocks unavailable to unprivileged users.\nAdditional free space was effectively lost due to GB -\u003e GiB conversion.\nThis often created an illusion of remaining space while the real\nfilesystem was already at 100% usage and databases were close to\nfailure or read-only mode.\n\nThis change adds a real used percentage based on f_bavail and exposes\nextra PostgreSQL volume details (WAL and WAL archive sizes).\n(Changes in python-troveclient will follow)\n\nStory: 2011633\n\nChange-Id: I5d52370c38022a95dc2cc58c55a221104adb0a03\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/69444dd22016f834f3b8068a8806a4eb29bcb4e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/69444dd22016f834f3b8068a8806a4eb29bcb4e6"}]},"branch":"refs/heads/master"},"cd32a92f483484653cd6f6280b78591aa782ff9c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-04-15 20:02:13.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"real_uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/05/973105/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/05/973105/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/973105/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/973105/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/973105/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/05/973105/3"}}},"commit":{"parents":[{"commit":"7a2cb21e8b3035835ae1260c9c7e676191360562","subject":"Improve instance list method by using Nova tags feature.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7a2cb21e8b3035835ae1260c9c7e676191360562"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-01-12 14:30:03.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-04-15 20:02:13.000000000","tz":0},"subject":"Fix misleading volume usage reporting and expose real disk exhaustion","message":"Fix misleading volume usage reporting and expose real disk exhaustion\n\nPreviously the API reported only total/used volume values calculated\nfrom f_bfree, ignoring blocks unavailable to unprivileged users.\nAdditional free space was effectively lost due to GB -\u003e GiB conversion.\nThis often created an illusion of remaining space while the real\nfilesystem was already at 100% usage and databases were close to\nfailure or read-only mode.\n\nThis change adds a real used percentage based on f_bavail and exposes\nextra PostgreSQL volume details (WAL and WAL archive sizes).\n(Changes in python-troveclient will follow)\n\nStory: 2011633\n\nChange-Id: I5d52370c38022a95dc2cc58c55a221104adb0a03\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cd32a92f483484653cd6f6280b78591aa782ff9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cd32a92f483484653cd6f6280b78591aa782ff9c"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"69444dd22016f834f3b8068a8806a4eb29bcb4e6","theirs":"7a2cb21e8b3035835ae1260c9c7e676191360562","contains_conflicts":false}},"73980ef4803a73635a12724d895377c8cbf61cf4":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-06-16 12:43:48.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/05/973105/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/05/973105/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/973105/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/973105/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/973105/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/05/973105/4"}}},"commit":{"parents":[{"commit":"7a2cb21e8b3035835ae1260c9c7e676191360562","subject":"Improve instance list method by using Nova tags feature.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7a2cb21e8b3035835ae1260c9c7e676191360562"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-01-12 14:30:03.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-16 12:43:48.000000000","tz":0},"subject":"Improve data volume response in Postgres","message":"Improve data volume response in Postgres\n\nPreviously, the API reported only total/used volume values calculated\nfrom f_bfree, ignoring blocks unavailable to unprivileged users.\nAdditional free space was effectively lost due to GB -\u003e GiB conversion.\nThis often created an illusion of remaining space while the real\nfilesystem was already at 100% usage, and databases were close to\nfailure or read-only mode.\n\nThis change adds a real used percentage based on f_bavail and exposes\nextra PostgreSQL volume details (WAL and WAL archive sizes).\n\nStory: 2011633\n\nDepends-On: https://review.opendev.org/c/openstack/python-troveclient/+/973106\nChange-Id: I5d52370c38022a95dc2cc58c55a221104adb0a03\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/73980ef4803a73635a12724d895377c8cbf61cf4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/73980ef4803a73635a12724d895377c8cbf61cf4"}]},"branch":"refs/heads/master","description":"Edit commit message","conflicts":{"ours":"69444dd22016f834f3b8068a8806a4eb29bcb4e6","theirs":"7a2cb21e8b3035835ae1260c9c7e676191360562","contains_conflicts":false}},"61131f20a73d3cef2f7e657d4a3582c86b3c2d8c":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-06-16 12:46:16.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/05/973105/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/05/973105/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/973105/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/973105/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/973105/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/05/973105/5"}}},"commit":{"parents":[{"commit":"7a2cb21e8b3035835ae1260c9c7e676191360562","subject":"Improve instance list method by using Nova tags feature.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7a2cb21e8b3035835ae1260c9c7e676191360562"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-01-12 14:30:03.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-16 12:46:16.000000000","tz":0},"subject":"Improve data volume response in Postgres","message":"Improve data volume response in Postgres\n\nPreviously, the API reported only total/used volume values calculated\nfrom f_bfree, ignoring blocks unavailable to unprivileged users.\nAdditional free space was effectively lost due to GB -\u003e GiB conversion.\nThis often created an illusion of remaining space while the real\nfilesystem was already at 100% usage, and databases were close to\nfailure or read-only mode.\n\nThis change adds a real used percentage based on f_bavail and exposes\nextra PostgreSQL volume details (WAL and WAL archive sizes).\n\nStory: 2011633\n\nChange-Id: I5d52370c38022a95dc2cc58c55a221104adb0a03\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/61131f20a73d3cef2f7e657d4a3582c86b3c2d8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/61131f20a73d3cef2f7e657d4a3582c86b3c2d8c"}]},"branch":"refs/heads/master","description":"Edit commit message","conflicts":{"ours":"69444dd22016f834f3b8068a8806a4eb29bcb4e6","theirs":"7a2cb21e8b3035835ae1260c9c7e676191360562","contains_conflicts":false}},"01547ec9a6aeb2bad50aa19ba10288baeb2ac478":{"kind":"REWORK","_number":6,"created":"2026-06-16 12:54:01.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/05/973105/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/05/973105/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/973105/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/973105/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/973105/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/05/973105/6"}}},"commit":{"parents":[{"commit":"d038f57505851d6681d437d12439942102a0f17c","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d038f57505851d6681d437d12439942102a0f17c"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-03-18 16:16:33.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-16 12:53:20.000000000","tz":300},"subject":"Improve data volume response in PostgreSQL","message":"Improve data volume response in PostgreSQL\n\nPreviously the API reported only total/used volume values calculated\nfrom f_bfree, ignoring blocks unavailable to unprivileged users.\nAdditional free space was effectively lost due to GB -\u003e GiB conversion.\nThis often created an illusion of remaining space while the real\nfilesystem was already at 100% usage and databases were close to\nfailure or read-only mode.\n\nThis change adds a real used percentage based on f_bavail and exposes\nextra PostgreSQL volume details (WAL and WAL archive sizes).\n(Changes in python-troveclient will follow)\n\nStory: 2011633\n\nChange-Id: I5d52370c38022a95dc2cc58c55a221104adb0a03\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/01547ec9a6aeb2bad50aa19ba10288baeb2ac478"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/01547ec9a6aeb2bad50aa19ba10288baeb2ac478"}]},"branch":"refs/heads/master"},"e66c2f417e50b32519aa9b9212c898f9a63bd7e8":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-06-16 12:57:27.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/05/973105/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/05/973105/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/973105/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/973105/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/973105/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/05/973105/7"}}},"commit":{"parents":[{"commit":"187af7eb8e32f58bcd4b2b7825dbd3e3272b2a86","subject":"Merge \"Use common utility to clone/install source\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/187af7eb8e32f58bcd4b2b7825dbd3e3272b2a86"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-03-18 16:16:33.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-16 12:57:14.000000000","tz":300},"subject":"Improve data volume response in PostgreSQL","message":"Improve data volume response in PostgreSQL\n\nPreviously the API reported only total/used volume values calculated\nfrom f_bfree, ignoring blocks unavailable to unprivileged users.\nAdditional free space was effectively lost due to GB -\u003e GiB conversion.\nThis often created an illusion of remaining space while the real\nfilesystem was already at 100% usage and databases were close to\nfailure or read-only mode.\n\nThis change adds a real used percentage based on f_bavail and exposes\nextra PostgreSQL volume details (WAL and WAL archive sizes).\n(Changes in python-troveclient will follow)\n\nStory: 2011633\n\nChange-Id: I5d52370c38022a95dc2cc58c55a221104adb0a03\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e66c2f417e50b32519aa9b9212c898f9a63bd7e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e66c2f417e50b32519aa9b9212c898f9a63bd7e8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
