)]}'
{"id":"openstack%2Ftrove~973510","triplet_id":"openstack%2Ftrove~master~I5eb51890cad2a7c93b816186792e0ad668806385","project":"openstack/trove","branch":"master","topic":"fix-wal-archiving-after-no-space-left-recovery","attention_set":{},"removed_from_attention_set":{"36080":{"account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"last_update":"2026-01-23 07:53:01.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2026-01-23 07:04:46.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"I5eb51890cad2a7c93b816186792e0ad668806385","subject":"Fix WAL archiving stall caused by partial archive files","status":"MERGED","created":"2026-01-15 15:51:53.000000000","updated":"2026-01-23 07:53:50.000000000","submitted":"2026-01-23 07:53:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973510-fix-wal-archiving-after-no-space-left-recovery","meta_rev_id":"db6805d11c3abf47b110860270db086e78b4635a","_number":973510,"virtual_id_number":973510,"owner":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-23 07:53:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-01-23 07:04:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":1,"date":"2026-01-22 12:14:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-01-23 07:04:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-15 16:24:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-01-15 17:44:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-19 10:13:07.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-01-19 10:13:07.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"893fb5c46038febd3cebebe43dd3b70cd65ff554","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-01-15 15:51:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ae6ae617aa63f8ad893e4abc4575b0dd92b42ec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 16:24:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/583844a850df4004809d4c5d298fc938\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2a8c1fdc2f244b539233cbfe6f509ae6 : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ba36387371fb47de846a7e9a649e9a48 : SUCCESS in 9m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b297c8101c9d4f69b7c962c195a42f8a : SUCCESS in 11m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2a40a527d394e89cdceb238d25102e5b056a0430","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 17:44:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c554e4cf3d0e44569f3452eb64769517\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7507d74eaca74b8bafa6f6be39a52824 : SUCCESS in 8m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/266db64677b24e77bbd49f3d784d3bd0 : SUCCESS in 5m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1b9cdd67141a4713b76e29f4099003d2 : SUCCESS in 3m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3bf5e99d21174d28a0a161db0a54411e : SUCCESS in 7m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57896a30606b4ac2a6ab227eaa23ca5a : SUCCESS in 5m 36s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/a3408e7460794c2d846afa8c7f337958 : SUCCESS in 47m 01s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/ee155821731540449d65f7cb24c01f0b : SUCCESS in 1h 49m 28s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/d39c9db4215d4374a0cf6d3066402a23 : SUCCESS in 1h 29m 45s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/5efe0ba564c94125a6523b61a11bae9a : SUCCESS in 1h 51m 28s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/66d8c21d7cbe4cbba3f2fc3de0b28d19 : SUCCESS in 1h 03m 00s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/19835209db1545e583a9522a6c247c7a : SUCCESS in 1h 39m 26s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/3c5a98d8f953460cb7a16107acb202e0 : SUCCESS in 1h 18m 52s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/80326ae89daf4d33a879bdd60cc8e1d1 : SUCCESS in 1h 36m 04s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/e197f622310f4407b5762e0eb232a670 : SUCCESS in 49m 54s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/cfd92fa83fbd40fabf06deb49261503c : SUCCESS in 1h 39m 18s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/62547bdbe910427586ddd375a3ae71bf : SUCCESS in 1h 17m 05s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/69888a3d15e847c7ad665cc02d925ff7 : SUCCESS in 1h 24m 10s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/63c50d76fd00455e906ab13bdc8a6da7 : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":1},{"id":"dfe5ad86d2f894a659c4738258c037bfe2ebf091","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-19 10:13:07.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2194b6db16cbee100f21e16e84f5e04690af2899","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-01-21 03:37:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cc823dd2c1c01f2fc21cd8ca7356f495e0f86965","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 03:38:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ad27b20f46617db8575e3a7ce2882f818a431062","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 05:02:24.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3473216c675d4bf7b8b1a171cf9846c9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b08d787e0e354c22b86dfe72301adba1 : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9885581f4fd24d04bb8c9ced5f3a3e5d : SUCCESS in 5m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0c06cd1b62db4d78aee23fb165a32b59 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e158dace847f4b848c6647e59c13c9da : SUCCESS in 5m 02s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/767329b031e346f88ad94966c4f78be8 : FAILURE in 43m 51s\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/16d0b96803654349aae72558a63b50c7 : FAILURE in 40m 28s\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/73c3178da1f643d982e8d3d1eaa28a4e : FAILURE in 43m 14s","accounts_in_message":[],"_revision_number":1},{"id":"c37e1f7035866935e0ad9c5bc2179b6023d02001","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"real_author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-01-21 11:57:22.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_36080\u003e\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}],"_revision_number":2},{"id":"b479cf15a6929299ef155cf5a8e9b96d5912ea21","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 12:57:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03a8773a9a2f4cf4babde4b5ce4a3eda\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5b2c970593bd44b6b759b662f058c833 : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d69d03ae04484ff392f7f915ea9999e6 : SUCCESS in 11m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ab2f8eaaaa8540e88d39833ce295db17 : RETRY_LIMIT in 5m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3b0d77979e6d05be25ddd06743bdf5534adf29b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 13:40:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1f610527ee046f39f92e7ba66dd82d2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cbdc3c5cf5ff4787959a0d41d4bafa2b : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed378460613040bdb958282860258bfd : SUCCESS in 5m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9a037a9f810944c0980518764cca429c : SUCCESS in 5m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a270a4447c2d444f8f0508d4419c7f60 : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/032e6387ff00447aa5d11375b3da90bb : SUCCESS in 6m 37s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/78431152285e4679a88660cd1f589451 : SUCCESS in 50m 36s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/60c604d18f29466ab5626557adfb383c : SUCCESS in 1h 38m 38s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/3fe9698803574c428d29fca2d3704ad5 : SUCCESS in 1h 25m 33s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/41078b2f18f14102ab2d85a5b5672dd9 : SUCCESS in 1h 41m 41s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/c16c6ff9704f40c3b9e4a099229f94b9 : SUCCESS in 44m 21s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/52bda24d54114539b6679011d28c395e : SUCCESS in 1h 32m 20s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/1d73275022e94274a42df5da397a3761 : SUCCESS in 1h 17m 54s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/5cc3457f33934caf8a93048e6e284f24 : SUCCESS in 1h 32m 27s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/59a226d9f2ef4fd0a12b2bbee423b71b : SUCCESS in 49m 06s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/53d60ca9dcc943a09dfc97a3a5b9ccf7 : SUCCESS in 1h 35m 09s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/3d93b4a7583e47b9a4b38f44dd874a58 : SUCCESS in 1h 13m 11s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/89d473aa8ac443db882e1e30be86af8f : SUCCESS in 1h 19m 17s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d94c7f38281242368be6b2e24f72faca : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":2},{"id":"05fee036d6f72a6edaadf5b6396ebf01ab5f2d07","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-22 12:14:12.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3bce191eeeba73aa3a5762fa94281e769b6fb685","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-01-23 07:04:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"389d50c969f0aeb6f4731cfcf93beab7772ca15f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 07:05:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"45f1a4abdd523b55a1778bd7b4d6cb2afba33ffe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 07:53:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1eb714ed002947818b9b9f9d0200b037\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b75ca74a563f4c69bcffaea8a2baf566 : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/75335a61780a448a99e087ae4e0d58f9 : SUCCESS in 5m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0fc2f9b6a2cd434fb313b2b534812063 : SUCCESS in 8m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/baca4a4a3ebe4969832b1c92fce6c387 : SUCCESS in 7m 00s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/2339de83dfcd423b835e09db3f52dc2b : SUCCESS in 44m 42s\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/690265cf88b748af84575401d0ba791b : SUCCESS in 43m 17s\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/29e3cef1944a4174a0cc0dc56b1e1536 : SUCCESS in 46m 34s","accounts_in_message":[],"_revision_number":2},{"id":"f13352cf1790808886dbc8579bc5c14163c73b9d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 07:53:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"db6805d11c3abf47b110860270db086e78b4635a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 07:53:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd0ca595c0084790a69ddb135102cd23\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f40e7f5f793d4808ad155d6479f7cf64 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e557e24d8b373bbfb11a06f866dd0a1e4ae9ca7b","revisions":{"6c3ea1e031a0e4655262fde9a88f2834bedd6c9c":{"kind":"REWORK","_number":1,"created":"2026-01-15 15:51:53.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/10/973510/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/10/973510/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/10/973510/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/10/973510/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/10/973510/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/10/973510/1"}}},"commit":{"parents":[{"commit":"9d6895646ed7a6f4a1da348e7d668a48fd151d47","subject":"Merge \"Fix: Prevent config removal on instance update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9d6895646ed7a6f4a1da348e7d668a48fd151d47"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-01-15 15:34:03.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-01-15 15:34:03.000000000","tz":300},"subject":"Fix WAL archiving stall caused by partial archive files","message":"Fix WAL archiving stall caused by partial archive files\n\nIf disk space is exhausted, PostgreSQL database can fail at least\nin two ways:\n\n- In most cases, space runs out during archive_command, leaving a\n  partially copied WAL segment in wal_archive. Subsequent archive\n  attempts after recovery fail the \"test ! -f\" check resulting in\n  pg_wal infinite grow.\n  Recovery requires manual SSH into the host by an administrator;\n  users cannot fix this themselves.\n- In some cases, the PostgreSQL container crashes before any WAL\n  segment is archived, leaving no corrupted archive files.\n\nThe modified archive_command ensures WAL segments are fully copied\nand that normal archiving resumes correctly after recovery from\nENOSPC. This approach meets PostgreSQL documentation requirements\nfor archive_command execution.\n\nChange-Id: I5eb51890cad2a7c93b816186792e0ad668806385\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6c3ea1e031a0e4655262fde9a88f2834bedd6c9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6c3ea1e031a0e4655262fde9a88f2834bedd6c9c"}]},"branch":"refs/heads/master"},"e557e24d8b373bbfb11a06f866dd0a1e4ae9ca7b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-01-21 11:57:22.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"real_uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/10/973510/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/10/973510/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/10/973510/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/10/973510/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/10/973510/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/10/973510/2"}}},"commit":{"parents":[{"commit":"83e21e695fcb9c27833481cfbbccde12dbfce1f2","subject":"Merge \"Inconsistent behavior after root disable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/83e21e695fcb9c27833481cfbbccde12dbfce1f2"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-01-15 15:34:03.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-01-21 11:57:22.000000000","tz":0},"subject":"Fix WAL archiving stall caused by partial archive files","message":"Fix WAL archiving stall caused by partial archive files\n\nIf disk space is exhausted, PostgreSQL database can fail at least\nin two ways:\n\n- In most cases, space runs out during archive_command, leaving a\n  partially copied WAL segment in wal_archive. Subsequent archive\n  attempts after recovery fail the \"test ! -f\" check resulting in\n  pg_wal infinite grow.\n  Recovery requires manual SSH into the host by an administrator;\n  users cannot fix this themselves.\n- In some cases, the PostgreSQL container crashes before any WAL\n  segment is archived, leaving no corrupted archive files.\n\nThe modified archive_command ensures WAL segments are fully copied\nand that normal archiving resumes correctly after recovery from\nENOSPC. This approach meets PostgreSQL documentation requirements\nfor archive_command execution.\n\nChange-Id: I5eb51890cad2a7c93b816186792e0ad668806385\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e557e24d8b373bbfb11a06f866dd0a1e4ae9ca7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e557e24d8b373bbfb11a06f866dd0a1e4ae9ca7b"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
