)]}'
{"id":"openstack%2Ftrove~974051","triplet_id":"openstack%2Ftrove~stable%2F2025.1~Ie527d34e62448c3e529c3c88fc4329fe807864b9","project":"openstack/trove","branch":"stable/2025.1","topic":"bug/disable-root-inconsistent-behavior-stable/2025.1","attention_set":{},"removed_from_attention_set":{"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2026-01-22 13:58:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie527d34e62448c3e529c3c88fc4329fe807864b9","subject":"Inconsistent behavior after root disable","status":"MERGED","created":"2026-01-21 12:54:27.000000000","updated":"2026-01-22 14:00:02.000000000","submitted":"2026-01-22 13:58:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974051-bug/disable-root-inconsistent-behavior-stable/2025.1","cherry_pick_of_change":970804,"cherry_pick_of_patch_set":5,"meta_rev_id":"5ad3fda8b2816ca47ffac25214cccc390a196add","_number":974051,"virtual_id_number":974051,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-22 13:58:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-01-22 01:46:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":1,"date":"2026-01-22 12:15:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-01-22 01:46:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-21 12:54:27.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"state":"CC"},{"updated":"2026-01-21 13:45:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-22 12:15:09.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"90e52ea0f70c384991aea995f46eee23e066d33a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-01-21 12:54:27.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1cc32831fa3f302fdafcdf805232e9df5ac47227","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 13:45:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ecb8a976b9e44884ade0fc8ddd01c3e6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1dccbc641f634309bc1b2e0408055bcb : SUCCESS in 6m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ed63290bd9e4cefaca80269c7ae41c0 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/338d60f89839400fb9bc99b29f73bd08 : SUCCESS in 5m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e6838dbaee3a445bb9fd7c7ec80a1986 : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0f2605a4b314d7f8076bb9d03fabc6b : SUCCESS in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38afe52b4bda46c0a255ee5abf08c71a : SUCCESS in 4m 15s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/b5bf7d13e2354ab481d8660c4d9a9187 : SUCCESS in 39m 59s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/b9c1c368535040fd82efea28a96cd730 : SUCCESS in 43m 36s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/9b84232c6f3e4481ae3d7ec4335f4ba5 : FAILURE in 40m 44s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/c398414c24b747e9a3a81e72cc935af6 : FAILURE in 41m 30s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/0cd8e2f77e3c4a229755f75807dae173 : SUCCESS in 48m 20s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/9688ea027e1045a89e16c76b21e66d0a : FAILURE in 35m 28s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/b2045797d6594e8f8b3662e3697df62d : FAILURE in 33m 11s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/5738a2feed464a79aba7fba36171602a : SUCCESS in 49m 39s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/e99fd4a7fa3546918f8005a9cf170d76 : FAILURE in 38m 15s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/d5ae75bced734bb2918d165e22036340 : FAILURE in 37m 55s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6ee27001c5264c4cb6d4c1227a43d431 : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":1},{"id":"18f62e15de8f2d05117a6e599ac574c8fc69500c","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-01-22 01:46:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f4787af4b8cdba7486696807cb99cd54569e51f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 01:46:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2706061203999723ed62a350abc693312c4d1657","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 02:43:55.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/48feae2cafb64ad689a56dadcf479fba\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25b41d6850474e2c81a69a7470b24316 : SUCCESS in 2m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0eac4ba6ef154ab891c357bb78cb888b : SUCCESS in 5m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d79bbb997d164ed6b2694eb12e0e89ad : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e22ad7a210634c65a569562fe6b6f6ca : SUCCESS in 3m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab4bc7c3253e42448a7d719906b4363f : SUCCESS in 6m 18s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/6cb18fac9cab4e00b4afcd8a33950b8e : SUCCESS in 25m 38s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/f7642457ad474293a45af6d81a289f7c : SUCCESS in 29m 04s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/876605a3588f46dd8539144f2e19cbf7 : FAILURE in 28m 10s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/6a81d881e6f74e3fa16c044118c378c2 : FAILURE in 37m 50s","accounts_in_message":[],"_revision_number":1},{"id":"fed9dfc044bb28877eb7d878c0f0240ca0620e1e","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-22 12:15:09.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"44f6bf0225925a64d9314b2693c8d17195659d5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 13:06:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9c894496d324e7da3ab80198ac61ce0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/712761ca1c45425d8a4208476882dafa : SUCCESS in 7m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6446f72eda04151a14c0e003506176e : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a2f3c8a21a143c39ef3a51bfe1bc3fc : SUCCESS in 3m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/48c6e916be8749e297de55c3747d3860 : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cb50139591a4028934f1062ab21deac : SUCCESS in 2m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7945d7ccb4e64eccbb60b69b2c4769c2 : SUCCESS in 6m 34s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/ee9c9866797445ff92162c0edce9130a : SUCCESS in 42m 59s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/efdcfd7618494e7f97644c7e9efee395 : SUCCESS in 39m 03s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/b511f477d91649039127212b27bc157c : FAILURE in 36m 49s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/e69245e8a10c4124a0135cf990ebc962 : FAILURE in 37m 12s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/3a3d16f108f14ff9a9b64efbc181e936 : SUCCESS in 46m 26s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/8bddf6d82cc541dc9f739d5c795763a1 : FAILURE in 37m 06s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/c0c9b135dde14bf3b10b1c572d757a0a : FAILURE in 37m 00s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/79dbffbbdb4b43f983b63446aa251130 : SUCCESS in 50m 02s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/344b9c1439b44e41be0550ae37f35922 : FAILURE in 41m 18s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/fd6993bf0f3b44d9a4428686f407d174 : FAILURE in 41m 20s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/363bb453d1994379a5ac6a034420847b : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":1},{"id":"af8dbc52c682b0447537487be275e5faa34b6dec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 13:06:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a632bb838f8ab44ff01cab60260d0219040e1320","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 13:58:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3188c7248e343d982a8682f15223cf5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a5e2d84889d474c9d1f29ea49b1817c : SUCCESS in 7m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b1cfd57e9a6d4530a06c2fe681215cab : SUCCESS in 9m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/16fca72f6f004528b73737fa57394998 : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b12ffaa5af2d4cd8a505ba88c7deba52 : SUCCESS in 5m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55754b3208b4400dbe7c6b2e9906e879 : SUCCESS in 8m 01s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/d964e5f00d8247558d311930e67f34dd : SUCCESS in 43m 29s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/682537ac40f1425aa19a5c9731c5f8bb : SUCCESS in 40m 30s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/bc82e2f3a15e422db0cf50e4f0f408f6 : SUCCESS in 40m 02s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/a975ab6ea1614326ba26c8879e6818c4 : SUCCESS in 50m 28s","accounts_in_message":[],"_revision_number":1},{"id":"fed73193c180a73c4061cd4ed3a565344817c17d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 13:58:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5ad3fda8b2816ca47ffac25214cccc390a196add","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 14:00:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/012bc294e9d94b36a42358a568767180\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/641d26c360a7423e937abb1072b6fef5 : SUCCESS in 1m 11s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/faf6cc47fb0f47e4a3180640ebfa8dae : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2e76bee374d7ef60607d3452036be366c053d32e","revisions":{"2e76bee374d7ef60607d3452036be366c053d32e":{"kind":"REWORK","_number":1,"created":"2026-01-21 12:54:27.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/51/974051/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/51/974051/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/51/974051/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/51/974051/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/51/974051/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/51/974051/1"}}},"commit":{"parents":[{"commit":"f2227f744ef305ec62323a42384affce66ef3d5b","subject":"Skip remaining Swift checks when not in use","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f2227f744ef305ec62323a42384affce66ef3d5b"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2025-12-12 09:21:51.000000000","tz":300},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2026-01-21 12:54:27.000000000","tz":0},"subject":"Inconsistent behavior after root disable","message":"Inconsistent behavior after root disable\n\n`root show` incorrectly displays \"is_root_enabled\": True even after\nroot was successfully disabled on the target instance. This patch\nensures the status reflects the actual state of the instance.\nLogic behind root_enabled_history data is preserved.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/973931\n\nChange-Id: Ie527d34e62448c3e529c3c88fc4329fe807864b9\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n(cherry picked from commit d563d002165b5d19d2d41798c5da8c2ec5705d7d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2e76bee374d7ef60607d3452036be366c053d32e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2e76bee374d7ef60607d3452036be366c053d32e"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
