)]}'
{"id":"openstack%2Ftrove~974397","triplet_id":"openstack%2Ftrove~stable%2F2025.2~I5eb51890cad2a7c93b816186792e0ad668806385","project":"openstack/trove","branch":"stable/2025.2","topic":"fix-wal-archiving-after-no-space-left-recovery-stable/2025.2","attention_set":{},"removed_from_attention_set":{"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2026-01-26 02:10:49.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-01-26 03:07:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5eb51890cad2a7c93b816186792e0ad668806385","subject":"Fix WAL archiving stall caused by partial archive files","status":"MERGED","created":"2026-01-23 07:59:15.000000000","updated":"2026-01-26 03:08:40.000000000","submitted":"2026-01-26 03:07:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974397-fix-wal-archiving-after-no-space-left-recovery-stable/2025.2","cherry_pick_of_change":973510,"cherry_pick_of_patch_set":2,"meta_rev_id":"e5d0570545795f785fc1b224737f40ca1985bab1","_number":974397,"virtual_id_number":974397,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-26 03:07:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-01-26 02:10:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-01-26 02:10:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-23 07:59:15.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"state":"CC"},{"updated":"2026-01-23 07:59:41.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2026-01-23 10:00:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dc0f8ae1fae2136b3187f8dedfad9614d7350e87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-23 07:59:15.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"eec149e328d959945931cbc2af005c44abc9c3e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 10:00:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3af1582ed8b1442b830f2d49a47340b1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e0c4a1745c34e8ba6f2c3368ad07f5d : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5922a72d99243bcb02e53cb183a9980 : SUCCESS in 5m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b4505f1727e34701924cd3dc684e440e : SUCCESS in 3m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/be310f89aa0a48faabee70fdd4b6fb9e : SUCCESS in 3m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/98c94566356243c3a0d9b020bc502aad : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfa7108cb90a45609895e7f228f21b4d : SUCCESS in 5m 40s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/9f5f08792d984200a6a7b77f8b93859a : SUCCESS in 47m 07s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/ed1088e83a7149a78888d76fe08b1393 : SUCCESS in 1h 43m 14s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/7bb5efa9f2384807b948109aa869d37f : SUCCESS in 1h 32m 26s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/c40aa3dbf3ef49298983878732e398b5 : SUCCESS in 1h 52m 34s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/c149e456d40f43a6bd49b6d9dd79804b : SUCCESS in 55m 23s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/63d0c9c1c3064800aecce7db4b1f84f5 : SUCCESS in 1h 50m 43s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/59a1a8a2d3c044c19105af88de0e0885 : SUCCESS in 1h 31m 41s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/40e1f23a2796447ab5079fde36a54231 : FAILURE in 1h 59m 29s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/a1d9a9777311475f9b5562db9b0f5e3d : SUCCESS in 55m 06s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/a95fea3d1c0248879aa597ba3ca5b396 : SUCCESS in 1h 49m 38s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/9b26cd3ea762456fb5fa9884a44949c3 : SUCCESS in 1h 32m 57s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/5ec86eaffdfb403a8c6628ca7bddf36f : SUCCESS in 1h 37m 42s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4e7141871ada494281fb5b5bb7f5ddab : SUCCESS in 6m 57s","accounts_in_message":[],"_revision_number":1},{"id":"f06b9ce6b12b96149170110fee98f18caf240269","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-01-26 02:10:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5b2b2265713732bdd8cdae3cd3906c5a5fcfee0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 02:16:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"45a6a9e2cea6528b957df5b9fba7d9fbd18bbcb6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 03:07:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c164ec9473c0443995798d7a9de1fa39\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/721075f8fd13478d9c362cc01f50a6d5 : SUCCESS in 5m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ae5c0e4b042b4b96ae7065323b5fa42e : SUCCESS in 6m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/29fd9759e1b24c259f77faec19f7a571 : SUCCESS in 8m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49b6bfee60ae434d8b17358ce8c4080a : SUCCESS in 6m 28s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/e9c039b52a1645789933ad24fff31f10 : SUCCESS in 50m 03s\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/25634445c49f4e9ab504553666c7d87b : SUCCESS in 44m 20s\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/4c1b45664a894a848f437a8b112ae069 : SUCCESS in 43m 15s","accounts_in_message":[],"_revision_number":1},{"id":"58a153794586101bc5029ae95c09468e4184a06e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 03:07:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e5d0570545795f785fc1b224737f40ca1985bab1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 03:08:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ba656c656944b56be25a0ff88351273\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90c3ae6a16694c2296338dde1fbe1023 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ff7c933c05cfd29317ce5a8d79e693f7d91c649d","revisions":{"ff7c933c05cfd29317ce5a8d79e693f7d91c649d":{"kind":"REWORK","_number":1,"created":"2026-01-23 07:59:15.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/97/974397/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/97/974397/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/974397/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/974397/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/974397/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/97/974397/1"}}},"commit":{"parents":[{"commit":"479b03f5041d59d46699d148613cb935d6a4e037","subject":"Inconsistent behavior after root disable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/479b03f5041d59d46699d148613cb935d6a4e037"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-01-15 15:34:03.000000000","tz":300},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-23 07:59:15.000000000","tz":0},"subject":"Fix WAL archiving stall caused by partial archive files","message":"Fix WAL archiving stall caused by partial archive files\n\nIf disk space is exhausted, PostgreSQL database can fail at least\nin two ways:\n\n- In most cases, space runs out during archive_command, leaving a\n  partially copied WAL segment in wal_archive. Subsequent archive\n  attempts after recovery fail the \"test ! -f\" check resulting in\n  pg_wal infinite grow.\n  Recovery requires manual SSH into the host by an administrator;\n  users cannot fix this themselves.\n- In some cases, the PostgreSQL container crashes before any WAL\n  segment is archived, leaving no corrupted archive files.\n\nThe modified archive_command ensures WAL segments are fully copied\nand that normal archiving resumes correctly after recovery from\nENOSPC. This approach meets PostgreSQL documentation requirements\nfor archive_command execution.\n\nChange-Id: I5eb51890cad2a7c93b816186792e0ad668806385\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n(cherry picked from commit e557e24d8b373bbfb11a06f866dd0a1e4ae9ca7b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ff7c933c05cfd29317ce5a8d79e693f7d91c649d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ff7c933c05cfd29317ce5a8d79e693f7d91c649d"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
