)]}'
{"id":"openstack%2Ftrove~974398","triplet_id":"openstack%2Ftrove~stable%2F2025.1~I5eb51890cad2a7c93b816186792e0ad668806385","project":"openstack/trove","branch":"stable/2025.1","topic":"fix-wal-archiving-after-no-space-left-recovery-stable/2025.1","attention_set":{},"removed_from_attention_set":{"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2026-01-26 02:10:55.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-01-26 03:07:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5eb51890cad2a7c93b816186792e0ad668806385","subject":"Fix WAL archiving stall caused by partial archive files","status":"MERGED","created":"2026-01-23 07:59:50.000000000","updated":"2026-01-26 03:08:49.000000000","submitted":"2026-01-26 03:07:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974398-fix-wal-archiving-after-no-space-left-recovery-stable/2025.1","cherry_pick_of_change":973510,"cherry_pick_of_patch_set":2,"meta_rev_id":"9f63a8aeefd08e2bdaa16e8c509c5b84175ccd6d","_number":974398,"virtual_id_number":974398,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-26 03:07:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-01-26 02:10:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-01-26 02:10:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"CC":[{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-23 07:59:50.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"state":"CC"},{"updated":"2026-01-23 08:00:13.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2026-01-23 08:55:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0e95409e6e13d25e0019d59403d7110e2b64d548","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-23 07:59:50.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"95ff8d3997af2afef0099b3110f92b90f1087275","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 08:55:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6269a302cd7e4772b84b4189dd701ff2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/88eb81d1d53f4d05801167783ac36ff6 : SUCCESS in 6m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d037528242f243a6951b94f734bcc9da : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/23f335f6516843a59b84a179b97a85c9 : SUCCESS in 7m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5c71e1e1331c403a8d3d8488874a80eb : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/597362233633490ea43a5123bc9396e8 : SUCCESS in 5m 50s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/4c7d9a7064d041069f7ef9f02899176c : SUCCESS in 54m 06s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/f9a26be624a546249c6cebc6ce197312 : SUCCESS in 49m 25s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/8bb4199b298c489283d1d80b74754e5d : FAILURE in 45m 26s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/d78b33f007914a5c9b92a7fda0bba201 : FAILURE in 45m 56s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/a44a9443f47a497498fba0a3ebdd07a4 : SUCCESS in 38m 37s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/6ac285ffd23d4894b18453a2476d8303 : FAILURE in 46m 36s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/1abc33ea507547a9829beaef0bd62b31 : FAILURE in 48m 39s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/12d90a4000eb4949a284b254b86d998f : SUCCESS in 52m 17s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/2bbdc0e7297a4772a701ca56f6ac40a3 : FAILURE in 34m 18s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/2e34ee5229594442a37d778a40a0ec13 : FAILURE in 46m 39s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e1a4dcf48d2d4618b408a233969851f9 : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":1},{"id":"7408348c193bf365d122d69341edce9dd1c7cb4d","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-01-26 02:10:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dd3e7e64becb76be601c2baafbdf4b3bdb7b6223","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 02:16:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3217d6f3bdb1c5f543df9b7dab5202c2ab15b4eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 03:07:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b216042d2c74616ad5bd21cc9f53cc3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2210434107ce4d5c8fba37df4dcf3621 : SUCCESS in 7m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4fedcd5dfaa0417296278f1c22d06433 : SUCCESS in 6m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/02f85e77d08a446182dc8ec6ffe7aeb9 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97cf4004e3d948c688712f1e37b2217d : SUCCESS in 6m 07s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/3de1adbe68e741998716eaa9c412c9cd : SUCCESS in 44m 16s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/a70d883f9d9e4fe7b04e02c7e7c64d74 : SUCCESS in 42m 06s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/6213e333362c454dbffc04b39593e23b : SUCCESS in 42m 22s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/fe7683488f7d44a68a06b12d9de805b0 : SUCCESS in 42m 56s","accounts_in_message":[],"_revision_number":1},{"id":"ebd2d155d4aa0c15d36056a20a3ae49dfde39b78","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 03:07:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9f63a8aeefd08e2bdaa16e8c509c5b84175ccd6d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 03:08:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f02db313f7dc4c57ad92c69b3b63077c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/711d849367dc4f0b827ef4e0d1d85584 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"425dd8c086b14e6538116ec9c2af18ca722dfe5e","revisions":{"425dd8c086b14e6538116ec9c2af18ca722dfe5e":{"kind":"REWORK","_number":1,"created":"2026-01-23 07:59:50.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/98/974398/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/98/974398/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/974398/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/974398/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/974398/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/98/974398/1"}}},"commit":{"parents":[{"commit":"2e76bee374d7ef60607d3452036be366c053d32e","subject":"Inconsistent behavior after root disable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2e76bee374d7ef60607d3452036be366c053d32e"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-01-15 15:34:03.000000000","tz":300},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-23 07:59:50.000000000","tz":0},"subject":"Fix WAL archiving stall caused by partial archive files","message":"Fix WAL archiving stall caused by partial archive files\n\nIf disk space is exhausted, PostgreSQL database can fail at least\nin two ways:\n\n- In most cases, space runs out during archive_command, leaving a\n  partially copied WAL segment in wal_archive. Subsequent archive\n  attempts after recovery fail the \"test ! -f\" check resulting in\n  pg_wal infinite grow.\n  Recovery requires manual SSH into the host by an administrator;\n  users cannot fix this themselves.\n- In some cases, the PostgreSQL container crashes before any WAL\n  segment is archived, leaving no corrupted archive files.\n\nThe modified archive_command ensures WAL segments are fully copied\nand that normal archiving resumes correctly after recovery from\nENOSPC. This approach meets PostgreSQL documentation requirements\nfor archive_command execution.\n\nChange-Id: I5eb51890cad2a7c93b816186792e0ad668806385\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n(cherry picked from commit e557e24d8b373bbfb11a06f866dd0a1e4ae9ca7b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/425dd8c086b14e6538116ec9c2af18ca722dfe5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/425dd8c086b14e6538116ec9c2af18ca722dfe5e"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
