)]}'
{"id":"openstack%2Ftrove~976724","triplet_id":"openstack%2Ftrove~stable%2F2025.1~Ice1ffd36a37f72fa668802b0bfb95e451d4d0dfc","project":"openstack/trove","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2026-03-03 02:26:43.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-03-03 02:43:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ice1ffd36a37f72fa668802b0bfb95e451d4d0dfc","subject":"Pin setuptools","status":"MERGED","created":"2026-02-13 09:32:41.000000000","updated":"2026-03-03 02:45:23.000000000","submitted":"2026-03-03 02:43:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"976724","meta_rev_id":"02e486041401662b18bb4f96ba3e622e423a1920","_number":976724,"virtual_id_number":976724,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-03 02:43:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-03-03 02:26:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-03-03 02:26:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-13 09:54:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-13 10:11:10.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"0eccd4f042b9522291593703ca02e8529904cebd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-13 09:32:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e0a85066caf471a07ce3cbd8c1876ea8ec5cadb2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-13 09:37:21.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"eb168130c9c44aa6390bdd4bbfb87d93d4cf3898","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 09:54:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b53ac58f26724f21ad601c5f6f0d06c2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5430e3c3259c414ea354be0ce6302231 : FAILURE in 2m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fab07e1bbf1c47da9d3c5bc65961ac46 : FAILURE in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd16be0e18b149fdb7c4d11add2d1244 : FAILURE in 6m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a9274ec27f4143819dd20c88bc377bbe : FAILURE in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0ab147fdc9543b39261c8dc947a7df7 : FAILURE in 4m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4c2d85dbf444b31b8813832b6e30ef2 : SUCCESS in 6m 48s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6e15df73429149f7a84794a240cf69f5 : FAILURE in 5m 12s","accounts_in_message":[],"_revision_number":2},{"id":"ff07aa5eaa48becdb5e804b2fe55dea107ab6566","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-13 09:57:14.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"33164e93d0648149fbbc208b76d0c6f202842494","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 10:10:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a0fcff1e8044beabd87a087d7f589b1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc5ef32647874d649321b9aad4ead6ce : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/658d3a775adc4cd4931e2b5291b76661 : SUCCESS in 2m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2bb4a7ec8c314b0994a5c6aa4a74a307 : SUCCESS in 6m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b8d7dc7fab044415a6569f3d2f629a0b : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a7474c7a3f243419a12d1688780940c : SUCCESS in 3m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2eea650c19a34344880e486366de8c61 : SUCCESS in 6m 23s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/51198922c3ad4b2286792fbb040812c6 : SUCCESS in 5m 56s","accounts_in_message":[],"_revision_number":3},{"id":"92c40f462933fec729305f5135e879e0d95847fe","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-13 10:11:05.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"20308abf85e015111ef5cdd399bfa21da5d18fe1","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-13 10:11:10.000000000","message":"Patch Set 4:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"7d2e53b995f097ea0b3c415b46d6d792a6527452","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 10:21:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47be045122f14745918f3c8e6b9edfe5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff3b879877b34ba78028b7c9e0c02803 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35a1cac535cb46a58a72a6046c4c386a : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/04368edf519145eb89142e12b51c5b59 : SUCCESS in 8m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8902b7f2c04140caa9c82e42a0f6fa29 : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/991e4011123540f988d239884bc7059b : SUCCESS in 5m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d2a59b21901424082cefdb019a994e0 : SUCCESS in 6m 04s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0552d4afbe8a44c19cd6c4985f33c490 : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":4},{"id":"4c9fb2aec31abf0ec0b7c41a76f04c33710b1568","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-28 12:34:03.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"009cc06748a8613df918dc61cb685d4c64a96721","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-02 11:38:54.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"906c770c7353b932dd0a6f730a46e9375a886721","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-03-03 02:26:43.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"ef7809c62c6d9090f18ff12ed54309878517c571","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 02:26:55.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"cab77d22fed28dd021012b8444322d380aef1324","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 02:43:36.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e501a376f49f415d9d4ad4a15fd2c354\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/992abb7966c242c4b36d3ebb996b397d : SUCCESS in 6m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb3bec5ec3914766a6eefd5cf1e52fa2 : SUCCESS in 4m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/22f5453cee0940aa97153fa3427f5cc7 : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b81b39c5a9f7422cb2148592da2fc1fa : SUCCESS in 4m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c3dee45347848b4a1f099c8bea736c7 : SUCCESS in 7m 52s","accounts_in_message":[],"_revision_number":4},{"id":"d24966007b66232197a47e95ddadeb0a446dbb9b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 02:43:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"02e486041401662b18bb4f96ba3e622e423a1920","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 02:45:23.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb805deb34374a068e8a06d893b88e18\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01784453f5a04d7baece0884661fd591 : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/30127b49163e434e9843db6f19981644 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"6128da27d2994d23b83e4424c5e9def838acfbc4","revisions":{"5a3a1be3b80aff9e36fb1080c142c0f3dac336eb":{"kind":"REWORK","_number":1,"created":"2026-02-13 09:32:41.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/976724/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/24/976724/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/976724/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/976724/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/976724/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/24/976724/1"}}},"commit":{"parents":[{"commit":"0b346d62f5f3fb0fd0b270326a112c52ef5d3110","subject":"Fix Keystone SSL verification handling in Trove service session","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0b346d62f5f3fb0fd0b270326a112c52ef5d3110"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-13 09:32:41.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-13 09:32:41.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Ice1ffd36a37f72fa668802b0bfb95e451d4d0dfc\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5a3a1be3b80aff9e36fb1080c142c0f3dac336eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5a3a1be3b80aff9e36fb1080c142c0f3dac336eb"}]},"branch":"refs/heads/stable/2025.1"},"b56423425b278184c501a125ed334bd6fb1c8a0b":{"kind":"REWORK","_number":2,"created":"2026-02-13 09:37:21.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/976724/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/24/976724/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/976724/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/976724/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/976724/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/24/976724/2"}}},"commit":{"parents":[{"commit":"0b346d62f5f3fb0fd0b270326a112c52ef5d3110","subject":"Fix Keystone SSL verification handling in Trove service session","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0b346d62f5f3fb0fd0b270326a112c52ef5d3110"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-13 09:32:41.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-13 09:37:20.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Ice1ffd36a37f72fa668802b0bfb95e451d4d0dfc\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b56423425b278184c501a125ed334bd6fb1c8a0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b56423425b278184c501a125ed334bd6fb1c8a0b"}]},"branch":"refs/heads/stable/2025.1"},"d81d597ed81e3f9a637936efd3064ffe2295ce86":{"kind":"REWORK","_number":3,"created":"2026-02-13 09:57:14.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/976724/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/24/976724/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/976724/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/976724/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/976724/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/24/976724/3"}}},"commit":{"parents":[{"commit":"0b346d62f5f3fb0fd0b270326a112c52ef5d3110","subject":"Fix Keystone SSL verification handling in Trove service session","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0b346d62f5f3fb0fd0b270326a112c52ef5d3110"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-13 09:32:41.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-13 09:57:14.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Ice1ffd36a37f72fa668802b0bfb95e451d4d0dfc\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d81d597ed81e3f9a637936efd3064ffe2295ce86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d81d597ed81e3f9a637936efd3064ffe2295ce86"}]},"branch":"refs/heads/stable/2025.1"},"6128da27d2994d23b83e4424c5e9def838acfbc4":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-02-13 10:11:05.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/976724/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/24/976724/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/976724/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/976724/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/976724/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/24/976724/4"}}},"commit":{"parents":[{"commit":"0b346d62f5f3fb0fd0b270326a112c52ef5d3110","subject":"Fix Keystone SSL verification handling in Trove service session","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0b346d62f5f3fb0fd0b270326a112c52ef5d3110"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-13 09:32:41.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-13 10:11:05.000000000","tz":0},"subject":"Pin setuptools","message":"Pin setuptools\n\nsetuptools 82.0.0 dropped support for the legacy \u0027pkg_resources\u0027\nlibrary. While most users of this (including the all important pbr\nlibrary) have been updated to remove references to \u0027pkg_resources\u0027 on\ncurrent master, this is not true for historical releases. Pin the\nsetuptools version used on these historic branches and disable build\nisolation to work around this.\n\nChange-Id: Ice1ffd36a37f72fa668802b0bfb95e451d4d0dfc\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6128da27d2994d23b83e4424c5e9def838acfbc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6128da27d2994d23b83e4424c5e9def838acfbc4"}]},"branch":"refs/heads/stable/2025.1","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
