)]}'
{"id":"openstack%2Ftrove~978312","triplet_id":"openstack%2Ftrove~stable%2F2025.2~I65edce7f6304860d0e38b26d7abb8b9d40653026","project":"openstack/trove","branch":"stable/2025.2","topic":"feature/allow-configuration-create-for-disabled-datastores-stable/2025.2","attention_set":{},"removed_from_attention_set":{"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2026-03-30 02:12:01.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-03-30 03:14:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I65edce7f6304860d0e38b26d7abb8b9d40653026","subject":"Feature: allow configuration create for disabled datastores","status":"MERGED","created":"2026-02-28 12:34:54.000000000","updated":"2026-03-30 03:15:56.000000000","submitted":"2026-03-30 03:14:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"978312-feature/allow-configuration-create-for-disabled-datastores-stable/2025.2","cherry_pick_of_change":972834,"cherry_pick_of_patch_set":3,"meta_rev_id":"0de2d2c9738122497396fb9463b99990ceafcbee","_number":978312,"virtual_id_number":978312,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-30 03:14:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-03-30 02:12:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-03-30 02:12:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-28 12:34:54.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"state":"CC"},{"updated":"2026-02-28 12:35:20.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2026-02-28 14:29:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1bacc0821942a27a04c7423adaee33226d5fe525","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-28 12:34:54.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d5a83165a97ef38bb1421966815741b31d846003","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 14:29:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d425e39920184e1e97452a6db0a76bd3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0fbfee5b22d434696c532a4270e24d1 : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffca41c16f034ec7af07b379c3b13725 : SUCCESS in 5m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4e6053088f074a50a8f273e6f4429063 : SUCCESS in 5m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9ad0cb8e744243878bb907eb808a4762 : SUCCESS in 5m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dee3b86a253040549f1b8c67a3b78978 : SUCCESS in 8m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edef02425c52427a8d0279ad59eb41c7 : SUCCESS in 5m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1075bc0d49ed476e89d71ec3368eb843 : SUCCESS in 6m 07s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/50ae564f6b6f4dbfa7773102ce283646 : SUCCESS in 46m 28s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/5a30fd7dc9274f369221c7d4a31c7c02 : SUCCESS in 1h 39m 26s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/d6ff7f1a6cd84432ab81a7921f2b7dce : SUCCESS in 1h 23m 01s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/8f37f75c5b6e4281ad25836cc11ece98 : SUCCESS in 1h 52m 04s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/958bf74f464f4dd2a71604d69996984c : SUCCESS in 43m 58s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/22dbdf34da2243548bed3c6cd5a13bb8 : SUCCESS in 1h 30m 33s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/71b4e17f058a46d4a3dda73bbfa6d3a5 : SUCCESS in 1h 19m 44s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/e908018f7c094d13b009564def80e678 : SUCCESS in 1h 36m 43s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/eae13bb21ca64bb7b453edf0e0c24869 : SUCCESS in 48m 50s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/9cc91c55709f4f3ca1fa6067c82cf404 : SUCCESS in 1h 33m 29s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/89a83b7bd42b44f5b18194cd88fb8e9b : SUCCESS in 1h 12m 07s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/2637f068e7e3449ba699699848ff7b49 : SUCCESS in 1h 19m 35s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1aeb897d74424467a22280b7989802e3 : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":1},{"id":"bd7e151c858244bdbf0e6884ef1f20cf1a2d0894","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-15 21:47:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"92eee0b909aad02d791e295af87ef9c454604dd3","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-03-30 02:12:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c4674cfae7b5a2d806abd3d87d8ef6920150980c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 02:16:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"77374801467b166a29a93678540050f580e223d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 03:14:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8d435f107394cefad57011dde4cf0a3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85192a2610694dc29efd257f46abb916 : SUCCESS in 5m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7e19fb9b3a5841739834ccb6e5533489 : SUCCESS in 6m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dfd7da22f4fe420f83ecfc9d810d9b07 : SUCCESS in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87f22c982256456497cecf3b234b8e0a : SUCCESS in 5m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2acc08e1a3a94ced9bdf24a3e31bf476 : SUCCESS in 4m 11s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/8d8b53baa2ae438ba7c456738d345e72 : SUCCESS in 47m 06s\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/d010e2c45093433b9767c3d8281e2f91 : SUCCESS in 57m 50s\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/2bddc6690822460181e318600389651f : SUCCESS in 45m 18s","accounts_in_message":[],"_revision_number":1},{"id":"e7cde7747c9cad9ff27d86bca59d78ced3ff5497","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 03:14:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0de2d2c9738122497396fb9463b99990ceafcbee","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 03:15:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8ce73d9b48b43e4bd96be5034476502\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cbef051afbf415196f97b73d837bf56 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d52a06afe7f4fa98210e647db40808a : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dbd8396b04d7bb3a7478e7c08144ebb25ad325bf","revisions":{"dbd8396b04d7bb3a7478e7c08144ebb25ad325bf":{"kind":"REWORK","_number":1,"created":"2026-02-28 12:34:54.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/12/978312/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/12/978312/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/12/978312/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/12/978312/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/12/978312/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/12/978312/1"}}},"commit":{"parents":[{"commit":"e1f43bdfff8dc9de1bdc510ade7602dd8ebcc8c9","subject":"Postgres: Fix user password change and access","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e1f43bdfff8dc9de1bdc510ade7602dd8ebcc8c9"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-01-09 10:17:01.000000000","tz":300},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-28 12:34:54.000000000","tz":0},"subject":"Feature: allow configuration create for disabled datastores","message":"Feature: allow configuration create for disabled datastores\n\nFor our production Openstack DBaaS the lifecycle for a datastore version is:\n- create datastore, e.g. postgresql 16.10; datastore is enabled, users may create instances for it\n- datastore is disabled due to the next minor version is available, e.g. postgresql 16.11 was released\n- legacy stage: old instances on disabled postgresql 16.10 still works; version update for old datastore instances is advised but not a mandatory\n- cleanup stage: all user instances was upgraded to upper versions, datastore 16.10 may be gracefully removed\n\nSo the problem here is that users are not able to create configuration group for disabled datastore versions and fully control their running database instances. This leads to a bad user experience forcing them to upgrade their instance to the latest version, for which they may not be ready yet.\nIn our opinion, there should be a possibility to modify configuration for the running instances even on disabled datastores.\nInstance creation for a disabled datastore should not be available though.\n\nChange-Id: I65edce7f6304860d0e38b26d7abb8b9d40653026\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n(cherry picked from commit 319518285a858a1be5fcd78f56bfb5331857dc57)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dbd8396b04d7bb3a7478e7c08144ebb25ad325bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dbd8396b04d7bb3a7478e7c08144ebb25ad325bf"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
