)]}'
{"id":"openstack%2Ftrove~978315","triplet_id":"openstack%2Ftrove~stable%2F2025.1~I96fa125fa2044e40e3fbeb685cc826b62b21bd4a","project":"openstack/trove","branch":"stable/2025.1","topic":"fix-incorrect-wal-archive-cleanup-logic-stable/2025.1","attention_set":{},"removed_from_attention_set":{"36080":{"account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"last_update":"2026-03-15 08:54:12.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2026-03-13 02:38:55.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-03-15 08:54:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I96fa125fa2044e40e3fbeb685cc826b62b21bd4a","subject":"Fix hardcoded WAL archive cleanup threshold in PostgreSQL","status":"MERGED","created":"2026-02-28 12:36:38.000000000","updated":"2026-03-15 08:55:01.000000000","submitted":"2026-03-15 08:54:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"978315-fix-incorrect-wal-archive-cleanup-logic-stable/2025.1","cherry_pick_of_change":973430,"cherry_pick_of_patch_set":11,"meta_rev_id":"498715b482ddf12c5cd589532c8c83452fa902c6","_number":978315,"virtual_id_number":978315,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-15 08:54:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-03-13 02:38:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-03-13 02:38:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"CC":[{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-28 12:36:38.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"state":"CC"},{"updated":"2026-02-28 12:38:13.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2026-02-28 13:25:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-14 20:12:10.000000000","updated_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"CC"}],"messages":[{"id":"f40ecb1d7657b67fba62e5c6d6c352a69eb1669b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-28 12:36:38.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"47bfd433e8086e92926facc5759829317a2c64ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 13:25:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/23b6f18ea3e14cfcbfe47fcdbeb42e7a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be37158a6bf645cda10971e6ee05b441 : SUCCESS in 7m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5525437dc9204861853a95c3195b0a17 : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4919f0eb48a248f798c96d11a1cda955 : SUCCESS in 5m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/160abcf44ccd415ab3199ac95485c009 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32c590d14f5042e38e7d26298786ff55 : FAILURE in 4m 55s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/91fbc681d51d4108a18c59b6666d729e : SUCCESS in 44m 54s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/4a478badb0ab42408b5daace3132f88a : SUCCESS in 44m 56s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/e5422ceb9ab94546b15b8eaa39c00033 : FAILURE in 38m 22s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/896f40f654e1417f932dcdc7cedca02d : FAILURE in 36m 26s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/817de003dcb84b8d851c5d131cd20f85 : SUCCESS in 40m 41s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/7347a7236dcc4dcebb83f5505b600363 : FAILURE in 36m 25s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/a3ecbb899d164c8ba6a5597e7be07cd8 : FAILURE in 40m 13s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/d0da1b64e8b64e7dbeda65644457f83a : FAILURE in 47m 28s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/4e3a38f70e2e4c29ad4078e59a9b6d22 : FAILURE in 36m 00s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/934e4f3fe0514c28a34b4b656fc6f3ec : FAILURE in 38m 56s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/67e6a52bde554b32948448ca452fcf1b : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":1},{"id":"1bccacef387a0c85cee3c8db9be1fde44535fd8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-28 13:28:48.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"0d8e9d58713c139996d1fab889de54fe6bdbfb1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 18:11:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba58081d5f2d47e9909bd02d7a481c58\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ce5dea8f4ca4a0eb9379380c4f583bc : FAILURE in 8m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97a946a001be4592bafbc26173552330 : SUCCESS in 7m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/68edfc151a42422bab350ef8d2a2eeb5 : FAILURE in 6m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/22472e90f474487289233c400fa32fae : FAILURE in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cbb86b4c3e646cc8a381e0c140ddd7d : SUCCESS in 6m 40s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/0cb21e9ccd3842b6876f065acb920e9f : SUCCESS in 1h 02m 01s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/32e1a8a04d9a40ebac2477017183d9f5 : SUCCESS in 43m 23s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/ddff7d54a120449e97850eedcaae8491 : FAILURE in 36m 40s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/fd967503a0314656938321094880e0af : FAILURE in 41m 24s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/58e62200c3e14144b3dfea33a768550e : SUCCESS in 49m 17s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/5cffd18d1bfd4a44b22bc6b8f9b77ab9 : FAILURE in 42m 44s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/cd6314f8537e445891bbad01ed2629c9 : FAILURE in 42m 36s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/55214edc157c4f578a2a8dba7746311e : FAILURE in 51m 13s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/0d48d97aaea249be96bfd259bb99bcaf : FAILURE in 36m 13s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/4f9a02bcdf5e49da8d97fe796bcc2bdb : FAILURE in 42m 32s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2eb6992b07204d2c959ae03ac97a4bf0 : SUCCESS in 5m 58s","accounts_in_message":[],"_revision_number":2},{"id":"2a011902da8588656ba24695c9f83714f2d3c19a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-02 07:51:34.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"6b740635099dddb669000ea1d8d9fa0b5b4872b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 08:57:37.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4ca470137860484c87e2984609436528\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c532687e7be4535bb9f0bc3a3a3db86 : SUCCESS in 7m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33552588149542fd82f6cf4b07eaf57d : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da4e9b9edc404a469c0b972b0836fd7c : SUCCESS in 5m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d39464e1f2fd4383b2df4870def1265e : SUCCESS in 3m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db441da4e5544193ad8ad14db9f59379 : SUCCESS in 5m 33s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/9fb1ba3b92554e45a507d8fab0d5cbe1 : SUCCESS in 47m 41s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/dc19a0ed019f403d93ad8f98d99e42ff : SUCCESS in 41m 58s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/dd755edd34944cb09dad8f743c317e28 : FAILURE in 40m 06s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/dbe545d39c86441fb8c50142370e4430 : FAILURE in 42m 02s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/3102220ac1744b7aad0fa520db9d4fdd : SUCCESS in 57m 43s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/062f7a5665614f079b80b4cda20ec3fe : FAILURE in 51m 58s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/b8f11b122cd14d8095dbd0238161b5b5 : FAILURE in 1h 05m 09s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/e563462143e54cd1b331d97786c3ab27 : FAILURE in 46m 48s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/5fb9467e89b14c93a8e498016cef8fa4 : FAILURE in 37m 24s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/a7c39c9b21124f1a92441318768a5ca5 : FAILURE in 41m 09s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a1c34c5e073b40e297a8a94f1691b073 : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":3},{"id":"85dcc5bd4cebccb13da8da3f399f88c4e8554ab2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-03-02 11:32:41.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"07419c27b0658427be7e3f063275ef7fa22320d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-03-02 11:34:40.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"fccd07ffde4597f7ee5c4718fcdb58de31ad6579","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 13:27:19.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/44917137f9034c6abe75158c372ac23e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff8f7e259966403ebdd4665038d3053a : SUCCESS in 7m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5a3855cd35a4595b0bc132c3fdb9ab4 : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f9ff2d4c0bc41c1a9ee8abbcb6b9773 : SUCCESS in 5m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/859a5df03e784a938b3abd012e5576ef : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4823310474de486295025773f16a9605 : SUCCESS in 5m 50s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/d57e886001ec48c6a6707ba907b54c57 : SUCCESS in 44m 16s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/c68a0436c5f94cb59ed2d554a82a37e7 : SUCCESS in 37m 52s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/35ee15ad04b546719fe86625332edb52 : SUCCESS in 1h 34m 29s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/5592a605c742491f9ae177b9248e5c89 : SUCCESS in 1h 19m 17s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/3cdae4edc998426ea7fe6d96ecc59599 : SUCCESS in 37m 35s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/8a2628f34323404b8bc019747a7a61ab : FAILURE in 1h 17m 23s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/0687d0f51121438b9f1d92c59b4d8a43 : FAILURE in 56m 29s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/2d69696ddfae4264aa2412dab9d70248 : FAILURE in 48m 02s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/92607467053d4dd3bdcc5f9e93755b65 : FAILURE in 54m 12s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/b9b4298ccdc24db2acde5bd7a188c25f : SUCCESS in 1h 10m 30s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/02519e254de041d9bd9f5a1f13fc3a90 : SUCCESS in 7m 07s","accounts_in_message":[],"_revision_number":5},{"id":"56ba217d2d8444f94b51a9f9c023146f9d75b5fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-03-02 17:40:37.000000000","message":"Patch Set 6: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"a47ce4510ee0ca90a6e636c1f4146a2c08547c0a","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-03-02 17:41:39.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9c4099d267adbecf00a71e27f36ae3f251665231","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 20:16:26.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/728c0e5c8bf8478ba4378b72787f4f19\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6bf41657f504662b9311f93c590f848 : SUCCESS in 8m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef7dfedea3264bb988aeb40716416169 : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fe6986aaff5c4d7a9bb9a1b5860a7c79 : SUCCESS in 6m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/40694423f594460694942ccdc1dced95 : SUCCESS in 7m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90246cbbf93740c2ae67d0b1cd51d5bd : SUCCESS in 3m 54s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/5d1d2341a8344b2b9f06076a869074ad : SUCCESS in 1h 01m 42s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/5cedfe21360040b5977df0af0582e1b5 : SUCCESS in 45m 57s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/deaefc41a06a475f9002f7e74b59dba0 : SUCCESS in 2h 00m 55s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/b1799bbafafd44fa91c9bffe7871be12 : SUCCESS in 2h 26m 05s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/c0366d6e13ac4b709be4aa5674b70c20 : SUCCESS in 26m 40s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/c1ee9970c4f6480c9c7c4ed04d21c5f8 : FAILURE in 1h 17m 00s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/6faf91e70fcc4b62af27bf6baecde240 : FAILURE in 49m 54s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/dc126c6b93784edf9b4ea704ccc79540 : SUCCESS in 1h 16m 04s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/0fbba7fd79184db4b3ecfab272a1c8d8 : SUCCESS in 1h 44m 07s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/892a71b3a172493a99493878125a1a0e : SUCCESS in 1h 42m 32s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f1c4d0235ba0475da2458cc115576d7b : SUCCESS in 5m 47s","accounts_in_message":[],"_revision_number":6},{"id":"66e95d4593e7602ff5d625b2a9311c98d6410d44","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-03-13 02:38:55.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"9214fd2de540196a23d6de10cb522e5d5acf28a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 02:39:40.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"1d22fa31cbd6a9e5c3b9f4cb9eae74bc1f96263d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 04:24:15.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9b5e65f27c8c4424bc51d308bcee8d57\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43c4d9d9c92b447aacd76b277145ad0c : SUCCESS in 6m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/996f9f4f185544cc9acacc5ac2579c4e : SUCCESS in 6m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bd11033b49bf4257b4b8a50799d86bd9 : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09931fc11b30404fbf72901164836519 : SUCCESS in 5m 29s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/d362956cc9254022bdca83e4456b884f : FAILURE in 31m 05s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/078fb36db4e1465ca1a563507457df58 : FAILURE in 27m 33s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/ddeceb447f454e46bd7081c73dbf7726 : FAILURE in 30m 36s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/baf6f735f6fc4cfa866b998a3f694074 : FAILURE in 31m 05s","accounts_in_message":[],"_revision_number":6},{"id":"6bc41dc75c209c3cfe4ba6643af65d733d1b62e4","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-03-14 20:12:10.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"52241da45378ddd5cc4e97c5213d20088c0125e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-14 22:07:03.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89594abbb74d4e94b8378b0dddeb0bbb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/184a5c32336744f3b7071c0e351bd0aa : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6829964efe5d408c9e00069b050110f8 : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e566f91356e43fe95099d8ea0378d77 : SUCCESS in 5m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/79d446c4efa94e7ca435e70123700a82 : SUCCESS in 6m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55933cab3f8c4a1795268d613ee0f175 : SUCCESS in 5m 18s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/3750cddeaabf4ad997e9d227905b637c : SUCCESS in 45m 07s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/a44ef97c9a8744959fe3502f8a219c9d : SUCCESS in 41m 32s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/9b2df59954664c85aa895a7e62c11d68 : SUCCESS in 1h 47m 50s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/a828814aa44e4830b86560da0cac83a5 : SUCCESS in 1h 19m 51s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/2828f9539e9a40ff8bf2c32fd2fb6079 : SUCCESS in 41m 50s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/8c3b58166656457cb2ed24a8e47e1597 : FAILURE in 1h 01m 24s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/ee15969a9c5e48f185456a5ae7ee1887 : FAILURE in 54m 32s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/f856368e48d741148706954c16db3963 : SUCCESS in 46m 55s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/622a4cd27a844158b234e3c368d43ddb : SUCCESS in 1h 42m 22s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/b41bd59e6b7b4408b222d8b7faeadb68 : SUCCESS in 1h 15m 47s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0b375ec5660a4121a931e127b5df56c3 : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":6},{"id":"c56a4980f89e06cc547eefd10e55442d564f00ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-14 22:07:23.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"14a65194393350ae31d99ecea5d02b6c11d8b84e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-14 22:55:54.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/40a6f09696224b918a7599e73d34964f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5b5dff94b354683bccd9cdc4b6dd957 : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/82694f2e5bff4d89b8b9a5e64761d4fc : SUCCESS in 3m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a7a3e8a12f9143dbb95c4e02602c8d30 : SUCCESS in 7m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea0202aa3417465b9207c78a0c4d24b0 : SUCCESS in 4m 21s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/487e743a7c514a6694f54b6a427f25e5 : SUCCESS in 44m 28s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/67a66405466a4b198f73f122f065fad8 : SUCCESS in 41m 54s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/e90007103c494c1aaf710680c983a32c : FAILURE in 27m 00s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/8d56269e06714bc38eb494303070c9d6 : SUCCESS in 30m 33s","accounts_in_message":[],"_revision_number":6},{"id":"f571e2543833162aa4afcb9c36058c8f0b14ea11","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-03-15 06:09:10.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"697215ad5cc17b91be6eaeeaed6b82483d980877","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-15 07:50:58.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68e3b3a405ec458baa1b4239fda33910\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f318a4e16414d3e850fe0c43962ce2c : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e09a0efc64154c7d81d69bb9cfa26bec : SUCCESS in 2m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/303d34485d6e4ab5ab7260bfa1c835ff : SUCCESS in 3m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/84f67aa17ee742b1bea70918c54525f1 : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8d8476b79c84dc4a5d123592095b725 : SUCCESS in 5m 47s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/4953aabb2b83409ba227482f49c326fd : SUCCESS in 48m 47s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/a78a53b440544375a0773d4a1dc8cefc : SUCCESS in 45m 03s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/ebb9c22f6a0a4d3886d886eaa1065cf8 : SUCCESS in 1h 40m 53s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/4fbc95854aef47a28034ac4856aa9521 : SUCCESS in 1h 26m 40s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/1ba3c42912f0458ba905fa634b24058e : SUCCESS in 47m 06s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/370e6aca65444077958d9cbba4228388 : FAILURE in 1h 05m 22s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/eceb1c3b8ae84f30b6703dbaa628a484 : FAILURE in 1h 06m 39s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/2ed0fb4a26cb468a89277d5ee222dce8 : SUCCESS in 51m 24s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/6e0ca8bea7784af98e01e2861b1f255e : SUCCESS in 1h 35m 15s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/39a6970528074a2095c1c97ec3509cd0 : FAILURE in 1h 05m 28s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c7af29844ce74162af82ecda4672489d : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":6},{"id":"ad4cfc6cb4344976be5fc5f05ad85fc6a96d77d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-15 07:51:17.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"07ab5d240aec6afc02ae707cc530269cdb8f3041","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-15 08:54:12.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/beff4a21bb4b4004b291d1c01dd85e57\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b265f3f9ff34258933c52ad9e1073b6 : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4fa0ea24ca8b42a68f6a7120582c2366 : SUCCESS in 5m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b1606da51f504622bb95716460025e60 : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e345590d447f46c09eb6cde398c90619 : SUCCESS in 4m 53s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/d7a4ba3fd86d46c49a47f2a37eeac288 : SUCCESS in 42m 17s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/2816094d57134f21ae97a7e41f23281e : SUCCESS in 39m 25s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/ea009782ebf8480fbf4e107ed49b3b1a : SUCCESS in 1h 01m 35s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/2331c7baeae947c88eb8ba3135e61a15 : SUCCESS in 47m 56s","accounts_in_message":[],"_revision_number":6},{"id":"4cdab67e0f96ad99e8f9a889c49c98e8752737b2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-15 08:54:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"498715b482ddf12c5cd589532c8c83452fa902c6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-15 08:55:01.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc6c71b81fce4e44ae4a656810302845\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ed5b8b1aeca45d28dff8507854041d8 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"cef1677216cee6ac5fa16270ea025a04bdf89cb6","revisions":{"8c584d1d8bf086161e4e4836d37349b2e200ffdf":{"kind":"REWORK","_number":1,"created":"2026-02-28 12:36:38.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/15/978315/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/15/978315/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/978315/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/978315/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/978315/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/15/978315/1"}}},"commit":{"parents":[{"commit":"0b346d62f5f3fb0fd0b270326a112c52ef5d3110","subject":"Fix Keystone SSL verification handling in Trove service session","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0b346d62f5f3fb0fd0b270326a112c52ef5d3110"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-01-14 19:22:19.000000000","tz":300},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-28 12:36:38.000000000","tz":0},"subject":"Fix hardcoded WAL archive cleanup threshold in PostgreSQL","message":"Fix hardcoded WAL archive cleanup threshold in PostgreSQL\n\nThe current WAL archive cleanup logic triggers only when the archive\ndirectory exceeds 50% of the total volume size. On instances with\nlarge volumes and relatively small effective data size, this allows\nthe WAL archive to grow excessively and consume most of the disk.\n\nFor example, on a 200GB volume with ~10GB of actual database data,\nWAL cleanup starts only after the archive reaches ~100GB. With regular\nscheduled backups, the archive size then remains close to this value,\nunnecessarily reserving disk space and increasing the risk of running\nout of free space as the database grows.\n\nThis change replaces the hardcoded volume-based threshold with a limit\nderived from max_wal_size and a configurable safety factor:\n\n  max_archive_size \u003d max_wal_size * WAL_SAFETY_FACTOR\n\nThe proposed default WAL_SAFETY_FACTOR is 5, which provides sufficient\nheadroom for typical workloads while keeping archive growth bounded.\nUsers with write-heavy workloads can increase max_wal_size as needed.\n\nThe patch also introduces a mechanism to disable the default WAL\narchive cleanup for advanced use cases by detecting a special keyword\nin archive_command. Documentation includes an example of implementing\na custom mtime-based cleanup routine.\n\nSee also the security note regarding archive_command handling in\nvalidation-rules.json.\n\nDepends-On: https://review.opendev.org/c/openstack/trove/+/978130\nDepends-On: https://review.opendev.org/c/openstack/trove/+/978131\nChange-Id: I96fa125fa2044e40e3fbeb685cc826b62b21bd4a\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n(cherry picked from commit 0604f0ecc97504062fbf335eccc8c65c6d9b1099)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8c584d1d8bf086161e4e4836d37349b2e200ffdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8c584d1d8bf086161e4e4836d37349b2e200ffdf"}]},"branch":"refs/heads/stable/2025.1"},"19cdcb79c258b3ad741c3d39a4ed0468b441c4fa":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-02-28 13:28:48.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/15/978315/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/15/978315/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/978315/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/978315/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/978315/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/15/978315/2"}}},"commit":{"parents":[{"commit":"0b346d62f5f3fb0fd0b270326a112c52ef5d3110","subject":"Fix Keystone SSL verification handling in Trove service session","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0b346d62f5f3fb0fd0b270326a112c52ef5d3110"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-01-14 19:22:19.000000000","tz":300},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-28 13:28:48.000000000","tz":0},"subject":"Fix hardcoded WAL archive cleanup threshold in PostgreSQL","message":"Fix hardcoded WAL archive cleanup threshold in PostgreSQL\n\nThe current WAL archive cleanup logic triggers only when the archive\ndirectory exceeds 50% of the total volume size. On instances with\nlarge volumes and relatively small effective data size, this allows\nthe WAL archive to grow excessively and consume most of the disk.\n\nFor example, on a 200GB volume with ~10GB of actual database data,\nWAL cleanup starts only after the archive reaches ~100GB. With regular\nscheduled backups, the archive size then remains close to this value,\nunnecessarily reserving disk space and increasing the risk of running\nout of free space as the database grows.\n\nThis change replaces the hardcoded volume-based threshold with a limit\nderived from max_wal_size and a configurable safety factor:\n\n  max_archive_size \u003d max_wal_size * WAL_SAFETY_FACTOR\n\nThe proposed default WAL_SAFETY_FACTOR is 5, which provides sufficient\nheadroom for typical workloads while keeping archive growth bounded.\nUsers with write-heavy workloads can increase max_wal_size as needed.\n\nThe patch also introduces a mechanism to disable the default WAL\narchive cleanup for advanced use cases by detecting a special keyword\nin archive_command. Documentation includes an example of implementing\na custom mtime-based cleanup routine.\n\nSee also the security note regarding archive_command handling in\nvalidation-rules.json.\n\nDepends-On: https://review.opendev.org/c/openstack/trove/+/978328\nDepends-On: https://review.opendev.org/c/openstack/trove/+/978327\nDepends-On: https://review.opendev.org/c/openstack/trove/+/976724\nChange-Id: I96fa125fa2044e40e3fbeb685cc826b62b21bd4a\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n(cherry picked from commit 0604f0ecc97504062fbf335eccc8c65c6d9b1099)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/19cdcb79c258b3ad741c3d39a4ed0468b441c4fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/19cdcb79c258b3ad741c3d39a4ed0468b441c4fa"}]},"branch":"refs/heads/stable/2025.1","description":"Edit commit message"},"aa0403106ee5c091b523886c5c5f93ba6cd34709":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-03-02 07:51:34.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/15/978315/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/15/978315/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/978315/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/978315/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/978315/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/15/978315/3"}}},"commit":{"parents":[{"commit":"0b346d62f5f3fb0fd0b270326a112c52ef5d3110","subject":"Fix Keystone SSL verification handling in Trove service session","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0b346d62f5f3fb0fd0b270326a112c52ef5d3110"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-01-14 19:22:19.000000000","tz":300},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-03-02 07:51:34.000000000","tz":0},"subject":"Fix hardcoded WAL archive cleanup threshold in PostgreSQL","message":"Fix hardcoded WAL archive cleanup threshold in PostgreSQL\n\nThe current WAL archive cleanup logic triggers only when the archive\ndirectory exceeds 50% of the total volume size. On instances with\nlarge volumes and relatively small effective data size, this allows\nthe WAL archive to grow excessively and consume most of the disk.\n\nFor example, on a 200GB volume with ~10GB of actual database data,\nWAL cleanup starts only after the archive reaches ~100GB. With regular\nscheduled backups, the archive size then remains close to this value,\nunnecessarily reserving disk space and increasing the risk of running\nout of free space as the database grows.\n\nThis change replaces the hardcoded volume-based threshold with a limit\nderived from max_wal_size and a configurable safety factor:\n\n  max_archive_size \u003d max_wal_size * WAL_SAFETY_FACTOR\n\nThe proposed default WAL_SAFETY_FACTOR is 5, which provides sufficient\nheadroom for typical workloads while keeping archive growth bounded.\nUsers with write-heavy workloads can increase max_wal_size as needed.\n\nThe patch also introduces a mechanism to disable the default WAL\narchive cleanup for advanced use cases by detecting a special keyword\nin archive_command. Documentation includes an example of implementing\na custom mtime-based cleanup routine.\n\nSee also the security note regarding archive_command handling in\nvalidation-rules.json.\n\nDepends-On: https://review.opendev.org/c/openstack/trove/+/976724\nChange-Id: I96fa125fa2044e40e3fbeb685cc826b62b21bd4a\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n(cherry picked from commit 0604f0ecc97504062fbf335eccc8c65c6d9b1099)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/aa0403106ee5c091b523886c5c5f93ba6cd34709"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/aa0403106ee5c091b523886c5c5f93ba6cd34709"}]},"branch":"refs/heads/stable/2025.1","description":"Edit commit message"},"6b759365e978f493abedd83b44e5eb82cb466420":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-03-02 11:32:41.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/15/978315/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/15/978315/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/978315/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/978315/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/978315/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/15/978315/4"}}},"commit":{"parents":[{"commit":"0b346d62f5f3fb0fd0b270326a112c52ef5d3110","subject":"Fix Keystone SSL verification handling in Trove service session","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0b346d62f5f3fb0fd0b270326a112c52ef5d3110"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-01-14 19:22:19.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-03-02 11:32:41.000000000","tz":0},"subject":"Fix hardcoded WAL archive cleanup threshold in PostgreSQL","message":"Fix hardcoded WAL archive cleanup threshold in PostgreSQL\n\nThe current WAL archive cleanup logic triggers only when the archive\ndirectory exceeds 50% of the total volume size. On instances with\nlarge volumes and relatively small effective data size, this allows\nthe WAL archive to grow excessively and consume most of the disk.\n\nFor example, on a 200GB volume with ~10GB of actual database data,\nWAL cleanup starts only after the archive reaches ~100GB. With regular\nscheduled backups, the archive size then remains close to this value,\nunnecessarily reserving disk space and increasing the risk of running\nout of free space as the database grows.\n\nThis change replaces the hardcoded volume-based threshold with a limit\nderived from max_wal_size and a configurable safety factor:\n\n  max_archive_size \u003d max_wal_size * WAL_SAFETY_FACTOR\n\nThe proposed default WAL_SAFETY_FACTOR is 5, which provides sufficient\nheadroom for typical workloads while keeping archive growth bounded.\nUsers with write-heavy workloads can increase max_wal_size as needed.\n\nThe patch also introduces a mechanism to disable the default WAL\narchive cleanup for advanced use cases by detecting a special keyword\nin archive_command. Documentation includes an example of implementing\na custom mtime-based cleanup routine.\n\nSee also the security note regarding archive_command handling in\nvalidation-rules.json.\n\nDepends-On: https://review.opendev.org/c/openstack/trove/+/976724\nDepends-On: https://review.opendev.org/c/openstack/trove/+/978437\nChange-Id: I96fa125fa2044e40e3fbeb685cc826b62b21bd4a\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n(cherry picked from commit 0604f0ecc97504062fbf335eccc8c65c6d9b1099)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6b759365e978f493abedd83b44e5eb82cb466420"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6b759365e978f493abedd83b44e5eb82cb466420"}]},"branch":"refs/heads/stable/2025.1","description":"Edit commit message"},"863ce024725bad33682fa809f517f8a9ac2d12bd":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-03-02 11:34:40.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/15/978315/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/15/978315/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/978315/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/978315/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/978315/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/15/978315/5"}}},"commit":{"parents":[{"commit":"0b346d62f5f3fb0fd0b270326a112c52ef5d3110","subject":"Fix Keystone SSL verification handling in Trove service session","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0b346d62f5f3fb0fd0b270326a112c52ef5d3110"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-01-14 19:22:19.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-03-02 11:34:40.000000000","tz":0},"subject":"Fix hardcoded WAL archive cleanup threshold in PostgreSQL","message":"Fix hardcoded WAL archive cleanup threshold in PostgreSQL\n\nThe current WAL archive cleanup logic triggers only when the archive\ndirectory exceeds 50% of the total volume size. On instances with\nlarge volumes and relatively small effective data size, this allows\nthe WAL archive to grow excessively and consume most of the disk.\n\nFor example, on a 200GB volume with ~10GB of actual database data,\nWAL cleanup starts only after the archive reaches ~100GB. With regular\nscheduled backups, the archive size then remains close to this value,\nunnecessarily reserving disk space and increasing the risk of running\nout of free space as the database grows.\n\nThis change replaces the hardcoded volume-based threshold with a limit\nderived from max_wal_size and a configurable safety factor:\n\n  max_archive_size \u003d max_wal_size * WAL_SAFETY_FACTOR\n\nThe proposed default WAL_SAFETY_FACTOR is 5, which provides sufficient\nheadroom for typical workloads while keeping archive growth bounded.\nUsers with write-heavy workloads can increase max_wal_size as needed.\n\nThe patch also introduces a mechanism to disable the default WAL\narchive cleanup for advanced use cases by detecting a special keyword\nin archive_command. Documentation includes an example of implementing\na custom mtime-based cleanup routine.\n\nSee also the security note regarding archive_command handling in\nvalidation-rules.json.\n\nDepends-On: https://review.opendev.org/c/openstack/trove/+/976724\nDepends-On: https://review.opendev.org/c/openstack/trove/+/975587\nChange-Id: I96fa125fa2044e40e3fbeb685cc826b62b21bd4a\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n(cherry picked from commit 0604f0ecc97504062fbf335eccc8c65c6d9b1099)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/863ce024725bad33682fa809f517f8a9ac2d12bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/863ce024725bad33682fa809f517f8a9ac2d12bd"}]},"branch":"refs/heads/stable/2025.1","description":"Edit commit message"},"cef1677216cee6ac5fa16270ea025a04bdf89cb6":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2026-03-02 17:40:37.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/15/978315/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/15/978315/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/978315/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/978315/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/978315/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/15/978315/6"}}},"commit":{"parents":[{"commit":"0b346d62f5f3fb0fd0b270326a112c52ef5d3110","subject":"Fix Keystone SSL verification handling in Trove service session","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0b346d62f5f3fb0fd0b270326a112c52ef5d3110"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-01-14 19:22:19.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-03-02 17:40:37.000000000","tz":0},"subject":"Fix hardcoded WAL archive cleanup threshold in PostgreSQL","message":"Fix hardcoded WAL archive cleanup threshold in PostgreSQL\n\nThe current WAL archive cleanup logic triggers only when the archive\ndirectory exceeds 50% of the total volume size. On instances with\nlarge volumes and relatively small effective data size, this allows\nthe WAL archive to grow excessively and consume most of the disk.\n\nFor example, on a 200GB volume with ~10GB of actual database data,\nWAL cleanup starts only after the archive reaches ~100GB. With regular\nscheduled backups, the archive size then remains close to this value,\nunnecessarily reserving disk space and increasing the risk of running\nout of free space as the database grows.\n\nThis change replaces the hardcoded volume-based threshold with a limit\nderived from max_wal_size and a configurable safety factor:\n\n  max_archive_size \u003d max_wal_size * WAL_SAFETY_FACTOR\n\nThe proposed default WAL_SAFETY_FACTOR is 5, which provides sufficient\nheadroom for typical workloads while keeping archive growth bounded.\nUsers with write-heavy workloads can increase max_wal_size as needed.\n\nThe patch also introduces a mechanism to disable the default WAL\narchive cleanup for advanced use cases by detecting a special keyword\nin archive_command. Documentation includes an example of implementing\na custom mtime-based cleanup routine.\n\nSee also the security note regarding archive_command handling in\nvalidation-rules.json.\n\nDepends-On: https://review.opendev.org/c/openstack/trove/+/976724\nDepends-On: https://review.opendev.org/c/openstack/trove/+/975587\nDepends-On: https://review.opendev.org/c/openstack/trove/+/976649\nChange-Id: I96fa125fa2044e40e3fbeb685cc826b62b21bd4a\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n(cherry picked from commit 0604f0ecc97504062fbf335eccc8c65c6d9b1099)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cef1677216cee6ac5fa16270ea025a04bdf89cb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cef1677216cee6ac5fa16270ea025a04bdf89cb6"}]},"branch":"refs/heads/stable/2025.1","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
