)]}'
{"id":"openstack%2Ftrove~981109","triplet_id":"openstack%2Ftrove~master~If6fd1f65354def972f4a7e568a1763ee486c310a","project":"openstack/trove","branch":"master","topic":"feature/allow-upgrade-to-same-version","attention_set":{},"removed_from_attention_set":{"36080":{"account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"last_update":"2026-05-13 03:28:40.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2026-05-13 02:39:46.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"If6fd1f65354def972f4a7e568a1763ee486c310a","subject":"Allow upgrade to the same version","status":"MERGED","created":"2026-03-18 13:36:00.000000000","updated":"2026-05-13 03:30:46.000000000","submitted":"2026-05-13 03:28:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"981109-feature/allow-upgrade-to-same-version","meta_rev_id":"e21e579b448685914bd25f8cb68e444d854782bf","_number":981109,"virtual_id_number":981109,"owner":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-13 03:28:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-05-13 02:39:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-05-13 02:39:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"CC":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-18 14:17:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-18 16:43:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-11 18:59:54.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-05-11 18:59:54.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"af6d04de27d7c57c1f1430587c85f2a9b4d80456","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-03-18 13:36:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"46ddf89ff4069e2145e1bb546eaf46540129b6bc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 14:17:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68a8c8cea1c141dcb278dcc1415ec031\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/960c7c835019456cbba8123a4c1309ee : SUCCESS in 9m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/42741332166d4605a9d24f439f37bb80 : SUCCESS in 7m 30s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/51bda2aa5c1c40e0b9268dc6e6ede69f : SUCCESS in 16m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"26f2b3afe30d32d8271215dad7ac4052b124ae62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 16:43:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f710165ce87f44b79c3597611abaf898\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1e5ca1787334627b2346d1623cd7457 : SUCCESS in 7m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb9a49ee43354bc596fa501a172af9a6 : SUCCESS in 5m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a48119407cea4f7d9143a67cdb8816ed : SUCCESS in 4m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8aa5d832d0994ac0991f49616b230a7a : SUCCESS in 4m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4639f9a21de744b5bd84813b0af3b306 : SUCCESS in 7m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7912135eeb424213893716d620acb1ad : SUCCESS in 5m 29s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/dc85d966d45e499cb666951b3c14d371 : TIMED_OUT in 3h 02m 59s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/3ede1afe08fd448384f477f1f7a9ba7f : FAILURE in 22m 45s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/d44f8a45ec7844ed851360e19916df32 : SUCCESS in 2h 40m 01s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/e19591a0a94a4ea78141eaae499d2621 : TIMED_OUT in 3h 04m 15s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/5860ad59b8834ff38d8db60041cc7fef : TIMED_OUT in 3h 04m 31s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/bc7ea9c9aa524b888d26fa6ee1681961 : SUCCESS in 2h 11m 14s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0d0ee010d3c24d5ebe5200dd39bee9ad : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":1},{"id":"57b6f541eedbfd3abe784252f673dd124b87799f","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-09 09:26:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c076f03efb47cf38e0c04267e836089f25cc4ba3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-09 09:29:00.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"7a2a561935a30b4c13c369f5fa54c9af68d403a3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-09 09:39:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d99d67c1e32c470a849ab4dc4b207592\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8ce5164f16a74905b89aa553dc200300 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6123fb6386d04a36b34840ff07939ea7 : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/69dc85529f254929b4b81d7931e3252e : SUCCESS in 8m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c87aa7b2274dad4702b3b4f6135eeb1f161eddd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-09 11:14:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f82fae12e1b456190f17db52c3348f2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8dbd813b48864b9aa8312d526b2b9b9e : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66710d6e0a134a7e99ad2204b957f1e4 : SUCCESS in 5m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a90df0c420d642129f8526af3788d2d0 : SUCCESS in 4m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/026b071de0524fe9b0e01b34797498d5 : SUCCESS in 4m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/17208679e17b475f956800d789aec2d8 : SUCCESS in 9m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddfcd1c3c6b54438bda4e52d3b61c798 : SUCCESS in 10m 56s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/564310d41bd741f2a940b4b93c3c1ba6 : SUCCESS in 44m 52s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/70b35d636b6049129b1108050cba450d : SUCCESS in 1h 41m 12s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/6f3d1e67df724b0bbc2b588931a71df7 : SUCCESS in 1h 24m 23s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/20a11a5f606640299f4e39f43b38615d : SUCCESS in 1h 44m 12s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/34743c0e2cfa40c7a15422fcbac5514c : SUCCESS in 1h 23m 39s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/f4dee829563b4097b39fea9f0ca56002 : SUCCESS in 1h 16m 37s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f1816b30278c4d1d9367a306b5907a76 : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":2},{"id":"f17b499121a73788271f7d806139195fd7644b12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-11 10:40:11.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"22aa9bc695d2874109f06559c4792dcf35b31421","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 10:56:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b68549aa68441e980045f9d9c59b895\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/55704c4afb2549f0b41022c32cc06710 : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/be5b9669a6eb4bf88a77a11c3e000560 : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/06baa89265db4ae4b8fd3852cef6d169 : SUCCESS in 10m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a3e98472d1c75f20eebe401dd1af64e977971ed2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 11:18:24.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/trove/+/981115 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"0b1abcf9f54c8e138c84fc3e40a2e6fdc707a161","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-11 14:38:34.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5cf4492f1787fa95fc131db97e908870d8fef813","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 14:53:52.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ef528ea299b40f0bdeeefd47a6b0f97\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/69f4c424c4e6432eacc94c0eb4080fa2 : SUCCESS in 8m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e166794b23284f48893b0663cc01de25 : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/398ecf59d56449ad8de1884695de5f77 : SUCCESS in 9m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2d769ec4b63a15971972643a15d2e46392862c25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 16:33:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e67e0c76ec55462bbae69f8b6bcb2eb6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/386ab91333204dd18f3f2d6eab58d1e5 : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b613e7e174446c8aaa5390d6bbcd201 : SUCCESS in 5m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8714d537d36d4484a1754facc878da44 : SUCCESS in 5m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c3795b6d34074e39aed88b5814cb590a : SUCCESS in 5m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9737cf296ecc4a77b5ab505cc9a2ad3f : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7ef65308cc74526913cf89b98612f65 : SUCCESS in 11m 04s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/a848b70642bc432bbe5f968810651d9f : SUCCESS in 52m 25s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/befd1be1abfc442da120a52f4576b5fa : SUCCESS in 1h 48m 38s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/8f38aeb144934936a1a5d0b91ddbc554 : SUCCESS in 1h 32m 34s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/95b6cbf3b4b54006a1648e5843f97ea6 : SUCCESS in 1h 49m 02s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/cf04caa2b32248199c5a55bdc0e7a7a7 : SUCCESS in 1h 24m 44s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/5bd01d71c0e24d33ba6bbcb402556359 : SUCCESS in 1h 20m 02s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5c3339dd9b51428896f7cd30df6040c7 : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":3},{"id":"4aa12c0a26d191392bd4e3f08c975cfb72c1d257","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-05-13 02:39:46.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b73199cb8bd1e610fe0c2185030fc7559dad9390","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 02:40:21.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a0d683c90db95f8bfed2fcad4cb729304827349b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 03:28:40.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11eaca07f5f7485aac29ce20289de671\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/214f2dd106c2448691b3477b5f85ecd3 : SUCCESS in 5m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6136abf0f35549c7acf375d0ced79674 : SUCCESS in 4m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/584ceecefdee4f45b12e3ba44c33395c : SUCCESS in 6m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/462848668df645808d87bb8472fbdcf0 : SUCCESS in 7m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d61c99c853c44dcab46d99d34511703e : SUCCESS in 8m 38s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/21605af8b5da4eee9dcc40699edd179a : SUCCESS in 45m 55s","accounts_in_message":[],"_revision_number":3},{"id":"1c15ffb858e25e76ebdb523a6bfb91b99a16466b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 03:28:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"e21e579b448685914bd25f8cb68e444d854782bf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 03:30:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67a842b73a164f0abe7ee2e1005633ab\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3adaa388a931423a8040f142c63d14a5 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"da358b6047890b2ca1ea0deecf76ee6cdcc37639","revisions":{"4764d6eea55975f0c34ef5774c908f030b92028a":{"kind":"REWORK","_number":1,"created":"2026-03-18 13:36:00.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/09/981109/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/09/981109/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/09/981109/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/09/981109/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/09/981109/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/09/981109/1"}}},"commit":{"parents":[{"commit":"8a6bf89072a0eda7c3e50dc0ee2a2c9651d09409","subject":"Remove unused replication_user options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8a6bf89072a0eda7c3e50dc0ee2a2c9651d09409"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-03-18 13:29:35.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-03-18 13:35:57.000000000","tz":300},"subject":"Allow upgrade to the same version","message":"Allow upgrade to the same version\n\nCurrent manager logic also blocks upgrade to the same version,\npreventing actual execution in tests without pre-upgrade datastores\ndefined in configuration. This change removes that restriction,\nallowing same-version upgrades (e.g. to refresh images), and enables\nTempest to validate the upgrade flow properly.\n\nChange-Id: If6fd1f65354def972f4a7e568a1763ee486c310a\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4764d6eea55975f0c34ef5774c908f030b92028a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4764d6eea55975f0c34ef5774c908f030b92028a"}]},"branch":"refs/heads/master"},"5add1e76b9c79e57f1cf5f0d01d82afa3a3e71d8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-05-09 09:29:00.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/09/981109/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/09/981109/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/09/981109/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/09/981109/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/09/981109/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/09/981109/2"}}},"commit":{"parents":[{"commit":"01d94858288755b725293d19ece02093b3591337","subject":"Merge \"Add PDF building\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/01d94858288755b725293d19ece02093b3591337"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-03-18 13:29:35.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-09 09:29:00.000000000","tz":0},"subject":"Allow upgrade to the same version","message":"Allow upgrade to the same version\n\nCurrent manager logic also blocks upgrade to the same version,\npreventing actual execution in tests without pre-upgrade datastores\ndefined in configuration. This change removes that restriction,\nallowing same-version upgrades (e.g. to refresh images), and enables\nTempest to validate the upgrade flow properly.\n\nChange-Id: If6fd1f65354def972f4a7e568a1763ee486c310a\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5add1e76b9c79e57f1cf5f0d01d82afa3a3e71d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5add1e76b9c79e57f1cf5f0d01d82afa3a3e71d8"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"4764d6eea55975f0c34ef5774c908f030b92028a","theirs":"01d94858288755b725293d19ece02093b3591337","contains_conflicts":false}},"da358b6047890b2ca1ea0deecf76ee6cdcc37639":{"kind":"REWORK","_number":3,"created":"2026-05-11 10:40:11.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/09/981109/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/09/981109/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/09/981109/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/09/981109/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/09/981109/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/09/981109/3"}}},"commit":{"parents":[{"commit":"01d94858288755b725293d19ece02093b3591337","subject":"Merge \"Add PDF building\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/01d94858288755b725293d19ece02093b3591337"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-03-18 13:29:35.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-11 10:39:10.000000000","tz":300},"subject":"Allow upgrade to the same version","message":"Allow upgrade to the same version\n\nCurrent manager logic also blocks upgrade to the same version,\npreventing actual execution in tests without pre-upgrade datastores\ndefined in configuration. This change removes that restriction,\nallowing same-version upgrades (e.g. to refresh images), and enables\nTempest to validate the upgrade flow properly.\n\nDepends-On: https://review.opendev.org/c/openstack/trove/+/981115\nChange-Id: If6fd1f65354def972f4a7e568a1763ee486c310a\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/da358b6047890b2ca1ea0deecf76ee6cdcc37639"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/da358b6047890b2ca1ea0deecf76ee6cdcc37639"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
