)]}'
{"id":"openstack%2Ftrove~981222","triplet_id":"openstack%2Ftrove~master~Iebbd4e33475c68ac078b2aa10186024017d1f6a8","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"36080":{"account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"last_update":"2026-05-25 16:00:45.000000000","reason":"\u003cGERRIT_ACCOUNT_36080\u003e replied on the change","reason_account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-05-25 17:41:59.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2026-05-25 19:07:14.000000000","reason":"Change was submitted"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-05-25 15:50:46.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"Iebbd4e33475c68ac078b2aa10186024017d1f6a8","subject":"Remove version number from jobs","status":"MERGED","created":"2026-03-19 09:09:40.000000000","updated":"2026-05-25 19:08:54.000000000","submitted":"2026-05-25 19:07:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"981222","meta_rev_id":"1972f840093dc69803698acc2541ecfad024d558","_number":981222,"virtual_id_number":981222,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-25 19:07:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":2,"date":"2026-05-25 16:00:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-05-25 15:50:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":1,"date":"2026-05-25 16:00:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":-1,"date":"2026-05-25 16:00:45.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-19 09:20:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-19 11:12:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-25 13:52:35.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-03-25 15:20:37.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-03-25 15:20:37.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"state":"REVIEWER"},{"updated":"2026-03-25 15:20:37.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"f33c88685678a5aa911b13f6d3c8bd6c069fed37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-19 09:09:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"02740af77c5603a6aa48e7ba11452b6b59f05116","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 09:20:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be17155854564384a1cd8c3cc0928c17\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/38c8c4590d8946d5be2fb1a7236741bd : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5dbc20e9d9cb4919b371435335c86fa2 : SUCCESS in 7m 39s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2e8f99fa3ee043638ae5e768c82cc3e5 : SUCCESS in 8m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"70789d44a2dccb879fa2d497e8b934026f906ab0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 11:12:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/361c9bb13c904551b03d782668e1b1c9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b1d35db0e744ab8aad411d61e7a09d1 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdea5a63b88a455790e96dbee3915b48 : SUCCESS in 5m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6f2bd8ef69e248cda129c78907c8d40a : SUCCESS in 5m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8f0d155737374281bbed267db19d2ce6 : SUCCESS in 2m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/54027f21b79b4855859f3a788eef0520 : SUCCESS in 3m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fb33a297802462588e35bd3adfad50e : SUCCESS in 5m 23s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/adbe6feca33841958e36da7724191a2b : SUCCESS in 54m 03s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/4962f113e3794c86a4e6f597ef74afd8 : SUCCESS in 2h 01m 13s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/4a3adbf146f44d2eadb96a9079f48102 : SUCCESS in 1h 34m 32s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/fa1cefce26d14dcd86186df28d273399 : SUCCESS in 1h 52m 17s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/3008235d91ca4d2693a5a85ed413ff06 : SUCCESS in 40m 16s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/2b6daa96f2a04ffba1c7e87d2181f16f : SUCCESS in 1h 35m 10s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/f79f8e286ee74c528ac9c7b70317f33c : SUCCESS in 1h 25m 13s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/1ea6471551e0474bb350a5024d76181e : SUCCESS in 1h 31m 45s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/99afd9d126114d86940c245ee9a37717 : SUCCESS in 54m 45s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/74d3e0d3a3454a1693c192644f2edc69 : SUCCESS in 1h 33m 30s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/a1af09de5c46496082412760f21c4645 : SUCCESS in 1h 24m 41s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/f8d2673fc1b141ffa5260d56f82b9835 : SUCCESS in 1h 20m 47s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/31f7ea43c93e4c449822ca277297f16e : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":1},{"id":"9f51a560edbf9bb1f5504fb10b9cef97525761f3","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-25 13:52:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"346fea77d90bc8f78cc923589d135364391370e9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-25 15:19:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fc2407ad5a0af0335679382ba8b1211dc9a78ddd","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-25 15:20:37.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9ce39039e7537aeb63d092bde0c5986b4c4420c5","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-03-29 11:20:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0bb78912c3ebf3ab62cefedf840f2bd6c68cc7e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-29 14:33:17.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"0662945033bfab0931f21a9228d37796864d876e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-29 14:34:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d36afe6bd0404a322d298dd99a2c5d857e2257fc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-29 14:46:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73ce574082644120b36de8e5f4f3c91f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c50caf8cff18477c9a4619ef7f1b19fd : SUCCESS in 9m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6119e360dbf7427790dd4a0d744d8a0a : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7f79413f16ec41beae00e7fdea82f2db : SUCCESS in 11m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"757002be4fd0e9e3a5fb775232c74732bb52d2ec","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-03-29 14:48:01.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"652b3b5bb3bca416251b5613f48c3b2c5f61beaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-29 17:13:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/445d4b7007074008b783d380d7b2a3ff\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2dc14882f3cf4466bf8dd03a133d19b6 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7587aba950734a8eb160cce36fbecd32 : SUCCESS in 3m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/20ca61d0d4bc4879a0f5a2fea5feb14a : SUCCESS in 5m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/39de5570e1a242708769719b29f213e9 : SUCCESS in 4m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/46177bfe44654b2597af91fa064bb2f3 : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/294ef62bc35541f48cbb68a0dfe389c9 : SUCCESS in 5m 02s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/5aad76ed69a3499b8d1b932b1c2d3a05 : SUCCESS in 1h 03m 11s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/01e63739806a4081983aba2aa17d3127 : SUCCESS in 2h 39m 29s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/36cbcabb16dd4bd3b93c3eefff9f4aa5 : SUCCESS in 2h 00m 57s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/28a11e58171c42b9910cbdf20ebca7e0 : SUCCESS in 2h 00m 22s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/f4dbd39463664651b42ae3f25c2d4522 : SUCCESS in 1h 17m 01s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/45a29d6c7d864efdaddafe13c9b3cfa0 : SUCCESS in 1h 59m 23s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/162b103bf6574304a29ed4e431865064 : SUCCESS in 1h 18m 56s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/8ec85ec8fec44691891d49f4833bbfa2 : SUCCESS in 2h 02m 30s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/0e0db109c16b4e7ab22f1acd2b9f9c7e : SUCCESS in 46m 18s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/3780856a6e2540969a9b98d269bff88f : SUCCESS in 2h 09m 51s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/255295bf58eb4964bda6fe2810438418 : SUCCESS in 1h 50m 25s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/a31b1346ed3146ccad027cec7bbe4cd2 : SUCCESS in 1h 23m 14s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5fcb380e41e14983ba1ded9e005fd697 : SUCCESS in 6m 33s","accounts_in_message":[],"_revision_number":2},{"id":"844aed55634546864bcde27ab8a164fa6cfc031b","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-29 21:11:32.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"6a83aa4af06d7e1763e5c8bfaa9cd94b8c447fa1","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-25 07:40:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ec2447f74b1769bb64d921a065ada3c5bd527152","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-25 15:48:51.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"98178f56ce721b1b71a103713a1543617a12651c","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-25 15:50:46.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"a3d5a6cb8fd9f19948c1f0a809bc352776b985de","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-25 16:00:45.000000000","message":"Patch Set 3: Backport-Candidate-1 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"491a935de1963433473780188833f8d46db6f065","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 16:23:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3998722590754cda84650e80606224dd\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2388cf0e3c0d47638c88ef4c948cebae : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/59e0d87209434a23a2187f22e54716e4 : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8b9fea6e170f48fc99a786754ea87ab2 : SUCCESS in 13m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b4b32141dea346b0e54fa2000bb2fef4db175e9e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-25 16:32:13.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c233eb5dfe8851d91261264435c3a7281d9deb9d","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-25 17:13:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"23bd04f2bf728c23f2c5a724ec3f73c346fc7436","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-25 17:41:59.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d1b23ac62cd20418fea237cf01663979553a7089","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 18:19:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec5cc5b5aa0d4659b4454461da5af6bb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/295c4f431b334879b69e5f9479177fe1 : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48b06561b7d84997811d3a5ba82559c6 : SUCCESS in 3m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0a790dce6d32452cb2c2ef501695507f : SUCCESS in 6m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1e4b8819942b463fbe34d497c83a4d6b : SUCCESS in 3m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/94cde09a5f9949909c6d759c922e3c53 : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebca8d02816c47e8850750b8ff0d5609 : SUCCESS in 10m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee92fc6f72014376b837776a54c48050 : SUCCESS in 7m 02s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/15ac20aad22e49eb92bd0ab31a7e1e17 : SUCCESS in 54m 48s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/b4d67ea79b4742bbb4a4f20ba6d52fa5 : SUCCESS in 2h 29m 34s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/76938ca0974d4b08b5810b3314c98f9a : SUCCESS in 1h 59m 19s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/3c3bcdef555d49f58665da65c86a51e8 : SUCCESS in 1h 58m 49s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/ada467a17bee47d0a6adcb89d29ee859 : SUCCESS in 50m 50s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/ec66577671ad465881ba48b9d2341fbf : SUCCESS in 1h 38m 00s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/21aa65400c294d18a47ca95b01f616f2 : SUCCESS in 1h 34m 17s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/2f6c90b1c9d14168b22653b46a803418 : SUCCESS in 1h 32m 09s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/9b326514be4640d48a86b9da42dca25f : SUCCESS in 53m 15s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/54791e1831314258a0a80c829cc62dde : FAILURE in 1h 00m 56s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/2833bf29f8074687ad4f2aa8d0ba9026 : SUCCESS in 1h 18m 11s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/daadb6c069594462adf7b1f0c72a85fb : SUCCESS in 1h 24m 47s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/956e753a6c26412ab2e5b86857fe483f : SUCCESS in 5m 43s","accounts_in_message":[],"_revision_number":3},{"id":"82c18b5e24b6837f9b0e4b058fa298b82291a971","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 18:20:25.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"aabf7b22ef249f4411222057325c7b03ed5b47db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 19:07:14.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c30eb0c689c4525a3d8b42850dc37db\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee3d2a8f41a4418583c1e244fb53f01c : SUCCESS in 5m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/395c5f496d134736b1b329eb95ad4f2c : SUCCESS in 4m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5523d37832fe4e4ebc5170295c4bcdac : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbd94fc579f44bbea1a13b3ae722fd00 : SUCCESS in 9m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a44c0ee975824d65b5d7c5090283b743 : SUCCESS in 7m 00s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/b5f52055fd564318a5e697417a2587f9 : SUCCESS in 46m 24s\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/83346ae4c8ec4965b14cd5b9b6b97b86 : SUCCESS in 41m 50s\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/9d6962af4e55405698e6527acc839f2a : SUCCESS in 45m 30s","accounts_in_message":[],"_revision_number":3},{"id":"b4341e116e3a1e68ad7a679d7ab32e0b87c964a0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 19:07:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"1972f840093dc69803698acc2541ecfad024d558","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 19:08:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b88f6f4e02cd4cda8ff6de4150979541\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8d52bd4bcec4adeb9088628c2e3a215 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a2a53d425aa0a93149ca3268248470ff0129cc60","revisions":{"9b960792dcd5df40f831118697ac81f8378413b9":{"kind":"REWORK","_number":1,"created":"2026-03-19 09:09:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/981222/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/22/981222/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/22/981222/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/22/981222/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/22/981222/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/22/981222/1"}}},"commit":{"parents":[{"commit":"8a6bf89072a0eda7c3e50dc0ee2a2c9651d09409","subject":"Remove unused replication_user options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8a6bf89072a0eda7c3e50dc0ee2a2c9651d09409"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-19 09:07:04.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-19 09:07:04.000000000","tz":540},"subject":"Remove version number from jobs","message":"Remove version number from jobs\n\nThese jobs are also used in trove-tempest-plugin. The current job name\nformat causes a few challenges.\n\n - We can\u0027t directly change the job name due to cross-dependency\n   between trove and trove-t-p\n - Adding stable branch job to trove-t-p requires maintaining\n   appropriate versions per stable branches.\n\nDrop the version number from job name clears these concerns.\n\nIn case two version should be tested, we keep the non-version job\nfor primary version and add a new one with specific version name.\n\nChange-Id: Iebbd4e33475c68ac078b2aa10186024017d1f6a8\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9b960792dcd5df40f831118697ac81f8378413b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9b960792dcd5df40f831118697ac81f8378413b9"}]},"branch":"refs/heads/master"},"0b7916e1d93996ab5fa05bb1068ed6b92dd76b1c":{"kind":"REWORK","_number":2,"created":"2026-03-29 14:33:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/981222/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/22/981222/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/22/981222/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/22/981222/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/22/981222/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/22/981222/2"}}},"commit":{"parents":[{"commit":"8a6bf89072a0eda7c3e50dc0ee2a2c9651d09409","subject":"Remove unused replication_user options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8a6bf89072a0eda7c3e50dc0ee2a2c9651d09409"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-19 09:07:04.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-29 14:33:07.000000000","tz":540},"subject":"Remove version number from jobs","message":"Remove version number from jobs\n\nThese jobs are also used in trove-tempest-plugin. The current job name\nformat causes a few challenges.\n\n - We can\u0027t directly change the job name due to cross-dependency\n   between trove and trove-t-p\n - Adding stable branch job to trove-t-p requires maintaining\n   appropriate versions per stable branches.\n\nDrop the version number from job name clears these concerns.\n\nIn case two version should be tested, we keep the non-version job\nfor primary version and add a new one with specific version name.\n\nChange-Id: Iebbd4e33475c68ac078b2aa10186024017d1f6a8\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0b7916e1d93996ab5fa05bb1068ed6b92dd76b1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0b7916e1d93996ab5fa05bb1068ed6b92dd76b1c"}]},"branch":"refs/heads/master"},"a2a53d425aa0a93149ca3268248470ff0129cc60":{"kind":"REWORK","_number":3,"created":"2026-05-25 15:48:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/981222/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/22/981222/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/22/981222/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/22/981222/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/22/981222/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/22/981222/3"}}},"commit":{"parents":[{"commit":"0a7462b4ffac6cf109a44e0443380f58a2fd56b8","subject":"Merge \"Add support PostgreSQL 18\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0a7462b4ffac6cf109a44e0443380f58a2fd56b8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-19 09:07:04.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-25 15:48:31.000000000","tz":540},"subject":"Remove version number from jobs","message":"Remove version number from jobs\n\nThese jobs are also used in trove-tempest-plugin. The current job name\nformat causes a few challenges.\n\n - We can\u0027t directly change the job name due to cross-dependency\n   between trove and trove-t-p\n - Adding stable branch job to trove-t-p requires maintaining\n   appropriate versions per stable branches.\n\nDrop the version number from job name clears these concerns.\n\nIn case two version should be tested, we keep the non-version job\nfor primary version and add a new one with specific version name.\n\nChange-Id: Iebbd4e33475c68ac078b2aa10186024017d1f6a8\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a2a53d425aa0a93149ca3268248470ff0129cc60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a2a53d425aa0a93149ca3268248470ff0129cc60"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
