)]}'
{"id":"openstack%2Ftrove~985930","triplet_id":"openstack%2Ftrove~stable%2F2025.1~I9c5fdb4a66e34d74151e5ce13395cb1e08d4d862","project":"openstack/trove","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"36080":{"account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"last_update":"2026-04-24 12:52:52.000000000","reason":"Change was abandoned"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2026-04-24 12:52:52.000000000","reason":"Change was abandoned"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-04-24 12:52:49.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I9c5fdb4a66e34d74151e5ce13395cb1e08d4d862","subject":"Fix backup failure when caller uses an application credential","status":"ABANDONED","created":"2026-04-23 08:01:59.000000000","updated":"2026-04-24 12:52:52.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":985929,"cherry_pick_of_patch_set":1,"meta_rev_id":"a23ba5aeb35d21b5da99593b776a177acac18cba","_number":985930,"virtual_id_number":985930,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-04-23 09:19:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"disliked":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":-1,"date":"2026-04-24 11:44:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":1,"date":"2026-04-23 13:17:22.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":39035,"name":"Kostiantyn Velychkovskyi","display_name":"K Velychkovskyi","email":"k.velychkovskyi@cloudfarm.tech","username":"velychkovsky","status":"CloudFarm Co-Founder / CTO"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-23 08:01:59.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":39035,"name":"Kostiantyn Velychkovskyi","display_name":"K Velychkovskyi","email":"k.velychkovskyi@cloudfarm.tech","username":"velychkovsky","status":"CloudFarm Co-Founder / CTO"},"state":"CC"},{"updated":"2026-04-23 08:02:29.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2026-04-23 09:19:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-23 13:17:22.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-04-24 11:44:16.000000000","updated_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"reviewer":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"state":"REVIEWER"}],"messages":[{"id":"5a40dba91b7b6925f87f8575548879e59428c83f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-23 08:01:59.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2025.2.","accounts_in_message":[],"_revision_number":1},{"id":"c2b3e50b24732e9b98c4bdf344d835f0903f05e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 09:19:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2b1f7506ded459b85a95c90afb12958\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f5d4a84736684aa6ab1a237697827722 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/707c1b1d4fa64a14be520bfdc0b0963f : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e21bd8264e834e1e8ce9e6c91ca8f253 : SUCCESS in 3m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5bce307fb0f84545ae471f04fc80d28f : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6775638b60e1443f95dc7af69bca0957 : SUCCESS in 5m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf322c89a8174145aa5b5dc195989ec3 : SUCCESS in 5m 06s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/56814053ad1c4df1818bd987346482f1 : FAILURE in 1h 01m 21s (non-voting)\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/e48a5cf61efb4383bc0eaa0b0df48e6b : FAILURE in 1h 16m 24s (non-voting)\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/9cff5e5a483a40378c37b6533e075c43 : FAILURE in 1h 15m 44s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4e51b6d7d60e4c56ac75f99b5fed669a : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":1},{"id":"8c3d1c4b8fa607155ac680222562dbcd10894bfd","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-23 13:17:22.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c697ad95a26bfde8747b95e3f72c6984914f3a00","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-04-24 11:44:16.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1825f8b303f8abb0ea2c24421b6bae219a60cc4c","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-24 12:52:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a23ba5aeb35d21b5da99593b776a177acac18cba","tag":"autogenerated:gerrit:abandon","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-24 12:52:52.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"175aaa9a269ba67fe99bd6e357ad5cdf987d8758","revisions":{"175aaa9a269ba67fe99bd6e357ad5cdf987d8758":{"kind":"REWORK","_number":1,"created":"2026-04-23 08:01:59.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/30/985930/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/30/985930/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/985930/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/985930/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/985930/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/30/985930/1"}}},"commit":{"parents":[{"commit":"5be9a891edce7231a7733116d1d2b908f21010e1","subject":"Merge \"Fix hardcoded WAL archive cleanup threshold in PostgreSQL\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5be9a891edce7231a7733116d1d2b908f21010e1"}]}],"author":{"name":"Kostiantyn Velychkovskyi","email":"k.velychkovskyi@cloudfarm.tech","date":"2026-04-19 08:54:59.000000000","tz":120},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-04-23 08:01:59.000000000","tz":0},"subject":"Fix backup failure when caller uses an application credential","message":"Fix backup failure when caller uses an application credential\n\nWhen the Trove API is called with an application-credential token,\nthe backup container opens Swift session by constructing\n``keystoneauth1.identity.v3.Token`` with an explicit ``project_id``\nand ``project_domain_name``. Keystone treats that as a rescope and\nrejects it with ``401 Application credentials cannot request a\nscope``. The backup process exits without \"Backup OK\" summary line,\nthe task manager reports ``Cannot parse backup\noutput``, and replica/clone bootstrap breaks along with it.\n\nThe forwarded ``context.auth_token`` is already project-scoped to\nthe instance owner\u0027s project, so the rescope parameters are\nredundant. Drop ``project_id`` and ``project_domain_name`` from\nthe ``v3.Token`` construction in ``backup/storage/swift.py``;\nre-authenticating a token without scope parameters works for\npassword, application-credential and trust-issued tokens alike.\n\nAdd unit tests for ``_get_user_keystone_session`` and\n``_get_service_client``\n\nStory: 2011763\nTask: 54182\nChange-Id: I9c5fdb4a66e34d74151e5ce13395cb1e08d4d862\nSigned-off-by: Kostiantyn Velychkovskyi \u003ck.velychkovskyi@cloudfarm.tech\u003e\n(cherry picked from commit a365950b48ba573cab3404e46e9e8d1ad5bdf698)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/175aaa9a269ba67fe99bd6e357ad5cdf987d8758"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/175aaa9a269ba67fe99bd6e357ad5cdf987d8758"}]},"branch":"refs/heads/stable/2025.1","conflicts":{"ours":"5be9a891edce7231a7733116d1d2b908f21010e1","theirs":"68753d20e77734c72140a54738cf6352c1df33ff","contains_conflicts":false}}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
