)]}'
{"id":"openstack%2Ftrove~986090","triplet_id":"openstack%2Ftrove~master~I558bec8bae5db9c244f9ba6d87cefd7478f12cb1","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"36080":{"account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"last_update":"2026-07-19 14:36:22.000000000","reason":"\u003cGERRIT_ACCOUNT_36080\u003e replied on the change","reason_account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2026-07-19 09:19:30.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-07-19 15:26:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I558bec8bae5db9c244f9ba6d87cefd7478f12cb1","subject":"Enable userdata refresh on instance rebuild (Nova ≥ 2.57)","status":"MERGED","created":"2026-04-24 13:46:50.000000000","updated":"2026-07-19 15:27:27.000000000","submitted":"2026-07-19 15:26:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"986090","meta_rev_id":"ab869b2f62fc22c14c78263e2ca068a1d58623ff","_number":986090,"virtual_id_number":986090,"owner":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-19 15:26:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":2,"date":"2026-07-19 14:36:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-07-19 09:19:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":1,"date":"2026-07-19 14:36:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":-1,"date":"2026-07-16 15:05:49.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-24 14:04:11.000000000","updated_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2026-04-24 14:48:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-04-24 16:22:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-24 22:28:15.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"b6c887be451462ab0070b82f18937be1f6895155","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-04-24 13:46:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4f364809fafe9d952d022ea6ae1babf41cfe4f65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-04-24 13:59:31.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"549534524828488383a4eaedc7db83d3f06769f0","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-04-24 14:04:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5664e1be6bb28758dd9e58d70e8c336fd03085d6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-24 14:48:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bf8fd7fe54c4cfba523b19983fc4457\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a43ba11699eb49f994167998f412c808 : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7531724ea6964d00b5a7e9376a6a4037 : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a7fec43628274a109e7134009b9da7bc : SUCCESS in 10m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8b006b2db757eae5087e4fd38f6448e3563d1880","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-24 16:22:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3790da52925749f7bacdeac1619b4dec\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84d938f318f74685a31492f28a12a916 : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c25f28beb2184fbb92811a1d9942399b : SUCCESS in 6m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3f440531145b42258723b8afc819d979 : SUCCESS in 4m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/16cf52323c014ecdae0d8d286d0e45b9 : SUCCESS in 3m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d3468167a2b8407db87ab22aac1ef611 : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86977ae88f7249ddb4c6188c0d496474 : SUCCESS in 6m 09s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/1e2702ac463e4c5190da3989ecfc4a85 : SUCCESS in 49m 01s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/503a619be19d4de68215000b88aeb82d : FAILURE in 1h 42m 23s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/e6f7e2d5cc3e4b3f942e225624cc85a1 : FAILURE in 1h 05m 41s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/523b4a84685e440a80abcec6d8f5e197 : SUCCESS in 2h 17m 50s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/3a3df01a9f554086899b498b7d48cda8 : SUCCESS in 49m 25s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/0a2c60bfc4ad4b9380490d833b5f9a63 : FAILURE in 1h 02m 14s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/0c7ea9b3a7f5436a93e563441c412b64 : FAILURE in 1h 02m 09s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/ae92aad2b6df49c5901f6a4854d8ad18 : SUCCESS in 1h 46m 20s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/aad732cda24a42e295f3af4ee395dd9e : SUCCESS in 1h 07m 42s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/4cea2c5825b84784bc4dcf5cf363b1fb : FAILURE in 58m 18s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/886c7f99f56d4aacb4ff0a80143e254e : FAILURE in 55m 27s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/73ff7e37a4b845679c41c2c97bc27488 : SUCCESS in 1h 45m 13s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7dcd6372ab47464185925d68b47b92b5 : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":2},{"id":"db2ceb8c75d72a2d5006b9e96983b012a975cc28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-04-27 06:14:47.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"352b81f96bd15be883e2dd8b44a8a3f2dd7cdcbb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 06:53:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff65a2aafe7f440794a60749b472c1a9\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/84fea663e394425fb71f8794d593daf2 : SUCCESS in 7m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/de9da58cf98a4cbf9942c12a335bbb09 : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/cc50155957ac4a7281d9b4109efb8f12 : SUCCESS in 9m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"68306ad0fe416031c5e18cb75a93a99dbeb03900","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 08:05:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be96ee849a074301988a5ab5e14a4478\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0cc4cec26db3434ebb7f7cbc70b5c22b : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d2e9cbcfcc6420ea1b5e85f8b0a1595 : SUCCESS in 5m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/174d18835b28473aafd8f358d44b5dd0 : SUCCESS in 3m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/64214d600a1841f495617f7f5c0804a0 : SUCCESS in 2m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ecfb1326e684456395cddedd550cfbb0 : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2520f72aef0d4bed941f775e3a0e5681 : SUCCESS in 5m 17s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/5c9294b0df75441aa41e03d4b51e7d2e : SUCCESS in 46m 52s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/c55288a58c3f4384b5b6b252b0b848a4 : SUCCESS in 1h 49m 07s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/0b5b7e28fada4386968116e9090bfad5 : SUCCESS in 1h 30m 59s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/fa3ff3f201e145c09140c79c671c34a1 : SUCCESS in 1h 42m 46s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/b0041613e8b641d69fd3c216e9ca84cc : SUCCESS in 40m 55s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/fe0a3502efd14358b0615c5cdc299c53 : SUCCESS in 1h 30m 41s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/3671e99a94d543ee9115f1033bdb6725 : SUCCESS in 1h 14m 47s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/e6b7322708574be6ba36e5782fada459 : SUCCESS in 1h 26m 51s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/b6c06cdf227a4d35aebfc3648fb73e2e : SUCCESS in 46m 26s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/9215809e42194c56b0066c8f059591cc : SUCCESS in 1h 33m 35s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/8e5f484d483b413fbf4332da78a3c9f8 : SUCCESS in 1h 16m 02s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/25572c7b626a43e88aca313df3f654b8 : SUCCESS in 1h 22m 37s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/38cabc4fa05c49ecacfc2a2e33830f08 : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":3},{"id":"3a00ad6a77c2e66e983aa54cc08e65f5efaa2ef1","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-24 22:28:15.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"8375c4be6704caf075aa3f885cc54016890b75d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-24 08:56:43.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"49588ed694447fb173099fd5e35c1f3162f69743","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 09:36:38.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd71a17d9a02427aa2919d2c3107f217\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/542288d8dbdb46a58c6f01e46ec45f6c : SUCCESS in 9m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ed83af6032224ad5a112a544532ed586 : SUCCESS in 7m 41s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0533a0f42c964d3b82c9b052818903cf : SUCCESS in 14m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"fc181d4886d34b10dbbf7ff0034c007888b570f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 11:09:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18d3d7502999470c87cfe5acb2c9adc8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f729ecac75694821a7b140272eeb23f5 : SUCCESS in 8m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e53722d3e2f47bb9cc4264545489d43 : SUCCESS in 5m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f9c720307523401a8244efa5b0af849f : SUCCESS in 4m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8becc3b310ba4bfcaba4efa05883946c : SUCCESS in 5m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4aa2bcc8111f4e08b2e87419fff81d6e : SUCCESS in 8m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42b8741a2b7c4dadaf09782d3bce8fe8 : SUCCESS in 11m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0dd1177d724c4bce8339028c5ea3657c : SUCCESS in 3m 47s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/dac8fbefd98b4bb58b77e0b56e9434b8 : SUCCESS in 56m 42s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/a0a51fafc4be417184b359b229a53d04 : SUCCESS in 2h 07m 27s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/aa0ef009db9e4f6089fb42c9cf0f1789 : SUCCESS in 1h 51m 02s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/eb217fda308544f7a5200d8f5edbbab3 : SUCCESS in 1h 50m 24s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/db31aba4eb5a490fa80114edb1618c54 : SUCCESS in 52m 21s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/5a78a7c17e084149874bb208adb2884b : SUCCESS in 2h 03m 54s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/e7a420c94c0d4c10b9ccdb563e87cc89 : SUCCESS in 1h 25m 50s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/eccb601c85cd4d3ba537a6016e0e7114 : SUCCESS in 2h 02m 37s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/3c5837d2bb9848cf915777d5da798f47 : SUCCESS in 51m 10s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/f8f260ba9ddc4fbb9662a5a16c6be9ea : FAILURE in 1h 30m 39s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/cc0a8a899ac0498ea249fac063a02f76 : SUCCESS in 1h 20m 47s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/b6f2e76a360440cb97ceaeb115df8c40 : SUCCESS in 1h 23m 06s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e941d8e668ac47d8b3ee907fb31b19d4 : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":4},{"id":"2afcb2b5542d3470b87f0829f62dccd47db9c05d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-12 09:42:06.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased. Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"bef4223afd85a867d0bfcb32becda836909a8f3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-12 09:52:27.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"cfcc68bc093c4573ea379bc8eb2e8e4c50fe6d27","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-12 10:02:56.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d501894f47ee87fc223aac95433118f2d19d8ccd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-12 10:08:52.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a102b6d49b2646d4a4fb4acb9c8d6e7e\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b9e22e91956041bf95415539d8feac35 : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b8b1c8a3f7b746eba95219101ccf6037 : SUCCESS in 8m 14s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/54add72b406349ffa3feb46ae0f9c2a6 : SUCCESS in 14m 21s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e9f84bac51731bf4245208032f5d257b5ca81b57","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-12 11:23:43.000000000","message":"Patch Set 6: Backport-Candidate-1","accounts_in_message":[],"_revision_number":6},{"id":"2e6b0f7dbe8cdf47a49567d06be9e981e5f69cad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-12 11:48:21.000000000","message":"Uploaded patch set 7.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR **is:MIN**\")\n","accounts_in_message":[],"_revision_number":7},{"id":"4fd58449e63ab9103a3231d1496a730bfc86c4e4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-12 12:04:29.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7070f87943804d11a4c3857fb3c82675\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4cd5168e4caa4ff39f01eeaedb83ab4b : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/31d0cdd04cbc4864879fcf7d6883bda9 : SUCCESS in 7m 06s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/725706f2e8fd4ca8bde3375c41b79e2a : SUCCESS in 13m 49s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"d04fc50e4b6de61b2fff4892d6da49fbc9e8183a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-12 13:55:06.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/adf7cd607d704a1b854e9497489d0c0d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf39cb90ea274055a200ab8b08dafc39 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aab5b19abd0646c095aac7c0699e210c : SUCCESS in 5m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5cd8052646194b39b906112869245a97 : SUCCESS in 4m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ca7601ac7b4a44aebc7c39fc127be78d : SUCCESS in 4m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/de916911179d46f0af6ecbc0b1dc837c : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e4c9e1b968b4049a7bd04664bf8f283 : SUCCESS in 10m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c661eb2ff71a4d5189bc6b52522b7aa6 : SUCCESS in 8m 12s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/76cc64f1d1af461a8f8786681d019e38 : SUCCESS in 1h 11m 25s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/89991527430f404bb8734ff8a09f1c86 : SUCCESS in 2h 05m 54s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/f0d7c7294b854f2f9b7377c7ae1b981d : SUCCESS in 1h 55m 28s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/8599adab116c4fa8a1ce62cf63d1a095 : SUCCESS in 1h 45m 15s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/54225306382647558b31c0d8aad12916 : SUCCESS in 59m 26s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/e97e1b9a197c43878a9491e8ac6a3410 : SUCCESS in 1h 46m 33s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/ba76ca2f2a9e4479a4fdc43dc00a4b5a : SUCCESS in 1h 34m 19s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/7737bd0bd4f348549ef49fec11a37cc3 : SUCCESS in 1h 35m 09s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/2666d22b6d91408292fb58af7db86ca8 : SUCCESS in 1h 04m 45s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/cb48889d932a4d729a35f16093360b68 : SUCCESS in 1h 50m 42s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/af137f6dff194809be6bc62965b7fc6b : SUCCESS in 1h 21m 41s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/0416c24fd8c042ce80465a6cfb238ca9 : SUCCESS in 1h 25m 15s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d7d441ccf9754ed68a710c3eee424a9a : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":7},{"id":"f145d22635657ed5262d27db08ce56d46db6393d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-15 15:52:13.000000000","message":"Uploaded patch set 8.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"e584a18a1e3e16c044e72d8f61949b3e3ce130a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 16:44:33.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2f036dc8ff845ee9680a75cc06cad7e\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1e5cf14beb4043c49050f32a469f334f : SUCCESS in 9m 20s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/95145d1918144220b044c669d00f0791 : SUCCESS in 8m 07s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/1d06b81e19d34db3a9695069eb020e96 : SUCCESS in 14m 42s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"831aff0c2107124b6064b1a77dbcc6b6daad7dd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 18:44:20.000000000","message":"Patch Set 8: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2436afaa6fcc45838c05c95317b069e7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd3641ee00ee42a2bbf6a6c766abcc77 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44f21d02ad0043568497d51447e36340 : FAILURE in 6m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/794fed5a71ce4c3893c00efbc79e9a91 : SUCCESS in 3m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/65cf261d6f6a4d7a9b5c6bb6211772e6 : SUCCESS in 4m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3633259ed30d4397b21f9dd277b169bb : SUCCESS in 8m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40d34dc556294ac1b01bdfc875b81b6d : SUCCESS in 7m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f2d95d260a974415ae83879c5a6e06a6 : SUCCESS in 5m 59s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/06f5fdad11b646639ad2b263926594c5 : SUCCESS in 1h 14m 22s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/f02177ea4ab24bf9a0c4a31b95ab329b : SUCCESS in 2h 29m 48s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/0436b0f2eb30436a95574df9130c6112 : SUCCESS in 1h 50m 34s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/72550523a26c4518981fe30300bfda14 : SUCCESS in 1h 53m 48s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/8acc4f0bba834d48bcdabc0f1d6e68e4 : SUCCESS in 59m 47s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/0acf964d1c314263b30ac9d00f976c34 : SUCCESS in 2h 01m 49s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/4411be18253d48f49bda168f8449ca1b : SUCCESS in 1h 44m 55s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/e5bbc25ab1a34f5595c94d087008e81a : SUCCESS in 2h 18m 40s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/650f7d6296e048d9802a4a14d58b9506 : SUCCESS in 57m 25s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/0a341767a38a4f2da92fe6305cbe9690 : FAILURE in 2h 51m 04s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/9d135bb000d84893977e80c1926a6bcd : SUCCESS in 1h 34m 29s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/b5eafb1d620042cb9412171a5f8e3744 : SUCCESS in 1h 47m 57s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c519328082e9417aadee1e39f06fca24 : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":8},{"id":"0192fa9eec8e41539543e3bc27bcc2a99ff0b3c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-16 04:39:26.000000000","message":"Uploaded patch set 9.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"98073daf79a4f7ddb402564031b491b4bdbcd494","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-16 04:39:41.000000000","message":"Topic implement-userdata-refresh-on-rebuild removed","accounts_in_message":[],"_revision_number":9},{"id":"af1fbb80a14b63a22acbb4db56023d80b4ae057e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-16 04:59:22.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/335c013924e84d13ba83ce29f85e3217\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f3f3c270e3d34820a3240f3edfc39edc : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1e612cc15e7c454995dec66390b2a764 : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e128ed1cea3e437b8e8a540e775f5d4c : SUCCESS in 10m 19s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"6b5385b7c517cb02b515276ba68645454db28b85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-16 06:43:19.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/365df6a204fb4d5aad1087b28d857577\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5b6716f5c62f469db2fe5d743287a76f : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dccacac14dea47e594be93fdaf05ede7 : SUCCESS in 2m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/33412119ed4443f5b9515be4cd8226d0 : SUCCESS in 3m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ee9a8b9708814141a5b978bf91ab840d : SUCCESS in 3m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/24919bfed4004f0d8b93798d5ea8d312 : SUCCESS in 9m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1108d532114429f9f1938512452c60e : SUCCESS in 4m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4bf328db59e54e6cba0612921b62a261 : SUCCESS in 8m 55s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/4c38135f66064276badd580eccfcab10 : SUCCESS in 51m 53s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/bd63e8cd525f46eca213b3439e7e533f : SUCCESS in 1h 58m 49s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/38c28f45b7824ee5b63e929d006a8f78 : SUCCESS in 1h 31m 35s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/92731cbf1e0f4eaebfe22b127527c0c6 : SUCCESS in 2h 02m 44s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/b6951c504b794ff99cceac35beb7923f : SUCCESS in 42m 19s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/bf4b0c307c994427a2125f8326653df9 : SUCCESS in 1h 28m 43s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/941aba5041ab47e1bd353fbb0fa4740c : SUCCESS in 1h 11m 28s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/a2561950b6604892885ef108f9cbeccb : SUCCESS in 1h 38m 34s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/e5110a82cd6d40c494fbf9cbcbb80a4e : SUCCESS in 1h 04m 34s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/91fa3bed6e874c269b3bccc847df16ba : SUCCESS in 1h 31m 33s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/2a00dcd351e24865a05029fd72ecc7ed : SUCCESS in 1h 13m 11s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/cfe34247127a494aa02cbf4293542d65 : SUCCESS in 1h 18m 34s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fb72b6ff464b4ac08e93ef630768173d : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":9},{"id":"c11aea9e589ddda1165699282b0380b9c667f3c4","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-16 12:15:08.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":9},{"id":"ad079eca79572f87d769dab9949cd9942274379c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-16 13:02:11.000000000","message":"Uploaded patch set 10.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"e5dc6854151b7fbb9529ed720a5f2d47cbcc5a5b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-16 13:32:34.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa3bef6024ff44379a095ae07f7095ce\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2c4d83dd22c34634834ef36b8513201d : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a2024aaf3b484dc1a3a53cd4a4519baf : NODE_FAILURE Node(set) request 3b9eda0e29e1407c82c4bf75616c5b81 failed in 0s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5e11828b876f4f4baf405acc653e5ceb : NODE_FAILURE Node(set) request 961e2d9b8c624f21b24dcf469d2e19be failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"83c7500b82a1a0da6682a491770c78c6289415c3","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-16 15:05:49.000000000","message":"Patch Set 10: Backport-Candidate-1\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":10},{"id":"604e1f7d9a834da4cc8f16c1ba7f67f9d0d9cabd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-16 15:18:20.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1166fad1d97421e95f3d600c9311916\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/53067a43f8d64888abcd3160c786c414 : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b35ffbe7a714808babfd228bc6bdc1d : SUCCESS in 4m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f033155a306f47faa14600cc703687e6 : SUCCESS in 5m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bafb044c0cd64eaeb23957b6f0b2fdae : SUCCESS in 6m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/34a28e6a35664c0b87b305fe1f326a40 : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd7fd0578bee4d9497c270b7eb91cc1a : SUCCESS in 9m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac308f2d291944e4a072474ef897d618 : SUCCESS in 4m 47s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/0a2a13facf214c86a34f496f76225866 : SUCCESS in 2h 01m 00s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/e14f4e5434b74439a35b7c7f33eed3bd : SUCCESS in 2h 14m 18s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/63a2f7c799fc46139ca77ad5082164c3 : SUCCESS in 1h 34m 00s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/d8774040560d4271930ba454a2057b4b : SUCCESS in 2h 03m 12s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/e504146a37954b678ce71f0e47b540f8 : SUCCESS in 45m 56s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/c8abd6e2f5c04b938632bc0aad89d701 : SUCCESS in 1h 29m 53s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/e720a00f4ff248bfbc8960f868914097 : SUCCESS in 1h 36m 04s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/55f9a49844964ad1ac153c375399a973 : SUCCESS in 2h 14m 53s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/c8f88502dc4846e397b9aa8154b452a9 : SUCCESS in 1h 24m 04s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/03d6e42625bf41e2820ecddd3dab590e : SUCCESS in 1h 36m 51s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/cd598523bde3468da4c445ebdfa3a4e2 : SUCCESS in 1h 29m 41s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/5b006103564a45d795778f3e47bbb1f2 : SUCCESS in 1h 33m 39s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/21b1a0386e7f4f69be28386a60c7fd84 : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":10},{"id":"e609c3c6585b36b29a4b8edac7d395cd13661bd9","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-07-19 09:19:30.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"583ceb9671cde23b5cb54b578eac44c9137bf00e","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-19 14:36:22.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"2c11a90f4290450159240e75f89cc1e795e69330","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-19 14:36:45.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"4e703623b978b697631fad4461826b834b17a884","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-19 15:26:36.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7a16a218c8144898bf596ba2de7c1d0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a59fcd1e1a2944ecbe3fb1aee904c73e : SUCCESS in 4m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f5f8c99f004e4f30ae181f017be1c30b : SUCCESS in 4m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d888b226980e4f83b293b4b921135bda : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3204531521734db18c1bba9da0b894ab : SUCCESS in 9m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e5cb727af8c435eb8e5aba33f1338d5 : SUCCESS in 7m 18s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/c8f985214f0f4975a24447f9957e71bc : SUCCESS in 49m 26s\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/a9a52e671d9448dd904a82b7b1c12298 : SUCCESS in 40m 16s\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/8819fc38ae494433ae01452fdd15976d : SUCCESS in 44m 49s","accounts_in_message":[],"_revision_number":10},{"id":"6f2849bcca8d8e5ab6bdf1ce4d6f83edfa924ae4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-19 15:26:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"ab869b2f62fc22c14c78263e2ca068a1d58623ff","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-19 15:27:27.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9842fc1c75ea45c2919294bf8bd61b79\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34388d73692b48809ebe57e7eb7a2955 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1820f4211e8a45b2840449a7e1b6d5f4 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"088c7f9844b54fb286adcbb8c4881e0d89466f5a","revisions":{"9bf03ddb6093ff24c92e29f008f2f3e5455bad20":{"kind":"REWORK","_number":1,"created":"2026-04-24 13:46:50.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/90/986090/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/90/986090/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/986090/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/986090/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/986090/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/90/986090/1"}}},"commit":{"parents":[{"commit":"7a2cb21e8b3035835ae1260c9c7e676191360562","subject":"Improve instance list method by using Nova tags feature.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7a2cb21e8b3035835ae1260c9c7e676191360562"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-04-24 13:42:12.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-04-24 13:42:12.000000000","tz":300},"subject":"Enable userdata refresh on instance rebuild (Nova ≥ 2.57)","message":"Enable userdata refresh on instance rebuild (Nova ≥ 2.57)\n\nPreviously, Trove could not apply updated cloud-init settings during\ninstance rebuild because Nova did not support passing userdata in\nrebuild requests. As a result, any changes to Trove configuration or\ncustom cloud-init overrides were only applied at instance creation\ntime.\n\nThis change introduces conditional support for rebuilding instances\nwith regenerated userdata when Nova microversion\u003e\u003d2.57 is configured.\nFor older microversions, the existing behavior using injected files\nis preserved.\n\nAlso:\n- Bump default nova_client_version to 2.57\n- Document Nova microversion requirements and related features\n\nThis allows updated cloud-init configuration to be applied on\nrebuild, resolving issues where configuration changes could not be\npropagated to existing instances.\n\nChange-Id: I558bec8bae5db9c244f9ba6d87cefd7478f12cb1\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9bf03ddb6093ff24c92e29f008f2f3e5455bad20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9bf03ddb6093ff24c92e29f008f2f3e5455bad20"}]},"branch":"refs/heads/master"},"9458f3bf0440dad847763e53a8de1ccf7304287a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-04-24 13:59:31.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/90/986090/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/90/986090/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/986090/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/986090/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/986090/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/90/986090/2"}}},"commit":{"parents":[{"commit":"7a2cb21e8b3035835ae1260c9c7e676191360562","subject":"Improve instance list method by using Nova tags feature.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7a2cb21e8b3035835ae1260c9c7e676191360562"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-04-24 13:42:12.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-04-24 13:59:31.000000000","tz":0},"subject":"Enable userdata refresh on instance rebuild (Nova\u003e\u003d2.57)","message":"Enable userdata refresh on instance rebuild (Nova\u003e\u003d2.57)\n\nPreviously, Trove could not apply updated cloud-init settings during\ninstance rebuild because Nova did not support passing userdata in\nrebuild requests. As a result, any changes to Trove configuration or\ncustom cloud-init overrides were only applied at instance creation\ntime.\n\nThis change introduces conditional support for rebuilding instances\nwith regenerated userdata when Nova microversion\u003e\u003d2.57 is configured.\nFor older microversions, the existing behavior using injected files\nis preserved.\n\nAlso:\n- Bump default nova_client_version to 2.57\n- Document Nova microversion requirements and related features\n\nThis allows updated cloud-init configuration to be applied on\nrebuild, resolving issues where configuration changes could not be\npropagated to existing instances.\n\nChange-Id: I558bec8bae5db9c244f9ba6d87cefd7478f12cb1\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9458f3bf0440dad847763e53a8de1ccf7304287a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9458f3bf0440dad847763e53a8de1ccf7304287a"}]},"branch":"refs/heads/master","description":"Edit commit message"},"9c786f984250dc402afd065e95b7b1326f2241e2":{"kind":"NO_CHANGE","_number":3,"created":"2026-04-27 06:14:47.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/90/986090/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/90/986090/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/986090/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/986090/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/986090/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/90/986090/3"}}},"commit":{"parents":[{"commit":"573cef64893ba7f8b0f777b49c26c9ce206c186e","subject":"Revert \"Fix backup failure when caller uses an application credential\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/573cef64893ba7f8b0f777b49c26c9ce206c186e"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-04-24 13:42:12.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-04-27 06:14:47.000000000","tz":0},"subject":"Enable userdata refresh on instance rebuild (Nova\u003e\u003d2.57)","message":"Enable userdata refresh on instance rebuild (Nova\u003e\u003d2.57)\n\nPreviously, Trove could not apply updated cloud-init settings during\ninstance rebuild because Nova did not support passing userdata in\nrebuild requests. As a result, any changes to Trove configuration or\ncustom cloud-init overrides were only applied at instance creation\ntime.\n\nThis change introduces conditional support for rebuilding instances\nwith regenerated userdata when Nova microversion\u003e\u003d2.57 is configured.\nFor older microversions, the existing behavior using injected files\nis preserved.\n\nAlso:\n- Bump default nova_client_version to 2.57\n- Document Nova microversion requirements and related features\n\nThis allows updated cloud-init configuration to be applied on\nrebuild, resolving issues where configuration changes could not be\npropagated to existing instances.\n\nChange-Id: I558bec8bae5db9c244f9ba6d87cefd7478f12cb1\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9c786f984250dc402afd065e95b7b1326f2241e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9c786f984250dc402afd065e95b7b1326f2241e2"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"9458f3bf0440dad847763e53a8de1ccf7304287a","theirs":"573cef64893ba7f8b0f777b49c26c9ce206c186e","contains_conflicts":false}},"f7440d70c4262c23f692c10a466c8f049a50d2a2":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-06-24 08:56:43.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/90/986090/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/90/986090/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/986090/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/986090/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/986090/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/90/986090/4"}}},"commit":{"parents":[{"commit":"573cef64893ba7f8b0f777b49c26c9ce206c186e","subject":"Revert \"Fix backup failure when caller uses an application credential\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/573cef64893ba7f8b0f777b49c26c9ce206c186e"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-04-24 13:42:12.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-24 08:56:43.000000000","tz":0},"subject":"Enable userdata refresh on instance rebuild (Nova\u003e\u003d2.57)","message":"Enable userdata refresh on instance rebuild (Nova\u003e\u003d2.57)\n\nCloses-Bug: #2158070\nChange-Id: I558bec8bae5db9c244f9ba6d87cefd7478f12cb1\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f7440d70c4262c23f692c10a466c8f049a50d2a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f7440d70c4262c23f692c10a466c8f049a50d2a2"}]},"branch":"refs/heads/master","description":"Edit commit message","conflicts":{"ours":"9458f3bf0440dad847763e53a8de1ccf7304287a","theirs":"573cef64893ba7f8b0f777b49c26c9ce206c186e","contains_conflicts":false}},"4921f57395dec1eaa09e784803cbcdee29045047":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":5,"created":"2026-07-12 09:42:06.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/90/986090/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/90/986090/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/986090/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/986090/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/986090/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/90/986090/5"}}},"commit":{"parents":[{"commit":"e9489ce73e2415ed277eb51ab45394f9247acc30","subject":"Merge \"Improve data volume response in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e9489ce73e2415ed277eb51ab45394f9247acc30"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-04-24 13:42:12.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-07-12 09:37:48.000000000","tz":300},"subject":"Enable userdata refresh on instance rebuild (Nova ≥ 2.57)","message":"Enable userdata refresh on instance rebuild (Nova ≥ 2.57)\n\nPreviously, Trove could not apply updated cloud-init settings during\ninstance rebuild because Nova did not support passing userdata in\nrebuild requests. As a result, any changes to Trove configuration or\ncustom cloud-init overrides were only applied at instance creation\ntime.\n\nThis change introduces conditional support for rebuilding instances\nwith regenerated userdata when Nova microversion\u003e\u003d2.57 is configured.\nFor older microversions, the existing behavior using injected files\nis preserved.\n\nAlso:\n- Bump default nova_client_version to 2.57\n- Document Nova microversion requirements and related features\n\nThis allows updated cloud-init configuration to be applied on\nrebuild, resolving issues where configuration changes could not be\npropagated to existing instances.\n\nChange-Id: I558bec8bae5db9c244f9ba6d87cefd7478f12cb1\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4921f57395dec1eaa09e784803cbcdee29045047"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4921f57395dec1eaa09e784803cbcdee29045047"}]},"branch":"refs/heads/master"},"c962275e46c76b7a8cfd18b8c98714ec1e2dbe7d":{"kind":"REWORK","_number":6,"created":"2026-07-12 09:52:27.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/90/986090/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/90/986090/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/986090/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/986090/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/986090/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/90/986090/6"}}},"commit":{"parents":[{"commit":"e9489ce73e2415ed277eb51ab45394f9247acc30","subject":"Merge \"Improve data volume response in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e9489ce73e2415ed277eb51ab45394f9247acc30"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-04-24 13:42:12.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-07-12 09:51:52.000000000","tz":300},"subject":"Enable userdata refresh on instance rebuild (Nova ≥ 2.57)","message":"Enable userdata refresh on instance rebuild (Nova ≥ 2.57)\n\nPreviously, Trove could not apply updated cloud-init settings during\ninstance rebuild because Nova did not support passing userdata in\nrebuild requests. As a result, any changes to Trove configuration or\ncustom cloud-init overrides were only applied at instance creation\ntime.\n\nThis change introduces conditional support for rebuilding instances\nwith regenerated userdata when Nova microversion\u003e\u003d2.57 is configured.\nFor older microversions, the existing behavior using injected files\nis preserved.\n\nAlso:\n- Bump default nova_client_version to 2.57\n- Document Nova microversion requirements and related features\n\nThis allows updated cloud-init configuration to be applied on\nrebuild, resolving issues where configuration changes could not be\npropagated to existing instances.\n\nChange-Id: I558bec8bae5db9c244f9ba6d87cefd7478f12cb1\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c962275e46c76b7a8cfd18b8c98714ec1e2dbe7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c962275e46c76b7a8cfd18b8c98714ec1e2dbe7d"}]},"branch":"refs/heads/master"},"46516cb88d5f81d8130e75fae92baf47352dbda7":{"kind":"REWORK","_number":7,"created":"2026-07-12 11:48:21.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/90/986090/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/90/986090/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/986090/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/986090/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/986090/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/90/986090/7"}}},"commit":{"parents":[{"commit":"e9489ce73e2415ed277eb51ab45394f9247acc30","subject":"Merge \"Improve data volume response in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e9489ce73e2415ed277eb51ab45394f9247acc30"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-04-24 13:42:12.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-07-12 11:47:54.000000000","tz":300},"subject":"Enable userdata refresh on instance rebuild (Nova ≥ 2.57)","message":"Enable userdata refresh on instance rebuild (Nova ≥ 2.57)\n\nPreviously, Trove could not apply updated cloud-init settings during\ninstance rebuild because Nova did not support passing userdata in\nrebuild requests. As a result, any changes to Trove configuration or\ncustom cloud-init overrides were only applied at instance creation\ntime.\n\nThis change introduces conditional support for rebuilding instances\nwith regenerated userdata when Nova microversion\u003e\u003d2.57 is configured.\nFor older microversions, the existing behavior using injected files\nis preserved.\n\nAlso:\n- Bump default nova_client_version to 2.57\n- Document Nova microversion requirements and related features\n\nThis allows updated cloud-init configuration to be applied on\nrebuild, resolving issues where configuration changes could not be\npropagated to existing instances.\n\nChange-Id: I558bec8bae5db9c244f9ba6d87cefd7478f12cb1\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/46516cb88d5f81d8130e75fae92baf47352dbda7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/46516cb88d5f81d8130e75fae92baf47352dbda7"}]},"branch":"refs/heads/master"},"06b3ca8f73a9aa00ce85c0abbcf9bd310a378cf1":{"kind":"REWORK","_number":8,"created":"2026-07-15 15:52:13.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/90/986090/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/90/986090/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/986090/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/986090/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/986090/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/90/986090/8"}}},"commit":{"parents":[{"commit":"33c603ccba5f99eeaea7b824e21d78cd32fafeea","subject":"Delay string interpolation in logs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/33c603ccba5f99eeaea7b824e21d78cd32fafeea"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-04-24 13:42:12.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-07-15 15:32:58.000000000","tz":300},"subject":"Enable userdata refresh on instance rebuild (Nova ≥ 2.57)","message":"Enable userdata refresh on instance rebuild (Nova ≥ 2.57)\n\nTrove can now regenerate and apply updated user_data during instance\nrebuild when using Nova microversion 2.57 or later, allowing updated\ncloud-init configuration and injected files to be applied to rebuilt\ninstances.\n\nBlueprint: update-user-data-on-rebuild\nAssisted-by: Olessya Khussainova \u003cfreckled.cellofun@gmail.com\u003e\nAssisted-by: GPT-5.5\nChange-Id: I558bec8bae5db9c244f9ba6d87cefd7478f12cb1\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/06b3ca8f73a9aa00ce85c0abbcf9bd310a378cf1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/06b3ca8f73a9aa00ce85c0abbcf9bd310a378cf1"}]},"branch":"refs/heads/master"},"09fe226667397b76bdf0518c65bbec9a692958ce":{"kind":"REWORK","_number":9,"created":"2026-07-16 04:39:26.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/90/986090/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/90/986090/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/986090/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/986090/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/986090/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/90/986090/9"}}},"commit":{"parents":[{"commit":"33c603ccba5f99eeaea7b824e21d78cd32fafeea","subject":"Delay string interpolation in logs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/33c603ccba5f99eeaea7b824e21d78cd32fafeea"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-04-24 13:42:12.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-07-16 04:39:08.000000000","tz":300},"subject":"Enable userdata refresh on instance rebuild (Nova ≥ 2.57)","message":"Enable userdata refresh on instance rebuild (Nova ≥ 2.57)\n\nTrove can now regenerate and apply updated user_data during instance\nrebuild when using Nova microversion 2.57 or later, allowing updated\ncloud-init configuration and injected files to be applied to rebuilt\ninstances.\n\nBlueprint: update-user-data-on-rebuild\nAssisted-by: Olessya Khussainova \u003cfreckled.cellofun@gmail.com\u003e\nAssisted-by: GPT-5.5\nChange-Id: I558bec8bae5db9c244f9ba6d87cefd7478f12cb1\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/09fe226667397b76bdf0518c65bbec9a692958ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/09fe226667397b76bdf0518c65bbec9a692958ce"}]},"branch":"refs/heads/master"},"088c7f9844b54fb286adcbb8c4881e0d89466f5a":{"kind":"REWORK","_number":10,"created":"2026-07-16 13:02:11.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/90/986090/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/90/986090/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/986090/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/986090/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/986090/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/90/986090/10"}}},"commit":{"parents":[{"commit":"33c603ccba5f99eeaea7b824e21d78cd32fafeea","subject":"Delay string interpolation in logs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/33c603ccba5f99eeaea7b824e21d78cd32fafeea"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-04-24 13:42:12.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-07-16 12:56:44.000000000","tz":300},"subject":"Enable userdata refresh on instance rebuild (Nova ≥ 2.57)","message":"Enable userdata refresh on instance rebuild (Nova ≥ 2.57)\n\nTrove can now regenerate and apply updated user_data during instance\nrebuild when using Nova microversion 2.57 or later, allowing updated\ncloud-init configuration and injected files to be applied to rebuilt\ninstances.\n\nBlueprint: update-user-data-on-rebuild\nAssisted-by: Olessya Khussainova \u003cfreckled.cellofun@gmail.com\u003e\nAssisted-by: GPT-5.5\nChange-Id: I558bec8bae5db9c244f9ba6d87cefd7478f12cb1\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/088c7f9844b54fb286adcbb8c4881e0d89466f5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/088c7f9844b54fb286adcbb8c4881e0d89466f5a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
