)]}'
{"id":"openstack%2Ftrove~986091","triplet_id":"openstack%2Ftrove~master~I1e88f9720ab1892bfa27aa1db381b16c6d7708d5","project":"openstack/trove","branch":"master","topic":"fix-resize-flavor-action-missing-flavor-id","attention_set":{},"removed_from_attention_set":{"36080":{"account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"last_update":"2026-05-08 04:06:43.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2026-05-08 03:18:46.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"I1e88f9720ab1892bfa27aa1db381b16c6d7708d5","subject":"Handle Nova microversion \u003e\u003d2.47 in resize validation","status":"MERGED","created":"2026-04-24 14:00:12.000000000","updated":"2026-05-08 04:07:38.000000000","submitted":"2026-05-08 04:06:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"986091-fix-resize-flavor-action-missing-flavor-id","meta_rev_id":"9650d7e9fa5615f01e32a204c40787074499840e","_number":986091,"virtual_id_number":986091,"owner":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-08 04:06:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-05-08 03:18:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-05-08 03:18:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-24 14:04:49.000000000","updated_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2026-04-24 15:06:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-04-24 16:39:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9cdb576d40392a7c0741083b76e6ae33901e8a08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-04-24 14:00:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b1f968153a627f719c0c7635074cc2765055914b","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-04-24 14:04:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"084bf6a00aa48fce92ab423955b32c49e4fb6fce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-24 15:06:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee719c946f454910942861cf25684309\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/148aa779ac7f4fd5a07021241ce62afe : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3c4fbb59f99545ca9c9873be8b8985db : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/da54c1cdff1344bf94c54808ea4959b4 : SUCCESS in 10m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9e9feadd6c53c81d0eea755c4adfb91060a7ba04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-24 16:39:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6684103fb9c84ed8b9738b8fa12d7698\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ce67dc9ff714d169f3c395c9b22f34b : SUCCESS in 7m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6caaeb55a7a419bbee282724f2f083e : SUCCESS in 3m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/01c446de57914234bed1b803ea80fdf7 : SUCCESS in 4m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8257cdcc3cfb4a2eaacfb0a42927bc70 : SUCCESS in 3m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/374806106729442a8128b666fef7f7b4 : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df44a4c8beba438893628cc81ad3dc36 : SUCCESS in 5m 37s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/9cce793e072d42729b4006cb2653d479 : SUCCESS in 1h 04m 17s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/5ae1c3bbbe6f485885094bb7747a5e4c : FAILURE in 1h 05m 34s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/ee0c8e4d6f1044b9b0ab0f164b95489e : FAILURE in 1h 06m 32s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/e28720e9ad7b4d349635b1ad6cd22cb7 : SUCCESS in 2h 38m 23s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/e32559932fd0495fa302ca48588da8fc : SUCCESS in 1h 09m 07s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/6e0c5fafe6d040efab6b94aef1ad8afd : FAILURE in 1h 03m 00s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/fcd6024f26b845d29c5ce41a42c6cf08 : FAILURE in 1h 12m 55s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/c687f9ed092d46df9748776479c10bd0 : SUCCESS in 1h 31m 43s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/9b186a9789b8498191cc7ad7e27ff20b : SUCCESS in 54m 12s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/840a92cae7f441d5a9ab580269d7b583 : FAILURE in 1h 33m 47s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/0ce9eb5a03944947ac71a4aa076a2bff : FAILURE in 1h 03m 39s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/c4e1de28cfa74231878bcdd7b01f03ce : SUCCESS in 1h 21m 46s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2b8487058cad4b65ad2ace78b218d284 : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":1},{"id":"31f29c027bcf2903876e23230416d72782cd8bc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-04-27 06:14:14.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"294c4a457d61e4a6eb1cfa7935b2620e30d5f378","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 06:52:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88f7d04d6c714ea0ac81fb6e8b693233\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f6d8d9e91acf437f850c28d64d84a2c6 : SUCCESS in 7m 53s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b2e7538be936471491c6ab3074b9d7d8 : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/270d64cdaba64cf9be74f4da0ff99c75 : SUCCESS in 12m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"58d0accb467b955a6438d5692ab4cefade17cd95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 08:12:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e51015c214f4cdf95d6b4fe3657c317\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d62a8287b38140ddbb72814b9747adc7 : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6345433af05e4dafa12d81136b14c543 : SUCCESS in 5m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7509058b59c04803a4e2c6ea3c965d84 : SUCCESS in 3m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/72507c6bb2b14beab821ceb7cf224283 : SUCCESS in 4m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8b9c65c0af134d9a9fdf79c590e4f9dd : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77ee9070bbe64459a03d3a9456129519 : SUCCESS in 5m 22s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/6ef5da3a8b3042e6bcebd869ce3120b0 : SUCCESS in 47m 55s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/010f938863a140b2930a5dfeea81837a : SUCCESS in 1h 49m 03s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/87c91ff0c2454e49b0b59cd4391a462b : SUCCESS in 1h 27m 13s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/7ffe5d28aea549d9a5d5756b973b1d09 : SUCCESS in 1h 57m 36s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/35b586c56e78494d88c088f181ae5d6d : SUCCESS in 43m 59s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/5fd008a9624f42bcbd33fa7a18ed21f4 : SUCCESS in 1h 36m 07s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/aeaa9140583c47308df442dc39b19300 : SUCCESS in 1h 13m 23s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/73c9fe8277444ce2ba7083d8f9f74af3 : SUCCESS in 1h 51m 44s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/42778d0d1dd34599b42b2ccbc9002851 : SUCCESS in 56m 00s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/437f714a85f34ba2a79ad0ac4979e549 : SUCCESS in 1h 45m 18s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/d2e6fe800753400e85e7078a0b0276f4 : SUCCESS in 1h 25m 12s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/c79dd07ea333495fa614ce53362d0f89 : SUCCESS in 1h 34m 11s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f7c43961d8ae498fb18191095ad139d2 : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":2},{"id":"25b00c8a4d6c16eb36c72f3e2717c9cc29c1c324","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-05-08 03:18:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"20876c200b5f41e37f586467b42cd30d6da2ccb4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 03:19:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1a7b4a0035df00290cd8eb57fd510955d762174e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 04:06:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6841a33fcacb49c9bc1665a9dc5fceed\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1f78b7b266b4b839509939d7992301b : SUCCESS in 2m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/84f1038570624d96b86a38809cbd7cc0 : SUCCESS in 2m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4f344ced3c6c410186ee99cb046b7137 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df6e55a444c5445aa5613b6bb0468321 : SUCCESS in 5m 34s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/1ab818a6d0d447d68c3609e49e2fbbfa : SUCCESS in 47m 06s\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/9599940cba2240dfa4382c612af49ab6 : SUCCESS in 40m 51s\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/caea519fda8240c190f24d9d9110d808 : SUCCESS in 43m 36s","accounts_in_message":[],"_revision_number":2},{"id":"7f4e3f898c5817a8ee244d6a5537a6b10e0df3e8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 04:06:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9650d7e9fa5615f01e32a204c40787074499840e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 04:07:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/419c1146031f475a972d57a37944919e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f25a1799c6b3412ca4cae4569413572e : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"513cc4dc0295aaf5c144c14f18adde96dbb8f945","revisions":{"a285a71d7782be139e76239c434ced87d3a31095":{"kind":"REWORK","_number":1,"created":"2026-04-24 14:00:12.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/91/986091/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/91/986091/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/91/986091/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/91/986091/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/91/986091/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/91/986091/1"}}},"commit":{"parents":[{"commit":"7a2cb21e8b3035835ae1260c9c7e676191360562","subject":"Improve instance list method by using Nova tags feature.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7a2cb21e8b3035835ae1260c9c7e676191360562"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-04-24 13:59:05.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-04-24 13:59:05.000000000","tz":300},"subject":"Handle Nova microversion \u003e\u003d2.47 in resize validation","message":"Handle Nova microversion \u003e\u003d2.47 in resize validation\n\nNova microversion 2.47 removes the id field from the server flavor\nobject and introduces original_name. This change updates resize\nvalidation logic to support both formats.\n\nChange-Id: I1e88f9720ab1892bfa27aa1db381b16c6d7708d5\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a285a71d7782be139e76239c434ced87d3a31095"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a285a71d7782be139e76239c434ced87d3a31095"}]},"branch":"refs/heads/master"},"513cc4dc0295aaf5c144c14f18adde96dbb8f945":{"kind":"NO_CHANGE","_number":2,"created":"2026-04-27 06:14:14.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/91/986091/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/91/986091/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/91/986091/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/91/986091/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/91/986091/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/91/986091/2"}}},"commit":{"parents":[{"commit":"573cef64893ba7f8b0f777b49c26c9ce206c186e","subject":"Revert \"Fix backup failure when caller uses an application credential\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/573cef64893ba7f8b0f777b49c26c9ce206c186e"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-04-24 13:59:05.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-04-27 06:14:14.000000000","tz":0},"subject":"Handle Nova microversion \u003e\u003d2.47 in resize validation","message":"Handle Nova microversion \u003e\u003d2.47 in resize validation\n\nNova microversion 2.47 removes the id field from the server flavor\nobject and introduces original_name. This change updates resize\nvalidation logic to support both formats.\n\nChange-Id: I1e88f9720ab1892bfa27aa1db381b16c6d7708d5\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/513cc4dc0295aaf5c144c14f18adde96dbb8f945"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/513cc4dc0295aaf5c144c14f18adde96dbb8f945"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"a285a71d7782be139e76239c434ced87d3a31095","theirs":"573cef64893ba7f8b0f777b49c26c9ce206c186e","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
