)]}'
{"id":"openstack%2Ftrove~986547","triplet_id":"openstack%2Ftrove~master~I84fe241877ffd889a74171ae7f778e43315fa777","project":"openstack/trove","branch":"master","topic":"use-swift-tokens-instead-of-credentials-in-guestagent","attention_set":{},"removed_from_attention_set":{"36080":{"account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"last_update":"2026-06-18 09:55:25.000000000","reason":"Change was submitted"},"35986":{"account":{"_account_id":35986,"name":"Adrian Jarvis","email":"adrian.r.jarvis@gmail.com","username":"adrian_jarvis"},"last_update":"2026-06-16 21:54:48.000000000","reason":"\u003cGERRIT_ACCOUNT_35986\u003e replied on the change","reason_account":{"_account_id":35986,"name":"Adrian Jarvis","email":"adrian.r.jarvis@gmail.com","username":"adrian_jarvis"}},"28691":{"account":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"last_update":"2026-06-18 09:55:25.000000000","reason":"Change was submitted"},"31737":{"account":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"last_update":"2026-06-18 09:55:25.000000000","reason":"Change was submitted"},"39035":{"account":{"_account_id":39035,"name":"Kostiantyn Velychkovskyi","display_name":"K Velychkovskyi","email":"k.velychkovskyi@cloudfarm.tech","username":"velychkovsky","status":"CloudFarm Co-Founder / CTO"},"last_update":"2026-06-18 09:55:25.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2026-06-18 08:44:01.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"I84fe241877ffd889a74171ae7f778e43315fa777","subject":"Use pre-auth Swift client, drop Keystone re-auth","status":"MERGED","created":"2026-04-28 15:06:45.000000000","updated":"2026-06-18 09:56:29.000000000","submitted":"2026-06-18 09:55:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":34,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"986547-use-swift-tokens-instead-of-credentials-in-guestagent","meta_rev_id":"2ba81c17608472b3fcb6118eb2f993ef67e090c2","_number":986547,"virtual_id_number":986547,"owner":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":35986,"name":"Adrian Jarvis","email":"adrian.r.jarvis@gmail.com","username":"adrian_jarvis"},{"value":0,"_account_id":39035,"name":"Kostiantyn Velychkovskyi","display_name":"K Velychkovskyi","email":"k.velychkovskyi@cloudfarm.tech","username":"velychkovsky","status":"CloudFarm Co-Founder / CTO"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-18 09:55:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"recommended":{"_account_id":35986,"name":"Adrian Jarvis","email":"adrian.r.jarvis@gmail.com","username":"adrian_jarvis"},"all":[{"value":0,"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":1,"date":"2026-06-16 21:54:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35986,"name":"Adrian Jarvis","email":"adrian.r.jarvis@gmail.com","username":"adrian_jarvis"},{"value":0,"_account_id":39035,"name":"Kostiantyn Velychkovskyi","display_name":"K Velychkovskyi","email":"k.velychkovskyi@cloudfarm.tech","username":"velychkovsky","status":"CloudFarm Co-Founder / CTO"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-06-18 08:44:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":35986,"name":"Adrian Jarvis","email":"adrian.r.jarvis@gmail.com","username":"adrian_jarvis"},{"value":0,"_account_id":39035,"name":"Kostiantyn Velychkovskyi","display_name":"K Velychkovskyi","email":"k.velychkovskyi@cloudfarm.tech","username":"velychkovsky","status":"CloudFarm Co-Founder / CTO"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-06-18 08:44:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":-1,"date":"2026-06-16 08:14:20.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":35986,"name":"Adrian Jarvis","email":"adrian.r.jarvis@gmail.com","username":"adrian_jarvis"},{"value":0,"_account_id":39035,"name":"Kostiantyn Velychkovskyi","display_name":"K Velychkovskyi","email":"k.velychkovskyi@cloudfarm.tech","username":"velychkovsky","status":"CloudFarm Co-Founder / CTO"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":35986,"name":"Adrian Jarvis","email":"adrian.r.jarvis@gmail.com","username":"adrian_jarvis"},{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"_account_id":39035,"name":"Kostiantyn Velychkovskyi","display_name":"K Velychkovskyi","email":"k.velychkovskyi@cloudfarm.tech","username":"velychkovsky","status":"CloudFarm Co-Founder / CTO"}],"CC":[{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-28 15:57:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-04-28 17:30:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-29 02:58:25.000000000","updated_by":{"_account_id":39035,"name":"Kostiantyn Velychkovskyi","display_name":"K Velychkovskyi","email":"k.velychkovskyi@cloudfarm.tech","username":"velychkovsky","status":"CloudFarm Co-Founder / CTO"},"reviewer":{"_account_id":39035,"name":"Kostiantyn Velychkovskyi","display_name":"K Velychkovskyi","email":"k.velychkovskyi@cloudfarm.tech","username":"velychkovsky","status":"CloudFarm Co-Founder / CTO"},"state":"REVIEWER"},{"updated":"2026-04-30 01:38:48.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2026-04-30 03:36:50.000000000","updated_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"CC"},{"updated":"2026-04-30 03:36:50.000000000","updated_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"reviewer":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"state":"CC"},{"updated":"2026-06-10 05:14:56.000000000","updated_by":{"_account_id":35986,"name":"Adrian Jarvis","email":"adrian.r.jarvis@gmail.com","username":"adrian_jarvis"},"reviewer":{"_account_id":35986,"name":"Adrian Jarvis","email":"adrian.r.jarvis@gmail.com","username":"adrian_jarvis"},"state":"CC"},{"updated":"2026-06-16 02:53:15.000000000","updated_by":{"_account_id":35986,"name":"Adrian Jarvis","email":"adrian.r.jarvis@gmail.com","username":"adrian_jarvis"},"reviewer":{"_account_id":35986,"name":"Adrian Jarvis","email":"adrian.r.jarvis@gmail.com","username":"adrian_jarvis"},"state":"REVIEWER"}],"messages":[{"id":"a7785b4a9ae843c6abe32d4e0a847bfb08f5ac19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-04-28 15:06:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"553fdea19b34b8725cdddc7719de646a9f806149","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-04-28 15:14:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"656320e845ede08318e8e8f18ea6d72a35618a50","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 15:57:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af4a8daf87264601b20e0c9df4136d06\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0f13a52dc01f452f9671d78f77acd0ee : FAILURE in 7m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/79129a6def254003a1fa967ee0709963 : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/aed2d57617eb4d6b8ce7baece5c8e3d9 : SUCCESS in 14m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0775bff4bf8bd5caa0804a065029b2a8a0f19653","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 17:30:51.000000000","message":"Patch Set 2: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5199eac01663421fbafbf3e501ff8f99\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/21c68ebc031b4ec0ae4afb61a2f5af31 : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e49aaffc468e4f3ea4188b9f7f98e9ba : FAILURE in 5m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7304a5ddc2d343f1983ce999ba656a80 : FAILURE in 4m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/05b8e9ec678e44da9412f966e56b2ad9 : SUCCESS in 4m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/28cc78eef78b4438bc1b05737885c3f0 : SUCCESS in 8m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa074530d7584eebac8aa253a2f5df9a : SUCCESS in 3m 03s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/692237226d814c56ba0aff009afc5df4 : SUCCESS in 57m 20s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/622b8287555a4a1cbfc890470329b22b : SUCCESS in 2h 13m 23s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/6c5f6624f1eb4578a035241d8a961b2c : SUCCESS in 1h 46m 12s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/fcd7686999b94117a9020dbced68e6eb : SUCCESS in 2h 15m 32s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/25f4a4f872de47c58eccdb89a657f8ed : SUCCESS in 1h 25m 49s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/b24b29018d3c48b0aabb4d29b6a1f118 : SUCCESS in 1h 46m 23s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/b1b7bc9e92824cf18b732850bd24d1e9 : SUCCESS in 1h 20m 27s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/745c815bb617424b9b84c7ffab8be022 : SUCCESS in 1h 36m 40s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/2fb61af6e4f24d82bda99243ae5dae58 : SUCCESS in 53m 21s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/6f5aeea8e26644a2b06c0727591ddd09 : SUCCESS in 1h 43m 13s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/461840eb9b7148c7b79aa063b6af7fce : SUCCESS in 1h 31m 47s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/a6a5269982d4439da99a0d314fc150ef : SUCCESS in 1h 37m 30s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/84bee014ad6c402d864fdade5268e983 : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":2},{"id":"c3a322f0af959ca5d970050777c20820b17b32cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-04-28 19:08:44.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"a8a498af62a94eb09f958762ca4314619308d2fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-04-28 19:12:48.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"0f82f572490e18963e34fbe230545207b1754265","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-04-28 19:15:08.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4a3f84852099027392bd9396988d7ae7cba6e10a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 19:41:19.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/436018b547b747398bdd1aed4fef5e8e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5a1ff05ec1884073bb5844ce36dac4c8 : FAILURE in 10m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/15811a26dc114450a1d579e74f4d23bc : SUCCESS in 8m 30s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e1b3a1d5a39942cf92ec73c909d9fc19 : SUCCESS in 12m 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"87fe8f38f5f300eab4c76ed222bb46558f4b5e7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 21:03:05.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/97686c952e144620923debef0e7c7c65\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/314357af0aa64d85a1e030722528bb15 : SUCCESS in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74972fc5288447c7a68c79d224323062 : SUCCESS in 5m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d10012ea568947d09acef635dd6aa169 : FAILURE in 5m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1ca3edcebca6471da01121eee12303cb : SUCCESS in 4m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ff111f196cc34e659056f802f507fc05 : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25cc3b5efc33457da5379e4fe437b797 : SUCCESS in 6m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77eb5b39df2e44728c1e4addf2471bcb : SUCCESS in 6m 40s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/ce8f9e13340c4567bc510cc6fb643749 : SUCCESS in 47m 53s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/413ad79d496b49c4836c3ba2046815cc : SUCCESS in 1h 49m 04s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/908e5a0ae6a14b7db302bf49414a7909 : SUCCESS in 1h 26m 59s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/183fd814f8d4482cb33280db1a2d36be : SUCCESS in 1h 47m 26s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/c68162c7c5b8491bb0c344de2fd16c53 : SUCCESS in 42m 50s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/4864b24a9fbf4fefb8fde7088797e4bb : SUCCESS in 1h 37m 51s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/49ec9d4d77ad47d0877885a515ae338a : SUCCESS in 1h 16m 44s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/d685268b9e4241d0924fc9dc28c875d2 : SUCCESS in 1h 31m 09s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/5a43f58f065b463b8e21d10993b7e98f : SUCCESS in 49m 08s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/726029798a46446db4ee058b41760ec6 : SUCCESS in 1h 45m 18s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/383caea4c2814f5eaf8e14106c3d2127 : SUCCESS in 1h 11m 36s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/8023c6d617334dc799a96b59dcab109e : SUCCESS in 1h 22m 58s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b5ef268f033c45f286453b47d170d816 : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":4},{"id":"e111ff572e4de6ea9f3baa6c92e0105ada37a758","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-04-29 02:18:56.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"5f161c2aa879a01d137b14a52e5f54064c4762e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 05:09:46.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a7c10921c76481ca1cefa04ab993e28\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1394cf60fc8a43f7921c7421258036e0 : SUCCESS in 10m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8b50686090b844f5a1e42302048cb745 : SUCCESS in 7m 46s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d4689c5dadad412aae670fa01fbd8250 : SUCCESS in 14m 08s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"65109af8f07a77724e61552efb691ce3fd561741","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 05:11:32.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/affdfc149af24804a3ba7e204498c428\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8cf585bb8b0f48368f07e981573d064d : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b22a2a792d548099e18170b97c9effd : SUCCESS in 5m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f408d54e054e428c939674c8aa9ec61c : SUCCESS in 5m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6d55dbf30d66404d8ee9b8f43a1d7dcb : SUCCESS in 6m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/321d61ba7f5d4a69838e3ac3269610f5 : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b9c263ec4b3468eba523e1fc6aec133 : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e55e41493f184da6bd59226d0fb05c5a : SUCCESS in 4m 07s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/a6c6e639a26d4a58a9787bc61f4fae86 : SUCCESS in 51m 40s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/d9c651fa5f45466e8eface4e491d5202 : SUCCESS in 2h 02m 52s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/ba7b2e1a181342fea34d7be72c39044f : SUCCESS in 1h 48m 19s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/d45d6a45b9eb4edb88c2b46e423a0a9d : SUCCESS in 2h 27m 36s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/aac1646f812c45b8980d28fba063b5ed : SUCCESS in 1h 31m 51s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/884709672bd447f59173dc48a2c633a5 : SUCCESS in 1h 51m 12s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/5d6610802fee419888bf96a6b61d123f : SUCCESS in 1h 22m 31s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/61e56d1b969640e48a5fe477c974b106 : SUCCESS in 1h 29m 42s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/81a964e539a9413793e8f60366807e09 : SUCCESS in 1h 02m 54s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/dab7b7d82a8a4d028fb39476190b862d : SUCCESS in 1h 35m 00s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/c71325388d8e431e856b47794d4afa2a : SUCCESS in 1h 33m 01s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/3ef90fc3f5f840dd8954053ce9556b21 : SUCCESS in 1h 38m 16s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0ae9fd5bcff1401082aabe344748c5a5 : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":5},{"id":"ef6ffaf8d585f13fdddc299c3d862f9b1d8390ce","author":{"_account_id":39035,"name":"Kostiantyn Velychkovskyi","display_name":"K Velychkovskyi","email":"k.velychkovskyi@cloudfarm.tech","username":"velychkovsky","status":"CloudFarm Co-Founder / CTO"},"date":"2026-04-29 09:04:27.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"a07f88fae7a5263ac0623adb360bde2a584faacd","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-04-30 01:38:48.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"b3707f89a6ed9fd02571b9eb0220764f9fcc2ef5","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-04-30 02:42:26.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"fc4a45fed9343e4ebe4685840ab2e76243e4a843","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-04-30 03:36:50.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"06ce0ccf0a930130c3cdb78923ed0db1b3eb885a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-04-30 07:40:49.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"000676693bd184cfdd8e69d9435e2a82bf15e8f2","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-04-30 07:43:27.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"169efcb8b70b0bef8aa012370b9802999f855263","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 08:23:39.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c578b0d47e8496f8ec996236bf0b94e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4eff60e8f59c48209d29c3df7df9a30b : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/441fca0bd26043489b2fcfd273307bab : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/cfd78e19a74a429198cb88cb846a3e9c : SUCCESS in 9m 56s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"755c54c14fb96aaa0a32a02b11f90cc9ed630d0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 10:05:16.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/332822f63fa34e23ae0bd7ea7eb9da3d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/70d13de85bf7409f87ebe47fa357bb83 : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b9f0b78e30e4f7094e5baae5a9a6dd1 : SUCCESS in 5m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a019906bc43a4cf988acc8a454dc6777 : SUCCESS in 4m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cda602cbf6364b0bb27b064ea7c519e5 : SUCCESS in 5m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6867166db1ee41d88344041a30255765 : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc5fedf5963c451cb572f3fab35aa498 : SUCCESS in 3m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7d3ae85e67c474698a3371c19501b2a : SUCCESS in 5m 12s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/da45b4ff1d8549878f59510268c05fe9 : SUCCESS in 54m 47s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/37a9d6ac0e3246c6940b3c71ab0c774f : SUCCESS in 2h 08m 26s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/c41e02de0664419596e5759051f5abd9 : SUCCESS in 1h 43m 04s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/4a71f95c14584de49a4a1a1f5fb19824 : SUCCESS in 2h 23m 26s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/1693ba70b07e4d1d855487bc9a0d32c9 : SUCCESS in 1h 17m 12s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/c94ea8c06ad043489693cd513845f4a9 : SUCCESS in 1h 40m 33s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/91349b996b3745809b587c3f1798f8d9 : SUCCESS in 1h 56m 54s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/ffa956a9505843af8cb3553ec21ba786 : SUCCESS in 1h 33m 26s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/ce19600242b84a61a2ebf3ec4d8a07be : SUCCESS in 51m 58s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/b2a1b94f003a4b10bbabbef9cbda344d : SUCCESS in 1h 40m 10s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/29aab86eb3804e50bd3a53292d1190e2 : SUCCESS in 1h 19m 36s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/d62f01552b7a4aff8771c1e8b92d50d1 : SUCCESS in 1h 46m 40s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a78fa9fd18cb436abc8f38cb7d16135e : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":6},{"id":"3ce0c5e2fb9080d2d6c9c8c3c969c634d8dccde6","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2026-05-06 02:44:02.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e32bc6f523dec0d2ca50c80acce2cc9b56f68132","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-06 19:15:51.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b4d95d57f90015846408f644b94004c147ba8f74","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2026-05-09 02:20:17.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0a27ff57bfdaf49dc81bd1f3e3b76c8f79a6a69b","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-05-09 07:32:47.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0df7fbc9d7a1ddbd58b8396e99988f26c4951d1a","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-09 07:50:08.000000000","message":"Patch Set 6: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"5ba98d7ec355f22ada73c4d6cfc64fddd8dba896","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-11 13:59:51.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"66705536b1c29cbf797a6b4aaf4b25e58548ad98","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 14:15:01.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94799f1f9ba347a49afc8c1c4c6f6cd2\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e53d7a703c79430d96387f634a66a091 : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ea3802003d7f41418734181916a32fe8 : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/6184ce5c7c684dd4af3348e6d66effb2 : SUCCESS in 10m 05s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"694fca499fbb3bfc4edae3d86fba9cb7a9799571","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 15:49:37.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e15d3d065a743e99f0ca810232c79d8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6780d7213b744f42bc9a9a2a99c38ece : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f51a0e82ee5d4dde8d90c1d0b1a8d5b8 : SUCCESS in 5m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/64d1f2e57b7d429ea86413531bfa13fa : SUCCESS in 5m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6d621b5c3a3a4266add9d8d0cdd410cc : SUCCESS in 6m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b7d6d521ad4d445da429ce958aa69d21 : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1345acded8648d2840e6c396dbe415a : SUCCESS in 8m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/978474380880484faedbfda31d9a543e : SUCCESS in 8m 29s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/48e3b7907c1c499cae027a38d9ba13c6 : SUCCESS in 46m 02s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/27810b27d5c843259d83f80486692682 : SUCCESS in 1h 48m 08s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/5f0e740f9d38481aab330951b704941d : SUCCESS in 1h 30m 10s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/2be5ff3a4334440ebd87a0e925c5119c : SUCCESS in 1h 45m 35s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/620a7d11904d40099be33945939174e5 : SUCCESS in 41m 33s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/98f90e6eb2c74494af2069a579108ba3 : SUCCESS in 1h 37m 45s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/e421ee9197dc48f5a3d55df04a439a39 : SUCCESS in 1h 17m 26s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/c0b2448615b34919ad9431ca08861f12 : SUCCESS in 1h 27m 00s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/caa7912740a44de4b6a9087342de3c1c : SUCCESS in 45m 43s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/1911e3315c3c4a28b9064b846646bf09 : SUCCESS in 1h 25m 32s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/839e4b180b81415183e4ddced0ee87d1 : SUCCESS in 1h 12m 44s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/7c0196578caa4dcf9f6956491a9d0054 : SUCCESS in 1h 22m 42s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/18744d6f9ab940078f7485c39e827c5a : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":7},{"id":"4ab526272f5a6005313d04a22077f639bef2c820","author":{"_account_id":35986,"name":"Adrian Jarvis","email":"adrian.r.jarvis@gmail.com","username":"adrian_jarvis"},"date":"2026-06-10 05:14:56.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f03cde435f8c827120d727074815f230af58eade","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-06-10 08:47:31.000000000","message":"Patch Set 7: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":7},{"id":"811ba2db1b67a353ca612171cca3eecd0caff4a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-10 19:10:43.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"273a3426220dcf200aa707cf8d1c5a5c5148fd39","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 19:11:10.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b7e9cd668c3498c9e31e8db4438b9a6\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/22a8cd3a407b49fba427bc80e8a505ba : NODE_FAILURE Node(set) request f580c8c0cff5405d86af1588ff1d16fe failed in 0s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d5e59c98a5144aef8ef7abe4e2a64cfd : NODE_FAILURE Node(set) request 898be578059f4acfbec8a10ba4b940bf failed in 0s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e740551eb8e74386b7111457e14f60f0 : NODE_FAILURE Node(set) request a33551cbd515469da59dd68f1e993024 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"d4b6c88d1c6696efbd8daa6ebd8d777a56ac2ae3","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-10 19:12:03.000000000","message":"Patch Set 7: Backport-Candidate-1\n\nCopied votes on follow-up patch sets have been updated:\n* Backport-Candidate-1 has been copied to patch set 8 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\").\n\n(7 comments)","accounts_in_message":[],"_revision_number":7},{"id":"27788017daf73b03d363897c45f1e6bb4c1a7827","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 21:07:28.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32ca835c7ffa4c64b92f62738d0f1e64\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/de511426110e401d8e3bd79eaeaea14c : SUCCESS in 3m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8bb6cd8b689449c1bf8aaa89b644c5dd : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28f9c2c381724d6eb1b8323f6faad6ef : SUCCESS in 4m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4bdf191bf40d403bae9fafa3ed811656 : SUCCESS in 4m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/168fa2a7834b412c8bbc95b4af673d0a : SUCCESS in 3m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2dc08816fefd46c09ac94b166c9258b5 : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8542c0d96ef4431a82d668e422ac9cc4 : SUCCESS in 5m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed9aa1584c2e40caae246393d62594f7 : SUCCESS in 6m 58s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/8cbfe76dd96448adbaa93e7e0691d604 : SUCCESS in 53m 48s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/cd69dd0a04024d719ced5889f6883a84 : FAILURE in 1h 20m 41s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/5ad177530ef0476e8d1786f664dec247 : FAILURE in 1h 26m 35s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/aa54714d2ea745708cac5ad7bd0ca5c1 : SUCCESS in 1h 55m 44s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/981861f9bc2049c083194a12f4de7bd3 : SUCCESS in 56m 50s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/1409a86ace1547d7ba68703229be4e76 : FAILURE in 1h 13m 01s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/74ba9ac27e9e4eb398342bd8f6c3c413 : FAILURE in 1h 12m 59s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/bccf63a1f85342ae9f96c7a8655b785b : SUCCESS in 1h 38m 46s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/ef6e3ec412da4828ba406596e8c44219 : SUCCESS in 58m 20s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/dd0e4209fa624a6ab5a3f3360c49f571 : FAILURE in 1h 16m 58s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/dd9b8a3e9d3748d5b820078e88e54af2 : FAILURE in 1h 18m 57s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/6eae07911d324ec088357d2805da94fc : SUCCESS in 1h 29m 44s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ff8bc4ef24ff4b4c916a962554290f2c : SUCCESS in 6m 15s","accounts_in_message":[],"_revision_number":8},{"id":"dc258bf776da989353e81d444d58da6ed0b3725d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-11 05:25:28.000000000","message":"Uploaded patch set 9.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"86c3c0ee1906f8b46de17af3cda6e832d43a50d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 07:27:08.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f34c3685860346f0b3c54deb1e80784c\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c796fb9942064f7b886f720b06cfe123 : SUCCESS in 6m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f2d485a513b846ca8b7178da70c7417c : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3177034aaefc48dfa1e45ae9a81eca87 : SUCCESS in 6m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/af84c121ede54d65821a6c073d61711d : SUCCESS in 4m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/69fda625af41401d8d3a25b23545354e : SUCCESS in 4m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5948219983ca402480103f46c41e18dc : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/026385e5885848debe0e9b4e361fa8f4 : SUCCESS in 11m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9537953356254dcc9ad4b0323b6fc546 : SUCCESS in 4m 32s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/b13e6c119fcf4cf5b176abfee14d9dde : SUCCESS in 51m 19s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/55c3c26d02ae464398de4f14ee0db484 : SUCCESS in 2h 00m 16s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/522f57c321e440f9baa62623b86b66b2 : SUCCESS in 1h 41m 39s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/2ae879b58468499798e2c0ae9e093ea8 : SUCCESS in 1h 50m 41s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/62d45956674448bd81694ef72de165eb : SUCCESS in 46m 04s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/4ecdcc9b11934aca93fa20a757a27a2f : SUCCESS in 1h 37m 42s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/c4c0c4a6ce6d4a2fb9f6573acc42d863 : SUCCESS in 1h 15m 11s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/41d546f46e9c4f7e90b76ad37602cc00 : SUCCESS in 1h 27m 25s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/578186700f15441bb21a97e73b27fa10 : SUCCESS in 47m 03s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/d59d4c6ea4ad42d9be2a6d7b2db4478e : SUCCESS in 1h 29m 57s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/6c60ea00d6e944c791591d100573a82a : SUCCESS in 1h 18m 24s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/58b24b7a901d4333b7e89dafe05346d6 : SUCCESS in 1h 20m 43s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3705fac9526d44059605ad02f1af4b54 : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":9},{"id":"79f663bf844d979103eba0a9a7d176e7397fc7fb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 18:17:42.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f41688467fc472b883ca2909bf75027\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cbe76872701f4c9e9eb6a3d42c2c25d7 : SUCCESS in 7m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c1e76e7effdd41ff801eceb7313f3dab : SUCCESS in 8m 16s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ece0257af9224937849a1e984e0d8796 : SUCCESS in 14m 50s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"36adbc20eee7ec8397339d2c6183569516944f81","author":{"_account_id":35986,"name":"Adrian Jarvis","email":"adrian.r.jarvis@gmail.com","username":"adrian_jarvis"},"date":"2026-06-16 02:53:15.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"668a9f8584b980aeefda4e8c8769200817e91d92","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-06-16 03:23:51.000000000","message":"Patch Set 9: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"8bda22b84fdb8cb8009db4cd91331de07544aec8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-16 08:09:16.000000000","message":"Uploaded patch set 10.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR **is:MIN**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"8c0ebc88ce9da2bcff0e863a8aef70375df1db12","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-16 08:11:28.000000000","message":"Patch Set 10: Backport-Candidate-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"be4954e8c57cef1fe02642bb896dc89a7ec3b611","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-16 08:14:20.000000000","message":"Uploaded patch set 11.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR **is:MIN**\")\n","accounts_in_message":[],"_revision_number":11},{"id":"5fd771d470a3cc797e6be28f0f2a86d23b3435d9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 08:46:14.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16981a22911e4d359274fdd104f6c293\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6506dee4f6eb4b6b8aa3d18686d8647b : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/dc046c9cf6d546cbbdfc8fa5350bb30e : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2e062a3d01a74cea88cf939c89d36ca7 : SUCCESS in 12m 30s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"8ce7ec5209de8369e21a671bd352b6f64eb66d0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 10:13:08.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25642692b4aa49f4af5f8f570def551b\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d9a6547f490d4c9e8be6c08bd512a0eb : SUCCESS in 5m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2842fedf38cc4c2bafb26f75ee841e05 : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48ae8823fe3443baa1ffe6aa7d5ff5af : SUCCESS in 5m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/288ea1c437a74650843d4bcab309e26a : SUCCESS in 4m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7ead0bf183b44b47b5c3a32cc85ee47e : SUCCESS in 4m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/19a07c68f15f4d06995c19d847fb998b : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f190538cb9224faa8c73fb9058d361ff : SUCCESS in 9m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61b29f51cb5b4411a1d2be964bb3473c : SUCCESS in 7m 32s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/a53abee1ef474ffc9b07e3156774026b : SUCCESS in 55m 19s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/125b071b95a54a4c84713312125e6327 : FAILURE in 1h 15m 38s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/304b83414d7c4149a7a40f8b20e43345 : FAILURE in 1h 19m 04s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/aea75374790b43a796abf32b5cea0d45 : SUCCESS in 1h 56m 33s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/f5417c5d280a4287a339a0389715c84c : SUCCESS in 48m 20s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/feacf946b28343b1b3530d6e2cb21425 : FAILURE in 59m 55s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/c6264da8088e44e1b402de02120faef3 : FAILURE in 1h 04m 10s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/9fcfb5e9e28d476ca2f6a657a9b6bf63 : SUCCESS in 1h 34m 29s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/0edab71aed8e4a60ba5a1a63f4171737 : SUCCESS in 50m 49s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/e7a1afa76115416db49f452412b9ee4c : FAILURE in 1h 02m 01s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/44477db4cc034336a06c1b22ea6fc7de : FAILURE in 1h 04m 29s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/2d73475d4d4a4f4281e47e4335568ca1 : SUCCESS in 1h 34m 39s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7a40d3430e654331b6dc7bd35fcfdc8a : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":11},{"id":"eb56d94e888a18e8613a96d47048ca109e0805d7","author":{"_account_id":35986,"name":"Adrian Jarvis","email":"adrian.r.jarvis@gmail.com","username":"adrian_jarvis"},"date":"2026-06-16 21:54:48.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"9148d901d2dda807d7d64a0909d16450a2a82783","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-06-18 08:44:01.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"7a47064ff042b8b4ee78ca5f38ec7a3951cdbdd0","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-06-18 08:44:23.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"761b591179d587239e37134d74645698325d3758","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 08:44:27.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"99199892869d883ff7c74879031e47d66f020c69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 09:55:25.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0be0f1c5b0914d94b7574c66e421de84\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1a41cf059eaa490c80c92ab91bfb0712 : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e038d4d8a19f477c81789172cdd1ad7d : SUCCESS in 6m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2bb5785fe7214b53b849f503e88e0bd7 : SUCCESS in 4m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ee6356ad2d0b422abbbe7a41e66ad467 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2c6d49b40654d73b5b7aa0f47a6744a : SUCCESS in 10m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/78db230fa662471eb13623b632939a73 : SUCCESS in 7m 27s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/6e2ed425834d4483a133b0ceb6ee5b6c : SUCCESS in 1h 10m 18s\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/1948505fd02d40bda96024440736210a : SUCCESS in 1h 04m 41s\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/459ce36cfb224bd88b5c16da66af8ae9 : SUCCESS in 1h 04m 45s","accounts_in_message":[],"_revision_number":11},{"id":"811e28cdc2c4999b557b51d85dbef5da3c31212b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 09:55:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11},{"id":"2ba81c17608472b3fcb6118eb2f993ef67e090c2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 09:56:29.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/623b74a8b6734dd297e8001e0f41e9a0\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/023d162f57334c419cae24116301c9bf : SUCCESS in 47s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61b4419692634d5dab9b0dedf7bc1978 : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36a87d3c0d554e3cafa41745fc362495 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"804f960b1dccf61b9ec5c78fa88223077eb5fc0d","revisions":{"9797a6df0638a48c01801fa6c1d2d05f3bdb1093":{"kind":"REWORK","_number":1,"created":"2026-04-28 15:06:45.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/47/986547/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/47/986547/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/986547/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/986547/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/986547/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/47/986547/1"}}},"commit":{"parents":[{"commit":"573cef64893ba7f8b0f777b49c26c9ce206c186e","subject":"Revert \"Fix backup failure when caller uses an application credential\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/573cef64893ba7f8b0f777b49c26c9ce206c186e"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-04-28 15:04:21.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-04-28 15:04:21.000000000","tz":300},"subject":"[WIP] Use pre-auth Swift client, drop Keystone re-auth","message":"[WIP] Use pre-auth Swift client, drop Keystone re-auth\n\nStop using Keystone session and token re-scoping in backup/restore flows.\nPass Swift endpoint (swift_url) from taskmanager to guestagent and use\nswiftclient with preauthtoken directly in backup container.\n\nThis removes dependency on service credentials in trove-guestagent.conf.\n\nChange-Id: I84fe241877ffd889a74171ae7f778e43315fa777\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9797a6df0638a48c01801fa6c1d2d05f3bdb1093"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9797a6df0638a48c01801fa6c1d2d05f3bdb1093"}]},"branch":"refs/heads/master"},"1b5b9b2722721719d1f53db61337d98a425e51dc":{"kind":"REWORK","_number":2,"created":"2026-04-28 15:14:06.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/47/986547/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/47/986547/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/986547/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/986547/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/986547/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/47/986547/2"}}},"commit":{"parents":[{"commit":"573cef64893ba7f8b0f777b49c26c9ce206c186e","subject":"Revert \"Fix backup failure when caller uses an application credential\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/573cef64893ba7f8b0f777b49c26c9ce206c186e"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-04-28 15:04:21.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-04-28 15:13:54.000000000","tz":300},"subject":"[WIP] Use pre-auth Swift client, drop Keystone re-auth","message":"[WIP] Use pre-auth Swift client, drop Keystone re-auth\n\nStop using Keystone session and token re-scoping in backup/restore flows.\nPass Swift endpoint (swift_url) from taskmanager to guestagent and use\nswiftclient with preauthtoken directly in backup container.\n\nThis removes dependency on service credentials in trove-guestagent.conf.\n\nChange-Id: I84fe241877ffd889a74171ae7f778e43315fa777\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1b5b9b2722721719d1f53db61337d98a425e51dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1b5b9b2722721719d1f53db61337d98a425e51dc"}]},"branch":"refs/heads/master"},"afd5246a09542d94072f5d3607277cdfa0aeadbb":{"kind":"REWORK","_number":3,"created":"2026-04-28 19:08:44.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/47/986547/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/47/986547/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/986547/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/986547/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/986547/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/47/986547/3"}}},"commit":{"parents":[{"commit":"573cef64893ba7f8b0f777b49c26c9ce206c186e","subject":"Revert \"Fix backup failure when caller uses an application credential\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/573cef64893ba7f8b0f777b49c26c9ce206c186e"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-04-28 15:04:21.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-04-28 19:08:29.000000000","tz":300},"subject":"[WIP] Use pre-auth Swift client, drop Keystone re-auth","message":"[WIP] Use pre-auth Swift client, drop Keystone re-auth\n\nStop using Keystone session and token re-scoping in backup/restore flows.\nPass Swift endpoint (swift_url) from taskmanager to guestagent and use\nswiftclient with preauthtoken directly in backup container.\n\nThis removes dependency on service credentials in trove-guestagent.conf.\n\nChange-Id: I84fe241877ffd889a74171ae7f778e43315fa777\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/afd5246a09542d94072f5d3607277cdfa0aeadbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/afd5246a09542d94072f5d3607277cdfa0aeadbb"}]},"branch":"refs/heads/master"},"44710535eca0234b85f15652d5c7ae17a72e2dbb":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-04-28 19:12:48.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/47/986547/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/47/986547/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/986547/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/986547/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/986547/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/47/986547/4"}}},"commit":{"parents":[{"commit":"573cef64893ba7f8b0f777b49c26c9ce206c186e","subject":"Revert \"Fix backup failure when caller uses an application credential\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/573cef64893ba7f8b0f777b49c26c9ce206c186e"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-04-28 15:04:21.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-04-28 19:12:34.000000000","tz":300},"subject":"Use pre-auth Swift client, drop Keystone re-auth","message":"Use pre-auth Swift client, drop Keystone re-auth\n\nStop using Keystone session and token re-scoping in backup/restore flows.\nPass Swift endpoint (swift_url) from taskmanager to guestagent and use\nswiftclient with preauthtoken directly in backup container.\n\nThis removes dependency on service credentials in trove-guestagent.conf.\n\nChange-Id: I84fe241877ffd889a74171ae7f778e43315fa777\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/44710535eca0234b85f15652d5c7ae17a72e2dbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/44710535eca0234b85f15652d5c7ae17a72e2dbb"}]},"branch":"refs/heads/master"},"e56f80a158e027803ae61f29feb0da168ab45fcd":{"kind":"REWORK","_number":5,"created":"2026-04-29 02:18:56.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/47/986547/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/47/986547/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/986547/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/986547/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/986547/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/47/986547/5"}}},"commit":{"parents":[{"commit":"573cef64893ba7f8b0f777b49c26c9ce206c186e","subject":"Revert \"Fix backup failure when caller uses an application credential\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/573cef64893ba7f8b0f777b49c26c9ce206c186e"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-04-28 15:04:21.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-04-29 02:18:44.000000000","tz":300},"subject":"Use pre-auth Swift client, drop Keystone re-auth","message":"Use pre-auth Swift client, drop Keystone re-auth\n\nStop using Keystone session and token re-scoping in backup/restore flows.\nPass Swift endpoint (swift_url) from taskmanager to guestagent and use\nswiftclient with preauthtoken directly in backup container.\n\nThis removes dependency on service credentials in trove-guestagent.conf.\n\nChange-Id: I84fe241877ffd889a74171ae7f778e43315fa777\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e56f80a158e027803ae61f29feb0da168ab45fcd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e56f80a158e027803ae61f29feb0da168ab45fcd"}]},"branch":"refs/heads/master"},"2d64d5d187f4dc48d268cfbf1022dbf6b5099f30":{"kind":"REWORK","_number":6,"created":"2026-04-30 07:40:49.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/47/986547/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/47/986547/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/986547/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/986547/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/986547/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/47/986547/6"}}},"commit":{"parents":[{"commit":"573cef64893ba7f8b0f777b49c26c9ce206c186e","subject":"Revert \"Fix backup failure when caller uses an application credential\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/573cef64893ba7f8b0f777b49c26c9ce206c186e"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-04-28 15:04:21.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-04-30 07:40:36.000000000","tz":300},"subject":"Use pre-auth Swift client, drop Keystone re-auth","message":"Use pre-auth Swift client, drop Keystone re-auth\n\nStop using Keystone session and token re-scoping in backup/restore flows.\nPass Swift endpoint (swift_url) from taskmanager to guestagent and use\nswiftclient with preauthtoken directly in backup container.\n\nThis removes dependency on service credentials in trove-guestagent.conf.\n\nChange-Id: I84fe241877ffd889a74171ae7f778e43315fa777\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2d64d5d187f4dc48d268cfbf1022dbf6b5099f30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2d64d5d187f4dc48d268cfbf1022dbf6b5099f30"}]},"branch":"refs/heads/master"},"9e37055d79b8bd24ccd7321795871166d86d7372":{"kind":"REWORK","_number":7,"created":"2026-05-11 13:59:51.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/47/986547/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/47/986547/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/986547/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/986547/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/986547/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/47/986547/7"}}},"commit":{"parents":[{"commit":"01d94858288755b725293d19ece02093b3591337","subject":"Merge \"Add PDF building\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/01d94858288755b725293d19ece02093b3591337"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-04-28 15:04:21.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-11 13:59:30.000000000","tz":300},"subject":"Use pre-auth Swift client, drop Keystone re-auth","message":"Use pre-auth Swift client, drop Keystone re-auth\n\nStop using Keystone session and token re-scoping in backup/restore flows.\nPass Swift endpoint (swift_url) from taskmanager to guestagent and use\nswiftclient with preauthtoken directly in backup container.\n\nThis removes dependency on service credentials in trove-guestagent.conf.\n\nChange-Id: I84fe241877ffd889a74171ae7f778e43315fa777\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9e37055d79b8bd24ccd7321795871166d86d7372"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9e37055d79b8bd24ccd7321795871166d86d7372"}]},"branch":"refs/heads/master"},"62bf1bf575c3729bd1c08b1c8bedafb610d85b50":{"kind":"REWORK","_number":8,"created":"2026-06-10 19:10:43.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/47/986547/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/47/986547/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/986547/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/986547/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/986547/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/47/986547/8"}}},"commit":{"parents":[{"commit":"187af7eb8e32f58bcd4b2b7825dbd3e3272b2a86","subject":"Merge \"Use common utility to clone/install source\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/187af7eb8e32f58bcd4b2b7825dbd3e3272b2a86"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-04-28 15:04:21.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-10 19:08:26.000000000","tz":300},"subject":"Use pre-auth Swift client, drop Keystone re-auth","message":"Use pre-auth Swift client, drop Keystone re-auth\n\nStop using Keystone session and token re-scoping in backup/restore flows.\nPass Swift endpoint (swift_url) from taskmanager to guestagent and use\nswiftclient with preauthtoken directly in backup container.\n\nThis removes dependency on service credentials in trove-guestagent.conf\nand resolves rescope error: \"401 Application credentials cannot request\na scope\".\n\nChange-Id: I84fe241877ffd889a74171ae7f778e43315fa777\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\nCo-authored-by: Adrian Jarvis \u003cadrian.r.jarvis@gmail.com\u003e\nCo-authored-by: K Velychkovskyi \u003ck.velychkovskyi@cloudfarm.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/62bf1bf575c3729bd1c08b1c8bedafb610d85b50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/62bf1bf575c3729bd1c08b1c8bedafb610d85b50"}]},"branch":"refs/heads/master"},"14f76970d41ae762d44243efa2ddf4ccd7916d13":{"kind":"REWORK","_number":9,"created":"2026-06-11 05:25:28.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/47/986547/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/47/986547/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/986547/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/986547/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/986547/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/47/986547/9"}}},"commit":{"parents":[{"commit":"187af7eb8e32f58bcd4b2b7825dbd3e3272b2a86","subject":"Merge \"Use common utility to clone/install source\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/187af7eb8e32f58bcd4b2b7825dbd3e3272b2a86"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-04-28 15:04:21.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-11 05:25:18.000000000","tz":300},"subject":"Use pre-auth Swift client, drop Keystone re-auth","message":"Use pre-auth Swift client, drop Keystone re-auth\n\nStop using Keystone session and token re-scoping in backup/restore flows.\nPass Swift endpoint (swift_url) from taskmanager to guestagent and use\nswiftclient with preauthtoken directly in backup container.\n\nThis removes dependency on service credentials in trove-guestagent.conf\nand resolves rescope error: \"401 Application credentials cannot request\na scope\".\n\nChange-Id: I84fe241877ffd889a74171ae7f778e43315fa777\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\nCo-authored-by: Adrian Jarvis \u003cadrian.r.jarvis@gmail.com\u003e\nCo-authored-by: K Velychkovskyi \u003ck.velychkovskyi@cloudfarm.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/14f76970d41ae762d44243efa2ddf4ccd7916d13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/14f76970d41ae762d44243efa2ddf4ccd7916d13"}]},"branch":"refs/heads/master"},"9309cfbbb4d8dc03c7b218e6689684d358375eb7":{"kind":"REWORK","_number":10,"created":"2026-06-16 08:09:16.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/47/986547/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/47/986547/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/986547/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/986547/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/986547/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/47/986547/10"}}},"commit":{"parents":[{"commit":"187af7eb8e32f58bcd4b2b7825dbd3e3272b2a86","subject":"Merge \"Use common utility to clone/install source\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/187af7eb8e32f58bcd4b2b7825dbd3e3272b2a86"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-16 07:59:09.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-16 08:08:37.000000000","tz":300},"subject":"Use pre-auth Swift client, drop Keystone re-auth","message":"Use pre-auth Swift client, drop Keystone re-auth\n\nStop using Keystone session and token re-scoping in backup/restore flows.\nPass Swift endpoint (swift_url) from taskmanager to guestagent and use\nswiftclient with preauthtoken directly in backup container.\n\nChange-Id: I84fe241877ffd889a74171ae7f778e43315fa777\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\nCo-authored-by: Adrian Jarvis \u003cadrian.r.jarvis@gmail.com\u003e\nCo-authored-by: K Velychkovskyi \u003ck.velychkovskyi@cloudfarm.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9309cfbbb4d8dc03c7b218e6689684d358375eb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9309cfbbb4d8dc03c7b218e6689684d358375eb7"}]},"branch":"refs/heads/master"},"804f960b1dccf61b9ec5c78fa88223077eb5fc0d":{"kind":"REWORK","_number":11,"created":"2026-06-16 08:14:20.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/47/986547/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/47/986547/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/986547/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/986547/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/986547/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/47/986547/11"}}},"commit":{"parents":[{"commit":"187af7eb8e32f58bcd4b2b7825dbd3e3272b2a86","subject":"Merge \"Use common utility to clone/install source\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/187af7eb8e32f58bcd4b2b7825dbd3e3272b2a86"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-16 07:59:09.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-16 08:14:10.000000000","tz":300},"subject":"Use pre-auth Swift client, drop Keystone re-auth","message":"Use pre-auth Swift client, drop Keystone re-auth\n\nStop using Keystone session and token re-scoping in backup/restore flows.\nPass Swift endpoint (swift_url) from taskmanager to guestagent and use\nswiftclient with preauthtoken directly in backup container.\n\nChange-Id: I84fe241877ffd889a74171ae7f778e43315fa777\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\nCo-authored-by: Adrian Jarvis \u003cadrian.r.jarvis@gmail.com\u003e\nCo-authored-by: K Velychkovskyi \u003ck.velychkovskyi@cloudfarm.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/804f960b1dccf61b9ec5c78fa88223077eb5fc0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/804f960b1dccf61b9ec5c78fa88223077eb5fc0d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
