)]}'
{"id":"openstack%2Ftrove~986685","triplet_id":"openstack%2Ftrove~master~I55217fda1dfd5541b27859480e29262278f3b4e2","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-04-30 02:59:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I55217fda1dfd5541b27859480e29262278f3b4e2","subject":"zuul: Use ansible variable to configure tempest plugins","status":"MERGED","created":"2026-04-29 15:50:39.000000000","updated":"2026-04-30 03:01:54.000000000","submitted":"2026-04-30 02:59:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"986685","meta_rev_id":"9ea98960a217eca87354dfb39ea4b58cb7340d50","_number":986685,"virtual_id_number":986685,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-30 02:59:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-04-30 01:40:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-04-30 01:40:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-29 19:51:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-04-29 19:53:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-30 01:40:08.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"a56a6c74787a198a6ff9bcbde2bb96c38de50c81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-29 15:50:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fff53f8969b33bc1f77e72c347ef876cdd95cdcc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-29 15:51:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c641370f4db4e682f31a2d1177a8b31bc127bc29","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 19:51:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e64b9418780b470a9111203f83e2b0a4\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bfe2ba712062433486045405a36299c4 : SUCCESS in 10m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/acffeb5fd168453094dc8292ecae49c5 : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a911cb584d284132a7cd091a4df46287 : SUCCESS in 8m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0f1031e1ed70ca633c4e32f4aef7efef3b42a62c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 19:53:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0648ed5391cb4a76b32e8543366b40e4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/378e8127b5e448c092417281b99697b3 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e4dadd6e4584776b754315401f5d3fb : SUCCESS in 4m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f13f0629ee224511a779625881834b1c : SUCCESS in 4m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4a237e3cdfa647718b28e9c25b7110d0 : SUCCESS in 5m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f96043fcf43c4e1e80108d6b508babac : SUCCESS in 7m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/194633ff79354654a146c48c74106956 : SUCCESS in 4m 00s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/08309e9c78db4f86ac66a1a685c57662 : SUCCESS in 49m 55s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/d79e98223e1e41e995bfceaa88a724ca : SUCCESS in 1h 49m 04s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/b01b65eadb8b4fe2b7b0211db166910a : SUCCESS in 1h 27m 54s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/c4427c94bf0b44fe84ac4349e33ff955 : SUCCESS in 1h 59m 36s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/f521271e64b24e9f8353506ab90fa30e : SUCCESS in 47m 47s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/12dbb1d18f0845f6a03b727e9452113b : SUCCESS in 1h 48m 50s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/7ce44d916be04c5497237d85c87778fa : SUCCESS in 1h 17m 51s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/66873b444a4d431796b1041faee3c0db : SUCCESS in 1h 25m 42s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/b8d7c67103464d8d9444ae64826c924a : SUCCESS in 1h 01m 48s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/d72539efb9d249dcb2a112cde0406aed : SUCCESS in 1h 43m 32s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/799a485eebc94c76ab212911e4be24df : SUCCESS in 1h 17m 03s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/9f5650e90d1b46ffb8623537746c05d1 : SUCCESS in 1h 23m 58s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3215a8cd8dd94ccbb35bfd92f6c4f211 : SUCCESS in 5m 43s","accounts_in_message":[],"_revision_number":2},{"id":"7c346bfbeeae618d1de3155858602cc517af744e","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-04-30 01:40:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"974f92540701ffe32a2d64191b42901c4f02768c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 01:40:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"23c0ee38607c4c7d50cd2675beb6d94d36514d2b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 02:59:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e47f895c2f44bd5ba59f6c2960b4a06\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a14e2ff3c614a9fbb9bb29400830871 : SUCCESS in 5m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/67faefceecbb40c9ad97fa0fe0f106c6 : SUCCESS in 4m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2bd7fd09a0324bcd98ffea731d67e1fd : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33cac3479f904e2d9aca95cc4adff4ce : SUCCESS in 5m 21s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/caf7b72601b04f4aaa9fcd937c416218 : SUCCESS in 1h 05m 37s\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/4970860093524d4a8b21cd868f74ade6 : SUCCESS in 1h 18m 01s\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/a0cc7f4690b144c28018ae5484eb0b2d : SUCCESS in 1h 06m 12s","accounts_in_message":[],"_revision_number":2},{"id":"6a082a1d58d4017dd8dee708b8d29cd10aa3c98d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 02:59:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9ea98960a217eca87354dfb39ea4b58cb7340d50","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 03:01:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/780b9fd95fc34340a8dcc3ac4ca55fcd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8962cee0a1a4c368a78aa63373dcd92 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"11d7e61ddfa3f06b8fea25b2814ee1c229dd2e63","revisions":{"3b084e40102ef2529559bcd866929eb763b7099b":{"kind":"REWORK","_number":1,"created":"2026-04-29 15:50:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/85/986685/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/85/986685/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/85/986685/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/85/986685/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/85/986685/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/85/986685/1"}}},"commit":{"parents":[{"commit":"7a2cb21e8b3035835ae1260c9c7e676191360562","subject":"Improve instance list method by using Nova tags feature.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7a2cb21e8b3035835ae1260c9c7e676191360562"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-29 15:50:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-29 15:50:24.000000000","tz":540},"subject":"zuul: Use ansible variable to configure tempest plugins","message":"zuul: Use ansible variable to configure tempest plugins\n\nThe variable was introduced 6+ years ago[1].\n\n[1] https://review.opendev.org/c/openstack/devstack/+/639374\n\nChange-Id: I55217fda1dfd5541b27859480e29262278f3b4e2\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3b084e40102ef2529559bcd866929eb763b7099b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3b084e40102ef2529559bcd866929eb763b7099b"}]},"branch":"refs/heads/master"},"11d7e61ddfa3f06b8fea25b2814ee1c229dd2e63":{"kind":"REWORK","_number":2,"created":"2026-04-29 15:51:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/85/986685/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/85/986685/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/85/986685/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/85/986685/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/85/986685/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/85/986685/2"}}},"commit":{"parents":[{"commit":"7a2cb21e8b3035835ae1260c9c7e676191360562","subject":"Improve instance list method by using Nova tags feature.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7a2cb21e8b3035835ae1260c9c7e676191360562"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-29 15:50:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-29 15:50:58.000000000","tz":540},"subject":"zuul: Use ansible variable to configure tempest plugins","message":"zuul: Use ansible variable to configure tempest plugins\n\nThe variable was introduced 6+ years ago[1].\n\n[1] https://review.opendev.org/c/openstack/devstack/+/639374\n\nChange-Id: I55217fda1dfd5541b27859480e29262278f3b4e2\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/11d7e61ddfa3f06b8fea25b2814ee1c229dd2e63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/11d7e61ddfa3f06b8fea25b2814ee1c229dd2e63"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
