)]}'
{"id":"openstack%2Ftrove~988059","triplet_id":"openstack%2Ftrove~master~I93b9647c30cfcc7823e7d293652816f9040ee201","project":"openstack/trove","branch":"master","topic":"ssl-feature","attention_set":{},"removed_from_attention_set":{"36080":{"account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"last_update":"2026-07-20 14:19:36.000000000","reason":"\u003cGERRIT_ACCOUNT_36080\u003e replied on the change","reason_account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2026-07-19 08:57:31.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"I93b9647c30cfcc7823e7d293652816f9040ee201","subject":"SSL management feature","status":"MERGED","created":"2026-05-11 09:07:13.000000000","updated":"2026-07-20 17:28:19.000000000","submitted":"2026-07-20 17:26:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":46,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"988059-ssl-feature","meta_rev_id":"4fbc5a9d3da1ea5547a558d52e1495ce160f0ade","_number":988059,"virtual_id_number":988059,"owner":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-20 17:26:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":2,"date":"2026-07-19 14:29:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-07-19 08:57:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":1,"date":"2026-07-19 14:29:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":-1,"date":"2026-07-19 14:29:40.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}],"CC":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-11 09:44:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-11 10:56:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-24 19:58:40.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-05-24 19:58:40.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"652d1be8cd27f540789380339f2b3390f35d0b31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-11 09:07:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f87a22cd008f3da7660ee40fae1178549884eefc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-11 09:09:21.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"f00431ce1e817c790cd72c80d26b52d37c747d34","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 09:44:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ca2a74d7bdd46faac7de67199b95457\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/40a031fa310d4bcea4cf31f44cea118a : SUCCESS in 7m 53s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b4d768385d1c4be1a2b1e568f1dd67fd : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7d0ddcf8c9624255ba0060d03aef1a8e : SUCCESS in 9m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"22607e6bebf02db25e5d1f3415d80c83bf3dd865","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 10:56:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5e986c2c1124a939ea9f0856762267f\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7bcc9cd79f6248a28a0a1975acbfd035 : SUCCESS in 5m 15s\n- requirements-check https://zuul.opendev.org/t/openstack/build/022b4844e39f43d1bd556a592c34a475 : SUCCESS in 3m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/94637ff9230e4600b0bfa7076c169035 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efd6b8742b4b4a889098e9724bd68152 : SUCCESS in 5m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b1a509d6c41640c09f0f096bc4768fc3 : SUCCESS in 4m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0e6e547af4fa4bedb168e08bd7027031 : SUCCESS in 3m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dbfc819c277f493e98fc7387914aa34f : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e5f8957ea3f43c9ab0507013a8f0034 : SUCCESS in 9m 57s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/0a8cf3ecf41842c094a871e589b6a3cb : SUCCESS in 44m 46s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/77642138fe994df8b9263c40af0e77a1 : SUCCESS in 1h 46m 07s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/a81287ba838e4d1282f8bcd8968f644a : SUCCESS in 1h 42m 15s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/e7c9d60f35244596a5307d68fb5b1e7e : FAILURE in 1h 17m 46s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/5fcfc6a1ba88466cbed7f447e01dcfa6 : SUCCESS in 49m 46s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/538661c5c18045899015106d95731172 : SUCCESS in 1h 33m 49s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/99811edd46ab4cf48b75bc0ea958a709 : SUCCESS in 1h 16m 11s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/c04689988209423eae3e2b555c65a3ff : SUCCESS in 1h 31m 39s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/0e4589d85fa04980a84f1b6072f850bc : SUCCESS in 43m 51s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/c1e983a770e8404b954f6b28151d05d0 : SUCCESS in 1h 28m 56s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/723782cb83624c48a4046f9b21eda965 : SUCCESS in 1h 11m 07s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/aa2a32a4f8f44e319206bc9120e72c2c : SUCCESS in 1h 26m 58s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3e9b9d3692c7434090443650da489c21 : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":2},{"id":"e811d06ffce1bda8a35adbf67d4093698f3ed58d","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-11 12:18:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3bd2372017591bdc121ddf18f9014cd9b3f5087b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-11 12:51:50.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"661a46820f235e5c92e7f5bff9612451a1c92a89","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 13:06:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b8f5c7c3ba6425ea5a40d33467a72ac\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f19c20634f57498e8e26d81e324e9a34 : SUCCESS in 8m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/aff9050d7e6b43f08190f3f569065e69 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/08e85e3c178c4debabcb25093b0e6857 : SUCCESS in 12m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5c6e747002aa46cf05aebb48a29f32ad12c94a13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 14:41:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84e5201cc2344f1ca941b26be705e73a\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/15cc92da3d274c0a94463d73a13f864b : SUCCESS in 6m 23s\n- requirements-check https://zuul.opendev.org/t/openstack/build/deddcf474f9140e4a601ed3cc3080d15 : SUCCESS in 3m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/70e402345620487482d91415f858b227 : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d926084e40b04e0099ddd12168c13915 : SUCCESS in 5m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eaea60792f884916b5eabc5f01bdfbaa : SUCCESS in 3m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cdcb4402601d4d39a88a01b880c9141b : SUCCESS in 4m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3a8f06735ee04d9f9c27deeb8bbe1f33 : SUCCESS in 8m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4311835a39c54f3fb76ee5c8ab3fcb47 : SUCCESS in 9m 49s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/081d6250842e405fa919ff254b7c0f19 : SUCCESS in 46m 57s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/6fa8fdf84b7c44499b452a763f7ee131 : SUCCESS in 1h 44m 21s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/0c4a001ae4454120b804e9e9cd42d6e4 : SUCCESS in 1h 27m 56s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/b836ea0f67304f4197597e4e51b49f73 : SUCCESS in 1h 46m 25s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/2235a698d41c424d86cd7ddf4f230379 : SUCCESS in 43m 09s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/8d8b0cafe4414d1f9fe0802ddfe30200 : SUCCESS in 1h 31m 48s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/b53b34cf646a4effaf5c38ab56098189 : SUCCESS in 1h 20m 38s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/0c2659d9731c4504b55a08f776122713 : SUCCESS in 1h 41m 04s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/d07c25b66d6449399ee15e20e106959c : SUCCESS in 44m 08s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/07b7f4bff2bb4119ba439feb301e6d9f : SUCCESS in 1h 34m 23s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/811ae0829a51448988a41fe46a1a33cf : SUCCESS in 1h 11m 45s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/d7067719425f4c0f82ee4d5eb21149ff : SUCCESS in 1h 22m 21s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9f6dfcd1322844f2abcfcb9ef5edc499 : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":3},{"id":"f1eb29d2489717c2431a36ed79cc4e613417fa53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-11 14:48:12.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"a3b80a990764e05c417e92c0757b14b2abbf68d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 15:08:09.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/614c2a19ad504623a2284af4d046ebba\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/52aeb84477b844418823e5a00b7a699a : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/69bf8c10ae40412cba586306deaf3d10 : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/af88b113085a45a7ae71acbfb50a3c15 : SUCCESS in 10m 30s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"85b8cd62906ebcfa2116f5e04b8b499af878512d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 17:06:03.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/442c27fd6e514fd183cbedf9133202cb\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b8ea4bb13db94bcba82dc5b54e0f0a3a : SUCCESS in 5m 59s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f2dc9238ddb045c58cf805aa76b1ec01 : SUCCESS in 5m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a35f004eaa174e4293e83f8bf485f847 : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d11bf95051fb44a8a7d7e925db525613 : SUCCESS in 6m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/345c5f62fa8d44fe9a305960bf1e4ce2 : SUCCESS in 5m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/80f01255d57743ea8b21531bba684f80 : SUCCESS in 5m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/935c620527ba488d982febcbf60ad1d6 : SUCCESS in 8m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fd37eacc1e449619a7cbe4797887f0f : SUCCESS in 7m 10s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/ba0df57fb1e94f28b071f19ea6a6fb5d : SUCCESS in 54m 14s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/473e756ce1604f3fb675d3136ef8f0a9 : SUCCESS in 2h 03m 48s (non-voting)\n- trove-tempest-ubuntu-ssl-mysql8.4 https://zuul.opendev.org/t/openstack/build/1be6a63c15c04e7d8408007589be193f : FAILURE in 54m 10s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/d39e85b71bf44eb1af060f0b8d696cfb : SUCCESS in 1h 28m 30s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/369f20e81c764ed4b9e4b25447a21e45 : SUCCESS in 1h 49m 41s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/c3ba26c49d9c4b33a09d100eaa4ffeb6 : SUCCESS in 46m 17s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/b3669dfd30974abe839014e8f0e1a2a3 : SUCCESS in 1h 41m 08s (non-voting)\n- trove-tempest-ubuntu-ssl-mariadb11.4 https://zuul.opendev.org/t/openstack/build/b09a7a09d77c491dba2d6c6f83ef9b51 : FAILURE in 51m 06s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/0b20f65f098147f3a9e0acbd13827035 : SUCCESS in 1h 22m 16s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/4dc20d169be440a8ac84b85db9c78588 : SUCCESS in 1h 45m 31s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/57b661ff5ca54d3cba00596f56876e83 : SUCCESS in 53m 28s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/0c100e07cfdf470599b48dbc56001e25 : SUCCESS in 1h 39m 26s (non-voting)\n- trove-tempest-ubuntu-ssl-postgresql17 https://zuul.opendev.org/t/openstack/build/7d4ebe3d4d1b437cbddd132e3dac95b0 : FAILURE in 47m 36s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/095ca968ab3643949402e81af5efa52b : SUCCESS in 1h 16m 05s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/922d197e75f74589a40ef59d86403d54 : SUCCESS in 1h 33m 43s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/20cf92e2b25d4f7989beee005d25faa8 : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":4},{"id":"50f948fcab2971fd5fd652c3f2d207fcc9164af2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-11 18:00:53.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"8d79227fb896ae602684427acb15bf76bfd903a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 18:12:01.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d34ba0a005ae4e6bb1a605649bbe501d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c1840e758c5d4bf0b87b5cd102ccd528 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a9d18424c9d64ec8a541c59453b00285 : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/12a3fbdb6ab441e886ea7f86a264cb55 : SUCCESS in 9m 02s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"493f512bde25133f7a83498f280725292640249d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 19:53:12.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be380e350ef049e7be1491fd4d69ef6e\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/eb1e90304c59402291e48cf5a2d43e06 : SUCCESS in 5m 08s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5404d284bedc412ab51dd6e6d04d7bb8 : SUCCESS in 3m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a140a6b7ae894fe8b83cdab380b89180 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25f7e48eb6da441eb71e2d638c00413f : SUCCESS in 5m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7cd35a04f65a4b3c95666748e581f180 : SUCCESS in 4m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/852b8bd374804925bec9272a68ecd888 : SUCCESS in 4m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ff365c1bbbe344a587304a1df48d46cf : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ad5a966e1244d4cad34fe5fbb355995 : SUCCESS in 11m 03s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/00be39803bf342b290b6dffbd717082a : SUCCESS in 44m 34s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/6339a1e8bd554468bf08564ab176179f : SUCCESS in 1h 43m 02s (non-voting)\n- trove-tempest-ubuntu-ssl-mysql8.4 https://zuul.opendev.org/t/openstack/build/7626345cd41b4a2b83be4248798a2adf : FAILURE in 54m 59s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/5a3be8619d554a36b4963e5902152b0c : SUCCESS in 1h 27m 13s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/7b7fdabebecc421c85f9fb8ec986cf8f : SUCCESS in 1h 50m 21s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/59e2ee35c4724bda867f80ae685e0bf9 : SUCCESS in 51m 35s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/b698b98db58d4ece99f05e6696531e67 : SUCCESS in 1h 29m 50s (non-voting)\n- trove-tempest-ubuntu-ssl-mariadb11.4 https://zuul.opendev.org/t/openstack/build/6df0348726924bfc9b696e0a6966b1a8 : FAILURE in 57m 40s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/61b9820aae964cc296d4606330e1bd6e : SUCCESS in 1h 16m 07s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/81999458a9cf49c385f9fc2f623a2815 : SUCCESS in 1h 29m 44s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/c5d9a74d783f483f8a584334b38e1e7d : SUCCESS in 52m 27s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/e31ae495449c4d23bff6684d9e9daaf4 : SUCCESS in 1h 33m 00s (non-voting)\n- trove-tempest-ubuntu-ssl-postgresql17 https://zuul.opendev.org/t/openstack/build/0ae47844b6ac44909679bfd11b443cbf : FAILURE in 53m 38s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/d333e61a29a2466b840edf12b66a52c4 : SUCCESS in 1h 12m 18s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/14114cfd24d24fc9953acd1603a2d60e : SUCCESS in 1h 19m 31s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d122a583bd814e4c91f5007a0f4dca29 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":5},{"id":"85fa28cf811fa6ff8d637c19b655bd091f07ae08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-12 03:31:47.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"e7bb2363c224fa4188575dbd1d3fe8a7c84ba395","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-12 03:33:51.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"89b9864089f7d5d84ba4261ffe9bb94500d85b5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 05:25:26.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5d808db70544970aa27c96efc537c11\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d2207f736f5142cebd233ade5297ef0e : SUCCESS in 5m 21s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7e1e69096f9a4e548aeeb7c4ee22e110 : SUCCESS in 3m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/63665acc1aab499285580a66709a7393 : SUCCESS in 7m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/806db0072d084abd81193b861cf7b84e : SUCCESS in 6m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d0663ec1033542598eadc43ba52d220f : SUCCESS in 4m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2030b5e1b6c94a5bbeca1f91007516b7 : SUCCESS in 5m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c06b5e5e5a3a435d9f036a3703658c7a : SUCCESS in 7m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/963d832559e747ba9d9deb8a763da1db : SUCCESS in 4m 46s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/57fad6ca0dc14b4ea831ba97252dbffd : SUCCESS in 45m 59s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/27666cd220c0410bb6f4d5e29bfff42b : SUCCESS in 1h 47m 28s (non-voting)\n- trove-tempest-ubuntu-ssl-mysql8.4 https://zuul.opendev.org/t/openstack/build/45a2fb5f0d194a32a27bf96c6fb6b91a : SUCCESS in 1h 39m 04s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/8541ef811f004dca88252e997a7d7df4 : SUCCESS in 1h 30m 22s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/dc06e9070a1a480d9e693d91a2b7f1e6 : SUCCESS in 1h 49m 54s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/b0e41e12c4c246a1b5ea98644c069d74 : SUCCESS in 43m 33s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/a24d17afb7e2429cab9e2764d4253c85 : FAILURE in 53m 08s (non-voting)\n- trove-tempest-ubuntu-ssl-mariadb11.4 https://zuul.opendev.org/t/openstack/build/1b3c4b9f45734642ae3edc469756b7d4 : SUCCESS in 1h 20m 46s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/4faacf120a21493894870c4b2c248f66 : SUCCESS in 1h 18m 43s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/46b0e038919e41a3adf54a7187ed9ae1 : SUCCESS in 1h 34m 36s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/62354888ab39455c8d4ae1e09b8fe586 : SUCCESS in 46m 28s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/fdfb535f4c014cbb846ab0dfdfd3f97c : SUCCESS in 1h 30m 29s (non-voting)\n- trove-tempest-ubuntu-ssl-postgresql17 https://zuul.opendev.org/t/openstack/build/673301841f684c8fa6a0f17f32a4119f : SUCCESS in 1h 16m 15s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/a366d68c2bae443d9584bcc914d5f117 : SUCCESS in 1h 12m 24s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/45d97e800323401d9fed1366015ae15a : SUCCESS in 1h 19m 07s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e82460a5c949411383f57fb81c5d22b5 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":7},{"id":"e0558f964db8c5bff406cbb702afa7b1516c8f57","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 05:36:58.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c470d0c713b47a1bc5e45905987dfdc\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9ac28f300ae54b15b979acfe398c05cc : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/67d938ada0a04c92b2ee8e6f49274fbb : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e11a96400d90477a90a9fdb2c3d02fdf : SUCCESS in 13m 53s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"b0f7c89a6d21f4515ef15f05c313622232fd74a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-12 07:13:07.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"0a268568c3b07354b17a53cf951b9e79d587904e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-12 07:13:49.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"785eb6dcb75e8225851549a43df7b121d240fa93","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 07:29:46.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14c0428399b64322b5709a30f5fa3e32\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1034e5db6b61443a8f175b7d8f99a07b : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/52401244d7694565a497458781e5f8f8 : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/fb9369e33704451487fa36685e7375dc : SUCCESS in 13m 13s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"9ffd7d50ca24ff58173bbd8bf97b16987015de3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 09:08:29.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6aa55a377ea4b4599f854e76f9dc128\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7fa1307d59164bf7a3ac3b3a7cd9d493 : SUCCESS in 6m 03s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b8cd9f2f2c234be690550ebd9a430feb : SUCCESS in 3m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6becf6e10dbb460c8ae9f757e47569f6 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae5b3b7f0dbe400d89b8807497e0579a : SUCCESS in 6m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/592b31d8db98445cb93cb42338403563 : SUCCESS in 4m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4ed25f25f80b4d2c845bedf7b6d82588 : SUCCESS in 2m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c50501e8ce954ef2bc96df0171c53045 : SUCCESS in 8m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ccfd754230f47048e2996e535db07b9 : SUCCESS in 7m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5fb3d4d6b7c461b97c38550fe945a17 : SUCCESS in 5m 35s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/0051ef0f4761438cb956981b56bdee09 : SUCCESS in 47m 43s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/4f5e54c996d041a3855e70bdfe53003f : SUCCESS in 1h 44m 06s (non-voting)\n- trove-tempest-ubuntu-ssl-mysql8.4 https://zuul.opendev.org/t/openstack/build/c96dde0e48e24e4e80b57e8ab39e2189 : SUCCESS in 1h 37m 13s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/a9874ca3204d4051ae709923a427d478 : SUCCESS in 1h 24m 29s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/4ac9768cfc8c49179cf753f58ba26645 : SUCCESS in 1h 51m 13s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/b40be39355404269ab358f22c3af183f : SUCCESS in 48m 00s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/8a51903d8b614628b7ed25dc36152662 : SUCCESS in 1h 35m 54s (non-voting)\n- trove-tempest-ubuntu-ssl-mariadb11.4 https://zuul.opendev.org/t/openstack/build/1637980d20424055be17f8a9ebe7e2fb : SUCCESS in 1h 20m 50s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/a97d949056524939a54dba105dd1886e : SUCCESS in 1h 15m 21s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/9fc713a0b8b24d279b5cb6d61a223361 : SUCCESS in 1h 31m 50s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/de8d69433e6b4de2a6931da1cf1dfe2f : SUCCESS in 43m 57s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/b2e51696e8fa48c2a4f350eed02b7f1b : SUCCESS in 1h 33m 30s (non-voting)\n- trove-tempest-ubuntu-ssl-postgresql17 https://zuul.opendev.org/t/openstack/build/8dcf296849ff472baa43474131ca0c9e : SUCCESS in 1h 16m 35s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/22797629d98b492da02f1a269748c8a0 : SUCCESS in 1h 12m 19s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/292059f38a5d48cb8cab42cf17f5ab93 : SUCCESS in 1h 17m 56s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4466535255804edda43d3258d2d12af7 : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":9},{"id":"62c33da62ba6f889e57f096414a3bfb65ca8fa90","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-05-25 13:59:27.000000000","message":"Patch Set 9: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"eeeac739ce970f1ce255b217016d0ed8e4150369","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-25 14:21:32.000000000","message":"Patch Set 9: Backport-Candidate-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"128a501fdf3e8ad9125269e2b8a7909c12d42bba","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-05-25 16:10:46.000000000","message":"Patch Set 9:\n\n(9 comments)","accounts_in_message":[],"_revision_number":9},{"id":"a9a3d16fd4d603f2148aa0e90f9e86f8b26b61ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-26 09:25:31.000000000","message":"Uploaded patch set 10.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR **is:MIN**\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"8182a2f7db7ce0f7ef023f9d6b41a87cd455d66c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 09:25:55.000000000","message":"Patch Set 10: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/trove on branch master.  The\nproblem was:\n\n  The key \"voting\" appears more than once; duplicate keys are not permitted.\n\n  voting: false\n              voting: false\n\n  in \"openstack/trove/zuul.d/projects.yaml@master\", line 23","accounts_in_message":[],"_revision_number":10},{"id":"1de2b3db4c5b35bf699bbb2dd4bdc4566cb56359","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-26 09:25:57.000000000","message":"Patch Set 10: Backport-Candidate-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":10},{"id":"8439f485a19457a0bb9a5e5fefb10f3f01b081d8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 09:26:13.000000000","message":"Patch Set 10:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/trove on branch master.  The\nproblem was:\n\n  The key \"voting\" appears more than once; duplicate keys are not permitted.\n\n  voting: false\n              voting: false\n\n  in \"openstack/trove/zuul.d/projects.yaml@master\", line 23","accounts_in_message":[],"_revision_number":10},{"id":"d1dacb518845847ea669b9e313bc962c6aad7dd3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-26 09:27:09.000000000","message":"Uploaded patch set 11.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"f35ea6c16f47200bf0919c4ee9d8ab4d8ccd274f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 09:55:39.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e3aa9b6b7d34ff790a111ab6f53241f\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7ab4cf7861f0429ba297b1a2b51fba5b : SUCCESS in 7m 13s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/027883793ef149499b1db6fb82edd173 : SUCCESS in 7m 41s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/6ea8516b732a4134b708cd441841e704 : SUCCESS in 8m 32s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"1869f9eefe1e4eda30858cfaabfe3d357e86542c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 11:49:54.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f86d4dc619cc41bfa28a639ff3178036\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8225f47dec224a768b22a13267c5e8b2 : SUCCESS in 5m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8f08b207566c44138124ba3b9c83b943 : SUCCESS in 3m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/31a80a280bf5431eaad242a495285100 : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/065952155ae843729bd836b3359750c0 : SUCCESS in 5m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0a0c4dbd9b5541a688f074103603b91f : SUCCESS in 4m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a6511b9ec2c34de4a51923a40941eea0 : SUCCESS in 5m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/004496e6033746fab864d41af81ff30d : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d75c0024be143fbacd74e0cee8eef55 : SUCCESS in 8m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ee6f83efc914eeb9ba1a04d7b815bd9 : SUCCESS in 7m 20s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/c27adca4f22e443b8b3091d0d5ba96e3 : SUCCESS in 45m 42s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/cbd4ba52663441b2a5b8cb446fb273ee : SUCCESS in 2h 20m 03s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/446f9c0390904ad99e859ae28a890370 : SUCCESS in 1h 37m 04s (non-voting)\n- trove-tempest-ubuntu-ssl-mysql https://zuul.opendev.org/t/openstack/build/409a8ed824204908b70fa72aca56b1f7 : FAILURE in 1h 09m 23s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/a29ca5520713451fabad0f2ff8de6446 : SUCCESS in 1h 51m 25s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/c0cb08326cc946bf96480b32b02b35b3 : FAILURE in 56m 19s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/3b3b8a564fd24dfa961b55fafb2b917c : FAILURE in 1h 03m 33s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/48a6e5c436d34451861bc3fea0e36e34 : SUCCESS in 1h 26m 54s (non-voting)\n- trove-tempest-ubuntu-ssl-mariadb https://zuul.opendev.org/t/openstack/build/5860e467809e4bf9837e7016151fe55f : SUCCESS in 1h 26m 42s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/4b645c19830543f0abe774b68bff9e3e : FAILURE in 23m 07s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/20d2eec5d1234562aedbc426efdd6f24 : SUCCESS in 45m 06s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/e8128a41e474496a99fc0d7ebbd7a69d : SUCCESS in 1h 37m 22s (non-voting)\n- trove-tempest-ubuntu-ssl-postgresql https://zuul.opendev.org/t/openstack/build/e3822dd627bb4de5b7e7b2eab9c0999c : SUCCESS in 1h 49m 33s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/7fc98e7110ff4e66a92326ecfcc8cabb : FAILURE in 28m 21s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/be3ac3d8442d4c92bbf47bf045f2aff1 : SUCCESS in 1h 34m 59s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b1cc7ce36c7d434ebc84647475174e50 : SUCCESS in 6m 55s","accounts_in_message":[],"_revision_number":11},{"id":"1a39d6c9e9a41a6315eed306b7fef43ce4da8939","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-26 12:53:02.000000000","message":"Patch Set 11: Backport-Candidate-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"6b32d71cfff63084940e75c7749638caa443d77c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 13:12:47.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb18d867e42843968db1a8b423fb979d\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/45d4f64bbf9b47d799205f7cd13574d7 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/62026278f61d432a88c97db0bb058157 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/33a6b1996a9a4816a6606cf131a8b2ad : SUCCESS in 7m 46s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"9e79c7a5ba1a315a46de564a370a33889ea6d923","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 14:59:33.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b2d8d7400b14333a9d92af4f4c8771e\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/eef217af8b1b40a0afaf44e18fb5c1fa : SUCCESS in 4m 18s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b4c0a35ea71e4531b9dc5102cdc9b5ed : SUCCESS in 3m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee810ed5c4bc4c759c37a5a44d6c429e : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4521c850b9224c2796ea6f8b6a955b71 : SUCCESS in 6m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/287f874462184ec2b349b14a835db4d3 : SUCCESS in 4m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b368eb0c12f241e88d3c28f3d54a2ffc : SUCCESS in 6m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9dd87210459d4739a10161c08f7dee00 : SUCCESS in 9m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/837f3b3fd9904748b54d8bafc688f06a : SUCCESS in 9m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/68b360581486426a8363b11ceb8f5160 : SUCCESS in 7m 53s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/d5f91e0d997945099b5de4ef57b0a2f8 : SUCCESS in 47m 53s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/cdff01e4951e40af9e4071d82bbf7f89 : SUCCESS in 2h 04m 34s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/2e55c1c905dc4db2a1e40f5f0795020d : SUCCESS in 1h 36m 49s (non-voting)\n- trove-tempest-ubuntu-ssl-mysql https://zuul.opendev.org/t/openstack/build/643cae9197124a7d86145e0044e9840c : FAILURE in 1h 07m 27s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/d36b1d47a3df4f4fb88b867663de20fe : SUCCESS in 1h 45m 52s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/b6dfd17cb339458d9e6933e1b3946e41 : SUCCESS in 41m 48s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/d5e4282dd5b24df092bd9f1fa69814e4 : SUCCESS in 1h 37m 14s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/a7da5bcf51da471c95cfa57c2873d075 : SUCCESS in 1h 12m 26s (non-voting)\n- trove-tempest-ubuntu-ssl-mariadb https://zuul.opendev.org/t/openstack/build/7b178d1fcd20482ba8f58b74e376d7f2 : SUCCESS in 1h 19m 42s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/e90609312b3d4cb7bb5706b5e88c57d8 : SUCCESS in 1h 26m 50s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/4f41d924cf02492da316611c44f7ff91 : SUCCESS in 43m 59s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/aa8f862cd636405fbc969ab4b9b1677f : SUCCESS in 1h 38m 52s (non-voting)\n- trove-tempest-ubuntu-ssl-postgresql https://zuul.opendev.org/t/openstack/build/05f1927b99e447908992d7c4c9957fa2 : SUCCESS in 1h 20m 26s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/8ece6499a1974cdfb1e0b2eecc389cee : SUCCESS in 1h 14m 46s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/b675d6a76f614376b8a679bf1cb03432 : SUCCESS in 1h 18m 30s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/905dd51de20447c0b5e7df56a6308223 : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":11},{"id":"01e0ce1858290c0798cbbeb1c34bc44c779e3cc4","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-26 19:40:17.000000000","message":"Patch Set 11: Workflow-1","accounts_in_message":[],"_revision_number":11},{"id":"63960423636a5fc6768bbfbcd7363480ff2464d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-27 13:01:31.000000000","message":"Uploaded patch set 12.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"5cf8572615b0d07302ba23d88101b1de8df4ac1c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 13:34:15.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7fd6d495e7514cce9853c99cf1ee4d3f\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8fb350b5cd774911be5319dc5aa419f7 : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/93b3d0cb969e49808e2c3d68edb0a2f4 : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/884a3640626a4ab1a6efda4bd5d4a5d3 : SUCCESS in 14m 58s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"f4672ce21b2c90ade213145aaf37ecbef2e6859a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 15:18:17.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f89537469c1643169d2800ad5bec8170\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/24079fa09a9c4cab9939a348ea9a5229 : SUCCESS in 3m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6f43b4b881de490c880ca2d806894c6e : SUCCESS in 4m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e9105df91a2a4476b187e1f6a6d563ec : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7573f91a62fc4fef99bdcee2a8957c08 : SUCCESS in 6m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/74f0fd6965ed46cc861c3c969e653e07 : SUCCESS in 5m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cfd4909e412d4110898c47bfad28cfe2 : SUCCESS in 4m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/193638fa0fa64865a028c0d72ed822b1 : SUCCESS in 7m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2685d2084ff54028bc6fb33c4c9f4dd6 : SUCCESS in 11m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5629403c017e492ea43ab1eb9e3e963c : SUCCESS in 4m 34s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/d8227f7554b843408dc9581c709e1bd8 : SUCCESS in 51m 23s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/40a49097053840d7b024398fffe2c04d : SUCCESS in 2h 14m 55s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/b9f0da7f05df4702aa8718dedb57d64e : SUCCESS in 1h 41m 16s (non-voting)\n- trove-tempest-ubuntu-ssl-mysql https://zuul.opendev.org/t/openstack/build/e9978718e4f6443a8461c62e6ca31629 : FAILURE in 1h 49m 34s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/952f052b6ecd470b915638431deaafcb : SUCCESS in 1h 53m 05s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/d4956fa18d9e4f938c49b67e5ac25991 : SUCCESS in 53m 44s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/b9df39a6d5384ad4aaa2f9b4dd2ab825 : SUCCESS in 1h 41m 20s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/6cb28ce6f106451b803db9c59a08c3d2 : SUCCESS in 1h 20m 06s (non-voting)\n- trove-tempest-ubuntu-ssl-mariadb https://zuul.opendev.org/t/openstack/build/6676df4bca0c47128424463115f44b43 : SUCCESS in 1h 29m 14s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/9772c540e7064ee68881962e9e49b68a : SUCCESS in 1h 36m 28s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/229e35caf21841deb125d5a36fdc537a : SUCCESS in 53m 24s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/da574c167fb445d4b8c57ddba8c6df31 : SUCCESS in 1h 48m 18s (non-voting)\n- trove-tempest-ubuntu-ssl-postgresql https://zuul.opendev.org/t/openstack/build/5cbd861a26ee4f529ef4b835fe0871f5 : SUCCESS in 1h 25m 59s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/52480c1b88934f3f88237d43f9cc544c : SUCCESS in 1h 29m 16s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/499a437929904fb69814a59529b5e6b1 : SUCCESS in 1h 25m 38s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f972901597804eceb60cdce2421627ad : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":12},{"id":"91d8258d67ad1b36c83388f5ee99bdd9a13614d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-27 16:54:52.000000000","message":"Uploaded patch set 13.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"9957cfb6a0d4bb1b857255c24bb4f7ffb12eae07","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 17:12:24.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36437a00585f40c186b216be15eac8e0\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b35a97f6054240a4b085a19d6821b30e : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2af3328753dd45358a26fe22afb60235 : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2611f7d9dff44bfca7e3eaf2384efcde : SUCCESS in 9m 47s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"516e4a8e18c38ffd36f69c7fcf5750573439c981","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 18:50:37.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37d4c5a83a3a4842942b66878d5df24b\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1d79a179f3bb42308e9a5a6de6f6abcd : SUCCESS in 3m 19s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8b5af9dcc6b04fd2be2c65b69351615b : SUCCESS in 3m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6bdbae70a196410ebd758eb5be4de323 : SUCCESS in 8m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b3ea994ee194290b27f6390ab2f3565 : SUCCESS in 5m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0d56ea37652f4c589804849ddb5b76c9 : SUCCESS in 4m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2ed7b33f4c9e4268914d3bb6f1cbd56a : SUCCESS in 6m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3dc6b8eb59d34ffea16afcddf160509c : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fce7e87302f34640b9271c32c924978d : SUCCESS in 9m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e27e268d833a4eb9840634236718a226 : SUCCESS in 8m 13s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/f684a1014dfc43aeb66153a184c40272 : SUCCESS in 55m 41s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/5ba3a9291a394b40804099d9269cbeef : SUCCESS in 1h 50m 15s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/23eff7ef19ad43028ba7ab049f4b7133 : SUCCESS in 1h 48m 34s (non-voting)\n- trove-tempest-ubuntu-ssl-mysql https://zuul.opendev.org/t/openstack/build/7122a5ee4eb5440f91ddebcc696c7899 : SUCCESS in 1h 52m 36s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/c62212abd4874186b1ff82236a61cc61 : SUCCESS in 1h 54m 08s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/3c691fac9d4b44048181a36e1a6851fd : SUCCESS in 48m 30s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/8b748a42b37e470a825ac92b40356ee7 : SUCCESS in 1h 39m 57s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/b49477f5c2434201852866a1b1ca1229 : SUCCESS in 1h 14m 49s (non-voting)\n- trove-tempest-ubuntu-ssl-mariadb https://zuul.opendev.org/t/openstack/build/e90a4f22159248efb7267865de6835d1 : SUCCESS in 1h 22m 21s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/892cffda9e1b4ae9bd86b6708f154036 : SUCCESS in 1h 34m 36s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/c59bdb3c3e4d430891ef3162a6006711 : SUCCESS in 46m 26s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/b490ef29997f4ab4bcef3dea466a1c0c : SUCCESS in 1h 25m 31s (non-voting)\n- trove-tempest-ubuntu-ssl-postgresql https://zuul.opendev.org/t/openstack/build/7e5a53fc956c49b998abe22bd63f2356 : SUCCESS in 1h 23m 25s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/c729e8f34ca7416ab1c74af5d99e99c3 : SUCCESS in 1h 15m 13s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/c810a64862b34c2ab156df5597a4bf91 : SUCCESS in 1h 17m 00s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6f97e53e97af40e2be7eaa3ce7a8a262 : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":13},{"id":"e1fbff8c5a2fde87d43ca3a0a7fb851b36581a54","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-27 19:56:05.000000000","message":"Patch Set 13: Backport-Candidate-1 Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"7f3cf703f3e48be51aafa3670a12d71d2ae47d39","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-28 16:58:50.000000000","message":"Uploaded patch set 14.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"86e3b66aaacb4f1d63f7606e1bd5dbc59a88d673","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 17:30:03.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd02c78c16c74971a380506a5a2f2ff3\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a54fc28a1bcc485aa3ea0e8964b33eb3 : SUCCESS in 9m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ae5d2276c38b43eba9f6b7b3729f883a : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/47210facbfbb4478a884c0ce55568fa8 : SUCCESS in 15m 35s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"a2f8f7174eaa0b9f9a01c81448643fddb44a82c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 19:07:54.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ac70cab88eb4c8a97b1882a19c1ce1f\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8b6347e24d2e4103b0bd852c744e57b9 : SUCCESS in 3m 32s\n- requirements-check https://zuul.opendev.org/t/openstack/build/fc7748f1a52a410ca2f1a4a0abd757eb : SUCCESS in 4m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b7054d7fa3554cd6b9e7b73e2e81b696 : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e056a9f85804be7908e092f8dea8342 : SUCCESS in 4m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/316f9305dc1149eca4e73610a65b2ee4 : SUCCESS in 3m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8b3f0605ddb14a9badd41becc64dfcc1 : SUCCESS in 4m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/83a24cf6137d430cb5a47ae7cf2cbbfc : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44d2fbb29808456a9e6cf3932da248b8 : SUCCESS in 10m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d774a342c2f457995a1b7b904a46b37 : SUCCESS in 6m 24s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/dc0c4ebd1f6c410e9af56b108300f648 : SUCCESS in 49m 25s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/0b37426ed8344bf7b362648b8b041b04 : SUCCESS in 2h 07m 39s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/f96b95cbff724729a048d3e9ebb14fae : SUCCESS in 1h 44m 26s (non-voting)\n- trove-tempest-ubuntu-ssl-mysql https://zuul.opendev.org/t/openstack/build/a195389f36e840008e54e75f5e07fbdb : SUCCESS in 1h 14m 28s (non-voting)\n- trove-tempest-ubuntu-ssl-replication-mysql https://zuul.opendev.org/t/openstack/build/0bfb47e004d544af9e92f2288eb3fa8c : SUCCESS in 1h 39m 02s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/579f53f15cab402d8862567631590d7b : SUCCESS in 1h 47m 06s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/49b47ed491f94207b9c3714f250c0b68 : SUCCESS in 45m 51s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/85673efa7888460fb83cd9b5e22ea224 : SUCCESS in 1h 35m 59s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/cfab256b9e3242d9900ebc9c5315610d : SUCCESS in 1h 15m 19s (non-voting)\n- trove-tempest-ubuntu-ssl-mariadb https://zuul.opendev.org/t/openstack/build/045ecf6a2ce34968981343fc71865bc4 : SUCCESS in 1h 01m 31s (non-voting)\n- trove-tempest-ubuntu-ssl-replication-mariadb https://zuul.opendev.org/t/openstack/build/2b0adebfed644a979d29c7a9ed04713f : SUCCESS in 1h 24m 12s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/d645398ff1044de283e7d6986038d761 : SUCCESS in 1h 33m 43s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/bc1ff66429d444cfb5f846f7288306f6 : SUCCESS in 45m 39s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/8d02bc857a8c4cefb6fafd0809006a1c : SUCCESS in 1h 40m 33s (non-voting)\n- trove-tempest-ubuntu-ssl-postgresql https://zuul.opendev.org/t/openstack/build/cdc4729cc8864f0e83f796469e0b4cef : SUCCESS in 1h 00m 27s (non-voting)\n- trove-tempest-ubuntu-ssl-replication-postgresql https://zuul.opendev.org/t/openstack/build/b915b1784e984d0abc51c2c212ccab37 : SUCCESS in 1h 26m 21s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/a7ace4b162a845c781fe18852149c5a4 : SUCCESS in 1h 23m 33s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/e083854fc20c412686e0dde97b95fec0 : SUCCESS in 1h 27m 30s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b9d9e080d3044baab77b2ecf7c875c7e : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":14},{"id":"dafeb4351ab5d4358347885b257a75349b4932fb","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-28 19:17:16.000000000","message":"Patch Set 14: Backport-Candidate-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"4bfc506b8ed42a9358dc1a084356bf41d3071f61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-28 19:21:07.000000000","message":"Uploaded patch set 15.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":15},{"id":"4b45c66b47a6d45ec0f252a4818e44ea35836dd3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 20:44:23.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1dde2bf86b4040f5a07a0fd403b54344\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/28ac4627ef5d40e98e78abb42a78e436 : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/98d3413c023840b69c65f90ee553a5e3 : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/6b1dfc62fa7e4a4cab896130d9643efa : SUCCESS in 17m 07s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"3e897d8298b59307c303b90223c826dbf640cdc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 21:25:06.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3ba0a15610e447e9d3b60c98ee1993c\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/49193f4f43044361a7eca6e4f700c918 : SUCCESS in 6m 54s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0f755385baff48b2aa06b91028e4894d : SUCCESS in 3m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8520339eaf643b5a4495e7ab026eaa7 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/986f5efd88a348cf873f4fb448b4716f : SUCCESS in 5m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ac32d1f2b88749dcadab1ec18c2227c6 : SUCCESS in 2m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/40952648e20949ce986ab95df177e88f : SUCCESS in 5m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f94181a90e214eceba4473d6c9d7b801 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86395f4e239749a3b416817333e227b5 : SUCCESS in 5m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c64dfff4f5c240fe9f2c66f085a4ee3f : SUCCESS in 7m 57s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/c2542865a158484cb1dd09dc4e2eb492 : SUCCESS in 55m 57s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/a55fc14f10a6483c8f60690c312d6152 : SUCCESS in 2h 01m 53s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/b5101b2330a14e14808b9979e4efc976 : SUCCESS in 1h 51m 18s (non-voting)\n- trove-tempest-ubuntu-ssl-mysql https://zuul.opendev.org/t/openstack/build/d1438d542fcd47a99fce09c8efb2ad87 : SUCCESS in 1h 22m 30s (non-voting)\n- trove-tempest-ubuntu-ssl-replication-mysql https://zuul.opendev.org/t/openstack/build/997e751891064bf8884e07b7236cd66a : SUCCESS in 1h 51m 53s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/1075d77e247d418891fcb2673ae47cd5 : SUCCESS in 1h 51m 23s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/fbd122cc94bb4372b0511feebb513021 : SUCCESS in 47m 33s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/56f07f32ce524d87b459dd893d1fcd92 : SUCCESS in 1h 38m 42s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/b185fb8bfc5645509a70b608bee49bb0 : SUCCESS in 1h 16m 31s (non-voting)\n- trove-tempest-ubuntu-ssl-mariadb https://zuul.opendev.org/t/openstack/build/87760a56d5c7422e8de76eee2538b1cb : SUCCESS in 58m 39s (non-voting)\n- trove-tempest-ubuntu-ssl-replication-mariadb https://zuul.opendev.org/t/openstack/build/897f03e29bbc4fe59c7f08f3e070bcf8 : SUCCESS in 1h 16m 41s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/1d89359a0c184e7084a6173ecad585fa : SUCCESS in 1h 38m 40s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/176c939a68b5412999f79188aecc275f : SUCCESS in 44m 10s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/188a5352ec2f4d4b803ce38e96dbaff0 : FAILURE in 59m 13s (non-voting)\n- trove-tempest-ubuntu-ssl-postgresql https://zuul.opendev.org/t/openstack/build/04fa7e90b5894d84ba85cd7e150e8624 : SUCCESS in 53m 16s (non-voting)\n- trove-tempest-ubuntu-ssl-replication-postgresql https://zuul.opendev.org/t/openstack/build/62f5fad313b84b2bbf26e7896d85bbe5 : SUCCESS in 1h 21m 28s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/48a21ceea7764edd9f5c488b18f96a8b : SUCCESS in 1h 24m 52s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/bd48606f3ee843c6a5f413e1755f4aee : SUCCESS in 1h 29m 10s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/63c93f86bbf74dd185cf8f5b213204ba : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":15},{"id":"ebbf812767abb2b3974e8857bdd7ba98daa89b4a","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-06-04 02:43:57.000000000","message":"Patch Set 15:\n\n(3 comments)","accounts_in_message":[],"_revision_number":15},{"id":"81d84dafeade6103811f319390d71f1503740ef3","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-06-04 02:45:10.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"515eff51963addc6b830841f0563bd94a12cbea1","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-04 06:16:26.000000000","message":"Patch Set 15: Backport-Candidate-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":15},{"id":"c3702e1064e728d084f6f145d45ebfccc413afeb","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-06-08 05:58:39.000000000","message":"Patch Set 15:\n\n(4 comments)","accounts_in_message":[],"_revision_number":15},{"id":"a2714d643df8c3ed77a778d4042632dfadf95575","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-08 06:31:14.000000000","message":"Uploaded patch set 16.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":16},{"id":"066e05545bda8b8ea897b88319747d6549c41f24","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-08 06:31:33.000000000","message":"Patch Set 15: Backport-Candidate-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"da44a458f0495ec215d2f3959ca558e482c1cc30","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 07:02:07.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8cffeabdbf234074a8abaeb76712dce9\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/af5c484c61a746f7a189a3b96b8c2628 : SUCCESS in 10m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9073352859c644bb9a8aef097783fc6d : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/95abb352f2cb4886a84add60e31116c5 : SUCCESS in 12m 45s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"a164ce86672eb7efddefd1fb16ccd0c444a67428","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 08:40:55.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fdd72301397463684f64c9a2ce262fd\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/45230f026f7448c3851c8fc293131479 : SUCCESS in 3m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/de8f68df06d249788de78078f53e9a2f : SUCCESS in 2m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/69886d0d3efa4b99aaae71434cf860ba : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93443993327e4ad98f5a4a51d3539735 : SUCCESS in 3m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6fa22a7f7c534404bccfb32d250fb0c8 : SUCCESS in 6m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8d57d0f33b8f470999e0f390287267d9 : SUCCESS in 4m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0d7efbcaa08f455ba69686fc47ed06fb : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b506c2492f74f11805d206685607c9f : SUCCESS in 5m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e81a367fbe5c4853a03598cc5728e5f6 : SUCCESS in 7m 34s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/630e61282a5c45dbb474a6e6efc7f97d : SUCCESS in 50m 32s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/41e27eeb82ee4ed3b56ece6200215082 : SUCCESS in 2h 01m 22s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/e201ab0ae73c413486a370547abe17c5 : SUCCESS in 1h 34m 45s (non-voting)\n- trove-tempest-ubuntu-ssl-mysql https://zuul.opendev.org/t/openstack/build/91fd6768333a44788ba345425834ff13 : SUCCESS in 1h 11m 01s (non-voting)\n- trove-tempest-ubuntu-ssl-replication-mysql https://zuul.opendev.org/t/openstack/build/e848c0436d0548739e95ad862ad77d46 : FAILURE in 2h 08m 11s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/2a368572ed2f497887793934d5b7fc30 : SUCCESS in 1h 46m 01s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/82f25a2dff124bc5907e00eeb6e0eefa : SUCCESS in 50m 21s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/ae31b88f60b74e3ebb3104c66940bcf6 : SUCCESS in 1h 43m 22s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/8fe113732378449aa2098e42fc5a4105 : SUCCESS in 1h 30m 51s (non-voting)\n- trove-tempest-ubuntu-ssl-mariadb https://zuul.opendev.org/t/openstack/build/6a26b1e8f1194d88a75ef50d3cfc7a42 : SUCCESS in 1h 04m 33s (non-voting)\n- trove-tempest-ubuntu-ssl-replication-mariadb https://zuul.opendev.org/t/openstack/build/7ea87708a7114af69967cf0db084325f : SUCCESS in 1h 29m 54s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/be96dd5ef88b421993265b14e74a0d43 : SUCCESS in 1h 24m 37s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/ac6eed451a354bedaa16510ff2725c8e : SUCCESS in 46m 38s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/3c0d3402901543d18cdd76571284c75c : FAILURE in 1h 02m 31s (non-voting)\n- trove-tempest-ubuntu-ssl-postgresql https://zuul.opendev.org/t/openstack/build/9d40a8c1e51c4c5eaafbfd6118009cdc : SUCCESS in 1h 05m 41s (non-voting)\n- trove-tempest-ubuntu-ssl-replication-postgresql https://zuul.opendev.org/t/openstack/build/c99924b91fe14d309f50bfab5f511c05 : FAILURE in 1h 06m 12s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/916bd47b672848368e643a2c4e55cf66 : FAILURE in 58m 12s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/c538a428e0f84d40a3e08b4768fbf46e : SUCCESS in 1h 23m 24s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2c352410f9744614aeb5f87d97758700 : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":16},{"id":"83d5a5adfc656c8ec79bdefe23b69f2f32add046","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-23 09:25:48.000000000","message":"Uploaded patch set 17.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":17},{"id":"47f54cb54a4f39ee266c12b9514cfc640d0c7820","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 09:44:08.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a673abcdbfc451981679a7cb4e2f6af\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e9b8f4ad6a1843958fdc8462013b6f45 : FAILURE in 7m 55s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a53283501ed841998879659d6e6012d2 : FAILURE in 8m 36s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e569358dd00d40739558169bf85bed76 : FAILURE in 15m 23s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"21d6869e36eb63e3845c45068ddcc429068260e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-23 11:17:29.000000000","message":"Uploaded patch set 18.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR **is:MIN**\")\n","accounts_in_message":[],"_revision_number":18},{"id":"85db67b6439908c7c3c0872016fb33558247ed97","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 11:35:04.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f018bc1000314de8994532f16543dd03\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f39e11982d044628a64518684a78370a : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/aab7f8bfe95c405fbb50276f2f31cd05 : SUCCESS in 8m 39s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e10d11556d974be79d7ecadb846b16d6 : SUCCESS in 14m 57s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"f3fd68014fcdce6f3e893d1100f1682e2d4ca367","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 13:22:52.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75b21e15036a45c09dd4e27c8dd4962e\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b1eada3e59be45a29064ee71859f260c : SUCCESS in 5m 20s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c4fb9d59d92b4aff8aba9f0be9641144 : SUCCESS in 3m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a7c64d583d854f82aeb40867584a5608 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f567729532c74322b5084c4fee76855b : SUCCESS in 5m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a9961a1c72e9445cbe1743f2c8b2cb0b : SUCCESS in 5m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f357582a6d534e2c8610bf73c18da973 : SUCCESS in 5m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ab272453c5f94e6887c3fdb8b0d09efe : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6da50893e09e47579ece04d79af3cdb9 : SUCCESS in 10m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a9174c530f9743d08d68b26c01b5a000 : SUCCESS in 5m 59s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/638d3276e36b4c0f975767c6fbd88529 : SUCCESS in 1h 03m 17s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/e8986a4b6d39449e8ec0bdf3356fcd03 : SUCCESS in 1h 56m 13s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/f60b71b6013c4640a504d23cb60e26e4 : SUCCESS in 1h 44m 40s (non-voting)\n- trove-tempest-ubuntu-ssl-mysql https://zuul.opendev.org/t/openstack/build/319fc4c416a244b4ad72e10cb759eefe : SUCCESS in 1h 01m 31s (non-voting)\n- trove-tempest-ubuntu-ssl-replication-mysql https://zuul.opendev.org/t/openstack/build/e4a7a35b1dea4a15bf1fe5dea73d0335 : FAILURE in 1h 08m 01s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/8e32edab325744e8bc07ce5972a2c8d3 : SUCCESS in 1h 44m 23s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/045ec3f09e054aacb94906860904402e : SUCCESS in 52m 53s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/4c9786f3af63427e94c43d388c8cf2cc : SUCCESS in 1h 35m 38s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/87076ffe190643c6a16c3ecde341ddc3 : SUCCESS in 1h 11m 57s (non-voting)\n- trove-tempest-ubuntu-ssl-mariadb https://zuul.opendev.org/t/openstack/build/123c4efcf3154925abdad06214b68500 : SUCCESS in 49m 29s (non-voting)\n- trove-tempest-ubuntu-ssl-replication-mariadb https://zuul.opendev.org/t/openstack/build/bbd786b47bd94adf8e33061089aaac24 : FAILURE in 48m 17s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/3b944e9a803f4f87995a174f7d92c7ca : SUCCESS in 1h 35m 19s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/af6b034f249d4c8ab32e4a7d4bfb687e : SUCCESS in 44m 26s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/3077cf74feaa422f85779b1f37aff7c4 : SUCCESS in 1h 34m 29s (non-voting)\n- trove-tempest-ubuntu-ssl-postgresql https://zuul.opendev.org/t/openstack/build/abdfe3cbd31f41f0b23b6ad79944fae9 : SUCCESS in 46m 56s (non-voting)\n- trove-tempest-ubuntu-ssl-replication-postgresql https://zuul.opendev.org/t/openstack/build/12fdda2b9bdf49cf9f51aa2cfa9c95d3 : FAILURE in 46m 54s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/cc41b4caf9b94ff69f71dc3e0994fa05 : SUCCESS in 1h 12m 39s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/9661b505c8504366bf48abe163b6a1e3 : SUCCESS in 1h 18m 40s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ea47cc92cf4f4f538bd80ac3ecacdef4 : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":18},{"id":"705bd8f22bb218b0d4553fd3d795e7b6142cdb92","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-23 13:34:51.000000000","message":"Uploaded patch set 19.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":19},{"id":"c4d3a079f9496cbf009ba74fcb012dddc6e76579","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 14:04:50.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff30b6717bdd4370b0ed0112befd95bd\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a3033bc06bfe4357aff74c07d2da61be : SUCCESS in 9m 54s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b9739fe811e04c1492901e6acf86d56b : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f223c00fe3d94d868aca854456edeb03 : SUCCESS in 17m 26s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"3eb31359c3cbaa2a30b79ddd3f1e239208bc7adc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 15:38:59.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a396c515dcb047cab42fac4c7699d292\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/cbc3335b07694d06913e9c557f81ca4b : SUCCESS in 4m 35s\n- requirements-check https://zuul.opendev.org/t/openstack/build/85a36899a4874b23be9075c9c3356307 : SUCCESS in 4m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f89d2abd2e334933894e2194dc831267 : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f02fd89431c420a8228e4a34e3f761d : SUCCESS in 4m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ee9fe5e8120c4da1b51cff491dece1fb : SUCCESS in 4m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6a4872acd6734148abb4f62c03ff3598 : SUCCESS in 4m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bafeeda5139d4affa6b648771f0cca10 : SUCCESS in 7m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e51e3d1e74854b588d2ef735160ccc15 : SUCCESS in 8m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d97d56e60d154744a355668a4be10c5c : SUCCESS in 7m 30s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/ebb528348fa0430b929d3338df991eef : SUCCESS in 58m 27s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/fec0a68a267b4b0d9517a7a338392e0f : SUCCESS in 2h 00m 18s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/825d0db502544049ba0947fbda23b0df : SUCCESS in 1h 31m 22s (non-voting)\n- trove-tempest-ubuntu-ssl-mysql https://zuul.opendev.org/t/openstack/build/cc99affa22a34ae0b74f22f193fe3bac : FAILURE in 56m 44s (non-voting)\n- trove-tempest-ubuntu-ssl-replication-mysql https://zuul.opendev.org/t/openstack/build/d7ca9dae8c934093af972dcdd78f1807 : FAILURE in 1h 04m 13s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/1fc6bc952e2c4525b807f3f6d47970a2 : SUCCESS in 1h 48m 07s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/f3b10f72a2ba47209de7539b264a995a : SUCCESS in 48m 08s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/21f76cf1e10040ffb9fb56b57a4e6502 : SUCCESS in 1h 35m 18s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/c89ca2e758c443c9bbb23d5e87e06ad0 : SUCCESS in 1h 19m 49s (non-voting)\n- trove-tempest-ubuntu-ssl-mariadb https://zuul.opendev.org/t/openstack/build/8cc39791e8474800841ed4256047b981 : FAILURE in 46m 07s (non-voting)\n- trove-tempest-ubuntu-ssl-replication-mariadb https://zuul.opendev.org/t/openstack/build/252733a33522466e81b29f6a232b288b : FAILURE in 55m 46s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/b176c4542c7240ae912c65584ea3bd71 : SUCCESS in 1h 41m 20s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/7a8e54893fe844fda15d67684a4d422b : SUCCESS in 50m 19s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/26ba4cca33524003a032f8595a5438bd : SUCCESS in 1h 42m 02s (non-voting)\n- trove-tempest-ubuntu-ssl-postgresql https://zuul.opendev.org/t/openstack/build/fae0a9e104b04d8e8b43756ac01672c3 : FAILURE in 50m 34s (non-voting)\n- trove-tempest-ubuntu-ssl-replication-postgresql https://zuul.opendev.org/t/openstack/build/6ccd15e166ff4438a90d46d7771c914f : FAILURE in 55m 35s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/6762df0b98864e1b8ab21f83553bad76 : SUCCESS in 1h 22m 06s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/33a4056584a74b34b8d6e23cc0c7f284 : SUCCESS in 1h 16m 51s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/45c86de3cb1d46f6bb31ae1696c8a1fb : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":19},{"id":"e89e6b24ce71bc19647782b42a91e8c5386e2fbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-23 16:53:13.000000000","message":"Uploaded patch set 20.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":20},{"id":"1f174d21fe697cbe37ec2062eb146964bb2d13f2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 17:04:50.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0377b219694841358fd2c6afcf649aa3\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2c261d51be494132b67aa7f607528d1d : SUCCESS in 7m 53s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e6273595bab6484196d808336720f4cd : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ede643c574c54aa88374ae6461c38256 : SUCCESS in 8m 41s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"0b10ee4383559511305b70c582d3ec0f0be3f3cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-23 17:28:15.000000000","message":"Uploaded patch set 21.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR **is:MIN**\")\n","accounts_in_message":[],"_revision_number":21},{"id":"690adf6e5b6e0aaea9dbd5827493c6d3ed9dcabd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-23 17:34:41.000000000","message":"Uploaded patch set 22.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR **is:MIN**\")\n","accounts_in_message":[],"_revision_number":22},{"id":"93699dc6af43a27d40bc402886c15df91d86429d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 17:57:15.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8c1b68bc0e04114aa9d7ede384b0bac\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/33329a406ba34d238477fdd52b275ac4 : SUCCESS in 10m 09s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/83708edb9e9545338a75dae33fdd0b7f : SUCCESS in 9m 00s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f1a5abf3decd4175b96f14c63284dc83 : SUCCESS in 18m 26s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"59cbd1a5d0d30091a2aded108d730f0f043a4e98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 19:45:07.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8deb13630fcd474f96c6473e4246eb66\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c5cbcf5f97554585b8afbdffb5d44859 : SUCCESS in 3m 38s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8859d7fa07f946af9dc94dd1f8fc6c52 : SUCCESS in 4m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3557576350df403b98aa9ab913df66f2 : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d676c0748f484cf9afa49f091700e600 : SUCCESS in 5m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6410ae30ed794ad9aa8c756a70ebdde2 : SUCCESS in 6m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7295b5f93753438f92d5d4edb1e0dda0 : SUCCESS in 5m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/829ad1168014466280501b4c8cab8ab3 : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/902e0ffb998a458fac30202e1cc29d25 : SUCCESS in 10m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3677b813b145418ca7faa4b4f95be519 : SUCCESS in 7m 55s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/51ccd2cb3062426b8bdfe76e7d6a4096 : SUCCESS in 56m 47s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/c6dbc0f492ab400e94493ce9f84da82a : SUCCESS in 2h 07m 25s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/d6d76c61f1154f96bf32fd7eae2b37a1 : SUCCESS in 1h 40m 44s (non-voting)\n- trove-tempest-ubuntu-ssl-mysql https://zuul.opendev.org/t/openstack/build/69147a5e65dd40678243e88bd5de31a5 : FAILURE in 57m 00s (non-voting)\n- trove-tempest-ubuntu-ssl-replication-mysql https://zuul.opendev.org/t/openstack/build/dbe5503f2fd043e5a179784c2e498048 : FAILURE in 1h 08m 08s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/a5331556229346fa949330a70a09d564 : SUCCESS in 1h 53m 32s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/255b558cc584465e93dc3018de6cfeef : SUCCESS in 46m 36s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/cf08debd33284392ae70d95efa6f72d8 : SUCCESS in 1h 45m 37s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/46f37ce3756143efaae9160755a19fc5 : SUCCESS in 1h 25m 19s (non-voting)\n- trove-tempest-ubuntu-ssl-mariadb https://zuul.opendev.org/t/openstack/build/e8147e46aeb14545acf48a9b7ecce7f1 : FAILURE in 47m 26s (non-voting)\n- trove-tempest-ubuntu-ssl-replication-mariadb https://zuul.opendev.org/t/openstack/build/5d03f9a65f4a4eab9658def79b97446e : FAILURE in 55m 21s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/4ff30431541c437ba5bd7533d6ddac0b : SUCCESS in 1h 47m 07s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/963e1fd18a8c4fa2a6b9ef15631eb67d : SUCCESS in 54m 33s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/8c9a60e9ab9348b8b905f1784d4dd30f : SUCCESS in 1h 45m 20s (non-voting)\n- trove-tempest-ubuntu-ssl-postgresql https://zuul.opendev.org/t/openstack/build/83a1a76303f044969d7ea7489e017f59 : FAILURE in 52m 38s (non-voting)\n- trove-tempest-ubuntu-ssl-replication-postgresql https://zuul.opendev.org/t/openstack/build/2382a8c6c10048d89ee866b72c454846 : FAILURE in 1h 00m 09s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/892b3638e5d1441f98f911b4bd7097f7 : SUCCESS in 1h 27m 13s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/b45ae627555e4fbcaca3986d94c829ab : SUCCESS in 1h 18m 11s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d486c69bbcc14b7cbddbd78e509e2a87 : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":22},{"id":"b620b7c06f2774657f459a1423b6a9d4c60a75d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-24 06:25:58.000000000","message":"Uploaded patch set 23.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":23},{"id":"aa5ba8c82583da51c30867e78b88d64af4602974","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 06:41:13.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7c455afc7114e76b5ab2b0e8f2f6985\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6ed63077669e47aa9ae489375afc9b32 : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e5c266b9a9bd414abe599e7a59280b4a : SUCCESS in 7m 51s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/bee434bacce94ffca9212c8ddbc80158 : SUCCESS in 11m 47s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"45c744797952a9afc61d1c5d6e09236bc18dbed1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 07:31:34.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a7ccd21c0fd54e29af816cfe512157f6\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/eec1f6a5aceb46989b5fff3077b5d5b7 : SUCCESS in 3m 01s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c68639de841145c9b92c8e7327c26add : SUCCESS in 4m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fdb2d254fecc4db8bfff981bfac931ce : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e258ac06d6d4cbda1aee7b5e36943ed : SUCCESS in 5m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/46c86fdc292a495fb1cf4a3e2b09577e : SUCCESS in 3m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/145cf6a18e0242488ccc972f9369bff9 : SUCCESS in 2m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7e98303c9a07498b9af948655bcc1335 : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30a29ebb333d45c4b7e0ab89fda74770 : SUCCESS in 10m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c82a0e8c9d56465ba9eeffe299657d8e : SUCCESS in 3m 28s\n- trove-tempest-ubuntu-ssl-mysql https://zuul.opendev.org/t/openstack/build/e93c2aeac34b4931a5d6fb31329914d8 : FAILURE in 55m 23s\n- trove-tempest-ubuntu-ssl-replication-mysql https://zuul.opendev.org/t/openstack/build/04c25958484d4682bb28dd004bbadeec : FAILURE in 1h 04m 00s (non-voting)\n- trove-tempest-ubuntu-ssl-mariadb https://zuul.opendev.org/t/openstack/build/040f6233d8ac4f52a5d3886bc1c4fc2f : FAILURE in 45m 56s (non-voting)\n- trove-tempest-ubuntu-ssl-replication-mariadb https://zuul.opendev.org/t/openstack/build/17a087f9067a4b4abf9284f881f9a111 : FAILURE in 55m 34s (non-voting)\n- trove-tempest-ubuntu-ssl-postgresql https://zuul.opendev.org/t/openstack/build/6faf48dd40ba4d959517f92b30faaf47 : FAILURE in 49m 13s (non-voting)\n- trove-tempest-ubuntu-ssl-replication-postgresql https://zuul.opendev.org/t/openstack/build/b869e05dd7b340d8bbb1cd9efdc5a8bc : FAILURE in 52m 22s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8c6413acd0e543e5baf7ca1f79d730c8 : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":23},{"id":"108a49da638784699feea14d131ba3864d8f3504","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-24 07:37:05.000000000","message":"Uploaded patch set 24.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":24},{"id":"62db2424a01a3744be7a12e3cf8897d038acff94","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 07:50:31.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f88a53f0256418185be374b8e352c91\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c5465260c08e489e9e2111ca9bd7767f : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/baf82279202e4f789d95ec56a59b7e15 : SUCCESS in 8m 02s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e3a069ea49fd4979b0ea754bc79e6177 : SUCCESS in 10m 39s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"eb5b6028132dce7bd405b29957c485444a6de60f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-24 08:27:51.000000000","message":"Uploaded patch set 25.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR **is:MIN**\")\n","accounts_in_message":[],"_revision_number":25},{"id":"76b6de90cf6418e0298781af7f8b982f50fc8a0d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 09:25:21.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df5db4c1debf4179988715054d79ffd0\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5527332e0b4b445a9a25c40dc7e5d742 : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2efed3fee39f4cb2a12d3885686e8e79 : SUCCESS in 7m 39s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/60d071936ef54693a62f706294d74ab2 : SUCCESS in 15m 55s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"8c0c31fe27706375ed72b81dbb003a2504a1cdbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 10:05:14.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d531fdc1bdea4286a12b2da55ddc4cfc\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4815924a2f7b45a89d705fce110d4a84 : SUCCESS in 5m 09s\n- requirements-check https://zuul.opendev.org/t/openstack/build/527698b1bd8842fead904514c0190641 : SUCCESS in 3m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8042aa86b9ca42e8be5261fd6a90c4ce : SUCCESS in 8m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6064fff6b7940ee9146fcbcdcf90ba3 : SUCCESS in 6m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/95d14e4e76334d7eaeaf765d2b150bb8 : SUCCESS in 3m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/31b68a6861f14df3b9775e58d6c434ec : SUCCESS in 5m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e3fdc251f47d479598333d52f62baa11 : SUCCESS in 10m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51d974162b0c43c7849597fdec997898 : SUCCESS in 10m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/253f08e1ec3a4aaaa45227a8d862c946 : SUCCESS in 7m 50s\n- trove-tempest-ubuntu-ssl-mysql https://zuul.opendev.org/t/openstack/build/e8dab672f32f47879f28ede95a7861e5 : FAILURE in 1h 17m 52s\n- trove-tempest-ubuntu-ssl-replication-mysql https://zuul.opendev.org/t/openstack/build/d4448e38022140f298d7fc582e75a71d : FAILURE in 1h 33m 58s (non-voting)\n- trove-tempest-ubuntu-ssl-mariadb https://zuul.opendev.org/t/openstack/build/0045554f7b7b473d98f0af30eebe642a : FAILURE in 49m 58s (non-voting)\n- trove-tempest-ubuntu-ssl-replication-mariadb https://zuul.opendev.org/t/openstack/build/75f0c3e7c8cc4612ba8373b3ef8d9953 : FAILURE in 53m 26s (non-voting)\n- trove-tempest-ubuntu-ssl-postgresql https://zuul.opendev.org/t/openstack/build/34db9d13eae5415d82c6652e8e063ce6 : FAILURE in 48m 48s (non-voting)\n- trove-tempest-ubuntu-ssl-replication-postgresql https://zuul.opendev.org/t/openstack/build/c782dca935bd4b4bb1d636ac326d2549 : FAILURE in 1h 26m 41s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/16260a3fb7e043bfa31964fe8ab8507f : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":25},{"id":"616464a88ca6471208b54f8ab05cb5dbdb7dd5b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-24 13:17:43.000000000","message":"Uploaded patch set 26.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":26},{"id":"c491bf24a3c59d163c9a19dc4bb08d56d1067e3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-24 14:12:52.000000000","message":"Uploaded patch set 27.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR **is:MIN**\")\n","accounts_in_message":[],"_revision_number":27},{"id":"127f7535d6908ee13fe0d59981dfa2a5f8bc70df","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 14:40:33.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/795850c616e246b2aa196ecb45bcea29\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f22c889371464cbfbcd6c0d67de30241 : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a1b2fc16319d481fa63191295405a9cc : SUCCESS in 9m 53s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4aa9fbd2d9144785959afa4280e0c6ec : SUCCESS in 15m 33s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"600d19a5a7137b4561dbc236e3c4575aec1ef4a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 15:14:52.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/adf421631133482cb6e21c84c7a5af24\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4b37c1f5f4c749c9b38169addd52a55a : SUCCESS in 5m 08s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e15cb8a7da5c4c628fc8f57f6530e346 : SUCCESS in 4m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/862d5b84e65b4ababcbce73978eff024 : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6fc964e03e14e85ac620352e582145d : SUCCESS in 8m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e41e0bf76b3d48fcb194c7c8f9c76ab5 : SUCCESS in 4m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/71c5a4ff102e467792f7b4467801d40f : SUCCESS in 4m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7e65cb0396934132a5e6a74e74211122 : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/961d70a9033240108c92c73aefcd4cb1 : SUCCESS in 11m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/336eaa18e0094bd694253cb971231215 : SUCCESS in 7m 47s\n- trove-tempest-ubuntu-ssl-mysql https://zuul.opendev.org/t/openstack/build/0985d870787c44d2aebb07a37fb8cb96 : FAILURE in 58m 28s\n- trove-tempest-ubuntu-ssl-replication-mysql https://zuul.opendev.org/t/openstack/build/5cf093f83e0e401f8a019d2c4f96123b : FAILURE in 54m 10s (non-voting)\n- trove-tempest-ubuntu-ssl-mariadb https://zuul.opendev.org/t/openstack/build/021aa41665184b25a80aea92197a06c0 : FAILURE in 42m 10s (non-voting)\n- trove-tempest-ubuntu-ssl-replication-mariadb https://zuul.opendev.org/t/openstack/build/7933cd70570a46c4a34a6c876e903bed : FAILURE in 49m 31s (non-voting)\n- trove-tempest-ubuntu-ssl-postgresql https://zuul.opendev.org/t/openstack/build/3d84f8a0b0434504b823f0208c52fde1 : FAILURE in 46m 04s (non-voting)\n- trove-tempest-ubuntu-ssl-replication-postgresql https://zuul.opendev.org/t/openstack/build/93b07f8ae48f4feb9e4a6c00f02b92fe : FAILURE in 54m 41s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fdd911a1a9b04a88b76ea673653daa38 : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":27},{"id":"0225065fb5cebcda3fd514198ca55cd7dccf89ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-24 15:52:54.000000000","message":"Uploaded patch set 28.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":28},{"id":"9ae5270b28a454515f0a4b017882cc750a891012","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-24 15:56:56.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":28},{"id":"84aa1c08981652876175330406ee9226239e34d6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 16:52:50.000000000","message":"Patch Set 28:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e907644ebb54415587b878ec709b7f83\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4418b52665154a74b00c732c5a4c6214 : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/10c4183e5cca434d90c33e559a1c1677 : SUCCESS in 8m 34s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/33e7c1e4fcc54805b70a12f1b51c4966 : SUCCESS in 13m 13s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"8ffe5234bcf8a17af126946b02467cab3f595737","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 17:10:01.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/07b71670be744b41a6f7e813d4933fb6\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/fe82a89f88924ba59132f4ce26bafeb1 : SUCCESS in 5m 35s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3be82e2e04c0491a92a847962ef532d4 : SUCCESS in 4m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d4842ab7e0b44fd89f4131b0382db5cc : SUCCESS in 7m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/037685914e7c430fbf9e0b01e134cb06 : SUCCESS in 3m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9fff3b67495844df856cc90e6ca6c386 : SUCCESS in 4m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/55d968a39cda4c77b2c4d6c0db5ecfa8 : SUCCESS in 5m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fae3e2d91ac84641a77b0e5ae14f8e2b : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/556e4575a3594505b8446ba38f9b01df : SUCCESS in 4m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb4ee86c87eb492a854b45cb5c0ac00b : SUCCESS in 8m 00s\n- trove-tempest-ubuntu-ssl-mysql https://zuul.opendev.org/t/openstack/build/c315b8589bc44f87a5f4a0e3d568a8f7 : FAILURE in 42m 30s\n- trove-tempest-ubuntu-ssl-replication-mysql https://zuul.opendev.org/t/openstack/build/efc40e0f9c244195902704fa320bdeb7 : FAILURE in 1h 12m 40s (non-voting)\n- trove-tempest-ubuntu-ssl-mariadb https://zuul.opendev.org/t/openstack/build/40fbd04283d14cbea1cb56ca5ba70fb0 : FAILURE in 38m 14s (non-voting)\n- trove-tempest-ubuntu-ssl-replication-mariadb https://zuul.opendev.org/t/openstack/build/49423a9d558f4b6cbbb78b562b94cbdc : FAILURE in 40m 57s (non-voting)\n- trove-tempest-ubuntu-ssl-postgresql https://zuul.opendev.org/t/openstack/build/97942f9a65594acc9ab385fe3397bd47 : FAILURE in 56m 46s (non-voting)\n- trove-tempest-ubuntu-ssl-replication-postgresql https://zuul.opendev.org/t/openstack/build/cd9a0778953d4d349b7b5cd68ebf07c2 : FAILURE in 1h 06m 25s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0f8f839ddd8541c48857a1141c59340e : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":28},{"id":"bfe05d430ae1cb6d44a84d2e18002a72d68938be","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-24 17:48:44.000000000","message":"Uploaded patch set 29.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":29},{"id":"4317744000f7927568ec47698570672fa1c73931","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 18:19:44.000000000","message":"Patch Set 29:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe8a4941103645fb81f5be55e73cae43\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/24217af5048240e5955519b491a77dfe : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/799c9a96bda0482b883486a2dffdbac6 : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d0f193ebb7214ac3bbfa244fe63a0344 : SUCCESS in 10m 25s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"46fb15ac0bd462245d9f467f2007b23e6f283903","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 18:42:04.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9b1787d80e2743888cebcc7bea557883\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/cd2b53d92c3e47cb836e7eecc657fcd6 : SUCCESS in 5m 13s\n- requirements-check https://zuul.opendev.org/t/openstack/build/993e32c27c414e0d860b679dce189fb5 : SUCCESS in 2m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee983e0ab4184e02b7a8563dc4d7cafb : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c693ed6157545ab8db2652a37363422 : SUCCESS in 3m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6a7f1f847e294039874d2959f4f66b18 : SUCCESS in 4m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6551c9832f354e72b8253adc01bb7b50 : SUCCESS in 3m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5b1a16a222d7470d88920b0137d5c1d8 : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85c823b73efd45758f845d2d31799641 : SUCCESS in 5m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e4a473ee280490a8bd8e5a93a3eb477 : SUCCESS in 6m 31s\n- trove-tempest-ubuntu-ssl-mysql https://zuul.opendev.org/t/openstack/build/5f265d5c02504c01be35e99b22f271ea : FAILURE in 45m 26s\n- trove-tempest-ubuntu-ssl-replication-mysql https://zuul.opendev.org/t/openstack/build/c23dede20fd5480db2fb132d6f95e2ce : FAILURE in 50m 34s (non-voting)\n- trove-tempest-ubuntu-ssl-mariadb https://zuul.opendev.org/t/openstack/build/0e362fab979347db9e959a39588b943a : FAILURE in 41m 05s (non-voting)\n- trove-tempest-ubuntu-ssl-replication-mariadb https://zuul.opendev.org/t/openstack/build/28569a5eba0b4775bce5be0257c91a6f : FAILURE in 46m 58s (non-voting)\n- trove-tempest-ubuntu-ssl-postgresql https://zuul.opendev.org/t/openstack/build/11dda853bae546499e8f5fa50b680d34 : FAILURE in 35m 57s (non-voting)\n- trove-tempest-ubuntu-ssl-replication-postgresql https://zuul.opendev.org/t/openstack/build/c4531ec3f92447a18d3dab2b8b9dc379 : FAILURE in 40m 49s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8ce0fdf701f34b4f912569d8ec9ba291 : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":29},{"id":"3df3117b9b6f0c715d5deeca76668b5c9d5a678f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-25 07:03:10.000000000","message":"Uploaded patch set 30.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":30},{"id":"f611df71e687c7531ca3f43e32c45ad541f4ad56","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 07:22:08.000000000","message":"Patch Set 30:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65f3e989491f4dc49e3f9937097615af\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/46418095af3d4581ba8ffbdced2407ac : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/18ba21c4395a49b3a1ee61e5020d89b4 : SUCCESS in 9m 16s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/1b6518dbf82f42a69bd3a773b2b16851 : SUCCESS in 15m 31s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"ebb22f9739eaadced7c52580e6b48021e08cfdcd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 08:16:17.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99abc6a3284344ffa84b971ad6f0e682\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e69ab70e835547dca53174078930ebc6 : SUCCESS in 5m 07s\n- requirements-check https://zuul.opendev.org/t/openstack/build/56615117d48c48ef9bef173f9750d1eb : SUCCESS in 3m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5d0121aba9f4270985ba398405640de : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bed32ff0ec8f4afea423b1ecfaf09e21 : SUCCESS in 3m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b5df5597e91b444089620ff70cd8b108 : SUCCESS in 4m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/491e3f142a1b4701b6c640cb08988432 : SUCCESS in 4m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6911d03230cc49d49c7e52240158d480 : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75e585fb27c44fb2915d5fa3a127bc81 : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f2ef8cb92714701bfb2123ec4b2a096 : SUCCESS in 7m 16s\n- trove-tempest-ubuntu-ssl-mysql https://zuul.opendev.org/t/openstack/build/2f216c83ba28465d9f5ea4cf98c8c104 : SUCCESS in 48m 56s\n- trove-tempest-ubuntu-ssl-replication-mysql https://zuul.opendev.org/t/openstack/build/52201404c7a84574b51a51a3e38c0f17 : SUCCESS in 1h 11m 09s (non-voting)\n- trove-tempest-ubuntu-ssl-mariadb https://zuul.opendev.org/t/openstack/build/2a6ac65193d141ac8e17c8bc2bfef5da : SUCCESS in 41m 51s (non-voting)\n- trove-tempest-ubuntu-ssl-replication-mariadb https://zuul.opendev.org/t/openstack/build/3d45f070072640d8b49d5911971cb99b : SUCCESS in 1h 01m 07s (non-voting)\n- trove-tempest-ubuntu-ssl-postgresql https://zuul.opendev.org/t/openstack/build/4d5e578349904e7f9cf246b802757082 : SUCCESS in 36m 53s (non-voting)\n- trove-tempest-ubuntu-ssl-replication-postgresql https://zuul.opendev.org/t/openstack/build/b945bf0bf34d4d96869d571fd86ba39a : SUCCESS in 54m 49s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b9a3513ee43b42a4b0048d3e7b955e82 : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":30},{"id":"b0cf86610bafe57c8506dab4e38c22d990b7299b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-25 08:23:26.000000000","message":"Uploaded patch set 31.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":31},{"id":"222e6d1a288585c1dd9508628c87f29ce301547e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 08:42:17.000000000","message":"Patch Set 31:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/245f74a30f064f09a0f3561768db7c2d\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/99949edace6a4ca196d846061c22cff8 : SUCCESS in 9m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ae63182acfd648ee9188199c74286d8a : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b36c47e314f04b5db7fd77b7e418ddcd : SUCCESS in 14m 19s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"d7fe12b8918edf80c71e0dd2bc295e14a522a6b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 09:52:11.000000000","message":"Patch Set 31: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/820e5978f11a436aa6e0b7c896966a34\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/67f8928051c749fcb005214e6675fcad : SUCCESS in 3m 45s\n- requirements-check https://zuul.opendev.org/t/openstack/build/367a2ee6babe4b3fbbf0fdc328bc283c : SUCCESS in 3m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9292c16f3b974804be410ca7cd6c49dd : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44aeb5a354c448bfa2ba76abf9680df0 : SUCCESS in 5m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4fbf59a7689244dfae0f49dbe00f0cc1 : SUCCESS in 3m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8f489156383a4283b26bb9cda290fd2e : SUCCESS in 3m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1bd34a01b8544933b6f5b5e81efc4f24 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff715f6616e44b9888bc9dfc46b17727 : SUCCESS in 10m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c78f1f1c6ab4eada3c7c5440d488478 : SUCCESS in 7m 27s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/4a71742c47094ef0af98295a0b47e894 : FAILURE in 31m 00s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/e7ea2360d88d4e138848b1c4c08b62ee : FAILURE in 32m 05s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/2eb7a3ec94224dc5b73b94a3d5243532 : FAILURE in 25m 48s (non-voting)\n- trove-tempest-ubuntu-ssl-mysql https://zuul.opendev.org/t/openstack/build/779b22feb48e4be8bb8726e9b3803a3d : SUCCESS in 59m 07s\n- trove-tempest-ubuntu-ssl-replication-mysql https://zuul.opendev.org/t/openstack/build/e963933ff9c94ada878f2d1423a94dcf : SUCCESS in 1h 26m 25s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/02fb08ffdfac46008d121139773f16c8 : FAILURE in 20m 27s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/2d82f5242c0c44769b0af3a1ba131d67 : FAILURE in 24m 59s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/99a5419835e6410f871ba13e57b34db6 : FAILURE in 26m 18s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/356c4b7f0f77469a9425ae1d874755f7 : FAILURE in 25m 40s (non-voting)\n- trove-tempest-ubuntu-ssl-mariadb https://zuul.opendev.org/t/openstack/build/4ee3754b1eff4b2dbf6ff8fef787509b : SUCCESS in 50m 27s\n- trove-tempest-ubuntu-ssl-replication-mariadb https://zuul.opendev.org/t/openstack/build/5fdf81aecc63431cac3e48fbd632af84 : SUCCESS in 1h 04m 55s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/b85e566293e94ebca3571910175188ed : FAILURE in 26m 17s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/5bef8bfbefdd41518e0701752af17bea : FAILURE in 30m 35s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/65c94efb32304823afabc3fb279d6ad6 : FAILURE in 23m 55s (non-voting)\n- trove-tempest-ubuntu-ssl-postgresql https://zuul.opendev.org/t/openstack/build/8fc1728259e8433a9488a32338280352 : SUCCESS in 44m 07s\n- trove-tempest-ubuntu-ssl-replication-postgresql https://zuul.opendev.org/t/openstack/build/84a7605c486647e8bc067936634d5f1a : SUCCESS in 1h 00m 04s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/434da57c4e2b4bae9b85fa3d5619b8b8 : FAILURE in 22m 22s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/6d6b93f4970446478f3a6694109e117d : FAILURE in 19m 34s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8ecaa20113c64c1b8324540956cc7ae7 : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":31},{"id":"13f96112fad47641a7e481d2dc57356596170524","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-25 11:14:02.000000000","message":"Uploaded patch set 32.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":32},{"id":"b825d848169f549efeddd887ea99afe389bae353","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 11:36:58.000000000","message":"Patch Set 32:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ee7576263f0411c8f1a0879636ab0b3\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/969ecbf755e84a21ae3594ab926d56fe : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4f2184832bed4bb485b9dffaccfc1dde : SUCCESS in 10m 06s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/db9ecab4a6094eb79fa7a5e474cb2aa0 : SUCCESS in 18m 49s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"3e5566f64748c815078c4d75932beb0e7cceb4d7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-25 11:49:47.000000000","message":"Uploaded patch set 33.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR **is:MIN**\")\n","accounts_in_message":[],"_revision_number":33},{"id":"902321702dac06566581249659e9c152eed4112e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 12:07:49.000000000","message":"Patch Set 33:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/671efbcdd69f4117ba9f2c4be2f0b192\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ba1a18c8561549f190e69726e58545c8 : SUCCESS in 7m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3aae05b3995b4049916f475b2721d46b : SUCCESS in 7m 53s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7967bdc8def3440aa4762785f76643d8 : SUCCESS in 13m 59s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"7e0f478d5e549b893db5af2357e94bd98f4ebcec","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-25 12:14:05.000000000","message":"Uploaded patch set 34.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR **is:MIN**\")\n","accounts_in_message":[],"_revision_number":34},{"id":"5c52fbc0e8147078c51623d605a8dbc8febcb10a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 12:29:14.000000000","message":"Patch Set 34:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61cf730d1efa4157b22a2346044e8bd1\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/de4735ccc0294b7fb8450a4ec076e15b : SUCCESS in 7m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/00f659e1dcd649f1b7fe6708ecd465cb : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/1c147227087741c4b85202d22aecce2c : SUCCESS in 12m 17s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"1f8d23556e38f57539638e73330a2a762fc4792a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-25 12:40:27.000000000","message":"Uploaded patch set 35.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR **is:MIN**\")\n","accounts_in_message":[],"_revision_number":35},{"id":"53c239f2ec9c0f745c9086fc4b5b03c93271db2a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 12:58:31.000000000","message":"Patch Set 35:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b7f36ff83804ddead5bf5d648faf26a\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/083988ccc6e140d59fc83afa937623af : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/485b591f86fd4d72b2f82702c73c54d1 : SUCCESS in 8m 44s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/cc57baf2830646789b29ca71f1bb9b82 : SUCCESS in 15m 05s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"1e73da07b580dcc71501008d7a8518ce70e961fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 15:10:26.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/876a65529d0c4252b22d8165faf8212e\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4f026a0330714f24a2081bbf47c6030f : SUCCESS in 3m 34s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8ad62a7f16b94b99925e02604312984e : SUCCESS in 2m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/29065c457e32439a86aef1b555e3ffda : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/961a47d85d6743139a1102ecf7dc9668 : SUCCESS in 3m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c56854aa52d5436b909d28e184ce7777 : SUCCESS in 4m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/97fe6dcfe2644b4f8c362f6e4a584c71 : SUCCESS in 4m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d21271c1221747688a3e49e41fa4b7ee : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5eb7fb17b91b4d09b142241f5d12b41a : SUCCESS in 6m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0fb5c43647c64f329693025f1da3a3cc : SUCCESS in 6m 31s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/95a33b4db71f4efb8797514d85427ff9 : SUCCESS in 45m 18s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/a6fb363f33a64177b38d9c8672e423c2 : SUCCESS in 2h 26m 12s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/9c1d1ecb87d94064af7b641044eef4d5 : SUCCESS in 1h 12m 39s (non-voting)\n- trove-tempest-ubuntu-ssl-mysql https://zuul.opendev.org/t/openstack/build/56d254cc11ee4af39c47f97611c381d7 : SUCCESS in 2h 01m 01s\n- trove-tempest-ubuntu-ssl-replication-mysql https://zuul.opendev.org/t/openstack/build/01686d1eb10f43ec8e4646eccaf91d83 : SUCCESS in 2h 27m 16s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/2b0196d68d6d41d0aaf49c7fac887376 : SUCCESS in 1h 44m 47s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/60af5811ca194d929ec466de84e40138 : SUCCESS in 47m 06s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/668e843237824db788427eb24e762353 : SUCCESS in 1h 24m 31s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/6a6be189114c49f3b492ebd79b889985 : SUCCESS in 57m 07s (non-voting)\n- trove-tempest-ubuntu-ssl-mariadb https://zuul.opendev.org/t/openstack/build/ccc50c4e0a8c416d937e7f2ff409b4ce : SUCCESS in 1h 55m 08s\n- trove-tempest-ubuntu-ssl-replication-mariadb https://zuul.opendev.org/t/openstack/build/608ca89cf9eb45b9844f1e26844f91ed : SUCCESS in 1h 12m 51s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/f0dd892c7ca74945bd28f8bad1b75589 : SUCCESS in 1h 43m 49s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/50586af8b15f4030b0f2aeaf85eaab00 : SUCCESS in 1h 46m 16s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/1b800e6671584eb287c979ed3cb2b1bc : SUCCESS in 1h 01m 28s (non-voting)\n- trove-tempest-ubuntu-ssl-postgresql https://zuul.opendev.org/t/openstack/build/988c747fd4624fd9b65d29993a1bffa5 : SUCCESS in 1h 48m 32s\n- trove-tempest-ubuntu-ssl-replication-postgresql https://zuul.opendev.org/t/openstack/build/86b3e7bc9dcc42a2bc683d30b68a99ed : SUCCESS in 1h 14m 48s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/36743108fbe74002a7aafe4b7ece15eb : SUCCESS in 2h 00m 51s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/a20f2bcb4cf84273aa1df24ac530da25 : SUCCESS in 2h 27m 16s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e07185933d25491f97ef8ce0e97c58d8 : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":35},{"id":"d926b83e4e6d43570919fd6293200baa8c46bad7","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-25 15:57:17.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":35},{"id":"e05a167db0201e0e9168990b48753f77719d2673","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-07-05 15:10:41.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"d18132fe258e0336c7aad1934e8f71ba27b2d9ac","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-08 05:52:29.000000000","message":"Patch Set 35: Backport-Candidate-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"46ee35ea89150541bebd1e68139370ff734a1f36","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-07-08 14:48:38.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"16b4cf556e13f0acac0421922afbe12fd7ae05ec","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-09 08:51:55.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"21809e21751b12c676aa209a2fefa6a1a6a20024","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-10 13:16:24.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"00c8a95eebebc9b95e1495f7b752b13f413b3e60","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 13:16:50.000000000","message":"Patch Set 35:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/trove for 994217,5","accounts_in_message":[],"_revision_number":35},{"id":"ef15249fb2377021c98db397042b6e99089148a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 13:17:58.000000000","message":"Patch Set 35: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/trove for 994217,5","accounts_in_message":[],"_revision_number":35},{"id":"e26a1fd131cc89fc8da8dc76a49ed227c7cee936","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-10 13:18:33.000000000","message":"Patch Set 36: Commit message was updated.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR changekind:TRIVIAL_REBASE OR is:MAX OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":36},{"id":"62c15bbae80a6f22aa8db3e7a4b0b8ff4141e32b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 13:26:00.000000000","message":"Patch Set 36:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/trove/+/992900 is needed.","accounts_in_message":[],"_revision_number":36},{"id":"cedf3cc466dd5d1252c4042700f14d8093c0ecd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 13:26:18.000000000","message":"Patch Set 36: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/trove/+/992900 is needed.","accounts_in_message":[],"_revision_number":36},{"id":"f58127a4afb2ef00cdbe4e41aaa75d3890b6c948","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-14 12:35:14.000000000","message":"Uploaded patch set 37.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":37},{"id":"6451d75199ab3d681915aebb15bf26279f4152d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-14 12:49:37.000000000","message":"Uploaded patch set 38.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR **is:MIN**\")\n","accounts_in_message":[],"_revision_number":38},{"id":"664541db8e68e832b32593da9ac95303c4e9f47b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 13:20:36.000000000","message":"Patch Set 38:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a0f7e0a04fa4c9ead4b3e4cee884ebf\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7338b09565b84a83bc6c168995682e03 : SUCCESS in 8m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ce14eb9dfe5a474fa0fa0eec948ae9f7 : SUCCESS in 8m 47s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/29ca716493734c989273404b7a451a29 : SUCCESS in 17m 17s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"be39a7d59091b06327d48605e7d63a5bd8d4dab2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 15:10:15.000000000","message":"Patch Set 38: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cc7a63d202a447292e1223f00db3ed1\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/de42bd05d6234a369dab2d61dbb1db93 : SUCCESS in 6m 18s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4c2c6f47fe7648068dd1a7caf31973f7 : SUCCESS in 3m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c7946a70a5d64dc8993f65060e9b9185 : SUCCESS in 7m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9fe8562dac24f97b165ed98d533a21e : SUCCESS in 5m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9b45985607a04c80806cefffeeecb793 : SUCCESS in 4m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bcae11eec4d043c7828a4af75fe80f2a : SUCCESS in 5m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6a14854afead439c8d9d59c90e331b72 : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08131b7b4ff44761b6b3838e198473c2 : SUCCESS in 6m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bdb034f3e10d4ff19898e2e262a35d43 : SUCCESS in 7m 48s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/a502e7f2b816413ca48bcaf89750c5c4 : SUCCESS in 1h 10m 18s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/e39187411e5a4a378ed5498245d37622 : SUCCESS in 2h 18m 26s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/13ee765813cb40e086668d02e39c66ed : SUCCESS in 1h 54m 42s (non-voting)\n- trove-tempest-ubuntu-ssl-mysql https://zuul.opendev.org/t/openstack/build/630c1ea1e985463893f5f648aff7d5f8 : SUCCESS in 1h 03m 19s\n- trove-tempest-ubuntu-ssl-replication-mysql https://zuul.opendev.org/t/openstack/build/f3e99525fb2b4fb0a86ed000ff6dd1b3 : SUCCESS in 1h 44m 17s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/0d5acd06bacd48d3a2536351f136914d : SUCCESS in 1h 46m 47s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/bfcf3c18de604e9b90a6480f902391b8 : SUCCESS in 45m 42s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/431401900198427893404a8892124df6 : SUCCESS in 1h 47m 20s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/581d09be891f4daba9c561cd0ac9ac68 : SUCCESS in 1h 36m 36s (non-voting)\n- trove-tempest-ubuntu-ssl-mariadb https://zuul.opendev.org/t/openstack/build/4a9a15d5d18e47258b8d9d117436b1ae : SUCCESS in 58m 08s\n- trove-tempest-ubuntu-ssl-replication-mariadb https://zuul.opendev.org/t/openstack/build/45c4c431b5da41fd9735091c2d3f96f6 : SUCCESS in 1h 18m 30s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/1bf1c729bf054ee19600387283ab9e84 : FAILURE in 1h 52m 43s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/1db8778dccbf4f32b51cbf4f4f1f7a93 : SUCCESS in 48m 20s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/82d8fcb8f95a4d3585bc2d8602fc3f0c : SUCCESS in 1h 50m 02s (non-voting)\n- trove-tempest-ubuntu-ssl-postgresql https://zuul.opendev.org/t/openstack/build/c8d90a0fa1e44274801b620558e0764f : SUCCESS in 1h 01m 39s\n- trove-tempest-ubuntu-ssl-replication-postgresql https://zuul.opendev.org/t/openstack/build/3238369df8e4488c911ac808a0f69f10 : SUCCESS in 1h 21m 38s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/ccd977624d9e48939ea0ec2d41c251da : SUCCESS in 1h 18m 39s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/fa792eb3fb42450187ee91171f652da5 : SUCCESS in 1h 52m 10s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a1506c1e1e9d41f3843dba68ae75edc3 : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":38},{"id":"db5b81b9855fd8fc44a7424099cbef536998c7ef","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-07-19 08:57:31.000000000","message":"Patch Set 38: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":38},{"id":"27d1528b28d8a54bc33fd4cb54d18f5e9e8a3e01","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-19 14:29:40.000000000","message":"Patch Set 38: Backport-Candidate-1 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":38},{"id":"ddc6e954e9bcb9d74edc3f9f650ef712a2a078cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-19 15:14:50.000000000","message":"Patch Set 38: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":38},{"id":"ad2eabaa66f851055f1126e4d5c7e17d4cfed872","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-19 15:26:40.000000000","message":"Patch Set 38: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/trove-tempest-plugin/+/988053 is needed.","accounts_in_message":[],"_revision_number":38},{"id":"c4c5142fabd6c90c83a2cdb692207fd8bcaa55b8","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-20 05:52:23.000000000","message":"Patch Set 38:\n\n(1 comment)","accounts_in_message":[],"_revision_number":38},{"id":"619dd098f05895e4669699d1b7d8bf146391e9a3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 06:03:03.000000000","message":"Patch Set 38:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9e29199a4f24221a2c8f948f6ebb78d\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3af554b8b7fd4c90add554602125c741 : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/14ce108b8141483fb633183e529a6391 : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7ee31f92097741038b901aefdd15d3c9 : SUCCESS in 8m 41s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"fce9b35aa8e0e6840720dc86a5ead686797240b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 06:36:58.000000000","message":"Patch Set 38: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/988a70c314874f259ce35879962e126a\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9d6af7ae02e6466da8a0f78449ef32ff : SUCCESS in 4m 56s\n- requirements-check https://zuul.opendev.org/t/openstack/build/415c00d2ca2a4bbe877133f84ad24c0d : SUCCESS in 3m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d51492d23f3463e81645472ccc9be99 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/928b0aab3b24494d826a81fb9eaf79a6 : SUCCESS in 5m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/028106d703d4486389be37477247b10d : SUCCESS in 4m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7ca584f5040a4c53a6ee2adb4a334e82 : SUCCESS in 4m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/14cadea81dd74952ad5cf6d1242a6272 : SUCCESS in 8m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eaba909653694fd0a6f6d902c361e013 : SUCCESS in 9m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2be8e93f317e4a60b8bac0f481e402f1 : SUCCESS in 7m 16s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/b95df0a3bce74e05af5983b558026a73 : FAILURE in 30m 07s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/21d9f1e5ff9b43c0bc20946bb0533121 : FAILURE in 37m 26s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/90e5283a98d743a2983d22fa7ac77b0d : FAILURE in 36m 34s (non-voting)\n- trove-tempest-ubuntu-ssl-mysql https://zuul.opendev.org/t/openstack/build/0caced7874c145fd954a9f6d58f54418 : FAILURE in 30m 43s\n- trove-tempest-ubuntu-ssl-replication-mysql https://zuul.opendev.org/t/openstack/build/62a4f3bbb942436ea5418e969a6026e3 : FAILURE in 37m 34s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/7f588312c1f04638ba448e1ab8c19f4d : FAILURE in 27m 21s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/55dec9d59a15465b9956b8921bbd1abd : FAILURE in 28m 26s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/48678129bb224253a5c32f95a9926fb0 : FAILURE in 37m 24s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/a9f3ff185baf4ac5b6114d8c2acba101 : FAILURE in 36m 17s (non-voting)\n- trove-tempest-ubuntu-ssl-mariadb https://zuul.opendev.org/t/openstack/build/99fe4e3ef3df4738a1fc3975765efe0b : FAILURE in 31m 16s\n- trove-tempest-ubuntu-ssl-replication-mariadb https://zuul.opendev.org/t/openstack/build/4296f41d1cdc445194574c0f4219ca4a : FAILURE in 37m 11s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/aa51d1cfb4af4577a17efb1f29b92424 : FAILURE in 30m 26s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/b0f6b795aecd4dca97ed1a6426c73f24 : FAILURE in 43m 26s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/eaddf9393a0e4eae90726e180e3d2bb8 : FAILURE in 36m 34s (non-voting)\n- trove-tempest-ubuntu-ssl-postgresql https://zuul.opendev.org/t/openstack/build/4a7370f0a17b4091a5a01e4f72dec99c : FAILURE in 30m 03s\n- trove-tempest-ubuntu-ssl-replication-postgresql https://zuul.opendev.org/t/openstack/build/6e1a1eaf10674334b2c7682e1e41ce7b : FAILURE in 35m 54s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/7a6c0db149444c4d9b156d8c77adbd27 : FAILURE in 36m 11s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/ea2a445080904660be0062f8de75b723 : FAILURE in 30m 26s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6217e89212c04566b1833a745f6f2449 : SUCCESS in 6m 07s","accounts_in_message":[],"_revision_number":38},{"id":"74cf13c986047bb85a72915206380d25ead458a2","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-20 07:04:24.000000000","message":"Patch Set 38:\n\n(1 comment)","accounts_in_message":[],"_revision_number":38},{"id":"58272eee22942c586864c5bd0d665863f1b6e81f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 07:19:42.000000000","message":"Patch Set 38:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d526224906f49c183a2823217767806\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bda4aba2b06a43e6ad1973727c023a05 : SUCCESS in 7m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/00156e03215e41f3b7fbc49741e672e1 : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/47965201351d4d2a840a5b263073f06d : SUCCESS in 8m 35s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"09e012dfd22bf63d4c761bdc18061858c783b5b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 09:58:29.000000000","message":"Patch Set 38:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/338a853ab47e4155ad3d712848679ed7\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e100e339c86d47dfbd390526d770ca5e : SUCCESS in 4m 52s\n- requirements-check https://zuul.opendev.org/t/openstack/build/08238c044eb5454bbe30ac497ddd192f : SUCCESS in 3m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/728549fea5794bb9be85f03b744804bb : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd89d3217f95495d8baa1d1053b4535a : SUCCESS in 5m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/65e7cebcffdc48cdad878b76ee0b73b6 : SUCCESS in 4m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9282a1ed51cd40e4901a0ff5f5568447 : SUCCESS in 4m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ae734306262b45f4a9c57d4e03765b2b : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/809fcd4563e848b89e3b862cee00be51 : SUCCESS in 9m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c20a471db32e43d58c96317c76e59343 : SUCCESS in 8m 01s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/b065bff02b604b52aed0f9371432b34f : FAILURE in 1h 15m 10s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/09918e26837740278a4723dc2733d6aa : FAILURE in 1h 30m 17s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/f8f50bf1c99a4286ba96e82b3bbb484e : FAILURE in 1h 24m 00s (non-voting)\n- trove-tempest-ubuntu-ssl-mysql https://zuul.opendev.org/t/openstack/build/d29eaccfdf864bf3bd8fe0af854bdd0c : SUCCESS in 1h 49m 04s\n- trove-tempest-ubuntu-ssl-replication-mysql https://zuul.opendev.org/t/openstack/build/220cb922ec9743d58bd6f9d21d7a5f10 : FAILURE in 1h 32m 58s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/1d7d02cf949640d1bb8f9b258a072ec0 : SUCCESS in 2h 40m 53s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/856180148c26404f8f6dfc2c10f950fd : SUCCESS in 50m 59s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/d44f8a9573374ef39a7d66fa935a5104 : FAILURE in 1h 00m 10s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/44386eb59b0346079b943a2699c9e736 : SUCCESS in 2h 14m 12s (non-voting)\n- trove-tempest-ubuntu-ssl-mariadb https://zuul.opendev.org/t/openstack/build/72d8cb6e2b414e0ebbcf254c0e413d22 : SUCCESS in 1h 14m 22s\n- trove-tempest-ubuntu-ssl-replication-mariadb https://zuul.opendev.org/t/openstack/build/3ca2756f4f8f4c7bbcbacf95144b58d6 : SUCCESS in 1h 47m 25s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/79754a3ad9dc457ea5b19c0677717c81 : FAILURE in 2h 28m 38s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/e995d54f9826496db215a244e96307f0 : SUCCESS in 1h 01m 46s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/f3da8e73628f4d669b6ef97902e3c576 : FAILURE in 1h 28m 24s (non-voting)\n- trove-tempest-ubuntu-ssl-postgresql https://zuul.opendev.org/t/openstack/build/5347401ca91344b388e35210f291c2a9 : SUCCESS in 1h 22m 43s\n- trove-tempest-ubuntu-ssl-replication-postgresql https://zuul.opendev.org/t/openstack/build/9a4aee5712044c07a7c4941b2a9b8419 : FAILURE in 1h 55m 42s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/5dd22c4522544765b558683285f82240 : SUCCESS in 2h 14m 10s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/9d3f593f72dd413abd0ad6a524d97804 : FAILURE in 1h 50m 26s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/779ee95696b54b35b83fe7ac479fa96f : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":38},{"id":"260956c8df5db2be512b73e3e3e478281a241ba0","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-20 14:19:36.000000000","message":"Patch Set 38:\n\n(1 comment)","accounts_in_message":[],"_revision_number":38},{"id":"b9f1f767b34810713879289559fb6142a5b28731","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 14:49:55.000000000","message":"Patch Set 38:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18b73e55cee64071bf4dadf1d927f4f4\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/67bd4d811dd948138bebd17f5ba186e0 : SUCCESS in 8m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/fe43b827f673407694807d4f9dba1bb7 : SUCCESS in 8m 36s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/118292e2576d44018ef7780938bb4df7 : SUCCESS in 16m 32s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"36758fd86d7ea37dc162b90d12130ff81df26620","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 16:14:41.000000000","message":"Patch Set 38: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d02011be1e04f14a762ce115c3d1553\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ba63eb993b7347b2ad74104920aef9b4 : SUCCESS in 5m 22s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c0cea63cec0c45598143f2b86ad446ae : SUCCESS in 3m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9f03a84cf744512bfd980a76034f105 : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfe06a2406514bcab87fac3a05d99439 : SUCCESS in 6m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1be6e6b8ef13457ab9eaa3a4086dbe9e : SUCCESS in 5m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/127df140bcf04e03902b19e2861c8ca2 : SUCCESS in 4m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3e8f73ca796d4f45be0d28e9e187a8a2 : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29f48d749f0445238e125270dc3372b8 : SUCCESS in 10m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c06f397dc67145f2b9574cbf6e4f154a : SUCCESS in 7m 18s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/2d06793135fa46c8ab0954a094e9ea85 : SUCCESS in 52m 43s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/6fca0cec5fed4c23883d67b8de358ba4 : SUCCESS in 1h 52m 46s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/643f501337574decba4f47c4ad874651 : SUCCESS in 1h 35m 48s (non-voting)\n- trove-tempest-ubuntu-ssl-mysql https://zuul.opendev.org/t/openstack/build/e2936356ccda4aa98455268b9a5eaa1b : SUCCESS in 1h 08m 43s\n- trove-tempest-ubuntu-ssl-replication-mysql https://zuul.opendev.org/t/openstack/build/036acdcf6e324ce699e3c6a666d585a4 : SUCCESS in 1h 45m 04s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/d2ac6467bad24419abf885ea27930185 : SUCCESS in 1h 52m 12s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/5ea1b39cf7f6473c86916df848c4ad6c : SUCCESS in 43m 41s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/420bd2ebc9f6465b995ad7e8e6a2c31a : SUCCESS in 1h 53m 13s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/1e8cc2dfe7794b379ff9aafccb165097 : SUCCESS in 1h 20m 21s (non-voting)\n- trove-tempest-ubuntu-ssl-mariadb https://zuul.opendev.org/t/openstack/build/88d96946e5e9464984f828a834e3b465 : SUCCESS in 56m 05s\n- trove-tempest-ubuntu-ssl-replication-mariadb https://zuul.opendev.org/t/openstack/build/f45e43e35ee74871a8e4360fa1a063f2 : SUCCESS in 1h 18m 57s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/17438a7a0b0a435c93de14a61fb58d08 : SUCCESS in 1h 34m 08s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/cb9b2b94d8904ee8bedb723bcd01c745 : SUCCESS in 46m 38s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/dbfae936ca6f4575b0dd4011c00f86ec : SUCCESS in 1h 53m 31s (non-voting)\n- trove-tempest-ubuntu-ssl-postgresql https://zuul.opendev.org/t/openstack/build/6a8c74aa1f644e95839c90d6392cae2e : SUCCESS in 1h 24m 33s\n- trove-tempest-ubuntu-ssl-replication-postgresql https://zuul.opendev.org/t/openstack/build/457d0fd01429432ba1f7eb0d8746842d : SUCCESS in 1h 49m 16s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/a6ce501059d64c51bd7cdd1fd532c48a : SUCCESS in 1h 17m 55s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/e17aeba0ceb146498b30f49a9d7e8273 : SUCCESS in 1h 19m 51s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6137b019eb7648e68a48dc97f985a0bd : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":38},{"id":"227f6b7356540cfef2edaadb95dcb12219a32df2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 16:15:26.000000000","message":"Patch Set 38: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":38},{"id":"0224f91b5b54dbca3e324d4a7e33c704480552b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 17:26:08.000000000","message":"Patch Set 38: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/532f6cc38d4f4e7cb7cfaea7b5e6b378\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c762bf2f0a2f4c96ae8f5ff227350423 : SUCCESS in 7m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/da6d0251da0c4dfbae1f86d22f9f0e51 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/242dc27592ce4e0f9071e620b58938c6 : SUCCESS in 5m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2cde901857654a36895b9051a3ebb671 : SUCCESS in 5m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6fe937200a804e4c94c57b48c510daa1 : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79b9d28e2e374da6b64a6902bfceb046 : SUCCESS in 10m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/710ddeef65dc47268e1778c12dae4cba : SUCCESS in 8m 04s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/04dd2bfcef4c4c82b7ca1da41fe5a730 : SUCCESS in 1h 03m 16s\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/6f7681ae807f4d7480a8aa066c2a4d1f : SUCCESS in 1h 08m 44s\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/6a9d6fdcac35462f9c9fef189b10e874 : SUCCESS in 56m 25s","accounts_in_message":[],"_revision_number":38},{"id":"93d7124a010b793feccb0a141fe192add68ba814","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 17:26:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":38},{"id":"4fbc5a9d3da1ea5547a558d52e1495ce160f0ade","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 17:28:19.000000000","message":"Patch Set 38:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c6b3b5191754659b27b45d15f04a090\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/bcdde344c31642bc8d2ef479d662bdee : SUCCESS in 45s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1d6012d91b14153bc4f65dd6905399a : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0ff538b6386427786c74d9a825e5384 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":38}],"current_revision_number":38,"current_revision":"729f614a73d426c9f2bd6162c4a6ab868b203dc8","revisions":{"7f62adcc77232f394673b2315a134314a61a689f":{"kind":"REWORK","_number":1,"created":"2026-05-11 09:07:13.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/59/988059/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/59/988059/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/59/988059/1"}}},"commit":{"parents":[{"commit":"01d94858288755b725293d19ece02093b3591337","subject":"Merge \"Add PDF building\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/01d94858288755b725293d19ece02093b3591337"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-11 08:55:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-11 09:06:33.000000000","tz":300},"subject":"SSL feature [WIP]","message":"SSL feature [WIP]\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/988053\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\nChange-Id: I93b9647c30cfcc7823e7d293652816f9040ee201\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7f62adcc77232f394673b2315a134314a61a689f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7f62adcc77232f394673b2315a134314a61a689f"}]},"branch":"refs/heads/master"},"0b690bfbfda75412570c91ce870961d8e5a2f415":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-05-11 09:09:21.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/59/988059/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/59/988059/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/59/988059/2"}}},"commit":{"parents":[{"commit":"01d94858288755b725293d19ece02093b3591337","subject":"Merge \"Add PDF building\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/01d94858288755b725293d19ece02093b3591337"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-11 08:55:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-11 09:09:06.000000000","tz":300},"subject":"SSL management feature [WIP]","message":"SSL management feature [WIP]\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/988053\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\nChange-Id: I93b9647c30cfcc7823e7d293652816f9040ee201\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0b690bfbfda75412570c91ce870961d8e5a2f415"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0b690bfbfda75412570c91ce870961d8e5a2f415"}]},"branch":"refs/heads/master"},"7be62737ee11f03d3ebc3e656e7be5b02411d354":{"kind":"REWORK","_number":3,"created":"2026-05-11 12:51:50.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/59/988059/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/59/988059/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/59/988059/3"}}},"commit":{"parents":[{"commit":"01d94858288755b725293d19ece02093b3591337","subject":"Merge \"Add PDF building\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/01d94858288755b725293d19ece02093b3591337"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-11 08:55:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-11 12:51:36.000000000","tz":300},"subject":"SSL management feature [WIP]","message":"SSL management feature [WIP]\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/988053\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\nChange-Id: I93b9647c30cfcc7823e7d293652816f9040ee201\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7be62737ee11f03d3ebc3e656e7be5b02411d354"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7be62737ee11f03d3ebc3e656e7be5b02411d354"}]},"branch":"refs/heads/master"},"aaee62a109076fa65b3f2d7b5535a4334aa622e2":{"kind":"REWORK","_number":4,"created":"2026-05-11 14:48:12.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/59/988059/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/59/988059/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/59/988059/4"}}},"commit":{"parents":[{"commit":"01d94858288755b725293d19ece02093b3591337","subject":"Merge \"Add PDF building\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/01d94858288755b725293d19ece02093b3591337"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-11 08:55:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-11 14:47:58.000000000","tz":300},"subject":"SSL management feature [WIP]","message":"SSL management feature [WIP]\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/988053\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\nChange-Id: I93b9647c30cfcc7823e7d293652816f9040ee201\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/aaee62a109076fa65b3f2d7b5535a4334aa622e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/aaee62a109076fa65b3f2d7b5535a4334aa622e2"}]},"branch":"refs/heads/master"},"403a6122e608d65b3594655cb21a1e0b66a49b15":{"kind":"REWORK","_number":5,"created":"2026-05-11 18:00:53.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/59/988059/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/59/988059/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/59/988059/5"}}},"commit":{"parents":[{"commit":"01d94858288755b725293d19ece02093b3591337","subject":"Merge \"Add PDF building\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/01d94858288755b725293d19ece02093b3591337"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-11 08:55:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-11 18:00:34.000000000","tz":300},"subject":"SSL management feature [WIP]","message":"SSL management feature [WIP]\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/988053\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\nChange-Id: I93b9647c30cfcc7823e7d293652816f9040ee201\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/403a6122e608d65b3594655cb21a1e0b66a49b15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/403a6122e608d65b3594655cb21a1e0b66a49b15"}]},"branch":"refs/heads/master"},"a547c810fa19b61469237b3b8078027e822a3832":{"kind":"REWORK","_number":6,"created":"2026-05-12 03:31:47.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/59/988059/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/59/988059/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/59/988059/6"}}},"commit":{"parents":[{"commit":"01d94858288755b725293d19ece02093b3591337","subject":"Merge \"Add PDF building\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/01d94858288755b725293d19ece02093b3591337"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-11 08:55:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-12 03:31:32.000000000","tz":300},"subject":"SSL management feature [WIP]","message":"SSL management feature [WIP]\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/988053\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\nChange-Id: I93b9647c30cfcc7823e7d293652816f9040ee201\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a547c810fa19b61469237b3b8078027e822a3832"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a547c810fa19b61469237b3b8078027e822a3832"}]},"branch":"refs/heads/master"},"d173d5d44f66e6b74a65b9093c2f7ff73e0678a4":{"kind":"REWORK","_number":7,"created":"2026-05-12 03:33:51.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/59/988059/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/59/988059/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/59/988059/7"}}},"commit":{"parents":[{"commit":"01d94858288755b725293d19ece02093b3591337","subject":"Merge \"Add PDF building\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/01d94858288755b725293d19ece02093b3591337"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-11 08:55:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-12 03:33:39.000000000","tz":300},"subject":"SSL management feature [WIP]","message":"SSL management feature [WIP]\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/988053\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\nChange-Id: I93b9647c30cfcc7823e7d293652816f9040ee201\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d173d5d44f66e6b74a65b9093c2f7ff73e0678a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d173d5d44f66e6b74a65b9093c2f7ff73e0678a4"}]},"branch":"refs/heads/master"},"fda969e16802ad13d47b8bbb31bb85d5a8cf5d92":{"kind":"REWORK","_number":8,"created":"2026-05-12 07:13:07.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/59/988059/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/59/988059/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/59/988059/8"}}},"commit":{"parents":[{"commit":"01d94858288755b725293d19ece02093b3591337","subject":"Merge \"Add PDF building\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/01d94858288755b725293d19ece02093b3591337"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-11 08:55:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-12 07:10:11.000000000","tz":300},"subject":"SSL management feature","message":"SSL management feature\n\nAPI-calls to manage SSL/TLS certificates for securing\nclient-server comminucations.\nPKCS#12 container and credentials are stored in Barbican.\nAvailable modes are: basic, enforced and mTLS.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/988053\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\nChange-Id: I93b9647c30cfcc7823e7d293652816f9040ee201\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fda969e16802ad13d47b8bbb31bb85d5a8cf5d92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fda969e16802ad13d47b8bbb31bb85d5a8cf5d92"}]},"branch":"refs/heads/master"},"c93c5b1832c9d44b6ffb42bbf03d68b501c84d35":{"kind":"REWORK","_number":9,"created":"2026-05-12 07:13:49.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/59/988059/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/59/988059/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/59/988059/9"}}},"commit":{"parents":[{"commit":"01d94858288755b725293d19ece02093b3591337","subject":"Merge \"Add PDF building\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/01d94858288755b725293d19ece02093b3591337"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-11 08:55:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-12 07:13:40.000000000","tz":300},"subject":"SSL management feature","message":"SSL management feature\n\nAPI-calls to manage SSL/TLS certificates for securing\nclient-server comminucations.\nPKCS#12 container and credentials are stored in Barbican.\nAvailable modes are: basic, enforced and mTLS.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/988053\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\nChange-Id: I93b9647c30cfcc7823e7d293652816f9040ee201\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c93c5b1832c9d44b6ffb42bbf03d68b501c84d35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c93c5b1832c9d44b6ffb42bbf03d68b501c84d35"}]},"branch":"refs/heads/master"},"56e338d1f91f6919307bc7d23a92e85cb8456d0a":{"kind":"REWORK","_number":10,"created":"2026-05-26 09:25:31.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/59/988059/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/59/988059/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/59/988059/10"}}},"commit":{"parents":[{"commit":"19db99328bd8929b76e5c726169f480877d02b4c","subject":"Merge \"update config parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/19db99328bd8929b76e5c726169f480877d02b4c"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-11 08:55:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-26 09:25:13.000000000","tz":300},"subject":"SSL management feature","message":"SSL management feature\n\nAPI-calls to manage SSL/TLS certificates for securing\nclient-server comminucations.\nPKCS#12 container and credentials are stored in Barbican.\nAvailable modes are: basic, enforced and mTLS.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/988053\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\nChange-Id: I93b9647c30cfcc7823e7d293652816f9040ee201\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/56e338d1f91f6919307bc7d23a92e85cb8456d0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/56e338d1f91f6919307bc7d23a92e85cb8456d0a"}]},"branch":"refs/heads/master"},"6e133899971354c14b9416441d9bdb01b55b2aac":{"kind":"REWORK","_number":11,"created":"2026-05-26 09:27:09.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/59/988059/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/59/988059/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/59/988059/11"}}},"commit":{"parents":[{"commit":"19db99328bd8929b76e5c726169f480877d02b4c","subject":"Merge \"update config parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/19db99328bd8929b76e5c726169f480877d02b4c"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-11 08:55:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-26 09:26:55.000000000","tz":300},"subject":"SSL management feature","message":"SSL management feature\n\nAPI-calls to manage SSL/TLS certificates for securing\nclient-server comminucations.\nPKCS#12 container and credentials are stored in Barbican.\nAvailable modes are: basic, enforced and mTLS.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/988053\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\nChange-Id: I93b9647c30cfcc7823e7d293652816f9040ee201\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6e133899971354c14b9416441d9bdb01b55b2aac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6e133899971354c14b9416441d9bdb01b55b2aac"}]},"branch":"refs/heads/master"},"af349dab670aadb84f13b51c90f334f09b6fb7a3":{"kind":"REWORK","_number":12,"created":"2026-05-27 13:01:31.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/59/988059/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/59/988059/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/59/988059/12"}}},"commit":{"parents":[{"commit":"19db99328bd8929b76e5c726169f480877d02b4c","subject":"Merge \"update config parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/19db99328bd8929b76e5c726169f480877d02b4c"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-11 08:55:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-27 13:00:05.000000000","tz":300},"subject":"SSL management feature","message":"SSL management feature\n\nAPI-calls to manage SSL/TLS certificates for securing\nclient-server comminucations.\nPKCS#12 container and credentials are stored in Barbican.\nAvailable modes are: basic, enforced and mTLS.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/988053\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\nChange-Id: I93b9647c30cfcc7823e7d293652816f9040ee201\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/af349dab670aadb84f13b51c90f334f09b6fb7a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/af349dab670aadb84f13b51c90f334f09b6fb7a3"}]},"branch":"refs/heads/master"},"ab3f950ff9c10aa430dfbd3eb1f26e63c2c56d7e":{"kind":"REWORK","_number":13,"created":"2026-05-27 16:54:52.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/59/988059/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/59/988059/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/59/988059/13"}}},"commit":{"parents":[{"commit":"19db99328bd8929b76e5c726169f480877d02b4c","subject":"Merge \"update config parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/19db99328bd8929b76e5c726169f480877d02b4c"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-11 08:55:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-27 16:54:39.000000000","tz":300},"subject":"SSL management feature","message":"SSL management feature\n\nAPI-calls to manage SSL/TLS certificates for securing\nclient-server comminucations.\nPKCS#12 container and credentials are stored in Barbican.\nAvailable modes are: basic, enforced and mTLS.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/988053\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\nChange-Id: I93b9647c30cfcc7823e7d293652816f9040ee201\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ab3f950ff9c10aa430dfbd3eb1f26e63c2c56d7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ab3f950ff9c10aa430dfbd3eb1f26e63c2c56d7e"}]},"branch":"refs/heads/master"},"54992f249da8c48e910ce58582876ce0a75df252":{"kind":"REWORK","_number":14,"created":"2026-05-28 16:58:50.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/59/988059/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/59/988059/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/59/988059/14"}}},"commit":{"parents":[{"commit":"b3ab18957a601524dd94e4085f959b16754cefdf","subject":"Fix checking for /dev/kvm device exists","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b3ab18957a601524dd94e4085f959b16754cefdf"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-11 08:55:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-28 16:58:19.000000000","tz":300},"subject":"SSL management feature","message":"SSL management feature\n\nAPI-calls to manage SSL/TLS certificates for securing\nclient-server comminucations.\nPKCS#12 container and credentials are stored in Barbican.\nAvailable modes are: basic, enforced and mTLS.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/988053\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\nChange-Id: I93b9647c30cfcc7823e7d293652816f9040ee201\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/54992f249da8c48e910ce58582876ce0a75df252"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/54992f249da8c48e910ce58582876ce0a75df252"}]},"branch":"refs/heads/master"},"382c46212aaaa07b8733bf2d66a274dda32d46e6":{"kind":"REWORK","_number":15,"created":"2026-05-28 19:21:07.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/59/988059/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/59/988059/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/59/988059/15"}}},"commit":{"parents":[{"commit":"b3ab18957a601524dd94e4085f959b16754cefdf","subject":"Fix checking for /dev/kvm device exists","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b3ab18957a601524dd94e4085f959b16754cefdf"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-11 08:55:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-28 19:20:57.000000000","tz":300},"subject":"SSL management feature","message":"SSL management feature\n\nAPI-calls to manage SSL/TLS certificates for securing\nclient-server comminucations.\nPKCS#12 container and credentials are stored in Barbican.\nAvailable modes are: basic, enforced and mTLS.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/988053\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\nChange-Id: I93b9647c30cfcc7823e7d293652816f9040ee201\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/382c46212aaaa07b8733bf2d66a274dda32d46e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/382c46212aaaa07b8733bf2d66a274dda32d46e6"}]},"branch":"refs/heads/master"},"213f45c52af878d33165a360d2691e8dd8caf762":{"kind":"REWORK","_number":16,"created":"2026-06-08 06:31:14.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/59/988059/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/59/988059/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/59/988059/16"}}},"commit":{"parents":[{"commit":"b3ab18957a601524dd94e4085f959b16754cefdf","subject":"Fix checking for /dev/kvm device exists","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b3ab18957a601524dd94e4085f959b16754cefdf"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-11 08:55:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-08 06:31:03.000000000","tz":300},"subject":"SSL management feature","message":"SSL management feature\n\nAPI-calls to manage SSL/TLS certificates for securing\nclient-server comminucations.\nPKCS#12 container and credentials are stored in Barbican.\nAvailable modes are: basic, enforced and mTLS.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/988053\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\nChange-Id: I93b9647c30cfcc7823e7d293652816f9040ee201\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/213f45c52af878d33165a360d2691e8dd8caf762"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/213f45c52af878d33165a360d2691e8dd8caf762"}]},"branch":"refs/heads/master"},"fada42f66d5c7b7ae3b0a003751ade463bc2161a":{"kind":"REWORK","_number":17,"created":"2026-06-23 09:25:48.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/59/988059/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/59/988059/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/59/988059/17"}}},"commit":{"parents":[{"commit":"e9489ce73e2415ed277eb51ab45394f9247acc30","subject":"Merge \"Improve data volume response in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e9489ce73e2415ed277eb51ab45394f9247acc30"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-11 08:55:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-23 09:25:19.000000000","tz":300},"subject":"SSL management feature","message":"SSL management feature\n\nAPI-calls to manage SSL/TLS certificates for securing\nclient-server comminucations.\nPKCS#12 container and credentials are stored in Barbican.\nAvailable modes are: basic, enforced and mTLS.\n\nBlueprint: ssl-management-feature\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/988053\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\nChange-Id: I93b9647c30cfcc7823e7d293652816f9040ee201\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fada42f66d5c7b7ae3b0a003751ade463bc2161a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fada42f66d5c7b7ae3b0a003751ade463bc2161a"}]},"branch":"refs/heads/master"},"c3dfc01ea9ff3d74ba08f95356ca10377e3da9f8":{"kind":"REWORK","_number":18,"created":"2026-06-23 11:17:29.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/59/988059/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/59/988059/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/59/988059/18"}}},"commit":{"parents":[{"commit":"e9489ce73e2415ed277eb51ab45394f9247acc30","subject":"Merge \"Improve data volume response in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e9489ce73e2415ed277eb51ab45394f9247acc30"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-11 08:55:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-23 11:17:17.000000000","tz":300},"subject":"SSL management feature","message":"SSL management feature\n\nAPI-calls to manage SSL/TLS certificates for securing\nclient-server comminucations.\nPKCS#12 container and credentials are stored in Barbican.\nAvailable modes are: basic, enforced and mTLS.\n\nBlueprint: ssl-management-feature\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/988053\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\nChange-Id: I93b9647c30cfcc7823e7d293652816f9040ee201\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c3dfc01ea9ff3d74ba08f95356ca10377e3da9f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c3dfc01ea9ff3d74ba08f95356ca10377e3da9f8"}]},"branch":"refs/heads/master"},"3ad10fd2b56284fb9dae7c94b0dd21fc326f36b6":{"kind":"REWORK","_number":19,"created":"2026-06-23 13:34:51.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/59/988059/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/59/988059/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/59/988059/19"}}},"commit":{"parents":[{"commit":"e9489ce73e2415ed277eb51ab45394f9247acc30","subject":"Merge \"Improve data volume response in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e9489ce73e2415ed277eb51ab45394f9247acc30"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-11 08:55:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-23 13:34:03.000000000","tz":300},"subject":"SSL management feature","message":"SSL management feature\n\nAPI-calls to manage SSL/TLS certificates for securing\nclient-server comminucations.\nPKCS#12 container and credentials are stored in Barbican.\nAvailable modes are: basic, enforced and mTLS.\n\nBlueprint: ssl-management-feature\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/988053\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\nChange-Id: I93b9647c30cfcc7823e7d293652816f9040ee201\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ad10fd2b56284fb9dae7c94b0dd21fc326f36b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ad10fd2b56284fb9dae7c94b0dd21fc326f36b6"}]},"branch":"refs/heads/master"},"ab010642aa0823b8445bf6a76a8900019c5fa065":{"kind":"REWORK","_number":20,"created":"2026-06-23 16:53:13.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/59/988059/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/59/988059/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/59/988059/20"}}},"commit":{"parents":[{"commit":"e9489ce73e2415ed277eb51ab45394f9247acc30","subject":"Merge \"Improve data volume response in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e9489ce73e2415ed277eb51ab45394f9247acc30"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-11 08:55:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-23 16:53:02.000000000","tz":300},"subject":"SSL management feature","message":"SSL management feature\n\nAPI-calls to manage SSL/TLS certificates for securing\nclient-server comminucations.\nPKCS#12 container and credentials are stored in Barbican.\nAvailable modes are: basic, enforced and mTLS.\n\nBlueprint: ssl-management-feature\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/988053\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\nChange-Id: I93b9647c30cfcc7823e7d293652816f9040ee201\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ab010642aa0823b8445bf6a76a8900019c5fa065"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ab010642aa0823b8445bf6a76a8900019c5fa065"}]},"branch":"refs/heads/master"},"7eaacd2f4293d7cf1ea4b1a08720fbbe35f60f18":{"kind":"REWORK","_number":21,"created":"2026-06-23 17:28:15.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/59/988059/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/59/988059/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/59/988059/21"}}},"commit":{"parents":[{"commit":"e9489ce73e2415ed277eb51ab45394f9247acc30","subject":"Merge \"Improve data volume response in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e9489ce73e2415ed277eb51ab45394f9247acc30"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-11 08:55:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-23 17:28:01.000000000","tz":300},"subject":"SSL management feature","message":"SSL management feature\n\nAPI-calls to manage SSL/TLS certificates for securing\nclient-server comminucations.\nPKCS#12 container and credentials are stored in Barbican.\nAvailable modes are: basic, enforced and mTLS.\n\nBlueprint: ssl-management-feature\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/988053\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\nChange-Id: I93b9647c30cfcc7823e7d293652816f9040ee201\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7eaacd2f4293d7cf1ea4b1a08720fbbe35f60f18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7eaacd2f4293d7cf1ea4b1a08720fbbe35f60f18"}]},"branch":"refs/heads/master"},"5b7dc24b231481ec092b120de0efe9d5fae70fac":{"kind":"REWORK","_number":22,"created":"2026-06-23 17:34:41.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/59/988059/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/59/988059/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/59/988059/22"}}},"commit":{"parents":[{"commit":"e9489ce73e2415ed277eb51ab45394f9247acc30","subject":"Merge \"Improve data volume response in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e9489ce73e2415ed277eb51ab45394f9247acc30"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-11 08:55:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-23 17:34:30.000000000","tz":300},"subject":"SSL management feature","message":"SSL management feature\n\nAPI-calls to manage SSL/TLS certificates for securing\nclient-server comminucations.\nPKCS#12 container and credentials are stored in Barbican.\nAvailable modes are: basic, enforced and mTLS.\n\nBlueprint: ssl-management-feature\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/988053\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\nChange-Id: I93b9647c30cfcc7823e7d293652816f9040ee201\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5b7dc24b231481ec092b120de0efe9d5fae70fac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5b7dc24b231481ec092b120de0efe9d5fae70fac"}]},"branch":"refs/heads/master"},"d41194b2f089c7d269499e4c2f80b38dbf326b30":{"kind":"REWORK","_number":23,"created":"2026-06-24 06:25:58.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/59/988059/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/59/988059/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/59/988059/23"}}},"commit":{"parents":[{"commit":"e9489ce73e2415ed277eb51ab45394f9247acc30","subject":"Merge \"Improve data volume response in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e9489ce73e2415ed277eb51ab45394f9247acc30"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-11 08:55:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-24 06:25:43.000000000","tz":300},"subject":"SSL management feature","message":"SSL management feature\n\nAPI-calls to manage SSL/TLS certificates for securing\nclient-server comminucations.\nPKCS#12 container and credentials are stored in Barbican.\nAvailable modes are: basic, enforced and mTLS.\n\nBlueprint: ssl-management-feature\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/988053\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\nChange-Id: I93b9647c30cfcc7823e7d293652816f9040ee201\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d41194b2f089c7d269499e4c2f80b38dbf326b30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d41194b2f089c7d269499e4c2f80b38dbf326b30"}]},"branch":"refs/heads/master"},"adaf26af57f7e025a9a1390aebd0f181f33662fc":{"kind":"REWORK","_number":24,"created":"2026-06-24 07:37:05.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/59/988059/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/59/988059/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/59/988059/24"}}},"commit":{"parents":[{"commit":"e9489ce73e2415ed277eb51ab45394f9247acc30","subject":"Merge \"Improve data volume response in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e9489ce73e2415ed277eb51ab45394f9247acc30"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-11 08:55:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-24 07:36:51.000000000","tz":300},"subject":"SSL management feature","message":"SSL management feature\n\nAPI-calls to manage SSL/TLS certificates for securing\nclient-server comminucations.\nPKCS#12 container and credentials are stored in Barbican.\nAvailable modes are: basic, enforced and mTLS.\n\nBlueprint: ssl-management-feature\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/988053\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\nChange-Id: I93b9647c30cfcc7823e7d293652816f9040ee201\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/adaf26af57f7e025a9a1390aebd0f181f33662fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/adaf26af57f7e025a9a1390aebd0f181f33662fc"}]},"branch":"refs/heads/master"},"1a87f328f85543b3224df20001c1e0ee2c66ee67":{"kind":"REWORK","_number":25,"created":"2026-06-24 08:27:51.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/59/988059/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/59/988059/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/59/988059/25"}}},"commit":{"parents":[{"commit":"e9489ce73e2415ed277eb51ab45394f9247acc30","subject":"Merge \"Improve data volume response in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e9489ce73e2415ed277eb51ab45394f9247acc30"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-11 08:55:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-24 08:27:39.000000000","tz":300},"subject":"SSL management feature","message":"SSL management feature\n\nAPI-calls to manage SSL/TLS certificates for securing\nclient-server comminucations.\nPKCS#12 container and credentials are stored in Barbican.\nAvailable modes are: basic, enforced and mTLS.\n\nBlueprint: ssl-management-feature\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/988053\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\nChange-Id: I93b9647c30cfcc7823e7d293652816f9040ee201\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1a87f328f85543b3224df20001c1e0ee2c66ee67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1a87f328f85543b3224df20001c1e0ee2c66ee67"}]},"branch":"refs/heads/master"},"f33d41b4ce08947b562a4eec310d3f0b2b144a02":{"kind":"REWORK","_number":26,"created":"2026-06-24 13:17:43.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/59/988059/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/59/988059/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/59/988059/26"}}},"commit":{"parents":[{"commit":"e9489ce73e2415ed277eb51ab45394f9247acc30","subject":"Merge \"Improve data volume response in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e9489ce73e2415ed277eb51ab45394f9247acc30"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-11 08:55:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-24 13:13:08.000000000","tz":300},"subject":"SSL management feature","message":"SSL management feature\n\nAPI-calls to manage SSL/TLS certificates for securing\nclient-server comminucations.\nPKCS#12 container and credentials are stored in Barbican.\nAvailable modes are: basic, enforced and mTLS.\n\nBlueprint: ssl-management-feature\nDepends-On: https://review.opendev.org/c/openstack/trove/+/992900\nDepends-On: https://review.opendev.org/c/openstack/trove/+/992028\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/988053\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\nChange-Id: I93b9647c30cfcc7823e7d293652816f9040ee201\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f33d41b4ce08947b562a4eec310d3f0b2b144a02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f33d41b4ce08947b562a4eec310d3f0b2b144a02"}]},"branch":"refs/heads/master"},"08b3dfb77928eb6bccc0d38addc581599e95e641":{"kind":"REWORK","_number":27,"created":"2026-06-24 14:12:52.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/59/988059/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/59/988059/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/59/988059/27"}}},"commit":{"parents":[{"commit":"e9489ce73e2415ed277eb51ab45394f9247acc30","subject":"Merge \"Improve data volume response in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e9489ce73e2415ed277eb51ab45394f9247acc30"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-11 08:55:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-24 14:12:42.000000000","tz":300},"subject":"SSL management feature","message":"SSL management feature\n\nAPI-calls to manage SSL/TLS certificates for securing\nclient-server comminucations.\nPKCS#12 container and credentials are stored in Barbican.\nAvailable modes are: basic, enforced and mTLS.\n\nBlueprint: ssl-management-feature\nDepends-On: https://review.opendev.org/c/openstack/trove/+/992900\nDepends-On: https://review.opendev.org/c/openstack/trove/+/992028\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/988053\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\nChange-Id: I93b9647c30cfcc7823e7d293652816f9040ee201\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/08b3dfb77928eb6bccc0d38addc581599e95e641"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/08b3dfb77928eb6bccc0d38addc581599e95e641"}]},"branch":"refs/heads/master"},"329a54069dd6e5b16f3bacba4762ff53853abe08":{"kind":"REWORK","_number":28,"created":"2026-06-24 15:52:54.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/59/988059/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/59/988059/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/59/988059/28"}}},"commit":{"parents":[{"commit":"e9489ce73e2415ed277eb51ab45394f9247acc30","subject":"Merge \"Improve data volume response in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e9489ce73e2415ed277eb51ab45394f9247acc30"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-11 08:55:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-24 15:52:11.000000000","tz":300},"subject":"SSL management feature","message":"SSL management feature\n\nAPI-calls to manage SSL/TLS certificates for securing\nclient-server comminucations.\nPKCS#12 container and credentials are stored in Barbican.\nAvailable modes are: basic, enforced and mTLS.\n\nBlueprint: ssl-management-feature\nDepends-On: https://review.opendev.org/c/openstack/trove/+/992900\nDepends-On: https://review.opendev.org/c/openstack/trove/+/992028\nDepends-On: https://review.opendev.org/c/openstack/trove/+/994217\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/988053\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\nChange-Id: I93b9647c30cfcc7823e7d293652816f9040ee201\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/329a54069dd6e5b16f3bacba4762ff53853abe08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/329a54069dd6e5b16f3bacba4762ff53853abe08"}]},"branch":"refs/heads/master"},"9e93ecdfc65b65baea4e7b8a755323a8771209c9":{"kind":"REWORK","_number":29,"created":"2026-06-24 17:48:44.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/59/988059/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/59/988059/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/59/988059/29"}}},"commit":{"parents":[{"commit":"e9489ce73e2415ed277eb51ab45394f9247acc30","subject":"Merge \"Improve data volume response in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e9489ce73e2415ed277eb51ab45394f9247acc30"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-11 08:55:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-24 17:48:34.000000000","tz":300},"subject":"SSL management feature","message":"SSL management feature\n\nAPI-calls to manage SSL/TLS certificates for securing\nclient-server comminucations.\nPKCS#12 container and credentials are stored in Barbican.\nAvailable modes are: basic, enforced and mTLS.\n\nBlueprint: ssl-management-feature\nDepends-On: https://review.opendev.org/c/openstack/trove/+/992900\nDepends-On: https://review.opendev.org/c/openstack/trove/+/992028\nDepends-On: https://review.opendev.org/c/openstack/trove/+/994217\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/988053\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\nChange-Id: I93b9647c30cfcc7823e7d293652816f9040ee201\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9e93ecdfc65b65baea4e7b8a755323a8771209c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9e93ecdfc65b65baea4e7b8a755323a8771209c9"}]},"branch":"refs/heads/master"},"94ceeda40b1cbc838955811ba0636ccb4f6cf7b2":{"kind":"REWORK","_number":30,"created":"2026-06-25 07:03:10.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/59/988059/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/59/988059/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/59/988059/30"}}},"commit":{"parents":[{"commit":"e9489ce73e2415ed277eb51ab45394f9247acc30","subject":"Merge \"Improve data volume response in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e9489ce73e2415ed277eb51ab45394f9247acc30"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-11 08:55:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-25 07:02:59.000000000","tz":300},"subject":"SSL management feature","message":"SSL management feature\n\nAPI-calls to manage SSL/TLS certificates for securing\nclient-server comminucations.\nPKCS#12 container and credentials are stored in Barbican.\nAvailable modes are: basic, enforced and mTLS.\n\nBlueprint: ssl-management-feature\nDepends-On: https://review.opendev.org/c/openstack/trove/+/992900\nDepends-On: https://review.opendev.org/c/openstack/trove/+/992028\nDepends-On: https://review.opendev.org/c/openstack/trove/+/994217\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/988053\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\nChange-Id: I93b9647c30cfcc7823e7d293652816f9040ee201\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/94ceeda40b1cbc838955811ba0636ccb4f6cf7b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/94ceeda40b1cbc838955811ba0636ccb4f6cf7b2"}]},"branch":"refs/heads/master"},"77eedfcf18d3417dfa8a8bb89495e1db5cf05dbf":{"kind":"REWORK","_number":31,"created":"2026-06-25 08:23:26.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/59/988059/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/59/988059/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/59/988059/31"}}},"commit":{"parents":[{"commit":"e9489ce73e2415ed277eb51ab45394f9247acc30","subject":"Merge \"Improve data volume response in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e9489ce73e2415ed277eb51ab45394f9247acc30"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-11 08:55:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-25 08:23:14.000000000","tz":300},"subject":"SSL management feature","message":"SSL management feature\n\nAPI-calls to manage SSL/TLS certificates for securing\nclient-server comminucations.\nPKCS#12 container and credentials are stored in Barbican.\nAvailable modes are: basic, enforced and mTLS.\n\nBlueprint: ssl-management-feature\nDepends-On: https://review.opendev.org/c/openstack/trove/+/992900\nDepends-On: https://review.opendev.org/c/openstack/trove/+/992028\nDepends-On: https://review.opendev.org/c/openstack/trove/+/994217\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/988053\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\nChange-Id: I93b9647c30cfcc7823e7d293652816f9040ee201\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/77eedfcf18d3417dfa8a8bb89495e1db5cf05dbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/77eedfcf18d3417dfa8a8bb89495e1db5cf05dbf"}]},"branch":"refs/heads/master"},"bc852454debd9568d958e0eacd54bb3a75aa7ced":{"kind":"REWORK","_number":32,"created":"2026-06-25 11:14:02.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/59/988059/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/59/988059/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/59/988059/32"}}},"commit":{"parents":[{"commit":"e9489ce73e2415ed277eb51ab45394f9247acc30","subject":"Merge \"Improve data volume response in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e9489ce73e2415ed277eb51ab45394f9247acc30"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-11 08:55:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-25 11:13:52.000000000","tz":300},"subject":"SSL management feature","message":"SSL management feature\n\nAPI-calls to manage SSL/TLS certificates for securing\nclient-server comminucations.\nPKCS#12 container and credentials are stored in Barbican.\nAvailable modes are: basic, enforced and mTLS.\n\nBlueprint: ssl-management-feature\nDepends-On: https://review.opendev.org/c/openstack/trove/+/992900\nDepends-On: https://review.opendev.org/c/openstack/trove/+/992028\nDepends-On: https://review.opendev.org/c/openstack/trove/+/994217\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/988053\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\nChange-Id: I93b9647c30cfcc7823e7d293652816f9040ee201\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bc852454debd9568d958e0eacd54bb3a75aa7ced"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bc852454debd9568d958e0eacd54bb3a75aa7ced"}]},"branch":"refs/heads/master"},"f2d1f65cf38a7507db4f971a1348a240d0aaf5ef":{"kind":"REWORK","_number":33,"created":"2026-06-25 11:49:47.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/59/988059/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/59/988059/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/59/988059/33"}}},"commit":{"parents":[{"commit":"e9489ce73e2415ed277eb51ab45394f9247acc30","subject":"Merge \"Improve data volume response in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e9489ce73e2415ed277eb51ab45394f9247acc30"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-11 08:55:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-25 11:49:37.000000000","tz":300},"subject":"SSL management feature","message":"SSL management feature\n\nAPI-calls to manage SSL/TLS certificates for securing\nclient-server comminucations.\nPKCS#12 container and credentials are stored in Barbican.\nAvailable modes are: basic, enforced and mTLS.\n\nBlueprint: ssl-management-feature\nDepends-On: https://review.opendev.org/c/openstack/trove/+/992900\nDepends-On: https://review.opendev.org/c/openstack/trove/+/992028\nDepends-On: https://review.opendev.org/c/openstack/trove/+/994217\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/988053\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\nChange-Id: I93b9647c30cfcc7823e7d293652816f9040ee201\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f2d1f65cf38a7507db4f971a1348a240d0aaf5ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f2d1f65cf38a7507db4f971a1348a240d0aaf5ef"}]},"branch":"refs/heads/master"},"3eb43225c47e6043a532252395e08b01bedd694f":{"kind":"REWORK","_number":34,"created":"2026-06-25 12:14:05.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/59/988059/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/59/988059/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/59/988059/34"}}},"commit":{"parents":[{"commit":"e9489ce73e2415ed277eb51ab45394f9247acc30","subject":"Merge \"Improve data volume response in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e9489ce73e2415ed277eb51ab45394f9247acc30"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-11 08:55:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-25 12:13:54.000000000","tz":300},"subject":"SSL management feature","message":"SSL management feature\n\nAPI-calls to manage SSL/TLS certificates for securing\nclient-server comminucations.\nPKCS#12 container and credentials are stored in Barbican.\nAvailable modes are: basic, enforced and mTLS.\n\nBlueprint: ssl-management-feature\nDepends-On: https://review.opendev.org/c/openstack/trove/+/992900\nDepends-On: https://review.opendev.org/c/openstack/trove/+/992028\nDepends-On: https://review.opendev.org/c/openstack/trove/+/994217\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/988053\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\nChange-Id: I93b9647c30cfcc7823e7d293652816f9040ee201\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3eb43225c47e6043a532252395e08b01bedd694f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3eb43225c47e6043a532252395e08b01bedd694f"}]},"branch":"refs/heads/master"},"dd90e68306783efd091f659e1c41453689b8814e":{"kind":"REWORK","_number":35,"created":"2026-06-25 12:40:27.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/59/988059/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/59/988059/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/59/988059/35"}}},"commit":{"parents":[{"commit":"e9489ce73e2415ed277eb51ab45394f9247acc30","subject":"Merge \"Improve data volume response in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e9489ce73e2415ed277eb51ab45394f9247acc30"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-11 08:55:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-25 12:40:15.000000000","tz":300},"subject":"SSL management feature","message":"SSL management feature\n\nAPI-calls to manage SSL/TLS certificates for securing\nclient-server comminucations.\nPKCS#12 container and credentials are stored in Barbican.\nAvailable modes are: basic, enforced and mTLS.\n\nBlueprint: ssl-management-feature\nDepends-On: https://review.opendev.org/c/openstack/trove/+/992900\nDepends-On: https://review.opendev.org/c/openstack/trove/+/992028\nDepends-On: https://review.opendev.org/c/openstack/trove/+/994217\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/988053\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\nChange-Id: I93b9647c30cfcc7823e7d293652816f9040ee201\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dd90e68306783efd091f659e1c41453689b8814e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dd90e68306783efd091f659e1c41453689b8814e"}]},"branch":"refs/heads/master"},"1f73cca92c334ab9965c9277d96f5022b5d0d936":{"kind":"NO_CODE_CHANGE","_number":36,"created":"2026-07-10 13:18:33.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/59/988059/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/59/988059/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/59/988059/36"}}},"commit":{"parents":[{"commit":"e9489ce73e2415ed277eb51ab45394f9247acc30","subject":"Merge \"Improve data volume response in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e9489ce73e2415ed277eb51ab45394f9247acc30"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-11 08:55:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-07-10 13:18:33.000000000","tz":0},"subject":"SSL management feature","message":"SSL management feature\n\nAPI-calls to manage SSL/TLS certificates for securing\nclient-server comminucations.\nPKCS#12 container and credentials are stored in Barbican.\nAvailable modes are: basic, enforced and mTLS.\n\nBlueprint: ssl-management-feature\nDepends-On: https://review.opendev.org/c/openstack/trove/+/992900\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/988053\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\nChange-Id: I93b9647c30cfcc7823e7d293652816f9040ee201\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1f73cca92c334ab9965c9277d96f5022b5d0d936"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1f73cca92c334ab9965c9277d96f5022b5d0d936"}]},"branch":"refs/heads/master","description":"Edit commit message"},"e594eee0ac269a4455205c79dc3b3204499b8781":{"kind":"REWORK","_number":37,"created":"2026-07-14 12:35:14.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/59/988059/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/59/988059/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/59/988059/37"}}},"commit":{"parents":[{"commit":"db36c51d5b0ad2519cb47574849c5b87c6e6c241","subject":"Unmount volume before instance rebuild","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/db36c51d5b0ad2519cb47574849c5b87c6e6c241"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-11 08:55:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-07-14 12:33:32.000000000","tz":300},"subject":"SSL management feature","message":"SSL management feature\n\nAdd instance API actions to show, enable, disable, and roll back SSL/TLS\nconfiguration. Certificates are supplied as PKCS#12 secrets stored in\nBarbican, and support basic, enforced, and mutual-TLS modes.\n\nImplement guest-agent support for MySQL, MariaDB, and PostgreSQL, including\ncertificate propagation to replicas and Barbican consumer lifecycle handling.\n\nBlueprint: ssl-management-feature\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/988053\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\nChange-Id: I93b9647c30cfcc7823e7d293652816f9040ee201\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e594eee0ac269a4455205c79dc3b3204499b8781"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e594eee0ac269a4455205c79dc3b3204499b8781"}]},"branch":"refs/heads/master"},"729f614a73d426c9f2bd6162c4a6ab868b203dc8":{"kind":"REWORK","_number":38,"created":"2026-07-14 12:49:37.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/59/988059/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/59/988059/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/988059/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/59/988059/38"}}},"commit":{"parents":[{"commit":"db36c51d5b0ad2519cb47574849c5b87c6e6c241","subject":"Unmount volume before instance rebuild","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/db36c51d5b0ad2519cb47574849c5b87c6e6c241"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-11 08:55:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-07-14 12:49:27.000000000","tz":300},"subject":"SSL management feature","message":"SSL management feature\n\nAdd instance API actions to show, enable, disable, and roll back SSL/TLS\nconfiguration. Certificates are supplied as PKCS#12 secrets stored in\nBarbican, and support basic, enforced, and mutual-TLS modes.\n\nImplement guest-agent support for MySQL, MariaDB, and PostgreSQL, including\ncertificate propagation to replicas and Barbican consumer lifecycle handling.\n\nBlueprint: ssl-management-feature\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/988053\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\nChange-Id: I93b9647c30cfcc7823e7d293652816f9040ee201\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/729f614a73d426c9f2bd6162c4a6ab868b203dc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/729f614a73d426c9f2bd6162c4a6ab868b203dc8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
