)]}'
{"id":"openstack%2Ftrove~988207","triplet_id":"openstack%2Ftrove~master~Ib9434f57f209042794985f15e2b7af423ad4919b","project":"openstack/trove","branch":"master","topic":"enable-default-tempest-concurrency","hashtags":[],"change_id":"Ib9434f57f209042794985f15e2b7af423ad4919b","subject":"[test] Use default tempest concurrency.","status":"ABANDONED","created":"2026-05-12 07:19:09.000000000","updated":"2026-05-12 09:47:20.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e703dc54bca6e9021a53e9924336bc320ba284bd","_number":988207,"virtual_id_number":988207,"owner":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-05-12 09:11:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-12 07:34:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-12 09:11:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"db12433bc4219f597321129ab6b5cd2b9f124f31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-12 07:19:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a202be74f093c508f99404dfc3c0d1cc76bed396","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 07:34:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe4de0ea809948babb4a2cd5e98d3f36\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/37fb167d98dd4a42a13ff26fc5e69eaf : SUCCESS in 8m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/77932e54aaeb4484859a3097aee3edcb : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/565afdf5f395421a865bb9bfbebb38cc : SUCCESS in 9m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ad789a22e3919d25f1ff0ee4c15ac76219b61eb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 09:11:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5677d8a96e144fe2af0ec407586de970\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ec0bbad691d45ed985d9b0042f72eef : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b8d0d439d0f40329a902809787725ba : SUCCESS in 3m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ee893b69e89846129a7ffab28e8e8d4b : SUCCESS in 4m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1376d13eeda44c85a8566a075e7c03d3 : SUCCESS in 4m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/910e0f5367d74eb7b11b3a68b074953a : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eda0de3c3ea64a629ddabd71bfbe04e1 : SUCCESS in 9m 52s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/9b3baa52894141e39db67d766498f70e : SUCCESS in 47m 25s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/ed132e4b7c9d444197f0bd615af02ac2 : SUCCESS in 1h 51m 27s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/2beee9e0e10d462c9f117ac7cf8d8973 : SUCCESS in 1h 27m 43s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/6039bafb98a44d1c8d5608e670a0ba7a : SUCCESS in 1h 08m 31s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/9fc852877b444acba5b829ed2f04aa40 : SUCCESS in 40m 14s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/002224c070724980ac613087cbd97e45 : SUCCESS in 1h 35m 33s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/ff65933e325a45c4a5d044d44c6d6b24 : SUCCESS in 1h 17m 20s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/750ffe205d494647ba6cf9ef50e2bc5d : SUCCESS in 57m 52s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/2476e6f9e8ca41ec8c20f70a7f0f2825 : SUCCESS in 44m 40s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/8f47435dd21c49e98c0fadc8e7d3d23b : SUCCESS in 1h 28m 29s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/2fba3f1f39d34920a5aa9fd78cc28d79 : SUCCESS in 1h 14m 05s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/9c45149f7001436a842bf2de5baf014e : SUCCESS in 53m 50s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/69d90941a0c64b138abf6fffadf0413d : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":1},{"id":"d717bf9ad9a040826bd4c725aabaa5928b1fef35","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-12 09:47:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e703dc54bca6e9021a53e9924336bc320ba284bd","tag":"autogenerated:gerrit:abandon","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-12 09:47:20.000000000","message":"Abandoned\n\nno results","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d9bfc3e77257d100b901eb8e527dc8208c765272","revisions":{"d9bfc3e77257d100b901eb8e527dc8208c765272":{"kind":"REWORK","_number":1,"created":"2026-05-12 07:19:09.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/07/988207/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/07/988207/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/07/988207/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/07/988207/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/07/988207/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/07/988207/1"}}},"commit":{"parents":[{"commit":"01d94858288755b725293d19ece02093b3591337","subject":"Merge \"Add PDF building\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/01d94858288755b725293d19ece02093b3591337"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-12 07:14:42.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-12 07:18:39.000000000","tz":300},"subject":"[test] Use default tempest concurrency.","message":"[test] Use default tempest concurrency.\n\nRight now, tempest_concurrency \u003d 1 which makes tests run very\nslow. Removing this parameter will utilize tempest\u0027s default\nconcurrency which equals to num_cores - 2.\n\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\nChange-Id: Ib9434f57f209042794985f15e2b7af423ad4919b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d9bfc3e77257d100b901eb8e527dc8208c765272"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d9bfc3e77257d100b901eb8e527dc8208c765272"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
