)]}'
{"id":"openstack%2Ftrove~989856","triplet_id":"openstack%2Ftrove~master~Ia99918ff3df2078fed8636d46a9c4309f402eaba","project":"openstack/trove","branch":"master","topic":"redis-based-keydb-valkey-datastores","attention_set":{"36080":{"account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"last_update":"2026-09-03 11:34:02.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{"28691":{"account":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"last_update":"2026-05-27 01:38:23.000000000","reason":"\u003cGERRIT_ACCOUNT_28691\u003e replied on the change","reason_account":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}},"5733":{"account":{"_account_id":5733,"name":"Kieran Spear","email":"kispear@gmail.com","username":"kspear"},"last_update":"2026-08-11 00:10:55.000000000","reason":"\u003cGERRIT_ACCOUNT_5733\u003e replied on the change","reason_account":{"_account_id":5733,"name":"Kieran Spear","email":"kispear@gmail.com","username":"kspear"}}},"hashtags":[],"change_id":"Ia99918ff3df2078fed8636d46a9c4309f402eaba","subject":"Add KeyDB and Valkey datastore support","status":"NEW","created":"2026-05-23 13:25:02.000000000","updated":"2026-09-03 15:02:11.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ba38ff5dd4d1a6a2dfe41aa7312e1f282b5f4f68","_number":989856,"virtual_id_number":989856,"owner":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-03 15:02:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":5733,"name":"Kieran Spear","email":"kispear@gmail.com","username":"kspear"},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-23 13:58:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-23 15:27:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-25 09:42:14.000000000","updated_by":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"CC"},{"updated":"2026-08-10 06:45:11.000000000","updated_by":{"_account_id":5733,"name":"Kieran Spear","email":"kispear@gmail.com","username":"kspear"},"reviewer":{"_account_id":5733,"name":"Kieran Spear","email":"kispear@gmail.com","username":"kspear"},"state":"CC"}],"messages":[{"id":"c2c74a594ac6316782cb82af43872450c08647c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-23 13:25:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e445a538c562e6ebaa84f0a081cad2e742074c30","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-23 13:58:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/261bd347232b434c833322064ebf56d8\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/28578a4d2da04cf8a3fa9c198362fba1 : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/78a3246a43354ff18473c85576bd4ba9 : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ccfc3f0587f44b5399e105a158d00f3a : SUCCESS in 10m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"90d041d9cb316a2d0d35af89d812fb555a96db81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-23 15:27:27.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a9dcf3da76d94bb68dbd9da20722e403\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/21aaa0e0c5e04083a0bc3cc089fec1e0 : SUCCESS in 2m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b03624354bf14245ae304c566ef8cdec : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6447831cc2846099a49272fb75c4454 : FAILURE in 4m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/17dbbbf02bf044dba25ce999cde0d7d1 : SUCCESS in 2m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0940160f88ed4ac4951a3f8fd54aae73 : SUCCESS in 4m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c4f507b131cb45b391a6f5c098ef1300 : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/442475f6d00246aca45e54828856872d : SUCCESS in 5m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/307a05787f5f48ee92e69dc2f6c1ea0a : SUCCESS in 5m 08s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/94e9ca97ce10467c91252117c9440c1c : SUCCESS in 57m 36s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/de6ac8f37d274a28a550a09feb054e3e : SUCCESS in 2h 01m 10s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/73b84272fe3c4408916cd93f5555e39a : SUCCESS in 50m 23s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/709263f0d6ab4ad48f018fba7b7fa9e3 : SUCCESS in 1h 57m 17s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/2d8a5e28ce5e4671877a8493d7ef8474 : SUCCESS in 55m 37s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/1240a42eb4bf4a9e8cf021d42bd72c9b : SUCCESS in 1h 41m 15s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/617f1fe80c9247809f6166a07ceb8baa : SUCCESS in 1h 24m 45s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/b349d72b0574417dbe1d3ca2cc65e273 : SUCCESS in 1h 45m 11s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/dfd8c62484f64078b8937ac27e02ded9 : SUCCESS in 57m 55s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/8c7fd71a96bd47b89cd3bf07e3716e36 : SUCCESS in 1h 42m 17s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/e131e6c893b247708c67d1f8b6774a0b : SUCCESS in 1h 24m 32s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/bd84db781185417099227b26f1e28bb6 : SUCCESS in 1h 32m 06s (non-voting)\n- trove-tempest-cinder-storage-driver-keydb https://zuul.opendev.org/t/openstack/build/bb1bcf1bdec84912a33d867fdd0ae898 : FAILURE in 40m 32s (non-voting)\n- trove-tempest-cinder-storage-driver-valkey https://zuul.opendev.org/t/openstack/build/7fba8971d9b745159beb779fb650e882 : SUCCESS in 1h 16m 03s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9d9eaa8cabe541e9a95d4b1b9b2a90eb : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":1},{"id":"70cf199ac5e6d5c8d3024667405c720eef788bcf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-23 18:09:41.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"6315ebeb46bc74f5cc65a29ec634f242cd1524bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-23 18:11:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e781915a6b9b2ef9e172e12a4f44b99063acff49","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-23 18:28:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5090715814d942c796c295ebb44091ff\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ed42e8db89a140c381bece63fa42583e : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/62386e44b70f494fb52990012b6c70a6 : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/70d5d1660eed4723bdc0b1377e0ff66b : SUCCESS in 9m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a8e68e87e2205a091ea87c3d874251affc883a94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-23 20:25:20.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4c963ab8dcc6486e98120f05d400d4ff\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/4a99488162e943d3a922102c42899f02 : SUCCESS in 3m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81bca5f39c2248219b7b7d83e283b9fb : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af58a0c03aa94d92a198acf595046de6 : SUCCESS in 5m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b10a5529f64447cb9292f41773bb2483 : SUCCESS in 3m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/04825e52187742fc8902c56d24186e08 : SUCCESS in 4m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d3d5a64cae4d4ef99f60c1dd1c6013da : SUCCESS in 7m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ad1350794ed43fe864facd37375ab86 : SUCCESS in 8m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/65a35c7768c94cf99c4bcf5f9379d98d : SUCCESS in 6m 57s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/390dee75af2e43048c525dc6139e00ef : SUCCESS in 48m 32s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/780f63ff5af04d6a81d565ee9cc2f77e : FAILURE in 33m 28s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/12a059fbe6e94c46b89a7870098b4c98 : FAILURE in 33m 06s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/4e1230f18832458e83adc48517dfede2 : SUCCESS in 2h 11m 02s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/2882d89899724cb0b41590ca7ea1add3 : SUCCESS in 45m 42s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/6bce64c2742e484d814fafc5e6354e7b : FAILURE in 35m 32s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/4a6646aff0b849ae9d4310111ca4b77d : FAILURE in 35m 40s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/6f3b7a8ad5294e92bf08983bc0528197 : SUCCESS in 1h 43m 47s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/f72e7bd874114318bd4157ab006f4a08 : SUCCESS in 47m 24s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/6fec166a4b7a43eda9290d3b2c93d52d : FAILURE in 34m 57s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/71c2dc83b9ef439dba48d181da01755c : FAILURE in 33m 26s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/cd471c969ca7423d95df1fab8445311f : SUCCESS in 1h 19m 10s (non-voting)\n- trove-tempest-ubuntu-base-keydb https://zuul.opendev.org/t/openstack/build/1c25b6a86dd342569b90a89f1ceb9b2a : FAILURE in 33m 15s (non-voting)\n- trove-tempest-cinder-storage-driver-keydb https://zuul.opendev.org/t/openstack/build/7ad4d42763524f549f53ecd84d66ed50 : FAILURE in 44m 02s (non-voting)\n- trove-tempest-cinder-storage-driver-valkey https://zuul.opendev.org/t/openstack/build/00d5e97e04ae46f1a27aa135ce0d08a3 : FAILURE in 44m 03s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0cdbfa3bf5a24ff99d1462a734e20924 : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":3},{"id":"369c02a70b8b1da202b5cdc5da70c8cbe160e81a","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-24 07:00:25.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7082ea14727dda878256ba735fb7155da7f882d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-24 07:15:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f6459776bf64fbd81a8c5f580141868\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4c86f768c5504f17ac4445480f4a93c0 : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/62999b7335d748da9da689e2426699b2 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b24cfbf64f4a42949519ac17177c22d9 : SUCCESS in 13m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fd4539c92991850117363ae13727cb47c9eec66b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-24 08:47:00.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2d3c9263ce7e496fb0d6f8fa41f08251\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/5f5401f2ef344cc2b360727c400aca38 : SUCCESS in 2m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/567651ac45b24839944d09845eca7839 : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc55d264fd49479489e48ee8a15f89e5 : SUCCESS in 3m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/367ef1387efd4ecca8cf2675be4d13ab : SUCCESS in 4m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3069e174dd5b47928570a57ba35e6730 : SUCCESS in 3m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6c1668f90c764af8a5ea5a5d82659521 : SUCCESS in 7m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98444fee968d4206a7772cd40b6f26e5 : SUCCESS in 5m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bbea2c318f8f491fb2ab1d97d347c397 : SUCCESS in 4m 19s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/c7e0ea570aca448abbaf7e29ca855713 : SUCCESS in 27m 22s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/c50ca007370e47abb5665a66005a66d4 : FAILURE in 33m 22s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/392b9dd2572a447cae90fc48d5635cf2 : FAILURE in 38m 06s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/cda5140d1a694726962001e14a51f2ef : SUCCESS in 1h 44m 23s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/1618832f16324b8b9c13fc170b28de04 : SUCCESS in 41m 05s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/a219c11754864157b6df02db68d06f4f : FAILURE in 32m 01s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/3c4e70a8cf924d3086b9c0100924328e : FAILURE in 38m 31s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/238d2491736d4b41ae8138a8b5ae471b : SUCCESS in 1h 36m 16s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/efd487b05df4497eabb5d5e509385a99 : SUCCESS in 45m 14s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/e1a3fb499af5424cbbaa616e54ff328f : FAILURE in 38m 24s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/7ee49b4fd7f043eb91c30df2fbc11f16 : FAILURE in 33m 57s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/614826baa4c34bf5a3da4b5b8c0812b6 : SUCCESS in 1h 19m 35s (non-voting)\n- trove-tempest-ubuntu-base-keydb https://zuul.opendev.org/t/openstack/build/3dbc3bb91b1c4933b242baf00937c441 : SUCCESS in 44m 27s (non-voting)\n- trove-tempest-cinder-storage-driver-keydb https://zuul.opendev.org/t/openstack/build/ed7cf1676d1942f2add522d11799ff7b : SUCCESS in 1h 05m 15s (non-voting)\n- trove-tempest-cinder-storage-driver-valkey https://zuul.opendev.org/t/openstack/build/36f7a776c69d4630a804cfef688aae64 : FAILURE in 59m 39s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/dc881f2d291b4cdcb6d4e855e6a9a387 : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":3},{"id":"f9b84145433fd4aa8b83d24436e6b634a5fef1da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-24 10:17:21.000000000","message":"Uploaded patch set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"3203c2bf5314270934b994467c140651894b0713","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-24 10:32:39.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d44f8ed8e854f65a01eb2ee2e96c094\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/aeb9f58a7ab845f8875dafae119d03c3 : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bddf10b8f6ef4643a8147086d8958350 : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a6b8aa0ae3c444aea4a2d07e2c4ef20a : SUCCESS in 12m 55s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"95dd63c22182198155378773a212aa0a7d1bbb5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-24 11:05:23.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8b00ca7a927f4c315d83411b4c41469702e5c6ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-24 11:23:26.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c9bb212d6ae463582b0e1f73b0e3503\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5edd7accd2d44c53ab5e63b9a54e4136 : SUCCESS in 7m 46s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9329213452134b23b0512cda9e7c757b : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a4da58c6f82c418d92de4498892001ce : SUCCESS in 15m 14s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"43e09ecf7c8a8436e258b7309871b8bb06c1ef20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-24 13:03:11.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6760dbff53164020bcdd214c6cdb9143\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/2908df2533a645da8b31eb9f9466da61 : SUCCESS in 3m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7907ce8a815c49a08d366916f75ea8a7 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/348f7f0455284c4d82a1a8270a7b59a3 : SUCCESS in 4m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b81c2233c2294fdaacfae4bbc16a144b : SUCCESS in 3m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a02d935b703e43fd9215eff5ff6a3917 : SUCCESS in 3m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/63b743b1731540cd881e468c243e62ee : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc0955df35de4194852af6791dca7444 : SUCCESS in 5m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d04e28b3b3548e2af6cfe07bf8469c5 : SUCCESS in 7m 08s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/a501d99e5d2a456ca3fa6c4461934c68 : SUCCESS in 32m 06s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/99247bad2bb5451284ef6966e1c9fc08 : SUCCESS in 1h 56m 14s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/792fdbd8bf4c4cd58f4931f3c14bff26 : SUCCESS in 1h 47m 25s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/a350c95f53f24becaaf2ba925fd61b65 : FAILURE in 1h 36m 58s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/c99959f906ff4c04b313459bed73d14b : SUCCESS in 44m 35s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/a17192a2bb9a469482a3b19db48ca7cf : SUCCESS in 1h 51m 23s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/a223a4f1f4d34590a3db76c320ba84c2 : SUCCESS in 1h 20m 13s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/b13e9755a19d44b4bdab22c35091daa9 : SUCCESS in 1h 36m 58s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/36d3056cf95345bfa20f03428dd57daa : FAILURE in 45m 02s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/184bf5a8e2f3433ebe4a9d4b70c5ddf0 : SUCCESS in 1h 35m 28s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/93ad22600d1e4e168b23e44d2d401924 : SUCCESS in 1h 23m 30s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/fa2e5ab0f13243829ce44bd5ed9a4705 : SUCCESS in 1h 42m 35s (non-voting)\n- trove-tempest-ubuntu-base-keydb https://zuul.opendev.org/t/openstack/build/e5421df55d4d44e0a299daacdd73d51e : SUCCESS in 47m 29s (non-voting)\n- trove-tempest-cinder-storage-driver-keydb https://zuul.opendev.org/t/openstack/build/cd42368ec4054f65a27e7e763350b431 : SUCCESS in 57m 20s (non-voting)\n- trove-tempest-cinder-storage-driver-valkey https://zuul.opendev.org/t/openstack/build/eda4bf5d156a4b43bcb94aaddd1a7f9a : SUCCESS in 57m 15s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/befb3f878e56445a852bf5f9fc20771e : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":5},{"id":"fe13c421a3aac10cf65d15fa0c2a284d27805e30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-24 13:46:46.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"23fb37e20df06ccab64ba9775144594ee5ede852","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-24 13:47:48.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"2332d95cc572edded0e2eebb6cc7a5a5cc47cb8c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-24 13:58:33.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cb3b2e706604eb28c78a46f6de5698b\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/24e3b06b864040949ed7d1dd2c1aed5d : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f3d59a96295541119c4eb1bfa5d18e7f : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3433527dc9ae478b82a8cf0205ae19a3 : SUCCESS in 9m 06s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"c5a9434dbb517a0c3de5be99a8f81e9efb9864ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-24 15:43:31.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12ccc285e8ea4b939a86c241395724db\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/9d370e139bd3413f968ba92cd4ec0d40 : SUCCESS in 3m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2c5f348ec4940ada3f8b12adb9debfe : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/214cda9f1bf44cd3a152557e5376e919 : SUCCESS in 4m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/77a61557be7f4c64aadc4190336cecbc : SUCCESS in 4m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a11795bdf5d645eeb4bf37fd010e59d9 : SUCCESS in 4m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/98a85f0322634554a31f58901d4de0ab : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eecf877a5e0f4b52a5f4d3f3686cb18a : SUCCESS in 9m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/96fb4f67edac47bca071ee60c65c8f05 : SUCCESS in 6m 59s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/76c1c224d2c045eb809e22bd80c93004 : SUCCESS in 47m 07s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/0ace4779f9b9407c8533e5060a7531a3 : SUCCESS in 1h 54m 35s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/af738429edaf4fe5973115e12eb004c1 : SUCCESS in 51m 08s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/70d22cfd59cb4a85a1627247c64c1eb1 : SUCCESS in 1h 45m 23s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/cd6dadeec0fb46308c45e8fbe77089f3 : SUCCESS in 43m 00s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/660a17c4a00b49a6964db9254b9a6e70 : SUCCESS in 1h 35m 36s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/804111232ce946f2bdd87378ba271e6c : SUCCESS in 1h 52m 08s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/d1e4b1cc037742389a04b2ddb215b105 : SUCCESS in 1h 32m 57s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/65efa568896843b89087bde665d847c9 : SUCCESS in 45m 54s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/5cb633ea43e54151b9d5f7e59044458c : SUCCESS in 1h 36m 45s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/17fa3cee32104aef9e3de8df322b91f8 : SUCCESS in 1h 15m 48s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/5cc42c1855bb40328b16509d5f0bee25 : SUCCESS in 50m 56s (non-voting)\n- trove-tempest-ubuntu-base-keydb https://zuul.opendev.org/t/openstack/build/4d82dbbe90c44cb3879f903375a0f8b7 : SUCCESS in 23m 35s (non-voting)\n- trove-tempest-cinder-storage-driver-keydb https://zuul.opendev.org/t/openstack/build/9bec46107df84f2e8c2f58ff665bbe4f : SUCCESS in 1h 15m 27s (non-voting)\n- trove-tempest-cinder-storage-driver-valkey https://zuul.opendev.org/t/openstack/build/14836e4648df446aa35e71a498b4ff71 : SUCCESS in 1h 00m 34s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/04fc65c182d1422d8aba711bfb0b3e25 : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":7},{"id":"df145ec93f610ca33f2a2dee8fa2a99b6b5605d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-24 17:59:34.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"c4a00ae543749537d8836a14b7dfab66e06247c1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-24 18:14:07.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2a3324f65ab421cac84991fa8917c71\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5817ac8eb0b54c6a9a17f8f5b9c8ac5c : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/689c592f59c048a68052dbaaf6ddc5ef : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2b0f794fe6374d77af2293f0b9628d39 : SUCCESS in 12m 29s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"b5636e9e4f9a7436aa92cfb221b42c93a6059808","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-24 20:00:06.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c5abe4cb6ad4bc285673d6d68b090dd\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/01c29e29b2a145458480cbe574728ecc : SUCCESS in 3m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07c5badfd9a348a691748dcc8b95fd6c : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3790f849a75641938de042d4c25295e5 : SUCCESS in 5m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b61512edbb2a4be0a90ce3f344552b33 : SUCCESS in 4m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e6ccc5c1a10e485ea5da0a0ec36e05dd : SUCCESS in 2m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c3284c3e5b614476aae35b2206929e03 : SUCCESS in 9m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/189b30a240ed4a6abdd5a20cdd5f83fd : SUCCESS in 5m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d459cf11e2bd4dd0a4e170676107853f : SUCCESS in 3m 24s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/f4186436343041f285900fdd013cc395 : SUCCESS in 1h 11m 33s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/ab3d7bdb56814034ae93c15773b2cd73 : SUCCESS in 1h 52m 03s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/bbad5522bb4a459783108318901cca35 : SUCCESS in 1h 28m 56s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/00303d6138ad474fba1f17e380f426ee : SUCCESS in 1h 45m 40s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/9bb7371c1a3843ae8cfcac22153e4065 : SUCCESS in 1h 02m 40s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/bc220859e7c24264977cfdec3c363df4 : SUCCESS in 1h 36m 19s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/d4d734c187c04f6abd84fa16c56f7921 : SUCCESS in 1h 30m 37s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/20231bd366324c9f88047ffed6850fbc : FAILURE in 1h 58m 45s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/20e1498ad24e4216a139ad57b91a019b : SUCCESS in 57m 59s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/c5accacaceff449baf04b91d603ecc5a : SUCCESS in 1h 47m 02s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/7834bf3dd6b94861a00aeec795beb40a : SUCCESS in 1h 12m 33s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/73178a6bd0574a7a8a1a62bf83ac4559 : SUCCESS in 1h 35m 26s (non-voting)\n- trove-tempest-ubuntu-base-keydb https://zuul.opendev.org/t/openstack/build/bf61b3ae1b6d4ebc8a2e68e0750c371b : SUCCESS in 51m 13s (non-voting)\n- trove-tempest-cinder-storage-driver-keydb https://zuul.opendev.org/t/openstack/build/541e533ebd134c50bb03a47cde8fb485 : SUCCESS in 52m 18s (non-voting)\n- trove-tempest-cinder-storage-driver-valkey https://zuul.opendev.org/t/openstack/build/aaa52a0c77ce4be8a6507f5f90da7f3a : SUCCESS in 1h 12m 11s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/407c03d24fd940ebb8fa162e61d10a4c : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":8},{"id":"43fa1024af3bf6ed0785a49a3f0baa404c755f19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-25 07:54:34.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"ab366ac1b0a645cfb0e8ec0d4e008166c27af14d","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2026-05-25 09:42:14.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"226dde952cab04bae06e2874e4eb454c9a7e2335","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-25 10:16:37.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"a513f5807a90ce184bc7aac20f7b7f145cff2649","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-25 11:45:30.000000000","message":"Uploaded patch set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"6a93841bb83fd0857bb9062982ee7f975cef2ee0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 12:09:19.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34782bf8f58241438f3eac187f6931fd\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7a84721e27d544388b7b792aa86cc4af : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/fa65f7bc868e4b19b0dd8428927d5f31 : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/13a7aaa876d24c31b145a1e0a195aef4 : SUCCESS in 14m 37s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"de57e4241e61d2366ee2384eeba6e2941aeb3a13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 13:55:31.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/331df7e18f8b40b4b078827d575655f9\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ac00afdd63554b70b7bf1dc9c9144e63 : SUCCESS in 2m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/782e25e9882040eeafac3746c4a700a5 : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49876192c24a4cd7a7f65bbe1b12bfc5 : SUCCESS in 3m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ae181bc566d342f98e62348e57c75eb8 : SUCCESS in 3m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1de46da6b8454fa7a5d9d00765eed2ac : SUCCESS in 5m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/51ea571cd52945e2b99bdb2e934ea5f6 : SUCCESS in 8m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c56e70167a844f2ea32843ee42a5481f : SUCCESS in 5m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c24baa998ce6454395de5614ddf03f64 : SUCCESS in 7m 50s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/8685860e97b84bb887ede83077b11347 : SUCCESS in 48m 27s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/3e861a5ba15e4fa0ab1f79c5d070edd7 : SUCCESS in 2h 08m 15s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/f9ffcc858be9426098530f59a57b2f6b : SUCCESS in 1h 40m 10s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/98d14c1c0fa14223b8ad438ddf4655a1 : SUCCESS in 1h 52m 01s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/a721ef33899448c9bc1fd29226c47f47 : FAILURE in 48m 18s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/775a15abb2c940daa27ddc7f79514257 : SUCCESS in 1h 41m 56s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/7bd163173a1d476c9bd4485c5bce51f2 : SUCCESS in 1h 16m 30s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/fe49c1a518b74d76af4d02a68935ac40 : SUCCESS in 1h 36m 28s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/3e0165182dd64f56ae8441e157513229 : SUCCESS in 46m 11s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/e05d085b23a54862930eddab0e21da81 : FAILURE in 1h 03m 54s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/bf7308f1f88549cda83d4bac425ba51b : SUCCESS in 1h 21m 46s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/ae0aaca525b34ec89f81775ac92bc518 : SUCCESS in 1h 25m 49s (non-voting)\n- trove-tempest-ubuntu-base-keydb https://zuul.opendev.org/t/openstack/build/b8305845f5b747d19a3690776567cdff : SUCCESS in 40m 01s (non-voting)\n- trove-tempest-cinder-storage-driver-keydb https://zuul.opendev.org/t/openstack/build/64832cbbda494f619732a0153af12f77 : SUCCESS in 1h 03m 06s (non-voting)\n- trove-tempest-cinder-storage-driver-valkey https://zuul.opendev.org/t/openstack/build/bb3b199d6be048f1b1e30fe51e12fad8 : SUCCESS in 1h 02m 28s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/84f15798abd44de68d7cf023525a8f7c : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":10},{"id":"04538ba29be3c472d0a727311a2409cf6aabea5a","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-25 14:35:35.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"12fff83c69dc2368e98573d2f52175b2594926f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 15:07:34.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16c178434c1a46f0b10cd799dbe5fdd2\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/89f5c0de8ba84107aae7788d9e64465a : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/03bfb03241644e9d84a2dc6487369bc9 : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5d36252bd4624a879c4acbc3427df789 : SUCCESS in 14m 13s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"91234b25a56b927d40ed3061f1aa0aab5809193b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 16:44:05.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79c779139c674351aab0db96d769d384\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/421e2f1d30a94f8ba3be13ef1fd2010e : SUCCESS in 2m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c277c5ce05647b798aa1fb300681831 : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ce54828a4ff412ab378e2bf2c8ae361 : SUCCESS in 5m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/655b18bc27ae47b3b34c8e4695573bfb : SUCCESS in 4m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e4ede721196f415682907b3f91df126f : SUCCESS in 4m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3bc9f0ee568d4159b1500991300c0048 : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6014cfa0b9534d7f96a64af65d16c0df : SUCCESS in 9m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/18a1c21316354d00ac8b4966497ccc73 : SUCCESS in 8m 13s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/281daeff439349e4bb36c0737b48e367 : SUCCESS in 56m 00s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/6b186413373b40b8a353893f23ccaa71 : SUCCESS in 2h 06m 42s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/fd0892215a2047c1ab7d09a7e8eb9606 : SUCCESS in 1h 42m 47s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/28071ca6a256421e978dd3377b69f573 : SUCCESS in 1h 47m 53s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/c02edfd8c114430e99712916982fdd97 : SUCCESS in 44m 55s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/ecb9df9926294f0f8e7a4448e6296934 : SUCCESS in 1h 41m 08s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/79383713b93348d5a79bc948bc80e9d2 : SUCCESS in 1h 20m 04s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/308b806990bc46d1b5727af2b9923638 : SUCCESS in 1h 40m 02s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/212adab4b50a424eb6cb64d1a35bbce8 : SUCCESS in 46m 36s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/d55bcb18a1e240149285adc262bbd06b : SUCCESS in 1h 41m 03s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/503108a53e6f4278a317414e9c85de49 : SUCCESS in 1h 28m 04s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/8dcbfd9fd1b740829ea71e3a536e6bf6 : SUCCESS in 1h 30m 54s (non-voting)\n- trove-tempest-ubuntu-base-keydb https://zuul.opendev.org/t/openstack/build/ede3de3c1f744a64bc79bd4bbdb643b0 : SUCCESS in 46m 12s (non-voting)\n- trove-tempest-cinder-storage-driver-keydb https://zuul.opendev.org/t/openstack/build/76d0c925b13a4ce9b8e87b6cfd2544cc : SUCCESS in 1h 09m 28s (non-voting)\n- trove-tempest-cinder-storage-driver-valkey https://zuul.opendev.org/t/openstack/build/4d431f0db8b345d4848c9c921d165102 : SUCCESS in 1h 05m 24s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b5eaa30ea13d44698ea4b33446f904cd : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":10},{"id":"d4ffeef94cfa0471ddcfda1b7083d56a2aa665db","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2026-05-26 06:55:15.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"d625272c400c2369096e135b96a23fc5f4df3c80","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-26 09:01:02.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"181a0618d0439a0e17239bd45a347d9e63e4f3ec","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2026-05-27 01:38:23.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"367dd9b38de220c5eed342eb888a221db2d9d793","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-29 13:13:25.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"b0b9756e68219e332c85cff0c51736f138207c24","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 14:40:17.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a8274e7a516450e8cf2e1c30fc0270f\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1604ef8879104feeacf72018c511bc53 : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a8dd65b21cac4461814e73854fb6472d : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0d38b3a6ae67482287f10db5b8e65f96 : SUCCESS in 14m 39s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"a35b461a82e7381558f1cde9809bd4b562212599","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 15:23:46.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ad32960095264ddc97ba0686b66cc90f\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3900e8cfe28e4d77951d52d9471a752e : SUCCESS in 4m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6511875193fd44388f18a007744b0d87 : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ffc6b1f56894fd1a57602fedba6703f : SUCCESS in 4m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e12d033e36044faf9d85a56043d461b3 : SUCCESS in 4m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d5fdeea5c1ce4a70aae94b95da26384d : SUCCESS in 2m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6651d522fe0b4d319e82cb4de9122560 : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a718eb4ad084570a635dd601303ffdc : SUCCESS in 9m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98f11d6efbbe4760a5b01cb9cac5dcd2 : SUCCESS in 7m 13s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/e69e9fa33099455c827dd11527075de7 : SUCCESS in 48m 25s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/6294313546ec433e8cb739ee8da457a9 : SUCCESS in 2h 08m 39s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/0a7137b05dbf491c91cf934993885034 : SUCCESS in 1h 40m 52s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/256642672bca49fdab3c12d489df960a : SUCCESS in 1h 50m 28s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/fdd22fc74d2943b38dd95fbba925040b : SUCCESS in 51m 03s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/65b3404d097041b69b120a7587083da4 : SUCCESS in 1h 40m 18s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/069413e173364ebb920f24e1f64f9934 : SUCCESS in 1h 19m 06s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/b072a6ca19ae4c00aea2bf4e43590adf : SUCCESS in 1h 39m 59s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/5d83d5bb92264031b16fe51fe3b03b6e : FAILURE in 26m 44s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/3fe93877ae6842e783742a912f739fe0 : FAILURE in 36m 32s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/1197a15b3b0b4cc7b78f8a3cf40d262f : FAILURE in 36m 56s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/41c63b8c10fe4bc29e107537af74afce : FAILURE in 28m 40s (non-voting)\n- trove-tempest-ubuntu-base-keydb https://zuul.opendev.org/t/openstack/build/f3569c747c3644b6bcd6bd740f39dfd9 : SUCCESS in 39m 29s\n- trove-tempest-ubuntu-backup-keydb https://zuul.opendev.org/t/openstack/build/9edec22701fa46028c0b117a6098d125 : FAILURE in 45m 35s (non-voting)\n- trove-tempest-ubuntu-replication-keydb https://zuul.opendev.org/t/openstack/build/68f70fd75ba542d9ba51c6e5c2caa86f : FAILURE in 45m 25s (non-voting)\n- trove-tempest-cinder-storage-driver-keydb https://zuul.opendev.org/t/openstack/build/e772ecdc68a4409ca76c7f88cd24e1a1 : SUCCESS in 57m 56s (non-voting)\n- trove-tempest-ubuntu-base-valkey https://zuul.opendev.org/t/openstack/build/c97570aa52ae4123a815ed8c024e07e5 : SUCCESS in 37m 40s (non-voting)\n- trove-tempest-cinder-storage-driver-valkey https://zuul.opendev.org/t/openstack/build/78a08ac9c1f14322b6cb396564e3deef : SUCCESS in 1h 02m 34s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b9a84c38cd894e9a95ca092ad6b7326c : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":11},{"id":"a68ac7acd4add9a3805d95eab0a83810487fa4d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-29 15:29:41.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"0f957d5d9d247f0b78f94a0d562940c164539262","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 16:00:46.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb81406198dc451e93467358a2f53799\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/90ecad926a3540ea977ce95e722a077e : SUCCESS in 7m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4437bc4131134d04a3f3aecf13a87ac6 : SUCCESS in 7m 33s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b8b4009531034eac9434287e637e94db : SUCCESS in 8m 40s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"217df837dde60bbcd036943f98f9e308e5c4b6d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 17:47:50.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/81539b19212a46419f41927b44fd2b18\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b7a4814d463646c9b7d1df3fc7a7acee : SUCCESS in 4m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fad01f0c3dc44235b127dd9fb6b87fcc : SUCCESS in 8m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6971e3e18d5446ee8d5026db7e09b6de : SUCCESS in 6m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/018212e7c09844729842b814aaa4fab9 : SUCCESS in 4m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d68d252319304913a06dbb5253de3bc9 : SUCCESS in 4m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9664c46d33b9491b915d14c8c02ee20f : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35123c8d5fa44bb1b5bc5604268af470 : SUCCESS in 10m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee7e7891361747ca95b257b3dc9e742c : SUCCESS in 7m 28s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/05e14775700f474d8d8460423e298a89 : SUCCESS in 47m 38s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/062dbc3317a042b0a2b2c698944e0984 : SUCCESS in 2h 15m 38s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/5bf2cfcdb80b4a9babc384f3c7d2dfb4 : SUCCESS in 1h 48m 17s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/115986a5f1f9443da12a8ba9c13f88e9 : SUCCESS in 1h 57m 26s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/23c235fd666547b1a2fe800e850db8f9 : SUCCESS in 46m 29s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/462b7e61982346528a2ac42553568a55 : SUCCESS in 1h 38m 30s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/980eaf2b7cf14f72aa297723718bd7ec : SUCCESS in 1h 26m 59s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/4fa28f2d56354587aabe874e65038b3e : SUCCESS in 1h 37m 32s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/835538995453496980ee81094498596b : FAILURE in 27m 45s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/b140dbba6f2a4e24af5ab438f8efb0f9 : FAILURE in 35m 12s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/1f4988403cec44dea8f6543375d784f0 : FAILURE in 36m 30s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/fdb44399e6ec4c1b8546c12d2199caba : FAILURE in 27m 12s (non-voting)\n- trove-tempest-ubuntu-base-keydb https://zuul.opendev.org/t/openstack/build/84eab80a33b94b0695df3489fe01b91e : SUCCESS in 40m 00s\n- trove-tempest-ubuntu-backup-keydb https://zuul.opendev.org/t/openstack/build/ec67de4408f5413cb3947604606141aa : FAILURE in 48m 38s (non-voting)\n- trove-tempest-ubuntu-replication-keydb https://zuul.opendev.org/t/openstack/build/6300f3526b1e4bdbaa35d5677a812ec9 : FAILURE in 45m 01s (non-voting)\n- trove-tempest-cinder-storage-driver-keydb https://zuul.opendev.org/t/openstack/build/8eea86507a9d41c8bc980b75340d6095 : SUCCESS in 1h 00m 24s (non-voting)\n- trove-tempest-ubuntu-base-valkey https://zuul.opendev.org/t/openstack/build/468b890e1ee54fa4a647458e2d9c8da7 : SUCCESS in 40m 03s (non-voting)\n- trove-tempest-cinder-storage-driver-valkey https://zuul.opendev.org/t/openstack/build/a0510d4fb3304b199483874867df5a82 : SUCCESS in 59m 48s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3e9dd1ad7ebc48b4bd859eeb506a4d43 : SUCCESS in 6m 06s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/c22baa9a9ab144f8b30b9055391a0962 : SUCCESS in 8m 38s","accounts_in_message":[],"_revision_number":12},{"id":"1c2baa609b42c31073083e9840878198a33b5b40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-29 19:48:47.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"146ebefd5270f806b5458a0f074300778b50b902","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 20:04:11.000000000","message":"Patch Set 13: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/trove/+/990725 is needed.","accounts_in_message":[],"_revision_number":13},{"id":"130f366faaf5f131e241f2494b0bee19f475100f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 20:04:15.000000000","message":"Patch Set 13:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/trove/+/990725 is needed.","accounts_in_message":[],"_revision_number":13},{"id":"99176f4fabfb652e7ccfceae2c974a0913aba82a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-02 09:11:43.000000000","message":"Uploaded patch set 14: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"63e2bd7a214bbaf646a9b6fbe357342df97a7119","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 09:50:47.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ba7e26c3a75477cbac06894b0cb662f\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/309b7799ecc64432935ecdda7fc38dba : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5d40c176d903487489828c390e0e1812 : SUCCESS in 7m 47s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b58bb0100ad042dd962daad349286005 : SUCCESS in 11m 20s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"084a9b7b7caad95b300f449e07d7051b61fcd43e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 10:56:51.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7bc099aa6ab346b192bf5e9607fa4ba5\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/9888ae49d2ac4d11932767b70da97d51 : SUCCESS in 5m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/66778e8745a24d128720bf3de87fda5b : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8d63ab2426449fb87ce5e22f9719975 : SUCCESS in 3m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/549af025372a410e989a1f44bac8909b : SUCCESS in 4m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ff37fd1f85e845d7a19604cf4534129e : SUCCESS in 4m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/041140e0e24741e7bfbcdd953fc3bb53 : SUCCESS in 9m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47022223837a473d9f8cbc452d124b79 : SUCCESS in 11m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/972eee3549cb40d2925d7cfdb9ccd874 : SUCCESS in 7m 03s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/f89101ddfb3846d38942a12cd2a59589 : SUCCESS in 43m 45s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/4f59444d7da8473f99711ea3f1685542 : SUCCESS in 1h 39m 26s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/eb7cb49736cc4c46840d99fbbb877a5f : SUCCESS in 1h 32m 29s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/c7a98b3e9db94add8dac23918e71c8b3 : SUCCESS in 1h 43m 37s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/81efdd5742794ea1aa932983207e4a59 : SUCCESS in 41m 47s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/fd18bb1df57a4f7ca30f57e510b9f6d7 : SUCCESS in 1h 41m 05s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/8c8c5c80a3ad4828ba9783310af66d84 : SUCCESS in 1h 19m 27s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/f50dab9fcc604e3fb14598cb7477873f : SUCCESS in 1h 37m 06s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/cd3082dd6dbe4ede9fb6a8937d74e378 : FAILURE in 26m 53s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/87cfd0793e954c5dad06553452211399 : FAILURE in 39m 10s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/7588ca8447a34a41bc42ba683ab9e93c : FAILURE in 50m 42s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/995c4432d0a94064a95789a36d377ee1 : FAILURE in 37m 54s (non-voting)\n- trove-tempest-ubuntu-base-keydb https://zuul.opendev.org/t/openstack/build/25960c700b6246ae9003e70c2dfa0e85 : FAILURE in 40m 45s\n- trove-tempest-ubuntu-backup-keydb https://zuul.opendev.org/t/openstack/build/0a1f639b58994bbbbf682ccf88bd887e : FAILURE in 43m 25s (non-voting)\n- trove-tempest-ubuntu-replication-keydb https://zuul.opendev.org/t/openstack/build/36bbd96a1c3147e5a3f60e42bc3c8849 : FAILURE in 57m 56s (non-voting)\n- trove-tempest-cinder-storage-driver-keydb https://zuul.opendev.org/t/openstack/build/e45e30f9b70143b4824d4bf45114cbaf : FAILURE in 44m 53s (non-voting)\n- trove-tempest-ubuntu-base-valkey https://zuul.opendev.org/t/openstack/build/02be0af623b7404685e464cf230f026f : FAILURE in 38m 34s (non-voting)\n- trove-tempest-cinder-storage-driver-valkey https://zuul.opendev.org/t/openstack/build/7fdba0844a1d48c5a364e692a8c0cd0e : FAILURE in 56m 00s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c46728cf59b9404fae3720218d23ce94 : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":14},{"id":"15e40e4d3e77fa14775510db35fb723ec5ef576e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-04 15:15:07.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":15},{"id":"600a6cea25c2e20c15a4728b35576c2896748d6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 15:15:38.000000000","message":"Patch Set 15: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/trove for 989856,15","accounts_in_message":[],"_revision_number":15},{"id":"421b5ec00981f4cd541f84bff93f9619c0ddf413","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 15:15:56.000000000","message":"Patch Set 15:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/trove for 989856,15","accounts_in_message":[],"_revision_number":15},{"id":"49e7478045692c964c6a1fc85a198058c50cc19f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-04 15:30:14.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":16},{"id":"8d9763eb6ff7069f3cfd8252140b8df5b5f8e747","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 15:49:45.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6a04bfe51cc4ca090a73c55e718be97\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/602ab10d445a44da835e4347bfb5d9c1 : SUCCESS in 7m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0e96534d9d664caaba892821568bd795 : SUCCESS in 8m 50s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/93a3093c09e64d69b2decb04b40dfbf7 : SUCCESS in 9m 26s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"1f6206a2b401c2d8e3a7ef0f989cb48eaa363424","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-04 15:56:34.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"c4e85fad6b3a7fae1e2f7cf03fbc6d116e898981","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 16:09:09.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d40fb9de4214b44ba6f2471370fcf25\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4477b4d9e35f4977b94257701c675513 : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c4c565e546ff486fa39538705c58e0c8 : SUCCESS in 7m 32s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/22c77ef766ce46c0803e81704071b3eb : SUCCESS in 9m 31s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"7c2b710e14c64420a1587f9ed2027c7553b6586f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-04 17:04:03.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"5d81042e3830572e2cbe8f4af14abbe68089bab6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 17:21:12.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23f6734230ea453b9ecea6e741aec4e6\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bcb1cd308c5c4c0aad92327f90d2de00 : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/32f7f803acf4450384e5dc5e766790cd : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/059b11eaee364fa292c61b3cbe4e180f : SUCCESS in 12m 28s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"cbcbb9500bd9bdfb315d5a7ab053dc5d9ea943e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 17:36:22.000000000","message":"Patch Set 18: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/trove/+/990725 is needed.","accounts_in_message":[],"_revision_number":18},{"id":"e125b3c863adf094de5fde12ca3ec498e5d58895","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-04 17:45:34.000000000","message":"Uploaded patch set 19: Patch Set 18 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":19},{"id":"2757e20302eeafee6addce525540055010e63be9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 18:04:56.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f8364c4663349f683449f5ef32f09ff\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3c5363c1b402437abcb0dec094cbf0f1 : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ec600c1a104c416b81a51d3a67b70472 : SUCCESS in 7m 53s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ec55aa46688846cbbab9c025a648f76c : SUCCESS in 7m 07s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"31c05eb129b339394eecec53351bf038241380fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 19:15:49.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/379fa81458e0436b8acd11376a8d2a08\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/af62b3d1cf8c4edd844367f64495fadb : SUCCESS in 2m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02065cf976ee4f84ae16acf5ac989522 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcb32d736ae44dc2afb121fe4def714e : SUCCESS in 5m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1ab68fcdcee045c78761a33e52c1fe06 : SUCCESS in 2m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/249fe9f36a484e93ab0ff0378d2ae3f8 : SUCCESS in 5m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b981536f465d4d658b665e79a174e34c : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e4428cb670e47a7844acd540073b2ed : SUCCESS in 10m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8185300ffdba439b9ebc2f75c6313692 : SUCCESS in 7m 15s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/9b644ae57e36438bbf15189ab11fc1fa : SUCCESS in 36m 21s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/2eace9e5932848299c5f712fbae9fe8b : SUCCESS in 1h 20m 56s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/2299ebe763e8421da2c209925a06fa73 : SUCCESS in 1h 03m 41s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/a57a1657ef9744d9ba4efdc5e247ee58 : SUCCESS in 1h 16m 26s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/262c83a0693e4a08b19d6411fe683d63 : SUCCESS in 42m 00s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/ff677f17449843eca001125d045f18d4 : SUCCESS in 1h 21m 27s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/2d87beec98804b1b9f5baa5d3787059e : SUCCESS in 1h 03m 59s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/4060cbf060d042a19b12d6ee07a7d6d1 : SUCCESS in 1h 26m 06s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/b32f4937e92647ea94c87ed14479744c : SUCCESS in 39m 10s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/facee950bd0e47148c3927d9cf45f218 : SUCCESS in 1h 10m 04s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/88e90dfa3b1f483a8cbd45ccaa2ec61c : SUCCESS in 47m 48s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/6263820f3df94e9a9a9ecdd6e9cf51e9 : SUCCESS in 1h 05m 54s (non-voting)\n- trove-tempest-ubuntu-base-keydb https://zuul.opendev.org/t/openstack/build/7a7a0d8b3bf34c4184ef755200ea6d33 : FAILURE in 38m 07s\n- trove-tempest-ubuntu-backup-keydb https://zuul.opendev.org/t/openstack/build/2efbac79d9f44972aed5de8c1540f119 : FAILURE in 37m 12s (non-voting)\n- trove-tempest-ubuntu-replication-keydb https://zuul.opendev.org/t/openstack/build/a852858ee97f412db9b1d97d67856755 : FAILURE in 39m 08s (non-voting)\n- trove-tempest-cinder-storage-driver-keydb https://zuul.opendev.org/t/openstack/build/74ea9fe072a54ed58600979fe08275d7 : FAILURE in 44m 10s (non-voting)\n- trove-tempest-ubuntu-base-valkey https://zuul.opendev.org/t/openstack/build/57c5de65e4f5485492b212e23c31d00d : FAILURE in 36m 44s (non-voting)\n- trove-tempest-cinder-storage-driver-valkey https://zuul.opendev.org/t/openstack/build/769d8a496db846b8b51a2157da3f39e3 : FAILURE in 42m 23s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7fc1f571f58f441e949791fb51b725cb : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":19},{"id":"6f084fd5128238cd926daf70d1659824fff854b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-05 06:30:35.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":20},{"id":"d8384824a4afa181799c03ee68d007706878dc1d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 06:42:01.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43b8f9ec29e443b9b15a2e5ab387afae\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bffdf007804a4fec9e28791df42a8885 : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/102ac661b15948a18e48fd18b18360bf : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/aedcfc76fee542e68161f1752e687290 : SUCCESS in 8m 38s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"f6b66254b1a8d33bc384b83c1d6c82307614f6dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-05 07:06:29.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"0c727268a0968cdfd8fefe1b62a5146598c32861","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 07:26:07.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/612ecc8ab41a4806a3b7eea6a636a389\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5f811612d21b4fccb22b0a465d0e97c3 : SUCCESS in 8m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8e1b746738d94960bbb73c9c102c32b9 : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/dc62f4f8858742758f3873e65a02cab0 : SUCCESS in 16m 19s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"7fba8a0bbab94a3b25563eb79769c6e84bbff624","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-05 08:34:55.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"a41ac80bb8f35fb2a3e559788b436756ad41a35c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 08:49:28.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e7ad43cde2b4bfc93912510ecade461\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4906e24e938a4e409fd45175572c9178 : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ad16328b3a20450881bfed19c2eb3988 : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0b8c3e459ab44d4eae664600587dad39 : SUCCESS in 12m 10s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"21801401e137bf25eefbe6f20ab797cf83d8be9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 09:04:06.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2fe68f8d35214e6c80638c0b91bb86c3\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/f2bded828db447f3b76e0f1b26c7df36 : SUCCESS in 3m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/499c40c7aacc40a29b004c76f426961e : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7aba3f274954f94bd12b354ea6e6ea4 : SUCCESS in 3m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a6a5c8a467d04826a684c66d16b62782 : SUCCESS in 4m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/56d3a86511ec4972a9462a06aca9f764 : SUCCESS in 6m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a8d3dfeddd8246ccbac3a3332951f3ff : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f40637a374aa4a7ba7d9a76e366ad52d : SUCCESS in 9m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/59e5a2df3fb44074b2bdab7499869492 : SUCCESS in 4m 31s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/406e453445f04c84b3f5c34bb64cbf5a : SUCCESS in 46m 33s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/3b1359f0bacd4f508fd5d116a84868d8 : FAILURE in 1h 15m 59s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/3d67e76f71b44109a728050809550a26 : SUCCESS in 1h 29m 29s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/dc0ab1e6d8104b3688b1916e6efd29e2 : SUCCESS in 1h 55m 31s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/9aa6e0e9f7554f8fbee16046895a65f9 : FAILURE in 48m 42s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/8626bf13f2ea4668925fc7511a235188 : SUCCESS in 1h 43m 41s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/3fa85768fa9a40f48d68853bf901279c : SUCCESS in 1h 16m 25s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/543db7f4508c40128cdc77aa393f6412 : SUCCESS in 1h 38m 10s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/8fb7c0c339c44adba89b81224f83416e : SUCCESS in 50m 16s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/c4e22b8480bb4b67936cbb3ad6a27310 : SUCCESS in 1h 29m 42s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/d855572ca67944d7aad3884a7c0a4b0d : SUCCESS in 1h 23m 35s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/0064051e1d0242b9a57c362bd6a69e80 : SUCCESS in 1h 34m 45s (non-voting)\n- trove-tempest-ubuntu-base-keydb https://zuul.opendev.org/t/openstack/build/d6d984e4d1ff491893c472ca92630385 : SUCCESS in 35m 22s\n- trove-tempest-ubuntu-backup-keydb https://zuul.opendev.org/t/openstack/build/e88dd6842fb74f74b7ec74b06cdd6cb6 : FAILURE in 1h 35m 48s (non-voting)\n- trove-tempest-ubuntu-replication-keydb https://zuul.opendev.org/t/openstack/build/5d8023e0029349b9895e208350a9aea7 : SUCCESS in 1h 03m 30s (non-voting)\n- trove-tempest-cinder-storage-driver-keydb https://zuul.opendev.org/t/openstack/build/f5f7397fc1ba47bf8cd3435434d89a9a : SUCCESS in 1h 12m 04s (non-voting)\n- trove-tempest-ubuntu-base-valkey https://zuul.opendev.org/t/openstack/build/f8774a67c5fa4b6f9491fccb434f6e8f : SUCCESS in 51m 04s (non-voting)\n- trove-tempest-cinder-storage-driver-valkey https://zuul.opendev.org/t/openstack/build/23aecd073e59405a8468c0f30048254d : FAILURE in 1h 21m 58s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/84686cbd3d02450bac11c976c4830462 : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":21},{"id":"46f0db4273b8eec6a1e37f1a0c39b7435546509e","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-05 12:58:51.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"1d8d7cb2bbf48b769e1bba1bbe82f0d5ff99aa35","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 13:17:28.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c8e56db1965454ab1807ab9f6fb7ff7\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2a54a9522db74953bbb91f4e6cf0fd8d : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9007ad6eb3464d0b83b17d520cf5a1af : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/6646fed81319405cbfd278d83bc5583a : SUCCESS in 15m 18s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"e7b70dc701d71af047b7054245ad56603eaf70d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 13:44:41.000000000","message":"Patch Set 21:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/trove/+/990725 is needed.","accounts_in_message":[],"_revision_number":21},{"id":"244953c5f915d6e08469ec2f48f0effa74b547f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-05 15:14:57.000000000","message":"Uploaded patch set 22: Patch Set 21 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":22},{"id":"ed2ba52d1c6e2923f44bc528bea07269886a22eb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 15:31:29.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/859e368a6f18443ba7876acb693ef540\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/562e72861ecc495a9bfb8f00421c96a5 : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/23edde8051704a469ac2e2912a55f1aa : SUCCESS in 8m 30s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/bed4302f097b4df0817ff8eb36883781 : SUCCESS in 10m 24s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"08bd2caae4d57e032726f3556c2de9f7c63807c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-05 17:40:57.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"6a51a1ccfa71bcca00e67c13844e7e1d7a5c7330","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 17:54:36.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b23101f0a6b740728132b4d5aff8112e\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c6b2858de42a4a2a831f9e5619b7bfbe : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/27614eba0fc24885bc038ee56dd02447 : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/606e51f7708344589ce38a7e3dd1b29a : SUCCESS in 10m 30s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"89cc67039ac6ec27aa770f70d5586cba4b053ec0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-05 19:42:17.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"a87b0151988dc60ae034dc537071b53c6a1e70b4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 19:56:55.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64e60cb3ab9f4f7ba66326eb362f2ff6\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/615f59e3369d4b8390c6b872cf8e1d66 : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/079d4eb28a94494aaddc828cca0c6c1f : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f697585dbda0409993d08c2c446b06cb : SUCCESS in 8m 01s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"841e6a69246c418c6394bcf9465424b91c2a7621","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 22:35:47.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3236297ac0a54364a54049cda9d364a1\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/eb5a8e263ef949d5a0d3b8c12b8f6aa0 : SUCCESS in 3m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a95ef66763f54519bf1f44d5a24a7f27 : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df093489720340e7a84277d0428cd4fb : SUCCESS in 6m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2932b02b64d743a18e46ea44d9353ac7 : SUCCESS in 3m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2cdaf87e73e2482db1c2907fcbe28970 : SUCCESS in 5m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/254bf9c0f84747b3a42bdc6077d16016 : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d378f79771943008eb1a6613c7ebfb2 : SUCCESS in 10m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/124f514d008f4a5d89c95f608b70ca21 : SUCCESS in 8m 58s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/ed918ba0702b4e5f917404ece9c0108a : FAILURE in 1h 34m 39s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/6b2de4b4970a441a88f58a02334a0c70 : FAILURE in 1h 37m 19s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/3e5f107d2b7342ea8432591c3f1d65e1 : FAILURE in 1h 36m 21s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/a6d0741773c148428bded3a2aefa0656 : FAILURE in 2h 15m 18s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/be38a856f763476fa8aeaa07c78d8ed7 : FAILURE in 1h 33m 53s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/a9060df5c1cc4348a38771c30e3dc019 : FAILURE in 1h 06m 41s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/59530e87342240febe52af2478e26947 : FAILURE in 1h 06m 14s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/9b0ed5bbefcb4821a977bda8addd9a16 : FAILURE in 1h 43m 40s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/36fa63b9424f4b738b132937ecb73435 : FAILURE in 1h 33m 33s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/d680d6cf64004d57bfb3554fc87cdea7 : FAILURE in 1h 37m 25s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/4931b34e1d974f078edc663800bffc95 : FAILURE in 1h 06m 59s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/22e6f24b4f0148cba07c613fc6b02a9f : FAILURE in 2h 16m 30s (non-voting)\n- trove-tempest-ubuntu-base-keydb https://zuul.opendev.org/t/openstack/build/96b35afb4eab474b86007b6ce42e3730 : FAILURE in 1h 35m 28s\n- trove-tempest-ubuntu-backup-keydb https://zuul.opendev.org/t/openstack/build/d18fdf616f7342c38d231b5f564bb3e0 : FAILURE in 1h 05m 57s (non-voting)\n- trove-tempest-ubuntu-replication-keydb https://zuul.opendev.org/t/openstack/build/f1046393947f40d6b3d8ccd3643d8d34 : FAILURE in 1h 07m 51s (non-voting)\n- trove-tempest-cinder-storage-driver-keydb https://zuul.opendev.org/t/openstack/build/eba88f80132840409769d9e02b012b70 : FAILURE in 2h 45m 50s (non-voting)\n- trove-tempest-ubuntu-base-valkey https://zuul.opendev.org/t/openstack/build/6503c43964344cb89b816dc252e68b67 : FAILURE in 1h 03m 03s (non-voting)\n- trove-tempest-cinder-storage-driver-valkey https://zuul.opendev.org/t/openstack/build/ce17b7585cbe48ff990a4e6c6cc6eec6 : FAILURE in 2h 15m 09s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/135070e5be024117bb4f90319f4f8f36 : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":24},{"id":"4190a3d984de81e5c4ede05fc8037a4d88ab9f01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-06 06:27:52.000000000","message":"Uploaded patch set 25.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":25},{"id":"bf6d582e1a8e21e19c81765b59a17bf589f28fbf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-06 06:38:43.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/366c6a1aad2141f098e7be8a002ea872\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/28ce42e350b64f19839860986c3bb57e : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/58473479a5d1452db9037fed906ea77f : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e19dcf57efa0482394bc518319be6b16 : SUCCESS in 8m 06s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"6e3c84821ab1d641836534fa2b039082cc1390a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-06 08:44:26.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7769176c53b0461cb5623391d288f036\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/1397ad6f33c742de971c9f817e2cd80e : FAILURE in 3m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dea7ab6afec146349a819df223b33134 : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6668e4fdffd04babbdcac14397a516cc : SUCCESS in 8m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/15a78e588f424d2b887c15b9d5452add : SUCCESS in 2m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d462cb43c9f04a34b5b19f7e85034644 : SUCCESS in 3m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dd123eaf48914b679aad5029e1a4b4eb : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f77af1b982cb46e9a25d8123cabc10ea : SUCCESS in 8m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/317fbc42136444e5b08d7cfaf2d52a85 : SUCCESS in 7m 51s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/3094a6dd88dc40f582e842a628a90c41 : FAILURE in 1h 00m 32s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/1ad3747d96eb4ffb8400775e628a01f3 : FAILURE in 1h 05m 51s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/2cf11136c96b497b9b52de67722bfa8d : FAILURE in 1h 36m 02s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/43f366d0714e4e749b76d99622fb48f6 : FAILURE in 1h 56m 41s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/2695cd7e75a340feb1bc486dbc3c6866 : FAILURE in 1h 33m 11s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/d967fdaeaf6b41448401ad66668477d9 : FAILURE in 1h 37m 17s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/71de1966da8843d5a70b66c915274d18 : FAILURE in 1h 11m 01s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/11329695ff6a437ebc3144f691a0419c : FAILURE in 2h 14m 13s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/9bb8386f94ab48d88f4437f02476702d : FAILURE in 1h 32m 31s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/122cecb1b16247b0bc46d31f017733ab : FAILURE in 1h 10m 30s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/5ff21d4a33894ccc95977395a408615d : FAILURE in 1h 13m 00s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/d0bdac18ad094aa581075f85c78a22d3 : FAILURE in 1h 43m 58s (non-voting)\n- trove-tempest-ubuntu-base-keydb https://zuul.opendev.org/t/openstack/build/cdfb164fc2c54cd29e181748877a4931 : FAILURE in 1h 38m 22s\n- trove-tempest-ubuntu-backup-keydb https://zuul.opendev.org/t/openstack/build/703bab0dc0934539bd211ecf9f16b084 : FAILURE in 1h 35m 59s (non-voting)\n- trove-tempest-ubuntu-replication-keydb https://zuul.opendev.org/t/openstack/build/ecd899e629a54ebf890b05857c8a359a : FAILURE in 1h 15m 31s (non-voting)\n- trove-tempest-cinder-storage-driver-keydb https://zuul.opendev.org/t/openstack/build/738b1adeb07e450aa645bf57025a8b4a : FAILURE in 2h 14m 49s (non-voting)\n- trove-tempest-ubuntu-base-valkey https://zuul.opendev.org/t/openstack/build/94b94a392024425da4acaabc9c11b6ba : FAILURE in 1h 07m 34s (non-voting)\n- trove-tempest-cinder-storage-driver-valkey https://zuul.opendev.org/t/openstack/build/260483362879421d8734ddff3210dbb0 : FAILURE in 2h 12m 53s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f424672fe3c042158d5c30ded913479f : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":25},{"id":"2df7eb8d0933d3a918859d2c6527253da671d5dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-06 15:14:09.000000000","message":"Uploaded patch set 26.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":26},{"id":"d8d0f1a345dac0d5e4eec4cabbf85307f5e50c99","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-06 15:32:03.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9470c2e7b16b43f29fd627d6e66447f5\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/212134aca97b4cb7bf44e0dd61b82dd5 : SUCCESS in 7m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a79ca8cf955944d49a9bcff8c21cfd9e : SUCCESS in 8m 13s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ede919ff23314897895e7099f4ea1f9a : SUCCESS in 15m 03s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"a04c3353ea434bf080925d6c3342683db5f2f48e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-06 16:31:05.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"23b20963c828cbe1e70b4afc5acd758d10634946","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-06 16:46:14.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de44309e1a934bc6a1067ea7b7d30b7f\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3c50a8230590415eaf105665a1117742 : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/374c7c6f17d94c6aa4fbb30b3934d8ce : SUCCESS in 7m 43s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ba6b4d700bf14ddaa7c8b51a85e90481 : SUCCESS in 11m 59s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"9744e27636353c6d9a621050b47c1b7a5f99a2c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-06 17:22:06.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"a2aa4c9c9d31e6cdf8f7d3ea508866c240382b6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-06 17:30:30.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"c7e6f88a980fe14f27fe52542b9a849728976b31","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-06 17:44:05.000000000","message":"Patch Set 29:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50ed21e9033549c1a81f6ff1648a2929\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4af82c222c9444cbb92e9a164cfef743 : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c495f099c3874bc1a2ab526d5be01dd4 : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e253c952084543bca7022495e966421f : SUCCESS in 10m 45s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"d0ce0ce355503eaeb7e2c05479893ea3e25e6bf4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-06 18:47:33.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"f7cb123465f6bb27a922079582b5968be4ba7fd6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-06 18:54:18.000000000","message":"Uploaded patch set 31.","accounts_in_message":[],"_revision_number":31},{"id":"c78c28fe91d1ed23ff86075a6938bbb0178cf4c3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-06 19:05:19.000000000","message":"Patch Set 31:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef62597a0346413a93d8bdfb2aaf25a4\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/175fd11b2dfb4c4f889f561415a16c35 : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e8e1ded161c84269a0412b4f8c8fb8fa : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/121946c33ada4d848ccb8b41b14b0749 : SUCCESS in 8m 21s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"93640eeec34302f8c6d72bcc8f87a7b8bbf6c2fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-06 19:09:22.000000000","message":"Patch Set 31: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f3d2663b42674b32a2d4ee6ea16967da\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d825524e006a448e80e98bbc1100eddf : SUCCESS in 3m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/817c170492c9444d9d592df8f006805e : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9cc8111536548e68786db68269b1a5e : SUCCESS in 5m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d309174544b2443dbd402c531d1fe22c : SUCCESS in 3m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9d2b7aa9981b4f5693fa8789f37e1f7f : SUCCESS in 2m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2d53bc50774e45fbbe68a4e016ff8d1a : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89e4453902e54298b1056dd016aa459a : SUCCESS in 9m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ca2928561774277bb791d1b7c329d77 : SUCCESS in 4m 35s\n- trove-tempest-ubuntu-base-keydb https://zuul.opendev.org/t/openstack/build/fbe142fdbf8b42ec9fd3bdd12d2bad30 : FAILURE in 6m 40s\n- trove-tempest-ubuntu-backup-keydb https://zuul.opendev.org/t/openstack/build/f23da2dbd602462aaaefd2f969a8189a : FAILURE in 8m 37s (non-voting)\n- trove-tempest-ubuntu-replication-keydb https://zuul.opendev.org/t/openstack/build/6d61784286d64d19b602c57e657e1b32 : FAILURE in 8m 50s (non-voting)\n- trove-tempest-cinder-storage-driver-keydb https://zuul.opendev.org/t/openstack/build/e0fc51e049e440178708547d31114938 : FAILURE in 9m 09s (non-voting)\n- trove-tempest-ubuntu-base-valkey https://zuul.opendev.org/t/openstack/build/934a3ea32a4d4dc2977484bf87d1b1b5 : FAILURE in 6m 26s (non-voting)\n- trove-tempest-cinder-storage-driver-valkey https://zuul.opendev.org/t/openstack/build/d1ff2300c740448588dca48dda173152 : FAILURE in 9m 04s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6b29cc783e0844d780f192064bc5e05f : SUCCESS in 3m 14s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/d7fd1b011dda4d4da84791e0a8d41842 : SUCCESS in 8m 12s","accounts_in_message":[],"_revision_number":31},{"id":"88ae1c7c5885a9cb9305da9d7a8d16941351e9ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-06 19:09:50.000000000","message":"Uploaded patch set 32.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":32},{"id":"0f9c9bc06c1f23c433398ae755c5faa2c1157f3b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-06 19:22:00.000000000","message":"Patch Set 32:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25d2057a197841c1951ac34454cbf5b7\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f9c76e3a646445d8aeadf637c1266799 : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ed4e61b7197146aa8b556737d3f4cf10 : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/fae2265335a24f199d5d01fc94a1d743 : SUCCESS in 9m 25s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"5fb5ea905457228f5fe0a9d9460036a8211b306e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-06 19:22:23.000000000","message":"Uploaded patch set 33.","accounts_in_message":[],"_revision_number":33},{"id":"4bb5293c12a769af6359f54a0add7d2b21fdcc7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-06 19:34:19.000000000","message":"Uploaded patch set 34.","accounts_in_message":[],"_revision_number":34},{"id":"0a40b3c42f796eff59743b78a7bbd005d5ec9ebc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-06 19:46:08.000000000","message":"Patch Set 34:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0ec433d560540a28be24c3173e176cb\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c72b63b39aae42afafd0ef0f6fd2e2fb : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ef0c56bfd68c4b78849fe0399ab6499d : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/483be59eadf64d96992ac9215450f4fa : SUCCESS in 9m 10s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"1d294225e2886276f52ff02865f89eb1675bb3ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-06 22:39:57.000000000","message":"Patch Set 34: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ccbc4d2216d3428193fe4567f55a7a34\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/cddedbef8bee4f17950ea9b7839c66a3 : SUCCESS in 2m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2299e5f08c74b3aa7d26eba9e7bafd8 : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6b6c943a2d546a09e16295485b8b6bd : SUCCESS in 4m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a5f2780f8ad848368f8fc2fc704fd2fe : SUCCESS in 4m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/860189a4ab784fcd88d7d13ae6510bfd : SUCCESS in 4m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7b61d49c9a6e47649b2040bb582b9397 : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36fd43f02bca4b9686f0f5b0e080561b : SUCCESS in 9m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/941e6704ef5e42548b30f4207c9c9937 : SUCCESS in 3m 26s\n- trove-tempest-ubuntu-base-keydb https://zuul.opendev.org/t/openstack/build/94ca4aeac0944ef28cb670284d9beaf8 : TIMED_OUT in 3h 03m 48s\n- trove-tempest-ubuntu-backup-keydb https://zuul.opendev.org/t/openstack/build/ab758e072e054e25900bb9b5c5c09bb6 : TIMED_OUT in 3h 02m 39s (non-voting)\n- trove-tempest-ubuntu-replication-keydb https://zuul.opendev.org/t/openstack/build/4bc16298a935417bb8b3ba1635f26e48 : TIMED_OUT in 3h 02m 59s (non-voting)\n- trove-tempest-cinder-storage-driver-keydb https://zuul.opendev.org/t/openstack/build/9a2bf090cad84c628eadcc2944c4cca4 : TIMED_OUT in 3h 02m 39s (non-voting)\n- trove-tempest-ubuntu-base-valkey https://zuul.opendev.org/t/openstack/build/ac53598c668644b9b42b879e4a5a8088 : TIMED_OUT in 3h 02m 42s (non-voting)\n- trove-tempest-cinder-storage-driver-valkey https://zuul.opendev.org/t/openstack/build/aa613b6506e04fef9f01c57f56ed3c26 : TIMED_OUT in 3h 02m 51s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cacec87fa3114388ac4713bfef90b649 : SUCCESS in 4m 47s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/f31173f876aa426589f7f2a9b6c301ec : FAILURE in 5m 29s","accounts_in_message":[],"_revision_number":34},{"id":"aa5b335969a66f6f47cade9b075810f5c39157a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-08 06:26:10.000000000","message":"Uploaded patch set 35.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":35},{"id":"f7c2ec14579fc604111ca3f50dae91c8bc560999","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 06:41:16.000000000","message":"Patch Set 35: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/43b56799172745f9bae8184a8c77eb15\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d454ef40cf874a5ea303776487f2539f : SUCCESS in 3m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/851b9cbf41334aac85bd587977927513 : SUCCESS in 7m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17caeb3f184d448e99b8f8fafd6806a4 : SUCCESS in 5m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/70b30c0835454b2eb0ff53ab339e10ac : SUCCESS in 3m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a77062e2adcf4991a9559668d830551f : SUCCESS in 2m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f9430e20f1654a07b21a204582c406a1 : SUCCESS in 8m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8159e11650a445f9084dc38cd592d41 : SUCCESS in 9m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/09c81767a8fb46518d6ad7de26013614 : SUCCESS in 3m 37s\n- trove-tempest-ubuntu-base-keydb https://zuul.opendev.org/t/openstack/build/9c313139288c42b8bdb427ed7544099f : FAILURE in 10m 36s\n- trove-tempest-ubuntu-backup-keydb https://zuul.opendev.org/t/openstack/build/a640c0979cab4af9a42e7a390a85ec31 : FAILURE in 13m 20s (non-voting)\n- trove-tempest-ubuntu-replication-keydb https://zuul.opendev.org/t/openstack/build/611b8bd3030446618f2c3a609bc50c9d : FAILURE in 12m 04s (non-voting)\n- trove-tempest-cinder-storage-driver-keydb https://zuul.opendev.org/t/openstack/build/b0f0d11a6e3841e6aa5a317376d50e7c : FAILURE in 10m 48s (non-voting)\n- trove-tempest-ubuntu-base-valkey https://zuul.opendev.org/t/openstack/build/316c5573b507423ba48de6006977f13b : FAILURE in 10m 23s (non-voting)\n- trove-tempest-cinder-storage-driver-valkey https://zuul.opendev.org/t/openstack/build/bab8921d34a34446bcace479ba3d47f4 : FAILURE in 10m 35s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/05d3ecf3f845475f8f2dda4e04e0b8c7 : SUCCESS in 3m 27s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/95d6c9d8d4504fee9d552ef2570e68f7 : FAILURE in 3m 07s","accounts_in_message":[],"_revision_number":35},{"id":"d78082e829c2e51aa51a9bc2230bbcb5abc79163","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 06:47:06.000000000","message":"Patch Set 35:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15fb504fa0b345a2a2311cc6224c3cb1\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/25c5a6de46b64a7eb78d8e0f73137d63 : SUCCESS in 8m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e856a86512ae4f26995985cbafd12e27 : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2d9313e9f0774e97872b18ffa3e59510 : SUCCESS in 13m 28s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"4ec26fff940d982d9b5176092fc10f4dda7126f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-08 07:03:54.000000000","message":"Uploaded patch set 36.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":36},{"id":"7b44fbc9d6714f4df0019d758137f0d7fb47e698","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 07:39:04.000000000","message":"Patch Set 36:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a86d45cc73084a30ac2a3c018b25c607\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c4db4d828bda4917bcb6fc1f4affd2dc : SUCCESS in 9m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/783e3113b63a4c8a9a199897f31c32a1 : SUCCESS in 8m 07s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/9744e38a9dfb464b9ac95d7d99f1b213 : SUCCESS in 17m 02s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"21a2fb1d7d6a66c90e1ed34a4283f44307e8ffc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 08:11:44.000000000","message":"Patch Set 36: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/29016d8572184eac95bbf37bfabfa8b8\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/98fb46dd6ba846048289809acff86f97 : SUCCESS in 2m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1dd0a864ab854f8890479f05a9e352c3 : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c65793739776404c86b8cb8633091cbc : SUCCESS in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b7840b031e0f4ed59c05a06b5eceada9 : SUCCESS in 4m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/79a3378db65d40e3ac49811b68f1e81e : SUCCESS in 5m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c6275b1f326a4c1cae8c653c1b6b6735 : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e859226d327947a581f4be27561ac078 : SUCCESS in 5m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8340328a14f4b369df8eee378d0569c : SUCCESS in 7m 50s\n- trove-tempest-ubuntu-base-keydb https://zuul.opendev.org/t/openstack/build/ed4c9adeaae047149d4b96493b97c428 : SUCCESS in 49m 28s\n- trove-tempest-ubuntu-backup-keydb https://zuul.opendev.org/t/openstack/build/9758ba25e3174b25ae4fd1830299752f : SUCCESS in 1h 03m 32s (non-voting)\n- trove-tempest-ubuntu-replication-keydb https://zuul.opendev.org/t/openstack/build/87fdc147e76d4e8788b3033f2339b7b1 : FAILURE in 8m 10s (non-voting)\n- trove-tempest-cinder-storage-driver-keydb https://zuul.opendev.org/t/openstack/build/9078e09f7f2147e4aa906f94c015cfae : FAILURE in 42m 44s (non-voting)\n- trove-tempest-ubuntu-base-valkey https://zuul.opendev.org/t/openstack/build/5a14b97a7caf4b56b0c8af57a110fbdd : SUCCESS in 1h 00m 16s (non-voting)\n- trove-tempest-cinder-storage-driver-valkey https://zuul.opendev.org/t/openstack/build/445db2d3c1d7466dab1dd4159eba66c3 : FAILURE in 48m 03s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/920319c0535e4028970e1276f475b8dd : SUCCESS in 3m 32s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/a97bf85052ad4c34869568d8dd826b48 : SUCCESS in 6m 42s","accounts_in_message":[],"_revision_number":36},{"id":"e30faeb40a245a3ac11c1539975e390ae95b6c0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-08 11:05:55.000000000","message":"Uploaded patch set 37.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":37},{"id":"5667afd02041525d6a4d29f7ffd0bdda5ae856a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 11:30:34.000000000","message":"Patch Set 37:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a23a9bd3a7644b959cf2f1fb3e71cf78\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f0a19d74588840eca83cc751a667a6c0 : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b23865aab8fc451bbe9af23b9b3c7975 : SUCCESS in 7m 34s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b1423778db5a4b7ba46b3bf6aaaeee95 : SUCCESS in 13m 38s (non-voting)","accounts_in_message":[],"_revision_number":37},{"id":"ec80099c6bdd12dcbdf5347bdc47da2199b8b6cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 12:02:23.000000000","message":"Patch Set 37: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ff6f475635804694a649bfb5b2d7c658\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/6cb5575b33544ea1bb0c6f8c78d8c889 : SUCCESS in 2m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/caa4b0f8d93643828bfa40de0fa98a11 : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cce17b7050b940bb9cef3bfc63085a19 : SUCCESS in 5m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1d44f0b5aeab49be9ec7759e74c0c24c : SUCCESS in 3m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fc63f5dc44c3480585add61f2033b53f : SUCCESS in 4m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c074340fafd44379be1ea70da76dc88b : SUCCESS in 8m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1a28fee535a43508fa9852560956500 : SUCCESS in 11m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aff3fe9dd0b84af99ab8232df6c8e215 : SUCCESS in 8m 06s\n- trove-tempest-ubuntu-base-keydb https://zuul.opendev.org/t/openstack/build/7517e14eec624b2ca649146a723fd688 : SUCCESS in 39m 55s\n- trove-tempest-ubuntu-backup-keydb https://zuul.opendev.org/t/openstack/build/7e0e31b26741444a9761793b8e1cb115 : SUCCESS in 52m 47s (non-voting)\n- trove-tempest-ubuntu-replication-keydb https://zuul.opendev.org/t/openstack/build/941e5d1a48ce4a269d7808871cbe72de : SUCCESS in 52m 09s (non-voting)\n- trove-tempest-cinder-storage-driver-keydb https://zuul.opendev.org/t/openstack/build/a24df78b26ad4ab699b13ee1b8ff4624 : FAILURE in 45m 02s (non-voting)\n- trove-tempest-ubuntu-base-valkey https://zuul.opendev.org/t/openstack/build/d7f534a74ca24087b526d0dcdac40c18 : SUCCESS in 46m 03s (non-voting)\n- trove-tempest-cinder-storage-driver-valkey https://zuul.opendev.org/t/openstack/build/7bd6ad963eab49489aac827bf39cac2d : FAILURE in 47m 06s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/97dfc0dc7eea4355b0ca3b7fa7f2a2a8 : SUCCESS in 5m 51s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/5e57efc785fb473391dfa2342675a6af : SUCCESS in 8m 08s","accounts_in_message":[],"_revision_number":37},{"id":"e53d5df2ea870b80c89f8368f16fd817f537aa51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-08 12:43:19.000000000","message":"Uploaded patch set 38.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":38},{"id":"81578e1735cdcb1a98e9371bf53180978117ea19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-08 12:44:22.000000000","message":"Uploaded patch set 39.","accounts_in_message":[],"_revision_number":39},{"id":"d98b4cedf56a59ba4bd575ff7253abb6890bd45f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 13:06:50.000000000","message":"Patch Set 39:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46d1261b73d5405a8caa9f15a4f47b81\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/761002ae19ce4f0eac6d193666ed4614 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bf7fe944366542e7b7932ade7a112eca : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/390e52c6fc1d416e8a9a218a36187245 : SUCCESS in 8m 18s (non-voting)","accounts_in_message":[],"_revision_number":39},{"id":"686610dff65436081c8b271fa79692eb586c87d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 13:51:16.000000000","message":"Patch Set 39: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6e4ef11f9256454494acb1df8f429fe0\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/0a31c5abe3334b6fadb751858c764141 : SUCCESS in 2m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ad6d0de7d844bffb7f88022667c7b34 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/237fbbe6f5fa4a5dad467da7780ca9e9 : SUCCESS in 5m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c2681653faee4d88a7a41714fbf6f0a1 : SUCCESS in 3m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3868cba0eee24548a06492cfa84571ae : SUCCESS in 4m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c97c4f59ea9a4725992cddfeda85021a : SUCCESS in 7m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bef994f0426f47f9924ae97bab9bed63 : SUCCESS in 10m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de332fc26cda4b3f918b6d921ca850fb : SUCCESS in 8m 22s\n- trove-tempest-ubuntu-base-keydb https://zuul.opendev.org/t/openstack/build/4623a021e29e470a9b14e437dbc7119c : SUCCESS in 40m 31s\n- trove-tempest-ubuntu-backup-keydb https://zuul.opendev.org/t/openstack/build/8d506eb02908434f915dece6fc2e671b : SUCCESS in 1h 02m 47s (non-voting)\n- trove-tempest-ubuntu-replication-keydb https://zuul.opendev.org/t/openstack/build/c057ea8a68914ff1b1d5aa9362c8b8b2 : SUCCESS in 56m 37s (non-voting)\n- trove-tempest-cinder-storage-driver-keydb https://zuul.opendev.org/t/openstack/build/abf06668bced4576af4eb2430adc395a : FAILURE in 44m 14s (non-voting)\n- trove-tempest-ubuntu-base-valkey https://zuul.opendev.org/t/openstack/build/d2cdeb4bd99048a197e9ddfd82f56e67 : SUCCESS in 39m 42s (non-voting)\n- trove-tempest-cinder-storage-driver-valkey https://zuul.opendev.org/t/openstack/build/38e74eb1016e47afa72a5c2d9fef2cd2 : FAILURE in 37m 24s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ec1bb05ca4c3471eab1ba85b80f54d5f : SUCCESS in 4m 52s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/fe065c5c97264b858a25d4465f799048 : SUCCESS in 11m 17s","accounts_in_message":[],"_revision_number":39},{"id":"d5e537f954bb1e65a9f6113ddec8e9e9ad14c164","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-09 07:29:06.000000000","message":"Uploaded patch set 40.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":40},{"id":"67f3a92b0a03a0cb181a7fc407280fb8765cdf5c","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-09 08:49:11.000000000","message":"Patch Set 40:\n\n(1 comment)","accounts_in_message":[],"_revision_number":40},{"id":"a59b68a981c049acf7ffd4c2acf3c309772824a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-26 06:13:38.000000000","message":"Uploaded patch set 41.","accounts_in_message":[],"_revision_number":41},{"id":"afc91d65932ee7f1651ee26299ee18ed2d1f98d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 07:02:49.000000000","message":"Patch Set 41:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16321ed61b3643fbaffd3be7dd12686d\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3fca53b666984a498e8456017926aabc : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/30fcc246e1de486587957c9825657051 : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8b90fd8d144645ec8711271a9be5e981 : SUCCESS in 16m 55s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"64a2ce5bc17eeac216a02f97bb8a8e91b44b9181","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 08:07:22.000000000","message":"Patch Set 41: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2eca34471304795861ce8b52b7da531\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/cc6a33308bda4d9a9651da952e527660 : SUCCESS in 3m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f487afe11dbf44708dc71c6142e436d3 : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93dd950c7665432591c1285698803764 : SUCCESS in 5m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d7c5771bc59449f099e04d7189a9c4ff : SUCCESS in 4m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a37f59b5a0d546ca823613a2fd57dd78 : SUCCESS in 6m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b4fe2a9510654664a6eabc931214c527 : SUCCESS in 7m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78aed06f39fc4c5aaadd2d0d7214576a : SUCCESS in 9m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d201c20fa9244be08b990cfef921644d : SUCCESS in 8m 13s\n- trove-tempest-ubuntu-base-keydb https://zuul.opendev.org/t/openstack/build/d0edfded289149df995f4f44483707c0 : SUCCESS in 49m 01s\n- trove-tempest-ubuntu-backup-keydb https://zuul.opendev.org/t/openstack/build/4f5095f8c8dd40ebacdff99857925882 : FAILURE in 1h 00m 43s (non-voting)\n- trove-tempest-ubuntu-replication-keydb https://zuul.opendev.org/t/openstack/build/8b0ce777d71a4e8b992a54f48159ea33 : FAILURE in 1h 00m 15s (non-voting)\n- trove-tempest-cinder-storage-driver-keydb https://zuul.opendev.org/t/openstack/build/00b5fd1ef2674599b4649f4458c712f0 : FAILURE in 1h 50m 56s (non-voting)\n- trove-tempest-ubuntu-base-valkey https://zuul.opendev.org/t/openstack/build/9b93501afb4d48baa00374a90f6a2e36 : SUCCESS in 40m 13s (non-voting)\n- trove-tempest-cinder-storage-driver-valkey https://zuul.opendev.org/t/openstack/build/7790a09070284471b7be1aef4fccb6e0 : SUCCESS in 1h 07m 50s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ebb4bd99fc28497eba6da9844fe163c5 : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":41},{"id":"53cb2241f6e6d2fd54e7e0c467172ce922d2ca24","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-26 08:37:30.000000000","message":"Patch Set 41:\n\n(1 comment)","accounts_in_message":[],"_revision_number":41},{"id":"49339b097d6f81b4aa21c21647fcac2074db2fe1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 08:56:58.000000000","message":"Patch Set 41:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52d44bd896b94083adeeda77e1226df3\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9b387bf90c534886896ce35bf8bb12dd : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b156dee111e744408a542456d12f9c81 : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/6f7a75269421430a91fa67f4661dc1d5 : SUCCESS in 13m 52s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"2bdca6cda0d77ffe71ad2c4cb806bfe28a5c4a21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 09:54:26.000000000","message":"Patch Set 41:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e62d25aa848459b9835dcabf6a02aae\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/915bdc33f68e4e928db532e31602bb1d : SUCCESS in 3m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d8bf718c26e4682a449b12c2adc4d80 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/723bf466c6424a6087663599b60cedb4 : SUCCESS in 2m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b802b545b782455cbe1e378012dcef6f : SUCCESS in 7m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f6b480a5e29a4a9eaf54ab28dd8a2633 : SUCCESS in 4m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dbb145a97ae64be6a1c208721721ae12 : SUCCESS in 8m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/505bc571d26e4f038fd580c76b13968c : SUCCESS in 10m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/261e78c0f87043618f6811b1e447c4a0 : SUCCESS in 7m 01s\n- trove-tempest-ubuntu-base-keydb https://zuul.opendev.org/t/openstack/build/28b23069c6a248b4baa9e1b5f98df72a : SUCCESS in 54m 02s\n- trove-tempest-ubuntu-backup-keydb https://zuul.opendev.org/t/openstack/build/a7119042bac349a69b79e60af1d31660 : FAILURE in 58m 49s (non-voting)\n- trove-tempest-ubuntu-replication-keydb https://zuul.opendev.org/t/openstack/build/f258cc09ed524650b33957f676ddc790 : FAILURE in 56m 16s (non-voting)\n- trove-tempest-cinder-storage-driver-keydb https://zuul.opendev.org/t/openstack/build/aab872ba79e94b2fa1b436bcc9cc9e28 : SUCCESS in 1h 13m 03s (non-voting)\n- trove-tempest-ubuntu-base-valkey https://zuul.opendev.org/t/openstack/build/073d544f2340470a82843478e07967d0 : SUCCESS in 44m 38s (non-voting)\n- trove-tempest-cinder-storage-driver-valkey https://zuul.opendev.org/t/openstack/build/d94e8ac06bf947d98b5294b057cf9cb2 : SUCCESS in 1h 15m 38s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/83d3deff09c94a8e8014da43e0309e0b : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":41},{"id":"a4f015d4b08f83d059c23f190cd2e1e7bd824a65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-26 13:34:15.000000000","message":"Uploaded patch set 42.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":42},{"id":"b4d94158f2a6e87203f110cb84f2045115a79364","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 13:51:05.000000000","message":"Patch Set 42:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72a23e6bda2a488d912fc644fa3f8304\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/568efd25dafd463599397439c3f6e3c7 : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9723ad240d8a4ee6861bf063bf13abdb : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/547251ff316b450fb0c47119f64ff719 : SUCCESS in 14m 09s (non-voting)","accounts_in_message":[],"_revision_number":42},{"id":"2e15c977b285141bb043dc4af98a47e756c0101b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 14:51:45.000000000","message":"Patch Set 42: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c567370efaf41d3840dd84b915cbbe4\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/7a9a227288d840e0a353485610a06ddf : SUCCESS in 3m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f133d015fce64592909f41c562b67379 : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4226db67af24d709608053a930ee3d8 : SUCCESS in 3m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f8b482e17f054e6daaf0ef792143c510 : SUCCESS in 4m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7769bc5d2bf545aa949019125a2b3f99 : SUCCESS in 3m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e65737adb2674a05a7da0414625f8932 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e3a9ebaab1f422592b8e8d9ed5be0c9 : SUCCESS in 9m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aea6714a4cd2407f871873a317223d6e : SUCCESS in 7m 38s\n- trove-tempest-ubuntu-base-keydb https://zuul.opendev.org/t/openstack/build/142742a43138406ab55e8f61ed320cc6 : SUCCESS in 39m 30s\n- trove-tempest-ubuntu-backup-keydb https://zuul.opendev.org/t/openstack/build/5aec5bcf8a3c4513aef671d0aee030c4 : SUCCESS in 57m 49s (non-voting)\n- trove-tempest-ubuntu-replication-keydb https://zuul.opendev.org/t/openstack/build/e99ec63f79184c3199b315363b49a9bc : SUCCESS in 59m 02s (non-voting)\n- trove-tempest-cinder-storage-driver-keydb https://zuul.opendev.org/t/openstack/build/1d5db24c27fb4c21bb0e3da35c7b8255 : SUCCESS in 1h 12m 01s (non-voting)\n- trove-tempest-ubuntu-base-valkey https://zuul.opendev.org/t/openstack/build/57ce270fb263490996a1b934af7ef418 : SUCCESS in 39m 42s (non-voting)\n- trove-tempest-cinder-storage-driver-valkey https://zuul.opendev.org/t/openstack/build/89dbfc466683447886829975dd489ad9 : SUCCESS in 1h 10m 25s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c606d55b3f054ca0a19052e91efa7617 : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":42},{"id":"3885ec1a1d230adb28bdbc9b91189285f3d65808","author":{"_account_id":5733,"name":"Kieran Spear","email":"kispear@gmail.com","username":"kspear"},"date":"2026-08-10 06:45:11.000000000","message":"Patch Set 42:\n\n(1 comment)","accounts_in_message":[],"_revision_number":42},{"id":"4c7c4ba0db22c898f0250a05b4db68933f93f1e4","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-08-10 14:03:25.000000000","message":"Patch Set 42:\n\n(1 comment)","accounts_in_message":[],"_revision_number":42},{"id":"810d07e96ad7182c11902f3113b48a261fb50534","author":{"_account_id":5733,"name":"Kieran Spear","email":"kispear@gmail.com","username":"kspear"},"date":"2026-08-11 00:10:55.000000000","message":"Patch Set 42:\n\n(1 comment)","accounts_in_message":[],"_revision_number":42},{"id":"dbb97988fa570595f478d5a3efebe75e1872a3f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-03 11:32:43.000000000","message":"Uploaded patch set 43.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":43},{"id":"265189dd6db59852956c15a0d62602dfcbe850e0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 11:32:51.000000000","message":"Patch Set 43:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/trove-tempest-plugin for 989855,2","accounts_in_message":[],"_revision_number":43},{"id":"817300cdfb4b4596e76f416ada6e7cccfb186778","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 11:34:02.000000000","message":"Patch Set 43: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855 is needed.","accounts_in_message":[],"_revision_number":43},{"id":"2688eaf9377519fbd21aaf82c019459537c9f560","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-03 12:20:24.000000000","message":"Uploaded patch set 44: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":44},{"id":"02c3cfe87f293ab70608c2f397b456473940f119","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 12:53:20.000000000","message":"Patch Set 44:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/086fcd694e3b47a8926296707a1c84de\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/64cc7afbabd54377991d9db491972ba1 : SUCCESS in 9m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8673ea4c65dc4e61876b796a5ce81c2f : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/adf051b428f743b3843b90049d10cf54 : SUCCESS in 14m 33s (non-voting)","accounts_in_message":[],"_revision_number":44},{"id":"ba38ff5dd4d1a6a2dfe41aa7312e1f282b5f4f68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 15:02:11.000000000","message":"Patch Set 44: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0015fb79fb91439ab674793149b9128c\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/72f79d443d994d55b5b999793842ab71 : SUCCESS in 3m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13f4528b13c64c9a8fd7932d22457276 : SUCCESS in 8m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5acc21410299447187c17f9903595227 : SUCCESS in 4m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/52458dd3a4b24a3382167ac9a96626c9 : SUCCESS in 6m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ad7d94cf6f804241acaa036ae2ace4ea : SUCCESS in 3m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ff9b57b4d5924f0ca16aca7492af6752 : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8508095a7f144d62a8b28096a1ba566e : SUCCESS in 6m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/85fec541c4ee40e583a4c17fa97e0a82 : SUCCESS in 4m 45s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/67f44f0fd1af48d9a71e763e4a048d64 : SUCCESS in 44m 30s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/0cff008933ed4b70824d920e8d6353e0 : FAILURE in 38m 11s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/1110f9624bb14771a9ae78265bd5c011 : SUCCESS in 1h 09m 08s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/f7c36a45a0f2429f9258508f38984ba7 : SUCCESS in 1h 17m 52s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/948615106d2c42dc855373e6df308edb : SUCCESS in 40m 43s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/0873cc620e294632a2dd8d5a25d0b630 : FAILURE in 25m 40s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/190c8a9754264ba1b362b80c4fa503d7 : SUCCESS in 1h 10m 16s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/329e3103a5324f0fbe8858753f5c3820 : SUCCESS in 1h 27m 08s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/aa33ad921c2c476f9530f2c319ea2ab3 : SUCCESS in 44m 09s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/e1643182e87147cbb137369f4655ff3b : FAILURE in 53m 14s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/d72f2ac2c4be43d3b3a861d6e373fd07 : SUCCESS in 59m 22s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/ab8acf5791a74b9bac13268cae77b7c2 : SUCCESS in 1h 04m 26s (non-voting)\n- trove-tempest-ubuntu-base-keydb https://zuul.opendev.org/t/openstack/build/0414ec5c8a9546fe9f747d5b1f5260a1 : SUCCESS in 43m 41s\n- trove-tempest-ubuntu-backup-keydb https://zuul.opendev.org/t/openstack/build/e377099dd602491181d851b9e03d3334 : SUCCESS in 59m 02s (non-voting)\n- trove-tempest-ubuntu-ssl-keydb https://zuul.opendev.org/t/openstack/build/17f96e05409e49a282a5147f8250aa64 : SUCCESS in 49m 35s\n- trove-tempest-ubuntu-ssl-replication-keydb https://zuul.opendev.org/t/openstack/build/ae220c7a0db748b0ad4b2e63cb792434 : SUCCESS in 1h 01m 41s (non-voting)\n- trove-tempest-ubuntu-replication-keydb https://zuul.opendev.org/t/openstack/build/4c1c9ea3934345429c44e9022b9dd733 : SUCCESS in 55m 42s (non-voting)\n- trove-tempest-cinder-storage-driver-keydb https://zuul.opendev.org/t/openstack/build/54d5802b00954c038b143c63a1bcbf28 : SUCCESS in 1h 12m 53s (non-voting)\n- trove-tempest-ubuntu-base-valkey https://zuul.opendev.org/t/openstack/build/ea4fec49f22948b28603e633b6e39798 : SUCCESS in 42m 20s\n- trove-tempest-ubuntu-backup-valkey https://zuul.opendev.org/t/openstack/build/474d3218aefb47439daef508c8bb9035 : SUCCESS in 59m 16s (non-voting)\n- trove-tempest-ubuntu-ssl-valkey https://zuul.opendev.org/t/openstack/build/236f69a2e87142f2bd99fc09a44015e3 : SUCCESS in 54m 08s\n- trove-tempest-ubuntu-ssl-replication-valkey https://zuul.opendev.org/t/openstack/build/d644e85d57d7442d810062e1568faf6a : SUCCESS in 1h 11m 22s (non-voting)\n- trove-tempest-ubuntu-replication-valkey https://zuul.opendev.org/t/openstack/build/b60960bdcf5e4733874a2258882e9854 : SUCCESS in 58m 53s (non-voting)\n- trove-tempest-cinder-storage-driver-valkey https://zuul.opendev.org/t/openstack/build/ec14404f7ebf468e972b3e3871b66141 : SUCCESS in 1h 00m 32s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/80dfb745352f442d9411ff385b47d49e : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":44}],"current_revision_number":44,"current_revision":"e73cf9c77a3d7ce0abde887d997975a7e3c11547","revisions":{"edc223e1d1ce3774c7d69661050403048d5d3d54":{"kind":"REWORK","_number":1,"created":"2026-05-23 13:25:02.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/56/989856/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/989856/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/989856/1"}}},"commit":{"parents":[{"commit":"ed9162d86054f3ccc83c54dffe541eb3b3c5796f","subject":"Merge \"Drop Python 3.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ed9162d86054f3ccc83c54dffe541eb3b3c5796f"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-13 08:49:07.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-23 13:23:39.000000000","tz":300},"subject":"WIP Add KeyDB and Valkey datastore support","message":"WIP Add KeyDB and Valkey datastore support\n\nThis patch introduces support for KeyDB and Valkey datastores.\nTo achieve this without code duplication, the new Redis common\nimplementation was created from a scratch. KeyDB and Valkey now\ninherit from this base. Existing legacy experimental Redis is\nuntouched.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nChange-Id: Ia99918ff3df2078fed8636d46a9c4309f402eaba\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/edc223e1d1ce3774c7d69661050403048d5d3d54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/edc223e1d1ce3774c7d69661050403048d5d3d54"}]},"branch":"refs/heads/master"},"0f4dd8f8bd03002feae558dc1cf71cf6e2da36c0":{"kind":"REWORK","_number":2,"created":"2026-05-23 18:09:41.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/56/989856/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/989856/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/989856/2"}}},"commit":{"parents":[{"commit":"ed9162d86054f3ccc83c54dffe541eb3b3c5796f","subject":"Merge \"Drop Python 3.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ed9162d86054f3ccc83c54dffe541eb3b3c5796f"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-13 08:49:07.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-23 18:09:27.000000000","tz":300},"subject":"WIP Add KeyDB and Valkey datastore support","message":"WIP Add KeyDB and Valkey datastore support\n\nThis patch introduces support for KeyDB and Valkey datastores.\nTo achieve this without code duplication, the new Redis common\nimplementation was created from a scratch. KeyDB and Valkey now\ninherit from this base. Existing legacy experimental Redis is\nuntouched.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nChange-Id: Ia99918ff3df2078fed8636d46a9c4309f402eaba\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0f4dd8f8bd03002feae558dc1cf71cf6e2da36c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0f4dd8f8bd03002feae558dc1cf71cf6e2da36c0"}]},"branch":"refs/heads/master"},"0aea42100c9fbbedb1c45a3cd2f034aabf17c788":{"kind":"REWORK","_number":3,"created":"2026-05-23 18:11:03.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/56/989856/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/989856/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/989856/3"}}},"commit":{"parents":[{"commit":"ed9162d86054f3ccc83c54dffe541eb3b3c5796f","subject":"Merge \"Drop Python 3.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ed9162d86054f3ccc83c54dffe541eb3b3c5796f"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-13 08:49:07.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-23 18:10:53.000000000","tz":300},"subject":"WIP Add KeyDB and Valkey datastore support","message":"WIP Add KeyDB and Valkey datastore support\n\nThis patch introduces support for KeyDB and Valkey datastores.\nTo achieve this without code duplication, the new Redis common\nimplementation was created from a scratch. KeyDB and Valkey now\ninherit from this base. Existing legacy experimental Redis is\nuntouched.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nChange-Id: Ia99918ff3df2078fed8636d46a9c4309f402eaba\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0aea42100c9fbbedb1c45a3cd2f034aabf17c788"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0aea42100c9fbbedb1c45a3cd2f034aabf17c788"}]},"branch":"refs/heads/master"},"649e77afabf702024ff431e5b2e3db05bc10ebc5":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-05-24 10:17:21.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/56/989856/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/989856/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/989856/4"}}},"commit":{"parents":[{"commit":"ed9162d86054f3ccc83c54dffe541eb3b3c5796f","subject":"Merge \"Drop Python 3.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ed9162d86054f3ccc83c54dffe541eb3b3c5796f"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-13 08:49:07.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-24 10:16:52.000000000","tz":300},"subject":"WIP Add KeyDB and Valkey datastore support","message":"WIP Add KeyDB and Valkey datastore support\n\nThis patch introduces support for KeyDB and Valkey datastores.\nTo achieve this without code duplication, the new Redis common\nimplementation was created from a scratch. KeyDB and Valkey now\ninherit from this base. Existing legacy experimental Redis is\nuntouched.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nDepends-On: https://review.opendev.org/c/openstack/trove/+/989888\nChange-Id: Ia99918ff3df2078fed8636d46a9c4309f402eaba\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/649e77afabf702024ff431e5b2e3db05bc10ebc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/649e77afabf702024ff431e5b2e3db05bc10ebc5"}]},"branch":"refs/heads/master"},"d2e0138e7c28835fbee4a330fccfbc95a0424041":{"kind":"REWORK","_number":5,"created":"2026-05-24 11:05:23.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/56/989856/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/989856/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/989856/5"}}},"commit":{"parents":[{"commit":"ed9162d86054f3ccc83c54dffe541eb3b3c5796f","subject":"Merge \"Drop Python 3.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ed9162d86054f3ccc83c54dffe541eb3b3c5796f"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-13 08:49:07.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-24 11:05:11.000000000","tz":300},"subject":"WIP Add KeyDB and Valkey datastore support","message":"WIP Add KeyDB and Valkey datastore support\n\nThis patch introduces support for KeyDB and Valkey datastores.\nTo achieve this without code duplication, the new Redis common\nimplementation was created from a scratch. KeyDB and Valkey now\ninherit from this base. Existing legacy experimental Redis is\nuntouched.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nDepends-On: https://review.opendev.org/c/openstack/trove/+/989888\nChange-Id: Ia99918ff3df2078fed8636d46a9c4309f402eaba\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d2e0138e7c28835fbee4a330fccfbc95a0424041"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d2e0138e7c28835fbee4a330fccfbc95a0424041"}]},"branch":"refs/heads/master"},"2cd522e7f3ee6946390296db5103066c8f1e14f4":{"kind":"REWORK","_number":6,"created":"2026-05-24 13:46:46.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/56/989856/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/989856/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/989856/6"}}},"commit":{"parents":[{"commit":"ed9162d86054f3ccc83c54dffe541eb3b3c5796f","subject":"Merge \"Drop Python 3.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ed9162d86054f3ccc83c54dffe541eb3b3c5796f"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-13 08:49:07.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-24 13:46:31.000000000","tz":300},"subject":"WIP Add KeyDB and Valkey datastore support","message":"WIP Add KeyDB and Valkey datastore support\n\nThis patch introduces support for KeyDB and Valkey datastores.\nTo achieve this without code duplication, the new Redis common\nimplementation was created from a scratch. KeyDB and Valkey now\ninherit from this base. Existing legacy experimental Redis is\nuntouched.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nDepends-On: https://review.opendev.org/c/openstack/trove/+/989888\nChange-Id: Ia99918ff3df2078fed8636d46a9c4309f402eaba\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2cd522e7f3ee6946390296db5103066c8f1e14f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2cd522e7f3ee6946390296db5103066c8f1e14f4"}]},"branch":"refs/heads/master"},"b8cd2c1da1efc2605af50dd7a39f1d243e09cbb4":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-05-24 13:47:48.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/56/989856/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/989856/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/989856/7"}}},"commit":{"parents":[{"commit":"0994e6b1511359dedd20a4d1bf01608527f4c65a","subject":"Mark hashlib.md5 for non security purpose","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0994e6b1511359dedd20a4d1bf01608527f4c65a"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-13 08:49:07.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-24 13:47:37.000000000","tz":300},"subject":"WIP Add KeyDB and Valkey datastore support","message":"WIP Add KeyDB and Valkey datastore support\n\nThis patch introduces support for KeyDB and Valkey datastores.\nTo achieve this without code duplication, the new Redis common\nimplementation was created from a scratch. KeyDB and Valkey now\ninherit from this base. Existing legacy experimental Redis is\nuntouched.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nDepends-On: https://review.opendev.org/c/openstack/trove/+/989888\nChange-Id: Ia99918ff3df2078fed8636d46a9c4309f402eaba\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b8cd2c1da1efc2605af50dd7a39f1d243e09cbb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b8cd2c1da1efc2605af50dd7a39f1d243e09cbb4"}]},"branch":"refs/heads/master"},"382c1ffc93450762edf2c0c66246b6eeade43463":{"kind":"REWORK","_number":8,"created":"2026-05-24 17:59:34.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/56/989856/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/989856/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/989856/8"}}},"commit":{"parents":[{"commit":"0994e6b1511359dedd20a4d1bf01608527f4c65a","subject":"Mark hashlib.md5 for non security purpose","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0994e6b1511359dedd20a4d1bf01608527f4c65a"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-13 08:49:07.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-24 17:59:12.000000000","tz":300},"subject":"WIP Add KeyDB and Valkey datastore support","message":"WIP Add KeyDB and Valkey datastore support\n\nThis patch introduces support for KeyDB and Valkey datastores.\nTo achieve this without code duplication, the new Redis common\nimplementation was created from a scratch. KeyDB and Valkey now\ninherit from this base. Existing legacy experimental Redis is\nuntouched.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nDepends-On: https://review.opendev.org/c/openstack/trove/+/989888\nChange-Id: Ia99918ff3df2078fed8636d46a9c4309f402eaba\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/382c1ffc93450762edf2c0c66246b6eeade43463"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/382c1ffc93450762edf2c0c66246b6eeade43463"}]},"branch":"refs/heads/master"},"26f182f09dd22401bd04886da8de789278198eab":{"kind":"REWORK","_number":9,"created":"2026-05-25 07:54:34.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/56/989856/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/989856/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/989856/9"}}},"commit":{"parents":[{"commit":"25dde62d8f0bbc7429054966f62171d565c9e984","subject":"Actualize DevStack installation + docs update","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/25dde62d8f0bbc7429054966f62171d565c9e984"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-13 08:49:07.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-25 07:54:11.000000000","tz":300},"subject":"WIP Add KeyDB and Valkey datastore support","message":"WIP Add KeyDB and Valkey datastore support\n\nThis patch introduces support for KeyDB and Valkey datastores.\nTo achieve this without code duplication, the new Redis common\nimplementation was created from a scratch. KeyDB and Valkey now\ninherit from this base. Existing legacy experimental Redis is\nuntouched.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nDepends-On: https://review.opendev.org/c/openstack/trove/+/989888\nChange-Id: Ia99918ff3df2078fed8636d46a9c4309f402eaba\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/26f182f09dd22401bd04886da8de789278198eab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/26f182f09dd22401bd04886da8de789278198eab"}]},"branch":"refs/heads/master"},"a2cf0f7133235faa5007390b3553c37dda97f7af":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2026-05-25 11:45:30.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/56/989856/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/989856/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/989856/10"}}},"commit":{"parents":[{"commit":"25dde62d8f0bbc7429054966f62171d565c9e984","subject":"Actualize DevStack installation + docs update","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/25dde62d8f0bbc7429054966f62171d565c9e984"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-13 08:49:07.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-25 11:45:18.000000000","tz":300},"subject":"WIP Add KeyDB and Valkey datastore support","message":"WIP Add KeyDB and Valkey datastore support\n\nThis patch introduces support for KeyDB and Valkey datastores.\nTo achieve this without code duplication, the new Redis common\nimplementation was created from a scratch. KeyDB and Valkey now\ninherit from this base. Existing legacy experimental Redis is\nuntouched.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nChange-Id: Ia99918ff3df2078fed8636d46a9c4309f402eaba\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a2cf0f7133235faa5007390b3553c37dda97f7af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a2cf0f7133235faa5007390b3553c37dda97f7af"}]},"branch":"refs/heads/master"},"2d9c7d2ee10ec0326c49eca9754a72e3cbd53507":{"kind":"REWORK","_number":11,"created":"2026-05-29 13:13:25.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/56/989856/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/989856/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/989856/11"}}},"commit":{"parents":[{"commit":"e816cd850ca3baad6bf3f675d40224a6eeb6dc42","subject":"Preparation for removing eventlet","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e816cd850ca3baad6bf3f675d40224a6eeb6dc42"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-13 08:49:07.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-29 13:12:43.000000000","tz":300},"subject":"WIP Add KeyDB and Valkey datastore support","message":"WIP Add KeyDB and Valkey datastore support\n\nThis patch introduces support for KeyDB and Valkey datastores.\nTo achieve this without code duplication, the new Redis common\nimplementation was created from a scratch. KeyDB and Valkey now\ninherit from this base. Existing legacy experimental Redis is\nuntouched.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nChange-Id: Ia99918ff3df2078fed8636d46a9c4309f402eaba\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2d9c7d2ee10ec0326c49eca9754a72e3cbd53507"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2d9c7d2ee10ec0326c49eca9754a72e3cbd53507"}]},"branch":"refs/heads/master"},"0b9449f3e00c788af8bd32afe12f8804212392dc":{"kind":"REWORK","_number":12,"created":"2026-05-29 15:29:41.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/56/989856/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/989856/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/989856/12"}}},"commit":{"parents":[{"commit":"e816cd850ca3baad6bf3f675d40224a6eeb6dc42","subject":"Preparation for removing eventlet","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e816cd850ca3baad6bf3f675d40224a6eeb6dc42"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-13 08:49:07.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-29 15:24:41.000000000","tz":300},"subject":"WIP Add KeyDB and Valkey datastore support","message":"WIP Add KeyDB and Valkey datastore support\n\nThis patch introduces support for KeyDB and Valkey datastores.\nTo achieve this without code duplication, the new Redis common\nimplementation was created from a scratch. KeyDB and Valkey now\ninherit from this base. Existing legacy experimental Redis is\nuntouched.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nChange-Id: Ia99918ff3df2078fed8636d46a9c4309f402eaba\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0b9449f3e00c788af8bd32afe12f8804212392dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0b9449f3e00c788af8bd32afe12f8804212392dc"}]},"branch":"refs/heads/master"},"95903280ce94cab459c031c9ce41c09c2529b8cc":{"kind":"REWORK","_number":13,"created":"2026-05-29 19:48:47.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/56/989856/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/989856/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/989856/13"}}},"commit":{"parents":[{"commit":"e816cd850ca3baad6bf3f675d40224a6eeb6dc42","subject":"Preparation for removing eventlet","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e816cd850ca3baad6bf3f675d40224a6eeb6dc42"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-13 08:49:07.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-29 19:48:20.000000000","tz":300},"subject":"WIP Add KeyDB and Valkey datastore support","message":"WIP Add KeyDB and Valkey datastore support\n\nThis patch introduces support for KeyDB and Valkey datastores.\nTo achieve this without code duplication, the new Redis common\nimplementation was created from a scratch. KeyDB and Valkey now\ninherit from this base. Existing legacy experimental Redis is\nuntouched.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nDepends-On: https://review.opendev.org/c/openstack/trove/+/990725\nChange-Id: Ia99918ff3df2078fed8636d46a9c4309f402eaba\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/95903280ce94cab459c031c9ce41c09c2529b8cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/95903280ce94cab459c031c9ce41c09c2529b8cc"}]},"branch":"refs/heads/master"},"d31864084567077f5614839676e44c3c2a13ffd0":{"kind":"NO_CODE_CHANGE","_number":14,"created":"2026-06-02 09:11:43.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/56/989856/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/989856/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/989856/14"}}},"commit":{"parents":[{"commit":"e816cd850ca3baad6bf3f675d40224a6eeb6dc42","subject":"Preparation for removing eventlet","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e816cd850ca3baad6bf3f675d40224a6eeb6dc42"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-13 08:49:07.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-02 09:11:25.000000000","tz":300},"subject":"WIP Add KeyDB and Valkey datastore support","message":"WIP Add KeyDB and Valkey datastore support\n\nThis patch introduces support for KeyDB and Valkey datastores.\nTo achieve this without code duplication, the new Redis common\nimplementation was created from a scratch. KeyDB and Valkey now\ninherit from this base. Existing legacy experimental Redis is\nuntouched.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nChange-Id: Ia99918ff3df2078fed8636d46a9c4309f402eaba\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d31864084567077f5614839676e44c3c2a13ffd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d31864084567077f5614839676e44c3c2a13ffd0"}]},"branch":"refs/heads/master"},"8d121920752bc33e4f77d69a3626961cc0e38243":{"kind":"REWORK","_number":15,"created":"2026-06-04 15:15:07.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/56/989856/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/989856/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/989856/15"}}},"commit":{"parents":[{"commit":"034b1eecff21d51490ac6c902b189bd64490972c","subject":"Fix trove-tempest-plugin setup bug in devstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/034b1eecff21d51490ac6c902b189bd64490972c"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-13 08:49:07.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-04 15:14:43.000000000","tz":300},"subject":"WIP Add KeyDB and Valkey datastore support","message":"WIP Add KeyDB and Valkey datastore support\n\nThis patch introduces support for KeyDB and Valkey datastores.\nTo achieve this without code duplication, the new Redis common\nimplementation was created from a scratch. KeyDB and Valkey now\ninherit from this base. Existing legacy experimental Redis is\nuntouched.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nDepends-On: https://review.opendev.org/c/openstack/trove/+/990725\nChange-Id: Ia99918ff3df2078fed8636d46a9c4309f402eaba\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8d121920752bc33e4f77d69a3626961cc0e38243"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8d121920752bc33e4f77d69a3626961cc0e38243"}]},"branch":"refs/heads/master"},"fc50426900768854186e4b3bf46182ab4de9c9d8":{"kind":"REWORK","_number":16,"created":"2026-06-04 15:30:14.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/56/989856/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/989856/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/989856/16"}}},"commit":{"parents":[{"commit":"f96ba259257149197456324c3186a12c15c18788","subject":"Optimizing Zuul Jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f96ba259257149197456324c3186a12c15c18788"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-13 08:49:07.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-04 15:29:51.000000000","tz":300},"subject":"WIP Add KeyDB and Valkey datastore support","message":"WIP Add KeyDB and Valkey datastore support\n\nThis patch introduces support for KeyDB and Valkey datastores.\nTo achieve this without code duplication, the new Redis common\nimplementation was created from a scratch. KeyDB and Valkey now\ninherit from this base. Existing legacy experimental Redis is\nuntouched.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nDepends-On: https://review.opendev.org/c/openstack/trove/+/990725\nChange-Id: Ia99918ff3df2078fed8636d46a9c4309f402eaba\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fc50426900768854186e4b3bf46182ab4de9c9d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fc50426900768854186e4b3bf46182ab4de9c9d8"}]},"branch":"refs/heads/master"},"33de3811837b9999b842118830b6d48ac66c75ad":{"kind":"REWORK","_number":17,"created":"2026-06-04 15:56:34.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/56/989856/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/989856/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/989856/17"}}},"commit":{"parents":[{"commit":"f96ba259257149197456324c3186a12c15c18788","subject":"Optimizing Zuul Jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f96ba259257149197456324c3186a12c15c18788"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-13 08:49:07.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-04 15:56:04.000000000","tz":300},"subject":"WIP Add KeyDB and Valkey datastore support","message":"WIP Add KeyDB and Valkey datastore support\n\nThis patch introduces support for KeyDB and Valkey datastores.\nTo achieve this without code duplication, the new Redis common\nimplementation was created from a scratch. KeyDB and Valkey now\ninherit from this base. Existing legacy experimental Redis is\nuntouched.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nDepends-On: https://review.opendev.org/c/openstack/trove/+/990725\nChange-Id: Ia99918ff3df2078fed8636d46a9c4309f402eaba\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/33de3811837b9999b842118830b6d48ac66c75ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/33de3811837b9999b842118830b6d48ac66c75ad"}]},"branch":"refs/heads/master"},"2e1f1dbd6fd11f244c109dfcccea1c75e53fd776":{"kind":"REWORK","_number":18,"created":"2026-06-04 17:04:03.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/56/989856/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/989856/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/989856/18"}}},"commit":{"parents":[{"commit":"f96ba259257149197456324c3186a12c15c18788","subject":"Optimizing Zuul Jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f96ba259257149197456324c3186a12c15c18788"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-13 08:49:07.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-04 16:54:54.000000000","tz":300},"subject":"WIP Add KeyDB and Valkey datastore support","message":"WIP Add KeyDB and Valkey datastore support\n\nThis patch introduces support for KeyDB and Valkey datastores.\nTo achieve this without code duplication, the new Redis common\nimplementation was created from a scratch. KeyDB and Valkey now\ninherit from this base. Existing legacy experimental Redis is\nuntouched.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nDepends-On: https://review.opendev.org/c/openstack/trove/+/990725\nChange-Id: Ia99918ff3df2078fed8636d46a9c4309f402eaba\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2e1f1dbd6fd11f244c109dfcccea1c75e53fd776"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2e1f1dbd6fd11f244c109dfcccea1c75e53fd776"}]},"branch":"refs/heads/master"},"81a9e5d05bc741cdc268dcbb747d6f5063478f13":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2026-06-04 17:45:34.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/56/989856/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/989856/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/989856/19"}}},"commit":{"parents":[{"commit":"f7bc1eabb019a326472e1e30a39c0684db0642fd","subject":"Optimizing Zuul Jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f7bc1eabb019a326472e1e30a39c0684db0642fd"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-13 08:49:07.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-04 17:45:15.000000000","tz":300},"subject":"WIP Add KeyDB and Valkey datastore support","message":"WIP Add KeyDB and Valkey datastore support\n\nThis patch introduces support for KeyDB and Valkey datastores.\nTo achieve this without code duplication, the new Redis common\nimplementation was created from a scratch. KeyDB and Valkey now\ninherit from this base. Existing legacy experimental Redis is\nuntouched.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nDepends-On: https://review.opendev.org/c/openstack/trove/+/990725\nChange-Id: Ia99918ff3df2078fed8636d46a9c4309f402eaba\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/81a9e5d05bc741cdc268dcbb747d6f5063478f13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/81a9e5d05bc741cdc268dcbb747d6f5063478f13"}]},"branch":"refs/heads/master"},"a1c77b5c4a9548ca862d2de345f62986c57b1dae":{"kind":"REWORK","_number":20,"created":"2026-06-05 06:30:35.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/56/989856/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/989856/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/989856/20"}}},"commit":{"parents":[{"commit":"f7bc1eabb019a326472e1e30a39c0684db0642fd","subject":"Optimizing Zuul Jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f7bc1eabb019a326472e1e30a39c0684db0642fd"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-13 08:49:07.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-05 06:30:11.000000000","tz":300},"subject":"WIP Add KeyDB and Valkey datastore support","message":"WIP Add KeyDB and Valkey datastore support\n\nThis patch introduces support for KeyDB and Valkey datastores.\nTo achieve this without code duplication, the new Redis common\nimplementation was created from a scratch. KeyDB and Valkey now\ninherit from this base. Existing legacy experimental Redis is\nuntouched.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nDepends-On: https://review.opendev.org/c/openstack/trove/+/990725\nChange-Id: Ia99918ff3df2078fed8636d46a9c4309f402eaba\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a1c77b5c4a9548ca862d2de345f62986c57b1dae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a1c77b5c4a9548ca862d2de345f62986c57b1dae"}]},"branch":"refs/heads/master"},"a8ee1b9b415536d9c3a75676bf1fd8837d53239b":{"kind":"REWORK","_number":21,"created":"2026-06-05 07:06:29.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/56/989856/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/989856/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/989856/21"}}},"commit":{"parents":[{"commit":"f7bc1eabb019a326472e1e30a39c0684db0642fd","subject":"Optimizing Zuul Jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f7bc1eabb019a326472e1e30a39c0684db0642fd"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-13 08:49:07.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-05 07:05:50.000000000","tz":300},"subject":"WIP Add KeyDB and Valkey datastore support","message":"WIP Add KeyDB and Valkey datastore support\n\nThis patch introduces support for KeyDB and Valkey datastores.\nTo achieve this without code duplication, the new Redis common\nimplementation was created from a scratch. KeyDB and Valkey now\ninherit from this base. Existing legacy experimental Redis is\nuntouched.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nDepends-On: https://review.opendev.org/c/openstack/trove/+/990725\nChange-Id: Ia99918ff3df2078fed8636d46a9c4309f402eaba\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a8ee1b9b415536d9c3a75676bf1fd8837d53239b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a8ee1b9b415536d9c3a75676bf1fd8837d53239b"}]},"branch":"refs/heads/master"},"025e209e63a90ebe47319f12e0b9f5130413adf9":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2026-06-05 15:14:57.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/56/989856/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/989856/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/989856/22"}}},"commit":{"parents":[{"commit":"3b4cdb94d6a4e1fd304928dcc8fdb5b188653016","subject":"Optimizing Zuul Jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3b4cdb94d6a4e1fd304928dcc8fdb5b188653016"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-13 08:49:07.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-05 15:14:31.000000000","tz":300},"subject":"WIP Add KeyDB and Valkey datastore support","message":"WIP Add KeyDB and Valkey datastore support\n\nThis patch introduces support for KeyDB and Valkey datastores.\nTo achieve this without code duplication, the new Redis common\nimplementation was created from a scratch. KeyDB and Valkey now\ninherit from this base. Existing legacy experimental Redis is\nuntouched.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nDepends-On: https://review.opendev.org/c/openstack/trove/+/990725\nChange-Id: Ia99918ff3df2078fed8636d46a9c4309f402eaba\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/025e209e63a90ebe47319f12e0b9f5130413adf9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/025e209e63a90ebe47319f12e0b9f5130413adf9"}]},"branch":"refs/heads/master"},"45bbeb0fbca4f1032aa872b9f365acab85f0b7e2":{"kind":"REWORK","_number":23,"created":"2026-06-05 17:40:57.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/56/989856/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/989856/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/989856/23"}}},"commit":{"parents":[{"commit":"3b4cdb94d6a4e1fd304928dcc8fdb5b188653016","subject":"Optimizing Zuul Jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3b4cdb94d6a4e1fd304928dcc8fdb5b188653016"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-13 08:49:07.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-05 17:40:03.000000000","tz":300},"subject":"WIP Add KeyDB and Valkey datastore support","message":"WIP Add KeyDB and Valkey datastore support\n\nThis patch introduces support for KeyDB and Valkey datastores.\nTo achieve this without code duplication, the new Redis common\nimplementation was created from a scratch. KeyDB and Valkey now\ninherit from this base. Existing legacy experimental Redis is\nuntouched.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nDepends-On: https://review.opendev.org/c/openstack/trove/+/990725\nChange-Id: Ia99918ff3df2078fed8636d46a9c4309f402eaba\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/45bbeb0fbca4f1032aa872b9f365acab85f0b7e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/45bbeb0fbca4f1032aa872b9f365acab85f0b7e2"}]},"branch":"refs/heads/master"},"21a1f0cc259432bac19fd37443fb660c163e228c":{"kind":"REWORK","_number":24,"created":"2026-06-05 19:42:17.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/56/989856/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/989856/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/989856/24"}}},"commit":{"parents":[{"commit":"3b4cdb94d6a4e1fd304928dcc8fdb5b188653016","subject":"Optimizing Zuul Jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3b4cdb94d6a4e1fd304928dcc8fdb5b188653016"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-13 08:49:07.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-05 19:40:17.000000000","tz":300},"subject":"WIP Add KeyDB and Valkey datastore support","message":"WIP Add KeyDB and Valkey datastore support\n\nThis patch introduces support for KeyDB and Valkey datastores.\nTo achieve this without code duplication, the new Redis common\nimplementation was created from a scratch. KeyDB and Valkey now\ninherit from this base. Existing legacy experimental Redis is\nuntouched.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nDepends-On: https://review.opendev.org/c/openstack/trove/+/990725\nChange-Id: Ia99918ff3df2078fed8636d46a9c4309f402eaba\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/21a1f0cc259432bac19fd37443fb660c163e228c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/21a1f0cc259432bac19fd37443fb660c163e228c"}]},"branch":"refs/heads/master"},"61f1ee27c3addae9cb31de71b8d4592f6fa78b4b":{"kind":"REWORK","_number":25,"created":"2026-06-06 06:27:52.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/56/989856/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/989856/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/989856/25"}}},"commit":{"parents":[{"commit":"3b4cdb94d6a4e1fd304928dcc8fdb5b188653016","subject":"Optimizing Zuul Jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3b4cdb94d6a4e1fd304928dcc8fdb5b188653016"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-13 08:49:07.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-06 06:27:38.000000000","tz":300},"subject":"WIP Add KeyDB and Valkey datastore support","message":"WIP Add KeyDB and Valkey datastore support\n\nThis patch introduces support for KeyDB and Valkey datastores.\nTo achieve this without code duplication, the new Redis common\nimplementation was created from a scratch. KeyDB and Valkey now\ninherit from this base. Existing legacy experimental Redis is\nuntouched.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nDepends-On: https://review.opendev.org/c/openstack/trove/+/990725\nChange-Id: Ia99918ff3df2078fed8636d46a9c4309f402eaba\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/61f1ee27c3addae9cb31de71b8d4592f6fa78b4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/61f1ee27c3addae9cb31de71b8d4592f6fa78b4b"}]},"branch":"refs/heads/master"},"e84a47518b132528f29508b074a8e0d0c3a34689":{"kind":"REWORK","_number":26,"created":"2026-06-06 15:14:09.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/56/989856/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/989856/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/989856/26"}}},"commit":{"parents":[{"commit":"3b4cdb94d6a4e1fd304928dcc8fdb5b188653016","subject":"Optimizing Zuul Jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3b4cdb94d6a4e1fd304928dcc8fdb5b188653016"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-13 08:49:07.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-06 14:56:51.000000000","tz":300},"subject":"WIP Add KeyDB and Valkey datastore support","message":"WIP Add KeyDB and Valkey datastore support\n\nThis patch introduces support for KeyDB and Valkey datastores.\nTo achieve this without code duplication, the new Redis common\nimplementation was created from a scratch. KeyDB and Valkey now\ninherit from this base. Existing legacy experimental Redis is\nuntouched.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nDepends-On: https://review.opendev.org/c/openstack/trove/+/990725\nChange-Id: Ia99918ff3df2078fed8636d46a9c4309f402eaba\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e84a47518b132528f29508b074a8e0d0c3a34689"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e84a47518b132528f29508b074a8e0d0c3a34689"}]},"branch":"refs/heads/master"},"69b6ae4d034d06c62e23e5510a8e7d6a4125e9b1":{"kind":"REWORK","_number":27,"created":"2026-06-06 16:31:05.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/56/989856/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/989856/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/989856/27"}}},"commit":{"parents":[{"commit":"3b4cdb94d6a4e1fd304928dcc8fdb5b188653016","subject":"Optimizing Zuul Jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3b4cdb94d6a4e1fd304928dcc8fdb5b188653016"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-13 08:49:07.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-06 16:30:52.000000000","tz":300},"subject":"WIP Add KeyDB and Valkey datastore support","message":"WIP Add KeyDB and Valkey datastore support\n\nThis patch introduces support for KeyDB and Valkey datastores.\nTo achieve this without code duplication, the new Redis common\nimplementation was created from a scratch. KeyDB and Valkey now\ninherit from this base. Existing legacy experimental Redis is\nuntouched.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nDepends-On: https://review.opendev.org/c/openstack/trove/+/990725\nChange-Id: Ia99918ff3df2078fed8636d46a9c4309f402eaba\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/69b6ae4d034d06c62e23e5510a8e7d6a4125e9b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/69b6ae4d034d06c62e23e5510a8e7d6a4125e9b1"}]},"branch":"refs/heads/master"},"49fba687bda38d1fcf807d58b86c04bfe1858867":{"kind":"REWORK","_number":28,"created":"2026-06-06 17:22:06.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/56/989856/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/989856/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/989856/28"}}},"commit":{"parents":[{"commit":"3b4cdb94d6a4e1fd304928dcc8fdb5b188653016","subject":"Optimizing Zuul Jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3b4cdb94d6a4e1fd304928dcc8fdb5b188653016"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-13 08:49:07.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-06 17:21:47.000000000","tz":300},"subject":"WIP Add KeyDB and Valkey datastore support","message":"WIP Add KeyDB and Valkey datastore support\n\nThis patch introduces support for KeyDB and Valkey datastores.\nTo achieve this without code duplication, the new Redis common\nimplementation was created from a scratch. KeyDB and Valkey now\ninherit from this base. Existing legacy experimental Redis is\nuntouched.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nDepends-On: https://review.opendev.org/c/openstack/trove/+/990725\nChange-Id: Ia99918ff3df2078fed8636d46a9c4309f402eaba\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/49fba687bda38d1fcf807d58b86c04bfe1858867"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/49fba687bda38d1fcf807d58b86c04bfe1858867"}]},"branch":"refs/heads/master"},"103bb076fd0b87fed8220b66d65380b2bffd873a":{"kind":"REWORK","_number":29,"created":"2026-06-06 17:30:30.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/56/989856/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/989856/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/989856/29"}}},"commit":{"parents":[{"commit":"3b4cdb94d6a4e1fd304928dcc8fdb5b188653016","subject":"Optimizing Zuul Jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3b4cdb94d6a4e1fd304928dcc8fdb5b188653016"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-13 08:49:07.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-06 17:30:18.000000000","tz":300},"subject":"WIP Add KeyDB and Valkey datastore support","message":"WIP Add KeyDB and Valkey datastore support\n\nThis patch introduces support for KeyDB and Valkey datastores.\nTo achieve this without code duplication, the new Redis common\nimplementation was created from a scratch. KeyDB and Valkey now\ninherit from this base. Existing legacy experimental Redis is\nuntouched.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nDepends-On: https://review.opendev.org/c/openstack/trove/+/990725\nChange-Id: Ia99918ff3df2078fed8636d46a9c4309f402eaba\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/103bb076fd0b87fed8220b66d65380b2bffd873a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/103bb076fd0b87fed8220b66d65380b2bffd873a"}]},"branch":"refs/heads/master"},"b5cd304b4611372d50e94863534c9358387aaeda":{"kind":"REWORK","_number":30,"created":"2026-06-06 18:47:33.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/56/989856/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/989856/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/989856/30"}}},"commit":{"parents":[{"commit":"3b4cdb94d6a4e1fd304928dcc8fdb5b188653016","subject":"Optimizing Zuul Jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3b4cdb94d6a4e1fd304928dcc8fdb5b188653016"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-13 08:49:07.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-06 18:47:11.000000000","tz":300},"subject":"WIP Add KeyDB and Valkey datastore support","message":"WIP Add KeyDB and Valkey datastore support\n\nThis patch introduces support for KeyDB and Valkey datastores.\nTo achieve this without code duplication, the new Redis common\nimplementation was created from a scratch. KeyDB and Valkey now\ninherit from this base. Existing legacy experimental Redis is\nuntouched.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nDepends-On: https://review.opendev.org/c/openstack/trove/+/990725\nChange-Id: Ia99918ff3df2078fed8636d46a9c4309f402eaba\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b5cd304b4611372d50e94863534c9358387aaeda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b5cd304b4611372d50e94863534c9358387aaeda"}]},"branch":"refs/heads/master"},"a4967e5f735864487f8c990c8cf0b47b53fbb1cd":{"kind":"REWORK","_number":31,"created":"2026-06-06 18:54:18.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/56/989856/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/989856/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/989856/31"}}},"commit":{"parents":[{"commit":"3b4cdb94d6a4e1fd304928dcc8fdb5b188653016","subject":"Optimizing Zuul Jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3b4cdb94d6a4e1fd304928dcc8fdb5b188653016"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-13 08:49:07.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-06 18:54:03.000000000","tz":300},"subject":"WIP Add KeyDB and Valkey datastore support","message":"WIP Add KeyDB and Valkey datastore support\n\nThis patch introduces support for KeyDB and Valkey datastores.\nTo achieve this without code duplication, the new Redis common\nimplementation was created from a scratch. KeyDB and Valkey now\ninherit from this base. Existing legacy experimental Redis is\nuntouched.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nDepends-On: https://review.opendev.org/c/openstack/trove/+/990725\nChange-Id: Ia99918ff3df2078fed8636d46a9c4309f402eaba\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a4967e5f735864487f8c990c8cf0b47b53fbb1cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a4967e5f735864487f8c990c8cf0b47b53fbb1cd"}]},"branch":"refs/heads/master"},"c8dfc0a52efd2eb0b62ac29ea9f775d6b5a6a272":{"kind":"REWORK","_number":32,"created":"2026-06-06 19:09:50.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/56/989856/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/989856/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/989856/32"}}},"commit":{"parents":[{"commit":"3b4cdb94d6a4e1fd304928dcc8fdb5b188653016","subject":"Optimizing Zuul Jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3b4cdb94d6a4e1fd304928dcc8fdb5b188653016"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-13 08:49:07.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-06 19:09:35.000000000","tz":300},"subject":"WIP Add KeyDB and Valkey datastore support","message":"WIP Add KeyDB and Valkey datastore support\n\nThis patch introduces support for KeyDB and Valkey datastores.\nTo achieve this without code duplication, the new Redis common\nimplementation was created from a scratch. KeyDB and Valkey now\ninherit from this base. Existing legacy experimental Redis is\nuntouched.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nDepends-On: https://review.opendev.org/c/openstack/trove/+/990725\nChange-Id: Ia99918ff3df2078fed8636d46a9c4309f402eaba\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c8dfc0a52efd2eb0b62ac29ea9f775d6b5a6a272"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c8dfc0a52efd2eb0b62ac29ea9f775d6b5a6a272"}]},"branch":"refs/heads/master"},"f4d3a68195703369687bd4d9c1e801661299bf23":{"kind":"REWORK","_number":33,"created":"2026-06-06 19:22:23.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/56/989856/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/989856/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/989856/33"}}},"commit":{"parents":[{"commit":"3b4cdb94d6a4e1fd304928dcc8fdb5b188653016","subject":"Optimizing Zuul Jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3b4cdb94d6a4e1fd304928dcc8fdb5b188653016"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-13 08:49:07.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-06 19:22:11.000000000","tz":300},"subject":"WIP Add KeyDB and Valkey datastore support","message":"WIP Add KeyDB and Valkey datastore support\n\nThis patch introduces support for KeyDB and Valkey datastores.\nTo achieve this without code duplication, the new Redis common\nimplementation was created from a scratch. KeyDB and Valkey now\ninherit from this base. Existing legacy experimental Redis is\nuntouched.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nDepends-On: https://review.opendev.org/c/openstack/trove/+/990725\nChange-Id: Ia99918ff3df2078fed8636d46a9c4309f402eaba\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f4d3a68195703369687bd4d9c1e801661299bf23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f4d3a68195703369687bd4d9c1e801661299bf23"}]},"branch":"refs/heads/master"},"9764f3a2b3b8954770b9c671348a1bf4b9ef5421":{"kind":"REWORK","_number":34,"created":"2026-06-06 19:34:19.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/56/989856/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/989856/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/989856/34"}}},"commit":{"parents":[{"commit":"3b4cdb94d6a4e1fd304928dcc8fdb5b188653016","subject":"Optimizing Zuul Jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3b4cdb94d6a4e1fd304928dcc8fdb5b188653016"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-13 08:49:07.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-06 19:34:03.000000000","tz":300},"subject":"WIP Add KeyDB and Valkey datastore support","message":"WIP Add KeyDB and Valkey datastore support\n\nThis patch introduces support for KeyDB and Valkey datastores.\nTo achieve this without code duplication, the new Redis common\nimplementation was created from a scratch. KeyDB and Valkey now\ninherit from this base. Existing legacy experimental Redis is\nuntouched.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nDepends-On: https://review.opendev.org/c/openstack/trove/+/990725\nChange-Id: Ia99918ff3df2078fed8636d46a9c4309f402eaba\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9764f3a2b3b8954770b9c671348a1bf4b9ef5421"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9764f3a2b3b8954770b9c671348a1bf4b9ef5421"}]},"branch":"refs/heads/master"},"d69e809f219dc60883eac663322edc60890ea7b8":{"kind":"REWORK","_number":35,"created":"2026-06-08 06:26:10.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/56/989856/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/989856/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/989856/35"}}},"commit":{"parents":[{"commit":"3b4cdb94d6a4e1fd304928dcc8fdb5b188653016","subject":"Optimizing Zuul Jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3b4cdb94d6a4e1fd304928dcc8fdb5b188653016"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-13 08:49:07.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-08 06:25:54.000000000","tz":300},"subject":"WIP Add KeyDB and Valkey datastore support","message":"WIP Add KeyDB and Valkey datastore support\n\nThis patch introduces support for KeyDB and Valkey datastores.\nTo achieve this without code duplication, the new Redis common\nimplementation was created from a scratch. KeyDB and Valkey now\ninherit from this base. Existing legacy experimental Redis is\nuntouched.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nDepends-On: https://review.opendev.org/c/openstack/trove/+/990725\nChange-Id: Ia99918ff3df2078fed8636d46a9c4309f402eaba\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d69e809f219dc60883eac663322edc60890ea7b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d69e809f219dc60883eac663322edc60890ea7b8"}]},"branch":"refs/heads/master"},"f1cb082fdf86bc7d34c9eea7ca5c75091f8552d6":{"kind":"REWORK","_number":36,"created":"2026-06-08 07:03:54.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/56/989856/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/989856/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/989856/36"}}},"commit":{"parents":[{"commit":"3b4cdb94d6a4e1fd304928dcc8fdb5b188653016","subject":"Optimizing Zuul Jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3b4cdb94d6a4e1fd304928dcc8fdb5b188653016"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-13 08:49:07.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-08 07:03:12.000000000","tz":300},"subject":"WIP Add KeyDB and Valkey datastore support","message":"WIP Add KeyDB and Valkey datastore support\n\nThis patch introduces support for KeyDB and Valkey datastores.\nTo achieve this without code duplication, the new Redis common\nimplementation was created from a scratch. KeyDB and Valkey now\ninherit from this base. Existing legacy experimental Redis is\nuntouched.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nDepends-On: https://review.opendev.org/c/openstack/trove/+/990725\nChange-Id: Ia99918ff3df2078fed8636d46a9c4309f402eaba\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f1cb082fdf86bc7d34c9eea7ca5c75091f8552d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f1cb082fdf86bc7d34c9eea7ca5c75091f8552d6"}]},"branch":"refs/heads/master"},"688fb0d266c96c313eabf50a057af84e8608a1f1":{"kind":"REWORK","_number":37,"created":"2026-06-08 11:05:55.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/56/989856/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/989856/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/989856/37"}}},"commit":{"parents":[{"commit":"3b4cdb94d6a4e1fd304928dcc8fdb5b188653016","subject":"Optimizing Zuul Jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3b4cdb94d6a4e1fd304928dcc8fdb5b188653016"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-13 08:49:07.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-08 11:05:20.000000000","tz":300},"subject":"WIP Add KeyDB and Valkey datastore support","message":"WIP Add KeyDB and Valkey datastore support\n\nThis patch introduces support for KeyDB and Valkey datastores.\nTo achieve this without code duplication, the new Redis common\nimplementation was created from a scratch. KeyDB and Valkey now\ninherit from this base. Existing legacy experimental Redis is\nuntouched.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nDepends-On: https://review.opendev.org/c/openstack/trove/+/990725\nDepends-On: https://review.opendev.org/c/openstack/trove/+/992028\n\nChange-Id: Ia99918ff3df2078fed8636d46a9c4309f402eaba\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/688fb0d266c96c313eabf50a057af84e8608a1f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/688fb0d266c96c313eabf50a057af84e8608a1f1"}]},"branch":"refs/heads/master"},"1bd417237d14d662bc0921913de997a9fd68220b":{"kind":"REWORK","_number":38,"created":"2026-06-08 12:43:19.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/56/989856/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/989856/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/989856/38"}}},"commit":{"parents":[{"commit":"3b4cdb94d6a4e1fd304928dcc8fdb5b188653016","subject":"Optimizing Zuul Jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3b4cdb94d6a4e1fd304928dcc8fdb5b188653016"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-13 08:49:07.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-08 12:40:05.000000000","tz":300},"subject":"WIP Add KeyDB and Valkey datastore support","message":"WIP Add KeyDB and Valkey datastore support\n\nThis patch introduces support for KeyDB and Valkey datastores.\nTo achieve this without code duplication, the new Redis common\nimplementation was created from a scratch. KeyDB and Valkey now\ninherit from this base. Existing legacy experimental Redis is\nuntouched.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nDepends-On: https://review.opendev.org/c/openstack/trove/+/990725\nDepends-On: https://review.opendev.org/c/openstack/trove/+/992028\n\nChange-Id: Ia99918ff3df2078fed8636d46a9c4309f402eaba\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1bd417237d14d662bc0921913de997a9fd68220b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1bd417237d14d662bc0921913de997a9fd68220b"}]},"branch":"refs/heads/master"},"a6ec99fdd0810f708c8b728c62845fa8f40df931":{"kind":"REWORK","_number":39,"created":"2026-06-08 12:44:22.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/56/989856/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/989856/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/989856/39"}}},"commit":{"parents":[{"commit":"3b4cdb94d6a4e1fd304928dcc8fdb5b188653016","subject":"Optimizing Zuul Jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3b4cdb94d6a4e1fd304928dcc8fdb5b188653016"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-13 08:49:07.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-08 12:44:11.000000000","tz":300},"subject":"WIP Add KeyDB and Valkey datastore support","message":"WIP Add KeyDB and Valkey datastore support\n\nThis patch introduces support for KeyDB and Valkey datastores.\nTo achieve this without code duplication, the new Redis common\nimplementation was created from a scratch. KeyDB and Valkey now\ninherit from this base. Existing legacy experimental Redis is\nuntouched.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nDepends-On: https://review.opendev.org/c/openstack/trove/+/990725\nDepends-On: https://review.opendev.org/c/openstack/trove/+/992028\n\nChange-Id: Ia99918ff3df2078fed8636d46a9c4309f402eaba\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a6ec99fdd0810f708c8b728c62845fa8f40df931"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a6ec99fdd0810f708c8b728c62845fa8f40df931"}]},"branch":"refs/heads/master"},"64fcd106b2072c99c0ee2ca52ea24e42d8002132":{"kind":"REWORK","_number":40,"created":"2026-06-09 07:29:06.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/56/989856/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/989856/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/989856/40"}}},"commit":{"parents":[{"commit":"3b4cdb94d6a4e1fd304928dcc8fdb5b188653016","subject":"Optimizing Zuul Jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3b4cdb94d6a4e1fd304928dcc8fdb5b188653016"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-13 08:49:07.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-09 07:28:53.000000000","tz":300},"subject":"WIP Add KeyDB and Valkey datastore support","message":"WIP Add KeyDB and Valkey datastore support\n\nThis patch introduces support for KeyDB and Valkey datastores.\nTo achieve this without code duplication, the new Redis common\nimplementation was created from a scratch. KeyDB and Valkey now\ninherit from this base. Existing legacy experimental Redis is\nuntouched.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nDepends-On: https://review.opendev.org/c/openstack/trove/+/990725\nDepends-On: https://review.opendev.org/c/openstack/trove/+/992028\n\nChange-Id: Ia99918ff3df2078fed8636d46a9c4309f402eaba\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/64fcd106b2072c99c0ee2ca52ea24e42d8002132"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/64fcd106b2072c99c0ee2ca52ea24e42d8002132"}]},"branch":"refs/heads/master"},"01524785ca5e9b0e457009f21e84a83f927f5317":{"kind":"REWORK","_number":41,"created":"2026-06-26 06:13:38.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/56/989856/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/989856/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/989856/41"}}},"commit":{"parents":[{"commit":"e9489ce73e2415ed277eb51ab45394f9247acc30","subject":"Merge \"Improve data volume response in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e9489ce73e2415ed277eb51ab45394f9247acc30"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-13 08:49:07.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-26 06:11:57.000000000","tz":300},"subject":"WIP Add KeyDB and Valkey datastore support","message":"WIP Add KeyDB and Valkey datastore support\n\nThis patch introduces support for KeyDB and Valkey datastores.\nTo achieve this without code duplication, the new Redis common\nimplementation was created from a scratch. KeyDB and Valkey now\ninherit from this base. Existing legacy experimental Redis is\nuntouched.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nDepends-On: https://review.opendev.org/c/openstack/trove/+/992900\nDepends-On: https://review.opendev.org/c/openstack/trove/+/994159\nDepends-On: https://review.opendev.org/c/openstack/trove/+/992028\n\nBlueprint: redis-based-valkey-and-keydb-datastores\nChange-Id: Ia99918ff3df2078fed8636d46a9c4309f402eaba\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/01524785ca5e9b0e457009f21e84a83f927f5317"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/01524785ca5e9b0e457009f21e84a83f927f5317"}]},"branch":"refs/heads/master"},"f3d2eb4168f437c02bca8de83a521238d75a6ee7":{"kind":"REWORK","_number":42,"created":"2026-06-26 13:34:15.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/56/989856/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/989856/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/989856/42"}}},"commit":{"parents":[{"commit":"e9489ce73e2415ed277eb51ab45394f9247acc30","subject":"Merge \"Improve data volume response in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e9489ce73e2415ed277eb51ab45394f9247acc30"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-13 08:49:07.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-26 13:34:02.000000000","tz":300},"subject":"WIP Add KeyDB and Valkey datastore support","message":"WIP Add KeyDB and Valkey datastore support\n\nThis patch introduces support for KeyDB and Valkey datastores.\nTo achieve this without code duplication, the new Redis common\nimplementation was created from a scratch. KeyDB and Valkey now\ninherit from this base. Existing legacy experimental Redis is\nuntouched.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nDepends-On: https://review.opendev.org/c/openstack/trove/+/992900\nDepends-On: https://review.opendev.org/c/openstack/trove/+/994159\nDepends-On: https://review.opendev.org/c/openstack/trove/+/992028\n\nBlueprint: redis-based-valkey-and-keydb-datastores\nChange-Id: Ia99918ff3df2078fed8636d46a9c4309f402eaba\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f3d2eb4168f437c02bca8de83a521238d75a6ee7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f3d2eb4168f437c02bca8de83a521238d75a6ee7"}]},"branch":"refs/heads/master"},"5b9bcc367735fa9f18b9237d7484981a4fc2d4cd":{"kind":"REWORK","_number":43,"created":"2026-09-03 11:32:43.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/56/989856/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/989856/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/989856/43"}}},"commit":{"parents":[{"commit":"ee27b65b88af6ccf4326087910245c5b06cfee56","subject":"Remove workaround that fixed max_connections in auto.conf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ee27b65b88af6ccf4326087910245c5b06cfee56"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-13 08:49:07.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-03 11:26:38.000000000","tz":300},"subject":"WIP Add KeyDB and Valkey datastore support","message":"WIP Add KeyDB and Valkey datastore support\n\nThis patch introduces support for KeyDB and Valkey datastores.\nTo achieve this without code duplication, the new Redis common\nimplementation was created from a scratch. KeyDB and Valkey now\ninherit from this base. Existing legacy experimental Redis is\nuntouched.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\n\nBlueprint: redis-based-valkey-and-keydb-datastores\nChange-Id: Ia99918ff3df2078fed8636d46a9c4309f402eaba\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5b9bcc367735fa9f18b9237d7484981a4fc2d4cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5b9bcc367735fa9f18b9237d7484981a4fc2d4cd"}]},"branch":"refs/heads/master"},"e73cf9c77a3d7ce0abde887d997975a7e3c11547":{"kind":"NO_CODE_CHANGE","_number":44,"created":"2026-09-03 12:20:24.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/56/989856/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/989856/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/989856/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/989856/44"}}},"commit":{"parents":[{"commit":"ee27b65b88af6ccf4326087910245c5b06cfee56","subject":"Remove workaround that fixed max_connections in auto.conf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ee27b65b88af6ccf4326087910245c5b06cfee56"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-13 08:49:07.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-03 12:20:09.000000000","tz":300},"subject":"Add KeyDB and Valkey datastore support","message":"Add KeyDB and Valkey datastore support\n\nThis patch introduces support for KeyDB and Valkey datastores.\nTo achieve this without code duplication, the new Redis common\nimplementation was created from a scratch. KeyDB and Valkey now\ninherit from this base. Existing legacy experimental Redis is\nuntouched.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\n\nBlueprint: redis-based-valkey-and-keydb-datastores\nChange-Id: Ia99918ff3df2078fed8636d46a9c4309f402eaba\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e73cf9c77a3d7ce0abde887d997975a7e3c11547"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e73cf9c77a3d7ce0abde887d997975a7e3c11547"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
