)]}'
{"id":"openstack%2Ftrove~989898","triplet_id":"openstack%2Ftrove~master~Ia0ef6790292f8e6ab56bec1834cec115dd1443ca","project":"openstack/trove","branch":"master","topic":"full-tests-keydb-valkey","attention_set":{},"removed_from_attention_set":{"36080":{"account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"last_update":"2026-05-25 05:29:45.000000000","reason":"\u003cGERRIT_ACCOUNT_36080\u003e replied on the change","reason_account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}}},"hashtags":[],"change_id":"Ia0ef6790292f8e6ab56bec1834cec115dd1443ca","subject":"[DNM] Full tests for keyDB \u0026 Valkey","status":"NEW","created":"2026-05-24 13:45:36.000000000","updated":"2026-05-25 12:23:10.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"17770af93aecab9844ff8fdb5bf571c4022d1366","_number":989898,"virtual_id_number":989898,"owner":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"actions":{},"labels":{"Verified":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-24 13:46:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-24 13:46:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c0ae0c4e4e797724faa898dfdd2cedb5ba651a66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-24 13:45:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f5653649f54ea9948308e2e46df1988a5feea677","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-24 13:46:13.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/trove for 989898,1","accounts_in_message":[],"_revision_number":1},{"id":"31731f38146fafc0261a7bb8f189f653f796ac3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-24 13:46:17.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/trove for 989898,1","accounts_in_message":[],"_revision_number":1},{"id":"3817b56c0367e0f89536d94d26e0e9653205570b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-24 13:50:14.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"947694cb3d324333c0dd160387cc2a99d6a685fe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-24 13:50:53.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/trove for 989898,2","accounts_in_message":[],"_revision_number":2},{"id":"b5f633f1cbb8f6b2f37f8343a554934a31891c7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-24 13:50:56.000000000","message":"Patch Set 2: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/trove for 989898,2","accounts_in_message":[],"_revision_number":2},{"id":"b7f8aade421745225d8d2cbb0ab7683abac80ef4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-24 13:53:25.000000000","message":"Uploaded patch set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"1e5fda7f8db496230b9b66903b99481a4b9e2944","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-24 13:54:08.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/trove for 989898,3","accounts_in_message":[],"_revision_number":3},{"id":"92018422d73d7188b61c5d6352ce90cdcc6d56f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-24 13:54:10.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/trove for 989898,3","accounts_in_message":[],"_revision_number":3},{"id":"8119c2791a9c8fab53d56cbd39b32e8e1db4a085","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-24 14:00:05.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"140d212623a27adf553d8db9807ce40ca5baedac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-24 14:15:34.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32ee601e24aa4cae9b7ed9d304205e1e\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2a666f4ffa3f41c7ac1baf1866f9602c : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6cc1024a79bb4c608068be86d5da85b7 : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c9ac08f28fef4f1589cf49022075ac31 : SUCCESS in 12m 44s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"949ade7e257943c5caf2b0326315bd34b33bbc53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-24 14:26:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03d152477eea458eaf58c77cc4d4fa01\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/921418df50494f6496d314278a304551 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de6be3b84b584599a904973c7045c840 : SUCCESS in 4m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9331b84f187c4ac89fb7ef17c7ddf9ee : SUCCESS in 4m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e9f725131f844918b208e5ae8295f6c3 : SUCCESS in 4m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/445ef5f70d0447f6a934abe498c2f6ce : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f31c2f591a54ee8a765b18062de29f8 : SUCCESS in 5m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c8e22bdfeacc45f6ac9a93f14b151c31 : SUCCESS in 5m 11s\n- trove-tempest-ubuntu-base-keydb https://zuul.opendev.org/t/openstack/build/c2c76c858b8f4ac9b19931522a09f915 : FAILURE in 22m 59s (non-voting)\n- trove-tempest-ubuntu-backup-keydb https://zuul.opendev.org/t/openstack/build/356046d2b95c498283253685c8ba813a : FAILURE in 24m 49s (non-voting)\n- trove-tempest-ubuntu-replication-keydb https://zuul.opendev.org/t/openstack/build/2a9ecec9080c45679a89a89416e82724 : FAILURE in 23m 53s (non-voting)\n- trove-tempest-ubuntu-actions-keydb https://zuul.opendev.org/t/openstack/build/868f62e6286c4d7a999937a8327e209c : FAILURE in 24m 55s (non-voting)\n- trove-tempest-cinder-storage-driver-keydb https://zuul.opendev.org/t/openstack/build/d6cf508a6669487e97c02b5efe3d3eb7 : FAILURE in 23m 27s (non-voting)\n- trove-tempest-ubuntu-base-valkey https://zuul.opendev.org/t/openstack/build/444c2daea25243c991544ac8f2c3b3ab : FAILURE in 23m 24s (non-voting)\n- trove-tempest-ubuntu-backup-valkey https://zuul.opendev.org/t/openstack/build/3f3a849ce55343afbb9c9db69f8c87e0 : FAILURE in 24m 06s (non-voting)\n- trove-tempest-ubuntu-replication-valkey https://zuul.opendev.org/t/openstack/build/3aa0d1d1c78c41edb7cbe60cea883e27 : FAILURE in 23m 54s (non-voting)\n- trove-tempest-ubuntu-actions-valkey https://zuul.opendev.org/t/openstack/build/0b67de18448147468c47f48cc5db2e7a : FAILURE in 23m 47s (non-voting)\n- trove-tempest-cinder-storage-driver-valkey https://zuul.opendev.org/t/openstack/build/f3d0ded123284e9e8fccc73f90c0f2c8 : FAILURE in 24m 42s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"68240aec04fe9e0b3b7099b57c83e103c8966790","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-24 14:39:05.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"fcdf59431741506ebe46dd5d13e680f4d6ae55a4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-24 14:49:06.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/008102074a1d469187e624895939ced9\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/693f697bb4f24233870ef7bc74f59005 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/47dfd486733745448ab3915b837dbe4a : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/65b8d3aa112b4600aa89e179ff80a139 : SUCCESS in 7m 25s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"fce9f24f84d70e14d469a6b512bded7c4943cfc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-24 15:24:17.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/21b2b0f208ae477dbf0e852b18ae68f8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f89cabf93e7a47ce83fb7a570aa770a0 : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/751fb15c4f784bf5a555083d8ba68ae9 : SUCCESS in 2m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c83b1242a039407a91739e6c3d72735a : SUCCESS in 4m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9a8ad561352b465bb7c5bddf3b2c079f : SUCCESS in 5m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9dba92cb865b4baca6e19b48561952df : SUCCESS in 7m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33074cf8b234498ca345e165906719bc : SUCCESS in 5m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4f4d06f67654eed85cd72103d4e822d : SUCCESS in 6m 02s\n- trove-tempest-ubuntu-base-keydb https://zuul.opendev.org/t/openstack/build/5ec90fa0b512441985f213b851fccb72 : FAILURE in 25m 21s\n- trove-tempest-ubuntu-backup-keydb https://zuul.opendev.org/t/openstack/build/42aac37c94ea440abca37bdcc8eafc0c : FAILURE in 43m 25s\n- trove-tempest-ubuntu-replication-keydb https://zuul.opendev.org/t/openstack/build/9bc5b020214a4808bd4bac5f440a5e43 : FAILURE in 33m 49s\n- trove-tempest-ubuntu-actions-keydb https://zuul.opendev.org/t/openstack/build/0ac2d9f8d3ae41468adf7cb062a88ccb : FAILURE in 43m 09s\n- trove-tempest-cinder-storage-driver-keydb https://zuul.opendev.org/t/openstack/build/39f3f83c6f1e4e74b602d3a3d63228e0 : FAILURE in 33m 42s\n- trove-tempest-ubuntu-base-valkey https://zuul.opendev.org/t/openstack/build/93870be7a3154440ba3453feaf05dd0e : FAILURE in 32m 48s\n- trove-tempest-ubuntu-backup-valkey https://zuul.opendev.org/t/openstack/build/0400fe1842344fb6bd4243172e94dd49 : FAILURE in 33m 58s\n- trove-tempest-ubuntu-replication-valkey https://zuul.opendev.org/t/openstack/build/595c420c77c94afa814fb16ea0158d54 : FAILURE in 34m 34s\n- trove-tempest-ubuntu-actions-valkey https://zuul.opendev.org/t/openstack/build/e88e72a0d1bf4333bb427c4f5e5fa5c1 : FAILURE in 43m 09s\n- trove-tempest-cinder-storage-driver-valkey https://zuul.opendev.org/t/openstack/build/039b4313322d4f5584e9eb8457995021 : FAILURE in 25m 20s","accounts_in_message":[],"_revision_number":5},{"id":"ecef97936cca56e4bcd15d4c5bef117d65e7b1b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-24 15:34:49.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"66d1c07c84a949394427f18d8958ed954630fa90","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-24 15:49:26.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa68d71960224fc886193f2d60800658\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f2890212e4824cc7bfd158ef992191ec : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/18136a8ef5684518967b3e4a65d83014 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/27fab2fb89a9464698556ec3145458ad : SUCCESS in 11m 02s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"81a8feb29f2cd2471825bf584c8103405083828c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-24 17:23:54.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c6fa5e18d6224e75942873278664bb2d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/45e2aa8336064e6c8ce1c8685f59064b : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1023a1e506dc4744b5c1302d4c7edc48 : SUCCESS in 4m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e30457c6708c4964a84f5ad9f475d659 : SUCCESS in 3m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cee43d70b0d74314b3428a6ab3a5c885 : SUCCESS in 4m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8ea0df4aa59a41278f91a385999baf6f : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eec2e1b899f340228d24490e3fd0398a : SUCCESS in 5m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/68f6784a119d49658a28c16f483d7925 : SUCCESS in 6m 49s\n- trove-tempest-ubuntu-base-keydb https://zuul.opendev.org/t/openstack/build/2f1e607c06ce44299e8308534130e575 : SUCCESS in 45m 19s\n- trove-tempest-ubuntu-backup-keydb https://zuul.opendev.org/t/openstack/build/baf583b3eec946598b67d85b7c9233df : SUCCESS in 1h 05m 53s\n- trove-tempest-ubuntu-replication-keydb https://zuul.opendev.org/t/openstack/build/583a4427bbe34e9bb22d419181dbddc3 : SUCCESS in 1h 08m 38s\n- trove-tempest-ubuntu-actions-keydb https://zuul.opendev.org/t/openstack/build/96856cf90b33457d892a2933723a988b : FAILURE in 1h 46m 41s\n- trove-tempest-cinder-storage-driver-keydb https://zuul.opendev.org/t/openstack/build/b8a205f22a6547ca98806513fa86f05b : SUCCESS in 1h 01m 57s\n- trove-tempest-ubuntu-base-valkey https://zuul.opendev.org/t/openstack/build/e9da9bff5d304654b44731fd2e1db06d : SUCCESS in 43m 35s\n- trove-tempest-ubuntu-backup-valkey https://zuul.opendev.org/t/openstack/build/dc32c4eda38f4724ad67bc105026745b : SUCCESS in 1h 09m 35s\n- trove-tempest-ubuntu-replication-valkey https://zuul.opendev.org/t/openstack/build/826433275097475eaddd6565928f15d1 : SUCCESS in 1h 06m 33s\n- trove-tempest-ubuntu-actions-valkey https://zuul.opendev.org/t/openstack/build/ad796ec8382b46bea5337a3351569f39 : FAILURE in 1h 46m 40s\n- trove-tempest-cinder-storage-driver-valkey https://zuul.opendev.org/t/openstack/build/bb2d61ef3e124e179bcb586f7b0509b0 : SUCCESS in 1h 00m 42s","accounts_in_message":[],"_revision_number":6},{"id":"62ed7ae6047c0b5a03134024252140369d2676f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-24 18:00:19.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"aa9344882664aacace9533832f02279992bcd233","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-24 18:13:59.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/890ec6108cc341c0828815e5018e8f7c\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4f5d97374cc74aa4bfd43e5aaa1d277f : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/dc02275759944f80a228094bbc28b8f2 : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4c4d9281af6340d3b0495b9b7b3cb43f : SUCCESS in 11m 01s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"43b3e5517b12b31f8d91eb3b650db6f041bb01af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-24 19:12:51.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/30834d42d82e41618b6a280cfdbbf946\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ec44a99a6fa45569d6f89322279e7a7 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4119693170e24c10900e91dfd8d0c66c : SUCCESS in 5m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/022089d122264a37be471b043efa13a3 : SUCCESS in 4m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cd9eaa2e9b0e45438246d5dd802ccee6 : SUCCESS in 4m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/74a53de364134459abf20907b4ce49a3 : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/498e0c98aef54372a8130d104b5828ff : SUCCESS in 9m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d69d93e35754ffcab79a9598cdb437c : SUCCESS in 6m 48s\n- trove-tempest-ubuntu-base-keydb https://zuul.opendev.org/t/openstack/build/13d02a5ec08f464daa62e059341cdcae : SUCCESS in 39m 25s\n- trove-tempest-ubuntu-backup-keydb https://zuul.opendev.org/t/openstack/build/2919d282ba5c4c538e763225afbeaa7d : SUCCESS in 1h 03m 09s\n- trove-tempest-ubuntu-replication-keydb https://zuul.opendev.org/t/openstack/build/58d9844c71ab4f0cae597c20cc88551e : SUCCESS in 1h 03m 14s\n- trove-tempest-ubuntu-actions-keydb https://zuul.opendev.org/t/openstack/build/29d68ba7af164e81ad81abaa5de25e39 : FAILURE in 1h 07m 49s\n- trove-tempest-cinder-storage-driver-keydb https://zuul.opendev.org/t/openstack/build/765640dedb4247ca920b937253eb95d0 : SUCCESS in 1h 02m 45s\n- trove-tempest-ubuntu-base-valkey https://zuul.opendev.org/t/openstack/build/be72d433a5144047b4518ad93b40b6de : SUCCESS in 42m 24s\n- trove-tempest-ubuntu-backup-valkey https://zuul.opendev.org/t/openstack/build/b50ae40e2a554ccbaac8176d2bdb0a06 : SUCCESS in 1h 04m 35s\n- trove-tempest-ubuntu-replication-valkey https://zuul.opendev.org/t/openstack/build/03af83ea351d4b2a8fb60aefead98e29 : SUCCESS in 1h 04m 14s\n- trove-tempest-ubuntu-actions-valkey https://zuul.opendev.org/t/openstack/build/27ddc10f6e694d6eb03e2b471518aae5 : FAILURE in 1h 10m 48s\n- trove-tempest-cinder-storage-driver-valkey https://zuul.opendev.org/t/openstack/build/d2cc345920e447928990ba5841a09754 : SUCCESS in 1h 02m 18s","accounts_in_message":[],"_revision_number":7},{"id":"e1e9d200cdcfd9dcd6f4e120a139e9cf1033b718","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-25 05:29:45.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"953fa85b554f0ac5c354582c2172d9e2026375ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-25 07:57:42.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"ba5d877bb6522fbb829b2e7d2651dfce187d2fa1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-25 11:43:30.000000000","message":"Uploaded patch set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"17770af93aecab9844ff8fdb5bf571c4022d1366","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-25 12:23:10.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"4d345a81d707565867104e0cef5e9d56611e2155","revisions":{"05808c728c2af4040f272964968bb89902c2543e":{"kind":"REWORK","_number":1,"created":"2026-05-24 13:45:36.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/98/989898/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/98/989898/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/989898/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/989898/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/989898/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/98/989898/1"}}},"commit":{"parents":[{"commit":"0994e6b1511359dedd20a4d1bf01608527f4c65a","subject":"Mark hashlib.md5 for non security purpose","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0994e6b1511359dedd20a4d1bf01608527f4c65a"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-24 13:43:42.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-24 13:45:12.000000000","tz":300},"subject":"[DNM] Full tests for keyDB \u0026 Valkey","message":"[DNM] Full tests for keyDB \u0026 Valkey\n\nDepends-On: https://review.opendev.org/c/openstack/trove/+/989856\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nDepends-On: https://review.opendev.org/c/openstack/trove/+/989888\nChange-Id: Ia0ef6790292f8e6ab56bec1834cec115dd1443ca\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/05808c728c2af4040f272964968bb89902c2543e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/05808c728c2af4040f272964968bb89902c2543e"}]},"branch":"refs/heads/master"},"43c071d6d11fbd33f0fe6b5a551cf7e2ef576956":{"kind":"REWORK","_number":2,"created":"2026-05-24 13:50:14.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/98/989898/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/98/989898/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/989898/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/989898/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/989898/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/98/989898/2"}}},"commit":{"parents":[{"commit":"0994e6b1511359dedd20a4d1bf01608527f4c65a","subject":"Mark hashlib.md5 for non security purpose","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0994e6b1511359dedd20a4d1bf01608527f4c65a"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-24 13:43:42.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-24 13:49:52.000000000","tz":300},"subject":"[DNM] Full tests for keyDB \u0026 Valkey","message":"[DNM] Full tests for keyDB \u0026 Valkey\n\nDepends-On: https://review.opendev.org/c/openstack/trove/+/989856\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nDepends-On: https://review.opendev.org/c/openstack/trove/+/989888\nChange-Id: Ia0ef6790292f8e6ab56bec1834cec115dd1443ca\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/43c071d6d11fbd33f0fe6b5a551cf7e2ef576956"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/43c071d6d11fbd33f0fe6b5a551cf7e2ef576956"}]},"branch":"refs/heads/master"},"52caaaa814f736be3830d2a89d5e5c76d2d1d67e":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-05-24 13:53:25.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/98/989898/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/98/989898/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/989898/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/989898/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/989898/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/98/989898/3"}}},"commit":{"parents":[{"commit":"0994e6b1511359dedd20a4d1bf01608527f4c65a","subject":"Mark hashlib.md5 for non security purpose","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0994e6b1511359dedd20a4d1bf01608527f4c65a"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-24 13:43:42.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-24 13:53:09.000000000","tz":300},"subject":"[DNM] Full tests for keyDB \u0026 Valkey","message":"[DNM] Full tests for keyDB \u0026 Valkey\n\nDepends-On: https://review.opendev.org/c/openstack/trove/+/989856\nChange-Id: Ia0ef6790292f8e6ab56bec1834cec115dd1443ca\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/52caaaa814f736be3830d2a89d5e5c76d2d1d67e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/52caaaa814f736be3830d2a89d5e5c76d2d1d67e"}]},"branch":"refs/heads/master"},"77f14562c7edc308c1a966aa7516f1bffb192765":{"kind":"REWORK","_number":4,"created":"2026-05-24 14:00:05.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/98/989898/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/98/989898/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/989898/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/989898/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/989898/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/98/989898/4"}}},"commit":{"parents":[{"commit":"b8cd2c1da1efc2605af50dd7a39f1d243e09cbb4","subject":"WIP Add KeyDB and Valkey datastore support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b8cd2c1da1efc2605af50dd7a39f1d243e09cbb4"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-24 13:43:42.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-24 13:58:05.000000000","tz":300},"subject":"[DNM] Full tests for keyDB \u0026 Valkey","message":"[DNM] Full tests for keyDB \u0026 Valkey\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nDepends-On: https://review.opendev.org/c/openstack/trove/+/989888\nChange-Id: Ia0ef6790292f8e6ab56bec1834cec115dd1443ca\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/77f14562c7edc308c1a966aa7516f1bffb192765"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/77f14562c7edc308c1a966aa7516f1bffb192765"}]},"branch":"refs/heads/master"},"5a96d830d826c6ec0ae3b400ef3d6f8e20c9c62b":{"kind":"REWORK","_number":5,"created":"2026-05-24 14:39:05.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/98/989898/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/98/989898/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/989898/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/989898/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/989898/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/98/989898/5"}}},"commit":{"parents":[{"commit":"b8cd2c1da1efc2605af50dd7a39f1d243e09cbb4","subject":"WIP Add KeyDB and Valkey datastore support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b8cd2c1da1efc2605af50dd7a39f1d243e09cbb4"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-24 13:43:42.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-24 14:38:52.000000000","tz":300},"subject":"[DNM] Full tests for keyDB \u0026 Valkey","message":"[DNM] Full tests for keyDB \u0026 Valkey\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nDepends-On: https://review.opendev.org/c/openstack/trove/+/989888\nChange-Id: Ia0ef6790292f8e6ab56bec1834cec115dd1443ca\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5a96d830d826c6ec0ae3b400ef3d6f8e20c9c62b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5a96d830d826c6ec0ae3b400ef3d6f8e20c9c62b"}]},"branch":"refs/heads/master"},"9ac900e70c5482808b3c87013c104fb8c23431b2":{"kind":"REWORK","_number":6,"created":"2026-05-24 15:34:49.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/98/989898/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/98/989898/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/989898/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/989898/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/989898/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/98/989898/6"}}},"commit":{"parents":[{"commit":"b8cd2c1da1efc2605af50dd7a39f1d243e09cbb4","subject":"WIP Add KeyDB and Valkey datastore support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b8cd2c1da1efc2605af50dd7a39f1d243e09cbb4"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-24 13:43:42.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-24 15:34:33.000000000","tz":300},"subject":"[DNM] Full tests for keyDB \u0026 Valkey","message":"[DNM] Full tests for keyDB \u0026 Valkey\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nDepends-On: https://review.opendev.org/c/openstack/trove/+/989888\nChange-Id: Ia0ef6790292f8e6ab56bec1834cec115dd1443ca\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9ac900e70c5482808b3c87013c104fb8c23431b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9ac900e70c5482808b3c87013c104fb8c23431b2"}]},"branch":"refs/heads/master"},"7d7a2424ca0eb5ecb227c0150621286b5856e770":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-05-24 18:00:19.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/98/989898/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/98/989898/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/989898/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/989898/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/989898/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/98/989898/7"}}},"commit":{"parents":[{"commit":"382c1ffc93450762edf2c0c66246b6eeade43463","subject":"WIP Add KeyDB and Valkey datastore support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/382c1ffc93450762edf2c0c66246b6eeade43463"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-24 13:43:42.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-24 17:59:49.000000000","tz":300},"subject":"[DNM] Full tests for keyDB \u0026 Valkey","message":"[DNM] Full tests for keyDB \u0026 Valkey\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nDepends-On: https://review.opendev.org/c/openstack/trove/+/989888\nChange-Id: Ia0ef6790292f8e6ab56bec1834cec115dd1443ca\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7d7a2424ca0eb5ecb227c0150621286b5856e770"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7d7a2424ca0eb5ecb227c0150621286b5856e770"}]},"branch":"refs/heads/master"},"6031f10f08c6d53b914fd1f7c8e561aa1fadd7fd":{"kind":"REWORK","_number":8,"created":"2026-05-25 07:57:42.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/98/989898/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/98/989898/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/989898/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/989898/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/989898/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/98/989898/8"}}},"commit":{"parents":[{"commit":"26f182f09dd22401bd04886da8de789278198eab","subject":"WIP Add KeyDB and Valkey datastore support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/26f182f09dd22401bd04886da8de789278198eab"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-24 13:43:42.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-25 07:57:00.000000000","tz":300},"subject":"[DNM] Full tests for keyDB \u0026 Valkey","message":"[DNM] Full tests for keyDB \u0026 Valkey\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nDepends-On: https://review.opendev.org/c/openstack/trove/+/989888\nChange-Id: Ia0ef6790292f8e6ab56bec1834cec115dd1443ca\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6031f10f08c6d53b914fd1f7c8e561aa1fadd7fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6031f10f08c6d53b914fd1f7c8e561aa1fadd7fd"}]},"branch":"refs/heads/master"},"4d345a81d707565867104e0cef5e9d56611e2155":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2026-05-25 11:43:30.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/98/989898/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/98/989898/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/989898/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/989898/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/989898/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/98/989898/9"}}},"commit":{"parents":[{"commit":"26f182f09dd22401bd04886da8de789278198eab","subject":"WIP Add KeyDB and Valkey datastore support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/26f182f09dd22401bd04886da8de789278198eab"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-24 13:43:42.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-25 11:43:11.000000000","tz":300},"subject":"[DNM] Full tests for keyDB \u0026 Valkey","message":"[DNM] Full tests for keyDB \u0026 Valkey\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/989855\nChange-Id: Ia0ef6790292f8e6ab56bec1834cec115dd1443ca\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4d345a81d707565867104e0cef5e9d56611e2155"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4d345a81d707565867104e0cef5e9d56611e2155"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY"},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
