)]}'
{"id":"openstack%2Ftrove~990019","triplet_id":"openstack%2Ftrove~master~I2907fc62377572c2f879d51405f442fba04dc346","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-05-26 11:56:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2907fc62377572c2f879d51405f442fba04dc346","subject":"docs: update for pdf-build","status":"MERGED","created":"2026-05-25 21:36:12.000000000","updated":"2026-05-26 11:57:52.000000000","submitted":"2026-05-26 11:56:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"990019","meta_rev_id":"24535fdff114fe9a99ee906441892a9652e97e75","_number":990019,"virtual_id_number":990019,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-26 11:56:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-05-26 11:43:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-05-26 11:44:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2026-05-26 11:44:24.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-25 22:05:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-26 11:43:41.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"62d9c79d487d3f72a1beb66af685b1d9cabfc220","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-25 21:36:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9c977a814e1181d959d91c8064d9f75c61d44a26","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-25 21:38:51.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"4fec413f40d9fcc60ee5a6fd17e3f58a647aab5a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-25 21:45:06.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"60f80f07d07e30cb21b36fd7704182a3af02563e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 22:05:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c27a087d4b249ecb0d213a0148948d5\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6a8ffa81daec4e86ac0614bc6133e0d8 : SUCCESS in 5m 32s\n- requirements-check https://zuul.opendev.org/t/openstack/build/507fc09fbe8a46b0a024d049359fd675 : SUCCESS in 3m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b484722ef2a443d88bce0ae6720346b8 : SUCCESS in 7m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5aa922c3e3d4ab8ad5101c773ed21f0 : SUCCESS in 5m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9c30e7567998416b9ab67346baf2a8d9 : SUCCESS in 6m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f4471990725e4ef1a0ab9a4f8d92ef34 : SUCCESS in 4m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6e5b965c12354b0cbde130edcf44af0c : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/930cf536059848b9b2e1b9ddcf7d1c10 : SUCCESS in 8m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8d9588b3e924b48925f235e604a5da7 : SUCCESS in 4m 47s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/79a21839025d4cc3aed4b424ea0dd192 : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":3},{"id":"589218828075dbeb164876f0e0e4aeee2dfad9d3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-25 22:10:24.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"6c126902ff8a7efe5b89bde2451a66b90434a839","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-25 22:10:27.000000000","message":"Patch Set 4:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"fd2dd292f42f20e76569574abe40facb42a7b8cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 22:30:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c203892aed0b440db78581b273e88afd\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e7dbba60f89b4747b939f16b7d8c8daf : SUCCESS in 5m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2b78fb01509948d88a2e1511935c54f8 : SUCCESS in 4m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2fefda3eb4114faabe3f85132d7829d8 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46065eaca099463c8aa97a0844da02a8 : SUCCESS in 5m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/24c7a3f0e6654e16b34eb97597fb6bf5 : SUCCESS in 4m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4de1d2998bc94e97ac57f33d05ea3c86 : SUCCESS in 4m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b7e9aae7b7454da09bbd91223295d2a6 : SUCCESS in 7m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fca009cb9394a14ac670af9e69b8f14 : SUCCESS in 10m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a135c1e1a884bc889cce6ceeca68209 : SUCCESS in 6m 02s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d64fb324fb274fdc9a28cdbda5ddc236 : SUCCESS in 6m 44s","accounts_in_message":[],"_revision_number":4},{"id":"41cba1c92db4893e55b618ce4a879def811bfdc1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 22:51:15.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e148f79c7d64b029a0a4faf5a5b51fc\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/42cad5f5afa644198019537968c26610 : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f16b36cc44184cfb8384624114f373dc : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/825c83242dfb42d9a1a3f38b6639ca3a : SUCCESS in 14m 04s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"51b8c70bc722c54943f147dabb6d223f22927bc8","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-05-26 11:43:41.000000000","message":"Patch Set 4: Backport-Candidate+1 Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"b46b17bb85ca254e56e42844d2d13038c381aecf","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-05-26 11:44:24.000000000","message":"Patch Set 4: Backport-Candidate-1 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"c1e9e78c9bb6f4554626e956f4a900f25bf1b7c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 11:44:35.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"f4c00149cd0f3c0ad016addc8388ff9ba54f6c5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 11:56:54.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/973771df48af47048dfe54e4ab6929e0\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/93032f737fd8485bad02a69919332cac : SUCCESS in 5m 35s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b0fcf32ebe04431cbcc236c78bcdbed5 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1e4154f972043d99d5334fcc042232e : SUCCESS in 4m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aa9ac4a337144d4fae2349b4aa3a2837 : SUCCESS in 6m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ef4e2e17ba7246f289b78faf29a43b74 : SUCCESS in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74f2e86332454ce299fe5247c6c2965b : SUCCESS in 5m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9855b45f38744bd784a6531fc944e51b : SUCCESS in 7m 00s","accounts_in_message":[],"_revision_number":4},{"id":"e6884322d7202fadef3d9cdefc0d47051068092d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 11:56:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"24535fdff114fe9a99ee906441892a9652e97e75","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 11:57:52.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/369f263449854c3b87a92449bfbb9dde\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/9f3dcfcde0f142db833978c5afcfeaa0 : SUCCESS in 44s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cc2663f0e3a43b8ad724840b510f56b : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5bb9f96b2c6f4769a3b7bac251d77973 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"9a7a9eb62ed710c30480c73fe08bcdc56898cead","revisions":{"b1bdf68feeb6ec96847d6dae6e897b9a575f9f51":{"kind":"REWORK","_number":1,"created":"2026-05-25 21:36:12.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/19/990019/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/19/990019/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/19/990019/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/19/990019/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/19/990019/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/19/990019/1"}}},"commit":{"parents":[{"commit":"c40a2d54de13d73630af9155f89c0cee3f0938c7","subject":"Merge \"Use version 3 of Cinder API by default\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c40a2d54de13d73630af9155f89c0cee3f0938c7"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-25 21:36:12.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-25 21:36:12.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I2907fc62377572c2f879d51405f442fba04dc346\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b1bdf68feeb6ec96847d6dae6e897b9a575f9f51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b1bdf68feeb6ec96847d6dae6e897b9a575f9f51"}]},"branch":"refs/heads/master","conflicts":{"contains_conflicts":false}},"e79a4581eea74afe0d4d0c78fc9ba7ea02c02cf0":{"kind":"REWORK","_number":2,"created":"2026-05-25 21:38:51.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/19/990019/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/19/990019/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/19/990019/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/19/990019/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/19/990019/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/19/990019/2"}}},"commit":{"parents":[{"commit":"c40a2d54de13d73630af9155f89c0cee3f0938c7","subject":"Merge \"Use version 3 of Cinder API by default\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c40a2d54de13d73630af9155f89c0cee3f0938c7"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-25 21:36:12.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-25 21:38:51.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I2907fc62377572c2f879d51405f442fba04dc346\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e79a4581eea74afe0d4d0c78fc9ba7ea02c02cf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e79a4581eea74afe0d4d0c78fc9ba7ea02c02cf0"}]},"branch":"refs/heads/master"},"8b19585dad08772df597ac2ab2727674a0e99730":{"kind":"REWORK","_number":3,"created":"2026-05-25 21:45:06.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/19/990019/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/19/990019/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/19/990019/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/19/990019/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/19/990019/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/19/990019/3"}}},"commit":{"parents":[{"commit":"c40a2d54de13d73630af9155f89c0cee3f0938c7","subject":"Merge \"Use version 3 of Cinder API by default\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c40a2d54de13d73630af9155f89c0cee3f0938c7"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-25 21:36:12.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-25 21:45:06.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I2907fc62377572c2f879d51405f442fba04dc346\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8b19585dad08772df597ac2ab2727674a0e99730"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8b19585dad08772df597ac2ab2727674a0e99730"}]},"branch":"refs/heads/master"},"9a7a9eb62ed710c30480c73fe08bcdc56898cead":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-05-25 22:10:24.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/19/990019/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/19/990019/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/19/990019/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/19/990019/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/19/990019/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/19/990019/4"}}},"commit":{"parents":[{"commit":"c40a2d54de13d73630af9155f89c0cee3f0938c7","subject":"Merge \"Use version 3 of Cinder API by default\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c40a2d54de13d73630af9155f89c0cee3f0938c7"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-25 21:36:12.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-25 22:10:24.000000000","tz":0},"subject":"docs: update for pdf-build","message":"docs: update for pdf-build\n\nChange-Id: I2907fc62377572c2f879d51405f442fba04dc346\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9a7a9eb62ed710c30480c73fe08bcdc56898cead"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9a7a9eb62ed710c30480c73fe08bcdc56898cead"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
