)]}'
{"id":"openstack%2Ftrove~990158","triplet_id":"openstack%2Ftrove~stable%2F2025.1~If6fd1f65354def972f4a7e568a1763ee486c310a","project":"openstack/trove","branch":"stable/2025.1","topic":"feature/allow-upgrade-to-same-version-stable/2026.1-stable/2025.2-stable/2025.1","attention_set":{},"removed_from_attention_set":{"36080":{"account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"last_update":"2026-06-15 09:24:12.000000000","reason":"\u003cGERRIT_ACCOUNT_36080\u003e replied on the change","reason_account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2026-06-15 18:24:13.000000000","reason":"Change was submitted"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-06-15 15:57:50.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"If6fd1f65354def972f4a7e568a1763ee486c310a","subject":"Allow upgrade to the same version","status":"MERGED","created":"2026-05-26 18:12:26.000000000","updated":"2026-06-15 18:25:10.000000000","submitted":"2026-06-15 18:24:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"990158-feature/allow-upgrade-to-same-version-stable/2026.1-stable/2025.2-stable/2025.1","cherry_pick_of_change":990155,"cherry_pick_of_patch_set":1,"meta_rev_id":"cfb68b639800b8199a06c1b8c96868d9bdc5bcb9","_number":990158,"virtual_id_number":990158,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-15 18:24:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":2,"date":"2026-06-15 09:24:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":1,"date":"2026-06-15 09:24:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-26 18:12:26.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"state":"CC"},{"updated":"2026-05-26 18:13:00.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"state":"REVIEWER"},{"updated":"2026-05-26 20:05:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-14 18:10:26.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"f47a3a7da4dd9f5bba714837c3a65f1cddb17220","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-26 18:12:26.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2025.2.","accounts_in_message":[],"_revision_number":1},{"id":"e5df0dca6f4acd7ef63a5655d007a5eb286076b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 20:05:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4fe9f6c444604703a84a4be6a4c0d34c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18219f97c85f4b28be138d2b75633ce5 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/463e97a6fdbf430985eda94d73160419 : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/222d8493b5ee47fdb6ab283a0fc92548 : SUCCESS in 6m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cc201fe8afbd48b69f24fa5f3d650560 : SUCCESS in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cd2804c2da84ff5a2a3be592c35791e : SUCCESS in 4m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/97b60115ad6d4ff98eb4241e5bec46e2 : SUCCESS in 7m 38s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/bff570afb3e1440997b4c320200d61b2 : SUCCESS in 44m 06s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/936c6f78c0d643e6bd453c84cd844b5e : FAILURE in 24m 48s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/0c2314d92efe4be5ac30a0fda1242436 : SUCCESS in 1h 47m 00s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/da63497e0c134b238e53ab12f77a0119 : SUCCESS in 1h 29m 15s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e954238dc5bc4c53bc74ad8e18021fbe : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":1},{"id":"2422eb74bc4dcc51ee8ab0d0e796c9be93a93a38","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-05 21:08:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bcce5f28fed78912b3cc42bdf30b2d24eb51a1ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 22:52:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ed355093aca648a6958616649076f9af\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8fe6fc75cd74759b758b8897b0d1d12 : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60167ea724334376b258dfebbd4b70a5 : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d32557826ae44520b3126ec1e4cbc594 : SUCCESS in 6m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7659fe84c1844b278ae6b24fef323d0a : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18c422070b4047a0bde4071e6998f781 : SUCCESS in 5m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b47426283f0f4c50b50534802aa9110f : SUCCESS in 7m 20s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/fdcd34794bd843778ce5574fe2f0ed0d : SUCCESS in 40m 18s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/121cc1870f434fd0bb19b158d20f7677 : FAILURE in 23m 05s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/723f911574594921852725bb53da407f : SUCCESS in 1h 41m 24s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/5b5fc3dddf0e4513be6c3810afc942c1 : SUCCESS in 1h 25m 56s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d8594059bf1e47639ef51b408e057e9d : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":1},{"id":"0f7948611af66588c5aa08e5c866c1319b772d12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-07 15:08:18.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"2965fe61c6df9526e1ee48274887039397077df8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-07 15:08:48.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/trove/+/990157 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"7d672e03ec7f5a48309b995071479e3e2da26de1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-07 15:08:49.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/trove/+/990157 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"44ee6065127943103497742e2a10a08838a0f8a5","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-07 16:48:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"222c05f9e96ae88b28135ba4407f481c866eaa2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-07 18:21:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33443164c82b489ca2b3719648c16e8d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78cfa81f4f2d463b9422188fad9bed1e : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8911666b529f4b70a58e25de3e601757 : SUCCESS in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a019bf8af0a4ae689f42e83f0efb5bd : SUCCESS in 6m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8ddb20d845ed417bacf797cc6c5cfe34 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ff5047519f146098a441a8ac7a13c1c : SUCCESS in 4m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b11c25c42167409a827ec6dec6ad81c1 : SUCCESS in 3m 29s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/b9c9fac8839247169ef5d41581040b02 : SUCCESS in 40m 48s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/ff2fce4e1a0a418ba6e05fc0a31c1e7c : SUCCESS in 37m 53s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/1a7efcd6332d43188e917c1cdb970914 : SUCCESS in 1h 31m 14s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/8cb2a53f5d7b44b7b435fcda3c26349d : SUCCESS in 1h 15m 17s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b6ace4387cee433aa84f9e69e87f2ede : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":2},{"id":"c494cbbb14f1d0709ce65e5f96a9af5616f093e7","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-15 09:24:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4bf3871a3aa080e12b50257154c7f7bb7ec6779b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 09:25:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fc66b7d3f1ebfd2d8525723c86eeb81280fc3950","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 10:48:20.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7dc4bfb802494217b7b9d18fc760b606\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b7bc7a6a0e84c4aa996d4b3cf19e3bd : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/26c697c08a874c16ae00c939ab6e77d6 : SUCCESS in 6m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6dad1f37af7948ceacf03c53bcba540f : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4ecaa6c1887451ba84010d75bfa785e : SUCCESS in 5m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/597b8f3ea77c489a8515075ed7811524 : SUCCESS in 7m 42s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/f990e0f158054668a75c99d02b90261d : POST_FAILURE in 40m 11s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/caa3e6cc80844ffd862d064e64cdc1aa : POST_FAILURE in 38m 16s","accounts_in_message":[],"_revision_number":2},{"id":"42cc3d53b066dc08b94c67ea02a1837df008bfad","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-15 15:57:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"967631774068d9357974a1b7fb0811507cb457a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 17:40:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81aa790354a6460fba9bc8036f7b23fc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dbe7e6c53e5b49539fdf08a3f389d0c3 : SUCCESS in 7m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bc49f32f28d4a0b8dee3a449a5bbf54 : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5b6a6c0e5df4410288c5d26c10c6fbe5 : SUCCESS in 5m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/be2b38eef23941618d94a5dbe96c778d : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53073c952357441297bd8a2bf4dff270 : SUCCESS in 6m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6bdb95f31f27414b8c93b7ffa1dfcf24 : SUCCESS in 8m 00s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/cd54cbfc5bf84024ae36aaa033ee044a : SUCCESS in 45m 02s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/0d74712b1fb449eda425e03cb011dc2f : SUCCESS in 40m 42s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/6c012abaad6b4bd4b0bfb2d06406e06e : SUCCESS in 1h 40m 25s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/7c0338acd19c4837b50aa8a1c46f3671 : SUCCESS in 1h 20m 50s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/78163117d45d4e3a958675b1af27790c : SUCCESS in 6m 50s","accounts_in_message":[],"_revision_number":2},{"id":"f774f55a383f7190a87980373a916ab18357b503","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 17:40:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"16a0d4c18b5c4ba0950468bf158dfdbdf1915bf0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 18:24:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42e208869ed34380be04ad4458894e98\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84f71841458b442eb528f9fb220a2189 : SUCCESS in 5m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d382a1e35a7d4c219cae70f7a40a5294 : SUCCESS in 6m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/95536509588f42b392088a3d0094489b : SUCCESS in 7m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc555c6877294031b9c9003cef520626 : SUCCESS in 5m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64f8ac3f2c97464bbd9644dea1ed57a9 : SUCCESS in 4m 10s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/b24be590234f45e0b56c9ca93f718153 : SUCCESS in 43m 08s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/e26549bb10a04a0fa8a63da0fdbac0c7 : SUCCESS in 40m 41s","accounts_in_message":[],"_revision_number":2},{"id":"ea916d9e5d72fbc8e737c54d0d3f818745c0a863","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 18:24:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"cfb68b639800b8199a06c1b8c96868d9bdc5bcb9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 18:25:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b53c8658b69a403380ae877b473504b7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da7ddab9df42458fa6b1cea7b1c3629a : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"33ebb4f6a7dd58138e7a2801d7132f5e7d116be4","revisions":{"6d5e5cabe229a31a5a8c4df3a60c5182adbb8c6b":{"kind":"REWORK","_number":1,"created":"2026-05-26 18:12:26.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/58/990158/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/58/990158/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/58/990158/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/58/990158/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/58/990158/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/58/990158/1"}}},"commit":{"parents":[{"commit":"5be9a891edce7231a7733116d1d2b908f21010e1","subject":"Merge \"Fix hardcoded WAL archive cleanup threshold in PostgreSQL\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5be9a891edce7231a7733116d1d2b908f21010e1"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-03-18 13:29:35.000000000","tz":300},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-26 18:12:26.000000000","tz":0},"subject":"Allow upgrade to the same version","message":"Allow upgrade to the same version\n\nCurrent manager logic also blocks upgrade to the same version,\npreventing actual execution in tests without pre-upgrade datastores\ndefined in configuration. This change removes that restriction,\nallowing same-version upgrades (e.g. to refresh images), and enables\nTempest to validate the upgrade flow properly.\n\nDepends-On: https://review.opendev.org/c/openstack/trove/+/990157\nChange-Id: If6fd1f65354def972f4a7e568a1763ee486c310a\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n(cherry picked from commit da358b6047890b2ca1ea0deecf76ee6cdcc37639)\n(cherry picked from commit 6b4992b75d4a4eccc2227d48d3c191bef514e1bb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6d5e5cabe229a31a5a8c4df3a60c5182adbb8c6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6d5e5cabe229a31a5a8c4df3a60c5182adbb8c6b"}]},"branch":"refs/heads/stable/2025.1","conflicts":{"ours":"5be9a891edce7231a7733116d1d2b908f21010e1","theirs":"fa57b0051b804b6989af1c5a4f1bb674f8fc17a8","contains_conflicts":false}},"33ebb4f6a7dd58138e7a2801d7132f5e7d116be4":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-06-07 15:08:18.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/58/990158/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/58/990158/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/58/990158/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/58/990158/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/58/990158/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/58/990158/2"}}},"commit":{"parents":[{"commit":"dc313dfbd682e8443fe0fddd3aafa573daef36dc","subject":"Use compatible tempest tag for 2025.1 ubuntu jammy jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dc313dfbd682e8443fe0fddd3aafa573daef36dc"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-03-18 13:29:35.000000000","tz":300},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-07 15:08:18.000000000","tz":0},"subject":"Allow upgrade to the same version","message":"Allow upgrade to the same version\n\nCurrent manager logic also blocks upgrade to the same version,\npreventing actual execution in tests without pre-upgrade datastores\ndefined in configuration. This change removes that restriction,\nallowing same-version upgrades (e.g. to refresh images), and enables\nTempest to validate the upgrade flow properly.\n\nDepends-On: https://review.opendev.org/c/openstack/trove/+/990157\nChange-Id: If6fd1f65354def972f4a7e568a1763ee486c310a\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n(cherry picked from commit da358b6047890b2ca1ea0deecf76ee6cdcc37639)\n(cherry picked from commit 6b4992b75d4a4eccc2227d48d3c191bef514e1bb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/33ebb4f6a7dd58138e7a2801d7132f5e7d116be4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/33ebb4f6a7dd58138e7a2801d7132f5e7d116be4"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase","conflicts":{"base":"5be9a891edce7231a7733116d1d2b908f21010e1","ours":"6d5e5cabe229a31a5a8c4df3a60c5182adbb8c6b","theirs":"dc313dfbd682e8443fe0fddd3aafa573daef36dc","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
