)]}'
{"id":"openstack%2Ftrove~990374","triplet_id":"openstack%2Ftrove~master~Ic1a47bcedd8963bd7aab3c0b112dc0fb814a7cd6","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"36080":{"account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"last_update":"2026-05-28 05:35:44.000000000","reason":"\u003cGERRIT_ACCOUNT_36080\u003e replied on the change","reason_account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},"31737":{"account":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"last_update":"2026-05-28 02:10:03.000000000","reason":"\u003cGERRIT_ACCOUNT_31737\u003e replied on the change","reason_account":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}}},"hashtags":[],"change_id":"Ic1a47bcedd8963bd7aab3c0b112dc0fb814a7cd6","subject":"Fix checking for /dev/kvm device exists","status":"MERGED","created":"2026-05-27 20:46:32.000000000","updated":"2026-05-28 06:29:24.000000000","submitted":"2026-05-28 06:28:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"990374","meta_rev_id":"25fc5ecc5c0973f9362822cb24490a37ea3ebc83","_number":990374,"virtual_id_number":990374,"owner":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-28 06:28:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":2,"date":"2026-05-28 05:35:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":1,"date":"2026-05-28 02:10:03.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":1,"date":"2026-05-28 05:35:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":-1,"date":"2026-05-28 05:35:44.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-27 20:48:41.000000000","updated_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"reviewer":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"state":"REVIEWER"},{"updated":"2026-05-27 21:11:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-27 22:46:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a5b6a0b7d5a9a6ad008b945284b0b2e644883995","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-27 20:46:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"21c891eeca3f1558aba023568e301902a8d3000b","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-27 20:48:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9988f955cb4f617e1135547eba25bb41ef6a4be6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 21:11:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d158090fc69846f0b27de2d86219c323\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f17f5e46847141148b5e32ac3902ebd7 : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c7d3be7459004e3a84644d44f83c6bfd : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4bbd3e19a5834453b0cd7ecba5e1f4cb : SUCCESS in 12m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"52b9ca14f9fb3dd8970c47a6fe44f2e31bac2d28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 22:46:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42649b396f714d42bc29c03783c2e0d3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5af81bbf749841ef81b8639a130c33fe : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d50286edbbf848c993dcad8c085fd99b : SUCCESS in 6m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b97285472f654e99816b97349a994577 : SUCCESS in 5m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9c15af594dd9435994e19fef689aaa22 : SUCCESS in 4m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c31a475b99444791a8ca5e889bc68ca2 : SUCCESS in 11m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/527a4ddae2334db29fe6a7941901c596 : SUCCESS in 4m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1436decd5c5a4a929b6e0e7c52e0356f : SUCCESS in 7m 32s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/878a14d2bdff4a808e2fe4c0a1139ae8 : SUCCESS in 53m 38s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/b5014ab465bf4081bec03513aef8bb52 : SUCCESS in 1h 57m 06s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/ca4a672890f844978efa904bb4bc5b68 : SUCCESS in 1h 38m 14s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/55d22f507bd0466f919ebf2a40be1d30 : SUCCESS in 1h 44m 41s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/371aeadda6e442a1a50d6c3909dfff7a : SUCCESS in 39m 17s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/fe0c731cccf14890bc107e72cb192497 : SUCCESS in 1h 42m 39s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/e8235c32f23b4cf1b3533f769d4b431d : SUCCESS in 1h 23m 23s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/544b5a46a3434ca08b0f9f659cc174bd : SUCCESS in 1h 26m 33s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/9933416d337c4c6c9cee56fd4fe630ef : SUCCESS in 49m 15s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/0be4b063965244a2a559bda3aa03370d : SUCCESS in 1h 30m 14s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/992002cb010e499ca03938f7b6d6f662 : SUCCESS in 1h 14m 01s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/31a76d65d8764eacab7947819e677280 : SUCCESS in 1h 27m 36s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8e30f526c666499694320194ed523966 : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":1},{"id":"a454e15c369324e9c534fb785f2401e4ac4b598b","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2026-05-28 02:10:03.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e92cd6f2cdd8ef5436dd439da8ff8fe11dfbe981","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2026-05-28 02:56:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e533cbdf3bb67a96241701aeab4f5b0a7d58b70c","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-28 05:35:44.000000000","message":"Patch Set 1: Backport-Candidate-1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"358170a31b7b5db632a45efe3085bb4bb2bc492d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 05:36:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3da9b39bced87be4384092d544bd0532a8392d60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 06:28:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/239ef8a0dce44f5c87e9f35a4b46efe4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cf2de31886c4cdb92af95423b52a9cc : SUCCESS in 5m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4a7a4399e2254b518065c196b3d6237b : SUCCESS in 2m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f78f8e6f2ab1475ba2a4ddb72088011e : SUCCESS in 7m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b8fd70fb3074127a7d8d7b6e559334d : SUCCESS in 9m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/04f51bcbe0324d7c85bb408974816f10 : SUCCESS in 3m 18s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/6e81585a9218450ab2ad7f84dca53d94 : SUCCESS in 51m 52s\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/6ba34d36f1bf4159912935baffbdd188 : SUCCESS in 46m 56s\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/b24b47bac8264b57be763c43515acff6 : SUCCESS in 48m 07s","accounts_in_message":[],"_revision_number":1},{"id":"79e3dadf4a6fbee3c1a2a00f598bd105d3a4d1ac","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 06:28:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"25fc5ecc5c0973f9362822cb24490a37ea3ebc83","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 06:29:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1ed8979d2eb40ae87a23290bd6cdf9e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3522588bad0e4d1e86071ef88a2ff150 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b3ab18957a601524dd94e4085f959b16754cefdf","revisions":{"b3ab18957a601524dd94e4085f959b16754cefdf":{"kind":"REWORK","_number":1,"created":"2026-05-27 20:46:32.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/74/990374/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/74/990374/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/74/990374/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/74/990374/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/74/990374/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/74/990374/1"}}},"commit":{"parents":[{"commit":"2b75a7f30d1ef4ea374a342738c089be23946315","subject":"Merge \"replace .values()[0] with .scalar() in _migrate_legacy_database\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2b75a7f30d1ef4ea374a342738c089be23946315"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-27 20:34:44.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-05-27 20:46:00.000000000","tz":300},"subject":"Fix checking for /dev/kvm device exists","message":"Fix checking for /dev/kvm device exists\n\nIn bash, device check should be -c or -e (any type).\nAlso, comment was modified to make it more precise.\n\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\nChange-Id: Ic1a47bcedd8963bd7aab3c0b112dc0fb814a7cd6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b3ab18957a601524dd94e4085f959b16754cefdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b3ab18957a601524dd94e4085f959b16754cefdf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
