)]}'
{"id":"openstack%2Ftrove~991547","triplet_id":"openstack%2Ftrove~master~Ie073c901f9634ef6a3ce7fe44174b008e4e08eb8","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"31737":{"account":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"last_update":"2026-06-04 11:08:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie073c901f9634ef6a3ce7fe44174b008e4e08eb8","subject":"Fix trove-tempest-plugin setup bug in devstack","status":"MERGED","created":"2026-06-04 07:58:38.000000000","updated":"2026-06-04 11:09:52.000000000","submitted":"2026-06-04 11:08:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"991547","meta_rev_id":"a7fa2d06321b7e37636bec382178f0e7c6aa82ce","_number":991547,"virtual_id_number":991547,"owner":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-04 11:08:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":2,"date":"2026-06-04 10:20:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":1,"date":"2026-06-04 10:20:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-04 08:34:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-04 09:44:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-04 10:20:43.000000000","updated_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"reviewer":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"state":"REVIEWER"}],"messages":[{"id":"6b19ffa7d38f8c6568656656f0d1fea99047f385","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2026-06-04 07:58:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a79b1db1423dbfdd80c49c2975427a42fe796900","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 08:34:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2ded11921f24d16b3e345495c1adcb7\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bc1483b17ef44a08933637d760d175a2 : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/07c2f8c4e2fb4d3988cb5c44b91e192e : SUCCESS in 8m 28s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/de39cb2e05c94af484bc10e3b482ec3c : SUCCESS in 16m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0959e818b11800881f552f1488eadc192cd58945","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 09:44:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47569a496d084131a805f0ca271d8247\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d61d32d65c104e0b89a967737a5ccc36 : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b09d6bb5b744edd83f08056354c67ac : SUCCESS in 4m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/307a43fc971e440c9a0e02f5cc4eabeb : SUCCESS in 4m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/538df9b0643d4b6d8e04b236e5a8a85e : SUCCESS in 4m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/574d44ad3ac54a3d8f659b39ef07ee7a : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1e270af4f724b7d80ae8c47f9bd448d : SUCCESS in 5m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87e3b3ac1f3b44b3b3efe207157a38af : SUCCESS in 7m 26s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/5c987691df4845c8bac4282d24342e07 : SUCCESS in 46m 26s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/53158c7bea0b42aea1a5fc3eacb71ab7 : SUCCESS in 1h 41m 45s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/84bc7672690040b09c349b3d74c739af : SUCCESS in 1h 20m 01s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/d14445284c6e4b61ae4f4a22eaaabcb2 : SUCCESS in 1h 38m 08s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/98840610e7bd428f8d32af976fcaab2b : SUCCESS in 43m 29s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/b5c61ddc8cc0424eb9e8d23a243ba682 : SUCCESS in 1h 37m 59s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/5c378f3d23f04a06bc9f8fa540e77bbe : SUCCESS in 1h 21m 53s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/831c29f9015b4992ae3ec0e2ca154ab6 : SUCCESS in 1h 30m 16s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/e83ab61bb700412a9803628986fe75a0 : SUCCESS in 42m 59s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/836d073b7b194ef19fbab9b6072faa1f : SUCCESS in 1h 29m 23s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/f58a1fd53e5f4a0fac33e6ac253a922e : SUCCESS in 1h 09m 37s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/86fa1c096354430daf8bca38d1ea015e : SUCCESS in 1h 06m 38s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/dfdaaa9de27f4c17ab064a4134c80068 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":1},{"id":"49140e9c93eff736510355ee0aceebe20ed508a0","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-04 10:20:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"17df9ff0ad793feac2db70e562a598145f51f7a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 10:21:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"09610320d79db454fe0c9f4c303a8aaf861a1c84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 11:08:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9d74b28700f4887ae999e66c1ea18a6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/705383181a574ae691896187242d3ed8 : SUCCESS in 6m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ea9d49e8e0e14702973dd80f73130d05 : SUCCESS in 6m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8cdf3e1c632d4bcb9f6b7c2c4dcccab1 : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9942039388f845f9947ef96801898778 : SUCCESS in 8m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/caa33fd6da1440a9af32f60e93603ded : SUCCESS in 5m 35s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/1ae1525efae8482e8b98286091bd251b : SUCCESS in 46m 43s\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/44b7e9a6ef604824b17007f6ad17abaf : SUCCESS in 42m 46s\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/52e03c62331c4799b7c39bffbe675bc2 : SUCCESS in 43m 25s","accounts_in_message":[],"_revision_number":1},{"id":"cb1a058d0c1efb55e0b5b4b47822413b52bd76cc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 11:08:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a7fa2d06321b7e37636bec382178f0e7c6aa82ce","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 11:09:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06bc6818e8ad43fcba5c0a37d202b2b3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcee393e8ac84cd6b6086b417a15fefc : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"034b1eecff21d51490ac6c902b189bd64490972c","revisions":{"034b1eecff21d51490ac6c902b189bd64490972c":{"kind":"REWORK","_number":1,"created":"2026-06-04 07:58:38.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/47/991547/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/47/991547/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/991547/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/991547/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/991547/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/47/991547/1"}}},"commit":{"parents":[{"commit":"e816cd850ca3baad6bf3f675d40224a6eeb6dc42","subject":"Preparation for removing eventlet","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e816cd850ca3baad6bf3f675d40224a6eeb6dc42"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2026-06-04 07:15:25.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2026-06-04 07:58:23.000000000","tz":540},"subject":"Fix trove-tempest-plugin setup bug in devstack","message":"Fix trove-tempest-plugin setup bug in devstack\n\nThis PR ensures the plugin is installed into the appropriately\nconfigured virtual environment using devstack\u0027s native utilities.\nThe current implementation directly calls pip using a hardcoded,\npotentially uninitialized path ($TEMPEST_DIR) and a non-existent\n`.tox` directory, causing setup failures.\n\nChange-Id: Ie073c901f9634ef6a3ce7fe44174b008e4e08eb8\nSigned-off-by: Hirotaka Wakabayashi \u003chiwkby@yahoo.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/034b1eecff21d51490ac6c902b189bd64490972c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/034b1eecff21d51490ac6c902b189bd64490972c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
