)]}'
{"id":"openstack%2Ftrove~992593","triplet_id":"openstack%2Ftrove~master~Ieb7b8382769e3abf761c1951ba9d68db5b9207d8","project":"openstack/trove","branch":"master","topic":"remove-trove-mgmt-gateway","hashtags":[],"change_id":"Ieb7b8382769e3abf761c1951ba9d68db5b9207d8","subject":"DNM: remove TROVE_MGMT_GATEWAY","status":"ABANDONED","created":"2026-06-10 08:10:36.000000000","updated":"2026-06-23 07:04:34.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"94210699b2c3a99a1cd521038a65f01b86a08168","_number":992593,"virtual_id_number":992593,"owner":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-06-10 10:19:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-10 08:38:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-10 10:19:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b65a6e665a3418836d5f003130b22a0357441bcd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-10 08:10:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e419631425744cac3d98fcbb0a57cfa9c25b1567","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 08:38:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83c12f3b92a348da845fed978111f87a\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/552c9cfc976b4149bad3f367ad1cc11a : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1130f1c9e961427db2ed4fe87c4039c0 : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c62b66819a7f4cf5be04397d99e2fd54 : SUCCESS in 15m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e34fdd12f9953dfb1ded035b167d06e42b568094","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 10:19:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ff316ac1c0344ffb2d664e89cdd5ce1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7882ae5a64f04b1d893abd5efc8f095a : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ada6d09378a4299abf288b2009ad30f : SUCCESS in 4m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b32c727795704e08a5f20e6f559e32e4 : SUCCESS in 3m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/25d6c6c60b9f4ee4b306bdd481a9fb11 : SUCCESS in 2m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/33610dcb26c04142862b3cea4d67bdc0 : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/174b589be6aa46289c03f3461d76bd27 : SUCCESS in 4m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee439903c4e94bc09e7485b812572bc1 : SUCCESS in 6m 11s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/33bccef518ad4bb586a4e660d7f1344b : SUCCESS in 58m 40s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/dc4914a0a09f45f0a5f1ac0ce41ea46b : SUCCESS in 1h 48m 31s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/c9cde4b64fe74862b45bbf16198d1533 : SUCCESS in 1h 26m 37s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/4554c134d79c41e2a00bdee7f8128631 : SUCCESS in 2h 06m 15s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/9ebf7d30d1a245248b2451d9dc1a69ee : SUCCESS in 49m 29s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/4f06c12dfff84a9c9a0443063c25cbf9 : SUCCESS in 1h 33m 14s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/02cc1e7902c4481eb71709024fded57d : SUCCESS in 1h 14m 45s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/b0a0eb4c1b8949fca13043448f6ee699 : SUCCESS in 1h 45m 56s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/4f4982e197fd4546bf05700637db8427 : SUCCESS in 55m 47s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/f43156275bb448e7a3b43b1257d033b4 : SUCCESS in 1h 32m 43s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/ad664ae147e44516815e6810826390f4 : SUCCESS in 1h 13m 14s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/6cc95b9a1369493987d9c03b25a9d830 : SUCCESS in 1h 42m 27s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c25f310ca68e465c9098a8f53f0d94ea : SUCCESS in 3m 17s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/56ec0920d45b4a34b55f39c401171ff0 : SUCCESS in 8m 05s","accounts_in_message":[],"_revision_number":1},{"id":"94210699b2c3a99a1cd521038a65f01b86a08168","tag":"autogenerated:gerrit:abandon","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-23 07:04:34.000000000","message":"Abandoned\n\nAnother solution proposed in 992900","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"32d9adde4a4a1c90f979b0858be8507cfe2f4546","revisions":{"32d9adde4a4a1c90f979b0858be8507cfe2f4546":{"kind":"REWORK","_number":1,"created":"2026-06-10 08:10:36.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/93/992593/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/93/992593/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/93/992593/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/93/992593/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/93/992593/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/93/992593/1"}}},"commit":{"parents":[{"commit":"b6934150f3a08b76be119e78fc7b7c39612cd086","subject":"Add DNS nameserver to private network during install","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b6934150f3a08b76be119e78fc7b7c39612cd086"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-10 07:49:19.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-10 08:10:04.000000000","tz":300},"subject":"DNM: remove TROVE_MGMT_GATEWAY","message":"DNM: remove TROVE_MGMT_GATEWAY\n\nThis patch removes the TROVE_MGMT_GATEWAY variable. The private network\nremains the primary route for Internet connectivity.\n\nChange-Id: Ieb7b8382769e3abf761c1951ba9d68db5b9207d8\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/32d9adde4a4a1c90f979b0858be8507cfe2f4546"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/32d9adde4a4a1c90f979b0858be8507cfe2f4546"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
