)]}'
{"id":"openstack%2Ftrove~992900","triplet_id":"openstack%2Ftrove~master~I0ed3e5b90e4b5200232a56e412956986e33d6154","project":"openstack/trove","branch":"master","topic":"optimize-zuul-jobs","attention_set":{},"removed_from_attention_set":{"36080":{"account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"last_update":"2026-07-22 05:26:11.000000000","reason":"\u003cGERRIT_ACCOUNT_36080\u003e replied on the change","reason_account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2026-07-22 01:52:24.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"I0ed3e5b90e4b5200232a56e412956986e33d6154","subject":"Avoid multiple default routes in Zuul guest images","status":"MERGED","created":"2026-06-11 13:53:25.000000000","updated":"2026-07-22 07:19:22.000000000","submitted":"2026-07-22 07:18:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"992900-optimize-zuul-jobs","meta_rev_id":"e3cd7030fc72ce62fb4c3c84dab26b72fbb582e1","_number":992900,"virtual_id_number":992900,"owner":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-22 07:18:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":2,"date":"2026-07-22 05:26:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-07-22 01:52:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":1,"date":"2026-07-22 05:26:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-07-22 01:52:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-11 16:32:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-05 15:06:05.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"CC"},{"updated":"2026-07-22 01:52:24.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"4301020353e16f97ca657ea2d1bc52686a612070","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-11 13:53:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"40e20ec3ffe4d9f163289e0fa70a0c25aaadd277","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-11 14:18:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6078cde96e017eddaa720f5143d67e83b00b1afd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 16:32:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9821d30fec154c798aa0cbb879c0b3cb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34ef5d06ba834b13b845ee4875f285d2 : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/969817ba92c7476bbb58d614c7cc7d08 : SUCCESS in 5m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/178af8f1de0f4815a7c2e3e91f1bab1e : SUCCESS in 4m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c02e37cc8f31453dbfb3fb17f79030ca : SUCCESS in 4m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/827729ff8c374e03b154df18bafee7dd : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c44d9fc03704290b0668a0931d47a23 : SUCCESS in 10m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/608fc360d1114fada43a7864051dccb1 : SUCCESS in 7m 19s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/1fcae3c662014de88752b7fce20afa29 : SUCCESS in 56m 12s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/b5b893b4e9e744db8efbc49eaa3df950 : SUCCESS in 1h 48m 54s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/1d146452f67e4686b742450ce4e893b9 : SUCCESS in 1h 32m 06s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/8a547a71b27547bd8d6f001843da6452 : SUCCESS in 2h 12m 07s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/d1b7856c5a9c43029ed24f240ffcec74 : SUCCESS in 55m 43s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/584eaa78466b4513b6cf909fb5591f08 : SUCCESS in 1h 34m 53s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/e397ef6c894443c494933fb502a3d727 : SUCCESS in 1h 25m 46s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/f0b1c9afcc0d40e98aee39eb2fda2e73 : SUCCESS in 1h 41m 15s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/366e3dcad87c4aa0ab0e48287f274e80 : SUCCESS in 1h 01m 34s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/ec3cc75b1c25430fbd4a01f2c0fa5b8b : SUCCESS in 1h 31m 17s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/6c49dcdf4c8a471b920ad6a49f636979 : SUCCESS in 1h 25m 36s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/410e021df2d34fb7af3d9ef89b9233a2 : SUCCESS in 1h 44m 03s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/17d0813ab99240fab84986432f8ab5d6 : SUCCESS in 6m 09s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/70f3aadb2e274830ab021b913f326289 : SUCCESS in 8m 32s","accounts_in_message":[],"_revision_number":2},{"id":"61fae2e72a161cd8459d1366a199d12c3856c0d8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 19:31:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23a6581d410a4021b3b3f2ebf3e924a6\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d7a6af9364bd4df2a1ee3ba69348c732 : SUCCESS in 8m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2d2ce9c83739423e959ed932bb159c12 : SUCCESS in 7m 34s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/05e02b5deafc4f57bbf097454b62efa0 : SUCCESS in 15m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"71868ca2f7c0a337715ef47a2ea36805cddfe6d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-22 15:49:59.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"0eafc949e127f69f63ddf19d2b1064fb4fb93425","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 16:34:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79f460793a8f427eb9abdb55429035c9\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/551f489fd8a649ee8330280d6b1839f7 : SUCCESS in 9m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f0bdc59f4b674b11a923cfdc2c2a4a74 : SUCCESS in 8m 14s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/113b0c12ee4a496fb87ccb91521917fc : SUCCESS in 8m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d306b5b3b738496df25c6ef61bb9aa87fd19b6e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 18:58:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4aa3ecd9a0c54510907b61ed8059b400\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3d8290a339f4836b464327991b6fc79 : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df57da76275e4d41a111a280ca5c92bd : SUCCESS in 6m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7465d691bb8d4b2cbb069f37b747373c : SUCCESS in 4m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ab397359c1b34448a8bfc7f18fc1adeb : SUCCESS in 4m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7c956841341a49fcbbf72fdb563cd35d : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66509aaaf1484c0fb4e569313aef6a11 : SUCCESS in 5m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/119dd4cabeb044ffbc04eecdb6df87ef : SUCCESS in 6m 40s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/988f629bfe5f4419b9d67ffdb23a0e81 : SUCCESS in 47m 34s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/09abd891689949dbb76d706908f772ab : SUCCESS in 1h 38m 57s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/b508a8974dd9492b972ef7faa061faa2 : FAILURE in 1h 29m 58s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/29a9dd40a5a8448081ca43c0ad602a8f : SUCCESS in 1h 58m 51s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/09ac11fd0d594e21913d1cd10a7f4267 : SUCCESS in 57m 28s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/328c1e3933fc4ca88799579a369e9d53 : SUCCESS in 2h 21m 47s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/8675641044ca499389a2193d7021b50b : SUCCESS in 1h 12m 37s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/6f27f35d75864214a4b452c311b554c0 : TIMED_OUT in 3h 05m 29s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/333a5e8e6fe049009b64e5ad9a3b6d5e : SUCCESS in 1h 29m 40s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/08097827a9234607a44e8b8d13521c13 : SUCCESS in 2h 54m 07s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/d1fd285ea55d4ede98f76ba956a44321 : SUCCESS in 1h 51m 58s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/573ec0887eac4ba8ae527fc4c92f89c1 : SUCCESS in 2h 30m 58s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c3d1313933634f86ac9cd3ea1bbe48c2 : SUCCESS in 5m 10s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/4c64b0bb680f4aaf97b678f724ead74f : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":3},{"id":"11ac08a507d74794bf3505af9fba96b0bff6e013","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-07-05 15:06:05.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"4e0c0914962b9c8dafa46bd5da0e0d7b34eab203","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-08 06:32:07.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e22dee0ec094e40781530b66fc621d96c2b2b44d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-08 09:14:52.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"1445941441dd9f5dea64531a4acf8ca5e995ac12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-08 09:18:36.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"005dd81268c535c60e4605d79cbb94083f9286df","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-08 09:19:34.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"812e23763fa8d1369b6a3295643ae161b1ac9f42","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 09:48:36.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78ed94441c54472a937141989009f5bb\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ad793c8314a1428fa3159169e5b3668a : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9cb4acdec6a94943a4613c60d7fb6d4f : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d92b1cc4db34408187facf81e59027f7 : SUCCESS in 12m 36s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b9bde93c916447f247c3687933e0a9ff12e017f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 11:34:17.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b038ed90634a44f39cd3be3f60048586\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a337d0d1b7794081aedd59ff2e945db7 : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8757c215bb5049c7ba7bbc158119c4de : SUCCESS in 5m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4bb051e7650c4657acaf286cf25bd3bb : SUCCESS in 4m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ba6702153ae1498aa8c64b946056c510 : SUCCESS in 6m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f2d700be82314e2fb36a1b34bdcff2fc : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd8bfb12423143278480f0f4046d7f3c : SUCCESS in 10m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53df6c9ee01d481eb607ed4430a7b2be : SUCCESS in 7m 39s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/f80f69c29a0143b4a286a2f45760c8a6 : SUCCESS in 51m 23s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/135acf7c7c354e9c8374384ec20363fa : SUCCESS in 1h 52m 31s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/bb86b2964c4243f58e149bef0362ce23 : SUCCESS in 1h 29m 04s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/61b21f4ca78c4ddab0f6211a78f12c66 : SUCCESS in 2h 08m 41s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/c40e7d2d6fdb46b5a1b65d4c16732d3c : SUCCESS in 47m 12s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/eecade1b89414060ab73bdf15c93e62e : SUCCESS in 1h 34m 56s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/ed02636c2d814913829eda5a223cc0ed : SUCCESS in 1h 22m 43s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/fd742a0d2f324bbb971c1c94b790f16e : SUCCESS in 1h 49m 56s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/2e1676f50c7b45099bdbf73323d24886 : SUCCESS in 52m 26s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/5b7a6bfe42c84481b480fc4752f7bdad : SUCCESS in 1h 35m 04s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/2123e5970742404b9866c2c73646d804 : SUCCESS in 1h 22m 07s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/513e61e921c340c283c6c70709cd4e55 : SUCCESS in 1h 50m 53s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6d2b80d4f925439cada77fd83c33dc31 : SUCCESS in 5m 15s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/0158471d6e8f439892335c24660abf25 : SUCCESS in 8m 37s","accounts_in_message":[],"_revision_number":5},{"id":"91449c53ac72b37dc9d177e90c49ed590dcc5ca0","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-07-08 14:09:20.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"1f0f0e5b04c21f8b751363d3a6cd3b8bd7386c46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-09 15:56:29.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"fcaa1c8f9996aa0bf8f010e83f5cac7663080121","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-09 15:57:40.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"566c9e28238549c161e7bf7aa8dc4dccd9f0590b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 16:40:07.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5e3306fa3184a8e92f4127cbaa219f9\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6444fd527e5048019dfb15ca9311f748 : SUCCESS in 7m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/98c44b716b7542e28e0ddd373c526c93 : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/515e93443842495d8068ee5b622a3665 : SUCCESS in 12m 35s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"89d326e11cef3ca5c7049fb06f01434d752d7497","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 18:07:52.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e96713aa11548edb85c214c03a1cc59\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c74df124c914cadb35c930c34c4a16e : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d944b1298a4c4c29b8718d3c3454b3e3 : SUCCESS in 5m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/966867903e684514bf81bacd2016b1ab : SUCCESS in 4m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dec881f15edd4dc8bb8316a9cdc9cf96 : SUCCESS in 4m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/135f769400a64aa386250b6672dea221 : SUCCESS in 7m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f491b6b82dfa4c229dacc3c86df808ee : SUCCESS in 9m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13b84923d56d4a2eace21903607839df : SUCCESS in 7m 04s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/c6fde5f45d5e4da69418e6f74eb014ca : SUCCESS in 53m 25s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/20795b7b189a459b9e7bfe245e2f4651 : SUCCESS in 2h 04m 08s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/184d32bada8140659fa6a8597a66483c : SUCCESS in 1h 49m 16s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/1fa01cf9e14b4bdd8b210c64f7bdc4bc : SUCCESS in 2h 04m 58s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/c024ed88729d4f01abb1e548af436451 : SUCCESS in 1h 09m 23s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/eaf287ef31f3456695ca485d5c5c5c17 : SUCCESS in 2h 09m 47s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/f498b4fe23574d2393c8771aa061e52e : SUCCESS in 1h 40m 58s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/2b973a1ef9784abfb29aca6bbfeb1695 : SUCCESS in 1h 37m 48s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/665c183fb5924fcc83e87675f71503e4 : SUCCESS in 1h 26m 23s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/dde08568227844258fdeae2b2e47a908 : SUCCESS in 1h 55m 42s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/a7d25067273d40f99a89af7f92f63c6c : SUCCESS in 1h 14m 20s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/4aafa722a5ea404aabcd1fb1a99b562c : SUCCESS in 1h 37m 05s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/40db90768e47493aac08fecc516023b9 : SUCCESS in 5m 11s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/87dc649cc45b4ba5a396770c370670e1 : SUCCESS in 11m 47s","accounts_in_message":[],"_revision_number":6},{"id":"c0ba7778d1f085a519a67141c705ae42a190da06","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-09 18:58:58.000000000","message":"Topic optimize-zuul-jobs/disable-client-network-on-startup removed","accounts_in_message":[],"_revision_number":6},{"id":"7c6a9d380fd4cd4f898c6be84a42dda6ed7c6011","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-09 18:59:19.000000000","message":"Topic set to optimize-zuul-jobs","accounts_in_message":[],"_revision_number":6},{"id":"9975baa68274cbb96c42b6e13a3693f4a1f741b9","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-10 12:49:48.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ed3fdcf089ef3cb3cdcbb90ae22587b9cf6d03f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-10 13:25:56.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"94b51f6210f59b8367771e6f767f83530a4f8c64","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 13:48:40.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b75666ad61343919395249c6e2f3f4e\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/73764640e1cf4d89bbe431343eeea622 : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/adb07f36285b4b60b2663fe0a03c989d : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7770fd24e8f64a9699e3d5a9f8d804e0 : SUCCESS in 15m 00s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"4d388c2bce65ddfe3a525fd6b5733650b8288f6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 15:47:55.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/348b6ad3b14249d792f0129cd2afd00e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/56818fdea10c4e4a8614e2cae1603d0d : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccc352e69d6941489d9a4ee0828de9c0 : SUCCESS in 5m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f6928c7c134141fa8e67340c9f9dac0d : SUCCESS in 4m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4f545709660c40958ab13793e515b1b5 : SUCCESS in 4m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7cd67583287c4b1eaccff263a13f3025 : SUCCESS in 8m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff429a8a980d460cb6d6bdf308b23ae6 : SUCCESS in 11m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1dce55b9e5ef41e89fef26a2ef87b915 : SUCCESS in 10m 16s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/a34146139d4d4b88afee1037773e8c51 : SUCCESS in 53m 06s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/f0fe285850a943d0b9581cfca0ce366e : SUCCESS in 2h 03m 24s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/0bf4c3af9c4d4d7b8333a3a60a4b6c05 : SUCCESS in 1h 40m 54s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/e3070b8415c44f5d93271bb5746a8606 : SUCCESS in 2h 17m 33s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/c0c94a0ee2194af5831ed929879481e5 : SUCCESS in 58m 24s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/0882c698ccc942caaf24dcedff3c6d40 : SUCCESS in 1h 32m 51s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/00ad71af66574adc96224c6f5f0c5576 : SUCCESS in 1h 13m 49s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/12eed8287ddd4f79a5828a3ea1fbf4f1 : SUCCESS in 1h 58m 17s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/d9510642e2cd420ba073645aaf31677a : SUCCESS in 48m 33s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/4e418317fde04313a338d57b829af6e5 : SUCCESS in 1h 34m 43s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/a4e89171acf0467487b8f853aecd13cd : SUCCESS in 1h 07m 23s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/95ea272c259942b6b3fa6017dd0ecc10 : SUCCESS in 2h 06m 10s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a90e2ea919104eef8d3cc85298d7678e : SUCCESS in 7m 37s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/dee6995d2bf74cccb18dd19caaadd065 : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":7},{"id":"26242a8b355f3ab7f9c132c2233800c6b769d436","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-07-12 13:23:58.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"c44d2e107f01763f1fc0cdeb46dfe37ea62dfb36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-12 13:25:45.000000000","message":"Patch Set 8: Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"12b7d33856de4d7b25a5553a9462b47e4a5a9c19","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-12 13:43:05.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68cdd8553b504301a2670def96719aec\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b1d6563be7a8464d8bd501dcbaf28579 : SUCCESS in 7m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b170a428d83e4aaeb7f1fb8deaabc9d7 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e04f10d772b44b8ea06529a9189484a1 : SUCCESS in 14m 51s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"1459ae1e3ae421c022580fab40c74817d86e7cb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-12 16:37:03.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/18fc32bc283f4805960728f37bc712aa\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d971e471b9b64c0b8bf341c34ef6330c : SUCCESS in 10m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b83d10ca87a44d7199949e2e6833ccae : SUCCESS in 4m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3728923182734db68cae912234bf0447 : SUCCESS in 3m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e46f3cb4b02046878568082f03e93826 : POST_FAILURE in 3m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0109e3300873423b905af59b081cb201 : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b898059cde3c419bb70ec9956a285d17 : SUCCESS in 8m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70a44cac2d0c4a46ba2c1d5a19c4d4fa : SUCCESS in 9m 11s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/192572091502416d9a05ef2502949ef3 : SUCCESS in 1h 03m 36s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/bd2d5f95e8c346a88ddd38bccfe3d6e3 : FAILURE in 1h 41m 55s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/e4f28cf439b942a1a0d82bff435df516 : SUCCESS in 2h 05m 58s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/de209b2a5dba441faa27d0d5e597e6d5 : TIMED_OUT in 3h 04m 14s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/2e8add06afa641d8a9586b938d485241 : SUCCESS in 1h 04m 08s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/81d270e5cbc744d88965d1efcf78b880 : SUCCESS in 1h 32m 28s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/39fb00dac29149b8bdb26f1a47751751 : SUCCESS in 1h 49m 36s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/273baacd8b274d818b889b968367b846 : SUCCESS in 1h 37m 01s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/0d276a34f2884f5487d7a1ec732fb8bc : SUCCESS in 1h 07m 27s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/900c7b0810c843baa6665b428b25be03 : SUCCESS in 1h 41m 09s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/1634966c17f44ee7af59464f9bf763be : FAILURE in 1h 27m 16s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/6a5f28c56f6d4f9dae23dbeb62953597 : SUCCESS in 1h 58m 48s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7d52e3f555f646b1a176a9f328a9e3dd : SUCCESS in 5m 04s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/ece42f3088ec4290bcc5b8b73a2628c2 : SUCCESS in 11m 49s","accounts_in_message":[],"_revision_number":8},{"id":"4910e7254c92b93cd1cf322573e4f7e4bb4fde90","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-14 04:41:59.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"bb2e590296d07bf8691a4ea29092afbf63ada5e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 05:10:11.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/340e24ef478f4a6eae2acadff0383f78\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b45d6a8501b04da0884839b7c8bb4b30 : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b227d012d8da46f598bcf8607090468c : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0957535e0d7d4ae29b4361354c93c7d3 : SUCCESS in 14m 42s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"2c2524abfeff761d4db890f6e7341dd184cc1396","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 06:57:02.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/864490ea869546e9a59fff4d7ddc9dbe\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de441105ec64416cbd5b34f70ac2327b : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0eff6e2be590443394b00845d6874445 : SUCCESS in 3m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/01b28ca18d68492d8131ff43db15bf5f : SUCCESS in 4m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/adf965b4fe4c4ac49b6cae6aa10d9b35 : SUCCESS in 6m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/157b24c380db4c46bf8f5b23301f5492 : SUCCESS in 7m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a243f5aaebd3405da874cf00c0cd15e0 : SUCCESS in 9m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/646eaeb765ea46499ace0cf93d36d7f6 : SUCCESS in 7m 27s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/5479be4054e041c888a61219ae77e68c : SUCCESS in 1h 06m 50s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/a96c6f3182744d6d8f64e043f2e0948b : SUCCESS in 2h 03m 45s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/66c778e4ee274c4b8f12294e607d8d02 : SUCCESS in 1h 28m 57s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/7fea965969e14f8abf34a045a4bc9c91 : SUCCESS in 2h 12m 33s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/83eceb34765e4125acdaaeddb5cc7e79 : SUCCESS in 46m 13s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/ffce3f59603a41df9a0ab72a345bd946 : SUCCESS in 1h 51m 36s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/77fcc021c6f041e79875c9aacd69d182 : SUCCESS in 1h 21m 46s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/7c8c5bf650754ac18ccdd3e3466c3251 : SUCCESS in 1h 40m 05s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/464752ade9f64589b7005efed17eb387 : SUCCESS in 1h 03m 23s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/bcf52100b58f416c837888bb4b16fdd4 : SUCCESS in 1h 44m 25s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/dc29e7c7034c4673ba96d57ea54549b6 : SUCCESS in 1h 33m 43s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/42c4d393fa7546478fb78fbb23a175ec : SUCCESS in 1h 34m 51s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a6e77aa0f7b542109b2c294671a032a9 : SUCCESS in 5m 11s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/529e306adff74fb7b62afb5dc8472a13 : SUCCESS in 11m 21s","accounts_in_message":[],"_revision_number":8},{"id":"3beaa0eb26767b47feff997ef6eec1d1235347c2","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-14 11:01:28.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"2cff014dcd26e786f7dd61e8cbce655ee806396d","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-07-19 07:58:43.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"04a553abe811b66f7e864cc2b39e44f12ae456e6","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-20 08:40:17.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"893a324f53a7760c92c460601d3857e481875dda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-20 09:33:35.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"342274ad31c58549839ea11e5bc05a614c141c82","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 10:23:37.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d7a99189f774ec1a92e0aba1bc18d01\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0604577cbf8847dd894ddf606ea05d13 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/572b26408b0f4a458b3153f20cfeba02 : SUCCESS in 8m 30s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2ef1d3fedeac4de98b2eb2ed4bae3db5 : SUCCESS in 14m 37s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"90741c31d31d8c744dcc3d7b6323e48a3b7dc635","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 12:40:39.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1d58130b4914ed9b5084f8b911da8af\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c8a60c21f0b245018dfab6f9a1296627 : SUCCESS in 7m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/721c2412047641a0b38130135d980a14 : SUCCESS in 5m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/45e60ccc7e12477ea971430fba96a7f8 : SUCCESS in 5m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5b9984b3106f445f8ad07771edf6a1a2 : SUCCESS in 6m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/611f5c8fae1b45e0b808fdcd81d52b6b : SUCCESS in 9m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21aa93ab12274a1cbf3c36270f3283d6 : SUCCESS in 10m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ea901b774b341298082550746666590 : SUCCESS in 7m 18s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/be154037c8a041fe99026ea33efcd7dc : SUCCESS in 1h 23m 20s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/5defe64733e74d9d931778989203b5d5 : SUCCESS in 1h 51m 47s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/a2f9c628d4cd4d00aada04e5f3e4bad2 : FAILURE in 1h 34m 57s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/62ab33dab1764ccfb11bb25947fa0628 : TIMED_OUT in 3h 04m 34s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/804e858e561344c5809d7c189e490d42 : SUCCESS in 45m 05s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/9ca454cb0c9945b7afced7ca56699eee : SUCCESS in 1h 48m 35s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/690b3ddd72e043b1b8b80d282f5db23f : SUCCESS in 2h 15m 42s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/8de51642d35c4cbb8bb3d34cdae3e63e : SUCCESS in 2h 42m 21s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/06df681f61214fbcb8401c9b187c88e1 : SUCCESS in 1h 00m 18s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/516ff463b3ef4bf2b1654786365db68e : FAILURE in 1h 30m 50s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/2b0c159624eb44b4ba4f645fa159dd09 : SUCCESS in 1h 43m 27s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/a410a528435849f084b388b6681aca9f : SUCCESS in 2h 32m 19s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/79ec568c3cb548059cf2ba34030e000a : SUCCESS in 7m 10s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/ed6a541bd4b642f19cb625dd3df7074d : SUCCESS in 8m 38s","accounts_in_message":[],"_revision_number":9},{"id":"54f6219cd6b6f2b104a23799bed883a33aaa42eb","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-07-22 01:52:24.000000000","message":"Patch Set 9: Backport-Candidate+1 Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"f04a0572b1d639666c2a8fc673801e36549600a4","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-22 05:26:11.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"703912d097f3c0625c97ce51c80d9ceb93410672","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 05:26:39.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"72f09dd35f69f5431f1b9d9cb28f71647a4f3718","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 07:18:26.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/406a3fc07be946fe833d212f15ff2fec\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c99ac108ecbe46d79bc36c2db8f4bfb6 : SUCCESS in 2m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7763e8ecc5f54b34a93830dc425f354d : SUCCESS in 3m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5de36ff0cd40406088c68ce9582d47e5 : SUCCESS in 2m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0fe1dd522d648d6bb041fe40731f6db : SUCCESS in 8m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb6ffcf58b5b43308a9183e8197879b1 : SUCCESS in 5m 46s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/1ecd73a161ee448083a1a64e92471167 : SUCCESS in 1h 51m 15s\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/97d14fc205e34c94b54271ea10c8a7cd : SUCCESS in 1h 17m 34s\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/4a4ff8134c8b4029aeb360493bade7fa : SUCCESS in 1h 19m 40s","accounts_in_message":[],"_revision_number":9},{"id":"00c4e302e818bd2d2cd907b1bc0ba257b9d35d07","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 07:18:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"e3cd7030fc72ce62fb4c3c84dab26b72fbb582e1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 07:19:22.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/941c3c2038384f68acf7f9d31426ce29\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3a83714331848f9b5637d9a1ffa9b36 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"02efba18ef10ac09aae3438b05cc1ddbdc431054","revisions":{"337a8025d820e40d11f90305520dc0da0b3bbbdf":{"kind":"REWORK","_number":1,"created":"2026-06-11 13:53:25.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/00/992900/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/00/992900/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/00/992900/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/00/992900/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/00/992900/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/00/992900/1"}}},"commit":{"parents":[{"commit":"187af7eb8e32f58bcd4b2b7825dbd3e3272b2a86","subject":"Merge \"Use common utility to clone/install source\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/187af7eb8e32f58bcd4b2b7825dbd3e3272b2a86"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-11 13:17:40.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-11 13:17:40.000000000","tz":300},"subject":"Avoid multiple default routes in Zuul guest images","message":"Avoid multiple default routes in Zuul guest images\n\nDEV and CI guest instances may be configured with multiple default\nroutes, leading to intermittent connectivity failures and unreliable\ntest results.\n\nRun a helper script before trove-image-loader starts to ensure that\nthe trove-mgmt network remains the only active default route when\nnetwork isolation is disabled.\n\nThis change applies only to DEV/zuul guest images and does not\naffect production deployments.\n\nChange-Id: I0ed3e5b90e4b5200232a56e412956986e33d6154\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/337a8025d820e40d11f90305520dc0da0b3bbbdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/337a8025d820e40d11f90305520dc0da0b3bbbdf"}]},"branch":"refs/heads/master"},"addb00b9149982563764129efd1af03ebc4cd0eb":{"kind":"REWORK","_number":2,"created":"2026-06-11 14:18:20.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/00/992900/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/00/992900/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/00/992900/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/00/992900/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/00/992900/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/00/992900/2"}}},"commit":{"parents":[{"commit":"187af7eb8e32f58bcd4b2b7825dbd3e3272b2a86","subject":"Merge \"Use common utility to clone/install source\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/187af7eb8e32f58bcd4b2b7825dbd3e3272b2a86"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-11 13:17:40.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-11 14:17:55.000000000","tz":300},"subject":"Avoid multiple default routes in Zuul guest images","message":"Avoid multiple default routes in Zuul guest images\n\nDEV and CI guest instances may be configured with multiple default\nroutes, leading to intermittent connectivity failures and unreliable\ntest results.\n\nRun a helper script before trove-image-loader starts to ensure that\nthe trove-mgmt network remains the only active default route when\nnetwork isolation is disabled.\n\nThis change applies only to DEV/zuul guest images and does not\naffect production deployments.\n\nChange-Id: I0ed3e5b90e4b5200232a56e412956986e33d6154\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/addb00b9149982563764129efd1af03ebc4cd0eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/addb00b9149982563764129efd1af03ebc4cd0eb"}]},"branch":"refs/heads/master"},"5c30def872632e122917a852c22e8a6b47485a5f":{"kind":"REWORK","_number":3,"created":"2026-06-22 15:49:59.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/00/992900/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/00/992900/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/00/992900/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/00/992900/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/00/992900/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/00/992900/3"}}},"commit":{"parents":[{"commit":"09eb40506371fcdd1864bdea4f4e1c05d01151e7","subject":"Merge \"Skip backup image build in non-backup jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/09eb40506371fcdd1864bdea4f4e1c05d01151e7"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-11 13:17:40.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-22 15:49:04.000000000","tz":300},"subject":"Avoid multiple default routes in Zuul guest images","message":"Avoid multiple default routes in Zuul guest images\n\nDEV and CI guest instances may be configured with multiple default\nroutes, leading to intermittent connectivity failures and unreliable\ntest results.\n\nRun a helper script before trove-image-loader starts to ensure that\nthe trove-mgmt network remains the only active default route when\nnetwork isolation is disabled.\n\nThis change applies only to DEV/zuul guest images and does not\naffect production deployments.\n\nBlueprint: zuul-jobs-optimiztion-2026\nChange-Id: I0ed3e5b90e4b5200232a56e412956986e33d6154\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5c30def872632e122917a852c22e8a6b47485a5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5c30def872632e122917a852c22e8a6b47485a5f"}]},"branch":"refs/heads/master"},"c2c605e5991da0f2d4a5892405d61e57b346c735":{"kind":"REWORK","_number":4,"created":"2026-07-08 09:14:52.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/00/992900/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/00/992900/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/00/992900/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/00/992900/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/00/992900/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/00/992900/4"}}},"commit":{"parents":[{"commit":"e9489ce73e2415ed277eb51ab45394f9247acc30","subject":"Merge \"Improve data volume response in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e9489ce73e2415ed277eb51ab45394f9247acc30"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-11 13:17:40.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-07-08 09:12:56.000000000","tz":300},"subject":"Avoid multiple default routes in Zuul guest images","message":"Avoid multiple default routes in Zuul guest images\n\nDEV and CI guest instances may be configured with multiple default\nroutes, leading to intermittent connectivity failures and unreliable\ntest results.\n\nRun a helper script before trove-image-loader starts to ensure that\nthe trove-mgmt network will be prioritized (set metric\u003d50) if the\nnetwork isolation is enabled and TROVE_MGMT_GATEWAY is set.\n\nThis change applies only to DEV/zuul guest images and does not\naffect production deployments.\n\nBlueprint: zuul-jobs-optimiztion-2026\nDepends-On: https://review.opendev.org/c/openstack/trove/+/994159\nChange-Id: I0ed3e5b90e4b5200232a56e412956986e33d6154\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c2c605e5991da0f2d4a5892405d61e57b346c735"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c2c605e5991da0f2d4a5892405d61e57b346c735"}]},"branch":"refs/heads/master"},"61f737a6f1f1681ddef041fa1e282a34e3ad0187":{"kind":"REWORK","_number":5,"created":"2026-07-08 09:18:36.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/00/992900/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/00/992900/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/00/992900/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/00/992900/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/00/992900/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/00/992900/5"}}},"commit":{"parents":[{"commit":"e9489ce73e2415ed277eb51ab45394f9247acc30","subject":"Merge \"Improve data volume response in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e9489ce73e2415ed277eb51ab45394f9247acc30"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-11 13:17:40.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-07-08 09:18:14.000000000","tz":300},"subject":"Avoid multiple default routes in Zuul guest images","message":"Avoid multiple default routes in Zuul guest images\n\nDEV and CI guest instances may be configured with multiple default\nroutes, leading to intermittent connectivity failures and unreliable\ntest results.\n\nRun a helper script before trove-image-loader starts to ensure that\nthe trove-mgmt network will be prioritized (set metric\u003d50) if the\nnetwork isolation is enabled and TROVE_MGMT_GATEWAY is set.\n\nThis change applies only to DEV/zuul guest images and does not\naffect production deployments.\n\nBlueprint: zuul-jobs-optimiztion-2026\nDepends-On: https://review.opendev.org/c/openstack/trove/+/994159\nChange-Id: I0ed3e5b90e4b5200232a56e412956986e33d6154\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/61f737a6f1f1681ddef041fa1e282a34e3ad0187"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/61f737a6f1f1681ddef041fa1e282a34e3ad0187"}]},"branch":"refs/heads/master"},"7bdd579c79508966bdbb8ed5b19ec9c971e5d033":{"kind":"REWORK","_number":6,"created":"2026-07-09 15:56:29.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/00/992900/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/00/992900/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/00/992900/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/00/992900/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/00/992900/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/00/992900/6"}}},"commit":{"parents":[{"commit":"e9489ce73e2415ed277eb51ab45394f9247acc30","subject":"Merge \"Improve data volume response in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e9489ce73e2415ed277eb51ab45394f9247acc30"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-11 13:17:40.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-07-09 15:56:10.000000000","tz":300},"subject":"Avoid multiple default routes in Zuul guest images","message":"Avoid multiple default routes in Zuul guest images\n\nDEV and CI guest instances may be configured with multiple default\nroutes, leading to intermittent connectivity failures and unreliable\ntest results.\n\nRun a helper script before trove-image-loader starts to ensure that\nthe trove-mgmt network will be prioritized (set metric\u003d50) if the\nnetwork isolation is enabled and TROVE_MGMT_GATEWAY is set.\n\nThis change applies only to DEV/zuul guest images and does not\naffect production deployments.\n\nBlueprint: zuul-jobs-optimiztion-2026\nDepends-On: https://review.opendev.org/c/openstack/trove/+/994159\nChange-Id: I0ed3e5b90e4b5200232a56e412956986e33d6154\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7bdd579c79508966bdbb8ed5b19ec9c971e5d033"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7bdd579c79508966bdbb8ed5b19ec9c971e5d033"}]},"branch":"refs/heads/master"},"36ddaaefef787a6fee70e304ba1fa06cdbaeaa24":{"kind":"REWORK","_number":7,"created":"2026-07-10 13:25:56.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/00/992900/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/00/992900/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/00/992900/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/00/992900/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/00/992900/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/00/992900/7"}}},"commit":{"parents":[{"commit":"e9489ce73e2415ed277eb51ab45394f9247acc30","subject":"Merge \"Improve data volume response in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e9489ce73e2415ed277eb51ab45394f9247acc30"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-11 13:17:40.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-07-10 13:25:30.000000000","tz":300},"subject":"Avoid multiple default routes in Zuul guest images","message":"Avoid multiple default routes in Zuul guest images\n\nDEV and CI guest instances may be configured with multiple default\nroutes, leading to intermittent connectivity failures and unreliable\ntest results.\n\nRun a helper script before trove-image-loader starts to ensure that\nthe trove-mgmt network will be prioritized (set metric\u003d50) if the\nnetwork isolation is enabled and TROVE_MGMT_GATEWAY is set.\n\nThis change applies only to DEV/zuul guest images and does not\naffect production deployments.\n\nBlueprint: zuul-jobs-optimiztion-2026\nDepends-On: https://review.opendev.org/c/openstack/trove/+/994159\nChange-Id: I0ed3e5b90e4b5200232a56e412956986e33d6154\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/36ddaaefef787a6fee70e304ba1fa06cdbaeaa24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/36ddaaefef787a6fee70e304ba1fa06cdbaeaa24"}]},"branch":"refs/heads/master"},"968f29bd0dfc2eaf7f50b186755d3a845a8fd939":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2026-07-12 13:25:45.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/00/992900/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/00/992900/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/00/992900/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/00/992900/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/00/992900/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/00/992900/8"}}},"commit":{"parents":[{"commit":"e9489ce73e2415ed277eb51ab45394f9247acc30","subject":"Merge \"Improve data volume response in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e9489ce73e2415ed277eb51ab45394f9247acc30"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-11 13:17:40.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-07-12 13:25:45.000000000","tz":0},"subject":"Avoid multiple default routes in Zuul guest images","message":"Avoid multiple default routes in Zuul guest images\n\nDEV and CI guest instances may be configured with multiple default\nroutes, leading to intermittent connectivity failures and unreliable\ntest results.\n\nRun a helper script before trove-image-loader starts to ensure that\nthe trove-mgmt network will be the only default route during startup\nif the network isolation is enabled and TROVE_MGMT_GATEWAY is set.\n\nThis change applies only to DEV/zuul guest images and does not\naffect production deployments.\n\nBlueprint: zuul-jobs-optimiztion-2026\nDepends-On: https://review.opendev.org/c/openstack/trove/+/994159\nChange-Id: I0ed3e5b90e4b5200232a56e412956986e33d6154\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/968f29bd0dfc2eaf7f50b186755d3a845a8fd939"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/968f29bd0dfc2eaf7f50b186755d3a845a8fd939"}]},"branch":"refs/heads/master","description":"Edit commit message"},"02efba18ef10ac09aae3438b05cc1ddbdc431054":{"kind":"REWORK","_number":9,"created":"2026-07-20 09:33:35.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/00/992900/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/00/992900/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/00/992900/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/00/992900/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/00/992900/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/00/992900/9"}}},"commit":{"parents":[{"commit":"72cf40aa93ea241f312f0759af8d79e03272d93d","subject":"Merge \"Enable userdata refresh on instance rebuild (Nova ≥ 2.57)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/72cf40aa93ea241f312f0759af8d79e03272d93d"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-11 13:17:40.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-07-20 09:31:55.000000000","tz":300},"subject":"Avoid multiple default routes in Zuul guest images","message":"Avoid multiple default routes in Zuul guest images\n\nDEV and CI guest instances may be configured with multiple default\nroutes, leading to intermittent connectivity failures and unreliable\ntest results.\n\nRun a helper script before trove-image-loader starts to ensure that\nthe trove-mgmt network will be the only default route during startup\nif the network isolation is enabled and TROVE_MGMT_GATEWAY is set.\n\nThis change applies only to DEV/zuul guest images and does not\naffect production deployments.\n\nBlueprint: zuul-jobs-optimiztion-2026\nChange-Id: I0ed3e5b90e4b5200232a56e412956986e33d6154\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/02efba18ef10ac09aae3438b05cc1ddbdc431054"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/02efba18ef10ac09aae3438b05cc1ddbdc431054"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
