)]}'
{"id":"openstack%2Ftrove~994177","triplet_id":"openstack%2Ftrove~master~I4bf288234765a8b85edac3d79fe3c7d270730a5a","project":"openstack/trove","branch":"master","topic":"mysql-fix-get-backup-image-arg","hashtags":[],"change_id":"I4bf288234765a8b85edac3d79fe3c7d270730a5a","subject":"Fix mysql get_backup_image arguments","status":"ABANDONED","created":"2026-06-20 13:25:41.000000000","updated":"2026-08-07 09:34:43.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8a610ab1fe8528b9cc28145dde455588b8a68429","_number":994177,"virtual_id_number":994177,"owner":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-06-20 15:18:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":5733,"name":"Kieran Spear","email":"kispear@gmail.com","username":"kspear"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-20 13:43:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-20 15:18:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-07 07:18:10.000000000","updated_by":{"_account_id":5733,"name":"Kieran Spear","email":"kispear@gmail.com","username":"kspear"},"reviewer":{"_account_id":5733,"name":"Kieran Spear","email":"kispear@gmail.com","username":"kspear"},"state":"CC"}],"messages":[{"id":"4f73e5957b2db5513cc64a7d19117cc77f171f38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-20 13:25:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7d03cfcfb7c2e8513460d03c9d20b80b06b16db1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-20 13:43:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae9f013dfeb44ecfb0476cda0ef8ff47\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e1b45d3e735c4a1e9ef2fe4383e29875 : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3aa8ca73927b4adbb6779a8cf080c876 : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8311e4e08f4040d58c83efa84377dd22 : SUCCESS in 15m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d26fe12e411ff6fdd9e375a275a52f8a6305a147","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-20 15:18:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8863665c1ecc4b43ae8e276971f8f139\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f2dfef1f0c7423999b14db2d0c736eb : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32daff2d929044c58f81d62867d7e71e : SUCCESS in 4m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c01fdec224084d87914f80c74d06d6a2 : SUCCESS in 3m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/da409c640ebd43bd86a541c616a9d054 : SUCCESS in 7m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/707b48ed630d42fa8973905fc10c95f6 : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a52794f0c0bf4a02819e72932ee082b1 : SUCCESS in 10m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/40aa1f3d4caa49ddb3329fa9bb7a91a3 : SUCCESS in 5m 12s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/81f6bcc5b11a4b46a2a0746a895c0a28 : SUCCESS in 45m 40s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/d806d606348048faa6e53c259e89ea9d : SUCCESS in 1h 51m 51s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/56e0bf97986840d19572a168ab857e8a : SUCCESS in 1h 33m 34s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/049dc7e0062444bf9924eebe6261d2ec : SUCCESS in 1h 42m 34s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8ae6e2ea1fbc48b78ab88458ee5386bf : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":1},{"id":"2bcf2f9b89d49672acca738804aa67b188a3b6d0","tag":"autogenerated:gerrit:abandon","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-23 14:28:38.000000000","message":"Abandoned\n\nShould be fixed in https://review.opendev.org/c/openstack/trove/+/948284","accounts_in_message":[],"_revision_number":1},{"id":"ed1d7c99a93496ab3b46aaa2bb17aa1f87383f18","author":{"_account_id":5733,"name":"Kieran Spear","email":"kispear@gmail.com","username":"kspear"},"date":"2026-08-07 07:18:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4bfaadf330cdd536f35c06076efb862f9101abfb","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-08-07 08:37:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8a610ab1fe8528b9cc28145dde455588b8a68429","author":{"_account_id":5733,"name":"Kieran Spear","email":"kispear@gmail.com","username":"kspear"},"date":"2026-08-07 09:34:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b76c234a131c6d7b8873a3247ee9949cd4a64b81","revisions":{"b76c234a131c6d7b8873a3247ee9949cd4a64b81":{"kind":"REWORK","_number":1,"created":"2026-06-20 13:25:41.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/77/994177/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/77/994177/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/77/994177/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/77/994177/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/77/994177/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/77/994177/1"}}},"commit":{"parents":[{"commit":"36db92ee9a6d53e129cb7c195326d1f1eee8f8bb","subject":"Merge \"Remove service credentials from trove-guestagent.conf\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/36db92ee9a6d53e129cb7c195326d1f1eee8f8bb"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-20 12:17:39.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-06-20 13:23:28.000000000","tz":300},"subject":"Fix mysql get_backup_image arguments","message":"Fix mysql get_backup_image arguments\n\nIn some edge cases MySqlApp triggers arguments error, due to\na lack of ds_version argument, which is present in parent method.\n\nChange-Id: I4bf288234765a8b85edac3d79fe3c7d270730a5a\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b76c234a131c6d7b8873a3247ee9949cd4a64b81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b76c234a131c6d7b8873a3247ee9949cd4a64b81"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
