)]}'
{"id":"openstack%2Ftrove~994709","triplet_id":"openstack%2Ftrove~master~I784e679d09d4615f9af05bb7636a764c25674aae","project":"openstack/trove","branch":"master","topic":"oslotest-autospec","attention_set":{"8213":{"account":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"last_update":"2026-06-24 14:41:44.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I784e679d09d4615f9af05bb7636a764c25674aae","subject":"tests: update Mock and MagicMock to mock.Mock and mock.MagicMock","status":"NEW","created":"2026-06-24 12:41:02.000000000","updated":"2026-06-28 16:19:28.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":46,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6f2c2f51de0b183c366aa08f9ec3ac376d4fe289","_number":994709,"virtual_id_number":994709,"owner":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-06-24 18:02:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-24 13:40:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-24 14:41:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5ea744828ad6c4e938ec33455bb2e51c9da25959","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"date":"2026-06-24 12:41:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b365c4e06dab0c152d36c7c8c1673c72d53da8d7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 13:40:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ed96f678a6048a3bc77ca59f1b9b36c\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bf36991a29dc47b88b73586f8f403c32 : FAILURE in 7m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/185d4557a52b40748a1d78b813b4fad0 : FAILURE in 8m 07s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0a0ff32af1ba42d4849d7f25d2030132 : FAILURE in 11m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a9d1a9871a75b30d96608e3fe569b75a0e64b132","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 14:41:44.000000000","message":"Patch Set 1: Verified-1\n\n(46 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2843632a6d7c491786e772a5845b67ee\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a7e3afce71e543b6b54c10c5d861676f : FAILURE in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1189e643da7e4b438ed2610d3fbda3ca : FAILURE in 5m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/42cc731dd34c4874902e9bf095b4aef8 : FAILURE in 4m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f18adf4accf84d8ba90e2f46d5d13fa2 : FAILURE in 3m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/39581302e2e748f2ad6672c71f30bf51 : FAILURE in 5m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55d0ac4a0f524a3c9ecd1075402b36dc : SUCCESS in 10m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5a2fc73124143cbb695a40d48802d27 : SUCCESS in 8m 58s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/ea7581727fe94f1ebe7838f97e053368 : SUCCESS in 1h 13m 19s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/50ebd3fbb4ce42a388f35090c80365b2 : SUCCESS in 1h 58m 13s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/403695ba47c443e4a684dbb9420aa5a4 : SUCCESS in 1h 41m 22s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/0166058a2e6748c3ae5c38e938a40adc : SUCCESS in 1h 51m 13s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/546e45f9e8494e58a0c6d45e345beea7 : SUCCESS in 47m 22s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/761bd3523e974bcd98905ad291b4d656 : SUCCESS in 1h 34m 16s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/dcf79349bf7e4dd99e150723e84849cb : SUCCESS in 1h 14m 31s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/680fb49dbfe844eaa540d763b79c705b : SUCCESS in 1h 37m 00s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/83d6a48cd0914f00b2b807485e59a318 : SUCCESS in 47m 50s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/b8ddc299c32c4a1aba42ca8c6ce0a456 : FAILURE in 1h 03m 09s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/b7d4d2bcf92b4ba6ad207c172a49557e : SUCCESS in 1h 30m 29s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/24afc3caf2124c75a9a57d3fe90b87eb : SUCCESS in 1h 32m 58s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e78a44d92cf14c6ab0a4c6999115d4f0 : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":1},{"id":"3d597410cbd7ecfcfccf2f58a77881f39c882115","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"date":"2026-06-24 15:33:15.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"bd5173c98d5e6635cc302f302264339a0bedde03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"date":"2026-06-24 15:41:44.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8721ab1599d38ace98d34b0e52c1a5cd5750adcf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"date":"2026-06-24 15:43:28.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"38f358b30fa3e8e729d1c32f5cf9e452b443559d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 16:17:36.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c800ec7b82004d8baf702dcf4620dfb1\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e798c910eba54cf192d9fdcd5bc5c7e9 : SUCCESS in 9m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1a51956a3b1e494daa5bb504c1281bef : SUCCESS in 7m 40s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a95acd17bd2b434fbdc6ad6b383867e3 : SUCCESS in 15m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"dd3c164af8fb2417a5a31b1124acc06f381b5042","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 18:02:48.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7bb301ad89d4cd5ba91563f70a6b79e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be71e38e42e84c8eab55c7b7c3a22f95 : SUCCESS in 8m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c52912d883904fd29acbab3a20f6f727 : SUCCESS in 5m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/de6232575e3c417c89a5614309d07a3d : SUCCESS in 6m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ad62093075ec443790c18448325bc906 : SUCCESS in 5m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a2568f35393848f8bc83e1b54f138606 : SUCCESS in 9m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d51ab2ea4f94d8ea5b2b64edfeee65f : SUCCESS in 12m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b8f42157bce48229c3db45ae4c99fae : SUCCESS in 7m 19s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/9fb085e357dd4821b8545ded3dbf8c7e : SUCCESS in 58m 20s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/0ac82a9f18d94d7281f81f3c1e201a58 : SUCCESS in 2h 18m 20s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/988817fe206a4528bb1b1bfe47ec20b3 : SUCCESS in 1h 41m 36s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/3bc0478f81064d12b1c122c8726edf69 : SUCCESS in 1h 54m 13s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/7455a209b0574344844653eef0ce1424 : SUCCESS in 42m 55s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/c586e357d25b469893491f70b06b28d3 : SUCCESS in 1h 28m 10s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/3508852f52c742d89b0c5f157fabb76b : SUCCESS in 1h 15m 42s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/4762826148af4ebda5d29ab6dea7d5fb : SUCCESS in 1h 58m 27s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/23984532c69e475eb6e7a0bccec5db84 : SUCCESS in 57m 59s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/79b8c06387de4ed1b2e5fdb3267a9253 : FAILURE in 1h 08m 37s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/6a167cd94f9541f08bf17ac7c9388dda : SUCCESS in 1h 22m 48s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/af4d178f4672425b8a46678ce1c0d5ee : SUCCESS in 1h 23m 03s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2208d9bbeca9b537e74a15ffda205fa496c56a5e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"date":"2026-06-28 16:19:23.000000000","message":"Topic tests-autospec removed","accounts_in_message":[],"_revision_number":4},{"id":"6f2c2f51de0b183c366aa08f9ec3ac376d4fe289","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"date":"2026-06-28 16:19:28.000000000","message":"Topic set to oslotest-autospec","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"853257cb11a3a879e7ab8d4742289a29bad04088","revisions":{"04cfad0ea59c023431af23d46c9d7dece76e39ee":{"kind":"REWORK","_number":1,"created":"2026-06-24 12:41:02.000000000","uploader":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"ref":"refs/changes/09/994709/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/09/994709/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/09/994709/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/09/994709/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/09/994709/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/09/994709/1"}}},"commit":{"parents":[{"commit":"a5dd3f6d1b873462378b27230a2c34991132ecfe","subject":"tests: Enforce autospec\u003dTrue on mock patches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a5dd3f6d1b873462378b27230a2c34991132ecfe"}]}],"author":{"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","date":"2026-06-24 10:41:38.000000000","tz":0},"committer":{"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","date":"2026-06-24 12:40:50.000000000","tz":0},"subject":"tests: update Mock and MagicMock to mock.Mock and mock.MagicMock","message":"tests: update Mock and MagicMock to mock.Mock and mock.MagicMock\n\nBeing set up in test setUp, mock_fixture.MockAutospecFixture() can only work\nas intended if Mock and MagicMock are not imported directly.\n\nNeeded for a follow up commit, in which tests are updated to use\nautospecs.\n\nChange-Id: I784e679d09d4615f9af05bb7636a764c25674aae\nSigned-off-by: Claudiu Belu \u003ccbelu@cloudbasesolutions.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/04cfad0ea59c023431af23d46c9d7dece76e39ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/04cfad0ea59c023431af23d46c9d7dece76e39ee"}]},"branch":"refs/heads/master"},"0dda51195baa47cc392c2c3d2815d08420d340f3":{"kind":"REWORK","_number":2,"created":"2026-06-24 15:33:15.000000000","uploader":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"ref":"refs/changes/09/994709/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/09/994709/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/09/994709/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/09/994709/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/09/994709/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/09/994709/2"}}},"commit":{"parents":[{"commit":"cedbb6185df8e180c187abe0bb701ad21084c690","subject":"tests: Enforce autospec\u003dTrue on mock patches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cedbb6185df8e180c187abe0bb701ad21084c690"}]}],"author":{"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","date":"2026-06-24 10:41:38.000000000","tz":0},"committer":{"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","date":"2026-06-24 15:31:57.000000000","tz":0},"subject":"tests: update Mock and MagicMock to mock.Mock and mock.MagicMock","message":"tests: update Mock and MagicMock to mock.Mock and mock.MagicMock\n\nBeing set up in test setUp, mock_fixture.MockAutospecFixture() can only work\nas intended if Mock and MagicMock are not imported directly.\n\nNeeded for a follow up commit, in which tests are updated to use\nautospecs.\n\nChange-Id: I784e679d09d4615f9af05bb7636a764c25674aae\nSigned-off-by: Claudiu Belu \u003ccbelu@cloudbasesolutions.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0dda51195baa47cc392c2c3d2815d08420d340f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0dda51195baa47cc392c2c3d2815d08420d340f3"}]},"branch":"refs/heads/master"},"ee7bacc62ef1c97d6c1d2c9fcbde2ddfd7168294":{"kind":"REWORK","_number":3,"created":"2026-06-24 15:41:44.000000000","uploader":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"ref":"refs/changes/09/994709/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/09/994709/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/09/994709/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/09/994709/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/09/994709/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/09/994709/3"}}},"commit":{"parents":[{"commit":"d752b1bf300509a7779d884f395a647429989163","subject":"tests: Enforce autospec\u003dTrue on mock patches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d752b1bf300509a7779d884f395a647429989163"}]}],"author":{"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","date":"2026-06-24 09:39:53.000000000","tz":0},"committer":{"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","date":"2026-06-24 15:41:23.000000000","tz":0},"subject":"tests: update Mock and MagicMock to mock.Mock and mock.MagicMock","message":"tests: update Mock and MagicMock to mock.Mock and mock.MagicMock\n\nBeing set up in test setUp, mock_fixture.MockAutospecFixture() can only work\nas intended if Mock and MagicMock are not imported directly.\n\nNeeded for a follow up commit, in which tests are updated to use\nautospecs.\n\nChange-Id: I784e679d09d4615f9af05bb7636a764c25674aae\nSigned-off-by: Claudiu Belu \u003ccbelu@cloudbasesolutions.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ee7bacc62ef1c97d6c1d2c9fcbde2ddfd7168294"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ee7bacc62ef1c97d6c1d2c9fcbde2ddfd7168294"}]},"branch":"refs/heads/master"},"853257cb11a3a879e7ab8d4742289a29bad04088":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-06-24 15:43:28.000000000","uploader":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"ref":"refs/changes/09/994709/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/09/994709/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/09/994709/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/09/994709/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/09/994709/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/09/994709/4"}}},"commit":{"parents":[{"commit":"a72459c07be012e923c3b5ea4b9818a09bf7051d","subject":"tests: Enforce autospec\u003dTrue on mock patches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a72459c07be012e923c3b5ea4b9818a09bf7051d"}]}],"author":{"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","date":"2026-06-24 09:39:53.000000000","tz":0},"committer":{"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","date":"2026-06-24 15:43:17.000000000","tz":0},"subject":"tests: update Mock and MagicMock to mock.Mock and mock.MagicMock","message":"tests: update Mock and MagicMock to mock.Mock and mock.MagicMock\n\nBeing set up in test setUp, mock_fixture.MockAutospecFixture() can only work\nas intended if Mock and MagicMock are not imported directly.\n\nNeeded for a follow up commit, in which tests are updated to use\nautospecs.\n\nChange-Id: I784e679d09d4615f9af05bb7636a764c25674aae\nSigned-off-by: Claudiu Belu \u003ccbelu@cloudbasesolutions.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/853257cb11a3a879e7ab8d4742289a29bad04088"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/853257cb11a3a879e7ab8d4742289a29bad04088"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
