)]}'
{"id":"openstack%2Ftrove~994710","triplet_id":"openstack%2Ftrove~master~Ic05725075768f8310cd59f40d0c61a9b13aeebae","project":"openstack/trove","branch":"master","topic":"oslotest-autospec","attention_set":{"8213":{"account":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"last_update":"2026-06-24 14:46:54.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ic05725075768f8310cd59f40d0c61a9b13aeebae","subject":"tests: Update tests to use mock autospecs","status":"NEW","created":"2026-06-24 12:41:02.000000000","updated":"2026-07-09 11:34:13.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b5ca92b7901697e9776937b43390f75c468d8059","_number":994710,"virtual_id_number":994710,"owner":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-06-24 20:44:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-24 13:55:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-24 14:46:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-26 15:50:36.000000000","updated_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"reviewer":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"state":"CC"}],"messages":[{"id":"50715de344727f15a81df6ed3ae95ba870d7585a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"date":"2026-06-24 12:41:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5891896e89c337e3764ebbbd3ba6944ca54ec04c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 13:55:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc6ec7b91c8e497b9bd9d60d54e4883b\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/807c7274e43c4b6b969edac0808d4e3e : FAILURE in 7m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2581c873acfd4d86a2334c6609149e97 : FAILURE in 8m 27s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/91815a587b8349a582b26e06ae74e7cc : FAILURE in 14m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cc922179a5398e7c8fb5f45cb44854fd14ccd954","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 14:46:54.000000000","message":"Patch Set 1: Verified-1\n\n(11 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0d58306d25624a278d2518602eaf4631\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0392144610424a7eae022babca0fb538 : FAILURE in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c00df3b6b51148029a51dc24a1dfef39 : FAILURE in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e290893c86c94d098d5102ec9d7a7b68 : FAILURE in 4m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dd87f80f59d0492b83ba18546214583d : FAILURE in 7m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/05e77de1bd644c06af2efd5cb4d70687 : FAILURE in 7m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3266116fd86544f4be2e5bd7dfa96d16 : SUCCESS in 11m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/696c834cf6c740aea37a7bf8d0d0875d : SUCCESS in 4m 43s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/f9257ad152504d8d809db8b207dc7c83 : SUCCESS in 51m 55s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/9500a3f6b79149c18ffbe3f4ed206729 : SUCCESS in 2h 02m 07s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/8e901ab789564e029490538325723bb4 : SUCCESS in 1h 39m 39s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/f9924321a09943eb8420e6d8e875dc7b : SUCCESS in 1h 54m 33s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/f223d13175f642ebb94daa1467da1476 : SUCCESS in 46m 08s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/3bd8460fd02c4469852dddaf1e051954 : SUCCESS in 1h 49m 18s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/ce29c5c823d1472b81a5f2c2ff384ff8 : SUCCESS in 1h 18m 15s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/54d0d7b96f4b4caaafae045a015f202e : SUCCESS in 1h 58m 41s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/1aa1b0e7067a4baba600779655dc3ecc : SUCCESS in 47m 28s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/6afb244fabf94144a2c05396d03482bb : FAILURE in 59m 32s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/8114a9e7b4a84ae8971afe57630711da : SUCCESS in 1h 19m 06s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/c28f0179bf804f76884a55a55a6a5f39 : SUCCESS in 1h 29m 08s (non-voting)\n\nWarning:\n  Comments left for invalid file trove/tests/fakes/swift.py\n  Comments left for invalid file trove/tests/unittests/cluster/test_cluster_controller.py\n  Comments left for invalid file trove/tests/unittests/cluster/test_cluster_pxc_controller.py\n  Comments left for invalid file trove/tests/unittests/cluster/test_cluster_redis_controller.py\n  Comments left for invalid file trove/tests/unittests/cluster/test_cluster_vertica_controller.py\n  Comments left for invalid file trove/tests/unittests/common/test_notification.py\n  Comments left for invalid file trove/tests/unittests/extensions/common/test_service.py\n  Comments left for invalid file trove/tests/unittests/extensions/mgmt/datastores/test_service.py\n  Comments left for invalid file trove/tests/unittests/flavor/test_flavor_views.py\n  Comments left for invalid file trove/tests/unittests/instance/test_instance_controller.py\n  Comments left for invalid file trove/tests/unittests/instance/test_instance_models.py\n  Comments left for invalid file trove/tests/unittests/module/test_module_views.py\n  Comments left for invalid file trove/tests/unittests/taskmanager/test_galera_clusters.py\n  Comments left for invalid file trove/tests/unittests/taskmanager/test_manager.py\n  Comments left for invalid file trove/tests/unittests/taskmanager/test_vertica_clusters.py","accounts_in_message":[],"_revision_number":1},{"id":"5123a840c930eda5bd0617e73635f41dc6759519","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"date":"2026-06-24 15:33:15.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"e90fdb8161c2b6940fd78f3bf1bf568bcee723f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"date":"2026-06-24 15:41:44.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"4c74de057a6d42a72a8d527d592f781846f1eff8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"date":"2026-06-24 15:43:28.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"0dd2bccdc881bbda81d62398c8a88dca884d1ffd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 16:37:11.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ea8d47cedad47c0ac2004be2a5095d6\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/325898ec21ca4bc389da5f5acfd52600 : FAILURE in 8m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/33f463efb17744728b0ce4983c778c2c : FAILURE in 6m 16s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/03d2d229dd694863bc3e0202e4ed7f4d : FAILURE in 9m 52s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"094fa52edd82526c0a38a1b67df6ee262bb9ab51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 17:57:21.000000000","message":"Patch Set 4: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aa9692b044054de18ecf0b654e4f493d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7140edff172f48dea657df99de5b4b68 : FAILURE in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8ed72ba739f4c39b8e9a580ca4b65bd : FAILURE in 4m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6be7fe476c6d48828f950d6362fb042c : FAILURE in 5m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/21f5336146d44b86ac2581b29f86f300 : FAILURE in 5m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ea6763b0cb444fa5a730db1a5f65cfe8 : FAILURE in 7m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d43d13364544457b8621892a2c207d0 : SUCCESS in 8m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cdd2abf455484c939b1e7e37ff1a8b2f : SUCCESS in 9m 40s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/18b4ea6e9403488d80afb3626fbc50af : SUCCESS in 1h 01m 43s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/9144745f8cdc4587adcdf778441b6d25 : SUCCESS in 1h 58m 35s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/2cffe2e2c2a34fcf925d9a79c6d64a6d : SUCCESS in 1h 47m 04s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/990fd4ed8ff440328faa258024f923b3 : SUCCESS in 1h 49m 49s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/e7fde58dec9d4ba8bacb6eff9a1fb714 : SUCCESS in 42m 36s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/3afd38928c1343e18c2c55b1d2bc440a : SUCCESS in 1h 41m 04s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/5fd2fca70cab45e382e61f91e629062e : SUCCESS in 1h 21m 27s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/303dbc1963db401dab8aa1d89bb8fba2 : SUCCESS in 1h 38m 58s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/9f01f4a1e8cb4da196c0bb934c05a88a : SUCCESS in 54m 10s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/304a75851e724fca801d6418e877ef45 : FAILURE in 1h 07m 18s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/b2824d7c0d4a4fc9822d9698bebb2b56 : SUCCESS in 1h 24m 16s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/caccdf52a02f4374b7c4f57c94513f06 : SUCCESS in 2h 11m 45s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2ef8d5ef9e9ecc5e43402557ccdca5e4940474c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"date":"2026-06-24 18:19:22.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"89ac651199bb1d286955f5a60eb968d9eac3023d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 18:42:03.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34a1607057dd4048a30fda8f4749040f\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/887bd11fef75489694610a851d1a39d8 : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/47c6b0ba8fea48069acb190130a0ac9d : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/28a698287e2047219b3f4dbb76d0946d : SUCCESS in 8m 11s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"eff2ba8cb42cad305140dd30e8dcd8d0a603a64c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 20:44:44.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50d9422f190345559874d71d371f12ce\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac6c6ada6ce24267baa6638dc6c15cac : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d204b07abf5482b8ddadc8d815c0e33 : SUCCESS in 6m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ac4205fe1f6e46b0a8cb1ac41e801ed9 : SUCCESS in 3m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/36bb5cbb82374cfaa4b4a8b5be143346 : SUCCESS in 3m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6938d1d8b80f44bfaad798c354bcbd2b : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b7436f331f74d479fc34e764ab37ca1 : SUCCESS in 10m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/54c542a3afa041bd8c6244ee37558430 : SUCCESS in 3m 44s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/f6f19b5edfea45eb87d1f6339e45c3a2 : SUCCESS in 1h 22m 37s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/d09a6774cecd4e95acf0a62538c5cd8e : SUCCESS in 2h 24m 10s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/c71ee8522bb64ea9b4d172647cdcf7c1 : SUCCESS in 1h 37m 41s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/e58334ee409c4275b2db55828d07630c : SUCCESS in 2h 15m 46s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/de6e8b4b197240e88fef17de97ad7bee : SUCCESS in 1h 09m 11s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/61f6f2380ae74e53a87f8dc9eb97ac82 : SUCCESS in 2h 00m 20s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/b75cb41064c944bf9f6cc02fb7414897 : SUCCESS in 1h 13m 02s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/13d51fc409504b8f844472350fd7a692 : SUCCESS in 1h 36m 17s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/8ba7f978a8e343c9b2f1927f128e2bf6 : SUCCESS in 1h 14m 08s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/b6970c78a4ef4b2ea0b4ad6449ff0849 : FAILURE in 1h 24m 15s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/16f88ab9429d44e798f9163f42c869e4 : SUCCESS in 1h 16m 01s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/8ddc78b306954611a3a21b9db49b21ed : SUCCESS in 1h 45m 22s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"bacf09fb500cd6d036c121bf6af542622c8abdbf","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-26 15:50:36.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a5d7e694b828a120e6507692dcc3508553272d90","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"date":"2026-07-09 11:33:58.000000000","message":"Topic tests-autospec removed","accounts_in_message":[],"_revision_number":5},{"id":"b5ca92b7901697e9776937b43390f75c468d8059","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"date":"2026-07-09 11:34:13.000000000","message":"Topic set to oslotest-autospec","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"3327830d6484c7bd759f8856b4b9e56e4a1dfe2f","revisions":{"05c4d5551899145f23345cbb51b1757ff3b964b8":{"kind":"REWORK","_number":1,"created":"2026-06-24 12:41:02.000000000","uploader":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"ref":"refs/changes/10/994710/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/10/994710/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/10/994710/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/10/994710/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/10/994710/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/10/994710/1"}}},"commit":{"parents":[{"commit":"04cfad0ea59c023431af23d46c9d7dece76e39ee","subject":"tests: update Mock and MagicMock to mock.Mock and mock.MagicMock","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/04cfad0ea59c023431af23d46c9d7dece76e39ee"}]}],"author":{"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","date":"2026-06-24 10:42:09.000000000","tz":0},"committer":{"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","date":"2026-06-24 12:40:51.000000000","tz":0},"subject":"tests: Update tests to use mock autospecs","message":"tests: Update tests to use mock autospecs\n\nAutospec-ing enforces call signature checks when used.\n\nChange-Id: Ic05725075768f8310cd59f40d0c61a9b13aeebae\nSigned-off-by: Claudiu Belu \u003ccbelu@cloudbasesolutions.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/05c4d5551899145f23345cbb51b1757ff3b964b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/05c4d5551899145f23345cbb51b1757ff3b964b8"}]},"branch":"refs/heads/master"},"0686615cc069915585c760cfdbd5737ebd71e74b":{"kind":"REWORK","_number":2,"created":"2026-06-24 15:33:15.000000000","uploader":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"ref":"refs/changes/10/994710/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/10/994710/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/10/994710/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/10/994710/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/10/994710/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/10/994710/2"}}},"commit":{"parents":[{"commit":"0dda51195baa47cc392c2c3d2815d08420d340f3","subject":"tests: update Mock and MagicMock to mock.Mock and mock.MagicMock","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0dda51195baa47cc392c2c3d2815d08420d340f3"}]}],"author":{"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","date":"2026-06-24 10:42:09.000000000","tz":0},"committer":{"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","date":"2026-06-24 15:33:03.000000000","tz":0},"subject":"tests: Update tests to use mock autospecs","message":"tests: Update tests to use mock autospecs\n\nAutospec-ing enforces call signature checks when used.\n\nChange-Id: Ic05725075768f8310cd59f40d0c61a9b13aeebae\nSigned-off-by: Claudiu Belu \u003ccbelu@cloudbasesolutions.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0686615cc069915585c760cfdbd5737ebd71e74b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0686615cc069915585c760cfdbd5737ebd71e74b"}]},"branch":"refs/heads/master"},"b44663d1f706a76930862a2e7460f9105cf956c4":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-06-24 15:41:44.000000000","uploader":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"ref":"refs/changes/10/994710/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/10/994710/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/10/994710/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/10/994710/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/10/994710/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/10/994710/3"}}},"commit":{"parents":[{"commit":"ee7bacc62ef1c97d6c1d2c9fcbde2ddfd7168294","subject":"tests: update Mock and MagicMock to mock.Mock and mock.MagicMock","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ee7bacc62ef1c97d6c1d2c9fcbde2ddfd7168294"}]}],"author":{"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","date":"2026-06-24 10:42:09.000000000","tz":0},"committer":{"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","date":"2026-06-24 15:41:33.000000000","tz":0},"subject":"tests: Update tests to use mock autospecs","message":"tests: Update tests to use mock autospecs\n\nAutospec-ing enforces call signature checks when used.\n\nChange-Id: Ic05725075768f8310cd59f40d0c61a9b13aeebae\nSigned-off-by: Claudiu Belu \u003ccbelu@cloudbasesolutions.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b44663d1f706a76930862a2e7460f9105cf956c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b44663d1f706a76930862a2e7460f9105cf956c4"}]},"branch":"refs/heads/master"},"ec8f5023be78fa829fade17c037b108d4957fb16":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-06-24 15:43:28.000000000","uploader":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"ref":"refs/changes/10/994710/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/10/994710/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/10/994710/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/10/994710/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/10/994710/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/10/994710/4"}}},"commit":{"parents":[{"commit":"853257cb11a3a879e7ab8d4742289a29bad04088","subject":"tests: update Mock and MagicMock to mock.Mock and mock.MagicMock","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/853257cb11a3a879e7ab8d4742289a29bad04088"}]}],"author":{"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","date":"2026-06-24 10:42:09.000000000","tz":0},"committer":{"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","date":"2026-06-24 15:43:17.000000000","tz":0},"subject":"tests: Update tests to use mock autospecs","message":"tests: Update tests to use mock autospecs\n\nAutospec-ing enforces call signature checks when used.\n\nChange-Id: Ic05725075768f8310cd59f40d0c61a9b13aeebae\nSigned-off-by: Claudiu Belu \u003ccbelu@cloudbasesolutions.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ec8f5023be78fa829fade17c037b108d4957fb16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ec8f5023be78fa829fade17c037b108d4957fb16"}]},"branch":"refs/heads/master"},"3327830d6484c7bd759f8856b4b9e56e4a1dfe2f":{"kind":"REWORK","_number":5,"created":"2026-06-24 18:19:22.000000000","uploader":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"ref":"refs/changes/10/994710/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/10/994710/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/10/994710/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/10/994710/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/10/994710/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/10/994710/5"}}},"commit":{"parents":[{"commit":"853257cb11a3a879e7ab8d4742289a29bad04088","subject":"tests: update Mock and MagicMock to mock.Mock and mock.MagicMock","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/853257cb11a3a879e7ab8d4742289a29bad04088"}]}],"author":{"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","date":"2026-06-24 10:42:09.000000000","tz":0},"committer":{"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","date":"2026-06-24 18:19:12.000000000","tz":0},"subject":"tests: Update tests to use mock autospecs","message":"tests: Update tests to use mock autospecs\n\nAutospec-ing enforces call signature checks when used.\n\nChange-Id: Ic05725075768f8310cd59f40d0c61a9b13aeebae\nSigned-off-by: Claudiu Belu \u003ccbelu@cloudbasesolutions.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3327830d6484c7bd759f8856b4b9e56e4a1dfe2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3327830d6484c7bd759f8856b4b9e56e4a1dfe2f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
