)]}'
{"id":"openstack%2Ftrove~996931","triplet_id":"openstack%2Ftrove~master~I531220347c65e3bd2575b017703569d59b55d964","project":"openstack/trove","branch":"master","topic":"clean-unmount-on-rebuild-and-resize","attention_set":{},"removed_from_attention_set":{"36080":{"account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"last_update":"2026-07-12 13:42:45.000000000","reason":"\u003cGERRIT_ACCOUNT_36080\u003e replied on the change","reason_account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}}},"hashtags":[],"change_id":"I531220347c65e3bd2575b017703569d59b55d964","subject":"Unmount volume before instance rebuild","status":"MERGED","created":"2026-07-12 10:29:28.000000000","updated":"2026-07-12 14:32:50.000000000","submitted":"2026-07-12 14:31:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"996931-clean-unmount-on-rebuild-and-resize","meta_rev_id":"2de65094cbc5ee09c53308c4f24ac55961f5d97f","_number":996931,"virtual_id_number":996931,"owner":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-12 14:31:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":2,"date":"2026-07-12 13:42:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-07-12 13:40:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":1,"date":"2026-07-12 13:42:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":1,"date":"2026-07-12 13:42:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-12 10:44:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-12 12:29:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-12 13:40:33.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"9911e4bd638a4dc6e92fc500d4b31e7eaa762ed2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-12 10:29:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb6044b655599f77a1696a86dfcd3a81a9e50d59","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-12 10:44:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34dbe13950884d53a02b5f0dffdc36bf\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6c715bb5c8604a349305fe04a588aa89 : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/174d5a943b3c42de8fe774467b5ad3ae : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/659dd7b595874b74b76b147df9e305e1 : SUCCESS in 13m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a891120b9bb44507249674badfbda9a00515559e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-12 12:29:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c7db14888e846e283eb0115cbc6d15e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/15358fba59cf492d917761dd2e2a633b : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/addbd40960da4494902f76656a4fe967 : SUCCESS in 6m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/be945c626a7b4aea8275babeca3c2758 : SUCCESS in 4m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b5deccffcbcb49638e89c436b6c14062 : SUCCESS in 4m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a3c8442d7dc14271a08282f3ff633331 : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d2a67c2b2524970b4e20bf06a72a530 : SUCCESS in 10m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b84a0da3afaa45a9b444b09fdd90b129 : SUCCESS in 7m 51s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/f1870ea8870f4ee281e2c1e242d5c5e0 : SUCCESS in 51m 32s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/5202de6bc7a84bf99cd622e2f7a7972e : SUCCESS in 1h 59m 41s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/858fc4e779184ef995ea73c7b695a196 : SUCCESS in 1h 28m 26s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/5395fc78ca624b7bab4b1823752d3900 : SUCCESS in 1h 46m 55s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/15edf297134e4ec7a9fcc84debd63178 : SUCCESS in 45m 09s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/34f6bcc153494317afc88f3ae574bbfc : SUCCESS in 1h 32m 12s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/6d1b7d7338034e27b0e442719fcb350e : SUCCESS in 1h 21m 01s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/b998aadb413840a6b393b4cc55090226 : SUCCESS in 1h 43m 49s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/4f8cd48bab7548338cb8a30f771e0b51 : SUCCESS in 49m 59s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/4910f2907ae3464683ad4387c721924a : FAILURE in 1h 02m 37s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/6be87999c2bb47528a093eaf6a1353d7 : SUCCESS in 1h 16m 15s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/94dbd42174ab447eafe13a0a35314725 : SUCCESS in 1h 37m 00s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b40c259cfedc4409b7a27c01d5d404b9 : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":1},{"id":"5e8d052e2138b7b779ff4ccb336a317a851fa30f","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-07-12 13:40:33.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"416ed6b24658e711e63548b32afba36e43dee8e4","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-12 13:42:45.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c5aafd775b80c73f76a32893deebb9d2245e9a93","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-12 13:43:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d45bcabcce4ef56c157a12376d0f09493e8b011b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-12 14:31:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/307bdf6862f146f0b01f668e818c1027\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a120d62df54d47d9bb50887f98da4c57 : SUCCESS in 5m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/33b15e8a9bef49138ce2618d86f530bb : SUCCESS in 4m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/07e644f4822642a4914ddf40b4702572 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62546fbba2ad485ea89e73f30e5f3ea8 : SUCCESS in 8m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d76fa7334eb40e59de970b3a1a48049 : SUCCESS in 7m 32s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/e7b91ceaa3b0473ca9c8c2557cbbdc2e : SUCCESS in 48m 27s\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/a4544928862347cc97e329777a5674f7 : SUCCESS in 39m 59s\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/065936158917497ea85aa8db9c409de5 : SUCCESS in 43m 21s","accounts_in_message":[],"_revision_number":1},{"id":"b5038cdf528e815e96c4802d28a76875074d57d1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-12 14:31:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2de65094cbc5ee09c53308c4f24ac55961f5d97f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-12 14:32:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b637f6531b8146a58dae157acb8e31cd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a687fe968cd74e6487da813843031c78 : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb964f6740514ab8ac4b71ee71e3fe46 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"db36c51d5b0ad2519cb47574849c5b87c6e6c241","revisions":{"db36c51d5b0ad2519cb47574849c5b87c6e6c241":{"kind":"REWORK","_number":1,"created":"2026-07-12 10:29:28.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/31/996931/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/31/996931/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/31/996931/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/31/996931/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/31/996931/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/31/996931/1"}}},"commit":{"parents":[{"commit":"e9489ce73e2415ed277eb51ab45394f9247acc30","subject":"Merge \"Improve data volume response in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e9489ce73e2415ed277eb51ab45394f9247acc30"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-07-12 10:12:33.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-07-12 10:28:49.000000000","tz":300},"subject":"Unmount volume before instance rebuild","message":"Unmount volume before instance rebuild\n\nReuse the existing volume unmount logic by moving it into a shared\nVolumeUnmountActionBase class, allowing it to be used by both volume\nresize and instance rebuild operations.\n\nBefore issuing a Nova rebuild request, RebuildAction now unmounts the\ndata volume after the database has been stopped, reducing the risk of\ndata corruption during the rebuild/resize processes.\n\nChange-Id: I531220347c65e3bd2575b017703569d59b55d964\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/db36c51d5b0ad2519cb47574849c5b87c6e6c241"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/db36c51d5b0ad2519cb47574849c5b87c6e6c241"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
