)]}'
{"id":"openstack%2Ftrove~997965","triplet_id":"openstack%2Ftrove~master~I446e6713ba0b009473a66f2029b022d3aefcd572","project":"openstack/trove","branch":"master","topic":"api-flavor-associations","attention_set":{},"removed_from_attention_set":{"36080":{"account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"last_update":"2026-09-17 06:38:41.000000000","reason":"\u003cGERRIT_ACCOUNT_36080\u003e replied on the change","reason_account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},"39300":{"account":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"last_update":"2026-09-17 07:07:52.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_39300\u003e using the hovercard menu","reason_account":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"}}},"hashtags":[],"change_id":"I446e6713ba0b009473a66f2029b022d3aefcd572","subject":"Add datastore flavor association API","status":"MERGED","created":"2026-07-20 11:53:12.000000000","updated":"2026-09-17 07:44:51.000000000","submitted":"2026-09-17 07:43:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"997965-api-flavor-associations","meta_rev_id":"3f4b30bc53fe5fd4e16435113dec502c80354e4d","_number":997965,"virtual_id_number":997965,"owner":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-17 07:43:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":2,"date":"2026-09-17 06:38:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":1,"date":"2026-09-17 06:38:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":-1,"date":"2026-09-17 06:38:41.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-20 12:18:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-20 14:58:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-20 15:28:48.000000000","updated_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"reviewer":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"state":"CC"},{"updated":"2026-09-17 06:38:41.000000000","updated_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"reviewer":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"state":"REVIEWER"}],"messages":[{"id":"61a5a05c44a440cfed8c958ed31bc5e482bbbf7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-07-20 11:53:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c81ed9e6e969f301f7fe3def023e6c32316caa82","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 12:18:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4257401990e94f9daecfc15478901b9c\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/97c3728b2857471e8c4619b73c59cb4a : SUCCESS in 8m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f604c74d83eb46bd97a8cd71596c9b34 : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3c63c2d7d910492dabbe842d1b64adaa : SUCCESS in 8m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5fc26dd4b321244c5bbc2ee5bfe2d14518047d60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 14:58:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ef617e620e3d4734b7ccda8c16b4db6c\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/08ca5d8f2f2b48b3a914f8e0f9a5a8f0 : SUCCESS in 6m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/851b4b4c29a44e95a2ac728cd5949156 : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6187cd06012417f98dc53e73e437d15 : SUCCESS in 5m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/30620789b4884fa7ae6c027a259f59e1 : SUCCESS in 5m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4ac280bdddff48b4a835c86a13bc3106 : SUCCESS in 4m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/57b4954355bc486c90710b90188fd912 : SUCCESS in 8m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91e05ac81fea48df96a9512932a36683 : SUCCESS in 9m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10113446309c4d0f9aa546dfeb8363e4 : SUCCESS in 7m 41s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/c2477f41d11942dbb7261b8d720ca0c2 : FAILURE in 1h 23m 44s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/a601600c7b234c348cac50a2d939a2f0 : SUCCESS in 2h 06m 36s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/d9787fadd72a4717a3f5435853eba80e : SUCCESS in 1h 47m 32s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/8d5384f8bb894467883273d0bf36219c : SUCCESS in 3h 02m 53s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/2a157161522941579e85b26784e7e65b : SUCCESS in 1h 05m 20s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/c54fcf7777b146a6a1372e0988230bdc : FAILURE in 1h 47m 39s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/494c846266734fbd92b8ede857249c72 : SUCCESS in 2h 22m 49s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/39cda2cbf955427296f73f64d84b5c65 : FAILURE in 2h 11m 33s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/38217ad7a5bf46f8a1e4f3b7db8f48da : SUCCESS in 1h 12m 49s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/f40c2519ae1b4fe394b98508dda4acaa : FAILURE in 1h 35m 58s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/4686a299ec82497b8d41881412e8520c : FAILURE in 2h 00m 51s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/8a57d6ffa274486198f516b9e4d8c957 : SUCCESS in 1h 35m 11s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c87b27e3ecf14f4f9c11b2800c7ca4f7 : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":1},{"id":"3e13452607efa7cf6d641116a37717cf27292eb0","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-20 15:28:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0ccb138600026719509cdbf25b14f02ede23d859","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-07-21 07:36:15.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"0ea2d42790675f588e9582c58e4e1d4f9a21f313","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-07-21 07:41:23.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"2a1fba7daf39f3b5d0512effa5785e70304e75c6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 07:56:40.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc9099b88a30491389ce2a0167cabf43\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/81e989b25a464669a573c4e084123194 : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/58f32b37ea094c7d88a0f2f47ce5c8aa : SUCCESS in 7m 59s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e4aaf49e57a1462caf3c5abdfc250a34 : SUCCESS in 13m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3792181e6f20662021988741904c8108021ef5a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 10:45:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72e57cc9616840ba9c5311f3643530f4\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/96907603f4c2424aa938353a500851dc : SUCCESS in 5m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/944ccce4b67f491d993422f62af4d716 : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d51909e699e04c89b0fc885eb124f228 : SUCCESS in 4m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/13b7153b0a7a4a95818ac7ebe6db1fdc : SUCCESS in 4m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/320afd97673349218419a55c1fd62497 : SUCCESS in 4m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1962d9f4fbcd46749040e4316a908f63 : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7dadda0a9474aaab13ee6e3670ee35e : SUCCESS in 9m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84405f5646da47fc8f13a91d72a1fbc0 : SUCCESS in 3m 37s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/c8b173d11d6b443d9f898dd81f275205 : SUCCESS in 1h 05m 53s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/fd25d6878c31420bacb049e709213fa1 : TIMED_OUT in 3h 03m 22s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/ef1a9e3123424e89a03b25ac0a7f432d : FAILURE in 2h 17m 13s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/ef580ce6e3834ef6b1bd02fdfeebf0b8 : SUCCESS in 2h 57m 05s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/b008d31b6b554d4aa42bccd7f836891a : SUCCESS in 43m 59s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/26f8c7e826844dc296a9dce400c5951d : SUCCESS in 1h 39m 26s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/95dd6861b92948ab9f5704d636083d6c : TIMED_OUT in 3h 03m 24s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/b28a91805174400290ab1ad55c9c512c : SUCCESS in 2h 03m 00s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/2e8b0669f5ae4baa95e9f68e75d6a123 : SUCCESS in 1h 06m 05s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/1c62610485764a4796b31ac92d54a6d5 : FAILURE in 1h 46m 52s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/74e659e90d3f4fedb2b6badab5f5d014 : SUCCESS in 2h 31m 41s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/ee1e20a4f6e946b99b8206a2da8a09ea : SUCCESS in 2h 53m 41s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d00722c3f8ed4decb57f4b7ecf5c3fb0 : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":3},{"id":"3333b9dd6c93fd37bb7abc2ecbf94db07610aae9","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-07-22 14:21:10.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":3},{"id":"73b50e024ef9c201af38446fd5daff9da5889707","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-07-22 15:49:48.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"ec529f9b9e9836659ba08502375baaa1a11e1ecf","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-07-22 16:01:48.000000000","message":"Patch Set 4:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"cdeece133d541ebb28a3a34e98b47a572a991d65","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 16:31:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e83580af9bb348dba0c6695e45970fc6\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/964dcb0cfb19471cb857b7129e077646 : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5344a12c69ac4c3880bde2f5c420614d : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/1eb39c64ea17495ca458df1b6e040dab : SUCCESS in 10m 30s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"cfb8898fd06b48213646f419216134465f68e4f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 19:00:17.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f53c2a868f24bb9ba7bb0639967686a\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a6f421e170e240ccac4702e4a00427bc : SUCCESS in 6m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d9e6786b9333415b87ac4dfd34c6b281 : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/147ea829629a463aa442a938542fbedb : SUCCESS in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/33825f14d46a48e9997a2a2ec710673c : SUCCESS in 4m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7adcf51b72cd473c82b0f382c59610c1 : SUCCESS in 4m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/820ec3204782498e92ba0e19eb28b62a : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c32108cd39114931871d5119086de301 : SUCCESS in 10m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/396572be390342c98418366cdd4e1db4 : SUCCESS in 7m 46s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/f8bfd9200115477d8e1c0925d7e3e830 : FAILURE in 1h 32m 05s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/87a608117ce14f7da2a80d05fd6c2a0b : FAILURE in 1h 45m 10s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/e2f091a7c662461ca36fa436296f6ddc : FAILURE in 1h 58m 48s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/a548c9d499244d10b93c9f457c3eaaf1 : TIMED_OUT in 3h 05m 19s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/1cf800d0485441699a3e12efce70b310 : SUCCESS in 1h 07m 57s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/0d3bbd1affb34377b5b2d4a1cb0ad9e3 : SUCCESS in 2h 45m 28s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/e459c4ef86764f98afe4b88fb51b280f : SUCCESS in 2h 26m 59s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/052d5aab796f4c718dec78ae7de485bb : TIMED_OUT in 3h 05m 46s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/1d6d2d0770614590ae2fd1df249b7eb2 : SUCCESS in 1h 36m 53s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/cdd586b7f8e44aee82c4fdb290b0c52d : SUCCESS in 2h 48m 50s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/dd0a9253595547209b98a997e70801a2 : FAILURE in 1h 54m 54s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/ed5bc742203b4ccebe7d2807bfdbbd7d : SUCCESS in 2h 48m 59s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/eedb026f1b844f3987c90598c1e3d83e : SUCCESS in 6m 58s","accounts_in_message":[],"_revision_number":4},{"id":"8e2217d33546ea29636c47b8286c637b16e5688d","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-07-23 03:45:48.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"476f55bfca745e4a5e7bc3e5425a42557cf4f6db","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 04:17:35.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a33defd79b8841b09f8d33f243cc03c3\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1104490ae3304c8bb930f53613f664fb : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8ad7a5229d67474a93d79977b1277ea3 : SUCCESS in 8m 03s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4c4cbf583c46449bafca755f765022cf : SUCCESS in 14m 32s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"40751460cb9948a3c860f19302ceb4dfc90f4b82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 05:53:55.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67733ad6f066491ba494dcd6558273fb\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7094505db58443ae92ee06edeaf40ef3 : SUCCESS in 4m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d1efbe4abaa34d8f9fa196a89c5883ce : SUCCESS in 7m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f427d397a0454b3a88911d01e6d8be27 : SUCCESS in 4m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ec7d6dbd52534138beb333b863c0d8f8 : SUCCESS in 4m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5155ee6e750e4304a50f84cb6a0dba92 : SUCCESS in 6m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4ae52bf92efe47ef8e061e47ff11d9b4 : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04006306c3f94436b8154afacd9da37b : SUCCESS in 6m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5da665c0661d46d5a6a77edcc0b601c6 : SUCCESS in 7m 53s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/a71e77630aae460ebfcfcb02c2c5f313 : SUCCESS in 47m 35s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/a25e393c276843adb67b67251dcc5659 : SUCCESS in 1h 46m 52s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/2f83b92846ce45dba61959df8e6fc542 : SUCCESS in 1h 21m 55s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/135e40a12f8d4b898c2802702665dd3a : SUCCESS in 2h 07m 06s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/9493e06e6b8a4eb7a01e2090abd5869a : SUCCESS in 45m 13s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/83f821eda48f43a0820f111c08abe329 : SUCCESS in 1h 36m 04s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/6463cb12f4d04c9aa41508af6f17e1c8 : SUCCESS in 1h 19m 31s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/4ccaee0dc47d40b9b73d557a8179de23 : SUCCESS in 1h 41m 36s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/805b4662092b4066b285040610b9e997 : SUCCESS in 50m 10s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/fd2080e6e0db4fdda598e85cbfab15db : SUCCESS in 1h 34m 22s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/4d2ef55669ae4a14a39a09502c04ebad : SUCCESS in 1h 15m 08s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/d5edcfbfbf574d52a1d0a8c825c4cf16 : SUCCESS in 1h 36m 57s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/95d941b4b7594e2aadc1916b9e359ced : SUCCESS in 6m 12s","accounts_in_message":[],"_revision_number":4},{"id":"0426647932acfac302aedd3a4d6ae925a991572f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-07-24 15:30:31.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"7e0e295af6f40c1d85718b5cdf148e1130dca721","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 16:04:07.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/700f79e8d73041428d4ca3bc5c240449\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/63768e5201ed4941832d2dc39f36f413 : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/76b28005749348c8bfb2d48dcf4b1bb2 : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0bff973835fa4b509bbe397f09ae2964 : SUCCESS in 16m 52s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"70ba17023d5861a77e2290d166b3c30be9c5a805","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 18:37:19.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c0d318e755cc4c4682cfd47f29f25f88\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8e2c664db7fb4d79b15fb74e761206ce : SUCCESS in 6m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/678e831319f74cb38b5a1e7e078aee48 : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3705320e10574274ba9107e78b8ddf7e : SUCCESS in 5m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6e15703b60fd491fbcfa3d05fa500b9c : SUCCESS in 5m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2415b2434e474f68bafca6b2229cf0d0 : SUCCESS in 4m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/33847cedc5b34200a707ab8f3d7f520e : SUCCESS in 7m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afd978bed79242168ba736d86a6536ca : SUCCESS in 9m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/37ffc080910a44c78151fdaff65bcfc9 : SUCCESS in 10m 26s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/5478334acbdc4bd2946dcd44189c8a6a : SUCCESS in 1h 32m 07s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/3f744cde98e54991aa758b17a7e16a35 : FAILURE in 1h 21m 33s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/6eadc10e5e9e4199b293c4cedf4e3767 : SUCCESS in 1h 55m 58s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/1d4eb691d35c42cbb83d4130f4ad6367 : TIMED_OUT in 3h 03m 21s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/c871476c6baa4333b36ebec467dd16e0 : FAILURE in 1h 29m 47s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/8651df38af7f4e3e87ee3301841b9446 : SUCCESS in 2h 40m 59s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/76b26fcc615b4e0da8f12401fadab0d7 : SUCCESS in 1h 48m 40s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/70e4d8acdd164001a1ccdbc0bb7fea45 : SUCCESS in 2h 52m 36s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/cd85ec00a764420eb48928d5e2dd5321 : SUCCESS in 49m 12s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/2ba569af2c2f4c0bb28461293b7d0a08 : SUCCESS in 2h 44m 53s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/eb0f40556cf04570b693c463e740e98a : SUCCESS in 2h 00m 02s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/471a2a721564465c94c4231a04d1a202 : FAILURE in 2h 47m 06s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/eadd90cf914d47a6bdea3e2b2e9fb4fd : SUCCESS in 6m 28s","accounts_in_message":[],"_revision_number":5},{"id":"855f9260b3fc0b6d75d908e1f9c6c801b694c387","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-07-25 02:12:05.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a7fe83765fd004348cd6d81e822205f43768b8a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-25 04:20:11.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13dfa11650cf4ef48aee00e0bc0e5096\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1f79476a45604477b21ac8c3308addcd : SUCCESS in 8m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/366fb5d3ccf049b985c55d44fc3c0861 : SUCCESS in 8m 56s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/48876dd307b34586b4d4a206e8c5036f : SUCCESS in 14m 12s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f69aa5a1a4e2e126e6ca3186b273e8e41331a562","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-25 05:05:43.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ea618e9c57245cf88f5751adf7c72d7\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b65db9da0e654504820c8c0395b5eb09 : SUCCESS in 5m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3711e8fe5283402cbf7bf24eca222631 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b259faeef2694cdc8a396538079e4ebd : SUCCESS in 6m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b80d857d5bad40c19ca1aeca25a7e572 : SUCCESS in 5m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dbeaa8d14f3a4af7b539d7deea91bb06 : SUCCESS in 7m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ae375074f9234053b1ae013f34010a9f : SUCCESS in 7m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3f90ada1830429780b3e2c722fad9b5 : SUCCESS in 10m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a1fffa76d1c4878a637a69495841746 : SUCCESS in 8m 03s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/bca084289c234114b9a40359722ac36f : SUCCESS in 59m 08s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/607b25708b8e4186886a8390d1eff9df : SUCCESS in 1h 46m 49s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/00391d3535d84d4d8119770464b5bf23 : SUCCESS in 1h 36m 59s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/414417a416db4c6c8e5efe2917d0f278 : SUCCESS in 2h 18m 39s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/f6556ea55490438d98bfa68b3ce0b7a9 : SUCCESS in 43m 15s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/e08fc88e8c82421386365ffd3683ea58 : SUCCESS in 1h 27m 35s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/eccdac628cc54782827af7035be0ae48 : SUCCESS in 1h 27m 43s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/b856233838634b8eb2c1af76e24a55f5 : SUCCESS in 1h 45m 58s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/9a62ca5c56414b4396f2e3385e027d56 : SUCCESS in 49m 45s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/8b0863b94e1847299f193f46b7eedc3f : SUCCESS in 1h 30m 51s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/d2a4dcedd3b04cc9826847847bd7eb4a : SUCCESS in 1h 15m 02s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/42e22f52b4dd4ad68f18a3ced2d52bf6 : SUCCESS in 1h 36m 00s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/230c43c5b70f4232b0225b8815057e6b : SUCCESS in 5m 44s","accounts_in_message":[],"_revision_number":5},{"id":"c60135c2a7944ef84e98f342238f4c6c4a5c093e","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-08-04 17:53:59.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ef6c1393bc1251cd4f87b58fd073a56d9a10f11f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-08-06 05:05:07.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"e4fee54d8a7dd74e0424bb3aa8197490d0957334","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 05:23:39.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ec06c09ca794a38909e6ead047f57b7\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a880a82448e94b3ea7161204f69b10ec : FAILURE in 6m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/61c4b51b66c6451184b0db27044d231f : FAILURE in 7m 08s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/baec5899c9ae4cda9aeeaa6c9bf44de0 : FAILURE in 16m 04s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f7dce42d9d3ed255d99e0af84665e849bdad123b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-08-06 06:58:43.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"28a2dc562e836cdeea186f7225465591c5bae9f4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 07:17:30.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91bd1874d8434093afea4fd0e521ae09\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6e1bdb6ba35f4e6c84235be185a598f9 : SUCCESS in 8m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2b26f6e53a1a4389a959e594fe832b00 : SUCCESS in 9m 11s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4f6ca8a159c94062a7a4036b40e5d84b : SUCCESS in 16m 25s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a5cc4ff12fca78d444e1721d6dfe8e7f22b28ae5","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-08-06 08:36:58.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"06da3d8cd0a9385904de2d7f85bc03aa70f22330","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 09:04:13.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a73b8d3df4a44774a39a033bb8a5a3e5\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c65437294c49463490ad9884f51a4d6d : SUCCESS in 5m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d5235ad11a3744c5ba6f175af2aa976f : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f33e15822c4e430d8d5335d814254fd2 : SUCCESS in 2m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a4ebd7ffdbf744bb906aeb6abbc57f7f : SUCCESS in 3m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/26b7455a0edc406f9b7f336f81e0d895 : SUCCESS in 5m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fda900e0854145afa6813b56919b9619 : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42d3bc5e69a741ba9d24cef01aa3f322 : SUCCESS in 5m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b597c88c79ff46fc9f75e7834193fcb0 : SUCCESS in 6m 52s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/2eb46c0bd8774cc2a0122b08f410b560 : FAILURE in 24m 14s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/6fa59d6297f745b59d77f5e96c7ba2da : FAILURE in 24m 04s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/484c27e2baef4005948405b6827a8ebe : FAILURE in 21m 56s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/72903a9749e240b4880d997391ba3cea : SUCCESS in 2h 03m 16s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/6591d34f82c04e46acdbc66fce122711 : SUCCESS in 55m 30s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/aac0266aee124d2e9329f83dd28ad12a : SUCCESS in 1h 35m 06s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/50a81805db0946b0afe0b228788da58c : FAILURE in 24m 09s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/9bd1c5269d7c46a797a5e5cc67afd545 : SUCCESS in 1h 41m 10s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/fe35868875c94f7cb4b09910fb22a564 : SUCCESS in 57m 40s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/faa686483d584dbdbeac25b0725a6bcc : FAILURE in 20m 12s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/32c00d6a63f74f50ae594ecdb6ceb8b4 : FAILURE in 21m 20s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/315e4c59acc44a72a7a0b66b1e6bb011 : SUCCESS in 2h 03m 10s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2df4acfc66a34d2d82083062230f3f5a : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":7},{"id":"9030574a692e634394263005fdf8f831414ff76f","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-08-07 12:30:27.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"0a7638e97097f366aabf4cb8bc898e7395aa11ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 12:54:20.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a66063f4fc574198a6c305bc1537306c\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/75d6565805d841f8907223792fceb43b : SUCCESS in 8m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/95586b9c7ef04197979859e30a8a5e75 : SUCCESS in 8m 55s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2aa1b13f8a2040cf8987c36fdc0e7d0c : SUCCESS in 11m 06s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6c5a38e5d3353ff2be100755b2ce08c39fefdcb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 15:35:00.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de27ca2890124332bfa08c748493d1f4\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b62591e3c1824aa7a7ad77aa61583dba : SUCCESS in 2m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ca03192f7d04470b81997cd4eda7727 : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eae59abe59a2454191ab697b515f2f37 : SUCCESS in 5m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/12ad333acc934da3986c57df36b3302d : SUCCESS in 4m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/df171736bba643b3ae4bcb558f08136d : SUCCESS in 2m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/eed49d4a225c463f854030edec8d1ada : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dbdaaefdfb04bbaa1ba6ee1b973e412 : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90c3747d49034c44ae16a0843b15a698 : SUCCESS in 7m 47s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/3ca0d1a3e56d41889446266277d1d26c : SUCCESS in 1h 29m 43s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/4263506a303741a5987cc4324d37f26b : SUCCESS in 1h 58m 23s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/13ae6f335965433f8891da35a6a26c20 : SUCCESS in 1h 39m 27s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/61fbf3df2f8140e18be07088b8c7dfff : SUCCESS in 2h 36m 42s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/02dc4b2f8f9741a983b73825a414459e : SUCCESS in 56m 20s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/bbe7f41586954503a41094ca2b043ba4 : SUCCESS in 1h 55m 18s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/b485c5a3162140b8b94230ef6b3634e8 : TIMED_OUT in 3h 02m 56s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/fbc54391ead34c17b7ab77f8e677665a : SUCCESS in 1h 34m 56s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/a8e1c65e782343089b08529ebfef3998 : SUCCESS in 1h 03m 47s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/1c07bfea2d83409093ca79506f0360d7 : SUCCESS in 1h 51m 49s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/88bfb7c86f85430fa8902a30c6dc8e6e : SUCCESS in 1h 31m 46s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/bf5a7bb560704e3c93d0f1537af99256 : SUCCESS in 1h 52m 03s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/96689b32581a4833b943f2e6764a3262 : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":7},{"id":"ef57f4308007f8574e619bcf93426bd396320d35","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-14 11:53:57.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"bf2ff17adb096a5502467df102089896ac8a866a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-09-15 07:48:59.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"adba5901fb3e3ed4b2e2f0cd57bc8142a802043a","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-09-15 07:51:54.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"aa3734f95841cc174d3469542af75e04a0e4dbb5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 07:59:25.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e361f96a7636421ea625641d30c3fbf8\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/22ac032681e247b18d964d00640ccf81 : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7078181604a44228b001c4329b2d5ebc : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/bf068b6b77c04de7baff116eee2a0e81 : SUCCESS in 8m 20s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"3ddec133af425fc82b2d297a30155b690ee65691","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 09:06:24.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9a6f2bcdfe0947a18ef3ca218843f499\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/fa449ac2d49e476fa6f1983907a8cd6e : SUCCESS in 4m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f785aee3445d41068784baec4df7459c : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b06f4874f8842a7b888cfcf9f6439bb : SUCCESS in 5m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dcd5ee98584a4bb6af8eebf147e08196 : SUCCESS in 3m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/19a851604ab84a12a7cad0f8d3e1f97c : SUCCESS in 5m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6d9707c748ea48ce94eb9e11079455e8 : SUCCESS in 8m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1729cb2a29b44069997fb3f228f48de : SUCCESS in 10m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77f2367ff64b4aeb86680648f85391dd : SUCCESS in 8m 13s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/ac4b9400addc465d85b6ce607f2cb715 : SUCCESS in 39m 28s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/eb3ea5445dc848f6a87c2bd71752bd0e : SUCCESS in 1h 12m 21s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/d31d3411676143c2982d31cb7a1256c9 : SUCCESS in 1h 04m 59s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/9bf980b3280a4eab84334912b0d7b366 : SUCCESS in 1h 16m 11s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/d3b29c0b3b6d4becbca68e8e6d233544 : SUCCESS in 40m 42s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/ab64a1f5286d47a69ac2446291615359 : SUCCESS in 1h 09m 37s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/53a9611d200f463486168a54dbd4f0eb : SUCCESS in 1h 03m 03s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/97e7e2adbe734ea6bd6d0274077d0289 : SUCCESS in 1h 10m 54s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/45c7b89a72c1446d883cd218d82a8b9c : FAILURE in 22m 08s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/6e268c9dadc94f7cac3aa2f74781a9fb : SUCCESS in 1h 04m 54s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/a8272aa3668846169b5f26fe5771c143 : SUCCESS in 57m 40s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/2eacb022ef374eaf8beb4f71520758b0 : SUCCESS in 1h 02m 52s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ce79db4a951c40bd854c24aa9c4d630b : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":8},{"id":"b06f2132d66add28689d1128eb9cdc99526c5bc6","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-09-16 03:16:35.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"3a9204694cfdf8e4612b5c654b79555cd9212929","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 04:32:17.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1f4812caf6b461e91f54309e99cff4b\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d594fda9d3d240ac8229f7e9652fd777 : SUCCESS in 5m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b6d03de5169147959dfae4e5f2179dfc : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0a8cc6824de47ffb25d5c58d9a0e33c : SUCCESS in 5m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/448a6d3775fa43c59e605c1defd1fff0 : SUCCESS in 4m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/47d3e6b5c2f043ec853c1844faebf062 : SUCCESS in 6m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d5b6c477a9184ff6b3c69097daff2b6a : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbc68efafcd1454b8352cfccf65c1d99 : SUCCESS in 9m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed376201449a467e85d05b47728a1606 : SUCCESS in 6m 39s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/16298c27529d476493aea52500ffd3e4 : SUCCESS in 43m 32s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/939228b41eb442d9ad697a756c8951e4 : SUCCESS in 1h 13m 40s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/bb141f404d174921b5484f94d96cd6e8 : SUCCESS in 1h 02m 13s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/7e4f782d83bf46c2974df0b804357849 : FAILURE in 34m 19s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/0e4dd7012fe44877b1963f214c7e9606 : SUCCESS in 42m 22s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/2bbe52dcddb9406896d74821ec523fd0 : SUCCESS in 1h 12m 12s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/7f07c694bafa43c1ba97b1544e4bf823 : SUCCESS in 1h 07m 48s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/9ac46f6215c04d2787ae7b21fd8b1bb9 : SUCCESS in 1h 11m 43s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/d75e74efef904488875e1a5bdeacfd4a : SUCCESS in 1h 08m 51s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/3b5a8782ee6140f49b2a02b7956bf997 : SUCCESS in 1h 05m 39s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/ce2d21d17d324e0784c2aff5a9b3634e : SUCCESS in 56m 32s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/7df419a9f81e49e5893937378f35d451 : SUCCESS in 1h 02m 06s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7845654b2d9744189a967a37d6e44bf6 : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":8},{"id":"06c33f47deeae08eac1ca9165ef20f9df5cf70b0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 04:49:49.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2b29f9443a34b20a6102232a2f5a1dd\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f434dcdd6f7b4fd5a157b3e45c62e138 : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0fe9e5394a3f4abe83750f53f586f86e : SUCCESS in 7m 47s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/1426c316aacd4345be571fe4000737a3 : SUCCESS in 10m 09s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"e39dc7a6e7093f88a83bf2ad9303ea70b5ad7f37","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-17 06:38:41.000000000","message":"Patch Set 8: Backport-Candidate-1 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"0683b43c4a2dfe3ce98e4475633dad360e8afb72","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 06:39:30.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"8257cbfa1ca94cef57a06b52866314c54c3358c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 07:43:27.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f368e681faa443787c43edd081327d0\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f6faa7972e3646aa8ccb6197c34bf15f : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/caf43b3f304a429f81516257dd0c0503 : SUCCESS in 5m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/922683f7da334660ac5612b2579b9248 : SUCCESS in 4m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ffa4e7e42430469bb9366802ff1ca070 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c5e21ec45174ee68e2203fe27e51a7e : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/964e8d3bddc840b0af4cb130547298ec : SUCCESS in 8m 18s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/0abaccf4fd364750aa280d5f6b0dee76 : SUCCESS in 55m 17s\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/ae49f2905f0e4112bbd00ddfdb8ccf9a : SUCCESS in 41m 37s\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/f8c3407698564191aaa68c4ed08db8f7 : SUCCESS in 1h 00m 35s","accounts_in_message":[],"_revision_number":8},{"id":"99d9abcc4197632494de9194b78d9bcdfacc05b6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 07:43:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"3f4b30bc53fe5fd4e16435113dec502c80354e4d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 07:44:51.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38fbe4cab20b416eadcfe79a9bcdd5b8\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/6b6d3378d01c4574b105366f27578b72 : SUCCESS in 1m 00s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e0f146fa6f8475fa3187d470142d609 : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/19cf33dc83b44498801588b598edf442 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"34d1921062aac3ee55b9b792a8494c09053e854e","revisions":{"b3eaeea716e002cc94e020f00d4fa8e856add1f2":{"kind":"REWORK","_number":1,"created":"2026-07-20 11:53:12.000000000","uploader":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"ref":"refs/changes/65/997965/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/65/997965/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/997965/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/997965/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/997965/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/65/997965/1"}}},"commit":{"parents":[{"commit":"72cf40aa93ea241f312f0759af8d79e03272d93d","subject":"Merge \"Enable userdata refresh on instance rebuild (Nova ≥ 2.57)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/72cf40aa93ea241f312f0759af8d79e03272d93d"}]}],"author":{"name":"Olessya Khussainova","email":"freckled.cellofun@gmail.com","date":"2026-07-20 11:51:44.000000000","tz":300},"committer":{"name":"Olessya Khussainova","email":"freckled.cellofun@gmail.com","date":"2026-07-20 11:53:04.000000000","tz":300},"subject":"Add datastore version flavor association API","message":"Add datastore version flavor association API\n\nAdd a management API endpoint for setting flavor associations for a\ndatastore version.\n\nThis allows operators to add flavor associations through the API instead\nof using trove-manage, and supports passing an empty flavor_ids list to\nremove all flavor associations for the datastore version.\n\nChange-Id: I446e6713ba0b009473a66f2029b022d3aefcd572\nSigned-off-by: Olessya Khussainova \u003cfreckled.cellofun@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b3eaeea716e002cc94e020f00d4fa8e856add1f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b3eaeea716e002cc94e020f00d4fa8e856add1f2"}]},"branch":"refs/heads/master"},"c5174350e55d47d26fd71b4633981000ca501abf":{"kind":"REWORK","_number":2,"created":"2026-07-21 07:36:15.000000000","uploader":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"ref":"refs/changes/65/997965/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/65/997965/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/997965/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/997965/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/997965/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/65/997965/2"}}},"commit":{"parents":[{"commit":"72cf40aa93ea241f312f0759af8d79e03272d93d","subject":"Merge \"Enable userdata refresh on instance rebuild (Nova ≥ 2.57)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/72cf40aa93ea241f312f0759af8d79e03272d93d"}]}],"author":{"name":"Olessya Khussainova","email":"freckled.cellofun@gmail.com","date":"2026-07-20 11:51:44.000000000","tz":300},"committer":{"name":"Olessya Khussainova","email":"freckled.cellofun@gmail.com","date":"2026-07-21 07:33:05.000000000","tz":300},"subject":"Add datastore flavor association API","message":"Add datastore flavor association API\n\nTrove supports associating flavors with datastore versions through\ntrove-manage. That workflow requires direct access to the Trove control\nplane, which is not practical for deployments where datastore version\nmaintenance is automated through API workflows or where control plane\naccess is restricted.\n\nAdd a management API endpoint to associate one or more flavors with a\ndatastore version. This allows operators to maintain datastore version\nflavor restrictions through the Trove API instead of running trove-manage\non the control plane.\n\nChange-Id: I446e6713ba0b009473a66f2029b022d3aefcd572\nSigned-off-by: Olessya Khussainova \u003cfreckled.cellofun@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c5174350e55d47d26fd71b4633981000ca501abf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c5174350e55d47d26fd71b4633981000ca501abf"}]},"branch":"refs/heads/master"},"4034373f931f901a5003cfafc50ea9c982b861f4":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-07-21 07:41:23.000000000","uploader":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"ref":"refs/changes/65/997965/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/65/997965/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/997965/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/997965/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/997965/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/65/997965/3"}}},"commit":{"parents":[{"commit":"72cf40aa93ea241f312f0759af8d79e03272d93d","subject":"Merge \"Enable userdata refresh on instance rebuild (Nova ≥ 2.57)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/72cf40aa93ea241f312f0759af8d79e03272d93d"}]}],"author":{"name":"Olessya Khussainova","email":"freckled.cellofun@gmail.com","date":"2026-07-20 11:51:44.000000000","tz":300},"committer":{"name":"Olessya Khussainova","email":"freckled.cellofun@gmail.com","date":"2026-07-21 07:40:48.000000000","tz":300},"subject":"Add datastore flavor association API","message":"Add datastore flavor association API\n\nTrove supports associating flavors with datastore versions through\ntrove-manage. That workflow requires direct access to the Trove control\nplane, which is not practical for deployments where datastore version\nmaintenance is automated through API workflows or where control plane\naccess is restricted.\n\nAdd a management API endpoint to associate one or more flavors with a\ndatastore version. This allows operators to maintain datastore version\nflavor restrictions through the Trove API instead of running trove-\nmanage on the control plane.\n\nChange-Id: I446e6713ba0b009473a66f2029b022d3aefcd572\nSigned-off-by: Olessya Khussainova \u003cfreckled.cellofun@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4034373f931f901a5003cfafc50ea9c982b861f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4034373f931f901a5003cfafc50ea9c982b861f4"}]},"branch":"refs/heads/master"},"2581f7f0521cbbd64e0776fb18731bc0cec20df5":{"kind":"REWORK","_number":4,"created":"2026-07-22 15:49:48.000000000","uploader":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"ref":"refs/changes/65/997965/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/65/997965/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/997965/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/997965/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/997965/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/65/997965/4"}}},"commit":{"parents":[{"commit":"6a20f4fded41fa614c8598a9b243c74baec4b237","subject":"Merge \"Avoid multiple default routes in Zuul guest images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6a20f4fded41fa614c8598a9b243c74baec4b237"}]}],"author":{"name":"Olessya Khussainova","email":"freckled.cellofun@gmail.com","date":"2026-07-20 11:51:44.000000000","tz":300},"committer":{"name":"Olessya Khussainova","email":"freckled.cellofun@gmail.com","date":"2026-07-22 13:56:55.000000000","tz":300},"subject":"Add datastore flavor association API","message":"Add datastore flavor association API\n\nTrove supports associating flavors with datastore versions through\ntrove-manage. That workflow requires direct access to the Trove control\nplane, which is not practical for deployments where datastore version\nmaintenance is automated through API workflows or where control plane\naccess is restricted.\n\nAdd a management API endpoint to associate one or more flavors with a\ndatastore version. This allows operators to maintain datastore version\nflavor restrictions through the Trove API instead of running trove-\nmanage on the control plane.\n\nChange-Id: I446e6713ba0b009473a66f2029b022d3aefcd572\nSigned-off-by: Olessya Khussainova \u003cfreckled.cellofun@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2581f7f0521cbbd64e0776fb18731bc0cec20df5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2581f7f0521cbbd64e0776fb18731bc0cec20df5"}]},"branch":"refs/heads/master"},"d4cdc276dd1a74698786b51a01a2eb956451e11a":{"kind":"REWORK","_number":5,"created":"2026-07-24 15:30:31.000000000","uploader":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"ref":"refs/changes/65/997965/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/65/997965/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/997965/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/997965/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/997965/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/65/997965/5"}}},"commit":{"parents":[{"commit":"61ec2341004e60b3630fbffc8f56aeb64ef1add1","subject":"Merge \"Install dumb-init via package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/61ec2341004e60b3630fbffc8f56aeb64ef1add1"}]}],"author":{"name":"Olessya Khussainova","email":"freckled.cellofun@gmail.com","date":"2026-07-20 11:51:44.000000000","tz":300},"committer":{"name":"Olessya Khussainova","email":"freckled.cellofun@gmail.com","date":"2026-07-24 15:29:56.000000000","tz":300},"subject":"Add datastore flavor association API","message":"Add datastore flavor association API\n\nTrove supports associating flavors with datastore versions through\ntrove-manage. That workflow requires direct access to the Trove control\nplane, which is not practical for deployments where datastore version\nmaintenance is automated through API workflows or where control plane\naccess is restricted.\n\nAdd a management API endpoint to associate one or more flavors with a\ndatastore version. This allows operators to maintain datastore version\nflavor restrictions through the Trove API instead of running trove-\nmanage on the control plane.\n\nChange-Id: I446e6713ba0b009473a66f2029b022d3aefcd572\nSigned-off-by: Olessya Khussainova \u003cfreckled.cellofun@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d4cdc276dd1a74698786b51a01a2eb956451e11a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d4cdc276dd1a74698786b51a01a2eb956451e11a"}]},"branch":"refs/heads/master"},"73c6ef6a14f05fc4cf9f50c657ae0d28d5b1b70e":{"kind":"REWORK","_number":6,"created":"2026-08-06 05:05:07.000000000","uploader":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"ref":"refs/changes/65/997965/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/65/997965/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/997965/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/997965/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/997965/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/65/997965/6"}}},"commit":{"parents":[{"commit":"61ec2341004e60b3630fbffc8f56aeb64ef1add1","subject":"Merge \"Install dumb-init via package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/61ec2341004e60b3630fbffc8f56aeb64ef1add1"}]}],"author":{"name":"Olessya Khussainova","email":"freckled.cellofun@gmail.com","date":"2026-07-20 11:51:44.000000000","tz":300},"committer":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-08-06 05:05:04.000000000","tz":300},"subject":"Add datastore flavor association API","message":"Add datastore flavor association API\n\nAdd management API endpoints to list, add, and delete flavor\nassociations for a datastore version.\n\nChange-Id: I446e6713ba0b009473a66f2029b022d3aefcd572\nSigned-off-by: Olessya Khussainova \u003colessya.khussainova@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/73c6ef6a14f05fc4cf9f50c657ae0d28d5b1b70e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/73c6ef6a14f05fc4cf9f50c657ae0d28d5b1b70e"}]},"branch":"refs/heads/master"},"ec2e82bc207b57b09b85bcca85f0aade9a522fd7":{"kind":"REWORK","_number":7,"created":"2026-08-06 06:58:43.000000000","uploader":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"ref":"refs/changes/65/997965/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/65/997965/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/997965/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/997965/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/997965/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/65/997965/7"}}},"commit":{"parents":[{"commit":"61ec2341004e60b3630fbffc8f56aeb64ef1add1","subject":"Merge \"Install dumb-init via package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/61ec2341004e60b3630fbffc8f56aeb64ef1add1"}]}],"author":{"name":"Olessya Khussainova","email":"freckled.cellofun@gmail.com","date":"2026-07-20 11:51:44.000000000","tz":300},"committer":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-08-06 06:58:36.000000000","tz":300},"subject":"Add datastore flavor association API","message":"Add datastore flavor association API\n\nAdd management API endpoints to list, add, and delete flavor\nassociations for a datastore version.\n\nChange-Id: I446e6713ba0b009473a66f2029b022d3aefcd572\nSigned-off-by: Olessya Khussainova \u003colessya.khussainova@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ec2e82bc207b57b09b85bcca85f0aade9a522fd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ec2e82bc207b57b09b85bcca85f0aade9a522fd7"}]},"branch":"refs/heads/master"},"34d1921062aac3ee55b9b792a8494c09053e854e":{"kind":"REWORK","_number":8,"created":"2026-09-15 07:48:59.000000000","uploader":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"ref":"refs/changes/65/997965/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/65/997965/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/997965/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/997965/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/997965/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/65/997965/8"}}},"commit":{"parents":[{"commit":"61ec2341004e60b3630fbffc8f56aeb64ef1add1","subject":"Merge \"Install dumb-init via package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/61ec2341004e60b3630fbffc8f56aeb64ef1add1"}]}],"author":{"name":"Olessya Khussainova","email":"freckled.cellofun@gmail.com","date":"2026-07-20 11:51:44.000000000","tz":300},"committer":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-09-15 07:48:27.000000000","tz":300},"subject":"Add datastore flavor association API","message":"Add datastore flavor association API\n\nFlavor associations can currently be managed only with trove-manage,\nwhich requires direct access to the Trove control plane. This is not\nsuitable for deployments where control plane access is restricted or\ndatastore version maintenance is automated through API workflows.\n\nAdd management API endpoints to list, add, and delete flavor\nassociations for a datastore version. Adding associations preserves\nexisting entries and ignores flavor associations that already exist.\n\nChange-Id: I446e6713ba0b009473a66f2029b022d3aefcd572\nSigned-off-by: Olessya Khussainova \u003colessya.khussainova@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/34d1921062aac3ee55b9b792a8494c09053e854e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/34d1921062aac3ee55b9b792a8494c09053e854e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
