)]}'
{"id":"openstack%2Ftrove~998305","triplet_id":"openstack%2Ftrove~master~I68822e8ca26704ddfc6cf3506476938afa7f394e","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"36080":{"account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"last_update":"2026-07-23 16:37:18.000000000","reason":"\u003cGERRIT_ACCOUNT_36080\u003e replied on the change","reason_account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}}},"hashtags":[],"change_id":"I68822e8ca26704ddfc6cf3506476938afa7f394e","subject":"[DNM] test postgresql max_connections issue","status":"ABANDONED","created":"2026-07-22 13:46:48.000000000","updated":"2026-09-17 17:52:11.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2140f3b1d510dd42c4629d1ba0a06b2be4c0fa9a","_number":998305,"virtual_id_number":998305,"owner":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-07-23 18:53:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-22 14:02:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-22 15:16:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e354dfd300a962e24d0ef8c24d64591964054534","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-22 13:46:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"008a5f553217656c1b7906b76972dcec6194402c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 14:02:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28572a01117640c3a031520d9b3e5794\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/30a2576401554b9486b546e37ac6cc8d : SUCCESS in 7m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9327cfb6732a4099b2715a1c5da6dfd4 : SUCCESS in 8m 06s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f9299efe71a4465483a77f9657e43623 : SUCCESS in 13m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0320165bf693f650615fd81cfda3e51efac0591c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 15:16:42.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/trove/+/970639 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"cf808628ce0172a472e28680ed168bc131979443","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-23 16:37:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"84fcec27fe6b44c0c71a4a5479a8d455ddd77d56","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 16:53:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f5620fe248f4175860e21b3948a4a46\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7f5e400449c342728135db6df0114829 : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e97ae27939aa429883734ef6c98202be : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/85c6abd83c354a78ae70eea7c206e522 : SUCCESS in 13m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f7e1e8c134d69cdebc18f923ebee645d9baca130","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 17:26:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fc99ee298d7452a871e59c7cd1c3e27\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d92918e4545846a3bba7a3da8ccafedc : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d06c9372b9b84ac4b95ee68742876639 : SUCCESS in 5m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a4ac5df69b184ebf8718779ca96f89ff : SUCCESS in 4m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/186a302bdb9d404484f55ecb5a94e21e : SUCCESS in 4m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/638700d1ca1243ebb564445576adad44 : SUCCESS in 7m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87a7b8702d3f426e92194ab5d35e3c0b : SUCCESS in 9m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/393bd1aa0cbf4a799545610ccd284c78 : SUCCESS in 7m 07s\n- trove-tempest-ubuntu-max-connections-base-postgresql https://zuul.opendev.org/t/openstack/build/f0919cfcde98472c8a3ff9792423721b : SUCCESS in 46m 36s\n- trove-tempest-ubuntu-max-connections-scenario1-postgresql https://zuul.opendev.org/t/openstack/build/7e5ad94237a44cf4a4af0529d50f3427 : FAILURE in 44m 22s (non-voting)\n- trove-tempest-ubuntu-max-connections-scenario2-postgresql https://zuul.opendev.org/t/openstack/build/6cbcc02be6cf4bcba65dc65918ead554 : FAILURE in 45m 09s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/80c9580afc704d37925d046efffe979f : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":1},{"id":"435a7c671a13b1319af0cad0bb14661e4a5f789c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-23 17:56:45.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"ed64594ce23f3e03a46f33caffd1e58ed6d4d61c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 17:57:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5fab7e8beb4496d85eeb2920470f3bc\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/691cea83d8f84fb2adad5339fa3afff3 : NODE_FAILURE Node(set) request 874afaa653d04235873921313080ce5a failed in 0s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e2d194b5cb8741f1a1b88a7398f86c35 : NODE_FAILURE Node(set) request 0631e64aacb248078ad3e253198edf62 failed in 0s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/18a170c924c7409aa3008bbe70e7880c : NODE_FAILURE Node(set) request 417fe864f3a344bd89cfa0d895277321 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"76286e98a3a6b4e648b390c8195dd400797dfa3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 18:53:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4b23281fb4444f78eb4be45e779a8d1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/495c25cddcff4afb80a632f86c4c7025 : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8dac1d1c0034cc0910ada27564a2c55 : SUCCESS in 3m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/70c482d1a9744acdba5fd5810fc60ed3 : SUCCESS in 3m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/96ac972f0d5b4c9095cb23e1c54f1f68 : SUCCESS in 5m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1f41d872f84d4747911fa58ee20689bf : SUCCESS in 8m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b549d5c44e3242aa91194985162726c7 : SUCCESS in 5m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6dc23e2cfa19464fb1690180f99cda56 : SUCCESS in 6m 46s\n- trove-tempest-ubuntu-max-connections-base-postgresql https://zuul.opendev.org/t/openstack/build/4f2bea48c62c4403aa798a3f46d95580 : SUCCESS in 44m 08s\n- trove-tempest-ubuntu-max-connections-scenario1-postgresql https://zuul.opendev.org/t/openstack/build/326b718198ea45db880591ba6883d4e1 : SUCCESS in 55m 42s (non-voting)\n- trove-tempest-ubuntu-max-connections-scenario2-postgresql https://zuul.opendev.org/t/openstack/build/579d4a93746d4ad6ad39d114a49f0d08 : FAILURE in 41m 39s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5e4c2345db3d42afb12c2544c816c246 : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":2},{"id":"2140f3b1d510dd42c4629d1ba0a06b2be4c0fa9a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-17 17:52:11.000000000","message":"Abandoned\n\nOutdated MR cleanup.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c77113f277fd695e98368c3b3119013bd90a3c1c","revisions":{"428b6089589f702b9141f24e8560b3b8243c0025":{"kind":"REWORK","_number":1,"created":"2026-07-22 13:46:48.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/05/998305/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/05/998305/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/998305/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/998305/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/998305/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/05/998305/1"}}},"commit":{"parents":[{"commit":"6a20f4fded41fa614c8598a9b243c74baec4b237","subject":"Merge \"Avoid multiple default routes in Zuul guest images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6a20f4fded41fa614c8598a9b243c74baec4b237"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-07-22 13:45:17.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-07-22 13:45:17.000000000","tz":300},"subject":"[DNM] test postgresql max_connections issue","message":"[DNM] test postgresql max_connections issue\n\nNote that scenario2 should fail (not yet implemented).\n\nDepends-On: https://review.opendev.org/c/openstack/trove/+/970639\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/998303\nChange-Id: I68822e8ca26704ddfc6cf3506476938afa7f394e\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/428b6089589f702b9141f24e8560b3b8243c0025"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/428b6089589f702b9141f24e8560b3b8243c0025"}]},"branch":"refs/heads/master"},"c77113f277fd695e98368c3b3119013bd90a3c1c":{"kind":"REWORK","_number":2,"created":"2026-07-23 17:56:45.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/05/998305/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/05/998305/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/998305/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/998305/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/998305/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/05/998305/2"}}},"commit":{"parents":[{"commit":"61ec2341004e60b3630fbffc8f56aeb64ef1add1","subject":"Merge \"Install dumb-init via package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/61ec2341004e60b3630fbffc8f56aeb64ef1add1"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-07-22 13:45:17.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-07-23 17:56:31.000000000","tz":300},"subject":"[DNM] test postgresql max_connections issue","message":"[DNM] test postgresql max_connections issue\n\nNote that scenario2 should fail (not yet implemented).\n\nDepends-On: https://review.opendev.org/c/openstack/trove/+/970639\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/998303\nChange-Id: I68822e8ca26704ddfc6cf3506476938afa7f394e\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c77113f277fd695e98368c3b3119013bd90a3c1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c77113f277fd695e98368c3b3119013bd90a3c1c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
