)]}'
{"id":"openstack%2Ftrove~998575","triplet_id":"openstack%2Ftrove~master~I8a18650c67d4e38a03abb5d7d9d4b652dea6be36","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"36080":{"account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"last_update":"2026-09-02 12:05:01.000000000","reason":"\u003cGERRIT_ACCOUNT_36080\u003e replied on the change","reason_account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2026-09-02 12:47:18.000000000","reason":"Change was submitted"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-09-02 12:05:52.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I8a18650c67d4e38a03abb5d7d9d4b652dea6be36","subject":"Fix typo","status":"MERGED","created":"2026-07-24 04:45:26.000000000","updated":"2026-09-02 12:48:16.000000000","submitted":"2026-09-02 12:47:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"998575","meta_rev_id":"2bd44a74bc2019a882828ab69b396fd17140b504","_number":998575,"virtual_id_number":998575,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-02 12:47:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":2,"date":"2026-09-02 12:05:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":1,"date":"2026-09-02 12:05:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":-1,"date":"2026-09-02 12:05:01.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-24 04:59:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-24 07:50:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-01 19:22:49.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"state":"REVIEWER"},{"updated":"2026-09-02 00:26:57.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"358ed4ad0fb7d73502d2cbcd3b5cb5015bb9b3f6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-24 04:45:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b283005685480998fb930b37f3552f7d7da98d7b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-24 04:46:30.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"eda449133e9e65af8225ecf7cad60703a36f04b0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 04:59:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44a8bcc563534491bedee99641c19ab4\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8d942726d4184fd089971cb5c046e083 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f28b539584a64d67b1b8ed66f10d5481 : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d79e7ea5db124c53a9740cd6b2a3832e : SUCCESS in 12m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"987aa4cb3823390caef7e8b0dd09e3a271bacc5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 07:50:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d51642bc607477aa995afafa58b3735\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0642b27857bb4ec89c0bd890369b57a9 : SUCCESS in 4m 46s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/637bffc0593c4e17ad08c5776a978d6d : SUCCESS in 2m 46s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ebe691bc1a904305991b2e09cdff7234 : SUCCESS in 2m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b647a36f794247b28c397bb048dcd546 : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e92181453dc54e39849066f4ce492223 : SUCCESS in 3m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c8d117ea0d6646028cd4f112b427a9a5 : SUCCESS in 6m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c6edba90d5144c7f9e6017b0756faf7a : SUCCESS in 2m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d2027f45bc9c4221acfe4ea053b12df7 : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74259acbb0044e2cb62f9455934ec80c : SUCCESS in 9m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2e670a33a4d4853ab4595799a072139 : SUCCESS in 4m 15s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/79911f28b86e40909eb20dc716a6ce51 : SUCCESS in 1h 06m 20s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/bd865bd8193f4134ab2f0c3f3043a42b : SUCCESS in 2h 11m 27s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/79669e3e91ab4996bd8bd22c4ce20090 : SUCCESS in 1h 44m 07s (non-voting)\n- trove-tempest-ubuntu-ssl-mysql https://zuul.opendev.org/t/openstack/build/55aad977013742048afc196f81ae6a68 : SUCCESS in 1h 20m 40s\n- trove-tempest-ubuntu-ssl-replication-mysql https://zuul.opendev.org/t/openstack/build/713996eb456f437c90264239b54b6c25 : SUCCESS in 1h 34m 29s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/ba72bbebebca4e39b7a23e4fd0140b05 : TIMED_OUT in 3h 03m 25s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/d6f0a8305a034797a51a8fce112caf06 : SUCCESS in 1h 30m 18s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/37694d16017c4feab7d82d7db82be17a : SUCCESS in 1h 56m 16s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/bc56c0cfb26f4df3954afdab9ab6ee48 : SUCCESS in 1h 22m 13s (non-voting)\n- trove-tempest-ubuntu-ssl-mariadb https://zuul.opendev.org/t/openstack/build/8a2454c8b8744af09bdb7bbcaabd5f8e : SUCCESS in 1h 32m 32s\n- trove-tempest-ubuntu-ssl-replication-mariadb https://zuul.opendev.org/t/openstack/build/f47ee3a86de94f11b1f03a1af9a6a441 : SUCCESS in 1h 57m 53s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/ac585a438f55496a8ec79e53bff28287 : SUCCESS in 1h 56m 57s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/3beaf90560b04165950c5401680baa16 : SUCCESS in 1h 02m 25s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/a453afc8e09c4ba6a83b2a0baeb0b7a7 : SUCCESS in 2h 09m 56s (non-voting)\n- trove-tempest-ubuntu-ssl-postgresql https://zuul.opendev.org/t/openstack/build/e76e416688d84edead7ae8f6a30999a4 : SUCCESS in 58m 40s\n- trove-tempest-ubuntu-ssl-replication-postgresql https://zuul.opendev.org/t/openstack/build/064c296d2bd74736b5129b0e63d86546 : SUCCESS in 1h 36m 22s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/6edeb709be9d4e04ba017fe78c631bec : SUCCESS in 1h 34m 29s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/8206fb65485849bbb7ea440055cd2334 : SUCCESS in 1h 52m 17s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4b60da0e24f748d5982223c14a98a1ef : SUCCESS in 4m 39s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/1987c546563d49429795eba094f03897 : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":2},{"id":"ffb1e8ca344af9cb982cb499cedb229a6b95538e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-31 20:26:50.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2},{"id":"e9e6e253a92baadf81e91f30cd0940887ddf7bea","tag":"autogenerated:gerrit:restore","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-01 19:21:31.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"c804b9b58998dbc3d9d607dc36462cd0ec91ed77","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-01 19:21:42.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"35a83a22cb4d601edc82785a71edc86d6673cde9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-01 19:21:48.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"b83baf6b5d3c658d429464f81c3fd406bf883f09","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-01 19:22:43.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"5387c6584eef5fcdb4aed08e1e63b7822d60acdb","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-01 19:22:49.000000000","message":"Patch Set 5:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":5},{"id":"02711d544e2f5d54ec32c29f7192d1268e70ec97","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 19:49:39.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f283e29914d844db967b50b9612cc52a\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3ead12a2089f4c84aee0bcd6a3cc9e09 : NODE_FAILURE Node(set) request d634ab7a14424b2880c36d105360292c failed in 0s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f14e294025b7426a82283b289e25495e : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/263511e742af4b3bb5da97a024689bcf : SUCCESS in 13m 30s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d162a0d6522c42550d378c7dcd56abd661d322f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 20:45:57.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/573cd99e517b48c49d6ad43a4a7c0e6f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/293971f042064c588331657b72a0e7a1 : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d8da93430414ce09407c29549e1cb84 : SUCCESS in 7m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d8855cc9d36b4ded8dd2db87764b99c7 : SUCCESS in 3m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1ea4f5c492664a739abad18cf992266e : SUCCESS in 5m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/828e8377aded4071bbeb27af15a1a672 : SUCCESS in 8m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3e29169397d4ff69dc1f68c25d22a6f : SUCCESS in 10m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f99b582f296487eb61e3aae07c0b6e9 : SUCCESS in 10m 24s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/3a15f65a163e4058875158dcfcada8f3 : SUCCESS in 43m 43s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/ac29f98410f040e6b4ff15e183c44e08 : SUCCESS in 1h 12m 05s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/88b78ece559e44789bdfa706a69e2676 : SUCCESS in 1h 04m 46s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/1361b234a6c74b2a9e00e048647361f4 : SUCCESS in 1h 17m 23s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/041b82ecdd4f4841ba2e3cbe284af7e6 : SUCCESS in 41m 27s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/4cc101e7b6964d54aef685fd59e0907a : SUCCESS in 1h 18m 17s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/1dcfc560486f4609af0402d2e84ed54a : SUCCESS in 1h 01m 33s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/f1940623451b4b15bcce7aa734db1491 : SUCCESS in 1h 12m 11s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/effff92d410947b8896aee0bfd20a394 : SUCCESS in 42m 52s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/a5ca47ab6f3f41c494d579a637f1394e : SUCCESS in 1h 08m 59s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/f862c7a2644945808e022c8e6c50c142 : SUCCESS in 55m 50s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/81051ed7324c410c880e6fee7e6f7a52 : SUCCESS in 1h 02m 05s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1798d13afa1543a6a18bc7ae82664aab : SUCCESS in 2m 56s","accounts_in_message":[],"_revision_number":5},{"id":"838f2720926d74ba848eaf04f86786520b49d517","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-02 12:05:01.000000000","message":"Patch Set 5: Backport-Candidate-1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"4e278ef65c0d45c4a9f57da32ea2af529b3c0f92","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 12:05:34.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"eddadc6ed9056cc0791d97fcc54e93dc6c1a5039","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 12:47:18.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2c3257782704343bee25b0d58d7899f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa78da62ff214119a900dae6e4bf421e : SUCCESS in 3m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/79c0500fc2264ce39238d3f2fafe2e9e : SUCCESS in 4m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/88e68b2bd76d4fe2a8dba273a437306d : SUCCESS in 2m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ce3d3c491814ae1a18496d98ecb51fd : SUCCESS in 10m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/02074f23b32841578abad96f2822236e : SUCCESS in 7m 42s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/ff50fec3742a4f388b09cf2f399d4fa7 : SUCCESS in 41m 09s\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/474937efe31c4b6e85ed0b4e7faef7fe : SUCCESS in 38m 33s\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/92fa076b0fe64e3486e8e116bb7a42ed : SUCCESS in 40m 46s","accounts_in_message":[],"_revision_number":5},{"id":"c5468d6ab61a39e04910729ebe0a477b0fd90026","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 12:47:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"2bd44a74bc2019a882828ab69b396fd17140b504","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 12:48:16.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8cf0b6dbff2843df8d2c66011f1e8a5b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59316c9d34234a6fa8db6ce47a94a667 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"435f705fbbee993cb740a05c56b519721d90f757","revisions":{"b05bd3fa30ac148d030b6e0d6f49711ad01c9c30":{"kind":"REWORK","_number":1,"created":"2026-07-24 04:45:26.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/75/998575/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/75/998575/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/998575/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/998575/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/998575/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/75/998575/1"}}},"commit":{"parents":[{"commit":"61ec2341004e60b3630fbffc8f56aeb64ef1add1","subject":"Merge \"Install dumb-init via package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/61ec2341004e60b3630fbffc8f56aeb64ef1add1"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-24 04:45:26.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-24 04:45:26.000000000","tz":0},"subject":"WIP: Building guest image using AlmaLinux 10","message":"WIP: Building guest image using AlmaLinux 10\n\nChange-Id: I8a18650c67d4e38a03abb5d7d9d4b652dea6be36\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b05bd3fa30ac148d030b6e0d6f49711ad01c9c30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b05bd3fa30ac148d030b6e0d6f49711ad01c9c30"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"b48ce45d1669dca68a36e230b55343cb7692da71":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-07-24 04:46:30.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/75/998575/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/75/998575/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/998575/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/998575/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/998575/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/75/998575/2"}}},"commit":{"parents":[{"commit":"61ec2341004e60b3630fbffc8f56aeb64ef1add1","subject":"Merge \"Install dumb-init via package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/61ec2341004e60b3630fbffc8f56aeb64ef1add1"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-24 04:45:26.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-24 04:46:30.000000000","tz":0},"subject":"WIP: Building guest image using AlmaLinux 10","message":"WIP: Building guest image using AlmaLinux 10\n\nChange-Id: I8a18650c67d4e38a03abb5d7d9d4b652dea6be36\nCo-authored-by: Hirotaka Wakabayashi \u003chiwkby@yahoo.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b48ce45d1669dca68a36e230b55343cb7692da71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b48ce45d1669dca68a36e230b55343cb7692da71"}]},"branch":"refs/heads/master","description":"Edit commit message","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"318cb2264fdcd73b6bdb13e0b2e1b56a2cd52467":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-09-01 19:21:42.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/75/998575/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/75/998575/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/998575/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/998575/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/998575/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/75/998575/3"}}},"commit":{"parents":[{"commit":"fb596fef67b9c1a2b12e3c4692d4ad39592d77bb","subject":"Merge \"Pull embedded docker images with ctr\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fb596fef67b9c1a2b12e3c4692d4ad39592d77bb"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-24 04:45:26.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 19:21:42.000000000","tz":0},"subject":"WIP: Building guest image using AlmaLinux 10","message":"WIP: Building guest image using AlmaLinux 10\n\nChange-Id: I8a18650c67d4e38a03abb5d7d9d4b652dea6be36\nCo-authored-by: Hirotaka Wakabayashi \u003chiwkby@yahoo.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/318cb2264fdcd73b6bdb13e0b2e1b56a2cd52467"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/318cb2264fdcd73b6bdb13e0b2e1b56a2cd52467"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"61ec2341004e60b3630fbffc8f56aeb64ef1add1","ours":"b48ce45d1669dca68a36e230b55343cb7692da71","theirs":"fb596fef67b9c1a2b12e3c4692d4ad39592d77bb","merge_strategy":"recursive","contains_conflicts":false}},"a022c023046e57acd05af6ad680c1657d836ce1c":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-09-01 19:21:48.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/75/998575/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/75/998575/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/998575/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/998575/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/998575/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/75/998575/4"}}},"commit":{"parents":[{"commit":"fb596fef67b9c1a2b12e3c4692d4ad39592d77bb","subject":"Merge \"Pull embedded docker images with ctr\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fb596fef67b9c1a2b12e3c4692d4ad39592d77bb"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-24 04:45:26.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 19:21:48.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: I8a18650c67d4e38a03abb5d7d9d4b652dea6be36\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a022c023046e57acd05af6ad680c1657d836ce1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a022c023046e57acd05af6ad680c1657d836ce1c"}]},"branch":"refs/heads/master","description":"Edit commit message","conflicts":{"base":"61ec2341004e60b3630fbffc8f56aeb64ef1add1","ours":"b48ce45d1669dca68a36e230b55343cb7692da71","theirs":"fb596fef67b9c1a2b12e3c4692d4ad39592d77bb","merge_strategy":"recursive","contains_conflicts":false}},"435f705fbbee993cb740a05c56b519721d90f757":{"kind":"REWORK","_number":5,"created":"2026-09-01 19:22:43.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/75/998575/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/75/998575/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/998575/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/998575/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/998575/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/75/998575/5"}}},"commit":{"parents":[{"commit":"fb596fef67b9c1a2b12e3c4692d4ad39592d77bb","subject":"Merge \"Pull embedded docker images with ctr\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fb596fef67b9c1a2b12e3c4692d4ad39592d77bb"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-24 04:45:26.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 19:22:43.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: I8a18650c67d4e38a03abb5d7d9d4b652dea6be36\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/435f705fbbee993cb740a05c56b519721d90f757"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/435f705fbbee993cb740a05c56b519721d90f757"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
