)]}'
{"id":"openstack%2Ftrove~998782","triplet_id":"openstack%2Ftrove~master~Icd0926a4f2e071abe50c7e4c540555530b6a50a9","project":"openstack/trove","branch":"master","topic":"support-special-chars","attention_set":{},"removed_from_attention_set":{"36080":{"account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"last_update":"2026-08-21 06:27:23.000000000","reason":"\u003cGERRIT_ACCOUNT_36080\u003e replied on the change","reason_account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},"39300":{"account":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"last_update":"2026-08-17 09:29:37.000000000","reason":"\u003cGERRIT_ACCOUNT_39300\u003e replied on the change","reason_account":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"}}},"hashtags":[],"change_id":"Icd0926a4f2e071abe50c7e4c540555530b6a50a9","subject":"API endpoints for supporting special characters","status":"NEW","created":"2026-07-27 11:23:23.000000000","updated":"2026-08-21 06:27:23.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e75761d67c2f19c0f42c67d126e0e17d7ae2db75","_number":998782,"virtual_id_number":998782,"owner":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-17 09:22:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"all":[{"value":1,"date":"2026-08-17 09:29:37.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-27 12:06:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-27 14:58:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-11 14:42:30.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-08-14 13:55:14.000000000","updated_by":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"reviewer":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"state":"CC"},{"updated":"2026-08-17 09:29:37.000000000","updated_by":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"reviewer":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"state":"REVIEWER"}],"messages":[{"id":"a7e461af182a09e67e0d4b762ea133c395b27d91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-27 11:23:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d2aaf3ee80d06e93590f01e8b7fc0c093a216431","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-27 11:24:46.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"fb2071e068490496c0def05bab5e19364b1fec17","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-27 11:50:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7bc9ebefce02027047438f5bb7dfdda21a91e6c8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-27 12:06:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/efcaec2db77744ff9af1ebd44e34bbb1\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/af232a75f832479abd0c16e937c7d608 : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/63968d9de4c3455a94b9035c94185d73 : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e4f4da4adc99423dbab82153ab7909a6 : SUCCESS in 13m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d13ad6708f42aa46ca12fcfd8e57da0abbfa33ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-27 14:58:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fa8cae6505c4b14b8987950271296f4\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e0e3fd517eb6487c8f3d8565a3133f89 : SUCCESS in 5m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2802c1eaf6a64c12b4abfb200772e8f1 : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74c09767ac5c4ab281a4e0510987fd32 : SUCCESS in 5m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/02754b19f463428e8be545c38484d2ba : SUCCESS in 4m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7762130e16544cb9a249fab9d2bc51ac : SUCCESS in 4m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2042eadf9b184393a7f0735943b1c9ac : SUCCESS in 7m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8eedfda92e9d4248868653edcd18e3a5 : SUCCESS in 8m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9481e9ec45f741b1902d8e401454336b : SUCCESS in 8m 33s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/1840d116f85b46a69d94d309a5e006ec : SUCCESS in 1h 16m 20s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/0c2c09accae74998876fdc3ce0c19073 : SUCCESS in 2h 35m 21s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/67a3792e2ef44b9087895bbc237dc824 : SUCCESS in 2h 17m 42s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/4e5d7c7ae0e14ab7b64dd38ae0110a49 : TIMED_OUT in 3h 04m 05s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/7442e8574fa44046b0f6b4cfaef44ee6 : SUCCESS in 1h 11m 03s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/95e2387b22da4d4eb017ec23f70db310 : SUCCESS in 1h 51m 32s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/2fb60c54d03244aeae0fe72a24afd3aa : SUCCESS in 1h 51m 51s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/1634b2df40b04cec9cb09507b4378e41 : SUCCESS in 2h 07m 24s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/20fb247830b84e1480467f8609515ee5 : SUCCESS in 1h 16m 33s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/59d4fbb0990e4b2dbe84b1295db713b6 : SUCCESS in 2h 23m 24s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/05c943e6face4aa88a26b1d49c697ff9 : TIMED_OUT in 3h 03m 32s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/938304217be24010a1b11969b48c8876 : SUCCESS in 2h 10m 56s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4784cc0bf3494fff97f2a6fd71f67b13 : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":2},{"id":"730a6446a7884871936d7620b61db6938a1d6652","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-08-11 14:40:03.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"d44f7d3738c71c7021d907b0172ef940063860bc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-08-11 14:40:45.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"c2faa0fad02f346a766d6c0a0bdd6ea05b2ae234","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-11 14:42:30.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e6975e13fe28bbb7cb1b67ad2dfb4af7368d4d11","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-08-11 14:44:47.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"636b138acfee719b9f37fd960a57604d4c803b05","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-08-11 14:45:32.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"db850e02dc62cd1188bff9ff07a0149b3a2dc327","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 15:11:59.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed4b4cd552a74cc79d8012b526caf3bd\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/71485160898b4009a007b05c5737df43 : SUCCESS in 8m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3e6abb63f459427abc28ab2756aa9a1c : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/91dfbbf5d40048409c35bb664ce55c86 : SUCCESS in 16m 18s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"fcbf133002ab805caeac29b2021c43bc8b72d5b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 16:33:33.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61e28d60add54533b0b3c3048b8ea93f\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/eafd4dbf8ca54644bc52fb1f58106929 : SUCCESS in 4m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65c6ef4bba5a4f0fa0c6fde142dbbf17 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d885fb31f6c44612a98c688c75d85974 : SUCCESS in 4m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0c130587808a40e0a4106c06873af88f : SUCCESS in 2m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bbde3acdcca042bc9349c1ed8ca4acff : SUCCESS in 4m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9a7948a3429345199fd4248e7bcf444e : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfe1cbbcbc1444d7a05e3eb1f570d298 : SUCCESS in 5m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6de963df15854aa79b08660db200fd9c : SUCCESS in 8m 07s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/78c09e25bf274d25904964723167e6d3 : SUCCESS in 39m 31s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/a835b3a28ae141a2bb8a6e03a94d79d0 : SUCCESS in 1h 12m 32s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/5bd6975db0ff49f998e1d07113232677 : SUCCESS in 1h 03m 00s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/663c88d617bc4745b03c25ec1c7dce89 : SUCCESS in 1h 16m 44s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/e889ceac0f444e13920c8a653fd0ecaa : SUCCESS in 37m 58s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/0fe08ea29d2b464cb1b767e844c579ac : SUCCESS in 1h 12m 56s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/426e0aa1ed224a0586f79404964b0839 : SUCCESS in 1h 02m 16s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/0762a5cb6475425591e762ac8df674fa : SUCCESS in 1h 37m 07s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/0143811fe6ad4d3594abd51c1d9ed24f : SUCCESS in 1h 45m 42s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/c0a5c26e3e764e728ced69ee0ac50185 : FAILURE in 1h 14m 33s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/0b201b8dc6974a15a3c70bdfb7545312 : SUCCESS in 54m 21s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/f0e1380e9e1e420ebb950ad44c910dbc : FAILURE in 1h 14m 34s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5af9c87c42fb49029cb8087acd41c190 : SUCCESS in 6m 44s","accounts_in_message":[],"_revision_number":5},{"id":"e310bd619b15b47f57dbcd5b2db7dcdf70092ae3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-08-12 17:18:53.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"84e74d16be2cb18d415bffde7e6a1c133d603dd0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-08-12 17:39:12.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"793c7fc4d6a5ea573a2a08e59e54b9214a12d09e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 18:54:09.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21e7e41c1ed34e52814126a2bf3ab76a\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9cda7974059f4a17aeb37ef7ed4a1e7d : SUCCESS in 5m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6aecc46066164f53a063d4be394d1d4a : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f9c43806818416a8466ae9f4b10780a : SUCCESS in 5m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b4d5425f5a004d05930715f58bbf13d4 : SUCCESS in 4m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/65dc1da10a634f2883cee9d5f9d6293e : SUCCESS in 4m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5d54e7e97f484252a92ad39123004971 : SUCCESS in 9m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/304215c273a84a78aea1abb255e2cc14 : SUCCESS in 5m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c4155ec45d5f4137bc9789c11f32e64d : SUCCESS in 7m 12s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/adc69bb0b27c44498adcccacb765691b : SUCCESS in 38m 38s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/ed3a3cf9fad84af29e9a779cc1a15537 : SUCCESS in 1h 09m 02s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/f40bc58446714f0e96a2428af1d59e82 : SUCCESS in 1h 00m 32s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/962e4f0c7c284503be1608eae72e84f0 : SUCCESS in 1h 13m 40s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/b90b63eee5d0499b90fdbed615b63230 : SUCCESS in 37m 15s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/036853e06edd4cdb84ba0b5892906814 : SUCCESS in 1h 12m 35s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/bd0ceebec1774cf0b870693344f76605 : SUCCESS in 1h 00m 00s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/3cd19657a1a04229aa437c57df706068 : SUCCESS in 1h 10m 30s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/a5a456959dea441c82bcba24ac3ce405 : SUCCESS in 39m 52s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/54af3b56b7614e1d97ebabd5c501caad : SUCCESS in 1h 04m 46s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/bbe84faf5885443a8d744fc0590504bb : SUCCESS in 56m 05s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/4f3146708b7c4789bf6e0baf43b7f2dc : SUCCESS in 1h 01m 16s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/94d5e431bac345d3a7cb2c398b3ff263 : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":7},{"id":"b70bc189c6a2788fccb8c67b6ca8891ef66c8638","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 19:16:12.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11290e77834741c7b606d5009db2069c\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0a6013e7d89c4515b25f069b6e781eea : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/fbb9b03b34c4450683749cf29c4892d4 : SUCCESS in 7m 41s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b2133874bd1f4699a6b1e42e3db699ea : SUCCESS in 14m 53s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"1cc100cef8086b9c47e30780e8c8403d1611f28f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-08-13 16:31:26.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"8df1411c24c361dce6ee3a3b3edd63da66437810","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 17:09:15.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/273ab388d50c4ece8f0b27c715fe91a0\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9962ce6f15ba42798790952d345deab6 : SUCCESS in 9m 46s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e29461d67e3b4193b8a32b1161d5de14 : SUCCESS in 8m 34s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c710de1745c1409180112f84044d8165 : SUCCESS in 15m 46s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"0c4c01585d365dc3208302a73f865bb619fdfdc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 17:54:42.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ed61e95bbe44bdabee67af3eb8d6a68\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c28e7caf97e243a893335e8272de1200 : SUCCESS in 5m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/00a40352d1b14e4a86b29d5f4d1f1a57 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1073fa00caad4d81b67e6cb86737af41 : SUCCESS in 4m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4641c6276241489ea37fc10b963f81aa : SUCCESS in 4m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e760ef6a5b4b45de801d34789d0e4023 : SUCCESS in 5m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e0cefb8641244f5d894324034306c1a7 : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44ef76a265e7493db0885e0d23fdff47 : SUCCESS in 7m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d0fe6f9eaa143a8877214c8eb9c0a2a : SUCCESS in 7m 18s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/8a15d82d3b3f471184b9e7afc5bbb663 : SUCCESS in 40m 44s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/a13a541d3704414ca80575584396c48f : SUCCESS in 1h 13m 06s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/e689f2e43aef451888ae67e2ac697f9e : SUCCESS in 1h 07m 34s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/ad60a1e9ed904d5582f35df1e9e38e32 : SUCCESS in 1h 21m 53s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/2bed0a4287634617afe6d880d2a6afe2 : SUCCESS in 40m 26s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/5591f3f7cfab493fb14de3290ae1e3de : SUCCESS in 1h 10m 59s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/7b6c28d2530e4b369bd0cde9435eb029 : SUCCESS in 1h 03m 25s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/6734bd5166964e39a3ec875fa8ce470a : SUCCESS in 1h 15m 48s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/46cc41427a474ca48621b74c7c14fd72 : SUCCESS in 44m 57s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/51c5f23506124ccda39af7fbd530b216 : SUCCESS in 1h 06m 00s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/33aec4d7dc2f4fbfa4f17c93c48a3bdb : FAILURE in 37m 22s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/661a3438baf74b66921adf4f994d833e : SUCCESS in 1h 04m 50s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0d1d0d9bd16348f9b5adf5d8db707f17 : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":8},{"id":"23e07dbdaf1b33e38bc3527d3b3d332a13f1184b","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-08-14 13:55:14.000000000","message":"Patch Set 8:\n\n(5 comments)","accounts_in_message":[],"_revision_number":8},{"id":"3c385462db159b58c58a098a847b957dcce866f0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-08-15 12:17:38.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"e73281235c4679a17de887ca798a0af12bb70e9c","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-08-15 12:18:27.000000000","message":"Patch Set 9:\n\n(5 comments)","accounts_in_message":[],"_revision_number":9},{"id":"1981c2d8ced48fef836d3800d5ca4960e4cc3206","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-08-15 12:19:23.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"a3d989657963142c04a7ba4a179590d5b92f4b3e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-15 12:33:19.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7936bc17458f405e8b47a1acec84bef7\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2eaf8358141d4601983c9d0df743d400 : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e77a25c1fa5b4bd4a8442faa13e90774 : FAILURE in 5m 16s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f4e57e8fd5714c5c9e80db74aaace497 : SUCCESS in 13m 30s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"d259f2d7b7c5b963fa27267b93d685a71c7ddaff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-15 15:23:00.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/28136cd574174574abb148583e08fab1\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4e169f5f010842768aaf69e397e8c0c2 : SUCCESS in 4m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce1308fbd48043448638c89b3993deb7 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3496c3461e2242d9bead372f56337849 : SUCCESS in 4m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f70efcab043945e0ae454985cfdeba7a : SUCCESS in 4m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9b9db654485048c69a458564f40da94d : SUCCESS in 4m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f4882586f30440f88d5bc968298c817e : SUCCESS in 7m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44b0bc4cffa3459eac7a52947f4b30f6 : SUCCESS in 4m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c470dbdb74e49368761d2d1fa00260a : SUCCESS in 4m 00s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/3a1e171048e7458cbcdf507e5da0d4d1 : FAILURE in 25m 24s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/f88ac88adf734bf08f04be4bfd7aeb37 : SUCCESS in 1h 10m 53s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/3350ada4a2054e1bb38426643bed2dd8 : FAILURE in 25m 09s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/f30d73f3db0b453cbc7c40164b74078c : FAILURE in 20m 29s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/42238278bb8d441cb9dccff343fb7e8f : SUCCESS in 38m 05s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/82ff942fb3fd486087c1651c32f6c1cd : SUCCESS in 1h 09m 44s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/5fed85568d9a469593c5c88d737b906e : SUCCESS in 58m 10s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/15a28475c5e84dfa9e4ceb4c301effd6 : SUCCESS in 1h 10m 55s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/e831960350fb4ad58c3d12e551811176 : SUCCESS in 40m 13s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/e71c4a7bd298494f80c6c412446629ba : TIMED_OUT in 3h 02m 43s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/227a4c206010443283aa51991275932e : SUCCESS in 1h 02m 06s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/36b7fd29a6c0471a9514e3c4700c947d : SUCCESS in 1h 07m 47s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3f4561df5ea24c6f9e936ecb0b0895db : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":10},{"id":"182a1ed0af1894ab141715c465aac9ffd9464d1a","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-08-17 08:03:09.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"c277b06e177981328aa033b09f10b9eb69432954","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 08:14:44.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3eeeae457ebb4cc8b2e6d715e251aac7\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e777a3268680421383cca5fec73e0fe9 : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d0a57bcbf1794214bb01c3fc90bde154 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/be5c30e7767b4bf198b37fa44a1f3403 : SUCCESS in 8m 23s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"ed6709f31fddd43d0fe146b20e80542f6ab572af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 09:22:14.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/520eacf2a22a4526b84e32cde41a1162\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6e23292f093944cea7dd4d33858dd532 : SUCCESS in 4m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d652def2b644cb8a7b20743b37d78b3 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a10a5aa99ca4327a4b008e235d22294 : SUCCESS in 4m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6f4e8a854032430c8627296e27c17194 : SUCCESS in 4m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b444b6e2310c4ad7a4d84feb08e5f918 : SUCCESS in 4m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6fcb2fd7127c4deb95c6e9258c751cbb : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6df8a47a19a04d96b37eabb0b96deb1c : SUCCESS in 5m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e89b892194041aabc005e03f4b6d6cf : SUCCESS in 3m 22s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/490ce025359640fc8090996b7a4f5074 : SUCCESS in 39m 49s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/62a069a9e1b147fd9483d774f500d5b0 : SUCCESS in 1h 08m 41s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/0aa4443a1e0e4b3eb64303445e704df5 : SUCCESS in 1h 03m 09s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/573d3f080efa4d5cbc0b71973c841234 : FAILURE in 1h 16m 57s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/42784fefe6bd49c19133d704a4800631 : SUCCESS in 38m 53s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/797a129f6283448e9283b8ec126bac05 : SUCCESS in 1h 12m 56s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/bbb1c0e82e864e9a8bd2a2a4a80bbcb0 : SUCCESS in 1h 01m 47s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/e38a6bb08d6d40f0924c13c567e42ac2 : SUCCESS in 1h 12m 27s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/6e35ae0d8d694d16b047520b7cbfd281 : SUCCESS in 42m 05s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/5771cbabf1134364b4de6fecc10b011d : SUCCESS in 1h 04m 20s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/9c62eb72e50940a5bf3190e44413da52 : SUCCESS in 54m 48s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/75606da7dd2c4cea96de66597293317f : SUCCESS in 1h 02m 49s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0ee158dbd8c542e2b72bc8dff6f6da63 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":10},{"id":"5a99a9dc083d953a59cc7bb2adfb2ac1d610a0b9","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-08-17 09:29:37.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"e75761d67c2f19c0f42c67d126e0e17d7ae2db75","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-08-21 06:27:23.000000000","message":"Patch Set 10:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"59c0ce0f4e91ff3a564f61fc1656034d366126cc","revisions":{"f26cb95a725e21eafab2c599fa4f8941fc836bc3":{"kind":"REWORK","_number":1,"created":"2026-07-27 11:23:23.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/82/998782/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/82/998782/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/998782/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/998782/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/998782/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/82/998782/1"}}},"commit":{"parents":[{"commit":"61ec2341004e60b3630fbffc8f56aeb64ef1add1","subject":"Merge \"Install dumb-init via package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/61ec2341004e60b3630fbffc8f56aeb64ef1add1"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-07-27 07:20:53.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-07-27 11:22:45.000000000","tz":300},"subject":"Preparation for supporting special chars","message":"Preparation for supporting special chars\n\nAdd non_empty_utf8_sequence for database name validations.\nUnescape and decode special characters.\n\nChange-Id: Icd0926a4f2e071abe50c7e4c540555530b6a50a9\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f26cb95a725e21eafab2c599fa4f8941fc836bc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f26cb95a725e21eafab2c599fa4f8941fc836bc3"}]},"branch":"refs/heads/master"},"53fe9bc7670a7e1e63fcfd69676eac42904da0a6":{"kind":"REWORK","_number":2,"created":"2026-07-27 11:50:04.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/82/998782/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/82/998782/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/998782/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/998782/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/998782/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/82/998782/2"}}},"commit":{"parents":[{"commit":"61ec2341004e60b3630fbffc8f56aeb64ef1add1","subject":"Merge \"Install dumb-init via package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/61ec2341004e60b3630fbffc8f56aeb64ef1add1"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-07-27 07:20:53.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-07-27 11:49:52.000000000","tz":300},"subject":"Preparation for supporting special chars","message":"Preparation for supporting special chars\n\nAdd non_empty_utf8_sequence for database name validations.\nUnescape and decode special characters.\n\nChange-Id: Icd0926a4f2e071abe50c7e4c540555530b6a50a9\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/53fe9bc7670a7e1e63fcfd69676eac42904da0a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/53fe9bc7670a7e1e63fcfd69676eac42904da0a6"}]},"branch":"refs/heads/master"},"aebbb4055875d3c8e660cd78053d3a8c39b456e4":{"kind":"REWORK","_number":3,"created":"2026-08-11 14:40:03.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/82/998782/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/82/998782/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/998782/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/998782/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/998782/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/82/998782/3"}}},"commit":{"parents":[{"commit":"fff318de6332df36535012be7cda486655c27149","subject":"Merge \"Skip tempest tests when only unit tests are updated\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fff318de6332df36535012be7cda486655c27149"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-07-27 07:20:53.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-08-11 14:39:38.000000000","tz":300},"subject":"API endpoints for supporting special characters","message":"API endpoints for supporting special characters\n\nRemove `{user_name}` and `{database_name}` parameters from URLs by\nintroducing POST-based `/action` endpoints instead.\n\nThe legacy API endpoints are deprecated and will be removed in a\nfuture release.\n\nChange-Id: Icd0926a4f2e071abe50c7e4c540555530b6a50a9\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/aebbb4055875d3c8e660cd78053d3a8c39b456e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/aebbb4055875d3c8e660cd78053d3a8c39b456e4"}]},"branch":"refs/heads/master"},"9dd6d4527c2aea14ff0f13ba2eeb3c2173ac0ce6":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-08-11 14:40:45.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/82/998782/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/82/998782/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/998782/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/998782/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/998782/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/82/998782/4"}}},"commit":{"parents":[{"commit":"d0f365fe95954c6f237eb7bb90f073105191164c","subject":"Merge \"Don\u0027t call guest agent for instances in SHUTDOWN\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d0f365fe95954c6f237eb7bb90f073105191164c"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-07-27 07:20:53.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-08-11 14:40:22.000000000","tz":300},"subject":"API endpoints for supporting special characters","message":"API endpoints for supporting special characters\n\nRemove `{user_name}` and `{database_name}` parameters from URLs by\nintroducing POST-based `/action` endpoints instead.\n\nThe legacy API endpoints are deprecated and will be removed in a\nfuture release.\n\nChange-Id: Icd0926a4f2e071abe50c7e4c540555530b6a50a9\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9dd6d4527c2aea14ff0f13ba2eeb3c2173ac0ce6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9dd6d4527c2aea14ff0f13ba2eeb3c2173ac0ce6"}]},"branch":"refs/heads/master"},"c12e7ea22d7f11c39f3b3ecdf94467c461beb2e5":{"kind":"REWORK","_number":5,"created":"2026-08-11 14:44:47.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/82/998782/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/82/998782/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/998782/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/998782/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/998782/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/82/998782/5"}}},"commit":{"parents":[{"commit":"d0f365fe95954c6f237eb7bb90f073105191164c","subject":"Merge \"Don\u0027t call guest agent for instances in SHUTDOWN\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d0f365fe95954c6f237eb7bb90f073105191164c"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-07-27 07:20:53.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-08-11 14:44:28.000000000","tz":300},"subject":"API endpoints for supporting special characters","message":"API endpoints for supporting special characters\n\nRemove `{user_name}` and `{database_name}` parameters from URLs by\nintroducing POST-based `/action` endpoints instead.\n\nThe legacy API endpoints are deprecated and will be removed in a\nfuture release.\n\nChange-Id: Icd0926a4f2e071abe50c7e4c540555530b6a50a9\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c12e7ea22d7f11c39f3b3ecdf94467c461beb2e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c12e7ea22d7f11c39f3b3ecdf94467c461beb2e5"}]},"branch":"refs/heads/master"},"cfcf5074ee59661eb3756c11dc9fc45bb3f29c5c":{"kind":"REWORK","_number":6,"created":"2026-08-12 17:18:53.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/82/998782/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/82/998782/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/998782/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/998782/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/998782/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/82/998782/6"}}},"commit":{"parents":[{"commit":"d0f365fe95954c6f237eb7bb90f073105191164c","subject":"Merge \"Don\u0027t call guest agent for instances in SHUTDOWN\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d0f365fe95954c6f237eb7bb90f073105191164c"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-07-27 07:20:53.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-08-12 17:18:25.000000000","tz":300},"subject":"API endpoints for supporting special characters","message":"API endpoints for supporting special characters\n\nRemove `{user_name}` and `{database_name}` parameters from URLs by\nintroducing POST-based `/action` endpoints instead.\n\nThe legacy API endpoints are deprecated and will be removed in a\nfuture release.\n\nChange-Id: Icd0926a4f2e071abe50c7e4c540555530b6a50a9\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cfcf5074ee59661eb3756c11dc9fc45bb3f29c5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cfcf5074ee59661eb3756c11dc9fc45bb3f29c5c"}]},"branch":"refs/heads/master"},"83f938efbc31fc57e338bf51140a8dc857bd844f":{"kind":"REWORK","_number":7,"created":"2026-08-12 17:39:12.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/82/998782/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/82/998782/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/998782/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/998782/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/998782/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/82/998782/7"}}},"commit":{"parents":[{"commit":"d0f365fe95954c6f237eb7bb90f073105191164c","subject":"Merge \"Don\u0027t call guest agent for instances in SHUTDOWN\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d0f365fe95954c6f237eb7bb90f073105191164c"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-07-27 07:20:53.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-08-12 17:38:48.000000000","tz":300},"subject":"API endpoints for supporting special characters","message":"API endpoints for supporting special characters\n\nRemove `{user_name}` and `{database_name}` parameters from URLs by\nintroducing POST-based `/action` endpoints instead.\n\nThe legacy API endpoints are deprecated and will be removed in a\nfuture release.\n\nChange-Id: Icd0926a4f2e071abe50c7e4c540555530b6a50a9\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/83f938efbc31fc57e338bf51140a8dc857bd844f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/83f938efbc31fc57e338bf51140a8dc857bd844f"}]},"branch":"refs/heads/master"},"114391a7c80e60d89267c0a4524d9bf79f6dcf59":{"kind":"REWORK","_number":8,"created":"2026-08-13 16:31:26.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/82/998782/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/82/998782/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/998782/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/998782/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/998782/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/82/998782/8"}}},"commit":{"parents":[{"commit":"af8ebb3604d092b522d9376afe50ac01e58f1e31","subject":"Merge \"Build guest image from local repo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/af8ebb3604d092b522d9376afe50ac01e58f1e31"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-07-27 07:20:53.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-08-13 07:01:51.000000000","tz":300},"subject":"API endpoints for supporting special characters","message":"API endpoints for supporting special characters\n\nRemove `{user_name}` and `{database_name}` parameters from URLs by\nintroducing POST-based `/action` endpoints instead.\n\nThe legacy API endpoints are deprecated and will be removed in a\nfuture release.\n\nChange-Id: Icd0926a4f2e071abe50c7e4c540555530b6a50a9\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/114391a7c80e60d89267c0a4524d9bf79f6dcf59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/114391a7c80e60d89267c0a4524d9bf79f6dcf59"}]},"branch":"refs/heads/master"},"d8860c283115931eb0c373b5049c808e359b2d16":{"kind":"REWORK","_number":9,"created":"2026-08-15 12:17:38.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/82/998782/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/82/998782/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/998782/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/998782/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/998782/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/82/998782/9"}}},"commit":{"parents":[{"commit":"4cdc0cdadc76229aaa16195696d3aa2bc295649c","subject":"Merge \"Fix \"no space left on device\" for backup test jobs.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4cdc0cdadc76229aaa16195696d3aa2bc295649c"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-07-27 07:20:53.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-08-15 12:17:11.000000000","tz":300},"subject":"API endpoints for supporting special characters","message":"API endpoints for supporting special characters\n\nRemove `{user_name}` and `{database_name}` parameters from URLs by\nintroducing POST-based `/action` endpoints instead.\n\nThe legacy API endpoints are deprecated and will be removed in a\nfuture release.\n\nChange-Id: Icd0926a4f2e071abe50c7e4c540555530b6a50a9\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d8860c283115931eb0c373b5049c808e359b2d16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d8860c283115931eb0c373b5049c808e359b2d16"}]},"branch":"refs/heads/master"},"59c0ce0f4e91ff3a564f61fc1656034d366126cc":{"kind":"REWORK","_number":10,"created":"2026-08-15 12:19:23.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/82/998782/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/82/998782/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/998782/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/998782/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/998782/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/82/998782/10"}}},"commit":{"parents":[{"commit":"4cdc0cdadc76229aaa16195696d3aa2bc295649c","subject":"Merge \"Fix \"no space left on device\" for backup test jobs.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4cdc0cdadc76229aaa16195696d3aa2bc295649c"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-07-27 07:20:53.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-08-15 12:19:11.000000000","tz":300},"subject":"API endpoints for supporting special characters","message":"API endpoints for supporting special characters\n\nRemove `{user_name}` and `{database_name}` parameters from URLs by\nintroducing POST-based `/action` endpoints instead.\n\nThe legacy API endpoints are deprecated and will be removed in a\nfuture release.\n\nChange-Id: Icd0926a4f2e071abe50c7e4c540555530b6a50a9\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/59c0ce0f4e91ff3a564f61fc1656034d366126cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/59c0ce0f4e91ff3a564f61fc1656034d366126cc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
