)]}'
{"id":"openstack%2Ftrove~998789","triplet_id":"openstack%2Ftrove~master~If9b270a6fe8956092a8bf3590fe045f802c1099d","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"36080":{"account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"last_update":"2026-07-27 12:55:44.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"If9b270a6fe8956092a8bf3590fe045f802c1099d","subject":"[DNM] Test special chars","status":"NEW","created":"2026-07-27 11:46:09.000000000","updated":"2026-07-28 18:06:40.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":5,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"5c492c61023be41596d626378464a20a6974874e","_number":998789,"virtual_id_number":998789,"owner":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-07-28 18:06:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2026-07-27 11:50:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-27 11:50:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"25663c077a0118dc0d0cb963df58363d868511dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-27 11:46:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a3372e31ddd9bcc13b9d1ce8a46f5375c0924354","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-27 11:50:12.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/c/openstack/trove/+/998782, https://review.opendev.org/c/openstack/trove/+/971797, https://review.opendev.org/c/openstack/trove/+/998785 are needed.","accounts_in_message":[],"_revision_number":1},{"id":"02223504f3d8694ce253cbc02f36fd11a0f39b1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-27 11:50:18.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/c/openstack/trove/+/998782, https://review.opendev.org/c/openstack/trove/+/971797, https://review.opendev.org/c/openstack/trove/+/998785 are needed.","accounts_in_message":[],"_revision_number":1},{"id":"682ac485684b47a64518898b3e8730cd43078bab","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-27 12:55:44.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"5ae5bf71c284bbf654d34f8d1c711dd4a3612e39","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-27 16:41:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"790af36b260c44c6a815b0ded1ea52556e612136","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-27 16:57:34.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"0b810f44d05ab5eea690810ce3f24df0bfa62f7e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-27 17:00:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1e2f0d66c718d431c3ffcb354bd408994710eacd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-27 17:38:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a40acc93afc46fba8bd5ca99b0fea43\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/19495efb2ac24db6ac8bfd86d33e5771 : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d1bd3f4e500349f9845f584de5036f0c : SUCCESS in 7m 45s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4eea7472afd643989d9c051fa648fe3d : SUCCESS in 14m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4a0b964eb7dac0d3b700fb100bedd1fff0de4166","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-27 18:35:02.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0744a3764e2c41bb805eea94b582c29b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/054cbcad07e848f3af4ca1ce96d703bb : SUCCESS in 7m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11368930c0784eccba9edeab562f9eef : SUCCESS in 6m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e3bc6fb3bcec4fbf8692a417cc193051 : SUCCESS in 5m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8502b421a61e4f179b0147f2b6e6dc8c : SUCCESS in 5m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/83d9c6dea8514f34b15ebc1a2d02db2c : SUCCESS in 8m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50d0e56d88a54097a63b2bae2bfc25c1 : SUCCESS in 10m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87b682e8dc314c118b842662a749e9bf : SUCCESS in 8m 00s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/56cdd5489cbb482f8a79a6f9368cbeb0 : SUCCESS in 1h 32m 36s\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/374965addcf54c8baa61d756b233ba09 : FAILURE in 1h 09m 15s\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/dbbe16f354734ee0986b2aff27fb1552 : FAILURE in 1h 12m 28s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8b74ab5d76d144f4967956d8794a20dc : SUCCESS in 6m 49s","accounts_in_message":[],"_revision_number":3},{"id":"5727da4cfa8bfa9ec10e96ceb127c60b2ce5057d","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-28 10:10:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ef0b2d661ddc7b4353bf67c3fe9cb8c87a36f165","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-28 10:35:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef2f8c09a2dd46f0942274b19f5a3651\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9dc06622a0354d6291ae8b6673350ff2 : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e213ea8d58254eea8e5d9ac5e9f82c06 : SUCCESS in 7m 32s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d2ab42a7e71047a2b9bc9c0286ffb657 : SUCCESS in 13m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8ea83cc7c745ae724d692c9259b03e68b18e5ba1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-28 10:58:01.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fe18099839584db2bc0837736e3b4508\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a8c8191810ce4828b4210f3c23bf8adf : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/212fdd3002504be59c04bb6798ad1b57 : SUCCESS in 5m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/74833934a0a34fbf8d1aeaf71857ddf7 : SUCCESS in 5m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c02a75fc64f34f1ebbbc8b2a2bad6a83 : SUCCESS in 4m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b85d4a3366ce479997dad89eb600777e : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/773001c1dd484885a68d4cd1de6183cb : SUCCESS in 8m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf9b332ec15f48538880a5783b7e6e3c : SUCCESS in 5m 24s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/6ca4876f47694ad2a8d2ac4661195b79 : SUCCESS in 45m 51s\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/ff0d60d367a84e26a3a28b46c1b7ca45 : SUCCESS in 43m 01s\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/49b7a88483a943ce8d92cca48afc1bda : FAILURE in 43m 06s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/098cf7a288fb4c5ea75f1066839cf855 : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":3},{"id":"1ffa3883289899a938e5143e0a6aa947596b5145","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-28 11:54:59.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4f82f40b78437096ea57356002b9b911fadb6b1e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-28 12:13:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/816efe1bfe1742d38daf6aa576853a55\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/22d6ecbb8a964054ae7b2da6d6f734c3 : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8679af76deef42aebf7c3f6c0837f56d : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/51f18428937f45de8387ada7c7fae457 : SUCCESS in 12m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7a7e690c7337395c8467014b4388ea7647e18f9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-28 12:58:27.000000000","message":"Patch Set 3:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/trove-tempest-plugin/+/998786 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"c4ad45a6aaca6c514cd958ddcaf5b57322a93d04","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-28 12:59:07.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7d73ed0feab4ea83db4dc1ccb6804f73f6adce55","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-28 13:25:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/617f0da5f5da44fd97ea5fbeb36eeb97\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2650b584b993400697ed3260cc66340f : NODE_FAILURE Node(set) request 07d26bc09d6e411f8842af54199a7030 failed in 0s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a32fb2b6d261410384a938a0f8113293 : NODE_FAILURE Node(set) request 5c494a49951c4a848a04533d7fbc2ebe failed in 0s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0d92c874941e46e59e77843e0d4ec1bd : SUCCESS in 15m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b399d577ff4b5c1f448e993003c3bdd29ef4c58a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-28 14:24:38.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1d66f82db00246bd9c099facab8c60c9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2fe7b04739640b18613f1fa95f7f93b : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f914a2064a84490896d5ca21e2ac4252 : SUCCESS in 7m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4b9d763c30924e30833a167400d8c651 : SUCCESS in 3m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e4b33927c4c44bb58b3addfec5482244 : SUCCESS in 3m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c7dd657cb1894e0588d380ff15c79f6b : SUCCESS in 7m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c25e1363842c4a36a41f1015ef920ed2 : SUCCESS in 11m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2f603bdb1a64acbb4de0491231d4d6e : SUCCESS in 9m 53s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/ab021becc80c4de3b9504313326bd714 : SUCCESS in 49m 13s\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/91cfce228320427bab8ad095f182bc82 : SUCCESS in 42m 42s\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/04f3b8b90164427aa1c86d2f4085c9ab : FAILURE in 1h 22m 03s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/23ba1c53ec3540d9b19b2494bc79e127 : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":3},{"id":"490e92d44d16b79c61679f069846ee4aada4306f","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-28 17:04:40.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e2cb7dcad5b9b2787dab4e7703281278b64c86f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-28 17:37:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57b99655d226475783a3acf8fd027498\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/80062646eaa644a3af34e81bf3e8bce6 : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9e1903e216834678bc343c18cc8ace07 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/280d140ddde14d53a759a69e00ee6d40 : SUCCESS in 13m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5c492c61023be41596d626378464a20a6974874e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-28 18:06:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/124bb133cd324dd18355ab0c44d2f3b4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/68daa1654be2462591d13de0241bc141 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29b8bbb20afa41ca91a8e47fd2874eb3 : SUCCESS in 3m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e460a43e7bf84d98946dd74f789f95ca : SUCCESS in 3m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/818cac168cc54a60b3f836656ccab85c : SUCCESS in 4m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1dc91eb291a64ec4a9f039245e4d1633 : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcb17ffdd39a4b9d9b3352733e01ec50 : SUCCESS in 6m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d3532b198a34ad791a349c3ca816d35 : SUCCESS in 6m 05s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/ab48ddc864314f109df796c458a87981 : SUCCESS in 59m 35s\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/9bbbad2226d44a9581b75961ec3ab53a : SUCCESS in 54m 26s\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/1c4129b1b1bc4a4e9f9a704a0de01ea8 : SUCCESS in 58m 07s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1a320d96093e446d9257d317afce2ea0 : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f294fe56b0aab475360892c464b91d8098f85e0f","revisions":{"c5ce1ba8fd285b4299ad186b3588d70ffba8f978":{"kind":"REWORK","_number":1,"created":"2026-07-27 11:46:09.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/89/998789/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/89/998789/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/998789/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/998789/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/998789/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/89/998789/1"}}},"commit":{"parents":[{"commit":"61ec2341004e60b3630fbffc8f56aeb64ef1add1","subject":"Merge \"Install dumb-init via package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/61ec2341004e60b3630fbffc8f56aeb64ef1add1"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-07-27 11:35:43.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-07-27 11:35:43.000000000","tz":300},"subject":"[DNM] Test special chars","message":"[DNM] Test special chars\n\nDepends-On: https://review.opendev.org/c/openstack/trove/+/998782\nDepends-On: https://review.opendev.org/c/openstack/trove/+/971797\nDepends-On: https://review.opendev.org/c/openstack/trove/+/998785\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/998786\nChange-Id: If9b270a6fe8956092a8bf3590fe045f802c1099d\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c5ce1ba8fd285b4299ad186b3588d70ffba8f978"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c5ce1ba8fd285b4299ad186b3588d70ffba8f978"}]},"branch":"refs/heads/master"},"f4dcd84122240ff299f9ccd4aeae92322f9de9a8":{"kind":"REWORK","_number":2,"created":"2026-07-27 16:57:34.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/89/998789/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/89/998789/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/998789/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/998789/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/998789/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/89/998789/2"}}},"commit":{"parents":[{"commit":"61ec2341004e60b3630fbffc8f56aeb64ef1add1","subject":"Merge \"Install dumb-init via package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/61ec2341004e60b3630fbffc8f56aeb64ef1add1"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-07-27 11:35:43.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-07-27 16:57:23.000000000","tz":300},"subject":"[DNM] Test special chars","message":"[DNM] Test special chars\n\nDepends-On: https://review.opendev.org/c/openstack/trove/+/998782\nDepends-On: https://review.opendev.org/c/openstack/trove/+/971797\nDepends-On: https://review.opendev.org/c/openstack/trove/+/998785\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/998786\nChange-Id: If9b270a6fe8956092a8bf3590fe045f802c1099d\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f4dcd84122240ff299f9ccd4aeae92322f9de9a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f4dcd84122240ff299f9ccd4aeae92322f9de9a8"}]},"branch":"refs/heads/master"},"f294fe56b0aab475360892c464b91d8098f85e0f":{"kind":"REWORK","_number":3,"created":"2026-07-27 17:00:28.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/89/998789/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/89/998789/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/998789/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/998789/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/998789/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/89/998789/3"}}},"commit":{"parents":[{"commit":"61ec2341004e60b3630fbffc8f56aeb64ef1add1","subject":"Merge \"Install dumb-init via package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/61ec2341004e60b3630fbffc8f56aeb64ef1add1"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-07-27 11:35:43.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-07-27 17:00:17.000000000","tz":300},"subject":"[DNM] Test special chars","message":"[DNM] Test special chars\n\nDepends-On: https://review.opendev.org/c/openstack/trove/+/998782\nDepends-On: https://review.opendev.org/c/openstack/trove/+/971797\nDepends-On: https://review.opendev.org/c/openstack/trove/+/998785\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/998786\nChange-Id: If9b270a6fe8956092a8bf3590fe045f802c1099d\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f294fe56b0aab475360892c464b91d8098f85e0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f294fe56b0aab475360892c464b91d8098f85e0f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
