)]}'
{"id":"openstack%2Ftrove~998996","triplet_id":"openstack%2Ftrove~master~If46914a1283e44f3ddf260b04369e010dce655ca","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-08-04 16:08:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If46914a1283e44f3ddf260b04369e010dce655ca","subject":"Skip tempest tests when only unit tests are updated","status":"MERGED","created":"2026-07-28 18:05:08.000000000","updated":"2026-08-04 16:10:41.000000000","submitted":"2026-08-04 16:08:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"998996","meta_rev_id":"3091f583954573860ccf14d96f546a45047d5c00","_number":998996,"virtual_id_number":998996,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-04 16:08:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":2,"date":"2026-08-04 14:17:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":1,"date":"2026-08-04 14:17:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":-1,"date":"2026-07-29 15:39:32.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-28 18:59:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-28 21:29:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-29 15:39:32.000000000","updated_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"reviewer":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"state":"REVIEWER"}],"messages":[{"id":"8bee39c52366d977863f582d6463b380facf853d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-28 18:05:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f0f935102e5224429ea07c3adefad58c165e267","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-28 18:27:45.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"9c3c0689a4f65671174431276495881a54a30c60","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-28 18:59:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84017abf5fa34899a45286057786ee40\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7369a3a2090d426f97cb1ae878cafcc6 : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cfc92b6bff2c4ecd81177e8a2e3295ba : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c8491c7dc18e4e93863cd1e28f667c20 : SUCCESS in 13m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2b56b4bf5e839bbfc82455843dfbe935e1816a51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-28 21:29:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ac7d13dd8634863ba48673ff84c4f41\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/133c6d58fe924e1a8cd71ef9b57d0923 : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0085962d5edf4874a8ed73410ec4ad04 : SUCCESS in 6m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/22399fd743d54328bebe3f45b3efbbd7 : SUCCESS in 4m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/403edd1385cb4036b4698d6f10f2d8cb : SUCCESS in 2m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2c0b47d80c344545862a3383a90def09 : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c2bb4c816b4484598abe875da337316 : SUCCESS in 9m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22da09ed7df04ed797bb764df15715e3 : SUCCESS in 4m 05s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/8a2e859230794f5496fd71a041d48780 : SUCCESS in 1h 20m 00s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/2e90e135135f42ffb728b63093718d50 : SUCCESS in 2h 27m 08s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/9132206294de4b90b344200add98aab9 : SUCCESS in 2h 08m 31s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/813d64273aef4a06980b3acc7c277eb1 : SUCCESS in 2h 58m 27s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/185f09763afe4923b8bbea7087797e01 : SUCCESS in 1h 14m 29s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/c4ed5b40edef40eb9f92505a3be90319 : SUCCESS in 2h 09m 29s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/75a8443521ec4b0992dc46ecc2c11b74 : SUCCESS in 1h 50m 50s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/1a8c902e6af14239b2960dcebf26f90e : SUCCESS in 2h 12m 34s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/ba163f8c3bcd40638f3759bb7a27492c : SUCCESS in 1h 19m 00s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/76a9905ef3d14cbc86113a43f70bcf03 : SUCCESS in 1h 28m 43s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/0024064f33814f9aba5de48eafccb481 : SUCCESS in 1h 09m 02s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/01bc4d58aed0451fae80feb6c617f471 : SUCCESS in 1h 33m 38s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/943e1a6ff7864bc6ab16192bd49a7033 : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":2},{"id":"d63650bc1e6baa45f999effd2b53f5a261a43f0f","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-29 15:39:32.000000000","message":"Patch Set 2: Backport-Candidate-1 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c03a25218b4388f1f90224583757f372a7bb2fbe","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-08-04 11:35:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"eba09d951b257012bf2de569fa9ffa2b0c048131","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 11:51:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8da752c1059a4b35a82430528e729151\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ff0e6eaf735944818df8f273c0138b14 : SUCCESS in 7m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9ad01e61430348d79a3f885d0a303ec6 : SUCCESS in 7m 55s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f4d4905861174f8f9048265ab7785115 : SUCCESS in 13m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4b20af2b7882e6ec8a2fd466547b243d7c921f24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 13:37:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a49f9a9e09d420583ca5c94e90748aa\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dbd0a962c48d42e3a36e491f66a9fad5 : SUCCESS in 7m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36b4b0d4770b4a5cb648fdb262220fda : SUCCESS in 4m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/62d2e3df53514b56b2b08d71c247cac1 : SUCCESS in 3m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/29404b880a744402bf093caeb0781da2 : SUCCESS in 4m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/712b56a9ac144b96babfa8aba05e655f : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/326df8eb168e45bfa338291bb6b783cc : SUCCESS in 12m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd0a81733b2043819c7fef43327e2bc5 : SUCCESS in 7m 15s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/c2f984566bb14a6680ffcffa85a8780b : SUCCESS in 1h 02m 52s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/0067e2bffb5f4aaea23667ba67020cee : SUCCESS in 1h 48m 48s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/3519159ba12d4685a7d6a15bd3dd1c65 : SUCCESS in 1h 24m 00s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/37a46b21d2de44e7a1a2fec8b8b673e9 : SUCCESS in 1h 59m 14s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/9f67ac70cbac43c1b715badccf9dabda : SUCCESS in 1h 14m 27s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/4e04bf4dc9854ae7a6907974db84a3af : SUCCESS in 1h 29m 42s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/fc7285686d5e4f7ea53820742baa2e2c : SUCCESS in 1h 16m 37s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/bd36d31a77ca432aa8ae15621d18a760 : SUCCESS in 1h 36m 27s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/04ce476914a641c88a8670a6005479a4 : SUCCESS in 1h 04m 36s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/15494571f2794518950237b6869c20ff : SUCCESS in 1h 52m 26s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/d00af520b8f64012a4ec218759ec637d : SUCCESS in 1h 15m 38s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/75fc588c81924a519b9a972f5a8bed51 : SUCCESS in 2h 00m 12s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c8d5a35b94c8410d8f2fb383fb74f6bb : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":2},{"id":"a9f64d9fc6eb793c7c8ff87b48ae1ba41056e344","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-08-04 14:17:19.000000000","message":"Patch Set 2: -Code-Review -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"9a040f100d483d5fa29ad7c3d3120d6c93e64322","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-08-04 14:17:25.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"29d78e0dbbcaa0f5891c37d8a8b217ebec354f7e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 14:20:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5587762f6cf80ea6f492bf5e14c89bb1d1958d42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 16:08:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c557ce5735140cfa9b9e738e42bd50b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39af0878d71d4f6dbd2f823a9f4cc5fa : SUCCESS in 4m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1850491c61ad42ecb87017b6bfcc2662 : SUCCESS in 4m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b364bcfc69864bd7b2b8e8828f603ffe : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93161b0740ea49ea910dfa1d80024503 : SUCCESS in 8m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb6f4598f985436ab0f2bbb06b3b44fd : SUCCESS in 5m 57s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/c7b5d344c77c41f3951a353c8a66f9d7 : SUCCESS in 50m 11s\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/6f181cf3b1a24d0b97c3424b2f5d6ef0 : SUCCESS in 44m 48s\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/3abb4f48bbf74cfa99b4084c678bfffc : SUCCESS in 1h 47m 47s","accounts_in_message":[],"_revision_number":2},{"id":"de49dbc4266cb134d3ab72a61ebb1421fb50d3cd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 16:08:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"3091f583954573860ccf14d96f546a45047d5c00","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 16:10:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0203ba231b01491abd1fd96acec3cfbb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/357622e1123b4477867ef01556ba9227 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5741d95bbcbeb049f5b979abda365d8a0b2b60a8","revisions":{"75cec40c78650f3cffb98a5dd1474c73c3d4c82c":{"kind":"REWORK","_number":1,"created":"2026-07-28 18:05:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/998996/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/96/998996/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/96/998996/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/96/998996/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/96/998996/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/96/998996/1"}}},"commit":{"parents":[{"commit":"9019ff423fa05c8404d7ed973633197abe44e3d1","subject":"Do not skip tempest tests for CI config update","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9019ff423fa05c8404d7ed973633197abe44e3d1"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-28 18:04:18.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-28 18:04:18.000000000","tz":540},"subject":"Skip tempest tests when only unit tests are updated","message":"Skip tempest tests when only unit tests are updated\n\n... to optimize CI resource usage.\n\nAlso ignore the whole tests dirctory for bandit check because\nthe tests directory is excluded from the check target.\n\nChange-Id: If46914a1283e44f3ddf260b04369e010dce655ca\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/75cec40c78650f3cffb98a5dd1474c73c3d4c82c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/75cec40c78650f3cffb98a5dd1474c73c3d4c82c"}]},"branch":"refs/heads/master"},"5741d95bbcbeb049f5b979abda365d8a0b2b60a8":{"kind":"NO_CHANGE","_number":2,"created":"2026-07-28 18:27:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/998996/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/96/998996/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/96/998996/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/96/998996/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/96/998996/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/96/998996/2"}}},"commit":{"parents":[{"commit":"e73e9f696330254d5608f015d8a4e85c9b3a5940","subject":"Do not skip tempest tests for CI config update","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e73e9f696330254d5608f015d8a4e85c9b3a5940"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-28 18:04:18.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-28 18:27:45.000000000","tz":0},"subject":"Skip tempest tests when only unit tests are updated","message":"Skip tempest tests when only unit tests are updated\n\n... to optimize CI resource usage.\n\nAlso ignore the whole tests dirctory for bandit check because\nthe tests directory is excluded from the check target.\n\nChange-Id: If46914a1283e44f3ddf260b04369e010dce655ca\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5741d95bbcbeb049f5b979abda365d8a0b2b60a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5741d95bbcbeb049f5b979abda365d8a0b2b60a8"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"9019ff423fa05c8404d7ed973633197abe44e3d1","ours":"75cec40c78650f3cffb98a5dd1474c73c3d4c82c","theirs":"e73e9f696330254d5608f015d8a4e85c9b3a5940","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
