)]}'
{"id":"openstack%2Fui-cookiecutter~630203","triplet_id":"openstack%2Fui-cookiecutter~master~Ibb84c16106abfd4a9d264bdf038b9916e2d1a33a","project":"openstack/ui-cookiecutter","branch":"master","hashtags":[],"change_id":"Ibb84c16106abfd4a9d264bdf038b9916e2d1a33a","subject":"Update home-page","status":"NEW","created":"2019-01-11 12:40:29.000000000","updated":"2019-11-06 11:47:42.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7c0e695159bb266df98e3406f493f84604d3573c","_number":630203,"virtual_id_number":630203,"owner":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-11-06 11:47:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"all":[{"value":-2,"date":"2019-11-06 11:31:28.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-01-11 13:00:26.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2019-07-31 16:40:50.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-11-06 11:47:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a10f5141bf976afc1bd2fedcbb5cf689c4da792c","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-01-11 12:40:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e87d4b443e8e86eb602632878a58d0a0c997121e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-11 12:55:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- ui-cookiecutter-tox-checkbuild http://logs.openstack.org/03/630203/1/check/ui-cookiecutter-tox-checkbuild/d528fbc/ : SUCCESS in 14m 08s","accounts_in_message":[],"_revision_number":1},{"id":"a7f4c7fae582ec32df04884651d285b65b2cf69e","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-01-11 13:00:26.000000000","message":"Patch Set 1: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6279aca901a60f5915402384e9e734b11b256aff","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-07-31 16:40:50.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"d23eb214952ee0b4cc2b0a835056d9f39022a3f7","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-11-06 11:31:28.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"7c0e695159bb266df98e3406f493f84604d3573c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 11:47:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- ui-cookiecutter-tox-checkbuild https://zuul.opendev.org/t/openstack/build/f5867be720394dfaa76a04640c9ffb66 : SUCCESS in 14m 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d6503648d839de508078bfaaee4c9ec17f83d727","revisions":{"e91842a6f13c95d08204996cb84d54feac1b28a6":{"kind":"REWORK","_number":1,"created":"2019-01-11 12:40:29.000000000","uploader":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"ref":"refs/changes/03/630203/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ui-cookiecutter","ref":"refs/changes/03/630203/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/03/630203/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/03/630203/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/03/630203/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ui-cookiecutter refs/changes/03/630203/1"}}},"commit":{"parents":[{"commit":"af2835fd4f8a0de46007199b9afcfb6bb381fce8","subject":"Update contact address to openstack-discuss ML","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/af2835fd4f8a0de46007199b9afcfb6bb381fce8"}]}],"author":{"name":"huang.zhiping","email":"huang.zhiping@99cloud.net","date":"2019-01-11 12:40:16.000000000","tz":0},"committer":{"name":"huang.zhiping","email":"huang.zhiping@99cloud.net","date":"2019-01-11 12:40:28.000000000","tz":0},"subject":"Update home-page","message":"Update home-page\n\nChange-Id: Ibb84c16106abfd4a9d264bdf038b9916e2d1a33a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/e91842a6f13c95d08204996cb84d54feac1b28a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/e91842a6f13c95d08204996cb84d54feac1b28a6"}]},"branch":"refs/heads/master"},"d6503648d839de508078bfaaee4c9ec17f83d727":{"kind":"REWORK","_number":2,"created":"2019-11-06 11:31:28.000000000","uploader":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"ref":"refs/changes/03/630203/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ui-cookiecutter","ref":"refs/changes/03/630203/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/03/630203/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/03/630203/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/03/630203/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ui-cookiecutter refs/changes/03/630203/2"}}},"commit":{"parents":[{"commit":"af2835fd4f8a0de46007199b9afcfb6bb381fce8","subject":"Update contact address to openstack-discuss ML","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/af2835fd4f8a0de46007199b9afcfb6bb381fce8"}]}],"author":{"name":"huang.zhiping","email":"huang.zhiping@99cloud.net","date":"2019-01-11 12:40:16.000000000","tz":0},"committer":{"name":"Andriy Shevchenko","email":"huang.zhiping@99cloud.net","date":"2019-11-06 11:31:24.000000000","tz":0},"subject":"Update home-page","message":"Update home-page\n\nChange-Id: Ibb84c16106abfd4a9d264bdf038b9916e2d1a33a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/d6503648d839de508078bfaaee4c9ec17f83d727"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/d6503648d839de508078bfaaee4c9ec17f83d727"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
