)]}'
{"id":"openstack%2Fui-cookiecutter~719334","triplet_id":"openstack%2Fui-cookiecutter~master~I1106ad9f78be7c2ba0fc8413252e3ef051d752b5","project":"openstack/ui-cookiecutter","branch":"master","topic":"ensure-repositories","hashtags":[],"change_id":"I1106ad9f78be7c2ba0fc8413252e3ef051d752b5","subject":"Use ensure-* roles","status":"MERGED","created":"2020-04-12 13:30:52.000000000","updated":"2020-04-14 00:01:21.000000000","submitted":"2020-04-14 00:01:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"719334-1586822481688-b16e9fe7","meta_rev_id":"109e9af104a6ecc86075fd19f25c26365e8638da","_number":719334,"virtual_id_number":719334,"owner":{"_account_id":29671,"name":"Albin Vass","email":"opendev@albinvass.com","username":"albin_vass"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"date":"2020-04-13 14:25:02.000000000","_account_id":29671,"name":"Albin Vass","email":"opendev@albinvass.com","username":"albin_vass"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-14 00:01:21.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},"all":[{"value":2,"date":"2020-04-13 23:42:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},{"value":1,"date":"2020-04-13 13:46:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":29671,"name":"Albin Vass","email":"opendev@albinvass.com","username":"albin_vass"},{"value":0,"date":"2020-04-13 15:11:39.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},"all":[{"value":1,"date":"2020-04-13 23:42:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":29671,"name":"Albin Vass","email":"opendev@albinvass.com","username":"albin_vass"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29671,"name":"Albin Vass","email":"opendev@albinvass.com","username":"albin_vass"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-13 13:46:57.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-04-13 15:11:39.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-04-13 23:42:32.000000000","updated_by":{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},"reviewer":{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},"state":"REVIEWER"},{"updated":"2020-04-14 00:01:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"be3fcc79cc4f8056debfea318608925e87b89f99","author":{"_account_id":29671,"name":"Albin Vass","email":"opendev@albinvass.com","username":"albin_vass"},"date":"2020-04-12 13:30:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4774a25cab860085e22371f868ed2fbcea231d03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-12 13:49:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- ui-cookiecutter-tox-checkbuild https://zuul.opendev.org/t/openstack/build/45212d156a104cca858969c25b687f54 : FAILURE in 14m 41s","accounts_in_message":[],"_revision_number":1},{"id":"9cf2c25e1e4f5250d189ead3b67339a1f3342645","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-12 13:55:21.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a349c23b004520b9c5309cde910895799034aade","author":{"_account_id":29671,"name":"Albin Vass","email":"opendev@albinvass.com","username":"albin_vass"},"date":"2020-04-13 07:31:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"be58f3433e77853ffb28a00413fdd71065f0f75f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 07:48:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- ui-cookiecutter-tox-checkbuild https://zuul.opendev.org/t/openstack/build/e1a26e1929764a958ea59174e0e08c08 : FAILURE in 16m 56s","accounts_in_message":[],"_revision_number":1},{"id":"07233b5b115b3df29eb59fbf57107725dd6b21d2","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-13 08:41:48.000000000","message":"Patch Set 1:\n\nFixing in https://review.opendev.org/719411","accounts_in_message":[],"_revision_number":1},{"id":"2d5214656e17e558aa55340c2c0649ff430b5dd3","author":{"_account_id":29671,"name":"Albin Vass","email":"opendev@albinvass.com","username":"albin_vass"},"date":"2020-04-13 13:00:04.000000000","message":"Patch Set 1:\n\nAh thanks, the error didn\u0027t look relevant to the change so I didn\u0027t look deeper into it.","accounts_in_message":[],"_revision_number":1},{"id":"ebc923b3181037954b29e298d4e1b9bd97866f76","author":{"_account_id":29671,"name":"Albin Vass","email":"opendev@albinvass.com","username":"albin_vass"},"date":"2020-04-13 13:01:02.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"0ca11f594c073351569bc38a639c69a594a9768b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 13:19:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- ui-cookiecutter-tox-checkbuild https://zuul.opendev.org/t/openstack/build/7bd4097e3a244ec7913c19fa3f628fb9 : SUCCESS in 17m 20s","accounts_in_message":[],"_revision_number":2},{"id":"5cffc742162bd1e79ac4292e50be93c417602018","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-04-13 13:36:09.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bde4508f922da5c36d6392596f23d9e90bb7f018","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-13 13:46:57.000000000","message":"Patch Set 2: Code-Review+1\n\nIvan, depends-on works as well and is just fine IMHO.","accounts_in_message":[],"_revision_number":2},{"id":"4ee3d241bd3c002d3a0c14f592e7192da6e7fdae","author":{"_account_id":29671,"name":"Albin Vass","email":"opendev@albinvass.com","username":"albin_vass"},"date":"2020-04-13 14:25:02.000000000","message":"Patch Set 2:\n\nI think depends-on is handled a bit better in zuul as well since it doesn\u0027t care about patch sets.","accounts_in_message":[],"_revision_number":2},{"id":"311829e856a987e451eaba5e3389d00b85c3df8c","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-04-13 15:11:39.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"4e2807491ba891c3fc7e7768f0cc89b751c6859f","author":{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},"date":"2020-04-13 23:42:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"177733b758efc6c0d5e60a9bf850e22f79784602","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 23:42:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5da89c02dd796cf640e244358c4d930d34c9abe6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 00:01:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"8ee70dddbc9f15c59f3052e48d559859db3290d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 00:01:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- ui-cookiecutter-tox-checkbuild https://zuul.opendev.org/t/openstack/build/a27f7ea5afe24b9e9ac8fcdbf8698042 : SUCCESS in 14m 25s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"653ed276148fdda49dcbf6e02261cbd750bdb297","revisions":{"cc1b42200cf835d0a97175251349a16233ce94d0":{"kind":"REWORK","_number":1,"created":"2020-04-12 13:30:52.000000000","uploader":{"_account_id":29671,"name":"Albin Vass","email":"opendev@albinvass.com","username":"albin_vass"},"ref":"refs/changes/34/719334/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ui-cookiecutter","ref":"refs/changes/34/719334/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/34/719334/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/34/719334/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/34/719334/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ui-cookiecutter refs/changes/34/719334/1"}}},"commit":{"parents":[{"commit":"f655ea66c81d5487b8c6698f1080ff3ce74ae420","subject":"Use Horizon project template for django jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/f655ea66c81d5487b8c6698f1080ff3ce74ae420"}]}],"author":{"name":"vass","email":"albin.vass@gmail.com","date":"2020-04-12 13:30:22.000000000","tz":120},"committer":{"name":"vass","email":"albin.vass@gmail.com","date":"2020-04-12 13:30:49.000000000","tz":120},"subject":"Use ensure-* roles","message":"Use ensure-* roles\n\nThe install-* roles in zuul-jobs have been renamed to ensure-*, follow\nthe rename.\n\nChange-Id: I1106ad9f78be7c2ba0fc8413252e3ef051d752b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/cc1b42200cf835d0a97175251349a16233ce94d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/cc1b42200cf835d0a97175251349a16233ce94d0"}]},"branch":"refs/heads/master"},"653ed276148fdda49dcbf6e02261cbd750bdb297":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-04-13 13:01:02.000000000","uploader":{"_account_id":29671,"name":"Albin Vass","email":"opendev@albinvass.com","username":"albin_vass"},"ref":"refs/changes/34/719334/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ui-cookiecutter","ref":"refs/changes/34/719334/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/34/719334/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/34/719334/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/34/719334/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ui-cookiecutter refs/changes/34/719334/2"}}},"commit":{"parents":[{"commit":"f655ea66c81d5487b8c6698f1080ff3ce74ae420","subject":"Use Horizon project template for django jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/f655ea66c81d5487b8c6698f1080ff3ce74ae420"}]}],"author":{"name":"vass","email":"albin.vass@gmail.com","date":"2020-04-12 13:30:22.000000000","tz":120},"committer":{"name":"vass","email":"albin.vass@gmail.com","date":"2020-04-13 13:00:44.000000000","tz":120},"subject":"Use ensure-* roles","message":"Use ensure-* roles\n\nThe install-* roles in zuul-jobs have been renamed to ensure-*, follow\nthe rename.\n\nDepends-On: https://review.opendev.org/#/c/719411\nChange-Id: I1106ad9f78be7c2ba0fc8413252e3ef051d752b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/653ed276148fdda49dcbf6e02261cbd750bdb297"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/653ed276148fdda49dcbf6e02261cbd750bdb297"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
