)]}'
{"id":"openstack%2Fui-cookiecutter~719411","triplet_id":"openstack%2Fui-cookiecutter~master~Icde1a7be481a56ca2d0c64a651ad7d2a7f895441","project":"openstack/ui-cookiecutter","branch":"master","topic":"drop-py27","hashtags":[],"change_id":"Icde1a7be481a56ca2d0c64a651ad7d2a7f895441","subject":"Drop py27/py35","status":"MERGED","created":"2020-04-13 08:41:30.000000000","updated":"2020-04-13 23:55:30.000000000","submitted":"2020-04-13 23:55:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"719411-1586822130275-11ac1cc1","meta_rev_id":"9491e2f945e2eabe6094182c64df56d95cf8c323","_number":719411,"virtual_id_number":719411,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-13 23:55:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},"all":[{"value":2,"date":"2020-04-13 23:40:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},{"value":2,"date":"2020-04-13 10:03:04.000000000","_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},"all":[{"value":1,"date":"2020-04-13 23:40:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-13 10:03:04.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-04-13 23:40:42.000000000","updated_by":{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},"reviewer":{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},"state":"REVIEWER"},{"updated":"2020-04-13 23:55:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4b8af0a23afa62e68c2845c8f3465d5c591de777","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-13 08:41:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7ff058c88acc56fe26554b01a68e5f3641a6aa39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 09:04:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- ui-cookiecutter-tox-checkbuild https://zuul.opendev.org/t/openstack/build/917550709d1e4f61a4bd492fea40b255 : SUCCESS in 17m 59s","accounts_in_message":[],"_revision_number":1},{"id":"75ca064d0e4b47ef99a1ba9708f75e1910309ab9","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-04-13 10:03:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1826f1f72ef67b88ea79feba919ae9a2cfa6a6ce","author":{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},"date":"2020-04-13 23:40:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7077cbe64902c035768b91e02c678787244281fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 23:40:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d0465b0fba0f283bf29f32df743e5650e7949fc6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 23:55:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f24ebbb6d5922bda597035773b45c7fe2050e6e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 23:55:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- ui-cookiecutter-tox-checkbuild https://zuul.opendev.org/t/openstack/build/cd54af50546d4ce7ab8bdbbc92991c26 : SUCCESS in 13m 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"32ec8b47b1ad0e5c51b8bb2b9ffc455537efeac2","revisions":{"32ec8b47b1ad0e5c51b8bb2b9ffc455537efeac2":{"kind":"REWORK","_number":1,"created":"2020-04-13 08:41:30.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/11/719411/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ui-cookiecutter","ref":"refs/changes/11/719411/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/11/719411/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/11/719411/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/11/719411/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ui-cookiecutter refs/changes/11/719411/1"}}},"commit":{"parents":[{"commit":"f655ea66c81d5487b8c6698f1080ff3ce74ae420","subject":"Use Horizon project template for django jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/f655ea66c81d5487b8c6698f1080ff3ce74ae420"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-13 08:34:59.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-13 08:41:25.000000000","tz":120},"subject":"Drop py27/py35","message":"Drop py27/py35\n\nDrop py27 and py35 testing, for Ussuri those are dropped already.\nUpdate for current django 22 testing.\nRemove py27 cruft.\n\nChange-Id: Icde1a7be481a56ca2d0c64a651ad7d2a7f895441\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/32ec8b47b1ad0e5c51b8bb2b9ffc455537efeac2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/32ec8b47b1ad0e5c51b8bb2b9ffc455537efeac2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
