)]}'
{"id":"openstack%2Fui-cookiecutter~836997","triplet_id":"openstack%2Fui-cookiecutter~master~If4bae3fde1f29eeb1d4d43942acc21ff2ef19482","project":"openstack/ui-cookiecutter","branch":"master","topic":"fix-checkbuild","attention_set":{},"removed_from_attention_set":{"29313":{"account":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"last_update":"2022-04-14 11:53:33.000000000","reason":"Vishal Manchanda replied on the change"},"1736":{"account":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"last_update":"2022-04-20 15:23:18.000000000","reason":"Ivan Kolodyazhny replied on the change"},"841":{"account":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"last_update":"2022-04-20 15:42:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If4bae3fde1f29eeb1d4d43942acc21ff2ef19482","subject":"Fix tox-checkbuild CI job","status":"MERGED","created":"2022-04-07 16:33:49.000000000","updated":"2022-04-20 15:42:16.000000000","submitted":"2022-04-20 15:42:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"836997-fix-checkbuild","meta_rev_id":"fbe0dff31e0894d0a35210931f92fc683cf5edfa","_number":836997,"virtual_id_number":836997,"owner":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-20 15:42:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":2,"date":"2022-04-14 11:53:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":2,"date":"2022-04-20 15:23:18.000000000","_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":1,"date":"2022-04-20 15:23:18.000000000","_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-07 16:41:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-12 02:21:24.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2022-04-12 02:21:24.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"}],"messages":[{"id":"3acdfa59551ad8d3a3d9e0b1af3ac2ffe8b7e043","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2022-04-07 16:33:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a013536872c43193f618544021cbf3a929089abd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 16:41:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- ui-cookiecutter-tox-checkbuild https://zuul.opendev.org/t/openstack/build/a046fea517174549b0df6d2a00d07c5f : FAILURE in 6m 27s","accounts_in_message":[],"_revision_number":1},{"id":"c2c8928b0b7cf9b9acad3a2fe3cf6eacd10d9353","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2022-04-07 16:45:56.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"ae60861951ef8347c8be8c089d5b12b8ae47fcb7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2022-04-07 16:54:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b649f0f661c2e39b15cbc04cf78c703d87bff08b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 17:11:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- ui-cookiecutter-tox-checkbuild https://zuul.opendev.org/t/openstack/build/10f5fd5297d74aae92c5286a25739419 : FAILURE in 16m 03s","accounts_in_message":[],"_revision_number":2},{"id":"2cfe7477e40fff16361e1c91baeeccd5bd427d6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2022-04-07 17:35:00.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"3404f02101d1c509f2d41e6ed2238ad9afbd3b2c","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2022-04-07 17:36:38.000000000","message":"Patch Set 3: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"75c312e79a74e2177df5ff4ee16b3737949ca3c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 17:51:19.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- ui-cookiecutter-tox-checkbuild https://zuul.opendev.org/t/openstack/build/ef3636db0a68472bb0654a5cbd45299a : FAILURE in 10m 57s","accounts_in_message":[],"_revision_number":3},{"id":"c5cb2173b10cff4feb2801014948855c230e91d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2022-04-08 16:12:47.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a0156f15a000c7cc8631bf67df83291bec3521df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 16:34:03.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- ui-cookiecutter-tox-checkbuild https://zuul.opendev.org/t/openstack/build/fee8b0534f8a454a9db5aeaaa4caa4df : FAILURE in 15m 44s","accounts_in_message":[],"_revision_number":4},{"id":"8a0e4b738feaf3c1c12124b83d25783432ef68f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2022-04-12 00:46:53.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1665751d4d0550668a2b7028560e1fc10d0e5b10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 01:13:59.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- ui-cookiecutter-tox-checkbuild https://zuul.opendev.org/t/openstack/build/32354eb2e9c34ef583829778404ffdf9 : SUCCESS in 22m 20s","accounts_in_message":[],"_revision_number":5},{"id":"3368ded81b850b3c5372f72338abbacd64cdafd9","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2022-04-12 02:21:07.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"23c117f6c65e96ff7d12cb47044035056c766718","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2022-04-14 11:53:33.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c16a3aa2f9eca6eb581e01165b11f587dc65e6ce","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2022-04-20 15:23:18.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f349535564f95ae771c94875529e7e35e6e0ca66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 15:24:04.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"4f04b1fbda79af86be4fa22ada028848aa8f6442","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 15:42:15.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- ui-cookiecutter-tox-checkbuild https://zuul.opendev.org/t/openstack/build/0829cb124b174e749388d3b90f12c36b : SUCCESS in 16m 41s","accounts_in_message":[],"_revision_number":5},{"id":"fbe0dff31e0894d0a35210931f92fc683cf5edfa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 15:42:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"8390e9d6d2feb933d9e153b2ef045fbda63197e6","revisions":{"39051d1e03ccdb5565a12cb7052ac59ee5c4893c":{"kind":"REWORK","_number":1,"created":"2022-04-07 16:33:49.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/97/836997/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ui-cookiecutter","ref":"refs/changes/97/836997/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/97/836997/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/97/836997/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/97/836997/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ui-cookiecutter refs/changes/97/836997/1"}}},"commit":{"parents":[{"commit":"3b5d1a5de3090b1014b201668af87bb108f06636","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/3b5d1a5de3090b1014b201668af87bb108f06636"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2022-04-07 16:27:11.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2022-04-07 16:33:33.000000000","tz":540},"subject":"Fix tox-checkbuild CI job","message":"Fix tox-checkbuild CI job\n\nplaybooks/checkbuild/pre.yaml calls ensure-nodejs role, but the default\nnodejs version defined in ensure-nodejs role is still 6. nodejs6 repo is\nno longer available. This is the cause of the job failure.\nLet\u0027s specify nodejs_version used in horizon by default.\n\nChange-Id: If4bae3fde1f29eeb1d4d43942acc21ff2ef19482\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/39051d1e03ccdb5565a12cb7052ac59ee5c4893c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/39051d1e03ccdb5565a12cb7052ac59ee5c4893c"}]},"branch":"refs/heads/master"},"a0c2bdc6531b874f1b6f2428c7ee152e67e3ed38":{"kind":"REWORK","_number":2,"created":"2022-04-07 16:54:31.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/97/836997/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ui-cookiecutter","ref":"refs/changes/97/836997/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/97/836997/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/97/836997/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/97/836997/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ui-cookiecutter refs/changes/97/836997/2"}}},"commit":{"parents":[{"commit":"3b5d1a5de3090b1014b201668af87bb108f06636","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/3b5d1a5de3090b1014b201668af87bb108f06636"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2022-04-07 16:27:11.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2022-04-07 16:54:20.000000000","tz":540},"subject":"Fix tox-checkbuild CI job","message":"Fix tox-checkbuild CI job\n\nplaybooks/checkbuild/pre.yaml calls ensure-nodejs role, but the default\nnodejs version defined in ensure-nodejs role is still 6. nodejs6 repo is\nno longer available. This is the cause of the job failure.\nLet\u0027s specify nodejs_version used in horizon by default.\n\nChange-Id: If4bae3fde1f29eeb1d4d43942acc21ff2ef19482\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/a0c2bdc6531b874f1b6f2428c7ee152e67e3ed38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/a0c2bdc6531b874f1b6f2428c7ee152e67e3ed38"}]},"branch":"refs/heads/master"},"b93d6aa714cec5865fae4456cfc2e295e8974568":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-04-07 17:35:00.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/97/836997/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ui-cookiecutter","ref":"refs/changes/97/836997/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/97/836997/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/97/836997/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/97/836997/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ui-cookiecutter refs/changes/97/836997/3"}}},"commit":{"parents":[{"commit":"3b5d1a5de3090b1014b201668af87bb108f06636","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/3b5d1a5de3090b1014b201668af87bb108f06636"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2022-04-07 16:27:11.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2022-04-07 17:32:32.000000000","tz":540},"subject":"Fix tox-checkbuild CI job","message":"Fix tox-checkbuild CI job\n\n[Changes in zuul configurations]\n- playbooks/checkbuild/pre.yaml calls ensure-nodejs role, but the default\n  nodejs version defined in ensure-nodejs role is still 6. nodejs6 repo\n  is no longer available. This is the cause of the job failure.\n  Let\u0027s specify nodejs_version used in horizon by default.\n- Ensure to install tox globally in the system.\n  tox -e checkbuild calls \"tox\" as the test commands,\n  so it needs to be installed globally.\n\nWIP:\nMore UT fixes are required to pass the CI....\nI hope someone take it over.\n\nChange-Id: If4bae3fde1f29eeb1d4d43942acc21ff2ef19482\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/b93d6aa714cec5865fae4456cfc2e295e8974568"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/b93d6aa714cec5865fae4456cfc2e295e8974568"}]},"branch":"refs/heads/master"},"cafbfa408f3d39bbc3349371ed6dfd879118d9d6":{"kind":"REWORK","_number":4,"created":"2022-04-08 16:12:47.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/97/836997/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ui-cookiecutter","ref":"refs/changes/97/836997/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/97/836997/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/97/836997/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/97/836997/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ui-cookiecutter refs/changes/97/836997/4"}}},"commit":{"parents":[{"commit":"3b5d1a5de3090b1014b201668af87bb108f06636","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/3b5d1a5de3090b1014b201668af87bb108f06636"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2022-04-07 16:27:11.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2022-04-08 15:58:56.000000000","tz":540},"subject":"Fix tox-checkbuild CI job","message":"Fix tox-checkbuild CI job\n\n[Changes in zuul configurations]\n- playbooks/checkbuild/pre.yaml calls ensure-nodejs role, but the default\n  nodejs version defined in ensure-nodejs role is still 6. nodejs6 repo\n  is no longer available. This is the cause of the job failure.\n  Let\u0027s specify nodejs_version used in horizon by default.\n- Ensure to install tox globally in the system.\n  tox -e checkbuild calls \"tox\" as the test commands,\n  so it needs to be installed globally.\n- Specify \"python_version\" so that zuul ensures to install a required\n  version of python interpreter.\n\n[Changes in cookiecutter template]\n- hacking version in test-requirements.txt was too old.\n  It is updated to use more recent version which horizon uses.\n- Update python version to 3.9 in tox.ini.\n- More UT fixes are required to pass the CI....\n\nChange-Id: If4bae3fde1f29eeb1d4d43942acc21ff2ef19482\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/cafbfa408f3d39bbc3349371ed6dfd879118d9d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/cafbfa408f3d39bbc3349371ed6dfd879118d9d6"}]},"branch":"refs/heads/master"},"8390e9d6d2feb933d9e153b2ef045fbda63197e6":{"kind":"REWORK","_number":5,"created":"2022-04-12 00:46:53.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/97/836997/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ui-cookiecutter","ref":"refs/changes/97/836997/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/97/836997/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/97/836997/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/97/836997/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ui-cookiecutter refs/changes/97/836997/5"}}},"commit":{"parents":[{"commit":"3b5d1a5de3090b1014b201668af87bb108f06636","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/3b5d1a5de3090b1014b201668af87bb108f06636"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2022-04-07 16:27:11.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2022-04-12 00:46:36.000000000","tz":540},"subject":"Fix tox-checkbuild CI job","message":"Fix tox-checkbuild CI job\n\n[Changes in zuul configurations]\n- playbooks/checkbuild/pre.yaml calls ensure-nodejs role, but the default\n  nodejs version defined in ensure-nodejs role is still 6. nodejs6 repo\n  is no longer available. This is the cause of the job failure.\n  Let\u0027s specify nodejs_version used in horizon by default.\n- Ensure to install tox globally in the system.\n  tox -e checkbuild calls \"tox\" as the test commands,\n  so it needs to be installed globally.\n- Specify \"python_version\" so that zuul ensures to install a required\n  version of python interpreter.\n\n[Changes in cookiecutter template]\n- hacking version in test-requirements.txt was too old.\n  It is updated to use more recent version which horizon uses.\n- Update python version to 3.9 in tox.ini.\n- Install horizon and its dependencies separately.\n  The constraints file contains horizon, so installing horizon along with\n  the constraints file results in a version conflict error.\n\nChange-Id: If4bae3fde1f29eeb1d4d43942acc21ff2ef19482\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/8390e9d6d2feb933d9e153b2ef045fbda63197e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/8390e9d6d2feb933d9e153b2ef045fbda63197e6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
