)]}'
{"id":"openstack%2Fui-cookiecutter~865953","triplet_id":"openstack%2Fui-cookiecutter~master~Ia8ce252bf7ad6e5fb5cb92214c7d91d99d9ca8f8","project":"openstack/ui-cookiecutter","branch":"master","topic":"865953","attention_set":{},"removed_from_attention_set":{"29313":{"account":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"last_update":"2025-12-02 14:25:39.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ia8ce252bf7ad6e5fb5cb92214c7d91d99d9ca8f8","subject":"[DNM] Test ui-cookiecutter CI","status":"ABANDONED","created":"2022-11-29 08:32:09.000000000","updated":"2025-12-02 14:25:39.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ba8b9a5e6d071a92ee4c2a2752196ed42dfe2bcf","_number":865953,"virtual_id_number":865953,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-12-07 04:06:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-29 08:52:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"785cfb2a1d90aeea75dee928a10ca70e1acb44e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2022-11-29 08:32:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ab4e104ae52796cabee0ffce8c3dd60b03026d24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 08:52:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b2d14e17a3254c7bb6483d0bec300adf\n\n- ui-cookiecutter-tox-checkbuild https://zuul.opendev.org/t/openstack/build/6ee9fce9fc9b4b8f9dbaa74be1e3fdc2 : RETRY_LIMIT in 1m 33s","accounts_in_message":[],"_revision_number":1},{"id":"29f48486e638b556f52a11cd3986866fff2d1be9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2022-11-29 09:54:24.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"e138e1f2c264dadcea90db5d412c354d4c43bc48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 10:11:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/19a4eb8f32874f77a448dd659bc6635b\n\n- ui-cookiecutter-tox-checkbuild https://zuul.opendev.org/t/openstack/build/d984d6c7a1e94775aab9457b3ea606af : RETRY_LIMIT in 3m 15s","accounts_in_message":[],"_revision_number":2},{"id":"706e499e5ae4a14afa2e9edb83e9ccdd728c1bf8","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2022-11-29 18:27:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7c4b68350b1071702bdb631427e69bc342494df9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 18:46:07.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f6675a85814a4f578308c1cb9b4407b6\n\n- ui-cookiecutter-tox-checkbuild https://zuul.opendev.org/t/openstack/build/9b84933f66e4460b8bb4beea87605c2b : RETRY_LIMIT in 3m 13s","accounts_in_message":[],"_revision_number":2},{"id":"ffe5426377e26e2439a4afbcd3b5fce458a277c7","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2022-12-05 11:13:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3ef94f496dfc401af3ab38c1d25659fab91d0a93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-05 11:30:05.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c089b80016734d179283a64617c06270\n\n- ui-cookiecutter-tox-checkbuild https://zuul.opendev.org/t/openstack/build/064575494246483698f2300c02e4cfc1 : RETRY_LIMIT in 3m 16s","accounts_in_message":[],"_revision_number":2},{"id":"180bd11bfd01dc7a6526d84d58f6f31a41ce258d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2022-12-06 18:03:21.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2bbf7b2d26b0deee8949aaad3f2e198d4c13e236","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 18:17:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a9e80c0b095a47cca0a812156a93897f\n\n- ui-cookiecutter-tox-checkbuild https://zuul.opendev.org/t/openstack/build/cc0fa61a7a51497187d56a4d9f49ce4f : RETRY_LIMIT in 1m 21s","accounts_in_message":[],"_revision_number":3},{"id":"774110a27dd4ef27aaab28b558bea545b4e97a01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2022-12-06 18:30:14.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"366ad73304b49145d638a6594af9f664ed44ba6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2022-12-06 18:44:54.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"273aec14379cc2318d1a7f8b2dfd34891c232c59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 19:02:07.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fa3f4d1922ab4137986083d70452fcf4\n\n- ui-cookiecutter-tox-checkbuild https://zuul.opendev.org/t/openstack/build/850c6fb1c68d4187800d20d1677284e7 : RETRY_LIMIT in 1m 27s","accounts_in_message":[],"_revision_number":5},{"id":"bb80cb97a9e0422c199316eeca500d8337d2f018","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2022-12-06 19:23:48.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"7e99d330b34ed7aa46bde7938a9e3a351c02b641","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 19:39:09.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/da10613971434b74afeee6ea05c9b6c9\n\n- ui-cookiecutter-tox-checkbuild https://zuul.opendev.org/t/openstack/build/bbbd6285c85a4039ad90420f3ed37984 : RETRY_LIMIT in 3m 22s","accounts_in_message":[],"_revision_number":6},{"id":"5439d9b3e02f1f921e07b482d436564166abb7c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2022-12-07 03:17:17.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"73fb20e35f34d295d21458d35132cc007de94dcd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2022-12-07 03:37:36.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"cb45e61c208e0a12d35645a5f973e3fbda4a2ada","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2022-12-07 03:49:07.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"27804bcb48e0b7d8b17d2b8bac4add8b001198a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 04:06:29.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a8ca9ac28aea4aa7bb80381142baa8fb\n\n- ui-cookiecutter-tox-checkbuild https://zuul.opendev.org/t/openstack/build/582d9bc5d5884c84b9c9e81a25d4d749 : RETRY_LIMIT in 3m 59s","accounts_in_message":[],"_revision_number":9},{"id":"ba8b9a5e6d071a92ee4c2a2752196ed42dfe2bcf","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2025-12-02 14:25:39.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"03636edd0014a86b596fc0ceca5765a1fef45c3d","revisions":{"168c4e51b10a75afdc4b793d09a47febb9c83c9e":{"kind":"REWORK","_number":1,"created":"2022-11-29 08:32:09.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/53/865953/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ui-cookiecutter","ref":"refs/changes/53/865953/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/53/865953/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/53/865953/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/53/865953/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ui-cookiecutter refs/changes/53/865953/1"}}},"commit":{"parents":[{"commit":"8390e9d6d2feb933d9e153b2ef045fbda63197e6","subject":"Fix tox-checkbuild CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/8390e9d6d2feb933d9e153b2ef045fbda63197e6"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2022-11-29 08:31:19.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2022-11-29 08:32:05.000000000","tz":330},"subject":"[DNM] Test ui-cookiecutter CI","message":"[DNM] Test ui-cookiecutter CI\n\nChange-Id: Ia8ce252bf7ad6e5fb5cb92214c7d91d99d9ca8f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/168c4e51b10a75afdc4b793d09a47febb9c83c9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/168c4e51b10a75afdc4b793d09a47febb9c83c9e"}]},"branch":"refs/heads/master"},"67cf01cf28d93d1bb4b4ed97a2e5a96ed974c94a":{"kind":"REWORK","_number":2,"created":"2022-11-29 09:54:24.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/53/865953/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ui-cookiecutter","ref":"refs/changes/53/865953/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/53/865953/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/53/865953/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/53/865953/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ui-cookiecutter refs/changes/53/865953/2"}}},"commit":{"parents":[{"commit":"8390e9d6d2feb933d9e153b2ef045fbda63197e6","subject":"Fix tox-checkbuild CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/8390e9d6d2feb933d9e153b2ef045fbda63197e6"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2022-11-29 08:31:19.000000000","tz":330},"committer":{"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","date":"2022-11-29 09:54:24.000000000","tz":0},"subject":"[DNM] Test ui-cookiecutter CI","message":"[DNM] Test ui-cookiecutter CI\n\nChange-Id: Ia8ce252bf7ad6e5fb5cb92214c7d91d99d9ca8f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/67cf01cf28d93d1bb4b4ed97a2e5a96ed974c94a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/67cf01cf28d93d1bb4b4ed97a2e5a96ed974c94a"}]},"branch":"refs/heads/master"},"4e6a03773a2d9951acdd2af92bb97a4353fd2177":{"kind":"REWORK","_number":3,"created":"2022-12-06 18:03:21.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/53/865953/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ui-cookiecutter","ref":"refs/changes/53/865953/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/53/865953/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/53/865953/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/53/865953/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ui-cookiecutter refs/changes/53/865953/3"}}},"commit":{"parents":[{"commit":"8390e9d6d2feb933d9e153b2ef045fbda63197e6","subject":"Fix tox-checkbuild CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/8390e9d6d2feb933d9e153b2ef045fbda63197e6"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2022-11-29 08:31:19.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2022-12-06 18:03:17.000000000","tz":330},"subject":"[DNM] Test ui-cookiecutter CI","message":"[DNM] Test ui-cookiecutter CI\n\nChange-Id: Ia8ce252bf7ad6e5fb5cb92214c7d91d99d9ca8f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/4e6a03773a2d9951acdd2af92bb97a4353fd2177"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/4e6a03773a2d9951acdd2af92bb97a4353fd2177"}]},"branch":"refs/heads/master"},"94f93bd70d3ac5817173780ad3e588b951fa74c9":{"kind":"REWORK","_number":4,"created":"2022-12-06 18:30:14.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/53/865953/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ui-cookiecutter","ref":"refs/changes/53/865953/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/53/865953/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/53/865953/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/53/865953/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ui-cookiecutter refs/changes/53/865953/4"}}},"commit":{"parents":[{"commit":"8390e9d6d2feb933d9e153b2ef045fbda63197e6","subject":"Fix tox-checkbuild CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/8390e9d6d2feb933d9e153b2ef045fbda63197e6"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2022-11-29 08:31:19.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2022-12-06 18:30:00.000000000","tz":330},"subject":"[DNM] Test ui-cookiecutter CI","message":"[DNM] Test ui-cookiecutter CI\n\nChange-Id: Ia8ce252bf7ad6e5fb5cb92214c7d91d99d9ca8f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/94f93bd70d3ac5817173780ad3e588b951fa74c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/94f93bd70d3ac5817173780ad3e588b951fa74c9"}]},"branch":"refs/heads/master"},"77c820716feed563e9fd0b0045af36ba3a1cefa6":{"kind":"REWORK","_number":5,"created":"2022-12-06 18:44:54.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/53/865953/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ui-cookiecutter","ref":"refs/changes/53/865953/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/53/865953/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/53/865953/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/53/865953/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ui-cookiecutter refs/changes/53/865953/5"}}},"commit":{"parents":[{"commit":"8390e9d6d2feb933d9e153b2ef045fbda63197e6","subject":"Fix tox-checkbuild CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/8390e9d6d2feb933d9e153b2ef045fbda63197e6"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2022-11-29 08:31:19.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2022-12-06 18:44:41.000000000","tz":330},"subject":"[DNM] Test ui-cookiecutter CI","message":"[DNM] Test ui-cookiecutter CI\n\nChange-Id: Ia8ce252bf7ad6e5fb5cb92214c7d91d99d9ca8f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/77c820716feed563e9fd0b0045af36ba3a1cefa6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/77c820716feed563e9fd0b0045af36ba3a1cefa6"}]},"branch":"refs/heads/master"},"62ecf710ad5bbb4ce362df7c68f218aa9d00a659":{"kind":"REWORK","_number":6,"created":"2022-12-06 19:23:48.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/53/865953/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ui-cookiecutter","ref":"refs/changes/53/865953/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/53/865953/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/53/865953/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/53/865953/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ui-cookiecutter refs/changes/53/865953/6"}}},"commit":{"parents":[{"commit":"8390e9d6d2feb933d9e153b2ef045fbda63197e6","subject":"Fix tox-checkbuild CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/8390e9d6d2feb933d9e153b2ef045fbda63197e6"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2022-11-29 08:31:19.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2022-12-06 18:54:25.000000000","tz":330},"subject":"[DNM] Test ui-cookiecutter CI","message":"[DNM] Test ui-cookiecutter CI\n\nChange-Id: Ia8ce252bf7ad6e5fb5cb92214c7d91d99d9ca8f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/62ecf710ad5bbb4ce362df7c68f218aa9d00a659"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/62ecf710ad5bbb4ce362df7c68f218aa9d00a659"}]},"branch":"refs/heads/master"},"45e578a5cfb1c0e773a5ad4d286cb605021d7686":{"kind":"REWORK","_number":7,"created":"2022-12-07 03:17:17.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/53/865953/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ui-cookiecutter","ref":"refs/changes/53/865953/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/53/865953/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/53/865953/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/53/865953/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ui-cookiecutter refs/changes/53/865953/7"}}},"commit":{"parents":[{"commit":"8390e9d6d2feb933d9e153b2ef045fbda63197e6","subject":"Fix tox-checkbuild CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/8390e9d6d2feb933d9e153b2ef045fbda63197e6"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2022-11-29 08:31:19.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2022-12-07 03:17:03.000000000","tz":330},"subject":"[DNM] Test ui-cookiecutter CI","message":"[DNM] Test ui-cookiecutter CI\n\nChange-Id: Ia8ce252bf7ad6e5fb5cb92214c7d91d99d9ca8f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/45e578a5cfb1c0e773a5ad4d286cb605021d7686"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/45e578a5cfb1c0e773a5ad4d286cb605021d7686"}]},"branch":"refs/heads/master"},"1551550186776eb76066b81d92bd6828fc1e3f04":{"kind":"REWORK","_number":8,"created":"2022-12-07 03:37:36.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/53/865953/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ui-cookiecutter","ref":"refs/changes/53/865953/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/53/865953/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/53/865953/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/53/865953/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ui-cookiecutter refs/changes/53/865953/8"}}},"commit":{"parents":[{"commit":"8390e9d6d2feb933d9e153b2ef045fbda63197e6","subject":"Fix tox-checkbuild CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/8390e9d6d2feb933d9e153b2ef045fbda63197e6"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2022-11-29 08:31:19.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2022-12-07 03:37:26.000000000","tz":330},"subject":"[DNM] Test ui-cookiecutter CI","message":"[DNM] Test ui-cookiecutter CI\n\nChange-Id: Ia8ce252bf7ad6e5fb5cb92214c7d91d99d9ca8f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/1551550186776eb76066b81d92bd6828fc1e3f04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/1551550186776eb76066b81d92bd6828fc1e3f04"}]},"branch":"refs/heads/master"},"03636edd0014a86b596fc0ceca5765a1fef45c3d":{"kind":"REWORK","_number":9,"created":"2022-12-07 03:49:07.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/53/865953/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ui-cookiecutter","ref":"refs/changes/53/865953/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/53/865953/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/53/865953/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ui-cookiecutter refs/changes/53/865953/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ui-cookiecutter refs/changes/53/865953/9"}}},"commit":{"parents":[{"commit":"8390e9d6d2feb933d9e153b2ef045fbda63197e6","subject":"Fix tox-checkbuild CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/8390e9d6d2feb933d9e153b2ef045fbda63197e6"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2022-11-29 08:31:19.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2022-12-07 03:48:57.000000000","tz":330},"subject":"[DNM] Test ui-cookiecutter CI","message":"[DNM] Test ui-cookiecutter CI\n\nChange-Id: Ia8ce252bf7ad6e5fb5cb92214c7d91d99d9ca8f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/03636edd0014a86b596fc0ceca5765a1fef45c3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ui-cookiecutter/commit/03636edd0014a86b596fc0ceca5765a1fef45c3d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
