)]}'
{"id":"openstack%2Fvenus-dashboard~891007","triplet_id":"openstack%2Fvenus-dashboard~master~I47373bb4117c0bf61a9c6b024cabcea53f604925","project":"openstack/venus-dashboard","branch":"master","attention_set":{},"removed_from_attention_set":{"29602":{"account":{"_account_id":29602,"name":"cuizhengcheng","email":"cuizc@inspur.com","username":"cuizhengcheng"},"last_update":"2023-08-11 03:42:46.000000000","reason":"Change was submitted"},"31412":{"account":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"last_update":"2023-08-11 03:31:39.000000000","reason":"\u003cGERRIT_ACCOUNT_31412\u003e replied on the change","reason_account":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}},"30455":{"account":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"last_update":"2023-08-11 03:42:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I47373bb4117c0bf61a9c6b024cabcea53f604925","subject":"Fix bug: log analyse chart data get failed","status":"MERGED","created":"2023-08-10 02:43:03.000000000","updated":"2023-08-11 03:42:46.000000000","submitted":"2023-08-11 03:42:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"891007","meta_rev_id":"074d6d697c1dd8e5af4ed04d2459a323da629f28","_number":891007,"virtual_id_number":891007,"owner":{"_account_id":29602,"name":"cuizhengcheng","email":"cuizc@inspur.com","username":"cuizhengcheng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-11 03:42:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"all":[{"value":2,"date":"2023-08-11 03:31:39.000000000","_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"all":[{"value":1,"date":"2023-08-11 03:31:39.000000000","_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-10 02:47:33.000000000","updated_by":{"_account_id":29602,"name":"cuizhengcheng","email":"cuizc@inspur.com","username":"cuizhengcheng"},"reviewer":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"state":"REVIEWER"},{"updated":"2023-08-10 02:47:33.000000000","updated_by":{"_account_id":29602,"name":"cuizhengcheng","email":"cuizc@inspur.com","username":"cuizhengcheng"},"reviewer":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"state":"REVIEWER"},{"updated":"2023-08-10 03:00:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"257d8fac19388ff41d564a28b0c06e3000f05b74","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29602,"name":"cuizhengcheng","email":"cuizc@inspur.com","username":"cuizhengcheng"},"date":"2023-08-10 02:43:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"88c12e2e48e2f660ee75ed6b30b8d9c606ec13da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 03:00:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36547f7216e547e5943fc4db12e48029\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5abc07c6e7d4650a0bcbc4114920402 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9eeebb3a39814a0fb130739d72df961f : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ffe2f9b7ef34163bec92573c0933576 : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0a544ae723ed4102857ab8448a3cd355 : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":1},{"id":"83b4ab300489593a04760e9314a0d688e8229668","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2023-08-10 03:03:41.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fba3e03f73c46b17c714c0659af1f7881781c4ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29602,"name":"cuizhengcheng","email":"cuizc@inspur.com","username":"cuizhengcheng"},"date":"2023-08-10 03:17:13.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"797d45683afcd82dc06f4cb69312651f498c8d61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 03:27:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41735238b33c4090b4e05b07da65bed2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da09a93be36d46f79546d6b106ba9b07 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b3d273e19ae4af98c4b29b5226fa01a : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d671962f72b9407f9c9137a6b498ab80 : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f9b1c7f7640143e7aaea6904b08e6ee4 : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":2},{"id":"d3503c5314495051c865dd83c9626917c2f9d08e","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2023-08-10 05:35:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5402b68a2629ba77799d1e29bea954f7e0b1c0fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29602,"name":"cuizhengcheng","email":"cuizc@inspur.com","username":"cuizhengcheng"},"date":"2023-08-11 02:58:19.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"9ff562059df94bb1f6f599039890652254967f2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-11 03:14:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea81fbf275a04953b58a4d34c50391fd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/956f5e3fece6477ba3d1b6b71d3c7786 : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/53da31756a234d09a4861d8ac31f1a9a : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ca9858ce0d434a26b06bc3910f990a17 : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/21deab74b6df432ea6e2fb07e8dd57d1 : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":3},{"id":"6df388e2164d8cd8b807ef36a87ede801e5d7df8","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2023-08-11 03:31:39.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c4b59a2e8607ed5c042b29a2913a2b74314cbac8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-11 03:32:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"dad42d715594e146707669ab0f2a1ab9584e98b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-11 03:42:46.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a84adfc6f2e54725acf75dd879c2d58f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/095e36e86d4a4dfdb7170e32d7481d05 : SUCCESS in 3m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ec4dcab04154a4089a8c3bd8d66d36f : SUCCESS in 3m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2118e9c71f7944b2a500456e3f3b35ab : SUCCESS in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/387ef76a5cf04751aa343e97627313ea : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":3},{"id":"074d6d697c1dd8e5af4ed04d2459a323da629f28","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-11 03:42:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c4e614f1c249d65395efb14e31fad2c12e8fba34","revisions":{"6368fbbcb9057f2c7fa37c53b8f9c7a10de1176c":{"kind":"REWORK","_number":1,"created":"2023-08-10 02:43:03.000000000","uploader":{"_account_id":29602,"name":"cuizhengcheng","email":"cuizc@inspur.com","username":"cuizhengcheng"},"ref":"refs/changes/07/891007/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus-dashboard","ref":"refs/changes/07/891007/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus-dashboard refs/changes/07/891007/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus-dashboard refs/changes/07/891007/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus-dashboard refs/changes/07/891007/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus-dashboard refs/changes/07/891007/1"}}},"commit":{"parents":[{"commit":"27182dd698aa71368d28a69bafbd9e7e24ce4af6","subject":"log analysis api params change","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus-dashboard/commit/27182dd698aa71368d28a69bafbd9e7e24ce4af6"}]}],"author":{"name":"cuizc","email":"cuizc@inspur.com","date":"2023-08-10 02:42:33.000000000","tz":480},"committer":{"name":"cuizc","email":"cuizc@inspur.com","date":"2023-08-10 02:42:33.000000000","tz":480},"subject":"log analyse chart data get failed: bug2030857","message":"log analyse chart data get failed: bug2030857\n\nChange-Id: I47373bb4117c0bf61a9c6b024cabcea53f604925\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus-dashboard/commit/6368fbbcb9057f2c7fa37c53b8f9c7a10de1176c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus-dashboard/commit/6368fbbcb9057f2c7fa37c53b8f9c7a10de1176c"}]},"branch":"refs/heads/master"},"36fcce49e11e91f3fa0b345813e9c3eb3e76ae0e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-08-10 03:17:13.000000000","uploader":{"_account_id":29602,"name":"cuizhengcheng","email":"cuizc@inspur.com","username":"cuizhengcheng"},"ref":"refs/changes/07/891007/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus-dashboard","ref":"refs/changes/07/891007/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus-dashboard refs/changes/07/891007/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus-dashboard refs/changes/07/891007/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus-dashboard refs/changes/07/891007/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus-dashboard refs/changes/07/891007/2"}}},"commit":{"parents":[{"commit":"27182dd698aa71368d28a69bafbd9e7e24ce4af6","subject":"log analysis api params change","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus-dashboard/commit/27182dd698aa71368d28a69bafbd9e7e24ce4af6"}]}],"author":{"name":"cuizc","email":"cuizc@inspur.com","date":"2023-08-10 02:42:33.000000000","tz":480},"committer":{"name":"cuizc","email":"cuizc@inspur.com","date":"2023-08-10 03:13:43.000000000","tz":480},"subject":"Fix bug: log analyse chart data get failed","message":"Fix bug: log analyse chart data get failed\n\nCloses-Bug: #2030857\n\nChange-Id: I47373bb4117c0bf61a9c6b024cabcea53f604925\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus-dashboard/commit/36fcce49e11e91f3fa0b345813e9c3eb3e76ae0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus-dashboard/commit/36fcce49e11e91f3fa0b345813e9c3eb3e76ae0e"}]},"branch":"refs/heads/master"},"c4e614f1c249d65395efb14e31fad2c12e8fba34":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-08-11 02:58:19.000000000","uploader":{"_account_id":29602,"name":"cuizhengcheng","email":"cuizc@inspur.com","username":"cuizhengcheng"},"ref":"refs/changes/07/891007/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus-dashboard","ref":"refs/changes/07/891007/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus-dashboard refs/changes/07/891007/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus-dashboard refs/changes/07/891007/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus-dashboard refs/changes/07/891007/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus-dashboard refs/changes/07/891007/3"}}},"commit":{"parents":[{"commit":"7298b52dc9ce5c448789de76072aad4c7f3f5647","subject":"log Error feature develop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus-dashboard/commit/7298b52dc9ce5c448789de76072aad4c7f3f5647"}]}],"author":{"name":"cuizc","email":"cuizc@inspur.com","date":"2023-08-10 02:42:33.000000000","tz":480},"committer":{"name":"cuizhengcheng","email":"cuizc@inspur.com","date":"2023-08-11 02:58:19.000000000","tz":0},"subject":"Fix bug: log analyse chart data get failed","message":"Fix bug: log analyse chart data get failed\n\nCloses-Bug: #2030857\n\nChange-Id: I47373bb4117c0bf61a9c6b024cabcea53f604925\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus-dashboard/commit/c4e614f1c249d65395efb14e31fad2c12e8fba34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus-dashboard/commit/c4e614f1c249d65395efb14e31fad2c12e8fba34"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
