)]}'
{"id":"openstack%2Fvenus-tempest-plugin~877924","triplet_id":"openstack%2Fvenus-tempest-plugin~master~I038a6610a348c46078540de0a6477ccc9ae9c5d7","project":"openstack/venus-tempest-plugin","branch":"master","topic":"master5","attention_set":{"30409":{"account":{"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},"last_update":"2023-03-21 03:17:02.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I038a6610a348c46078540de0a6477ccc9ae9c5d7","subject":"add openstack-tox-docs template","status":"NEW","created":"2023-03-21 03:06:51.000000000","updated":"2023-03-21 03:17:02.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"90eb7362250767b2dc97428f8e6fc06dff934455","_number":877924,"virtual_id_number":877924,"owner":{"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-03-21 03:17:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-21 03:17:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7ef9a8219f35ab69ebf49148009bf40ce2c552d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},"date":"2023-03-21 03:06:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"90eb7362250767b2dc97428f8e6fc06dff934455","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-21 03:17:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3a2fa404a8c848b19cb281192be260f0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a2909887fb340a8882530f48b66cd72 : FAILURE in 2m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad631a84ed6a4be9af29ce1081144cc4 : SUCCESS in 2m 32s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"976fc81ff63c588cc9cc68d3fccb4c9b42d6f19d","revisions":{"976fc81ff63c588cc9cc68d3fccb4c9b42d6f19d":{"kind":"REWORK","_number":1,"created":"2023-03-21 03:06:51.000000000","uploader":{"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},"ref":"refs/changes/24/877924/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus-tempest-plugin","ref":"refs/changes/24/877924/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus-tempest-plugin refs/changes/24/877924/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus-tempest-plugin refs/changes/24/877924/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus-tempest-plugin refs/changes/24/877924/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus-tempest-plugin refs/changes/24/877924/1"}}},"commit":{"parents":[{"commit":"e8103d86456eb2b42cfb22108e35d2e9d899136f","subject":"delete invalid test file and cyborg api service","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus-tempest-plugin/commit/e8103d86456eb2b42cfb22108e35d2e9d899136f"}]}],"author":{"name":"wangzhiguang","email":"wangzhiguang@inspur.com","date":"2023-03-21 03:05:13.000000000","tz":480},"committer":{"name":"wangzhiguang","email":"wangzhiguang@inspur.com","date":"2023-03-21 03:05:20.000000000","tz":480},"subject":"add openstack-tox-docs template","message":"add openstack-tox-docs template\n\nadd openstack-tox-docs template in .zuul.yaml file\n\nChange-Id: I038a6610a348c46078540de0a6477ccc9ae9c5d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus-tempest-plugin/commit/976fc81ff63c588cc9cc68d3fccb4c9b42d6f19d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus-tempest-plugin/commit/976fc81ff63c588cc9cc68d3fccb4c9b42d6f19d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
