)]}'
{"id":"openstack%2Fvenus-tempest-plugin~933359","triplet_id":"openstack%2Fvenus-tempest-plugin~master~Ia1e92968d88f64b78e8adc4a747bb0377251712d","project":"openstack/venus-tempest-plugin","branch":"master","topic":"support-py312","attention_set":{},"removed_from_attention_set":{"31412":{"account":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"last_update":"2024-12-05 18:19:56.000000000","reason":"Change was submitted"},"30455":{"account":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"last_update":"2024-12-05 00:56:55.000000000","reason":"\u003cGERRIT_ACCOUNT_30455\u003e replied on the change","reason_account":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-12-05 18:05:49.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"Ia1e92968d88f64b78e8adc4a747bb0377251712d","subject":"Support py3.12 and drop py3.8","status":"MERGED","created":"2024-10-24 19:20:27.000000000","updated":"2024-12-05 18:19:56.000000000","submitted":"2024-12-05 18:19:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933359-support-py312","meta_rev_id":"114712c1f70ccc80aa6df5a521e6e93a69945435","_number":933359,"virtual_id_number":933359,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-05 18:19:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"all":[{"value":2,"date":"2024-12-05 00:56:55.000000000","_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"all":[{"value":1,"date":"2024-12-05 00:56:55.000000000","_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-24 19:43:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-04 19:39:13.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"state":"REVIEWER"},{"updated":"2024-12-04 19:39:27.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"state":"REVIEWER"}],"messages":[{"id":"13cdceb8db39ba52ac1bbd746ae2ec59481e03a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-10-24 19:20:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e823d453d884685e8c63afa57f55c9da12d0c28a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 19:43:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb2a11414437467a9fec9386aa7fffd6\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/37a51df781e84117857ab39e8735eee3 : SUCCESS in 2m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5920b9f7d58f4f0192997ca3ababbde5 : SUCCESS in 2m 10s","accounts_in_message":[],"_revision_number":1},{"id":"f94582e93a2447766d51fa4165a7c5209d06cc3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-10-25 22:25:18.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"051906318e0ca9c39bd3f36e1ae84ef53272e3a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 22:32:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eafc5eaec4094c5a8dcb89605d6bf9c6\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b39ba2f00407442fb63f0dfe4810d2b6 : SUCCESS in 1m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0413e7efb48b4b85a97ced7f176e34b7 : SUCCESS in 2m 08s","accounts_in_message":[],"_revision_number":2},{"id":"f690cb2e485f5816fa161772137411ab4ce92b09","author":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"date":"2024-12-05 00:56:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"91a2c0e8f055d58f64599be46aebe27852467898","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 00:57:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e2a137b7c5b4eb6a47167b64d37f49c96cfa7dfc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 01:03:40.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b449db3f11c24580a3f9044f0c7b634b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c566f91760dd4e01adc31bfaaa7dd3e0 : FAILURE in 1m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67541adc233f4a14a429cdf9e853a552 : SUCCESS in 2m 27s","accounts_in_message":[],"_revision_number":2},{"id":"0bb876826eec88a7ef7d92ab16cbb19565988dd1","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-12-05 18:05:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c0a1e57972047ad80a5d45e6ce348376b32dc16f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 18:12:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d679c5840b5430a85b27ebd94a76520\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/22a2c604443e4c638f44323ef058e910 : SUCCESS in 1m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e669c5fcd8584ee6841937b5d66cd186 : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":2},{"id":"e674a6fa23fbf6b7837f9a14687604b88e0c3e12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 18:12:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4ec2ee7475d03a635d8e718ddcb1d146eb0d6e08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 18:19:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c88253e4cf334d968fa2aebef6f52eee\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ac90fd843ef24a87a03584b0c531d65b : SUCCESS in 1m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/704ae9e065b54b4a81194895b8b88f70 : SUCCESS in 2m 19s","accounts_in_message":[],"_revision_number":2},{"id":"114712c1f70ccc80aa6df5a521e6e93a69945435","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 18:19:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b0dae6ef0cb8643c5b9dea9b74e53d097cd29c02","revisions":{"6a5caffc798b69c8979d965d1ef3a2f92a1d9846":{"kind":"REWORK","_number":1,"created":"2024-10-24 19:20:27.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/59/933359/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus-tempest-plugin","ref":"refs/changes/59/933359/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus-tempest-plugin refs/changes/59/933359/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus-tempest-plugin refs/changes/59/933359/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus-tempest-plugin refs/changes/59/933359/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus-tempest-plugin refs/changes/59/933359/1"}}},"commit":{"parents":[{"commit":"af1753e6d5ce043543c321ba048852d26e6eae06","subject":"Merge \"add venus api testcase for list api versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus-tempest-plugin/commit/af1753e6d5ce043543c321ba048852d26e6eae06"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-10-24 19:03:16.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-10-24 19:20:26.000000000","tz":-420},"subject":"Update python classifier in setup.cfg","message":"Update python classifier in setup.cfg\n\nIn tempest plugin, we test from python 3.8 till\npython 3.12 so updating the same in python\nclassifier in setup.cfg\n\nChange-Id: Ia1e92968d88f64b78e8adc4a747bb0377251712d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus-tempest-plugin/commit/6a5caffc798b69c8979d965d1ef3a2f92a1d9846"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus-tempest-plugin/commit/6a5caffc798b69c8979d965d1ef3a2f92a1d9846"}]},"branch":"refs/heads/master"},"b0dae6ef0cb8643c5b9dea9b74e53d097cd29c02":{"kind":"REWORK","_number":2,"created":"2024-10-25 22:25:18.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/59/933359/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus-tempest-plugin","ref":"refs/changes/59/933359/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus-tempest-plugin refs/changes/59/933359/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus-tempest-plugin refs/changes/59/933359/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus-tempest-plugin refs/changes/59/933359/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus-tempest-plugin refs/changes/59/933359/2"}}},"commit":{"parents":[{"commit":"af1753e6d5ce043543c321ba048852d26e6eae06","subject":"Merge \"add venus api testcase for list api versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus-tempest-plugin/commit/af1753e6d5ce043543c321ba048852d26e6eae06"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-10-24 19:03:16.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-10-25 22:12:53.000000000","tz":-420},"subject":"Support py3.12 and drop py3.8","message":"Support py3.12 and drop py3.8\n\nAs per the current release tested runtime, we test\nfrom python 3.9 till python 3.12 so updating the same\nin python classifier in setup.cfg\n\nAlso, dropping the python3.8 support\n- https://lists.openstack.org/archives/list/openstack-discuss@li\u003e\n\nChange-Id: Ia1e92968d88f64b78e8adc4a747bb0377251712d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus-tempest-plugin/commit/b0dae6ef0cb8643c5b9dea9b74e53d097cd29c02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus-tempest-plugin/commit/b0dae6ef0cb8643c5b9dea9b74e53d097cd29c02"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
