)]}'
{"id":"openstack%2Fvenus~823040","triplet_id":"openstack%2Fvenus~master~Ic5007eab3bfd843d3147f16e28d721a0c1928df4","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"30562":{"account":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"last_update":"2021-12-28 06:27:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic5007eab3bfd843d3147f16e28d721a0c1928df4","subject":"Add missing elasticsearch module","status":"MERGED","created":"2021-12-28 05:48:16.000000000","updated":"2021-12-28 06:28:44.000000000","submitted":"2021-12-28 06:27:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823040","meta_rev_id":"9158f7e299e6b97434514ac2959eed75450d65b3","_number":823040,"virtual_id_number":823040,"owner":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-28 06:27:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"all":[{"value":2,"date":"2021-12-28 06:22:19.000000000","_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"all":[{"value":1,"date":"2021-12-28 06:22:19.000000000","_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-28 05:52:23.000000000","updated_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"reviewer":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"state":"REVIEWER"},{"updated":"2021-12-28 05:53:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4d0cd19cc502da39584ccda1acef2fd6ca319d80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"date":"2021-12-28 05:48:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e39140f5af5e059c16178b765a317e850cc96a98","author":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"date":"2021-12-28 05:52:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"08e4fa286ed6a4247f44e646b1fb80e06d8dde78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-28 05:53:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/9a5a59128cb046f1bf06eb16f1153b73 : FAILURE in 2m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb2853d5bd024829a683c10079dbf53d : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f42f3fc97279481bb2cee7186122c33c : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ea130b3119f4ca9998ad1a0d479d790 : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fe69a319da1446e8a22bdacc27b52f4 : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":1},{"id":"a4178972259592a5fd910242911fa03c3ba281ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"date":"2021-12-28 06:03:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3977be29ea0ba0827d10a498000620de90dd7d90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-28 06:13:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d52a1208b9184571a7c6fb6517e5f4b8 : FAILURE in 2m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e065bd3c5e304ecb93b7600a07959861 : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b3110fc3f804e2688750ace0aefbd16 : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c359817f95ab4632a62db0d69f0d4a1b : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9c9fa4377b74fb78722b0d280b095fe : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":2},{"id":"571f6dfcd3e31a3e3447ecafd3c5d912bcdc2f2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"date":"2021-12-28 06:15:34.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a230439699b51ec844bd956f16d6c80ab4da6294","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-28 06:21:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/29363cd14a8a4f8493b2fb64b290deeb : SUCCESS in 2m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36c6038356a14267bb978d2e95538306 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4978b35191284b2eac19369aee55e659 : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f23556ef00bf4574a17a624913e2643e : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51e98ce6647e4577ba39eef562b75b9a : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":3},{"id":"a06661f21331fc8ccbb9c656b472864036482a83","author":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"date":"2021-12-28 06:22:19.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4b3b9ec6f5c83d3ae8cef766f9dc02c6ad2e1583","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-28 06:22:42.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"04717922417f0614fd849c293482543cbfc1f08e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-28 06:27:19.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ed439bbde35b432e992c365b69ef95cd : SUCCESS in 2m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de6ce23ab5dd49a0baf0d78d52c4f6af : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32775a9376904a62a619c5760eff4e25 : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3a0360d8fdd34e26a3b3e1808ca6b142 : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80627c82e4974b0783aab50b5072b430 : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":3},{"id":"42cdc443806a156912016649c578a1b6eecefefc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-28 06:27:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"9158f7e299e6b97434514ac2959eed75450d65b3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-28 06:28:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/856864753bd3480191837b4cd83541d7 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"fd93810f55d12e78259fef4f3854eaa80c874408","revisions":{"be9e2da7c33ad3795f95382c20124597405f9b76":{"kind":"REWORK","_number":1,"created":"2021-12-28 05:48:16.000000000","uploader":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"ref":"refs/changes/40/823040/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/40/823040/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/40/823040/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/40/823040/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/40/823040/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/40/823040/1"}}},"commit":{"parents":[{"commit":"4ce191206b871f6af951b26622eb8c498ec8dc4c","subject":"Merge \"Completment readme doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/4ce191206b871f6af951b26622eb8c498ec8dc4c"}]}],"author":{"name":"baiziyu","email":"baiziyu@inspur.com","date":"2021-12-28 05:39:16.000000000","tz":480},"committer":{"name":"baiziyu","email":"baiziyu@inspur.com","date":"2021-12-28 05:48:02.000000000","tz":480},"subject":"Add missing elasticsearch5 module","message":"Add missing elasticsearch5 module\n\nChange-Id: Ic5007eab3bfd843d3147f16e28d721a0c1928df4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/be9e2da7c33ad3795f95382c20124597405f9b76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/be9e2da7c33ad3795f95382c20124597405f9b76"}]},"branch":"refs/heads/master"},"aed3c887086d3f11cd255ddd2788cacb75d92832":{"kind":"REWORK","_number":2,"created":"2021-12-28 06:03:21.000000000","uploader":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"ref":"refs/changes/40/823040/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/40/823040/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/40/823040/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/40/823040/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/40/823040/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/40/823040/2"}}},"commit":{"parents":[{"commit":"4ce191206b871f6af951b26622eb8c498ec8dc4c","subject":"Merge \"Completment readme doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/4ce191206b871f6af951b26622eb8c498ec8dc4c"}]}],"author":{"name":"baiziyu","email":"baiziyu@inspur.com","date":"2021-12-28 05:39:16.000000000","tz":480},"committer":{"name":"baiziyu","email":"baiziyu@inspur.com","date":"2021-12-28 06:03:02.000000000","tz":480},"subject":"Add missing elasticsearch module","message":"Add missing elasticsearch module\n\nChange-Id: Ic5007eab3bfd843d3147f16e28d721a0c1928df4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/aed3c887086d3f11cd255ddd2788cacb75d92832"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/aed3c887086d3f11cd255ddd2788cacb75d92832"}]},"branch":"refs/heads/master"},"fd93810f55d12e78259fef4f3854eaa80c874408":{"kind":"REWORK","_number":3,"created":"2021-12-28 06:15:34.000000000","uploader":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"ref":"refs/changes/40/823040/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/40/823040/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/40/823040/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/40/823040/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/40/823040/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/40/823040/3"}}},"commit":{"parents":[{"commit":"4ce191206b871f6af951b26622eb8c498ec8dc4c","subject":"Merge \"Completment readme doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/4ce191206b871f6af951b26622eb8c498ec8dc4c"}]}],"author":{"name":"baiziyu","email":"baiziyu@inspur.com","date":"2021-12-28 05:39:16.000000000","tz":480},"committer":{"name":"baiziyu","email":"baiziyu@inspur.com","date":"2021-12-28 06:15:21.000000000","tz":480},"subject":"Add missing elasticsearch module","message":"Add missing elasticsearch module\n\nChange-Id: Ic5007eab3bfd843d3147f16e28d721a0c1928df4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/fd93810f55d12e78259fef4f3854eaa80c874408"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/fd93810f55d12e78259fef4f3854eaa80c874408"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
